)]}'
{"id":"openstack%2Fcyborg~977887","triplet_id":"openstack%2Fcyborg~master~I33c7a27efa9afb379392c55e88787e2cb74cdacc","project":"openstack/cyborg","branch":"master","topic":"linters","attention_set":{},"removed_from_attention_set":{"11604":{"account":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"last_update":"2026-02-25 21:52:45.000000000","reason":"\u003cGERRIT_ACCOUNT_11604\u003e replied on the change","reason_account":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"}}},"hashtags":[],"change_id":"I33c7a27efa9afb379392c55e88787e2cb74cdacc","subject":"Fix pyproject.toml deprecations and remove wsgi_scripts","status":"ABANDONED","created":"2026-02-24 21:52:05.000000000","updated":"2026-03-02 18:19:33.000000000","total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"6e055fc1a728cf8fd6e4da5f1d2a8b11041dc14b","_number":977887,"virtual_id_number":977887,"owner":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2026-02-24 22:34:07.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"disliked":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"all":[{"value":-1,"date":"2026-02-25 21:52:45.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":-1,"default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":28006,"name":"teim-ci","display_name":"teim-ci","email":"ci@seanmooney.info","username":"ci-sean-mooney","status":"this is a third-party ci account run by sean-k-mooney on irc\nhosted at zuul.teim.app"}],"REVIEWER":[{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-02-24 21:52:56.000000000","updated_by":{"_account_id":28006,"name":"teim-ci","display_name":"teim-ci","email":"ci@seanmooney.info","username":"ci-sean-mooney","status":"this is a third-party ci account run by sean-k-mooney on irc\nhosted at zuul.teim.app"},"reviewer":{"_account_id":28006,"name":"teim-ci","display_name":"teim-ci","email":"ci@seanmooney.info","username":"ci-sean-mooney","status":"this is a third-party ci account run by sean-k-mooney on irc\nhosted at zuul.teim.app"},"state":"CC"},{"updated":"2026-02-24 22:34:07.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"1e27f6b51ab12b856297ed2d90f5201c2fe9196d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2026-02-24 21:52:05.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"20fc9f55647b3058e8089635b5c8ef67d20c7013","tag":"autogenerated:zuul:automatic-ci","author":{"_account_id":28006,"name":"teim-ci","display_name":"teim-ci","email":"ci@seanmooney.info","username":"ci-sean-mooney","status":"this is a third-party ci account run by sean-k-mooney on irc\nhosted at zuul.teim.app"},"date":"2026-02-24 21:52:56.000000000","message":"Patch Set 1:\n\nStarting automatic-ci jobs.","accounts_in_message":[],"_revision_number":1},{"id":"f240dea76f45b843ff18bdb22c0d0a467dda431d","tag":"autogenerated:zuul:automatic-ci","author":{"_account_id":28006,"name":"teim-ci","display_name":"teim-ci","email":"ci@seanmooney.info","username":"ci-sean-mooney","status":"this is a third-party ci account run by sean-k-mooney on irc\nhosted at zuul.teim.app"},"date":"2026-02-24 22:00:56.000000000","message":"Patch Set 1:\n\n(1 comment)\n\nBuild succeeded (automatic-ci pipeline).\nhttps://zuul.teim.app/t/main/buildset/686e2a7314ef4407a9201b96f683f03a\n\n- teim-code-review https://zuul.teim.app/t/main/build/31291155ef0e421a8d4a0fb790c6f155 : SUCCESS in 7m 49s\n\nWarning:\n  Comments left for invalid file doc/source/admin/config-wsgi.rst","accounts_in_message":[],"_revision_number":1},{"id":"dcb5abd7674a7bfb1fffb2fe6332beeb12eba3c3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-24 22:34:07.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/f954275302124a9c94ef1ac1bae1f795\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/c793d97f4b5744a9b366d2efb3e2ccbc : FAILURE in 1m 52s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a04a782a486b4dcbab5a337e48809307 : SUCCESS in 5m 06s\n- requirements-check https://zuul.opendev.org/t/openstack/build/5b453f94032a44b8a94c556d9b55c166 : SUCCESS in 2m 10s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/49d07a2e34ef468a908151107a18af1d : SUCCESS in 3m 13s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e2d6dc5051624913827452f6e7ce3582 : SUCCESS in 8m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0d36f823f1a5441587821d24d57ebe8b : SUCCESS in 4m 05s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/d37aaa8f3419406fabe06f3a44e42f73 : SUCCESS in 3m 15s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/f342d3c93223433cbba4527226dfe0e4 : SUCCESS in 4m 01s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/fd206b069f5f4b51b2e69bbef74903bf : SUCCESS in 4m 45s (non-voting)\n- cyborg-tempest https://zuul.opendev.org/t/openstack/build/b69ffa340baf45e2bb90afc05e5468da : SUCCESS in 30m 59s\n- cyborg-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/664f303aec164e5584974852a73e3f05 : SUCCESS in 31m 07s (non-voting)\n- cyborg-tox-bandit https://zuul.opendev.org/t/openstack/build/983d4d930abf41bfac4a0f76040be9ef : SUCCESS in 3m 46s","accounts_in_message":[],"_revision_number":1},{"id":"96c2844c427ac7e61cf0b3fd1b728bddcd349a97","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2026-02-25 21:52:45.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"6e055fc1a728cf8fd6e4da5f1d2a8b11041dc14b","tag":"autogenerated:gerrit:abandon","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2026-03-02 18:19:33.000000000","message":"Abandoned\n\nno longer requried. i can reivve this after our build pipleine is moved to somethign mroe moderen","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"854f04913802cecc925748fd78d9be2c8de6c1a4","revisions":{"854f04913802cecc925748fd78d9be2c8de6c1a4":{"kind":"REWORK","_number":1,"created":"2026-02-24 21:52:05.000000000","uploader":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"ref":"refs/changes/87/977887/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cyborg","ref":"refs/changes/87/977887/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/87/977887/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/87/977887/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/87/977887/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cyborg refs/changes/87/977887/1"}}},"commit":{"parents":[{"commit":"686d0f20b650932c368c4dac61e260f74f49b96e","subject":"Migrate project metadata to pyproject.toml","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/686d0f20b650932c368c4dac61e260f74f49b96e"}]}],"author":{"name":"Sean Mooney","email":"work@seanmooney.info","date":"2026-02-24 20:19:37.000000000","tz":0},"committer":{"name":"Sean Mooney","email":"work@seanmooney.info","date":"2026-02-24 20:19:46.000000000","tz":0},"subject":"Fix pyproject.toml deprecations and remove wsgi_scripts","message":"Fix pyproject.toml deprecations and remove wsgi_scripts\n\n- Change license field from TOML table syntax to plain SPDX\n  string (\"Apache-2.0\") as required by setuptools \u003e\u003d 77\n- Bump setuptools minimum to 77.0.0 to match\n- Remove redundant \"License :: OSI Approved :: Apache Software\n  License\" classifier (deprecated when SPDX license field present)\n- Remove wsgi_scripts entry point (cyborg-wsgi-api); deployers\n  should reference cyborg.wsgi.api:application directly\n\nGenerated-By: claude-code sonnet 4.6\nChange-Id: I33c7a27efa9afb379392c55e88787e2cb74cdacc\nSigned-off-by: Sean Mooney \u003cwork@seanmooney.info\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/854f04913802cecc925748fd78d9be2c8de6c1a4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/854f04913802cecc925748fd78d9be2c8de6c1a4"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
