)]}'
{"id":"openstack%2Fcyborg~979788","triplet_id":"openstack%2Fcyborg~master~I6c54beceee3dabb38eb581afde92c1c2b427c9e8","project":"openstack/cyborg","branch":"master","attention_set":{"12393":{"account":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"last_update":"2026-03-10 12:03:45.000000000","reason":"Someone else replied on the change"}},"removed_from_attention_set":{},"hashtags":[],"change_id":"I6c54beceee3dabb38eb581afde92c1c2b427c9e8","subject":"Added ansible-lint and bashate pre-commit hook","status":"NEW","created":"2026-03-10 07:12:48.000000000","updated":"2026-03-10 12:03:45.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":false,"submittable":false,"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"274b01cb8dff5880d76de082771a47e13873dae9","_number":979788,"virtual_id_number":979788,"owner":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2026-03-10 11:56:08.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":28006,"name":"teim-ci","display_name":"teim-ci","email":"ci@seanmooney.info","username":"ci-sean-mooney","status":"this is a third-party ci account run by sean-k-mooney on irc\nhosted at zuul.teim.app"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-03-10 07:12:55.000000000","updated_by":{"_account_id":28006,"name":"teim-ci","display_name":"teim-ci","email":"ci@seanmooney.info","username":"ci-sean-mooney","status":"this is a third-party ci account run by sean-k-mooney on irc\nhosted at zuul.teim.app"},"reviewer":{"_account_id":28006,"name":"teim-ci","display_name":"teim-ci","email":"ci@seanmooney.info","username":"ci-sean-mooney","status":"this is a third-party ci account run by sean-k-mooney on irc\nhosted at zuul.teim.app"},"state":"CC"},{"updated":"2026-03-10 07:48:22.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"860535bfeb7c125f19469197d5ff068dc7637e2f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"date":"2026-03-10 07:12:48.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"29e95f254701fc5b3f90d7912c7ba5bc688bd1fe","tag":"autogenerated:zuul:automatic-ci","author":{"_account_id":28006,"name":"teim-ci","display_name":"teim-ci","email":"ci@seanmooney.info","username":"ci-sean-mooney","status":"this is a third-party ci account run by sean-k-mooney on irc\nhosted at zuul.teim.app"},"date":"2026-03-10 07:12:55.000000000","message":"Patch Set 1:\n\nStarting automatic-ci jobs.","accounts_in_message":[],"_revision_number":1},{"id":"8d36afdb8f25d0c9ad3aac666d3aa704240bb609","tag":"autogenerated:zuul:automatic-ci","author":{"_account_id":28006,"name":"teim-ci","display_name":"teim-ci","email":"ci@seanmooney.info","username":"ci-sean-mooney","status":"this is a third-party ci account run by sean-k-mooney on irc\nhosted at zuul.teim.app"},"date":"2026-03-10 07:27:38.000000000","message":"Patch Set 1:\n\n(5 comments)\n\nBuild succeeded (automatic-ci pipeline).\nhttps://zuul.teim.app/t/main/buildset/efe9738a22604714841a6310ab9c791c\n\n- teim-code-review https://zuul.teim.app/t/main/build/83b0c9d3ce5b4958af98882dc6cf0e72 : SUCCESS in 14m 29s\n\nWarning:\n  Comments left for invalid file devstack/plugin.sh","accounts_in_message":[],"_revision_number":1},{"id":"174ab5d19eb34ac866dfa4518e914ef73818dfe3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-10 07:48:22.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/563ec7c7bf6c4aac8dcd37ad295cdc5a\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c001ab42bbf64c74929b8789a91fbb60 : SUCCESS in 4m 08s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/68c400d1ab82433082e6cfd0476c52d5 : SUCCESS in 7m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/04f4e2603e974118a5359e149456dfe1 : SUCCESS in 2m 33s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/96d5dad2dfef4b23ae4f156ee59bc4ba : SUCCESS in 2m 14s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/c1581964f9d242a68bff26ba8d549b03 : SUCCESS in 4m 13s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/4e67624a3d124f2aaa3a23e12ac440f7 : SUCCESS in 5m 57s (non-voting)\n- cyborg-tempest https://zuul.opendev.org/t/openstack/build/0b5484fae8154530a73de5529ac66a89 : SUCCESS in 29m 09s\n- cyborg-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/ea3646069330447f965e82bcafd41fa0 : SUCCESS in 29m 38s (non-voting)\n- cyborg-nvme-tempest https://zuul.opendev.org/t/openstack/build/3672e6d4f20d4df6aeb6bd2f6eb72451 : SUCCESS in 17m 29s","accounts_in_message":[],"_revision_number":1},{"id":"ee90f2d414d0a6d0a41d2d8531ad978ff60a0567","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"date":"2026-03-10 11:32:33.000000000","message":"Patch Set 2: Patch Set 1 was rebased\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"60343f89bba4b1fe12687ea3cf1c7780717ab359","tag":"autogenerated:zuul:automatic-ci","author":{"_account_id":28006,"name":"teim-ci","display_name":"teim-ci","email":"ci@seanmooney.info","username":"ci-sean-mooney","status":"this is a third-party ci account run by sean-k-mooney on irc\nhosted at zuul.teim.app"},"date":"2026-03-10 11:32:41.000000000","message":"Patch Set 2:\n\nStarting automatic-ci jobs.","accounts_in_message":[],"_revision_number":2},{"id":"b97e6dd5ef0ac9b48bec14c59916b935036385cd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-10 11:56:08.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/5803341ed4584a0896f373dad4699ea4\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c84e0d062fdc4a95a5b1f6c749208a32 : SUCCESS in 3m 57s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4626f7e62c1e4e7e99b3d68823be7fed : SUCCESS in 7m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a69955a02a794e2e8e5d791d3ab87afc : SUCCESS in 4m 02s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/a30ced51bfa54140a70d0d15c2b2ea60 : SUCCESS in 3m 54s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/a25e79d5a36a4c1e9794ffd708e8fee5 : SUCCESS in 2m 19s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/80d87cf9d9c44892b155ecaa09fec129 : SUCCESS in 6m 07s (non-voting)\n- cyborg-tempest https://zuul.opendev.org/t/openstack/build/4061ed653b06466f98e1b2919e538384 : SUCCESS in 21m 43s\n- cyborg-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/635d5413b60e41e8b22f5236c20ab70f : SUCCESS in 16m 02s (non-voting)\n- cyborg-nvme-tempest https://zuul.opendev.org/t/openstack/build/1b6a8b11b03248c192be4529d8b8a0e8 : FAILURE in 17m 14s","accounts_in_message":[],"_revision_number":2},{"id":"274b01cb8dff5880d76de082771a47e13873dae9","tag":"autogenerated:zuul:automatic-ci","author":{"_account_id":28006,"name":"teim-ci","display_name":"teim-ci","email":"ci@seanmooney.info","username":"ci-sean-mooney","status":"this is a third-party ci account run by sean-k-mooney on irc\nhosted at zuul.teim.app"},"date":"2026-03-10 12:03:45.000000000","message":"Patch Set 2:\n\nBuild failed (automatic-ci pipeline). To rerun just this ci comment\n\"teim-ci: auto\". To rerun other ci jobs comment \"recheck\".\nNote this ci does run on \"recheck\".\n\nhttps://zuul.teim.app/t/main/buildset/a93a39c64ffe45ee8221c5fad8847865\n\n- teim-code-review https://zuul.teim.app/t/main/build/7a170b849a9b4cc283f93d51fd0cbcbe : TIMED_OUT in 30m 53s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"c600c020bb3707017dd9b8c5482e88e5ff8f8138","revisions":{"f84022f4d7bd7d762d9aa78dc493860721921a86":{"kind":"REWORK","_number":1,"created":"2026-03-10 07:12:48.000000000","uploader":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"ref":"refs/changes/88/979788/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cyborg","ref":"refs/changes/88/979788/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/88/979788/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/88/979788/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/88/979788/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cyborg refs/changes/88/979788/1"}}},"commit":{"parents":[{"commit":"574a8ef85c66d845a10af508862c95ab89bb122c","subject":"Add NVMeVirt emulated NVMe device support for devstack","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/574a8ef85c66d845a10af508862c95ab89bb122c"}]}],"author":{"name":"Chandan Kumar (raukadah)","email":"chkumar@redhat.com","date":"2026-03-10 07:10:44.000000000","tz":0},"committer":{"name":"Chandan Kumar (raukadah)","email":"chkumar@redhat.com","date":"2026-03-10 07:10:44.000000000","tz":0},"subject":"Added ansible-lint and bashate pre-commit hook","message":"Added ansible-lint and bashate pre-commit hook\n\nansible-lint will help to lint playbook directory changes.\nbashate will help to lint devstack changes.\n\nIt also adds .ansible-lint file to exclude non ansible files and\nfixes nvmevirt-pre.yaml playbook.\n\nChange-Id: I6c54beceee3dabb38eb581afde92c1c2b427c9e8\nSigned-off-by: Chandan Kumar (raukadah) \u003cchkumar@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/f84022f4d7bd7d762d9aa78dc493860721921a86"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/f84022f4d7bd7d762d9aa78dc493860721921a86"}]},"branch":"refs/heads/master"},"c600c020bb3707017dd9b8c5482e88e5ff8f8138":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2026-03-10 11:32:33.000000000","uploader":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"ref":"refs/changes/88/979788/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cyborg","ref":"refs/changes/88/979788/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/88/979788/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/88/979788/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/88/979788/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cyborg refs/changes/88/979788/2"}}},"commit":{"parents":[{"commit":"1abec6fe5baa2b05b9a8af38f8e1b76af50a690a","subject":"Add NVMeVirt emulated NVMe device support for devstack","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/1abec6fe5baa2b05b9a8af38f8e1b76af50a690a"}]}],"author":{"name":"Chandan Kumar (raukadah)","email":"chkumar@redhat.com","date":"2026-03-10 07:10:44.000000000","tz":0},"committer":{"name":"chandan kumar","email":"chkumar@redhat.com","date":"2026-03-10 11:32:33.000000000","tz":0},"subject":"Added ansible-lint and bashate pre-commit hook","message":"Added ansible-lint and bashate pre-commit hook\n\nansible-lint will help to lint playbook directory changes.\nbashate will help to lint devstack changes.\n\nIt also adds .ansible-lint file to exclude non ansible files and\nfixes nvmevirt-pre.yaml playbook.\n\nChange-Id: I6c54beceee3dabb38eb581afde92c1c2b427c9e8\nSigned-off-by: Chandan Kumar (raukadah) \u003cchkumar@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/c600c020bb3707017dd9b8c5482e88e5ff8f8138"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/c600c020bb3707017dd9b8c5482e88e5ff8f8138"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY"},{"label":"Workflow","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
