)]}'
{"id":"openstack%2Fcyborg~979805","triplet_id":"openstack%2Fcyborg~master~I287974c02be44437ed865c5b4ea3e2b1f2389573","project":"openstack/cyborg","branch":"master","topic":"cyborg_nvmevirt","attention_set":{},"removed_from_attention_set":{"34452":{"account":{"_account_id":34452,"name":"Joan Gilabert","display_name":"jgilaber","email":"jgilaber@redhat.com","username":"jgilaber"},"last_update":"2026-06-18 07:44:36.000000000","reason":"\u003cGERRIT_ACCOUNT_34452\u003e replied on the change","reason_account":{"_account_id":34452,"name":"Joan Gilabert","display_name":"jgilaber","email":"jgilaber@redhat.com","username":"jgilaber"}},"11604":{"account":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"last_update":"2026-06-17 20:10:04.000000000","reason":"\u003cGERRIT_ACCOUNT_11604\u003e replied on the change","reason_account":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"}},"12393":{"account":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"last_update":"2026-06-30 12:03:51.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I287974c02be44437ed865c5b4ea3e2b1f2389573","subject":"Rename devstack CYBORG_CONF_FILE to CYBORG_CONF for consistency","status":"MERGED","created":"2026-03-10 11:30:48.000000000","updated":"2026-06-30 12:06:02.000000000","submitted":"2026-06-30 12:03:51.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":4,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"979805-cyborg_nvmevirt","meta_rev_id":"a07c0797b539efc149d4a73ef50e9d158fc791b1","_number":979805,"virtual_id_number":979805,"owner":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"value":0,"_account_id":34452,"name":"Joan Gilabert","display_name":"jgilaber","email":"jgilaber@redhat.com","username":"jgilaber"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-06-30 12:03:50.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"all":[{"value":2,"date":"2026-06-17 20:10:10.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"value":1,"date":"2026-06-18 07:44:36.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":34452,"name":"Joan Gilabert","display_name":"jgilaber","email":"jgilaber@redhat.com","username":"jgilaber"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"all":[{"value":1,"date":"2026-06-30 11:05:03.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"value":0,"_account_id":34452,"name":"Joan Gilabert","display_name":"jgilaber","email":"jgilaber@redhat.com","username":"jgilaber"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":28006,"name":"teim-ci","display_name":"teim-ci","email":"ci@seanmooney.info","username":"ci-sean-mooney","status":"this is a third-party ci account run by sean-k-mooney on irc\nhosted at zuul.teim.app"}],"REVIEWER":[{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":34452,"name":"Joan Gilabert","display_name":"jgilaber","email":"jgilaber@redhat.com","username":"jgilaber"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-03-10 11:30:54.000000000","updated_by":{"_account_id":28006,"name":"teim-ci","display_name":"teim-ci","email":"ci@seanmooney.info","username":"ci-sean-mooney","status":"this is a third-party ci account run by sean-k-mooney on irc\nhosted at zuul.teim.app"},"reviewer":{"_account_id":28006,"name":"teim-ci","display_name":"teim-ci","email":"ci@seanmooney.info","username":"ci-sean-mooney","status":"this is a third-party ci account run by sean-k-mooney on irc\nhosted at zuul.teim.app"},"state":"CC"},{"updated":"2026-03-10 12:06:18.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-06-12 08:23:23.000000000","updated_by":{"_account_id":34452,"name":"Joan Gilabert","display_name":"jgilaber","email":"jgilaber@redhat.com","username":"jgilaber"},"reviewer":{"_account_id":34452,"name":"Joan Gilabert","display_name":"jgilaber","email":"jgilaber@redhat.com","username":"jgilaber"},"state":"REVIEWER"},{"updated":"2026-06-17 15:01:30.000000000","updated_by":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"reviewer":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"state":"REVIEWER"}],"messages":[{"id":"73cea2b0255822367d42ae8df03932aed8f167bc","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"date":"2026-03-10 11:30:48.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"b3b6eda4dc618a3908d8ac040a35d24ffac2928e","tag":"autogenerated:zuul:automatic-ci","author":{"_account_id":28006,"name":"teim-ci","display_name":"teim-ci","email":"ci@seanmooney.info","username":"ci-sean-mooney","status":"this is a third-party ci account run by sean-k-mooney on irc\nhosted at zuul.teim.app"},"date":"2026-03-10 11:30:54.000000000","message":"Patch Set 1:\n\nStarting automatic-ci jobs.","accounts_in_message":[],"_revision_number":1},{"id":"dfb7c792a3859120cc4df81d36266c28687dfb61","tag":"autogenerated:zuul:automatic-ci","author":{"_account_id":28006,"name":"teim-ci","display_name":"teim-ci","email":"ci@seanmooney.info","username":"ci-sean-mooney","status":"this is a third-party ci account run by sean-k-mooney on irc\nhosted at zuul.teim.app"},"date":"2026-03-10 11:56:47.000000000","message":"Patch Set 1:\n\nBuild succeeded (automatic-ci pipeline).\nhttps://zuul.teim.app/t/main/buildset/295e414dd959472eacbfbdfbddd5652c\n\n- teim-code-review https://zuul.teim.app/t/main/build/a16ff5b1be32461ba5aec6c7c747c2fe : SUCCESS in 25m 44s\n\nWarning:\n  Job teim-code-review: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path commit-message in the commit\u0027\n  Job teim-code-review: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path commit-message in the commit\u0027\n  Comments left for invalid file commit-message","accounts_in_message":[],"_revision_number":1},{"id":"152e5f31df84b5f6ffe8e451193b248f569218c0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-10 12:06:18.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b9f82d85b4d347a4b6da615afa60ef37\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ab6ce391aa0948d2a9cd465b073ae069 : SUCCESS in 4m 25s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e5d243b1e3654ba8a0b59924b2c26487 : SUCCESS in 4m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2c0a90987c134bbb84462b197d39ff6e : SUCCESS in 3m 53s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/0318c80cd1554add9bcdf92da109d407 : SUCCESS in 3m 18s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/ee171c0c0e4045e1bc3fc39b0ce254fc : SUCCESS in 3m 55s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/02186a26dd3747aa9ed3cd72a35e3407 : SUCCESS in 6m 23s (non-voting)\n- cyborg-tempest https://zuul.opendev.org/t/openstack/build/a23386f548c8423981dd87274ae20173 : SUCCESS in 30m 27s\n- cyborg-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/f9fcb947002748b794811da8b48a91cf : SUCCESS in 12m 53s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"95549e580bce064f4537bf53a7baf5f67a15dd53","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"date":"2026-03-11 05:20:28.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"e673af92eccf52d45ba1ca9d5dc0d647441ac7d4","tag":"autogenerated:zuul:automatic-ci","author":{"_account_id":28006,"name":"teim-ci","display_name":"teim-ci","email":"ci@seanmooney.info","username":"ci-sean-mooney","status":"this is a third-party ci account run by sean-k-mooney on irc\nhosted at zuul.teim.app"},"date":"2026-03-11 05:20:34.000000000","message":"Patch Set 2:\n\nStarting automatic-ci jobs.","accounts_in_message":[],"_revision_number":2},{"id":"37b443c29aaf8d1b498a10e67802be87edf615de","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-11 05:43:04.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2528ec722f104e6ea97ab34d497e902f\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/cbd415ac6f7c4c37b9a9cba28f2170c9 : SUCCESS in 3m 31s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b5e54549ecd647a189573d4e94f8e015 : SUCCESS in 6m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0334fb2be09f4c828ef9f502b8d79dad : SUCCESS in 3m 16s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/7398bd2208584fbd94e1e80c75e3ca30 : SUCCESS in 3m 17s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/948b4ff14ca94beea81172c9ecca8030 : SUCCESS in 4m 23s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/3be38db7f1e548619f6584ffb9d7d40b : SUCCESS in 7m 44s (non-voting)\n- cyborg-tempest https://zuul.opendev.org/t/openstack/build/c67ee0ab7f6042e08cdad68dd842b25f : SUCCESS in 22m 09s\n- cyborg-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/af212b067156491fb2b6b7d1c6f26ded : SUCCESS in 21m 06s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"05841cbb5bc6130bed1519a19dbeec31cb8e620f","tag":"autogenerated:zuul:automatic-ci","author":{"_account_id":28006,"name":"teim-ci","display_name":"teim-ci","email":"ci@seanmooney.info","username":"ci-sean-mooney","status":"this is a third-party ci account run by sean-k-mooney on irc\nhosted at zuul.teim.app"},"date":"2026-03-11 05:45:22.000000000","message":"Patch Set 2:\n\n(1 comment)\n\nBuild succeeded (automatic-ci pipeline).\nhttps://zuul.teim.app/t/main/buildset/3f6b6f51fdc04a23ae05bad6386a2be3\n\n- teim-code-review https://zuul.teim.app/t/main/build/9c29f075fd8146ec8c587d5d2a54ae63 : SUCCESS in 24m 40s","accounts_in_message":[],"_revision_number":2},{"id":"2ef982ba3cc0850b9642cce27988ab77cb04af77","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"date":"2026-04-10 14:03:54.000000000","message":"Uploaded patch set 3: Patch Set 2 was rebased.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":3},{"id":"d42f66d5ebe7a3057eb1b88af15831eaf52a3fb3","tag":"autogenerated:zuul:automatic-ci","author":{"_account_id":28006,"name":"teim-ci","display_name":"teim-ci","email":"ci@seanmooney.info","username":"ci-sean-mooney","status":"this is a third-party ci account run by sean-k-mooney on irc\nhosted at zuul.teim.app"},"date":"2026-04-10 14:04:02.000000000","message":"Patch Set 3:\n\nStarting automatic-ci jobs.","accounts_in_message":[],"_revision_number":3},{"id":"fd788409dea96489283ee5938262d69fa7fad1d5","tag":"autogenerated:zuul:automatic-ci","author":{"_account_id":28006,"name":"teim-ci","display_name":"teim-ci","email":"ci@seanmooney.info","username":"ci-sean-mooney","status":"this is a third-party ci account run by sean-k-mooney on irc\nhosted at zuul.teim.app"},"date":"2026-04-10 14:19:28.000000000","message":"Patch Set 3:\n\n(3 comments)\n\nBuild succeeded (automatic-ci pipeline).\nhttps://zuul.teim.app/t/main/buildset/c6dee9b0c8ac459286ed8c516c31b20d\n\n- teim-code-review https://zuul.teim.app/t/main/build/86cf99e688fe4fae9227250f4caf5d35 : SUCCESS in 15m 11s\n\nWarning:\n  Comments left for invalid file devstack/upgrade/upgrade.sh","accounts_in_message":[],"_revision_number":3},{"id":"e84b79b4f2665dd271a4cb19a230eb76548cf270","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-10 15:08:53.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/36b4be9b36e84237bb53c97be6821707\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ed1c18636ae54cb5911ca03390b5f04b : SUCCESS in 4m 33s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cc318b3af3b348c588401ba2c1c09183 : SUCCESS in 7m 56s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a55bc6e6730d4e1f98d898a6eaf2230e : SUCCESS in 5m 46s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/7bba663deaf54cbf8ae2744c83ce838f : SUCCESS in 2m 42s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/0823337e0b1d4043a5ea962ad9476655 : SUCCESS in 3m 32s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/0848c976728c4cba9aa2551fa0501582 : SUCCESS in 7m 12s (non-voting)\n- cyborg-tempest https://zuul.opendev.org/t/openstack/build/161062081db047c79d2c10769cbcf14c : SUCCESS in 14m 51s\n- cyborg-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/bf07343389154fcd86fcb51be428fadd : SUCCESS in 42m 23s (non-voting)\n- cyborg-grenade https://zuul.opendev.org/t/openstack/build/fda15899b8a64f73a312b8ace64e946c : FAILURE in 48m 39s\n- cyborg-grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/29a44a7af61f488fbb36f94a56872903 : FAILURE in 50m 29s","accounts_in_message":[],"_revision_number":3},{"id":"22412790276c516f4d825dc66b902e24bd25c9c7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"date":"2026-06-08 09:16:25.000000000","message":"Uploaded patch set 4.\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":4},{"id":"3294b85564f832fad2abfa67432cd677d01249da","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-08 10:29:26.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/dbc8c1b3fcaf436a865443e3dad23901\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c1156239ae6842f8b13c00a74e097bd2 : SUCCESS in 6m 52s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/45aaf583fa34450d941ffb5fed1e7049 : SUCCESS in 5m 12s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/60a22face5fb48c18cfb7b1620b0d5d4 : SUCCESS in 10m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2db8f46b971546c78d8d938ba7da848a : SUCCESS in 4m 09s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/39b8b1f57f6640019c7c02053d73fbd1 : SUCCESS in 4m 40s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/5c6ed6c1f4b345feab93a91f6c07029e : SUCCESS in 6m 32s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/dedb96be7fab4753b46b4e68d7a5a0ee : SUCCESS in 4m 53s (non-voting)\n- cyborg-tempest-py3 https://zuul.opendev.org/t/openstack/build/38c257d5d2fd45039f36cb66660fa8ce : SUCCESS in 35m 30s\n- cyborg-tempest-ipv6 https://zuul.opendev.org/t/openstack/build/ce4278c5ce4a4041974ecb9a07822fcb : SUCCESS in 29m 54s\n- cyborg-grenade https://zuul.opendev.org/t/openstack/build/709d54dba5744091854697bc06f6a592 : SUCCESS in 25m 42s\n- cyborg-grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/85c2dc2d90654b2aae29a0e890874acb : SUCCESS in 56m 00s","accounts_in_message":[],"_revision_number":4},{"id":"19c6feba6185ec3923a8f55030526790f30c5230","author":{"_account_id":34452,"name":"Joan Gilabert","display_name":"jgilaber","email":"jgilaber@redhat.com","username":"jgilaber"},"date":"2026-06-12 08:23:23.000000000","message":"Patch Set 4: Code-Review+1","accounts_in_message":[],"_revision_number":4},{"id":"f5894ba895bf7100773bc86dba5a4742db559a60","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"date":"2026-06-17 14:55:43.000000000","message":"Uploaded patch set 5.\n\nOutdated Votes:\n* Code-Review+1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":5},{"id":"3e967636aaa7627e1d94c48fa8298c41d68306b8","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2026-06-17 15:01:30.000000000","message":"Patch Set 5: Code-Review-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":5},{"id":"918f7f642cbc8e06e40f70418900ea1411cf660d","author":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"date":"2026-06-17 15:30:23.000000000","message":"Patch Set 5:\n\n(2 comments)","accounts_in_message":[],"_revision_number":5},{"id":"5cef6a65c29c59994c379a35e6306bffda943753","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"date":"2026-06-17 15:30:27.000000000","message":"Uploaded patch set 6.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n","accounts_in_message":[],"_revision_number":6},{"id":"23dea8a31a1f0fce8df9de7a770a7da76fa3b9b6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-17 16:31:03.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/26f57646139c4af79a463a232ae23cf1\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/2322033d1a5a4510afd676cedcd06861 : SUCCESS in 4m 35s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b4c0e8aa873f441ea7f81e210d756030 : SUCCESS in 5m 01s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/45c51ab2e9724281b23b7fd52fb04871 : SUCCESS in 7m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/07dd99e238e94990ad00b21ab6aebbe5 : SUCCESS in 3m 56s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/0ff118f96fa048df8cd68ff419200446 : SUCCESS in 5m 17s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/5a081d09e75a4f6c9e46b7fca4f8a747 : SUCCESS in 5m 17s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/4a0a9a79b1fa47df89f84b1afac64fe9 : SUCCESS in 6m 06s (non-voting)\n- cyborg-tempest-py3 https://zuul.opendev.org/t/openstack/build/378dd713e15e452389c59b332793ffec : SUCCESS in 41m 18s\n- cyborg-tempest-ipv6 https://zuul.opendev.org/t/openstack/build/8c6024e5120d472eb6442f1e2f3e35ed : SUCCESS in 29m 27s\n- cyborg-grenade https://zuul.opendev.org/t/openstack/build/15629b38d8864e0d93af25b8a8007d6b : SUCCESS in 25m 28s\n- cyborg-grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/4b34e5cc57e14322b0828e699a1a6561 : SUCCESS in 54m 17s","accounts_in_message":[],"_revision_number":6},{"id":"bb62d1983c0eaecc1511538a934ffcc2a5e1d30a","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2026-06-17 20:10:10.000000000","message":"Patch Set 6: Code-Review+2","accounts_in_message":[],"_revision_number":6},{"id":"39d94afa41152c5cd0f5e4cff4689514b4c8cba9","author":{"_account_id":34452,"name":"Joan Gilabert","display_name":"jgilaber","email":"jgilaber@redhat.com","username":"jgilaber"},"date":"2026-06-18 07:44:36.000000000","message":"Patch Set 6: Code-Review+1","accounts_in_message":[],"_revision_number":6},{"id":"d00aff9571e8b763e5ba1ce88451994699235a69","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2026-06-30 11:05:03.000000000","message":"Patch Set 6: Workflow+1","accounts_in_message":[],"_revision_number":6},{"id":"c9c8ada07cb3b0b6c2d1ba5b478d06e5788a079c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-30 11:05:39.000000000","message":"Patch Set 6: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":6},{"id":"fbc6adbea16caf6fd193b40c319d6d8af55b4c84","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-30 12:03:50.000000000","message":"Patch Set 6: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1754a7b263074188ac500f14e9fa72e3\n\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2701b2c7d9cf4e359c65f91f8cd18693 : SUCCESS in 6m 24s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ee49a508bfa8429a9179eb8a6995f914 : SUCCESS in 8m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d929e1390b4e44e2a32b382afcc6bd6e : SUCCESS in 2m 36s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/95331961465d480487137069fd065a8d : SUCCESS in 3m 32s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/91c1e92dfe2040cea375a14e1c8020b8 : SUCCESS in 6m 36s\n- cyborg-tempest-py3 https://zuul.opendev.org/t/openstack/build/9d35fdbde7b44f31a4d3c3a40e4da862 : SUCCESS in 38m 59s\n- cyborg-tempest-ipv6 https://zuul.opendev.org/t/openstack/build/9f9d6f55dd834427ad8fec47864a33d5 : SUCCESS in 13m 11s\n- cyborg-grenade https://zuul.opendev.org/t/openstack/build/e9f3a289942b42a29e5aff2e98949453 : SUCCESS in 51m 04s\n- cyborg-grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/c977c037bc174ec5a291dfdb7402e225 : SUCCESS in 27m 33s","accounts_in_message":[],"_revision_number":6},{"id":"7d06de91120dc159ce09d8cbda30f8d4dc34d5c5","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-30 12:03:51.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":6},{"id":"a07c0797b539efc149d4a73ef50e9d158fc791b1","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-30 12:06:02.000000000","message":"Patch Set 6:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1d2edcc474994e9e8652824f5d9b4e30\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2b484dfe61ff4bd0a835327d5c66b6c3 : SUCCESS in 55s","accounts_in_message":[],"_revision_number":6}],"current_revision_number":6,"current_revision":"2bff6797cadfafc787322b6628b3c79f55d20b43","revisions":{"6d3f64682d20bf9bcaa5f146967c6e93a59c4062":{"kind":"REWORK","_number":1,"created":"2026-03-10 11:30:48.000000000","uploader":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"ref":"refs/changes/05/979805/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cyborg","ref":"refs/changes/05/979805/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/05/979805/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/05/979805/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/05/979805/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cyborg refs/changes/05/979805/1"}}},"commit":{"parents":[{"commit":"26dec650a90fe7874b6014ae492ec8cc872552e2","subject":"adopt isort","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/26dec650a90fe7874b6014ae492ec8cc872552e2"}]}],"author":{"name":"Chandan Kumar (raukadah)","email":"chkumar@redhat.com","date":"2026-03-10 11:27:04.000000000","tz":0},"committer":{"name":"Chandan Kumar (raukadah)","email":"chkumar@redhat.com","date":"2026-03-10 11:27:04.000000000","tz":0},"subject":"Rename devstack CYBORG_CONF_FILE to CYBORG_CONF for consistency","message":"Rename devstack CYBORG_CONF_FILE to CYBORG_CONF for consistency\n\nCYBORG_CONF_FILE var is used to point cyborg config file in devstack.\nIn DevStack, most of the services config file var is SERVICE_CONF.\n\nIn order to keep consistent with other devstack plugins, thic cr\nrenames the var CYBORG_CONF_FILE to CYBORG_CONF.\n\nChange-Id: I287974c02be44437ed865c5b4ea3e2b1f2389573\nSigned-off-by: Chandan Kumar (raukadah) \u003cchkumar@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/6d3f64682d20bf9bcaa5f146967c6e93a59c4062"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/6d3f64682d20bf9bcaa5f146967c6e93a59c4062"}]},"branch":"refs/heads/master"},"9bfe3ae2aa8fdb94fa9e74d16911c1e9cbd1e98a":{"kind":"REWORK","_number":2,"created":"2026-03-11 05:20:28.000000000","uploader":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"ref":"refs/changes/05/979805/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cyborg","ref":"refs/changes/05/979805/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/05/979805/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/05/979805/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/05/979805/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cyborg refs/changes/05/979805/2"}}},"commit":{"parents":[{"commit":"26dec650a90fe7874b6014ae492ec8cc872552e2","subject":"adopt isort","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/26dec650a90fe7874b6014ae492ec8cc872552e2"}]}],"author":{"name":"Chandan Kumar (raukadah)","email":"chkumar@redhat.com","date":"2026-03-10 11:27:04.000000000","tz":0},"committer":{"name":"Chandan Kumar (raukadah)","email":"chkumar@redhat.com","date":"2026-03-11 05:19:53.000000000","tz":0},"subject":"Rename devstack CYBORG_CONF_FILE to CYBORG_CONF for consistency","message":"Rename devstack CYBORG_CONF_FILE to CYBORG_CONF for consistency\n\nCYBORG_CONF_FILE var is used to point cyborg config file in devstack.\nIn DevStack, most of the services config file var is SERVICE_CONF.\n\nIn order to keep consistent with other devstack plugins, thic cr\nrenames the var CYBORG_CONF_FILE to CYBORG_CONF.\n\nNote: It also make sure it fallback to the old name.\n\nChange-Id: I287974c02be44437ed865c5b4ea3e2b1f2389573\nSigned-off-by: Chandan Kumar (raukadah) \u003cchkumar@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/9bfe3ae2aa8fdb94fa9e74d16911c1e9cbd1e98a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/9bfe3ae2aa8fdb94fa9e74d16911c1e9cbd1e98a"}]},"branch":"refs/heads/master"},"9cb01d4515e2f0bd983d12f715b4a568c8177450":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2026-04-10 14:03:54.000000000","uploader":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"ref":"refs/changes/05/979805/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cyborg","ref":"refs/changes/05/979805/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/05/979805/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/05/979805/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/05/979805/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cyborg refs/changes/05/979805/3"}}},"commit":{"parents":[{"commit":"f111946df6713aa64efa29dd025d47839241c529","subject":"Merge \"Add Grenade upgrade support for Cyborg\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/f111946df6713aa64efa29dd025d47839241c529"}]}],"author":{"name":"Chandan Kumar (raukadah)","email":"chkumar@redhat.com","date":"2026-03-10 11:27:04.000000000","tz":0},"committer":{"name":"Chandan Kumar (raukadah)","email":"chkumar@redhat.com","date":"2026-04-10 14:02:07.000000000","tz":330},"subject":"Rename devstack CYBORG_CONF_FILE to CYBORG_CONF for consistency","message":"Rename devstack CYBORG_CONF_FILE to CYBORG_CONF for consistency\n\nCYBORG_CONF_FILE var is used to point cyborg config file in devstack.\nIn DevStack, most of the services config file var is SERVICE_CONF.\n\nIn order to keep consistent with other devstack plugins, thic cr\nrenames the var CYBORG_CONF_FILE to CYBORG_CONF.\n\nNote: It also make sure it fallback to the old name.\n\nChange-Id: I287974c02be44437ed865c5b4ea3e2b1f2389573\nSigned-off-by: Chandan Kumar (raukadah) \u003cchkumar@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/9cb01d4515e2f0bd983d12f715b4a568c8177450"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/9cb01d4515e2f0bd983d12f715b4a568c8177450"}]},"branch":"refs/heads/master"},"5f2d96eec02b21bc1986a9fc9f790c2662199c23":{"kind":"REWORK","_number":4,"created":"2026-06-08 09:16:25.000000000","uploader":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"ref":"refs/changes/05/979805/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cyborg","ref":"refs/changes/05/979805/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/05/979805/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/05/979805/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/05/979805/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cyborg refs/changes/05/979805/4"}}},"commit":{"parents":[{"commit":"233f2a5b7e396c24bf15ae22f8cf64a480666f5c","subject":"Merge \"pci: add OWNER_CYBORG constant and fix trait\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/233f2a5b7e396c24bf15ae22f8cf64a480666f5c"}]}],"author":{"name":"Chandan Kumar (raukadah)","email":"chkumar@redhat.com","date":"2026-06-08 09:15:50.000000000","tz":330},"committer":{"name":"Chandan Kumar (raukadah)","email":"chkumar@redhat.com","date":"2026-06-08 09:15:50.000000000","tz":330},"subject":"Rename devstack CYBORG_CONF_FILE to CYBORG_CONF for consistency","message":"Rename devstack CYBORG_CONF_FILE to CYBORG_CONF for consistency\n\nCYBORG_CONF_FILE var is used to point cyborg config file in devstack.\nIn DevStack, most of the services config file var is SERVICE_CONF.\n\nIn order to keep consistent with other devstack plugins, thic cr\nrenames the var CYBORG_CONF_FILE to CYBORG_CONF.\n\nNote: It also make sure it fallback to the old name.\n\nChange-Id: I287974c02be44437ed865c5b4ea3e2b1f2389573\nSigned-off-by: Chandan Kumar (raukadah) \u003cchkumar@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/5f2d96eec02b21bc1986a9fc9f790c2662199c23"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/5f2d96eec02b21bc1986a9fc9f790c2662199c23"}]},"branch":"refs/heads/master"},"1d716df29ae41cef1de07f7bca05fb3e9749c3a2":{"kind":"REWORK","_number":5,"created":"2026-06-17 14:55:43.000000000","uploader":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"ref":"refs/changes/05/979805/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cyborg","ref":"refs/changes/05/979805/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/05/979805/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/05/979805/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/05/979805/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cyborg refs/changes/05/979805/5"}}},"commit":{"parents":[{"commit":"042a8dca4d54372c95347fea38d8417ac275e9f0","subject":"Merge \"Add pci-sim developer guide\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/042a8dca4d54372c95347fea38d8417ac275e9f0"}]}],"author":{"name":"Chandan Kumar (raukadah)","email":"chkumar@redhat.com","date":"2026-06-08 09:15:50.000000000","tz":330},"committer":{"name":"Chandan Kumar (raukadah)","email":"chkumar@redhat.com","date":"2026-06-17 14:26:39.000000000","tz":330},"subject":"Rename devstack CYBORG_CONF_FILE to CYBORG_CONF for consistency","message":"Rename devstack CYBORG_CONF_FILE to CYBORG_CONF for consistency\n\nCYBORG_CONF_FILE var is used to point cyborg config file in devstack.\nIn DevStack, most of the services config file var is SERVICE_CONF.\n\nIn order to keep consistent with other devstack plugins, thic cr\nrenames the var CYBORG_CONF_FILE to CYBORG_CONF.\n\nNote: It also make sure it fallback to the old name.\n\nChange-Id: I287974c02be44437ed865c5b4ea3e2b1f2389573\nSigned-off-by: Chandan Kumar (raukadah) \u003cchkumar@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/1d716df29ae41cef1de07f7bca05fb3e9749c3a2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/1d716df29ae41cef1de07f7bca05fb3e9749c3a2"}]},"branch":"refs/heads/master"},"2bff6797cadfafc787322b6628b3c79f55d20b43":{"kind":"REWORK","_number":6,"created":"2026-06-17 15:30:27.000000000","uploader":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"ref":"refs/changes/05/979805/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cyborg","ref":"refs/changes/05/979805/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/05/979805/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/05/979805/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/05/979805/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cyborg refs/changes/05/979805/6"}}},"commit":{"parents":[{"commit":"042a8dca4d54372c95347fea38d8417ac275e9f0","subject":"Merge \"Add pci-sim developer guide\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/042a8dca4d54372c95347fea38d8417ac275e9f0"}]}],"author":{"name":"Chandan Kumar (raukadah)","email":"chkumar@redhat.com","date":"2026-06-08 09:15:50.000000000","tz":330},"committer":{"name":"Chandan Kumar (raukadah)","email":"chkumar@redhat.com","date":"2026-06-17 15:28:29.000000000","tz":330},"subject":"Rename devstack CYBORG_CONF_FILE to CYBORG_CONF for consistency","message":"Rename devstack CYBORG_CONF_FILE to CYBORG_CONF for consistency\n\nCYBORG_CONF_FILE var is used to point cyborg config file in devstack.\nIn DevStack, most of the services config file var is SERVICE_CONF.\n\nIn order to keep consistent with other devstack plugins, thic cr\nrenames the var CYBORG_CONF_FILE to CYBORG_CONF.\n\nNote: It also make sure it fallback to the old name.\n\nChange-Id: I287974c02be44437ed865c5b4ea3e2b1f2389573\nSigned-off-by: Chandan Kumar (raukadah) \u003cchkumar@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/2bff6797cadfafc787322b6628b3c79f55d20b43"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/2bff6797cadfafc787322b6628b3c79f55d20b43"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":34452,"name":"Joan Gilabert","display_name":"jgilaber","email":"jgilaber@redhat.com","username":"jgilaber"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
