)]}'
{"id":"openstack%2Fcyborg~997364","triplet_id":"openstack%2Fcyborg~master~I55eb22a68d3e1582cbcc337b85481ac4aabff971","project":"openstack/cyborg","branch":"master","topic":"generic-mdev-driver","attention_set":{"4690":{"account":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},"last_update":"2026-07-30 11:05:43.000000000","reason":"\u003cGERRIT_ACCOUNT_34452\u003e replied on the change","reason_account":{"_account_id":34452,"name":"Joan Gilabert","display_name":"jgilaber","email":"jgilaber@redhat.com","username":"jgilaber"}},"34452":{"account":{"_account_id":34452,"name":"Joan Gilabert","display_name":"jgilaber","email":"jgilaber@redhat.com","username":"jgilaber"},"last_update":"2026-08-31 15:07:48.000000000","reason":"Someone else replied on the change"}},"removed_from_attention_set":{"28006":{"account":{"_account_id":28006,"name":"teim-ci","display_name":"teim-ci","email":"ci@seanmooney.info","username":"ci-sean-mooney","status":"this is a third-party ci account run by sean-k-mooney on irc\nhosted at zuul.teim.app"},"last_update":"2026-07-31 13:25:46.000000000","reason":"removed on reply"}},"hashtags":[],"change_id":"I55eb22a68d3e1582cbcc337b85481ac4aabff971","subject":"[WIP] Configure mdev driver lifecycle tests","status":"NEW","created":"2026-07-15 10:41:05.000000000","updated":"2026-09-01 11:24:19.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":true,"submittable":false,"total_comment_count":43,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"3d2ac4bbb7cb014e41e8dee560710e6c84d49584","_number":997364,"virtual_id_number":997364,"owner":{"_account_id":34452,"name":"Joan Gilabert","display_name":"jgilaber","email":"jgilaber@redhat.com","username":"jgilaber"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2026-09-01 11:09:50.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},{"_account_id":28006,"name":"teim-ci","display_name":"teim-ci","email":"ci@seanmooney.info","username":"ci-sean-mooney","status":"this is a third-party ci account run by sean-k-mooney on irc\nhosted at zuul.teim.app"}],"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-07-15 10:41:31.000000000","updated_by":{"_account_id":28006,"name":"teim-ci","display_name":"teim-ci","email":"ci@seanmooney.info","username":"ci-sean-mooney","status":"this is a third-party ci account run by sean-k-mooney on irc\nhosted at zuul.teim.app"},"reviewer":{"_account_id":28006,"name":"teim-ci","display_name":"teim-ci","email":"ci@seanmooney.info","username":"ci-sean-mooney","status":"this is a third-party ci account run by sean-k-mooney on irc\nhosted at zuul.teim.app"},"state":"CC"},{"updated":"2026-07-15 11:20:06.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-07-30 01:50:50.000000000","updated_by":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},"reviewer":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},"state":"CC"}],"messages":[{"id":"f9df174aa8afa07bb63e27fa7a87d039343ce967","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34452,"name":"Joan Gilabert","display_name":"jgilaber","email":"jgilaber@redhat.com","username":"jgilaber"},"date":"2026-07-15 10:41:05.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"c418745ca6b97c32b2d9861a13d520146bd680f2","tag":"autogenerated:zuul:automatic-ci","author":{"_account_id":28006,"name":"teim-ci","display_name":"teim-ci","email":"ci@seanmooney.info","username":"ci-sean-mooney","status":"this is a third-party ci account run by sean-k-mooney on irc\nhosted at zuul.teim.app"},"date":"2026-07-15 10:41:31.000000000","message":"Patch Set 1:\n\nStarting automatic-ci jobs.","accounts_in_message":[],"_revision_number":1},{"id":"41edf205a53cee8152138477624ccd9f8bb573b8","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":34452,"name":"Joan Gilabert","display_name":"jgilaber","email":"jgilaber@redhat.com","username":"jgilaber"},"date":"2026-07-15 10:47:02.000000000","message":"Topic set to generic-mdev-driver","accounts_in_message":[],"_revision_number":1},{"id":"91a537b20670d7cfe4249ed8f15d06348373ffb0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34452,"name":"Joan Gilabert","display_name":"jgilaber","email":"jgilaber@redhat.com","username":"jgilaber"},"date":"2026-07-15 10:48:14.000000000","message":"Uploaded patch set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"fc71dc9adae09720769f1816ff6254b2f1e9b26d","tag":"autogenerated:zuul:automatic-ci","author":{"_account_id":28006,"name":"teim-ci","display_name":"teim-ci","email":"ci@seanmooney.info","username":"ci-sean-mooney","status":"this is a third-party ci account run by sean-k-mooney on irc\nhosted at zuul.teim.app"},"date":"2026-07-15 10:48:49.000000000","message":"Patch Set 2:\n\nStarting automatic-ci jobs.","accounts_in_message":[],"_revision_number":2},{"id":"3b620f91d52733876e0f92c556353a770b5d1dad","tag":"autogenerated:zuul:automatic-ci","author":{"_account_id":28006,"name":"teim-ci","display_name":"teim-ci","email":"ci@seanmooney.info","username":"ci-sean-mooney","status":"this is a third-party ci account run by sean-k-mooney on irc\nhosted at zuul.teim.app"},"date":"2026-07-15 11:08:50.000000000","message":"Patch Set 2:\n\n(3 comments)\n\nBuild succeeded (automatic-ci pipeline).\nhttps://zuul.teim.app/t/main/buildset/c9de42bcd9ee414ca1f75459d42329ab\n\n- teim-code-review https://zuul.teim.app/t/main/build/c85418dd7eca475a988566ac6afb3fde : SUCCESS in 13m 07s\n\nWarning:\n  The cyborg-tempest-py3 job inherits devstack_localrc from cyborg-tempest-base which has ENABLE_PCI_SIM: true but not PCI_SIM_ENABLE_MDEV. PCI_SIM_ENABLE_MDEV: true is only set in the subnode group-vars. Without it on the primary, PCI_SIM_CONFIGURE_CYBORG_MDEV defaults to False and create_pci_sim_... Impact: Even after fixing the other config issues, the mdev device profile and flavor will not exist because the DevStack plugin skips their creation when PCI_SIM_CONFIGURE_CYBORG_MDEV is False on the primary node. All mdev lifecycle tests will fail. Recommendation: Add PCI_SIM_ENABLE_MDEV: true to the cyborg-tempest-py3 job\u0027s devstack_localrc (or to cyborg-multinode-tempest\u0027s primary node vars) so that create_pci_sim_cyborg_mdev_device_profile runs on the primary node where Tempest and test resources are created.","accounts_in_message":[],"_revision_number":2},{"id":"be609a5409f4ed8114aec45dc5cd571f8fd84fd9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-15 11:20:06.000000000","message":"Patch Set 2: Verified-1\n\nThis change depends on a change that failed to merge.\n\nChange https://review.opendev.org/c/openstack/cyborg-tempest-plugin/+/997352 is needed.","accounts_in_message":[],"_revision_number":2},{"id":"86b94aec5ac09664b830cc04794b5ebaa485e725","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34452,"name":"Joan Gilabert","display_name":"jgilaber","email":"jgilaber@redhat.com","username":"jgilaber"},"date":"2026-07-15 11:21:01.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":3},{"id":"3808a22825af1175e60117d041ba52a12bf27907","tag":"autogenerated:zuul:automatic-ci","author":{"_account_id":28006,"name":"teim-ci","display_name":"teim-ci","email":"ci@seanmooney.info","username":"ci-sean-mooney","status":"this is a third-party ci account run by sean-k-mooney on irc\nhosted at zuul.teim.app"},"date":"2026-07-15 11:21:26.000000000","message":"Patch Set 3:\n\nStarting automatic-ci jobs.","accounts_in_message":[],"_revision_number":3},{"id":"d5112206060e993ec5c563c61774c3e89af18a81","tag":"autogenerated:zuul:automatic-ci","author":{"_account_id":28006,"name":"teim-ci","display_name":"teim-ci","email":"ci@seanmooney.info","username":"ci-sean-mooney","status":"this is a third-party ci account run by sean-k-mooney on irc\nhosted at zuul.teim.app"},"date":"2026-07-15 11:38:35.000000000","message":"Patch Set 3:\n\n(1 comment)\n\nBuild succeeded (automatic-ci pipeline).\nhttps://zuul.teim.app/t/main/buildset/d38b7fc713994a9baad71b1f7c5bf09b\n\n- teim-code-review https://zuul.teim.app/t/main/build/29bb604932854c038545f8be6679029c : SUCCESS in 9m 03s","accounts_in_message":[],"_revision_number":3},{"id":"196ae4c2b30e2eae3220a0e45e29049d18ca9a60","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34452,"name":"Joan Gilabert","display_name":"jgilaber","email":"jgilaber@redhat.com","username":"jgilaber"},"date":"2026-07-15 12:02:11.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"6c3f02c23d8ef9203510bb10a93d7ffc077f0423","tag":"autogenerated:zuul:automatic-ci","author":{"_account_id":28006,"name":"teim-ci","display_name":"teim-ci","email":"ci@seanmooney.info","username":"ci-sean-mooney","status":"this is a third-party ci account run by sean-k-mooney on irc\nhosted at zuul.teim.app"},"date":"2026-07-15 12:02:36.000000000","message":"Patch Set 4:\n\nStarting automatic-ci jobs.","accounts_in_message":[],"_revision_number":4},{"id":"8cf0182f92b48d1290ecb950d3ca94c7e9fc0e71","tag":"autogenerated:zuul:automatic-ci","author":{"_account_id":28006,"name":"teim-ci","display_name":"teim-ci","email":"ci@seanmooney.info","username":"ci-sean-mooney","status":"this is a third-party ci account run by sean-k-mooney on irc\nhosted at zuul.teim.app"},"date":"2026-07-15 12:17:42.000000000","message":"Patch Set 4:\n\n(1 comment)\n\nBuild succeeded (automatic-ci pipeline).\nhttps://zuul.teim.app/t/main/buildset/41291adf24f04b4a9a42e2d59ef73fe4\n\n- teim-code-review https://zuul.teim.app/t/main/build/a55d765b0324403aa3d5ca8f4993fd38 : SUCCESS in 14m 55s","accounts_in_message":[],"_revision_number":4},{"id":"e775b275003a58322b927b3d341acb9a0e5330d9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-15 12:58:39.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/0952fe3524b046aeaf6fde129eefbd72\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/cc565ef5dd364818b0959d770d2c5622 : SUCCESS in 4m 26s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/768bb6fbb7e54ca7804ba45a6045c658 : SUCCESS in 4m 05s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7af4c9e3abe84aae947959f24ec6ad76 : SUCCESS in 7m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d15d8132236f48acb3bb69c11df07163 : SUCCESS in 5m 09s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/d80e1ec29bb4477e96b7cac9f0f4c1ff : SUCCESS in 3m 28s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/61affe0dc7454ffba828c45fc96fd5ce : SUCCESS in 3m 46s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/3a8b1ba36ff14321bb2ba2f297afdd0c : SUCCESS in 5m 34s (non-voting)\n- cyborg-tempest-py3 https://zuul.opendev.org/t/openstack/build/0fdbbaab55904033a11c2497c3125dd0 : FAILURE in 24m 07s\n- cyborg-tempest-ipv6 https://zuul.opendev.org/t/openstack/build/5e1314d01b2e4caa91287e3afe7d2694 : FAILURE in 27m 40s\n- cyborg-grenade https://zuul.opendev.org/t/openstack/build/127a381a41b446c4bf0e7dd16af0ba9c : FAILURE in 39m 13s\n- cyborg-grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/41c1bc5bdc254ce68ab0c8e26d0a52d8 : FAILURE in 48m 21s","accounts_in_message":[],"_revision_number":4},{"id":"6b7e3d062c5408b505230f5633f55b6ae74a07ba","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34452,"name":"Joan Gilabert","display_name":"jgilaber","email":"jgilaber@redhat.com","username":"jgilaber"},"date":"2026-07-15 13:12:26.000000000","message":"Uploaded patch set 5.\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":5},{"id":"c3f6b381b191e12ddb25890a78edf2020d7e1b29","tag":"autogenerated:zuul:automatic-ci","author":{"_account_id":28006,"name":"teim-ci","display_name":"teim-ci","email":"ci@seanmooney.info","username":"ci-sean-mooney","status":"this is a third-party ci account run by sean-k-mooney on irc\nhosted at zuul.teim.app"},"date":"2026-07-15 13:12:51.000000000","message":"Patch Set 5:\n\nStarting automatic-ci jobs.","accounts_in_message":[],"_revision_number":5},{"id":"4f8ebaddcb2e10776376905bfb1c48e316217f50","tag":"autogenerated:zuul:automatic-ci","author":{"_account_id":28006,"name":"teim-ci","display_name":"teim-ci","email":"ci@seanmooney.info","username":"ci-sean-mooney","status":"this is a third-party ci account run by sean-k-mooney on irc\nhosted at zuul.teim.app"},"date":"2026-07-15 13:36:44.000000000","message":"Patch Set 5:\n\nBuild succeeded (automatic-ci pipeline).\nhttps://zuul.teim.app/t/main/buildset/62f12cd819e64b1e871d55859730cc80\n\n- teim-code-review https://zuul.teim.app/t/main/build/bb1b2ceea40548be9919d77c47815380 : SUCCESS in 10m 56s","accounts_in_message":[],"_revision_number":5},{"id":"af09da1aee29e6e20598f6831a0d460d706c53ba","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-15 14:14:17.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/cd7d46ef7cbc439dab180ba681808ac8\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/9ee30bdd3ac14d25a246aae9e9af453d : SUCCESS in 5m 47s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/06e76a9827a948c593cb74fa1a6f08c6 : SUCCESS in 7m 21s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/30a94fb9506241e2a9aa1e75b0396a82 : SUCCESS in 6m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/afdceabbe5e64251b64dcfaa6ce5634f : SUCCESS in 4m 11s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/47fee14a91644d9ea358a0410e8eb16b : SUCCESS in 3m 44s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/f7103731d9db428fa22c547252024ce6 : SUCCESS in 4m 55s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/095752501a394244b5f8600462db9147 : SUCCESS in 6m 27s (non-voting)\n- cyborg-tempest-py3 https://zuul.opendev.org/t/openstack/build/4c16b198bf4349d5ab9a2b7fbd971009 : FAILURE in 23m 26s\n- cyborg-tempest-ipv6 https://zuul.opendev.org/t/openstack/build/c63eebfd48d44bffa53d92443276afd4 : SUCCESS in 28m 17s\n- cyborg-grenade https://zuul.opendev.org/t/openstack/build/ab395cb55c4a46dd9fd7672ea5726884 : SUCCESS in 35m 44s\n- cyborg-grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/df8d8896363c45c0acdac6e4310287a5 : SUCCESS in 55m 07s","accounts_in_message":[],"_revision_number":5},{"id":"0b90174a421219fac92d78bb18a35b071cf94494","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34452,"name":"Joan Gilabert","display_name":"jgilaber","email":"jgilaber@redhat.com","username":"jgilaber"},"date":"2026-07-15 14:42:56.000000000","message":"Uploaded patch set 6: Commit message was updated.\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":6},{"id":"e1eab5e708176e887bfd54b64ffa38879f7d947a","tag":"autogenerated:zuul:automatic-ci","author":{"_account_id":28006,"name":"teim-ci","display_name":"teim-ci","email":"ci@seanmooney.info","username":"ci-sean-mooney","status":"this is a third-party ci account run by sean-k-mooney on irc\nhosted at zuul.teim.app"},"date":"2026-07-15 14:43:31.000000000","message":"Patch Set 6:\n\nStarting automatic-ci jobs.","accounts_in_message":[],"_revision_number":6},{"id":"8436c76ff5dddf0795f921c0144c36716767529e","tag":"autogenerated:zuul:automatic-ci","author":{"_account_id":28006,"name":"teim-ci","display_name":"teim-ci","email":"ci@seanmooney.info","username":"ci-sean-mooney","status":"this is a third-party ci account run by sean-k-mooney on irc\nhosted at zuul.teim.app"},"date":"2026-07-15 14:55:02.000000000","message":"Patch Set 6:\n\nBuild succeeded (automatic-ci pipeline).\nhttps://zuul.teim.app/t/main/buildset/a99455b9e1c24487945a659444e9acb6\n\n- teim-code-review https://zuul.teim.app/t/main/build/80414a62dce24af2a56fc1eedbff1ee7 : SUCCESS in 11m 22s","accounts_in_message":[],"_revision_number":6},{"id":"d98f2bf3726f42db0b4379135588ea6448df162f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-15 15:57:28.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/76d6c8d1514545e3ab0de8f8f5a3352d\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/5f58c912884746b3ae650346e1a99cb5 : SUCCESS in 4m 33s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f24cb911342146cca9bedfba0005662f : SUCCESS in 4m 28s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a95bc0e0caf1422cae3f251ac21f461b : SUCCESS in 8m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/46f9498bdd9648c6b7b98e0721763bb6 : SUCCESS in 4m 13s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/a60ead3b162449e8a2ff00fd0b23be4d : SUCCESS in 4m 18s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/6c5757c9a0c94ed995f1228dd6c0eb55 : SUCCESS in 4m 41s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/93a0ca107f9a407fbdcea3b465265995 : SUCCESS in 6m 57s (non-voting)\n- cyborg-tempest-py3 https://zuul.opendev.org/t/openstack/build/894a941efe074310aff31ba29e8f3ff1 : FAILURE in 48m 53s\n- cyborg-tempest-ipv6 https://zuul.opendev.org/t/openstack/build/f0c4849395a24e6489375ba067cbccca : SUCCESS in 28m 17s\n- cyborg-grenade https://zuul.opendev.org/t/openstack/build/4c96f9685f7a4c45b9e87286e729f13d : SUCCESS in 56m 22s\n- cyborg-grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/dfdf73368c024fa393a9aff1802ce356 : SUCCESS in 1h 07m 22s","accounts_in_message":[],"_revision_number":6},{"id":"e305fb6930b84aeccbf9ea6b3595a87860f602d0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34452,"name":"Joan Gilabert","display_name":"jgilaber","email":"jgilaber@redhat.com","username":"jgilaber"},"date":"2026-07-15 16:05:11.000000000","message":"Uploaded patch set 7.\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":7},{"id":"31376ce600ca570717cc7b7c5507e2653676f801","author":{"_account_id":34452,"name":"Joan Gilabert","display_name":"jgilaber","email":"jgilaber@redhat.com","username":"jgilaber"},"date":"2026-07-15 17:25:31.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"88e3b39513fe837fb7484868cbf53d033e4eaeb2","author":{"_account_id":34452,"name":"Joan Gilabert","display_name":"jgilaber","email":"jgilaber@redhat.com","username":"jgilaber"},"date":"2026-07-16 07:07:30.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"a80f297d2d269acae4fdb722ab2058042bab0b50","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34452,"name":"Joan Gilabert","display_name":"jgilaber","email":"jgilaber@redhat.com","username":"jgilaber"},"date":"2026-07-16 07:24:37.000000000","message":"Uploaded patch set 8.","accounts_in_message":[],"_revision_number":8},{"id":"cc5f0f6ef288580f3d42b6c98ce88f8c25d766fd","tag":"autogenerated:zuul:automatic-ci","author":{"_account_id":28006,"name":"teim-ci","display_name":"teim-ci","email":"ci@seanmooney.info","username":"ci-sean-mooney","status":"this is a third-party ci account run by sean-k-mooney on irc\nhosted at zuul.teim.app"},"date":"2026-07-16 07:25:11.000000000","message":"Patch Set 8:\n\nStarting automatic-ci jobs.","accounts_in_message":[],"_revision_number":8},{"id":"a1389b624ebf8b5dc3ed4e7f149d8186ab8e2407","tag":"autogenerated:zuul:automatic-ci","author":{"_account_id":28006,"name":"teim-ci","display_name":"teim-ci","email":"ci@seanmooney.info","username":"ci-sean-mooney","status":"this is a third-party ci account run by sean-k-mooney on irc\nhosted at zuul.teim.app"},"date":"2026-07-16 07:44:41.000000000","message":"Patch Set 8:\n\n(1 comment)\n\nBuild succeeded (automatic-ci pipeline).\nhttps://zuul.teim.app/t/main/buildset/1a3ec6d38a314a02817727a4f8f77e86\n\n- teim-code-review https://zuul.teim.app/t/main/build/9d40c0b952d046978f3d1334669c8a78 : SUCCESS in 9m 45s","accounts_in_message":[],"_revision_number":8},{"id":"d704d95a3d080bec3f6192a1f8362704ad73a2fc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-16 07:48:11.000000000","message":"Patch Set 8: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/053a89feee524d9f9d76541c08db5aae\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/92e3ee71da974e4b8d478e8d5d5a6137 : SUCCESS in 5m 44s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/bfd6504b4a0f46a59258270e798c9204 : SUCCESS in 5m 40s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5a3a68d4b3dc4850a97693c7f4647cc9 : SUCCESS in 8m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5337e734cbf04f49964fc78d97a3e673 : SUCCESS in 2m 35s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/5c43ca2922f9483780b732c94611b65a : SUCCESS in 5m 01s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/cee69c2c93914fdf82c45935b067c54b : SUCCESS in 4m 58s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/2186546ce45d4790a319c2de4ccddb73 : SUCCESS in 4m 03s (non-voting)\n- cyborg-tempest-py3 https://zuul.opendev.org/t/openstack/build/80c7820a8b1b4f549c3277d3f310d0bf : FAILURE in 21m 24s","accounts_in_message":[],"_revision_number":8},{"id":"6ebb52b5571ca5793249801b56b87e962529393f","author":{"_account_id":34452,"name":"Joan Gilabert","display_name":"jgilaber","email":"jgilaber@redhat.com","username":"jgilaber"},"date":"2026-07-16 07:54:59.000000000","message":"Patch Set 8:\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"3493068e307c5c3aab11798fc1091ae87198ebb9","tag":"autogenerated:zuul:automatic-ci","author":{"_account_id":28006,"name":"teim-ci","display_name":"teim-ci","email":"ci@seanmooney.info","username":"ci-sean-mooney","status":"this is a third-party ci account run by sean-k-mooney on irc\nhosted at zuul.teim.app"},"date":"2026-07-16 07:55:36.000000000","message":"Patch Set 8:\n\nStarting automatic-ci jobs.","accounts_in_message":[],"_revision_number":8},{"id":"4f55e69b0fe7caaf9e6009a2678e68189e8ff80b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-16 08:25:35.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b40495506fd249139674ef4b5ee342a9\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/9be89e01f4594503996ce2b305c8c48a : SUCCESS in 4m 30s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7a6b7de53b4c436f92583f00f5db8e78 : SUCCESS in 4m 05s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f98e97a481634897af99965b1c00eb3a : SUCCESS in 7m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9b1ae4f593074df78d38355f31d0f185 : SUCCESS in 3m 48s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/2a183b436a8344288eda036ebd837b2c : SUCCESS in 5m 25s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/d9bcb6643a844a438f6def2d964570f7 : SUCCESS in 2m 34s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/d5885684c227410887b8d4249e39df8e : SUCCESS in 8m 26s (non-voting)\n- cyborg-tempest-py3 https://zuul.opendev.org/t/openstack/build/8fb522f7942f4063b158d7e67a1d5622 : SUCCESS in 27m 59s","accounts_in_message":[],"_revision_number":8},{"id":"e1d4340194dcc25ae8f2fc532279a7d6983f3db6","tag":"autogenerated:zuul:automatic-ci","author":{"_account_id":28006,"name":"teim-ci","display_name":"teim-ci","email":"ci@seanmooney.info","username":"ci-sean-mooney","status":"this is a third-party ci account run by sean-k-mooney on irc\nhosted at zuul.teim.app"},"date":"2026-07-16 08:33:47.000000000","message":"Patch Set 8:\n\n(1 comment)\n\nBuild succeeded (automatic-ci pipeline).\nhttps://zuul.teim.app/t/main/buildset/8f927cb3e8ed4163902b7a296baf51c8\n\n- teim-code-review https://zuul.teim.app/t/main/build/3308d92b10f84b5e99e8844afd4192f5 : SUCCESS in 10m 03s","accounts_in_message":[],"_revision_number":8},{"id":"c5ff060fd999f191a00ef75a14660d8f4e7b7ae0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34452,"name":"Joan Gilabert","display_name":"jgilaber","email":"jgilaber@redhat.com","username":"jgilaber"},"date":"2026-07-16 09:12:32.000000000","message":"Uploaded patch set 9.\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":9},{"id":"eead36113c3985717a832bf9a2793a3bd4438207","tag":"autogenerated:zuul:automatic-ci","author":{"_account_id":28006,"name":"teim-ci","display_name":"teim-ci","email":"ci@seanmooney.info","username":"ci-sean-mooney","status":"this is a third-party ci account run by sean-k-mooney on irc\nhosted at zuul.teim.app"},"date":"2026-07-16 09:13:05.000000000","message":"Patch Set 9:\n\nStarting automatic-ci jobs.","accounts_in_message":[],"_revision_number":9},{"id":"c2893d13a87a2066f350575b551786c98b309a51","tag":"autogenerated:zuul:automatic-ci","author":{"_account_id":28006,"name":"teim-ci","display_name":"teim-ci","email":"ci@seanmooney.info","username":"ci-sean-mooney","status":"this is a third-party ci account run by sean-k-mooney on irc\nhosted at zuul.teim.app"},"date":"2026-07-16 09:37:01.000000000","message":"Patch Set 9:\n\nBuild succeeded (automatic-ci pipeline).\nhttps://zuul.teim.app/t/main/buildset/dcbe87a32dd14b2db8117f0d42ca83f5\n\n- teim-code-review https://zuul.teim.app/t/main/build/8c9a26307c324ba29eef9b347a64e4bd : SUCCESS in 12m 18s","accounts_in_message":[],"_revision_number":9},{"id":"4f0701ed4564bbe566e3af83822322cb1144a55f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-16 10:12:29.000000000","message":"Patch Set 9: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/b4f5744387364030bd0046543dceeb32\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/bdd865b1b6554ee2ac66b672d8a11546 : SUCCESS in 4m 09s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/04aafffd29384088b384e5a57b52ef20 : SUCCESS in 4m 03s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cfbf0c181154470ba6b379060836c5ed : SUCCESS in 6m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1b88b4422cd544e5baff887546923701 : SUCCESS in 2m 29s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/b519f3008b884a1991565fd4610f38d4 : SUCCESS in 3m 31s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/c5fdced2853b4d918498575a34530ced : SUCCESS in 3m 00s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/c4adc8e020a848119013bc9df6665741 : SUCCESS in 4m 24s (non-voting)\n- cyborg-tempest-py3 https://zuul.opendev.org/t/openstack/build/a30729181bd747aba5a86ae7e9ecff9e : SUCCESS in 29m 03s\n- cyborg-tempest-ipv6 https://zuul.opendev.org/t/openstack/build/1928732448ca430bb56d50134f546ba1 : FAILURE in 42m 51s\n- cyborg-grenade https://zuul.opendev.org/t/openstack/build/e20091d3bfa947d9a1cc1635bdc171ec : SUCCESS in 27m 01s\n- cyborg-grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/3219a1e328304bc785f86eb4f36c9d31 : SUCCESS in 53m 00s","accounts_in_message":[],"_revision_number":9},{"id":"53279e71f4c1cccef4292fba1d39bdab067afad6","author":{"_account_id":34452,"name":"Joan Gilabert","display_name":"jgilaber","email":"jgilaber@redhat.com","username":"jgilaber"},"date":"2026-07-16 10:15:57.000000000","message":"Patch Set 9:\n\n(1 comment)","accounts_in_message":[],"_revision_number":9},{"id":"4014431c0a6115b1eb1829cf3702fd3304b68435","tag":"autogenerated:zuul:automatic-ci","author":{"_account_id":28006,"name":"teim-ci","display_name":"teim-ci","email":"ci@seanmooney.info","username":"ci-sean-mooney","status":"this is a third-party ci account run by sean-k-mooney on irc\nhosted at zuul.teim.app"},"date":"2026-07-16 10:16:30.000000000","message":"Patch Set 9:\n\nStarting automatic-ci jobs.","accounts_in_message":[],"_revision_number":9},{"id":"7281f3f70367605005a9c59ddf6a2cc7590e27fd","tag":"autogenerated:zuul:automatic-ci","author":{"_account_id":28006,"name":"teim-ci","display_name":"teim-ci","email":"ci@seanmooney.info","username":"ci-sean-mooney","status":"this is a third-party ci account run by sean-k-mooney on irc\nhosted at zuul.teim.app"},"date":"2026-07-16 10:30:41.000000000","message":"Patch Set 9:\n\n(1 comment)\n\nBuild succeeded (automatic-ci pipeline).\nhttps://zuul.teim.app/t/main/buildset/1afc68808ad04ac395aa084f420fd39a\n\n- teim-code-review https://zuul.teim.app/t/main/build/1830cd81c055452e92ef78a61829cef0 : SUCCESS in 14m 00s","accounts_in_message":[],"_revision_number":9},{"id":"ee89d9ed563645a5545fafdd7abe3f3cc78a4344","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-16 11:21:19.000000000","message":"Patch Set 9: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1beb9d62d8f14951a3e609626318474d\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f102b3f6dabe4a5f9189c6619a1057ce : SUCCESS in 4m 50s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/4f6c71029d8b4237972e0738320a006e : SUCCESS in 4m 29s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/91f5698244e54caea91d4ba2becdb574 : SUCCESS in 5m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e5520e47a29a406ea362e7f7e88d7a35 : SUCCESS in 2m 24s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/fa435e197e9b4e8eba99a331703d1568 : SUCCESS in 4m 06s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/41cd97958e6c4121891f233a6d1db3d1 : SUCCESS in 3m 48s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/6fceb2fe494f4af2a48cea0b8c16ccc5 : SUCCESS in 8m 51s (non-voting)\n- cyborg-tempest-py3 https://zuul.opendev.org/t/openstack/build/6fb7b9b5b9564a0fb66192bae03bb9e0 : SUCCESS in 52m 39s\n- cyborg-tempest-ipv6 https://zuul.opendev.org/t/openstack/build/a0717ae1df4f462d9f8ccf6622ebecbe : SUCCESS in 25m 50s\n- cyborg-grenade https://zuul.opendev.org/t/openstack/build/79b7b9652494415c965b5d61d36c69a7 : SUCCESS in 35m 59s\n- cyborg-grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/9c47ab1fa54446ef8f0eb677e1b55bfb : SUCCESS in 58m 17s","accounts_in_message":[],"_revision_number":9},{"id":"6bd6cfa3834a0046aba9528ad5b226923b28bc0d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34452,"name":"Joan Gilabert","display_name":"jgilaber","email":"jgilaber@redhat.com","username":"jgilaber"},"date":"2026-07-16 12:13:34.000000000","message":"Uploaded patch set 10.\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":10},{"id":"cbfc1d73a0aa398b2c5586c1e2894b842581ba5c","tag":"autogenerated:zuul:automatic-ci","author":{"_account_id":28006,"name":"teim-ci","display_name":"teim-ci","email":"ci@seanmooney.info","username":"ci-sean-mooney","status":"this is a third-party ci account run by sean-k-mooney on irc\nhosted at zuul.teim.app"},"date":"2026-07-16 12:14:28.000000000","message":"Patch Set 10:\n\nStarting automatic-ci jobs.","accounts_in_message":[],"_revision_number":10},{"id":"93ec08b91468be83136f6e066e4c16d54a9c83b6","tag":"autogenerated:zuul:automatic-ci","author":{"_account_id":28006,"name":"teim-ci","display_name":"teim-ci","email":"ci@seanmooney.info","username":"ci-sean-mooney","status":"this is a third-party ci account run by sean-k-mooney on irc\nhosted at zuul.teim.app"},"date":"2026-07-16 12:25:10.000000000","message":"Patch Set 10:\n\nBuild succeeded (automatic-ci pipeline).\nhttps://zuul.teim.app/t/main/buildset/de53d46e70d048b3825044851085d599\n\n- teim-code-review https://zuul.teim.app/t/main/build/01f673bb1ed24b869626b9dd0bbd2525 : SUCCESS in 10m 33s","accounts_in_message":[],"_revision_number":10},{"id":"dfbc88f05c40b7f5644a7b69627a8e5a5c95ce79","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-16 13:24:53.000000000","message":"Patch Set 10: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/31d60494ce334a76aa2ad6601de0f822\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d9f1bc764bcc46a48ff044a959f98115 : SUCCESS in 4m 28s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3d99484e750a4af7898d6263ded53f52 : SUCCESS in 4m 47s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/22cb943192ad445297d74db14c441d2e : SUCCESS in 10m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/35c2e58a245445878ec2d9eb7526674b : SUCCESS in 6m 02s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/6403674f27144bf3aa893da6372f6ce4 : SUCCESS in 4m 29s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/d0d36c2dc0e14a5c9c1677f6c14ebe5f : SUCCESS in 4m 37s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/fa43c1c6b6f449b6bcab76ac104955f9 : SUCCESS in 6m 14s (non-voting)\n- cyborg-tempest-py3 https://zuul.opendev.org/t/openstack/build/01ad138ac2da4c0a97d1198a3c7eb8f2 : FAILURE in 57m 44s\n- cyborg-tempest-ipv6 https://zuul.opendev.org/t/openstack/build/c53904182f004a138936e162016d4e72 : SUCCESS in 19m 28s\n- cyborg-grenade https://zuul.opendev.org/t/openstack/build/d3f268f342f84e388b51d688b12af26c : SUCCESS in 1h 03m 43s\n- cyborg-grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/b2e4b43a08094a62b5fa3e357f8b12f0 : SUCCESS in 57m 56s","accounts_in_message":[],"_revision_number":10},{"id":"966b67c4577cca78d3f76ff0ddbe598c830234b4","author":{"_account_id":34452,"name":"Joan Gilabert","display_name":"jgilaber","email":"jgilaber@redhat.com","username":"jgilaber"},"date":"2026-07-16 15:02:07.000000000","message":"Patch Set 10:\n\n(1 comment)","accounts_in_message":[],"_revision_number":10},{"id":"d233979957b976c7dee063de1422f5c0b4c287c5","tag":"autogenerated:zuul:automatic-ci","author":{"_account_id":28006,"name":"teim-ci","display_name":"teim-ci","email":"ci@seanmooney.info","username":"ci-sean-mooney","status":"this is a third-party ci account run by sean-k-mooney on irc\nhosted at zuul.teim.app"},"date":"2026-07-16 15:02:42.000000000","message":"Patch Set 10:\n\nStarting automatic-ci jobs.","accounts_in_message":[],"_revision_number":10},{"id":"95df53f9e7c670ab9a92998620e94758ecee5ffb","tag":"autogenerated:zuul:automatic-ci","author":{"_account_id":28006,"name":"teim-ci","display_name":"teim-ci","email":"ci@seanmooney.info","username":"ci-sean-mooney","status":"this is a third-party ci account run by sean-k-mooney on irc\nhosted at zuul.teim.app"},"date":"2026-07-16 15:22:51.000000000","message":"Patch Set 10:\n\nBuild succeeded (automatic-ci pipeline).\nhttps://zuul.teim.app/t/main/buildset/4fc15ccfbf8c409191d5bde0dc889283\n\n- teim-code-review https://zuul.teim.app/t/main/build/fec9b3c27d3c434c91b400a33a8a030f : SUCCESS in 11m 55s","accounts_in_message":[],"_revision_number":10},{"id":"4666a4a35e3f528a944c2ee774a62d461bbf7dfb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-16 16:25:49.000000000","message":"Patch Set 10:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/9dba91de3e11442db2c60bf24f6c9d6a\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/fb3fb27d2b314ec8932c7e66f4be57ed : FAILURE in 6m 25s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/69de100a3b0d4821847142b3df8374f9 : SUCCESS in 5m 00s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e192de2006244e228182a04de6c26fb5 : SUCCESS in 6m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/50da576bc6584857bb7752782cb5611f : SUCCESS in 2m 51s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/21bdd20fcbf84d41bb15dd7c3a47ac57 : SUCCESS in 2m 44s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/e860b988586c49db8f8a9e79e6f824d5 : SUCCESS in 5m 16s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/907f3c34bbc64412b67b5e9b7e5b974a : SUCCESS in 5m 31s (non-voting)\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/8911f66d05cb46869dfbbcfdd8cb716e : FAILURE in 4m 29s\n- cyborg-tempest-py3 https://zuul.opendev.org/t/openstack/build/8c04a3b3849247b7a2aa8a24e989a4a2 : SUCCESS in 57m 21s\n- cyborg-tempest-ipv6 https://zuul.opendev.org/t/openstack/build/7cafe0413db64cd48429b579262aea46 : SUCCESS in 51m 07s\n- cyborg-grenade https://zuul.opendev.org/t/openstack/build/61e3da5942394e91a3397b130d282023 : SUCCESS in 1h 11m 42s\n- cyborg-grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/6424fb1bbc5c432c8155a24bf934114c : SUCCESS in 58m 06s","accounts_in_message":[],"_revision_number":10},{"id":"7e25704da537a9f6d424be56d6bda9efba0ce581","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34452,"name":"Joan Gilabert","display_name":"jgilaber","email":"jgilaber@redhat.com","username":"jgilaber"},"date":"2026-07-16 16:37:26.000000000","message":"Uploaded patch set 11: Patch Set 10 was rebased.\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":11},{"id":"1e619de52d5250051eb1f3849fa445a04f275540","tag":"autogenerated:zuul:automatic-ci","author":{"_account_id":28006,"name":"teim-ci","display_name":"teim-ci","email":"ci@seanmooney.info","username":"ci-sean-mooney","status":"this is a third-party ci account run by sean-k-mooney on irc\nhosted at zuul.teim.app"},"date":"2026-07-16 16:39:03.000000000","message":"Patch Set 11:\n\nStarting automatic-ci jobs.","accounts_in_message":[],"_revision_number":11},{"id":"7ac38adfee6d2acc26c61cfded73324784602b48","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-16 17:41:24.000000000","message":"Patch Set 11: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/5546d063c45e4c47892c4a83cb7762f7\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/616d89bcd48a459fac50078187fad09d : FAILURE in 4m 52s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a4f19847d0114a478fdc69610a7a2f0d : SUCCESS in 5m 44s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/946b3f3a1abd41fca85fe3429a7354ea : SUCCESS in 9m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/329141ee4e2648348e6eee03d78a556c : SUCCESS in 5m 04s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/b7a5959171eb4c47b59b1bacc83f3328 : SUCCESS in 4m 24s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/bfcb827ec674465a9c6cc7646f3e9306 : SUCCESS in 6m 49s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/59223598448346139c023f37963894f0 : SUCCESS in 7m 13s (non-voting)\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/4ca63879f50a484c9bd3f86d263d40af : FAILURE in 4m 58s\n- cyborg-tempest-py3 https://zuul.opendev.org/t/openstack/build/bdb038a9e880437c8477beccf1977c7a : SUCCESS in 52m 13s\n- cyborg-tempest-ipv6 https://zuul.opendev.org/t/openstack/build/2c282c553b0d4d13b4d280e7944bd46a : SUCCESS in 53m 35s\n- cyborg-grenade https://zuul.opendev.org/t/openstack/build/ef2ec341e57b4cbead5a71961b16f648 : SUCCESS in 56m 32s\n- cyborg-grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/72b397fefa2841d78dfcd2dc0d6cc3d7 : SUCCESS in 41m 36s","accounts_in_message":[],"_revision_number":11},{"id":"3b9694fd32077347a7e84a7a3d0db581794965fc","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34452,"name":"Joan Gilabert","display_name":"jgilaber","email":"jgilaber@redhat.com","username":"jgilaber"},"date":"2026-07-16 17:45:15.000000000","message":"Uploaded patch set 12: Patch Set 11 was rebased.\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":12},{"id":"3b83cd26a0a87d637f451c16a78428ac0cbe3824","tag":"autogenerated:zuul:automatic-ci","author":{"_account_id":28006,"name":"teim-ci","display_name":"teim-ci","email":"ci@seanmooney.info","username":"ci-sean-mooney","status":"this is a third-party ci account run by sean-k-mooney on irc\nhosted at zuul.teim.app"},"date":"2026-07-16 17:46:36.000000000","message":"Patch Set 12:\n\nStarting automatic-ci jobs.","accounts_in_message":[],"_revision_number":12},{"id":"54ba597b4492dc3d99b509b3b8badb1f5897ec06","tag":"autogenerated:zuul:automatic-ci","author":{"_account_id":28006,"name":"teim-ci","display_name":"teim-ci","email":"ci@seanmooney.info","username":"ci-sean-mooney","status":"this is a third-party ci account run by sean-k-mooney on irc\nhosted at zuul.teim.app"},"date":"2026-07-16 18:36:39.000000000","message":"Patch Set 12:\n\nBuild succeeded (automatic-ci pipeline).\nhttps://zuul.teim.app/t/main/buildset/fb4d0643e1f1482f8543b0d8dd250b21\n\n- teim-code-review https://zuul.teim.app/t/main/build/19190323d1d244bd981c7ad3d7e82a0d : SUCCESS in 12m 54s","accounts_in_message":[],"_revision_number":12},{"id":"f0c7533ea505b765186b889794a66b856431e1ba","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-16 18:46:40.000000000","message":"Patch Set 12: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6f5bb06bbccc4f8fb3d9ff630d7e3b32\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f6c58f4d505d4121820da5351573a6e0 : SUCCESS in 5m 37s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/bceef9b8b7204fb4933a080760b59085 : SUCCESS in 4m 27s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fe50e17a74604d81bc614e11fe363b62 : SUCCESS in 8m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6ba93b3657004727ac4281a9c06c1770 : SUCCESS in 4m 13s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/768029cf734a46ad8e0a39a8f239d049 : SUCCESS in 5m 59s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/b3f6abd372b4487aa9313322b392fb49 : SUCCESS in 4m 19s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/98adc39b6a6d4dadafc1157597794629 : SUCCESS in 6m 29s (non-voting)\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/e895fcb0975b41b9983e016e8cdd45e9 : SUCCESS in 4m 23s\n- cyborg-tempest-py3 https://zuul.opendev.org/t/openstack/build/2c78b967aff9442b9c2f799a614e270b : SUCCESS in 51m 24s\n- cyborg-tempest-ipv6 https://zuul.opendev.org/t/openstack/build/1d72675b19a84772b506990a87ebb63e : SUCCESS in 45m 25s\n- cyborg-grenade https://zuul.opendev.org/t/openstack/build/22e2c35c8ac94cf2a3e418d891b18f92 : SUCCESS in 53m 48s\n- cyborg-grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/ac05f476dc544c57acae1009db4138e4 : SUCCESS in 36m 08s","accounts_in_message":[],"_revision_number":12},{"id":"bba34a9e6573ae057f18571f951f6d6bfe611dd1","author":{"_account_id":34452,"name":"Joan Gilabert","display_name":"jgilaber","email":"jgilaber@redhat.com","username":"jgilaber"},"date":"2026-07-17 06:46:12.000000000","message":"Patch Set 12:\n\n(1 comment)","accounts_in_message":[],"_revision_number":12},{"id":"9f624bdcdb2696948415f8466493a0e1ff06bb68","tag":"autogenerated:zuul:automatic-ci","author":{"_account_id":28006,"name":"teim-ci","display_name":"teim-ci","email":"ci@seanmooney.info","username":"ci-sean-mooney","status":"this is a third-party ci account run by sean-k-mooney on irc\nhosted at zuul.teim.app"},"date":"2026-07-17 06:46:43.000000000","message":"Patch Set 12:\n\nStarting automatic-ci jobs.","accounts_in_message":[],"_revision_number":12},{"id":"d0fe222f8d0c1370c49b7e44a17ae63d27962aa5","tag":"autogenerated:zuul:automatic-ci","author":{"_account_id":28006,"name":"teim-ci","display_name":"teim-ci","email":"ci@seanmooney.info","username":"ci-sean-mooney","status":"this is a third-party ci account run by sean-k-mooney on irc\nhosted at zuul.teim.app"},"date":"2026-07-17 07:24:14.000000000","message":"Patch Set 12:\n\n(1 comment)\n\nBuild succeeded (automatic-ci pipeline).\nhttps://zuul.teim.app/t/main/buildset/04fcd69ec5354a3bbf77af64e50132b6\n\n- teim-code-review https://zuul.teim.app/t/main/build/52d6f6760387416ead800818b1863db7 : SUCCESS in 12m 26s","accounts_in_message":[],"_revision_number":12},{"id":"16004ae614370dca1cf02d41c7a60a60e7b55f0d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-17 07:48:03.000000000","message":"Patch Set 12:\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c4578295a0584fed9f6f38bb05e04715\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/233b77dff147457194764b93c3fbc94c : SUCCESS in 3m 41s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7570faf3e98d4dff998ae5fa707dffb2 : SUCCESS in 4m 26s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c6e7d009901e40f49db26c062b2ef8a2 : SUCCESS in 12m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c0c0735701ab4677a4035269d1c9d784 : SUCCESS in 3m 38s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/94780fd5c5da4276a12579ddcddd4d5c : SUCCESS in 5m 22s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/476b6ece54464739a23ef99b229487da : SUCCESS in 3m 48s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/41c6d853593840aebde2fa940a0ccc54 : SUCCESS in 8m 25s (non-voting)\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/81b7696ddfe64313ba42cfe720ba479e : SUCCESS in 4m 01s\n- cyborg-tempest-py3 https://zuul.opendev.org/t/openstack/build/e65f9f4d59bf4853a421b9637a7cdd6b : SUCCESS in 36m 01s\n- cyborg-tempest-ipv6 https://zuul.opendev.org/t/openstack/build/95ac7e963bc045c184c7568d8fbc35e5 : SUCCESS in 46m 39s\n- cyborg-grenade https://zuul.opendev.org/t/openstack/build/538d753344e046d18b19d18799faf955 : SUCCESS in 55m 12s\n- cyborg-grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/a5e8960bf7774f27b222d11e382daec4 : SUCCESS in 53m 01s","accounts_in_message":[],"_revision_number":12},{"id":"2a84143f248f7dcf821d4e47831a4c5031e49bd8","author":{"_account_id":34452,"name":"Joan Gilabert","display_name":"jgilaber","email":"jgilaber@redhat.com","username":"jgilaber"},"date":"2026-07-17 08:48:57.000000000","message":"Patch Set 12:\n\n(1 comment)","accounts_in_message":[],"_revision_number":12},{"id":"f5df13efebaad962fc77b3a8316e567cb8862b9f","tag":"autogenerated:zuul:automatic-ci","author":{"_account_id":28006,"name":"teim-ci","display_name":"teim-ci","email":"ci@seanmooney.info","username":"ci-sean-mooney","status":"this is a third-party ci account run by sean-k-mooney on irc\nhosted at zuul.teim.app"},"date":"2026-07-17 08:49:34.000000000","message":"Patch Set 12:\n\nStarting automatic-ci jobs.","accounts_in_message":[],"_revision_number":12},{"id":"ea5e953e3205b08aaee60279e904f5252429ca88","tag":"autogenerated:zuul:automatic-ci","author":{"_account_id":28006,"name":"teim-ci","display_name":"teim-ci","email":"ci@seanmooney.info","username":"ci-sean-mooney","status":"this is a third-party ci account run by sean-k-mooney on irc\nhosted at zuul.teim.app"},"date":"2026-07-17 09:03:56.000000000","message":"Patch Set 12:\n\nBuild succeeded (automatic-ci pipeline).\nhttps://zuul.teim.app/t/main/buildset/cba631d7301a4a0bac62e6178f100ccb\n\n- teim-code-review https://zuul.teim.app/t/main/build/c7382ecd36764c8ca18266584b233a38 : SUCCESS in 14m 14s","accounts_in_message":[],"_revision_number":12},{"id":"6bfa7d8d4dfadacb5dbf0008d8b344ae96940e4b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-17 09:50:30.000000000","message":"Patch Set 12:\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/670c160ef21b44c89e65179bd740e803\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/42de26ffaa3e4edc93a0574db5cb28bc : SUCCESS in 5m 52s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/451291fcb3af4d35b0defb773943d8c7 : SUCCESS in 5m 00s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/afe1f607dea94446b0087bbb5e921807 : SUCCESS in 9m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/853b049959c94b54b6095dc40bc225c9 : SUCCESS in 3m 50s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/addd2c92904d43499cb9cbaa6be17533 : SUCCESS in 3m 52s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/4438d22ee01c4c2d9347d2d451888e57 : SUCCESS in 3m 46s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/b2c3a3fd63244e489dcd825d9fd4be1f : SUCCESS in 7m 02s (non-voting)\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/daf3e9e0112e493280c99f4dbee90281 : SUCCESS in 4m 21s\n- cyborg-tempest-py3 https://zuul.opendev.org/t/openstack/build/a66e9891fb4d45f7aa87e5c83a20ebd8 : SUCCESS in 54m 45s\n- cyborg-tempest-ipv6 https://zuul.opendev.org/t/openstack/build/418104e7d517431a9288b838b9a106be : SUCCESS in 41m 41s\n- cyborg-grenade https://zuul.opendev.org/t/openstack/build/a1e6a2228cca4fc683000a1a35c6bf3a : SUCCESS in 34m 30s\n- cyborg-grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/9e4cfe9a68f845d19113f48387f925f5 : SUCCESS in 28m 29s","accounts_in_message":[],"_revision_number":12},{"id":"d0b3accd5ad3ac1dafc32426d52053b58d2bf0d8","author":{"_account_id":34452,"name":"Joan Gilabert","display_name":"jgilaber","email":"jgilaber@redhat.com","username":"jgilaber"},"date":"2026-07-17 11:56:37.000000000","message":"Patch Set 12:\n\n(1 comment)","accounts_in_message":[],"_revision_number":12},{"id":"7c64cf55c66b6378e1a0ffaf5ed09fb7e6bc80b4","tag":"autogenerated:zuul:automatic-ci","author":{"_account_id":28006,"name":"teim-ci","display_name":"teim-ci","email":"ci@seanmooney.info","username":"ci-sean-mooney","status":"this is a third-party ci account run by sean-k-mooney on irc\nhosted at zuul.teim.app"},"date":"2026-07-17 11:57:09.000000000","message":"Patch Set 12:\n\nStarting automatic-ci jobs.","accounts_in_message":[],"_revision_number":12},{"id":"40313bee7d1f47b7b5325cd68f82ce071f050e92","tag":"autogenerated:zuul:automatic-ci","author":{"_account_id":28006,"name":"teim-ci","display_name":"teim-ci","email":"ci@seanmooney.info","username":"ci-sean-mooney","status":"this is a third-party ci account run by sean-k-mooney on irc\nhosted at zuul.teim.app"},"date":"2026-07-17 12:08:53.000000000","message":"Patch Set 12:\n\nBuild succeeded (automatic-ci pipeline).\nhttps://zuul.teim.app/t/main/buildset/850a318a508c4b2d8f343dffde03a0b8\n\n- teim-code-review https://zuul.teim.app/t/main/build/a82f2fba20974182827dc17c6e30d416 : SUCCESS in 11m 35s","accounts_in_message":[],"_revision_number":12},{"id":"7c9da28aa1878c042ba08ecb27c3b8df7b235b4a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-17 13:08:59.000000000","message":"Patch Set 12: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/12471a06107a43a5b93abe2ceb01d3ec\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1871d388f67044d5880a0b6b00f465f8 : SUCCESS in 5m 16s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/00f80aa896cf4a268a61fdf4733d33cf : SUCCESS in 4m 13s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e5d2000a0cca4e4882ae508bf7016c33 : SUCCESS in 8m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cb72ca03e93f4a1ca71a74e8305f2e3c : SUCCESS in 3m 50s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/c1e9d17baa294d9fb303dbaa42af94c3 : SUCCESS in 4m 19s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/0b45f968465d4c6997165a547a17eb39 : SUCCESS in 5m 28s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/b7f13e89259a4c8ea923806d69785167 : SUCCESS in 6m 08s (non-voting)\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/52502e8ff94246559a26f0bfbba2142c : SUCCESS in 5m 22s\n- cyborg-tempest-py3 https://zuul.opendev.org/t/openstack/build/8f47bc967d7042278b8afdceb2590d60 : SUCCESS in 51m 18s\n- cyborg-tempest-ipv6 https://zuul.opendev.org/t/openstack/build/87e90ed8ebbc4b7cad73c3d69d76c6a0 : FAILURE in 45m 25s\n- cyborg-grenade https://zuul.opendev.org/t/openstack/build/700edd220e3d46c6b8ad6d2dda2bc0ee : SUCCESS in 52m 44s\n- cyborg-grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/8bba335e9f81478bbc7a692f9bdf89c2 : SUCCESS in 1h 04m 15s","accounts_in_message":[],"_revision_number":12},{"id":"2f8cb7322f1d51c88c06e169fed5816d58242279","author":{"_account_id":34452,"name":"Joan Gilabert","display_name":"jgilaber","email":"jgilaber@redhat.com","username":"jgilaber"},"date":"2026-07-17 15:49:52.000000000","message":"Patch Set 12:\n\n(1 comment)","accounts_in_message":[],"_revision_number":12},{"id":"b1b4f3a22e275f1fa458ac90ba50ac17c7961533","tag":"autogenerated:zuul:automatic-ci","author":{"_account_id":28006,"name":"teim-ci","display_name":"teim-ci","email":"ci@seanmooney.info","username":"ci-sean-mooney","status":"this is a third-party ci account run by sean-k-mooney on irc\nhosted at zuul.teim.app"},"date":"2026-07-17 15:50:20.000000000","message":"Patch Set 12:\n\nStarting automatic-ci jobs.","accounts_in_message":[],"_revision_number":12},{"id":"db14b22abb5e08aa97d280304cc82adef93f207e","tag":"autogenerated:zuul:automatic-ci","author":{"_account_id":28006,"name":"teim-ci","display_name":"teim-ci","email":"ci@seanmooney.info","username":"ci-sean-mooney","status":"this is a third-party ci account run by sean-k-mooney on irc\nhosted at zuul.teim.app"},"date":"2026-07-17 16:17:45.000000000","message":"Patch Set 12:\n\nBuild succeeded (automatic-ci pipeline).\nhttps://zuul.teim.app/t/main/buildset/e78abbb1d02141a9be1ea5025f53b1f4\n\n- teim-code-review https://zuul.teim.app/t/main/build/66912ea55baf4b3ab8c4deebf6155585 : SUCCESS in 10m 58s","accounts_in_message":[],"_revision_number":12},{"id":"850454aac63c94f7c2f52747bd25e48e346b3c34","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-17 17:26:16.000000000","message":"Patch Set 12:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/d68dc750a40549e8a46c32a2f021a9dd\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/4eef336a32f24c898436aebd5a1d29ad : SUCCESS in 6m 27s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2f8e74f500804cc29aa061debcf23569 : SUCCESS in 5m 43s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ae3d1fa44bbc4477990fc1767db5d4f8 : SUCCESS in 8m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/927e927c15c846c881357d1481168b7e : SUCCESS in 4m 33s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/224b6ef7ad5441ab8879c1ce75c709eb : SUCCESS in 6m 03s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/bc021aa5ba9a42fa8007e98cfe76e316 : SUCCESS in 5m 33s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/9349f1080a704a69b0f3cb03a91eb674 : SUCCESS in 8m 13s (non-voting)\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/1885f8ab61f244f087a511f9aa2e2a6e : SUCCESS in 4m 51s\n- cyborg-tempest-py3 https://zuul.opendev.org/t/openstack/build/4f7620c77c6d4197bcef1b91d534c8e9 : FAILURE in 1h 27m 11s\n- cyborg-tempest-ipv6 https://zuul.opendev.org/t/openstack/build/ff21a681666249f9bf2e6af856e6ceac : FAILURE in 1h 06m 59s\n- cyborg-grenade https://zuul.opendev.org/t/openstack/build/3162cfdf463f45728af61c3eb67d7c2f : SUCCESS in 48m 59s\n- cyborg-grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/9d544d2749cf474297981d011caf47b9 : SUCCESS in 55m 35s","accounts_in_message":[],"_revision_number":12},{"id":"eaa19fd48a5eb586ec28cce533d4f90fc8849a3b","author":{"_account_id":34452,"name":"Joan Gilabert","display_name":"jgilaber","email":"jgilaber@redhat.com","username":"jgilaber"},"date":"2026-07-17 17:27:43.000000000","message":"Patch Set 12:\n\n(1 comment)","accounts_in_message":[],"_revision_number":12},{"id":"178fe1c554d0cb879076c896c4e07056952c1a20","tag":"autogenerated:zuul:automatic-ci","author":{"_account_id":28006,"name":"teim-ci","display_name":"teim-ci","email":"ci@seanmooney.info","username":"ci-sean-mooney","status":"this is a third-party ci account run by sean-k-mooney on irc\nhosted at zuul.teim.app"},"date":"2026-07-17 17:28:14.000000000","message":"Patch Set 12:\n\nStarting automatic-ci jobs.","accounts_in_message":[],"_revision_number":12},{"id":"604d755aec06cdc13d498f06580b95e3b92ebb19","tag":"autogenerated:zuul:automatic-ci","author":{"_account_id":28006,"name":"teim-ci","display_name":"teim-ci","email":"ci@seanmooney.info","username":"ci-sean-mooney","status":"this is a third-party ci account run by sean-k-mooney on irc\nhosted at zuul.teim.app"},"date":"2026-07-17 17:57:32.000000000","message":"Patch Set 12:\n\nBuild succeeded (automatic-ci pipeline).\nhttps://zuul.teim.app/t/main/buildset/7d41ae468d1c46fba24297ffc0ad0ba7\n\n- teim-code-review https://zuul.teim.app/t/main/build/4461167317534fb7876e7297b157c050 : SUCCESS in 13m 16s","accounts_in_message":[],"_revision_number":12},{"id":"82f5fb07cb9d3bbbf62ebe4eae5784507703505b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-17 18:33:17.000000000","message":"Patch Set 12: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/af0b334e09814d1e8db672fde18f095f\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/8ef4142ad00a4fe3ba140929c630edcc : SUCCESS in 4m 57s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/dbbbe0e965964d03aefa7dca69ba997e : SUCCESS in 4m 28s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/089d4a1f85dd48b7beb29184f86256f9 : SUCCESS in 7m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4324f904d5ef47ad91448b63c0b74e8a : SUCCESS in 4m 33s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/1af847340f4640918083fb72eb32bbb4 : SUCCESS in 3m 56s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/8c9d6dab3b29456b8cfc4975b27c8a38 : SUCCESS in 3m 40s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/622cea1096f346a78bd6e5a0e81dc0f3 : SUCCESS in 7m 39s (non-voting)\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/974e4bc0a975475398893a82cd546eba : SUCCESS in 5m 38s\n- cyborg-tempest-py3 https://zuul.opendev.org/t/openstack/build/234ac53086d543a8860e3bcd1990327f : SUCCESS in 46m 02s\n- cyborg-tempest-ipv6 https://zuul.opendev.org/t/openstack/build/9112005b481a446baf9ecd66e40ccecb : SUCCESS in 47m 26s\n- cyborg-grenade https://zuul.opendev.org/t/openstack/build/7ee177e680f5490ba6b436f095c25687 : SUCCESS in 53m 57s\n- cyborg-grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/188f07294dcf4f4bafdd7c09c4b48e74 : SUCCESS in 59m 01s","accounts_in_message":[],"_revision_number":12},{"id":"3f4a2791765a9f85c60e7df5623f70a31c325cdf","author":{"_account_id":34452,"name":"Joan Gilabert","display_name":"jgilaber","email":"jgilaber@redhat.com","username":"jgilaber"},"date":"2026-07-20 10:06:37.000000000","message":"Patch Set 12:\n\n(1 comment)","accounts_in_message":[],"_revision_number":12},{"id":"95ed2d318d0f0d7ab628551faba1500f9cc49aec","tag":"autogenerated:zuul:automatic-ci","author":{"_account_id":28006,"name":"teim-ci","display_name":"teim-ci","email":"ci@seanmooney.info","username":"ci-sean-mooney","status":"this is a third-party ci account run by sean-k-mooney on irc\nhosted at zuul.teim.app"},"date":"2026-07-20 10:07:16.000000000","message":"Patch Set 12:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.\nWarning:\n  Error merging opendev/openstack/cyborg for 997364,12","accounts_in_message":[],"_revision_number":12},{"id":"47294010d10fe21be04aef780276ec705c8a9dd7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-20 10:07:41.000000000","message":"Patch Set 12: Verified-1\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.\nWarning:\n  Error merging gerrit/openstack/cyborg for 997364,12","accounts_in_message":[],"_revision_number":12},{"id":"3fab388a171fb3937123006e07d4cccdebbde399","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34452,"name":"Joan Gilabert","display_name":"jgilaber","email":"jgilaber@redhat.com","username":"jgilaber"},"date":"2026-07-20 10:11:47.000000000","message":"Uploaded patch set 13.\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":13},{"id":"13c2f725ec85745e057534c4cf12ebb894cede17","tag":"autogenerated:zuul:automatic-ci","author":{"_account_id":28006,"name":"teim-ci","display_name":"teim-ci","email":"ci@seanmooney.info","username":"ci-sean-mooney","status":"this is a third-party ci account run by sean-k-mooney on irc\nhosted at zuul.teim.app"},"date":"2026-07-20 10:13:18.000000000","message":"Patch Set 13:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.\nWarning:\n  Error merging opendev/openstack/cyborg for 997048,2","accounts_in_message":[],"_revision_number":13},{"id":"1dd437858d5a9b21a7a32570bf329fdb679fffc0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34452,"name":"Joan Gilabert","display_name":"jgilaber","email":"jgilaber@redhat.com","username":"jgilaber"},"date":"2026-07-20 10:13:42.000000000","message":"Uploaded patch set 14.","accounts_in_message":[],"_revision_number":14},{"id":"5412fcea00515f3f84331a2922d5221d7d4946cd","tag":"autogenerated:zuul:automatic-ci","author":{"_account_id":28006,"name":"teim-ci","display_name":"teim-ci","email":"ci@seanmooney.info","username":"ci-sean-mooney","status":"this is a third-party ci account run by sean-k-mooney on irc\nhosted at zuul.teim.app"},"date":"2026-07-20 10:14:16.000000000","message":"Patch Set 14:\n\nStarting automatic-ci jobs.","accounts_in_message":[],"_revision_number":14},{"id":"5d9da3a7f08e02e62a8ec69ceb09616aa72cedba","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-20 11:17:06.000000000","message":"Patch Set 14: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/510dfc28e38849fe89ae45131aa44209\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7cf87eec5b7d403f9bd1e6144f3435e0 : SUCCESS in 5m 40s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/51bf83b0eb7942878ba13fe1bd07a2fe : SUCCESS in 3m 54s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6aae1ec124a846c1a4048bf4cea4151f : SUCCESS in 8m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3fa3a98d32d245d2bd7e8ea236995400 : SUCCESS in 4m 10s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/4f834e4451324357a0e51142ec1b6d98 : SUCCESS in 4m 03s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/529d44be30e64231bba985c2446b286a : SUCCESS in 3m 40s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/358f85daf93a4c5590bf213cacd26c7f : SUCCESS in 6m 06s (non-voting)\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/57327cf86dd34eaaad2e17da9599f23b : SUCCESS in 3m 37s\n- cyborg-tempest-py3 https://zuul.opendev.org/t/openstack/build/11f0b96881434732a9771786b45f1b26 : SUCCESS in 51m 21s\n- cyborg-tempest-ipv6 https://zuul.opendev.org/t/openstack/build/310b0bbd9107425bab887fe4eaeb0327 : SUCCESS in 40m 20s\n- cyborg-grenade https://zuul.opendev.org/t/openstack/build/6c4dafa20f1e42a1be7efc11079142aa : SUCCESS in 53m 24s\n- cyborg-grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/4c60453143214b25acd538c4aee6ee94 : SUCCESS in 51m 59s","accounts_in_message":[],"_revision_number":14},{"id":"c977cf0843359fae625b3f1dec1872cd837b00e5","tag":"autogenerated:zuul:automatic-ci","author":{"_account_id":28006,"name":"teim-ci","display_name":"teim-ci","email":"ci@seanmooney.info","username":"ci-sean-mooney","status":"this is a third-party ci account run by sean-k-mooney on irc\nhosted at zuul.teim.app"},"date":"2026-07-20 12:06:37.000000000","message":"Patch Set 14:\n\nBuild succeeded (automatic-ci pipeline).\nhttps://zuul.teim.app/t/main/buildset/b06eb1abc9c54e2e981d1e6c07e677dc\n\n- teim-code-review https://zuul.teim.app/t/main/build/64c74cd17f11429996da689a53c72309 : SUCCESS in 11m 04s","accounts_in_message":[],"_revision_number":14},{"id":"60696837f3f4c135dffffdda8941e4272b1c2b73","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34452,"name":"Joan Gilabert","display_name":"jgilaber","email":"jgilaber@redhat.com","username":"jgilaber"},"date":"2026-07-20 12:50:20.000000000","message":"Uploaded patch set 15: Patch Set 14 was rebased.\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":15},{"id":"36a0e3f4ac0b96dab0e39bb52f4f23f631bdce23","tag":"autogenerated:zuul:automatic-ci","author":{"_account_id":28006,"name":"teim-ci","display_name":"teim-ci","email":"ci@seanmooney.info","username":"ci-sean-mooney","status":"this is a third-party ci account run by sean-k-mooney on irc\nhosted at zuul.teim.app"},"date":"2026-07-20 12:51:55.000000000","message":"Patch Set 15:\n\nStarting automatic-ci jobs.","accounts_in_message":[],"_revision_number":15},{"id":"166e6d1c0dc5113a433b1e4c14d18080499669ea","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-20 13:56:34.000000000","message":"Patch Set 15: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8a4dac4db4a54571989b3c3394e50569\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d2f0a46ceafd4903b558812e1b243881 : SUCCESS in 7m 55s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d657d362a94a4481b50db28c2958f108 : SUCCESS in 6m 44s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2ddc7ef77da7443ab9c1fe05b519ae0e : SUCCESS in 9m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/eccff4b652f440fca9764d178a757593 : SUCCESS in 6m 18s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/76d8097bdfc54e64a8e0c428f7b76541 : SUCCESS in 5m 08s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/a40edf25bfd440bfbdc4132fd3522893 : SUCCESS in 6m 26s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/e8eb814898654ec489a4aef24e85718f : SUCCESS in 6m 55s (non-voting)\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/edf9b950393e4268ad83bff6e77d188b : SUCCESS in 5m 51s\n- cyborg-tempest-py3 https://zuul.opendev.org/t/openstack/build/eaca055c4ec8472b99e1bd0757d966cc : SUCCESS in 58m 37s\n- cyborg-tempest-ipv6 https://zuul.opendev.org/t/openstack/build/fbdc4f751e15489085da5a1df9e32abc : SUCCESS in 32m 32s\n- cyborg-grenade https://zuul.opendev.org/t/openstack/build/a1d547f969eb4dc29818f2bea3d4842a : SUCCESS in 54m 54s\n- cyborg-grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/bcf9093a9a084c8890d76e5468dea14e : SUCCESS in 55m 59s","accounts_in_message":[],"_revision_number":15},{"id":"71b8e7ef3b89369d530cd809d6f444afc9c25b5e","author":{"_account_id":34452,"name":"Joan Gilabert","display_name":"jgilaber","email":"jgilaber@redhat.com","username":"jgilaber"},"date":"2026-07-20 14:02:53.000000000","message":"Patch Set 15:\n\n(1 comment)","accounts_in_message":[],"_revision_number":15},{"id":"940e7ab112b0b667595f395d979d44d07b751a21","tag":"autogenerated:zuul:automatic-ci","author":{"_account_id":28006,"name":"teim-ci","display_name":"teim-ci","email":"ci@seanmooney.info","username":"ci-sean-mooney","status":"this is a third-party ci account run by sean-k-mooney on irc\nhosted at zuul.teim.app"},"date":"2026-07-20 14:55:17.000000000","message":"Patch Set 15:\n\nBuild succeeded (automatic-ci pipeline).\nhttps://zuul.teim.app/t/main/buildset/cfa1b6c224de4907b0cef283ccff1ab4\n\n- teim-code-review https://zuul.teim.app/t/main/build/5145fd4f988d44de98182151abe3138a : SUCCESS in 12m 11s","accounts_in_message":[],"_revision_number":15},{"id":"f155aba3e805cdbc870f9314ef061925dd0ee767","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-20 14:59:51.000000000","message":"Patch Set 15:\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/945cf4b10d0a4bd8974fff1a8a26eff0\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/605a6c116064444a877b4843cf303bb9 : SUCCESS in 6m 56s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/70cb8e3083d943f0932563f39504a6e6 : SUCCESS in 6m 33s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/79085686dc424bfd907b82584dd53073 : SUCCESS in 10m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/caa43b0633834db58fcce290a850295c : SUCCESS in 4m 22s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/6191a1313b90419fa39b48f0d7c67700 : SUCCESS in 4m 30s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/10b2ae58d10e41bb9f99b32eb4605aaf : SUCCESS in 6m 50s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/fa2f2fe625c04992bbfba35d32493c2c : SUCCESS in 9m 15s (non-voting)\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/0d4ee8cb894d4567a1a549618a9c6266 : SUCCESS in 6m 27s\n- cyborg-tempest-py3 https://zuul.opendev.org/t/openstack/build/76c77536ad6e47b9a811e03e8eb304ee : SUCCESS in 51m 09s\n- cyborg-tempest-ipv6 https://zuul.opendev.org/t/openstack/build/df4dc35766db41ddae8ead7c2c0c8aa7 : SUCCESS in 38m 48s\n- cyborg-grenade https://zuul.opendev.org/t/openstack/build/3e647f246e204d22b1b535811932c5c6 : SUCCESS in 45m 17s\n- cyborg-grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/3a12856898f343f1b901f0ef08643035 : SUCCESS in 39m 47s","accounts_in_message":[],"_revision_number":15},{"id":"4fec8219afcce8ffcf00e2591a2dd3624f4046d0","author":{"_account_id":34452,"name":"Joan Gilabert","display_name":"jgilaber","email":"jgilaber@redhat.com","username":"jgilaber"},"date":"2026-07-20 15:09:24.000000000","message":"Patch Set 15:\n\n(1 comment)","accounts_in_message":[],"_revision_number":15},{"id":"80d99e25d9df98ca824944fbc36689c382254028","tag":"autogenerated:zuul:automatic-ci","author":{"_account_id":28006,"name":"teim-ci","display_name":"teim-ci","email":"ci@seanmooney.info","username":"ci-sean-mooney","status":"this is a third-party ci account run by sean-k-mooney on irc\nhosted at zuul.teim.app"},"date":"2026-07-20 15:09:59.000000000","message":"Patch Set 15:\n\nStarting automatic-ci jobs.","accounts_in_message":[],"_revision_number":15},{"id":"ffc9c02c254ba33629e0dc28f5efd69b3f85c360","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-20 16:17:04.000000000","message":"Patch Set 15:\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f75730350fcf4b3f98131cc79991d7f8\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ff9584dc72c847fd807afd53207096b9 : SUCCESS in 4m 58s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/eaec69fbabf94848891727352e83847f : SUCCESS in 4m 46s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8710a20c7c294b958a23cf081ef5298e : SUCCESS in 8m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/56af26115afd4f2aa5f869765bcb3565 : SUCCESS in 5m 12s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/d06ea7c76b114d65bc1158a7570e1bbc : SUCCESS in 5m 51s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/56c66e8440e242c19f0e8a4f67e65ad4 : SUCCESS in 4m 46s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/ca891ad79a444418ac967a8cd9425e6d : SUCCESS in 7m 33s (non-voting)\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/a683cd731474489aa1adc4d55f8fc816 : SUCCESS in 5m 09s\n- cyborg-tempest-py3 https://zuul.opendev.org/t/openstack/build/9c4dba6f901b48c285adc97b376a9f45 : SUCCESS in 53m 38s\n- cyborg-tempest-ipv6 https://zuul.opendev.org/t/openstack/build/245056b4c1ed49ee8a9db2b6a92cfba2 : SUCCESS in 34m 07s\n- cyborg-grenade https://zuul.opendev.org/t/openstack/build/6839e5d2963c40f28c569a4ec28fb9e1 : SUCCESS in 1h 00m 30s\n- cyborg-grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/f2386df5ff5b4f5fa5f51e54d5621581 : SUCCESS in 53m 03s","accounts_in_message":[],"_revision_number":15},{"id":"974f4f6537cb60b38a550b5731c4251e31464eae","tag":"autogenerated:zuul:automatic-ci","author":{"_account_id":28006,"name":"teim-ci","display_name":"teim-ci","email":"ci@seanmooney.info","username":"ci-sean-mooney","status":"this is a third-party ci account run by sean-k-mooney on irc\nhosted at zuul.teim.app"},"date":"2026-07-20 16:50:37.000000000","message":"Patch Set 15:\n\n(1 comment)\n\nBuild succeeded (automatic-ci pipeline).\nhttps://zuul.teim.app/t/main/buildset/5d1178f6de3546439e2dac8d7babbde1\n\n- teim-code-review https://zuul.teim.app/t/main/build/41c0813d53f24f5e99578084e375c890 : SUCCESS in 9m 48s","accounts_in_message":[],"_revision_number":15},{"id":"c2d98013fd8094580e427cbcdda256d1752782a5","author":{"_account_id":34452,"name":"Joan Gilabert","display_name":"jgilaber","email":"jgilaber@redhat.com","username":"jgilaber"},"date":"2026-07-21 07:55:25.000000000","message":"Patch Set 15:\n\n(1 comment)","accounts_in_message":[],"_revision_number":15},{"id":"28c5b094401782cc11ac882b96c004795c2c6ebd","tag":"autogenerated:zuul:automatic-ci","author":{"_account_id":28006,"name":"teim-ci","display_name":"teim-ci","email":"ci@seanmooney.info","username":"ci-sean-mooney","status":"this is a third-party ci account run by sean-k-mooney on irc\nhosted at zuul.teim.app"},"date":"2026-07-21 07:55:59.000000000","message":"Patch Set 15:\n\nStarting automatic-ci jobs.","accounts_in_message":[],"_revision_number":15},{"id":"6c96773c122cff35eb938b9577c7f8f5de4c4a62","tag":"autogenerated:zuul:automatic-ci","author":{"_account_id":28006,"name":"teim-ci","display_name":"teim-ci","email":"ci@seanmooney.info","username":"ci-sean-mooney","status":"this is a third-party ci account run by sean-k-mooney on irc\nhosted at zuul.teim.app"},"date":"2026-07-21 08:30:51.000000000","message":"Patch Set 15:\n\nBuild succeeded (automatic-ci pipeline).\nhttps://zuul.teim.app/t/main/buildset/a7e97867f7ac4ce1965109379f332c37\n\n- teim-code-review https://zuul.teim.app/t/main/build/95c1415fd7b2475d986e01f44d4df6b6 : SUCCESS in 9m 41s","accounts_in_message":[],"_revision_number":15},{"id":"b1a0de0dbf4e8558ef08b78f2e933a7240c21dca","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-21 08:50:05.000000000","message":"Patch Set 15: Verified-1\n\nThis change depends on a change that failed to merge.\n\nChange https://review.opendev.org/c/openstack/cyborg-tempest-plugin/+/997352 is needed.","accounts_in_message":[],"_revision_number":15},{"id":"9f5b31bed3abcf8749b86206d8d64f635efda663","author":{"_account_id":34452,"name":"Joan Gilabert","display_name":"jgilaber","email":"jgilaber@redhat.com","username":"jgilaber"},"date":"2026-07-21 08:56:31.000000000","message":"Patch Set 15:\n\n(1 comment)","accounts_in_message":[],"_revision_number":15},{"id":"1e37ca2608cb056ee4bcbf30e54e16a7cc0a77af","tag":"autogenerated:zuul:automatic-ci","author":{"_account_id":28006,"name":"teim-ci","display_name":"teim-ci","email":"ci@seanmooney.info","username":"ci-sean-mooney","status":"this is a third-party ci account run by sean-k-mooney on irc\nhosted at zuul.teim.app"},"date":"2026-07-21 08:57:07.000000000","message":"Patch Set 15:\n\nStarting automatic-ci jobs.","accounts_in_message":[],"_revision_number":15},{"id":"b413aa81039f6a28838dd982680bb654b0cf7032","tag":"autogenerated:zuul:automatic-ci","author":{"_account_id":28006,"name":"teim-ci","display_name":"teim-ci","email":"ci@seanmooney.info","username":"ci-sean-mooney","status":"this is a third-party ci account run by sean-k-mooney on irc\nhosted at zuul.teim.app"},"date":"2026-07-21 09:28:36.000000000","message":"Patch Set 15:\n\nBuild succeeded (automatic-ci pipeline).\nhttps://zuul.teim.app/t/main/buildset/b470495a70144cc5bdd478a0d0a971f7\n\n- teim-code-review https://zuul.teim.app/t/main/build/e3f68faccb3e4093a573775279ce5031 : SUCCESS in 10m 55s\n\nWarning:\n  The commit message body contains \u0027appropiate\u0027 which should be \u0027appropriate\u0027. While minor, this is visible in the permanent git history and can be trivially corrected before merge. Impact: Typo in permanent git history. Minor but trivially fixable by amending the commit message. Recommendation: Amend the commit message to fix \u0027appropiate\u0027 to \u0027appropriate\u0027.","accounts_in_message":[],"_revision_number":15},{"id":"973eab698a05836ed8061d9978aa0e5dfde5c1a7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-21 10:14:11.000000000","message":"Patch Set 15: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/118a836150c34685acc7a729797558c0\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d3f20334737640a493ffa3e16eeb3de1 : SUCCESS in 6m 47s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b5fc3e6939644dcb9e919eaac30832cb : SUCCESS in 5m 11s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e7b8b5a4718649f7804ee230b840d28f : SUCCESS in 8m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d030ec2d6bbf49a9a4a09b7c4f01da96 : SUCCESS in 5m 04s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/f53cfd3569384b139d4db1fc41a2606f : SUCCESS in 5m 10s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/91a3b908f2ac418c8ee8c5ad13c23854 : SUCCESS in 4m 21s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/310530cc4dc24c6f8f98eaa7433e78cf : SUCCESS in 6m 46s (non-voting)\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/be63b321fb8d46dfa9467221bfefec8b : SUCCESS in 6m 30s\n- cyborg-tempest-py3 https://zuul.opendev.org/t/openstack/build/3b9e277274af44d8ba44f5f59592fceb : SUCCESS in 56m 17s\n- cyborg-tempest-ipv6 https://zuul.opendev.org/t/openstack/build/d1bd24e6d8964e09b3577c1d4eefa20a : SUCCESS in 54m 33s\n- cyborg-grenade https://zuul.opendev.org/t/openstack/build/5b67b963c8104590a88316f70c18b385 : SUCCESS in 1h 09m 53s\n- cyborg-grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/47f9a42e30044c5d9be5f059f00ff003 : SUCCESS in 1h 03m 51s","accounts_in_message":[],"_revision_number":15},{"id":"155c04900b1bd791deb43c3da87d32a1c4b6c356","author":{"_account_id":34452,"name":"Joan Gilabert","display_name":"jgilaber","email":"jgilaber@redhat.com","username":"jgilaber"},"date":"2026-07-21 17:02:54.000000000","message":"Patch Set 15:\n\n(1 comment)","accounts_in_message":[],"_revision_number":15},{"id":"a021c6449bb185bc103c5690e6824767c95440e4","tag":"autogenerated:zuul:automatic-ci","author":{"_account_id":28006,"name":"teim-ci","display_name":"teim-ci","email":"ci@seanmooney.info","username":"ci-sean-mooney","status":"this is a third-party ci account run by sean-k-mooney on irc\nhosted at zuul.teim.app"},"date":"2026-07-21 17:03:28.000000000","message":"Patch Set 15:\n\nStarting automatic-ci jobs.","accounts_in_message":[],"_revision_number":15},{"id":"443eedd62636151374b412d87259998dad601484","tag":"autogenerated:zuul:automatic-ci","author":{"_account_id":28006,"name":"teim-ci","display_name":"teim-ci","email":"ci@seanmooney.info","username":"ci-sean-mooney","status":"this is a third-party ci account run by sean-k-mooney on irc\nhosted at zuul.teim.app"},"date":"2026-07-21 17:30:39.000000000","message":"Patch Set 15:\n\nBuild succeeded (automatic-ci pipeline).\nhttps://zuul.teim.app/t/main/buildset/d85b12b0bbe542b7a6d216bdc221d42a\n\n- teim-code-review https://zuul.teim.app/t/main/build/873f4d8857004d03b1dc191a4d670ffa : SUCCESS in 10m 31s","accounts_in_message":[],"_revision_number":15},{"id":"655fdfb26a53e90a8df0b2cab3ea1112212aaa19","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-21 18:10:29.000000000","message":"Patch Set 15:\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6e28fa400d4143c5b7a9c1a056a9ffc3\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/4924dc8bd506410fb497a174bf7817bf : SUCCESS in 5m 10s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/15e5729148da42beb98d28aa416cc7ab : SUCCESS in 4m 28s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a0f4fed6dd234fc7b8fa5c9ec2963742 : SUCCESS in 4m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/15c380e6f32d48d2b0ca0fea28626ba1 : SUCCESS in 4m 12s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/c163e0df97c4450aa0b0775755d531cc : SUCCESS in 4m 14s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/a95c886d758747219596b3aa3b5247e1 : SUCCESS in 4m 04s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/45febc2740f74ae688c19da23dfb6d51 : SUCCESS in 6m 41s (non-voting)\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/484203d609394632a7b11ce449429918 : SUCCESS in 4m 37s\n- cyborg-tempest-py3 https://zuul.opendev.org/t/openstack/build/7884a8554bb74802b5fce869dedc03b7 : SUCCESS in 55m 20s\n- cyborg-tempest-ipv6 https://zuul.opendev.org/t/openstack/build/125180be57af47b59dfd282a08e58a66 : SUCCESS in 34m 54s\n- cyborg-grenade https://zuul.opendev.org/t/openstack/build/58a591eba7324b2f9397141e91acd1bb : SUCCESS in 43m 12s\n- cyborg-grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/cbeb7c16c7b54161a9cfdcf4e9a19deb : SUCCESS in 53m 03s","accounts_in_message":[],"_revision_number":15},{"id":"c79a58719f17f98183d66e6b866f81a91cc2de45","author":{"_account_id":34452,"name":"Joan Gilabert","display_name":"jgilaber","email":"jgilaber@redhat.com","username":"jgilaber"},"date":"2026-07-22 08:56:40.000000000","message":"Patch Set 15:\n\n(1 comment)","accounts_in_message":[],"_revision_number":15},{"id":"d0f02a327b6a809675c5115811b062f4f2cd96a1","tag":"autogenerated:zuul:automatic-ci","author":{"_account_id":28006,"name":"teim-ci","display_name":"teim-ci","email":"ci@seanmooney.info","username":"ci-sean-mooney","status":"this is a third-party ci account run by sean-k-mooney on irc\nhosted at zuul.teim.app"},"date":"2026-07-22 08:57:21.000000000","message":"Patch Set 15:\n\nStarting automatic-ci jobs.","accounts_in_message":[],"_revision_number":15},{"id":"d3115b3e1b01cc72ca046c7b1922c99577f28c2e","tag":"autogenerated:zuul:automatic-ci","author":{"_account_id":28006,"name":"teim-ci","display_name":"teim-ci","email":"ci@seanmooney.info","username":"ci-sean-mooney","status":"this is a third-party ci account run by sean-k-mooney on irc\nhosted at zuul.teim.app"},"date":"2026-07-22 09:35:07.000000000","message":"Patch Set 15:\n\nBuild succeeded (automatic-ci pipeline).\nhttps://zuul.teim.app/t/main/buildset/220782d0723c4e64b2f31fc3d8eaace6\n\n- teim-code-review https://zuul.teim.app/t/main/build/84e2dc995f0345d1aac3eea4784288bf : SUCCESS in 13m 02s","accounts_in_message":[],"_revision_number":15},{"id":"1f2a899f83197f1309e14eababc1c380e0033757","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-22 10:07:17.000000000","message":"Patch Set 15: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/d8b2d5edd7614bddbd8f88a9e1bde8da\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b507ef1d341e4a518ca7c1108afaaa8a : SUCCESS in 5m 14s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/151b50d48402455a9404d7d5c83cc9dd : SUCCESS in 4m 36s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fad9e1c221b945928ca970f12a2fe428 : SUCCESS in 4m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1d1b164451544f5abcd26a36cbefc7b9 : FAILURE in 3m 49s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/04e7d66d8d384a21a47ddd632dfedb8d : SUCCESS in 7m 02s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/3a0a90a71e884a40a0f3a895f11ef29f : SUCCESS in 3m 07s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/f64218df536b475d99c406d7cc451625 : SUCCESS in 6m 31s (non-voting)\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/028af178a59a455bab0858c62fc1c0a3 : SUCCESS in 6m 06s\n- cyborg-tempest-py3 https://zuul.opendev.org/t/openstack/build/4527e35848fc49a9bcb71b1b8faa2128 : FAILURE in 58m 30s\n- cyborg-tempest-ipv6 https://zuul.opendev.org/t/openstack/build/110e8722b4db4ca78f687dd7e823023f : FAILURE in 49m 07s\n- cyborg-grenade https://zuul.opendev.org/t/openstack/build/8ee8f0e3bc634b65ba1c72785a0e65b8 : SUCCESS in 55m 29s\n- cyborg-grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/cfaac9beff20446e8e53e7db06faf4d7 : SUCCESS in 45m 03s\n- cyborg-tempest-srbac https://zuul.opendev.org/t/openstack/build/defefe8d788f440f90bb4ab990ebcbdf : FAILURE in 1h 03m 35s\n\nWarning:\n  Comments left for invalid file pci-sim/run_fake_pci_mdev_smoke.sh\n  Comments left for invalid file pci-sim/run_fake_pci_qemu_mdev_smoke.sh","accounts_in_message":[],"_revision_number":15},{"id":"6caa792e614dd6a89a138518d947aba2e397f0dd","author":{"_account_id":34452,"name":"Joan Gilabert","display_name":"jgilaber","email":"jgilaber@redhat.com","username":"jgilaber"},"date":"2026-07-22 10:19:06.000000000","message":"Patch Set 15:\n\n(1 comment)","accounts_in_message":[],"_revision_number":15},{"id":"145a046d4353e8b0f9bc5a04d0d37d2db4a01a43","tag":"autogenerated:zuul:automatic-ci","author":{"_account_id":28006,"name":"teim-ci","display_name":"teim-ci","email":"ci@seanmooney.info","username":"ci-sean-mooney","status":"this is a third-party ci account run by sean-k-mooney on irc\nhosted at zuul.teim.app"},"date":"2026-07-22 10:19:41.000000000","message":"Patch Set 15:\n\nStarting automatic-ci jobs.","accounts_in_message":[],"_revision_number":15},{"id":"79fcaecc833db7f9989921d8fffd0bf1ff3df522","tag":"autogenerated:zuul:automatic-ci","author":{"_account_id":28006,"name":"teim-ci","display_name":"teim-ci","email":"ci@seanmooney.info","username":"ci-sean-mooney","status":"this is a third-party ci account run by sean-k-mooney on irc\nhosted at zuul.teim.app"},"date":"2026-07-22 11:09:14.000000000","message":"Patch Set 15:\n\nBuild succeeded (automatic-ci pipeline).\nhttps://zuul.teim.app/t/main/buildset/48073a0112f846f9bc04d8a494985670\n\n- teim-code-review https://zuul.teim.app/t/main/build/a1213fa8a65d48378e801a783d0f0b7b : SUCCESS in 23m 07s","accounts_in_message":[],"_revision_number":15},{"id":"6a707656d349d725c6d666f3f0055caca7e6ac35","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-22 11:23:43.000000000","message":"Patch Set 15:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/2b442e022bcc4c7e914a0f9a047b2c31\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b9b72641f1044174a83d616d94aa89a5 : SUCCESS in 6m 50s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/65e7a6a798a5455bbc132e808e2fbeac : SUCCESS in 5m 11s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/43858bacd90e4ecaa84de83271be5436 : SUCCESS in 9m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4340b04f5dbf4bac875cf61e254f7349 : FAILURE in 4m 54s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/c91954fd6d524ee9a305f5d0f1fc219d : SUCCESS in 3m 17s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/366aa69566c9406f92f05bbda293efb0 : SUCCESS in 4m 01s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/1b0234d6fe6c4ac793b8d97ec888d3ba : SUCCESS in 6m 35s (non-voting)\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/2a59190744ed464b8d0f6287fc4a3782 : SUCCESS in 5m 42s\n- cyborg-tempest-py3 https://zuul.opendev.org/t/openstack/build/acacc000796a4393a90cd87bd7acb34a : SUCCESS in 31m 31s\n- cyborg-tempest-ipv6 https://zuul.opendev.org/t/openstack/build/76875af28fa54c749196285982944bc5 : SUCCESS in 42m 34s\n- cyborg-grenade https://zuul.opendev.org/t/openstack/build/dd4eb6f0f911438985bbade644cf0a20 : SUCCESS in 58m 02s\n- cyborg-grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/e606ed1a3e61484ebe927f916c1cd925 : SUCCESS in 43m 59s\n- cyborg-tempest-srbac https://zuul.opendev.org/t/openstack/build/8c5f39c3caeb417da25e7c59579a7f8c : SUCCESS in 53m 23s\n\nWarning:\n  Comments left for invalid file pci-sim/run_fake_pci_mdev_smoke.sh\n  Comments left for invalid file pci-sim/run_fake_pci_qemu_mdev_smoke.sh","accounts_in_message":[],"_revision_number":15},{"id":"13a36e03f1e180c3e6a86e73c5d9fdb15abce2df","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34452,"name":"Joan Gilabert","display_name":"jgilaber","email":"jgilaber@redhat.com","username":"jgilaber"},"date":"2026-07-22 11:41:22.000000000","message":"Uploaded patch set 16: Patch Set 15 was rebased. Commit message was updated.\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":16},{"id":"7132f3fe98106a10c5d2aa085e5ad051c6354a62","tag":"autogenerated:zuul:automatic-ci","author":{"_account_id":28006,"name":"teim-ci","display_name":"teim-ci","email":"ci@seanmooney.info","username":"ci-sean-mooney","status":"this is a third-party ci account run by sean-k-mooney on irc\nhosted at zuul.teim.app"},"date":"2026-07-22 11:43:19.000000000","message":"Patch Set 16:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.\nWarning:\n  Error merging opendev/openstack/cyborg for 997048,4","accounts_in_message":[],"_revision_number":16},{"id":"f22f0a53bcede1e9e8cf31ba70ee6666e934cb88","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-22 12:47:58.000000000","message":"Patch Set 16: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6cbacb6566de4c40b3da96dfc36692ac\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f24c8ccd1b2844e8a117624ed6239df3 : SUCCESS in 5m 21s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1d334dec65ad4815825b956c2418e015 : SUCCESS in 4m 38s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6f5e6415045e44ba979839da75f37f8d : SUCCESS in 8m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/07fb64f447984e89bd68cc0dd28f3cf7 : SUCCESS in 4m 34s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/91543e9fcac64116a68f4cdd57577b36 : SUCCESS in 2m 50s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/16be8be38a894777ac7a4ba8baace951 : SUCCESS in 4m 17s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/318ebc6793b044f694fa2e58adf162c7 : SUCCESS in 7m 15s (non-voting)\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/6e02fcf85a0a4e6398e3e28950748457 : SUCCESS in 4m 34s\n- cyborg-tempest-py3 https://zuul.opendev.org/t/openstack/build/ec5617bc8e254ee0ad58608ac1f7b692 : SUCCESS in 53m 08s\n- cyborg-tempest-ipv6 https://zuul.opendev.org/t/openstack/build/e7c1004e452c40e9a02a3e1441a4474f : SUCCESS in 42m 08s\n- cyborg-grenade https://zuul.opendev.org/t/openstack/build/e62ea348b0a24824bee9565c844bddd2 : SUCCESS in 58m 42s\n- cyborg-grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/d853a78123c1422ab0cbb331b6234ea3 : SUCCESS in 55m 28s\n- cyborg-tempest-srbac https://zuul.opendev.org/t/openstack/build/7cbe37db527d419295afa1e73695294a : SUCCESS in 56m 06s","accounts_in_message":[],"_revision_number":16},{"id":"3e33460d00acbfe2d98c63716315b909e2bb12e0","author":{"_account_id":34452,"name":"Joan Gilabert","display_name":"jgilaber","email":"jgilaber@redhat.com","username":"jgilaber"},"date":"2026-07-22 13:21:22.000000000","message":"Patch Set 16:\n\n(1 comment)","accounts_in_message":[],"_revision_number":16},{"id":"1fffec44a46584100450d0e6e3a27c7e0c784bbe","tag":"autogenerated:zuul:automatic-ci","author":{"_account_id":28006,"name":"teim-ci","display_name":"teim-ci","email":"ci@seanmooney.info","username":"ci-sean-mooney","status":"this is a third-party ci account run by sean-k-mooney on irc\nhosted at zuul.teim.app"},"date":"2026-07-22 13:21:57.000000000","message":"Patch Set 16:\n\nStarting automatic-ci jobs.","accounts_in_message":[],"_revision_number":16},{"id":"2688607a0d93b354b65a64dfe295c2a352d641c5","tag":"autogenerated:zuul:automatic-ci","author":{"_account_id":28006,"name":"teim-ci","display_name":"teim-ci","email":"ci@seanmooney.info","username":"ci-sean-mooney","status":"this is a third-party ci account run by sean-k-mooney on irc\nhosted at zuul.teim.app"},"date":"2026-07-22 13:47:26.000000000","message":"Patch Set 16:\n\nBuild succeeded (automatic-ci pipeline).\nhttps://zuul.teim.app/t/main/buildset/a4748e6e1f024f269cbf36ad748fa991\n\n- teim-code-review https://zuul.teim.app/t/main/build/3ae739b70e6f4f08888b45945b9f7de9 : SUCCESS in 11m 07s","accounts_in_message":[],"_revision_number":16},{"id":"c804b45bd07faac1b4cecc6cee234bba9e7f8913","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-22 14:32:58.000000000","message":"Patch Set 16:\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a42acc11e3554079b0c8993762df5303\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c86b093394964d4599de9f3d3a6ff2b9 : SUCCESS in 6m 05s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d483ab0b544e46d6b7a741926e204d5c : SUCCESS in 4m 59s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e4a34dcba71b4028b7e1756960ba3b33 : SUCCESS in 8m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/303a930ebf2249b2950b6f821a03d771 : SUCCESS in 4m 46s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/94a866137b6d4232bea8e1bc37af7a1d : SUCCESS in 5m 30s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/879b624dc21d4ee386e16241573513cd : SUCCESS in 5m 18s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/9aea6f99c0be42119c130b801ab92742 : SUCCESS in 6m 48s (non-voting)\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/dadd45c3b8cd44a3875116fba5b0f52a : SUCCESS in 4m 25s\n- cyborg-tempest-py3 https://zuul.opendev.org/t/openstack/build/53634a7c9312460984898f350f6ebdfe : SUCCESS in 1h 02m 56s\n- cyborg-tempest-ipv6 https://zuul.opendev.org/t/openstack/build/2fc97636ddd9465686d388f79cbf0a5f : SUCCESS in 54m 02s\n- cyborg-grenade https://zuul.opendev.org/t/openstack/build/bf7f0b10d5b449f396c5feabfeeaea80 : SUCCESS in 58m 12s\n- cyborg-grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/8f76e0fae9854ef79073ff511287eb34 : SUCCESS in 1h 04m 21s\n- cyborg-tempest-srbac https://zuul.opendev.org/t/openstack/build/b26540bb379241dc92de0645096bd9e8 : SUCCESS in 1h 00m 34s","accounts_in_message":[],"_revision_number":16},{"id":"ab9bdaa4ff59047c563d07bfaa774dc2201ffb55","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34452,"name":"Joan Gilabert","display_name":"jgilaber","email":"jgilaber@redhat.com","username":"jgilaber"},"date":"2026-07-22 14:57:35.000000000","message":"Uploaded patch set 17: Patch Set 16 was rebased.\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":17},{"id":"1f5ea1a5aa9211788e2d47b2c8c43948184a940c","tag":"autogenerated:zuul:automatic-ci","author":{"_account_id":28006,"name":"teim-ci","display_name":"teim-ci","email":"ci@seanmooney.info","username":"ci-sean-mooney","status":"this is a third-party ci account run by sean-k-mooney on irc\nhosted at zuul.teim.app"},"date":"2026-07-22 14:59:13.000000000","message":"Patch Set 17:\n\nStarting automatic-ci jobs.","accounts_in_message":[],"_revision_number":17},{"id":"ca5763b5b6d16705cc899de2712a5acd27a6f5f4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-22 16:19:06.000000000","message":"Patch Set 17: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6ae97c1d504e4f77b53cb6d2a5ba3111\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/3572f470fbb743f4b5e0a264fc3d025c : SUCCESS in 5m 38s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/68347a3f4005435b837a9f7fcf8e8099 : SUCCESS in 4m 51s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3a26d0d5239d4a3889cd49b30e200705 : SUCCESS in 9m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ca41962b9ffa4677b6ca6caf31ce33ac : SUCCESS in 5m 33s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/46d99789a6f94fee838ddbcb113bdef8 : SUCCESS in 6m 37s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/972b6a107e4f494e84b509a4a7a7fe7c : SUCCESS in 6m 43s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/17984deff02445309ede9b1375f3610e : SUCCESS in 8m 48s (non-voting)\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/778b4176695d497da540dd164b1bb75d : SUCCESS in 6m 02s\n- cyborg-tempest-py3 https://zuul.opendev.org/t/openstack/build/a1b4ca0b6744448fb985931e727e4bbf : SUCCESS in 53m 24s\n- cyborg-tempest-ipv6 https://zuul.opendev.org/t/openstack/build/c356442ac95f49818d4ad2066775ad14 : SUCCESS in 45m 15s\n- cyborg-grenade https://zuul.opendev.org/t/openstack/build/9328507806f34a689348da67dec3ba93 : SUCCESS in 1h 03m 21s\n- cyborg-grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/84183f7ac038438f8481152075421a52 : SUCCESS in 42m 54s\n- cyborg-tempest-srbac https://zuul.opendev.org/t/openstack/build/6b7d528c3c974d689bf0d3aa1871d1a6 : SUCCESS in 1h 06m 20s","accounts_in_message":[],"_revision_number":17},{"id":"4e97e203da692e287e17a7907bbfe7727e806ed9","tag":"autogenerated:zuul:automatic-ci","author":{"_account_id":28006,"name":"teim-ci","display_name":"teim-ci","email":"ci@seanmooney.info","username":"ci-sean-mooney","status":"this is a third-party ci account run by sean-k-mooney on irc\nhosted at zuul.teim.app"},"date":"2026-07-22 16:43:41.000000000","message":"Patch Set 17:\n\n(1 comment)\n\nBuild succeeded (automatic-ci pipeline).\nhttps://zuul.teim.app/t/main/buildset/0a3428e3facb49228818d2bf2c351fd0\n\n- teim-code-review https://zuul.teim.app/t/main/build/9f4e56a1338e4e7ca1628e672db10289 : SUCCESS in 10m 33s","accounts_in_message":[],"_revision_number":17},{"id":"166ce70f1a85dc8bd301893b1a2d298e7bbbd038","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34452,"name":"Joan Gilabert","display_name":"jgilaber","email":"jgilaber@redhat.com","username":"jgilaber"},"date":"2026-07-23 10:16:02.000000000","message":"Uploaded patch set 18: Patch Set 17 was rebased.\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":18},{"id":"c732412e3edf24e01aeda63d0c3976c3b33d12ca","tag":"autogenerated:zuul:automatic-ci","author":{"_account_id":28006,"name":"teim-ci","display_name":"teim-ci","email":"ci@seanmooney.info","username":"ci-sean-mooney","status":"this is a third-party ci account run by sean-k-mooney on irc\nhosted at zuul.teim.app"},"date":"2026-07-23 10:17:15.000000000","message":"Patch Set 18:\n\nStarting automatic-ci jobs.","accounts_in_message":[],"_revision_number":18},{"id":"b8eba9f99af1adf297b79e3b15ed98b70d3d6c56","tag":"autogenerated:zuul:automatic-ci","author":{"_account_id":28006,"name":"teim-ci","display_name":"teim-ci","email":"ci@seanmooney.info","username":"ci-sean-mooney","status":"this is a third-party ci account run by sean-k-mooney on irc\nhosted at zuul.teim.app"},"date":"2026-07-23 10:47:29.000000000","message":"Patch Set 18:\n\nBuild succeeded (automatic-ci pipeline).\nhttps://zuul.teim.app/t/main/buildset/e3fba4c1ab7549498f63179b2e653e15\n\n- teim-code-review https://zuul.teim.app/t/main/build/748620138eec4af3b0fc61838e483dd7 : SUCCESS in 10m 32s","accounts_in_message":[],"_revision_number":18},{"id":"95fac71eb5eff65fb667a70ddfde24ab3eec96ec","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-23 11:33:15.000000000","message":"Patch Set 18: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/cbd4afd96438475991e7c22d37065104\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c1ce9348935943ec9c1e627e0fb9f332 : SUCCESS in 5m 46s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/28368e42834e4728a3db77946c87e000 : SUCCESS in 5m 30s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2450db6a6e1743f19f87950f18011cdb : SUCCESS in 8m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/083c0e05fe6c4ac18e7b297ab43b8aed : SUCCESS in 4m 42s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/1276976344df4c82b68912590ca50348 : SUCCESS in 4m 37s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/61e5832e457e4496a05418979286705f : SUCCESS in 4m 56s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/0eafffb4198e454d9699eeef03ece662 : SUCCESS in 8m 30s (non-voting)\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/7acd2afa0f424223bad2632bae7366d0 : SUCCESS in 6m 19s\n- cyborg-tempest-py3 https://zuul.opendev.org/t/openstack/build/6f752a066f324fbb84e33b1f45102a53 : FAILURE in 1h 01m 01s\n- cyborg-tempest-ipv6 https://zuul.opendev.org/t/openstack/build/cf46932b09cf4ec6986f669fc7cc938f : FAILURE in 45m 56s\n- cyborg-grenade https://zuul.opendev.org/t/openstack/build/632db6441aae49829a7fe7d9e94c0ea7 : SUCCESS in 1h 04m 06s\n- cyborg-grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/174947c7e7b24d388a601fb07fe16f43 : SUCCESS in 1h 01m 35s\n- cyborg-tempest-srbac https://zuul.opendev.org/t/openstack/build/f183679245d44afcbae5cd28b919dabe : FAILURE in 53m 04s","accounts_in_message":[],"_revision_number":18},{"id":"805af1b05a0d673ce427b23c5507af337eb70ec9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34452,"name":"Joan Gilabert","display_name":"jgilaber","email":"jgilaber@redhat.com","username":"jgilaber"},"date":"2026-07-23 13:01:51.000000000","message":"Uploaded patch set 19: Patch Set 18 was rebased.\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":19},{"id":"c19c2f5d1cf4f066b85a89ffcee3e27e536f87cb","tag":"autogenerated:zuul:automatic-ci","author":{"_account_id":28006,"name":"teim-ci","display_name":"teim-ci","email":"ci@seanmooney.info","username":"ci-sean-mooney","status":"this is a third-party ci account run by sean-k-mooney on irc\nhosted at zuul.teim.app"},"date":"2026-07-23 13:02:44.000000000","message":"Patch Set 19:\n\nStarting automatic-ci jobs.","accounts_in_message":[],"_revision_number":19},{"id":"2c6fa9ad6ea809a286efead57ba5797044a23b40","tag":"autogenerated:zuul:automatic-ci","author":{"_account_id":28006,"name":"teim-ci","display_name":"teim-ci","email":"ci@seanmooney.info","username":"ci-sean-mooney","status":"this is a third-party ci account run by sean-k-mooney on irc\nhosted at zuul.teim.app"},"date":"2026-07-23 13:33:53.000000000","message":"Patch Set 19:\n\nBuild succeeded (automatic-ci pipeline).\nhttps://zuul.teim.app/t/main/buildset/605892a10f9c4d12be6a53a302ac654b\n\n- teim-code-review https://zuul.teim.app/t/main/build/9f22e6269fea4bbb80d6a6ccb6a73721 : SUCCESS in 14m 48s","accounts_in_message":[],"_revision_number":19},{"id":"e6100503166d9bc764e3bd04fd1ee07d51151fd8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-23 14:19:38.000000000","message":"Patch Set 19: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/ea6de4c7f58644b785fc76fe00677837\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/71b1d20568384709bc62cb9a98aaa6c6 : SUCCESS in 7m 20s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/544351bea7974b87ba8299cc9abc8f2a : SUCCESS in 5m 22s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/79064295d42744ccac6bc4f83e7d1995 : SUCCESS in 10m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/da88d34693c04bdb82025109c98cb319 : SUCCESS in 4m 39s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/654e509bb23a4b9999525593a0b3f8b7 : SUCCESS in 4m 54s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/68236a4234d24f3bbfa856b46b250458 : SUCCESS in 4m 25s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/4987649afa5e4af7877fa0a0c2ae6a1a : SUCCESS in 8m 18s (non-voting)\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/b4f594f8a1834250b2a3ece179b6bbe7 : SUCCESS in 5m 20s\n- cyborg-tempest-py3 https://zuul.opendev.org/t/openstack/build/63942866ce224e5f806c949eab54c36f : FAILURE in 1h 01m 10s\n- cyborg-tempest-ipv6 https://zuul.opendev.org/t/openstack/build/2ac02e276ab34dfba214057a588d9acd : FAILURE in 47m 15s\n- cyborg-grenade https://zuul.opendev.org/t/openstack/build/79f3e73a396441e3b86165664fa89a23 : SUCCESS in 58m 38s\n- cyborg-grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/44d2663e4425407b84bf559f73dfda65 : SUCCESS in 45m 14s\n- cyborg-tempest-srbac https://zuul.opendev.org/t/openstack/build/16f496b5ceb7445394328ea8f3944141 : FAILURE in 44m 27s","accounts_in_message":[],"_revision_number":19},{"id":"b812a3527366f6a8cbf7239c37bef4558b1f149f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34452,"name":"Joan Gilabert","display_name":"jgilaber","email":"jgilaber@redhat.com","username":"jgilaber"},"date":"2026-07-23 14:54:34.000000000","message":"Uploaded patch set 20: Patch Set 19 was rebased.\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":20},{"id":"3ed2d0a6d9070c6fa83c020ef98428d6c315f8d2","tag":"autogenerated:zuul:automatic-ci","author":{"_account_id":28006,"name":"teim-ci","display_name":"teim-ci","email":"ci@seanmooney.info","username":"ci-sean-mooney","status":"this is a third-party ci account run by sean-k-mooney on irc\nhosted at zuul.teim.app"},"date":"2026-07-23 14:55:27.000000000","message":"Patch Set 20:\n\nStarting automatic-ci jobs.","accounts_in_message":[],"_revision_number":20},{"id":"6fbf449d7eb6dfd0ee5843a0e165db23625f7725","tag":"autogenerated:zuul:automatic-ci","author":{"_account_id":28006,"name":"teim-ci","display_name":"teim-ci","email":"ci@seanmooney.info","username":"ci-sean-mooney","status":"this is a third-party ci account run by sean-k-mooney on irc\nhosted at zuul.teim.app"},"date":"2026-07-23 15:23:07.000000000","message":"Patch Set 20:\n\nBuild succeeded (automatic-ci pipeline).\nhttps://zuul.teim.app/t/main/buildset/20355666b51449d1b8deb20ac5f449c2\n\n- teim-code-review https://zuul.teim.app/t/main/build/37bfde57369342bc80be46c0ae2228d2 : SUCCESS in 13m 03s","accounts_in_message":[],"_revision_number":20},{"id":"0cfb5eb9e6af2926c30e618631071740fd4bff16","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-23 16:05:38.000000000","message":"Patch Set 20: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/d8d4ae6ed39a4f59881fc2386ff2602a\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/00f0afe17376487fabeb44e60ce83f48 : SUCCESS in 5m 32s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ffe3aa383d9246fbb7b4b2cdb0d7c0b1 : SUCCESS in 4m 40s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/256607151b744e2295d6755a8bbcaaad : SUCCESS in 9m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7f1ba2d9d1f04138a7cc7a62c6eaed4e : SUCCESS in 6m 23s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/ac077d9f4cd34d8f8fcef505e2b2ae81 : SUCCESS in 4m 17s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/2fbb0f2c8e1b4acbb47a85b559c81445 : SUCCESS in 4m 34s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/4dc76b47de91447b9c0359c421f6c9fe : SUCCESS in 7m 10s (non-voting)\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/5e622d9c2302484bbdcaade8233b6779 : SUCCESS in 5m 33s\n- cyborg-tempest-py3 https://zuul.opendev.org/t/openstack/build/4f44a99f243840458ed94b84e386041a : FAILURE in 44m 31s\n- cyborg-tempest-ipv6 https://zuul.opendev.org/t/openstack/build/5b313a8cd68440eeb2988757e2b97107 : FAILURE in 48m 26s\n- cyborg-grenade https://zuul.opendev.org/t/openstack/build/35d1dd8c38034fb8b7ff47fbb7b333a2 : SUCCESS in 56m 59s\n- cyborg-grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/3a70615063bf4e83acc6268a2e0c0a72 : SUCCESS in 53m 04s\n- cyborg-tempest-srbac https://zuul.opendev.org/t/openstack/build/df00ea161b12412fbc67edda8a2223a2 : FAILURE in 54m 55s","accounts_in_message":[],"_revision_number":20},{"id":"6fc1b57302b5b9a423364ac530c987be89278579","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34452,"name":"Joan Gilabert","display_name":"jgilaber","email":"jgilaber@redhat.com","username":"jgilaber"},"date":"2026-07-23 16:19:30.000000000","message":"Uploaded patch set 21: Patch Set 20 was rebased.\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":21},{"id":"35abdf42828653ddcd6ea6408d1fff90acb7dd3a","tag":"autogenerated:zuul:automatic-ci","author":{"_account_id":28006,"name":"teim-ci","display_name":"teim-ci","email":"ci@seanmooney.info","username":"ci-sean-mooney","status":"this is a third-party ci account run by sean-k-mooney on irc\nhosted at zuul.teim.app"},"date":"2026-07-23 16:20:26.000000000","message":"Patch Set 21:\n\nStarting automatic-ci jobs.","accounts_in_message":[],"_revision_number":21},{"id":"246caa601bb6a20140126b491d07b5031a899ab6","tag":"autogenerated:zuul:automatic-ci","author":{"_account_id":28006,"name":"teim-ci","display_name":"teim-ci","email":"ci@seanmooney.info","username":"ci-sean-mooney","status":"this is a third-party ci account run by sean-k-mooney on irc\nhosted at zuul.teim.app"},"date":"2026-07-23 16:39:53.000000000","message":"Patch Set 21:\n\nBuild succeeded (automatic-ci pipeline).\nhttps://zuul.teim.app/t/main/buildset/73152441886346819a0762693e33d64b\n\n- teim-code-review https://zuul.teim.app/t/main/build/51779a5734e84f659c6c1023a0391f4b : SUCCESS in 10m 03s","accounts_in_message":[],"_revision_number":21},{"id":"90bc0877b94bb03b87fdb5589dd9fda317272285","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-23 17:32:57.000000000","message":"Patch Set 21: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/8bba4335179d43ab897d3a2f0033f040\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/37d920e7aec945bcb954aea571f7f554 : SUCCESS in 4m 59s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ac887328180e42a5a04edf24ad0c81a3 : SUCCESS in 4m 36s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/468401e25f5841909bba5ddf9390f2aa : SUCCESS in 8m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/81780d19ed8d46f5b05e168fc3ee041e : SUCCESS in 4m 37s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/72537a16c53a48c8bfa4a1ca914fe8ed : SUCCESS in 4m 44s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/5b9f2bbb516c46568ca7ec17470d4abd : SUCCESS in 4m 32s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/86552653d47b445b91d84acb4b868fda : SUCCESS in 7m 03s (non-voting)\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/709bc3ea173345dd89463c0fd982fbdc : SUCCESS in 4m 31s\n- cyborg-tempest-py3 https://zuul.opendev.org/t/openstack/build/e66d9213831c4be4ada33519ac184b30 : FAILURE in 43m 42s\n- cyborg-tempest-ipv6 https://zuul.opendev.org/t/openstack/build/04e1334e73fc481cba34c0d93541d25d : FAILURE in 37m 18s\n- cyborg-grenade https://zuul.opendev.org/t/openstack/build/5a618eedb7f4436bad81c9ac850b60fa : SUCCESS in 48m 21s\n- cyborg-grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/d3f73429967849a8891237f9f0aab0af : SUCCESS in 1h 06m 28s\n- cyborg-tempest-srbac https://zuul.opendev.org/t/openstack/build/0d03695448584efa81652b572f6eb408 : FAILURE in 36m 31s","accounts_in_message":[],"_revision_number":21},{"id":"90d5d82e55d0917542edee4e60cc7beb6148c062","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34452,"name":"Joan Gilabert","display_name":"jgilaber","email":"jgilaber@redhat.com","username":"jgilaber"},"date":"2026-07-23 17:37:07.000000000","message":"Uploaded patch set 22.\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":22},{"id":"fbdccd56f3978c4d7b211c5c84fabb9636956199","tag":"autogenerated:zuul:automatic-ci","author":{"_account_id":28006,"name":"teim-ci","display_name":"teim-ci","email":"ci@seanmooney.info","username":"ci-sean-mooney","status":"this is a third-party ci account run by sean-k-mooney on irc\nhosted at zuul.teim.app"},"date":"2026-07-23 17:38:02.000000000","message":"Patch Set 22:\n\nStarting automatic-ci jobs.","accounts_in_message":[],"_revision_number":22},{"id":"90dbbdf492dfdfcfe09a2460b254b67f79eb18ff","tag":"autogenerated:zuul:automatic-ci","author":{"_account_id":28006,"name":"teim-ci","display_name":"teim-ci","email":"ci@seanmooney.info","username":"ci-sean-mooney","status":"this is a third-party ci account run by sean-k-mooney on irc\nhosted at zuul.teim.app"},"date":"2026-07-23 18:02:45.000000000","message":"Patch Set 22:\n\n(1 comment)\n\nBuild succeeded (automatic-ci pipeline).\nhttps://zuul.teim.app/t/main/buildset/18a5f68d87864843b383308b60252abe\n\n- teim-code-review https://zuul.teim.app/t/main/build/9a6a2fe8fa0345b6a103d1a3b1ebd29d : SUCCESS in 14m 31s","accounts_in_message":[],"_revision_number":22},{"id":"871e4620e931efb95131f3c593687017c09b9376","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-23 18:46:17.000000000","message":"Patch Set 22: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/255c5a9f7271440e8a717f7d7ca48313\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/5bccea0aba3647718ae3b499d2a20c93 : SUCCESS in 5m 24s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/6d479b64c4de40c5915c58add96b6358 : SUCCESS in 4m 34s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/efd635129c7c45af87b27d8f007a206b : SUCCESS in 8m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b9cea1dc79c545e897bf51dd5420602f : FAILURE in 2m 49s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/d7e93e849d5f4c11ae6256e5329e6187 : SUCCESS in 4m 29s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/3f69ab5edf7343909f08a8e3a9f68bfe : SUCCESS in 4m 03s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/1135cde5282143949484fd92994faba4 : SUCCESS in 7m 29s (non-voting)\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/139a9e9b6ed64e16a47936464e5476c7 : SUCCESS in 4m 55s\n- cyborg-tempest-py3 https://zuul.opendev.org/t/openstack/build/15a7fea3309941d98dc7bbba26972864 : FAILURE in 52m 45s\n- cyborg-tempest-ipv6 https://zuul.opendev.org/t/openstack/build/64853bae76384c6a82bb968ecbec76d6 : FAILURE in 24m 28s\n- cyborg-grenade https://zuul.opendev.org/t/openstack/build/4cc8fccf6c304141a496155086ae1b60 : SUCCESS in 1h 01m 32s\n- cyborg-grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/82e3cae5f2e34f998fd49c8e0a86cdf5 : SUCCESS in 26m 33s\n- cyborg-tempest-srbac https://zuul.opendev.org/t/openstack/build/e791ae584acf4b3b830c1ab4885473de : FAILURE in 55m 12s\n\nWarning:\n  Comments left for invalid file devstack/plugin.sh","accounts_in_message":[],"_revision_number":22},{"id":"380f861e4afab16b3554e8ae4b5f000aee802122","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34452,"name":"Joan Gilabert","display_name":"jgilaber","email":"jgilaber@redhat.com","username":"jgilaber"},"date":"2026-07-24 06:37:58.000000000","message":"Uploaded patch set 23: Patch Set 22 was rebased.\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":23},{"id":"0621ade0bebcf264fb4b245c433bcf8225f21f1f","tag":"autogenerated:zuul:automatic-ci","author":{"_account_id":28006,"name":"teim-ci","display_name":"teim-ci","email":"ci@seanmooney.info","username":"ci-sean-mooney","status":"this is a third-party ci account run by sean-k-mooney on irc\nhosted at zuul.teim.app"},"date":"2026-07-24 06:39:10.000000000","message":"Patch Set 23:\n\nStarting automatic-ci jobs.","accounts_in_message":[],"_revision_number":23},{"id":"b6abaa56501e5a040c617ceab6b22db2cd20362a","tag":"autogenerated:zuul:automatic-ci","author":{"_account_id":28006,"name":"teim-ci","display_name":"teim-ci","email":"ci@seanmooney.info","username":"ci-sean-mooney","status":"this is a third-party ci account run by sean-k-mooney on irc\nhosted at zuul.teim.app"},"date":"2026-07-24 07:11:49.000000000","message":"Patch Set 23:\n\n(1 comment)\n\nBuild succeeded (automatic-ci pipeline).\nhttps://zuul.teim.app/t/main/buildset/55d25992a6154085b4bfe595a738de5e\n\n- teim-code-review https://zuul.teim.app/t/main/build/e1afa444ab13404ca68292793c12d91c : SUCCESS in 21m 24s","accounts_in_message":[],"_revision_number":23},{"id":"6fbe2eed3953e2589bb9478aa1c32e35bb4aa5a3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34452,"name":"Joan Gilabert","display_name":"jgilaber","email":"jgilaber@redhat.com","username":"jgilaber"},"date":"2026-07-24 07:19:49.000000000","message":"Uploaded patch set 24: Patch Set 23 was rebased.","accounts_in_message":[],"_revision_number":24},{"id":"907505c5738183bc896678aabd3627ebdcea008e","tag":"autogenerated:zuul:automatic-ci","author":{"_account_id":28006,"name":"teim-ci","display_name":"teim-ci","email":"ci@seanmooney.info","username":"ci-sean-mooney","status":"this is a third-party ci account run by sean-k-mooney on irc\nhosted at zuul.teim.app"},"date":"2026-07-24 07:20:45.000000000","message":"Patch Set 24:\n\nStarting automatic-ci jobs.","accounts_in_message":[],"_revision_number":24},{"id":"4c2e26f324e7e6f4a96d53a6320163ea2c461956","tag":"autogenerated:zuul:automatic-ci","author":{"_account_id":28006,"name":"teim-ci","display_name":"teim-ci","email":"ci@seanmooney.info","username":"ci-sean-mooney","status":"this is a third-party ci account run by sean-k-mooney on irc\nhosted at zuul.teim.app"},"date":"2026-07-24 07:38:53.000000000","message":"Patch Set 24:\n\nBuild succeeded (automatic-ci pipeline).\nhttps://zuul.teim.app/t/main/buildset/56440baeffbf4c5cb87be435261e83e9\n\n- teim-code-review https://zuul.teim.app/t/main/build/f5ef630fbdc8452ea3f826687c5fa594 : SUCCESS in 8m 59s","accounts_in_message":[],"_revision_number":24},{"id":"7dfcd005de3caccc90fc1d142913f5c13c843838","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-24 08:20:19.000000000","message":"Patch Set 24: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/f0599b4a8a014ffc9a408435d3f8e6ef\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7fe07a5fd8854b2d9827335313025015 : SUCCESS in 5m 30s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9a160d5b983f45b2ac05c882f967e64c : SUCCESS in 4m 08s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/51def5d89e6b43e689578078bc6a5b90 : SUCCESS in 8m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/baaf136d99fe4cf686a1d164deb56c7b : SUCCESS in 3m 23s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/1c72a093651f48519d3ead731898c664 : SUCCESS in 3m 15s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/91d280b46d1541d6a126847712ff1a08 : SUCCESS in 3m 09s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/33a33f85d75b41779532ee0a569c4efb : SUCCESS in 4m 49s (non-voting)\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/0dbbf4fe05f64db0b2e196a2f6124d01 : SUCCESS in 3m 55s\n- cyborg-tempest-py3 https://zuul.opendev.org/t/openstack/build/82d952af71db41db84761ab3244702b8 : SUCCESS in 43m 15s\n- cyborg-tempest-ipv6 https://zuul.opendev.org/t/openstack/build/7bc90ece715346a99d21abb08766f39d : FAILURE in 24m 31s\n- cyborg-grenade https://zuul.opendev.org/t/openstack/build/2a4f944a57bb4b0c93b9bf037dc2672f : SUCCESS in 53m 18s\n- cyborg-grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/0af48118ede9462397afd7686367d7b1 : SUCCESS in 28m 13s\n- cyborg-tempest-srbac https://zuul.opendev.org/t/openstack/build/a20b601ae9ed46a9b1f7224a1436e06f : SUCCESS in 31m 45s","accounts_in_message":[],"_revision_number":24},{"id":"dbbb15e36ec70d2b10bc3d1d813c8d0f7401b0df","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34452,"name":"Joan Gilabert","display_name":"jgilaber","email":"jgilaber@redhat.com","username":"jgilaber"},"date":"2026-07-24 08:31:59.000000000","message":"Uploaded patch set 25.\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":25},{"id":"64dc5fe44f8cf5b63b23e2226379c3dd0c115953","tag":"autogenerated:zuul:automatic-ci","author":{"_account_id":28006,"name":"teim-ci","display_name":"teim-ci","email":"ci@seanmooney.info","username":"ci-sean-mooney","status":"this is a third-party ci account run by sean-k-mooney on irc\nhosted at zuul.teim.app"},"date":"2026-07-24 08:32:34.000000000","message":"Patch Set 25:\n\nStarting automatic-ci jobs.","accounts_in_message":[],"_revision_number":25},{"id":"e835f3bb0371ac4db0d3fb671c84741c5fa743d4","tag":"autogenerated:zuul:automatic-ci","author":{"_account_id":28006,"name":"teim-ci","display_name":"teim-ci","email":"ci@seanmooney.info","username":"ci-sean-mooney","status":"this is a third-party ci account run by sean-k-mooney on irc\nhosted at zuul.teim.app"},"date":"2026-07-24 08:43:26.000000000","message":"Patch Set 25:\n\nBuild succeeded (automatic-ci pipeline).\nhttps://zuul.teim.app/t/main/buildset/6be8a5eef71f4b779a4b1fb8680af353\n\n- teim-code-review https://zuul.teim.app/t/main/build/c4d80e789cae47cd963cfb65733d2b32 : SUCCESS in 10m 43s","accounts_in_message":[],"_revision_number":25},{"id":"e6f833d29470b688738a6c9e8a4250bd043b3008","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-24 09:32:10.000000000","message":"Patch Set 25: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/38be03f2e4da4479b5da77327511c246\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b5a4f79d0d744b60b24fef2d6d95315a : SUCCESS in 5m 08s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ee8781b6bc114369ae9d82ebd0760f23 : SUCCESS in 5m 43s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bbc303d5cd9a488ea0729e04eed31f2d : SUCCESS in 9m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/69bbabc0bc9c452d961d6e623f5002cc : SUCCESS in 5m 05s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/155618a65d214252a72f263437c0cb54 : SUCCESS in 4m 52s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/bd951f53a009441cb8db9aba6849635c : SUCCESS in 5m 53s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/967134d01cd244a7a63041658f9d2015 : SUCCESS in 6m 35s (non-voting)\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/ad1f62c5cde741aebb9bcc94cd0d601a : SUCCESS in 4m 27s\n- cyborg-tempest-py3 https://zuul.opendev.org/t/openstack/build/0592e10e674141dd9e4f9de9dad863b8 : SUCCESS in 53m 47s\n- cyborg-tempest-ipv6 https://zuul.opendev.org/t/openstack/build/831695a7d3004a11824e9afe2a6f602c : SUCCESS in 46m 57s\n- cyborg-grenade https://zuul.opendev.org/t/openstack/build/e018db7bee574b1da2e6dc205c7b40f0 : SUCCESS in 48m 22s\n- cyborg-grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/ede4b5b9b8c64a89a54d941600c042bd : SUCCESS in 53m 40s\n- cyborg-tempest-srbac https://zuul.opendev.org/t/openstack/build/84636de407e943f3a680388b038df474 : SUCCESS in 53m 18s","accounts_in_message":[],"_revision_number":25},{"id":"4e791ceacca8a5a75ae8ff8457e31b01b5ed9d18","author":{"_account_id":34452,"name":"Joan Gilabert","display_name":"jgilaber","email":"jgilaber@redhat.com","username":"jgilaber"},"date":"2026-07-24 09:33:35.000000000","message":"Patch Set 25:\n\n(1 comment)","accounts_in_message":[],"_revision_number":25},{"id":"2ababcf001be2b252806e6d4af07cdde0dc96045","tag":"autogenerated:zuul:automatic-ci","author":{"_account_id":28006,"name":"teim-ci","display_name":"teim-ci","email":"ci@seanmooney.info","username":"ci-sean-mooney","status":"this is a third-party ci account run by sean-k-mooney on irc\nhosted at zuul.teim.app"},"date":"2026-07-24 09:34:09.000000000","message":"Patch Set 25:\n\nStarting automatic-ci jobs.","accounts_in_message":[],"_revision_number":25},{"id":"57e787fd4f7a43824a75c917fd39b51b2a10c329","tag":"autogenerated:zuul:automatic-ci","author":{"_account_id":28006,"name":"teim-ci","display_name":"teim-ci","email":"ci@seanmooney.info","username":"ci-sean-mooney","status":"this is a third-party ci account run by sean-k-mooney on irc\nhosted at zuul.teim.app"},"date":"2026-07-24 09:45:02.000000000","message":"Patch Set 25:\n\nBuild succeeded (automatic-ci pipeline).\nhttps://zuul.teim.app/t/main/buildset/08499c1f89ff42b2b0d057693debe300\n\n- teim-code-review https://zuul.teim.app/t/main/build/e0edf9b648dd4f7abe52ada018a0e5c4 : SUCCESS in 10m 42s","accounts_in_message":[],"_revision_number":25},{"id":"3bdb0ae5aca480565b88373350f8d7e766c99f2e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-24 10:50:03.000000000","message":"Patch Set 25:\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7d46b5d552e749be8642e6936c38520b\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/317fb678790d4b498c1b94d906725d48 : SUCCESS in 5m 25s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/6664939ba4d840b3aa31623aa67a67fe : SUCCESS in 4m 39s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/03bea8c52b9a4dce90bfefb69756f41c : SUCCESS in 8m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7b03a42728774fb59e90332363fc2257 : SUCCESS in 4m 48s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/1964c6b202e641abb557b14987ffd48d : SUCCESS in 4m 18s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/a1dbecac32b84bdf85713433f33d4ef5 : SUCCESS in 4m 38s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/9bd9dfe504624e33979308df79c25a84 : SUCCESS in 8m 15s (non-voting)\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/f323853a52d34fd6953e9fcb6d283df4 : SUCCESS in 4m 26s\n- cyborg-tempest-py3 https://zuul.opendev.org/t/openstack/build/d9cf72bad39a40168f7aeb025d8cfc57 : SUCCESS in 27m 28s\n- cyborg-tempest-ipv6 https://zuul.opendev.org/t/openstack/build/3884a4e51383471bb94d3267fa455d4f : SUCCESS in 22m 04s\n- cyborg-grenade https://zuul.opendev.org/t/openstack/build/3aa0832ff3f1412398bfea2a6703edbf : SUCCESS in 1h 09m 15s\n- cyborg-grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/5d341d6811d84e83b14239ddc756e8da : SUCCESS in 1h 00m 36s\n- cyborg-tempest-srbac https://zuul.opendev.org/t/openstack/build/238c52b16d4c4fbe8fbb4e2090b5235e : SUCCESS in 58m 03s","accounts_in_message":[],"_revision_number":25},{"id":"2169a97d51b4ae3699c75949cb87e5e320034a62","author":{"_account_id":34452,"name":"Joan Gilabert","display_name":"jgilaber","email":"jgilaber@redhat.com","username":"jgilaber"},"date":"2026-07-24 11:04:36.000000000","message":"Patch Set 25:\n\n(1 comment)","accounts_in_message":[],"_revision_number":25},{"id":"9182c6c3d1d5de053ba3f5e2ddfe3f6982d4c1c3","tag":"autogenerated:zuul:automatic-ci","author":{"_account_id":28006,"name":"teim-ci","display_name":"teim-ci","email":"ci@seanmooney.info","username":"ci-sean-mooney","status":"this is a third-party ci account run by sean-k-mooney on irc\nhosted at zuul.teim.app"},"date":"2026-07-24 11:05:10.000000000","message":"Patch Set 25:\n\nStarting automatic-ci jobs.","accounts_in_message":[],"_revision_number":25},{"id":"059c8bedc6095b4656e3bec02f7c9abc7ed71e87","tag":"autogenerated:zuul:automatic-ci","author":{"_account_id":28006,"name":"teim-ci","display_name":"teim-ci","email":"ci@seanmooney.info","username":"ci-sean-mooney","status":"this is a third-party ci account run by sean-k-mooney on irc\nhosted at zuul.teim.app"},"date":"2026-07-24 11:27:48.000000000","message":"Patch Set 25:\n\nBuild succeeded (automatic-ci pipeline).\nhttps://zuul.teim.app/t/main/buildset/1dcdd5edc6754e0d9b3acf9642bd3deb\n\n- teim-code-review https://zuul.teim.app/t/main/build/b837bf42f5f74e3fb2f8ad75de57c5d7 : SUCCESS in 9m 24s","accounts_in_message":[],"_revision_number":25},{"id":"fa3e140c784ffb55aaec458d8edff3b1c7172541","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-24 12:24:49.000000000","message":"Patch Set 25:\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a9297b31659b44a4a9e1eb9db607b712\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/76d2011919d14b05be00a7640e4fccbc : SUCCESS in 3m 37s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8febe16396eb47638cf2d3b55359a397 : SUCCESS in 5m 34s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/70e1c68370cd4b8b97a6c66c6f0fb93d : SUCCESS in 8m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2444324c8269439aadc9edcb4d688031 : SUCCESS in 2m 46s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/c6c9eeab08764eafabd5454d4084db5b : SUCCESS in 6m 10s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/b87a0293783a48369a1fb9961aa99c29 : SUCCESS in 3m 39s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/6f9d6a54a99f4ced8134d3968007d123 : SUCCESS in 4m 17s (non-voting)\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/9ad7b98213aa4b4c82731e3f84ee152a : SUCCESS in 4m 39s\n- cyborg-tempest-py3 https://zuul.opendev.org/t/openstack/build/724d19366a2e45b999b4dc06b1d32b1e : SUCCESS in 43m 39s\n- cyborg-tempest-ipv6 https://zuul.opendev.org/t/openstack/build/65ad2fb55d2a41c9af258f50c56db0cb : SUCCESS in 37m 24s\n- cyborg-grenade https://zuul.opendev.org/t/openstack/build/5c71758220f548149c252b0f51397930 : SUCCESS in 53m 25s\n- cyborg-grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/312000b4623a4e9d90929f044ab65d23 : SUCCESS in 28m 08s\n- cyborg-tempest-srbac https://zuul.opendev.org/t/openstack/build/9b10ad85c07e4d21ade81c1938442f6b : SUCCESS in 58m 18s","accounts_in_message":[],"_revision_number":25},{"id":"16acf3822d72416f4e27df3bd1489d477d1c6245","author":{"_account_id":34452,"name":"Joan Gilabert","display_name":"jgilaber","email":"jgilaber@redhat.com","username":"jgilaber"},"date":"2026-07-28 10:34:01.000000000","message":"Patch Set 25:\n\n(1 comment)","accounts_in_message":[],"_revision_number":25},{"id":"2fe6488ef682614c1c41caabbe94851a303cbfa9","tag":"autogenerated:zuul:automatic-ci","author":{"_account_id":28006,"name":"teim-ci","display_name":"teim-ci","email":"ci@seanmooney.info","username":"ci-sean-mooney","status":"this is a third-party ci account run by sean-k-mooney on irc\nhosted at zuul.teim.app"},"date":"2026-07-28 10:34:33.000000000","message":"Patch Set 25:\n\nStarting automatic-ci jobs.","accounts_in_message":[],"_revision_number":25},{"id":"95891a8ceb9bb4408e0c425f9c2d71f6d43693a6","tag":"autogenerated:zuul:automatic-ci","author":{"_account_id":28006,"name":"teim-ci","display_name":"teim-ci","email":"ci@seanmooney.info","username":"ci-sean-mooney","status":"this is a third-party ci account run by sean-k-mooney on irc\nhosted at zuul.teim.app"},"date":"2026-07-28 10:57:40.000000000","message":"Patch Set 25:\n\nBuild succeeded (automatic-ci pipeline).\nhttps://zuul.teim.app/t/main/buildset/34c273f5133442068d5d182a4d91e811\n\n- teim-code-review https://zuul.teim.app/t/main/build/e9d41c39100d4432852015faea1d0996 : SUCCESS in 9m 03s","accounts_in_message":[],"_revision_number":25},{"id":"dff2d6dd1e7df8632d499bac3bef1ef9f3d1e6d3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-28 11:16:36.000000000","message":"Patch Set 25:\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8473c6b8999c4ecb8c3af2da3b4adfde\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/5fb1ba3ec5524f70a701d95b490f8c54 : SUCCESS in 3m 39s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/904b233c0cb241cdb6ef9e8743367b6b : SUCCESS in 5m 41s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/aee5d67a6dc34ca1bd508c3389b0e233 : SUCCESS in 7m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a68baaf714c043aa96356014491dfad5 : SUCCESS in 4m 25s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/e48ce9d4dd4c45fdb5868e2c083d4e25 : SUCCESS in 3m 59s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/157a63f8eb264a889794d1a09341efaa : SUCCESS in 4m 25s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/7c2ec0c1065746219da009b8073d9ffd : SUCCESS in 6m 12s (non-voting)\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/6ea0f33829d9400e9b05f59ba40977d0 : SUCCESS in 4m 35s\n- cyborg-tempest-py3 https://zuul.opendev.org/t/openstack/build/eed26d34a3cf486d878c234b5af076c6 : SUCCESS in 39m 33s\n- cyborg-tempest-ipv6 https://zuul.opendev.org/t/openstack/build/6f8aa2e32d4c478b86bc4e5f475a2738 : SUCCESS in 24m 25s\n- cyborg-grenade https://zuul.opendev.org/t/openstack/build/fbaec6dd0e7a4f70a448084f3edd18b6 : SUCCESS in 25m 22s\n- cyborg-grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/3f791849bbc74c908ecb68832e5a0fdf : SUCCESS in 28m 37s\n- cyborg-tempest-srbac https://zuul.opendev.org/t/openstack/build/73e8571a80c7498c996b5d4d620badaf : SUCCESS in 28m 05s","accounts_in_message":[],"_revision_number":25},{"id":"7c14d37aa0301297b41c1c1dbab120aa8d715df1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34452,"name":"Joan Gilabert","display_name":"jgilaber","email":"jgilaber@redhat.com","username":"jgilaber"},"date":"2026-07-29 10:36:45.000000000","message":"Uploaded patch set 26: Patch Set 25 was rebased.\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":26},{"id":"12b94282ebffcd49e3969ed1ff6a7dc2c1e59660","tag":"autogenerated:zuul:automatic-ci","author":{"_account_id":28006,"name":"teim-ci","display_name":"teim-ci","email":"ci@seanmooney.info","username":"ci-sean-mooney","status":"this is a third-party ci account run by sean-k-mooney on irc\nhosted at zuul.teim.app"},"date":"2026-07-29 10:38:29.000000000","message":"Patch Set 26:\n\nStarting automatic-ci jobs.","accounts_in_message":[],"_revision_number":26},{"id":"f53f93f11eee2ebc307962b64b0fdae6c5ae94f9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-29 11:42:39.000000000","message":"Patch Set 26: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/440faafabad140eb887a215d383b1f66\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f687bb7e0b2e4a95adbfd5c67106a295 : SUCCESS in 7m 00s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8f1dbe4175fd41f2ad34227364d925ad : SUCCESS in 6m 05s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0617ab9a2ab64f7295fcc34aafa4ec18 : SUCCESS in 8m 56s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/25a0b0dcdb55416ebbd7c00b56b4c63a : SUCCESS in 5m 27s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/e40f7262a7b14853bc46939115f023e1 : SUCCESS in 4m 23s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/8271f3b069e14efbb6e95c686eab9e5d : SUCCESS in 6m 15s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/f92a18d493b3416b9b6075f6e45facde : SUCCESS in 8m 50s (non-voting)\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/33a4c3aed4874bc48b241f7ee7fb47bc : SUCCESS in 6m 47s\n- cyborg-tempest-py3 https://zuul.opendev.org/t/openstack/build/e1f645dd424249938fe9b277bc0c3c05 : SUCCESS in 55m 56s\n- cyborg-tempest-ipv6 https://zuul.opendev.org/t/openstack/build/8cf5a0d394c2430cb8c55544f402887c : SUCCESS in 37m 47s\n- cyborg-grenade https://zuul.opendev.org/t/openstack/build/5a9b4e2db07c494c8c4272d2f31dc5e8 : SUCCESS in 57m 48s\n- cyborg-grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/d2e150d4f21a4e7ba7556085e5dbd75f : SUCCESS in 57m 44s\n- cyborg-tempest-srbac https://zuul.opendev.org/t/openstack/build/784bdab7b4d94a37a467a6a0cfbcc965 : SUCCESS in 57m 37s","accounts_in_message":[],"_revision_number":26},{"id":"501e25d251546c5df88256e01efe67396aeced63","tag":"autogenerated:zuul:automatic-ci","author":{"_account_id":28006,"name":"teim-ci","display_name":"teim-ci","email":"ci@seanmooney.info","username":"ci-sean-mooney","status":"this is a third-party ci account run by sean-k-mooney on irc\nhosted at zuul.teim.app"},"date":"2026-07-29 12:30:43.000000000","message":"Patch Set 26:\n\n(1 comment)\n\nBuild succeeded (automatic-ci pipeline).\nhttps://zuul.teim.app/t/main/buildset/fd0f5a5463f4499c821e70bd4bfdbb8d\n\n- teim-code-review https://zuul.teim.app/t/main/build/773e9fc39f15468cbb1cd5c19cf6c33e : SUCCESS in 10m 32s","accounts_in_message":[],"_revision_number":26},{"id":"342cc79915bf315b4e178e9d348057a8ce26ec17","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34452,"name":"Joan Gilabert","display_name":"jgilaber","email":"jgilaber@redhat.com","username":"jgilaber"},"date":"2026-07-29 15:19:08.000000000","message":"Uploaded patch set 27: Patch Set 26 was rebased.\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":27},{"id":"a54f82b5c171845289da73b42e4fa8cfeb379c4a","tag":"autogenerated:zuul:automatic-ci","author":{"_account_id":28006,"name":"teim-ci","display_name":"teim-ci","email":"ci@seanmooney.info","username":"ci-sean-mooney","status":"this is a third-party ci account run by sean-k-mooney on irc\nhosted at zuul.teim.app"},"date":"2026-07-29 15:20:53.000000000","message":"Patch Set 27:\n\nStarting automatic-ci jobs.","accounts_in_message":[],"_revision_number":27},{"id":"bba2402f58eab03ed718b8777228354f6ff53b13","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-29 16:37:26.000000000","message":"Patch Set 27: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/bad92e6db1d84bc3aa47ead1581f3b9a\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/507abb45f8da4f3aa29a074e297392d3 : SUCCESS in 6m 07s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c4192267adf242a5bdf7eb6410606197 : SUCCESS in 4m 43s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/148adc2023b040f99974d33fdcaf32f0 : SUCCESS in 9m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/23553af3e31b42f198b7d90a431b188d : SUCCESS in 4m 42s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/0d7269cbceff4343a405508fb8374fc2 : SUCCESS in 5m 27s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/83968d57b42b4edba78f66cfab3fc0a3 : SUCCESS in 4m 50s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/1fc97eb7c8f0421cae85fac4bc0b1253 : SUCCESS in 7m 40s (non-voting)\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/7dfde6e386104d58b47d857844fe801e : SUCCESS in 5m 19s\n- cyborg-tempest-py3 https://zuul.opendev.org/t/openstack/build/c7798dff79e34c81ad609fd92f5ef884 : FAILURE in 24m 29s\n- cyborg-tempest-ipv6 https://zuul.opendev.org/t/openstack/build/ae557003eda34dd6a7bad803f255b7b9 : FAILURE in 15m 17s\n- cyborg-grenade https://zuul.opendev.org/t/openstack/build/b5e2943ea57f45e79f6042834df2ccb2 : SUCCESS in 42m 54s\n- cyborg-grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/79c94af0003d40c39e57be64d9e8fea8 : SUCCESS in 1h 08m 45s\n- cyborg-tempest-srbac https://zuul.opendev.org/t/openstack/build/e5af6bf246124a1f93b6eed04afccb9d : FAILURE in 25m 37s","accounts_in_message":[],"_revision_number":27},{"id":"fbef6896631b64946ce17bba0b3132f3f157c6a8","tag":"autogenerated:zuul:automatic-ci","author":{"_account_id":28006,"name":"teim-ci","display_name":"teim-ci","email":"ci@seanmooney.info","username":"ci-sean-mooney","status":"this is a third-party ci account run by sean-k-mooney on irc\nhosted at zuul.teim.app"},"date":"2026-07-29 17:34:43.000000000","message":"Patch Set 27:\n\n(1 comment)\n\nBuild succeeded (automatic-ci pipeline).\nhttps://zuul.teim.app/t/main/buildset/54aaacd05c2c4b30a9a25c5340b8a7ab\n\n- teim-code-review https://zuul.teim.app/t/main/build/476a8234ca884735ba87b019a6ec8f89 : SUCCESS in 10m 23s","accounts_in_message":[],"_revision_number":27},{"id":"ca58484522e6924a9b3b223b176f29e41d0657ff","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34452,"name":"Joan Gilabert","display_name":"jgilaber","email":"jgilaber@redhat.com","username":"jgilaber"},"date":"2026-07-29 17:35:53.000000000","message":"Uploaded patch set 28: Patch Set 27 was rebased.\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":28},{"id":"68dacf1a4f9db10deabcdab5b0b47d987f31e708","tag":"autogenerated:zuul:automatic-ci","author":{"_account_id":28006,"name":"teim-ci","display_name":"teim-ci","email":"ci@seanmooney.info","username":"ci-sean-mooney","status":"this is a third-party ci account run by sean-k-mooney on irc\nhosted at zuul.teim.app"},"date":"2026-07-29 17:36:41.000000000","message":"Patch Set 28:\n\nStarting automatic-ci jobs.","accounts_in_message":[],"_revision_number":28},{"id":"98710ae0757eb1649177028e95c880b542cd3408","tag":"autogenerated:zuul:automatic-ci","author":{"_account_id":28006,"name":"teim-ci","display_name":"teim-ci","email":"ci@seanmooney.info","username":"ci-sean-mooney","status":"this is a third-party ci account run by sean-k-mooney on irc\nhosted at zuul.teim.app"},"date":"2026-07-29 18:29:45.000000000","message":"Patch Set 28:\n\n(1 comment)\n\nBuild succeeded (automatic-ci pipeline).\nhttps://zuul.teim.app/t/main/buildset/c19caad0fbbf43b88fac172166b0f322\n\n- teim-code-review https://zuul.teim.app/t/main/build/706a605f206d4b76b51398509dcd3d07 : SUCCESS in 9m 30s","accounts_in_message":[],"_revision_number":28},{"id":"4f20038ea5cf69fc0f63b694ac2a7d19a337a43f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-29 18:41:34.000000000","message":"Patch Set 28: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/afa39da137ca438dabbce159613bbf06\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/91722046a6f441e484b4f493f467bff0 : SUCCESS in 5m 06s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7660489283e7415a96562cee94cbb011 : SUCCESS in 4m 51s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c723d9a78181495bb663d6f84180b2d1 : SUCCESS in 6m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/95f82ae5c5624489bb34ed82bcfcd26a : SUCCESS in 4m 34s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/97aff4e975ef43ea8c8d35df2c6cc06e : SUCCESS in 4m 06s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/ba97a865bd694bc7bbfef9fe93169737 : SUCCESS in 4m 10s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/bcd039b6f1284cec8dadc2ab5fa3b034 : SUCCESS in 6m 30s (non-voting)\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/ac0162227577409087f446182ad6fc01 : SUCCESS in 4m 45s\n- cyborg-tempest-py3 https://zuul.opendev.org/t/openstack/build/bbcd67eac98e4540ba4074108809ca1b : SUCCESS in 51m 04s\n- cyborg-tempest-ipv6 https://zuul.opendev.org/t/openstack/build/a539649f4cff4badb0620f6c38f0d605 : SUCCESS in 28m 45s\n- cyborg-grenade https://zuul.opendev.org/t/openstack/build/05f986c21cf346179df64afc58630c1c : SUCCESS in 58m 04s\n- cyborg-grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/a6343117feec41679d11b1e412184acf : SUCCESS in 58m 25s\n- cyborg-tempest-srbac https://zuul.opendev.org/t/openstack/build/1a62727a646c4f0d94fb57cf0f8fb0de : SUCCESS in 55m 10s","accounts_in_message":[],"_revision_number":28},{"id":"ee9dcc8bfe83869cb4b4569a47f3006ec3fcef6c","author":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},"date":"2026-07-30 01:50:50.000000000","message":"Patch Set 28:\n\n(1 comment)","accounts_in_message":[],"_revision_number":28},{"id":"7f60751bc80ba0fc9caee2d798ef7a604f0c6c68","author":{"_account_id":34452,"name":"Joan Gilabert","display_name":"jgilaber","email":"jgilaber@redhat.com","username":"jgilaber"},"date":"2026-07-30 11:05:43.000000000","message":"Patch Set 28:\n\n(1 comment)","accounts_in_message":[],"_revision_number":28},{"id":"df1565114cc647c31c09609b5d1bc414bc589b31","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34452,"name":"Joan Gilabert","display_name":"jgilaber","email":"jgilaber@redhat.com","username":"jgilaber"},"date":"2026-07-30 11:06:20.000000000","message":"Uploaded patch set 29: Patch Set 28 was rebased.\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":29},{"id":"e2deafdd321c59edf6bbd5940c1ebfd7cf9b185a","tag":"autogenerated:zuul:automatic-ci","author":{"_account_id":28006,"name":"teim-ci","display_name":"teim-ci","email":"ci@seanmooney.info","username":"ci-sean-mooney","status":"this is a third-party ci account run by sean-k-mooney on irc\nhosted at zuul.teim.app"},"date":"2026-07-30 11:08:06.000000000","message":"Patch Set 29:\n\nStarting automatic-ci jobs.","accounts_in_message":[],"_revision_number":29},{"id":"8589cc38ea3a684509babafe0e9d8f72fc28d028","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-30 12:27:56.000000000","message":"Patch Set 29: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/0fb9a894b79646a8a8c39e0260002253\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/9d7db19729424a20b4ac6035f883c1dd : FAILURE in 3m 07s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9b468de2c2cd48b0b505ccedccb8bc85 : SUCCESS in 5m 24s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/78308b1295d04d82987b1edb1b1e952d : SUCCESS in 8m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ba9d478592934cbc950ceddbed874611 : SUCCESS in 4m 30s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/b5a6859233994238922493afcfd5cf7f : FAILURE in 4m 30s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/bbfb5696f1a64500a108c772c914763c : FAILURE in 3m 59s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/54a525eb27344a55a56486316f8654fa : FAILURE in 4m 35s (non-voting)\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/6642fae57a894867982c2b7ee9b654db : SUCCESS in 4m 20s\n- cyborg-tempest-py3 https://zuul.opendev.org/t/openstack/build/8b21ba6f1c3340c88b89331e56aadf51 : SUCCESS in 51m 25s\n- cyborg-tempest-ipv6 https://zuul.opendev.org/t/openstack/build/26908ccfc2b14f9da4bc3330b843157b : SUCCESS in 25m 17s\n- cyborg-grenade https://zuul.opendev.org/t/openstack/build/168f2bac33464ad6986042ee45563764 : SUCCESS in 1h 12m 16s\n- cyborg-grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/f56f9cfa76fd45e08fed8f3f8f0652ba : SUCCESS in 31m 24s\n- cyborg-tempest-srbac https://zuul.opendev.org/t/openstack/build/9288eeac70a4453aae2054d9ed810a56 : SUCCESS in 53m 46s","accounts_in_message":[],"_revision_number":29},{"id":"1e2a79788b51fe37d7781d49993e6ab3f32b113b","tag":"autogenerated:zuul:automatic-ci","author":{"_account_id":28006,"name":"teim-ci","display_name":"teim-ci","email":"ci@seanmooney.info","username":"ci-sean-mooney","status":"this is a third-party ci account run by sean-k-mooney on irc\nhosted at zuul.teim.app"},"date":"2026-07-30 12:51:47.000000000","message":"Patch Set 29:\n\nBuild succeeded (automatic-ci pipeline).\nhttps://zuul.teim.app/t/main/buildset/21671e947fc64f39aed1c2f8a85213b7\n\n- teim-code-review https://zuul.teim.app/t/main/build/ab344de4cc0b4b5fba251c1248724429 : SUCCESS in 11m 24s","accounts_in_message":[],"_revision_number":29},{"id":"190a713c09cdce800cef3c8d5484840075857eca","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34452,"name":"Joan Gilabert","display_name":"jgilaber","email":"jgilaber@redhat.com","username":"jgilaber"},"date":"2026-07-31 11:34:40.000000000","message":"Uploaded patch set 30.\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":30},{"id":"a07c123371a163fff6d4c887cb59e4e7a1dff0bf","tag":"autogenerated:zuul:automatic-ci","author":{"_account_id":28006,"name":"teim-ci","display_name":"teim-ci","email":"ci@seanmooney.info","username":"ci-sean-mooney","status":"this is a third-party ci account run by sean-k-mooney on irc\nhosted at zuul.teim.app"},"date":"2026-07-31 11:36:24.000000000","message":"Patch Set 30:\n\nStarting automatic-ci jobs.","accounts_in_message":[],"_revision_number":30},{"id":"9981d1a17d20a34dfaf84dbb131082ad92cda24e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-31 13:00:52.000000000","message":"Patch Set 30: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d25450736f9e4a6a91b5705a03367f02\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/3423bb79e08d40f2812cfbd1457728cb : SUCCESS in 5m 07s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/87c9ef9dd802454c95e7ed6be3f961ed : SUCCESS in 3m 10s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2116f9185a9f452d8ff09be4071e0dd3 : SUCCESS in 7m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/87c17eb4d9014025b3bc027d65c274c7 : SUCCESS in 5m 05s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/c24bdcfd8ca04a4ca11af6715996068f : SUCCESS in 4m 32s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/f4b16a465cdc453fb3ca9bc38e807017 : SUCCESS in 2m 52s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/7c112addd0414e54be91536aab6863b7 : SUCCESS in 6m 09s (non-voting)\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/9ff4f8b5bf8146f88d841a52bfe6841e : SUCCESS in 5m 23s\n- cyborg-tempest-py3 https://zuul.opendev.org/t/openstack/build/6293a77f8fe448c4a6d6c55222c0f195 : SUCCESS in 38m 12s\n- cyborg-tempest-ipv6 https://zuul.opendev.org/t/openstack/build/33fadd3719f444ffaf356aeb0cc27c83 : SUCCESS in 46m 33s\n- cyborg-grenade https://zuul.opendev.org/t/openstack/build/99d2d61350204d33ac1583d6a00672c7 : SUCCESS in 48m 00s\n- cyborg-grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/a21672fa0a3e4852a1486167ada21e11 : SUCCESS in 43m 52s\n- cyborg-tempest-srbac https://zuul.opendev.org/t/openstack/build/84ae7183c2ee437b831d5438af6b7c97 : SUCCESS in 43m 36s","accounts_in_message":[],"_revision_number":30},{"id":"8295f911146900d376c476f62312295df23cab8b","author":{"_account_id":34452,"name":"Joan Gilabert","display_name":"jgilaber","email":"jgilaber@redhat.com","username":"jgilaber"},"date":"2026-07-31 13:09:14.000000000","message":"Patch Set 30:\n\n(1 comment)","accounts_in_message":[],"_revision_number":30},{"id":"65f088e0f34aa65259757233e86e741cd68f1d43","tag":"autogenerated:zuul:automatic-ci","author":{"_account_id":28006,"name":"teim-ci","display_name":"teim-ci","email":"ci@seanmooney.info","username":"ci-sean-mooney","status":"this is a third-party ci account run by sean-k-mooney on irc\nhosted at zuul.teim.app"},"date":"2026-07-31 13:25:46.000000000","message":"Patch Set 30:\n\nBuild succeeded (automatic-ci pipeline).\nhttps://zuul.teim.app/t/main/buildset/105704bb7091495088ff0905a8344921\n\n- teim-code-review https://zuul.teim.app/t/main/build/41060872117e46e786c477a93f5e18fa : SUCCESS in 9m 06s","accounts_in_message":[],"_revision_number":30},{"id":"0fbad824f88c3713262b74b1946be36859e2b875","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34452,"name":"Joan Gilabert","display_name":"jgilaber","email":"jgilaber@redhat.com","username":"jgilaber"},"date":"2026-08-03 08:02:03.000000000","message":"Uploaded patch set 31: Patch Set 30 was rebased.\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":31},{"id":"4f980a686dd648f1d822e2ead0085fa281717c28","tag":"autogenerated:zuul:automatic-ci","author":{"_account_id":28006,"name":"teim-ci","display_name":"teim-ci","email":"ci@seanmooney.info","username":"ci-sean-mooney","status":"this is a third-party ci account run by sean-k-mooney on irc\nhosted at zuul.teim.app"},"date":"2026-08-03 08:03:51.000000000","message":"Patch Set 31:\n\nStarting automatic-ci jobs.","accounts_in_message":[],"_revision_number":31},{"id":"b0d5644c8bb0c81a6c512384ffcfe4eaece459b3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-03 08:56:38.000000000","message":"Patch Set 31: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/594a0e6e2c8245139d667f289abcd0ab\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c6a9b6762be747fea6e3de94c1ab4ac2 : SUCCESS in 5m 52s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/33ca58ab0deb40718aa3c78324e112f6 : SUCCESS in 3m 17s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7b4adb4f595747adb3b90798848f8918 : SUCCESS in 8m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e9dcdbf1b1334015893b306767dff932 : SUCCESS in 5m 21s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/e337f364362041f3b59402a9498aba61 : SUCCESS in 2m 58s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/6692644f230247229b5bd5741897f429 : SUCCESS in 2m 39s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/80aa949149b441fa9587f080a640802e : SUCCESS in 8m 18s (non-voting)\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/b16378a1bbce4dd2a973c41a00fa4588 : SUCCESS in 4m 33s\n- cyborg-tempest-py3 https://zuul.opendev.org/t/openstack/build/bd2c68459a5843ae946b43ce3db64c58 : SUCCESS in 36m 30s\n- cyborg-tempest-ipv6 https://zuul.opendev.org/t/openstack/build/baab0c34fdb04c89811a445ef03045ae : SUCCESS in 46m 18s\n- cyborg-grenade https://zuul.opendev.org/t/openstack/build/4c521a8824934b4eae0b5310377dd672 : SUCCESS in 29m 49s\n- cyborg-grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/bca781b70a354e6f9adc3a12b3b809aa : SUCCESS in 31m 04s\n- cyborg-tempest-srbac https://zuul.opendev.org/t/openstack/build/7d562d5aa87f4dfa817fb3c4a72c6451 : SUCCESS in 27m 05s","accounts_in_message":[],"_revision_number":31},{"id":"0e76d671609570814c90b0717606a4dc3f5c4dc2","tag":"autogenerated:zuul:automatic-ci","author":{"_account_id":28006,"name":"teim-ci","display_name":"teim-ci","email":"ci@seanmooney.info","username":"ci-sean-mooney","status":"this is a third-party ci account run by sean-k-mooney on irc\nhosted at zuul.teim.app"},"date":"2026-08-03 09:49:32.000000000","message":"Patch Set 31:\n\nBuild succeeded (automatic-ci pipeline).\nhttps://zuul.teim.app/t/main/buildset/dbfefda723ae4aba99fdd80a2f6dc4bd\n\n- teim-code-review https://zuul.teim.app/t/main/build/d090ba5f2ff446a5aa79ddbb5ad6ec42 : SUCCESS in 8m 42s","accounts_in_message":[],"_revision_number":31},{"id":"738665e9256d88bec14405a3223a4a00d2ff43a3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34452,"name":"Joan Gilabert","display_name":"jgilaber","email":"jgilaber@redhat.com","username":"jgilaber"},"date":"2026-08-04 16:14:06.000000000","message":"Uploaded patch set 32: New patch set was added with same tree, parent tree, and commit message as Patch Set 31.\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":32},{"id":"2fd5065ba5f2366eb6efa43a48c58122aecee123","tag":"autogenerated:zuul:automatic-ci","author":{"_account_id":28006,"name":"teim-ci","display_name":"teim-ci","email":"ci@seanmooney.info","username":"ci-sean-mooney","status":"this is a third-party ci account run by sean-k-mooney on irc\nhosted at zuul.teim.app"},"date":"2026-08-04 16:15:50.000000000","message":"Patch Set 32:\n\nStarting automatic-ci jobs.","accounts_in_message":[],"_revision_number":32},{"id":"5cecaa8c9c86e0187edebe858b04c0f69c0aa268","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-04 17:28:00.000000000","message":"Patch Set 32: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/96da338047d941d491123d08d246d6ed\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c81acb9f7ff14320a4f468987c2966d2 : SUCCESS in 6m 02s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/083f1d4f3f7642df8e9278f9c40ef1b9 : SUCCESS in 4m 17s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4d796a8d3da546f4a5f964351d906cf8 : SUCCESS in 5m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e83cc06979da4b26b61c4f06644a6525 : SUCCESS in 4m 30s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/c9b1fd6f2d26432084e7798b4c41c350 : SUCCESS in 4m 09s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/596aaac3450b4d6b88a8d60d268d783c : SUCCESS in 4m 14s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/5643264bfb0c4557b3efcc93e92a5e56 : SUCCESS in 7m 01s (non-voting)\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/bd01e228bf564f9bb9dba6d94410e7ad : SUCCESS in 3m 11s\n- cyborg-tempest-py3 https://zuul.opendev.org/t/openstack/build/c09e0f94992848d190e888739bdfb301 : SUCCESS in 34m 26s\n- cyborg-tempest-ipv6 https://zuul.opendev.org/t/openstack/build/91cf51ec14344cc092b86b4b2997b6bb : SUCCESS in 35m 47s\n- cyborg-grenade https://zuul.opendev.org/t/openstack/build/9c023dcda48949dbb0f0cd0ac11d683d : SUCCESS in 51m 39s\n- cyborg-grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/c557250fb5c44c6fa172258ca67374e1 : SUCCESS in 42m 53s\n- cyborg-tempest-srbac https://zuul.opendev.org/t/openstack/build/100db326c4714764b272c44c440f4db4 : SUCCESS in 1h 03m 11s","accounts_in_message":[],"_revision_number":32},{"id":"dbdd1a2c19d6968e6c0ba852d84fa7cf37ae72d0","tag":"autogenerated:zuul:automatic-ci","author":{"_account_id":28006,"name":"teim-ci","display_name":"teim-ci","email":"ci@seanmooney.info","username":"ci-sean-mooney","status":"this is a third-party ci account run by sean-k-mooney on irc\nhosted at zuul.teim.app"},"date":"2026-08-04 18:11:57.000000000","message":"Patch Set 32:\n\nBuild succeeded (automatic-ci pipeline).\nhttps://zuul.teim.app/t/main/buildset/8bbd9802122e4d1ebf36e855ee862d37\n\n- teim-code-review https://zuul.teim.app/t/main/build/f7238f7afe2742a59751fd6b4f7e6cc5 : SUCCESS in 8m 42s","accounts_in_message":[],"_revision_number":32},{"id":"714c8fcdd8660afe10fc7daf14f2a1e0a19dcd60","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34452,"name":"Joan Gilabert","display_name":"jgilaber","email":"jgilaber@redhat.com","username":"jgilaber"},"date":"2026-08-31 12:33:18.000000000","message":"Uploaded patch set 33: Patch Set 32 was rebased.\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":33},{"id":"4a1af218bd60b6fe987f71703729b9cfc3661b67","tag":"autogenerated:zuul:automatic-ci","author":{"_account_id":28006,"name":"teim-ci","display_name":"teim-ci","email":"ci@seanmooney.info","username":"ci-sean-mooney","status":"this is a third-party ci account run by sean-k-mooney on irc\nhosted at zuul.teim.app"},"date":"2026-08-31 12:34:22.000000000","message":"Patch Set 33:\n\nStarting automatic-ci jobs.","accounts_in_message":[],"_revision_number":33},{"id":"bcbe3ba68cd3ee251a7168c0727ec64f703502b4","tag":"autogenerated:zuul:automatic-ci","author":{"_account_id":28006,"name":"teim-ci","display_name":"teim-ci","email":"ci@seanmooney.info","username":"ci-sean-mooney","status":"this is a third-party ci account run by sean-k-mooney on irc\nhosted at zuul.teim.app"},"date":"2026-08-31 14:11:00.000000000","message":"Patch Set 33:\n\n(1 comment)\n\nBuild succeeded (automatic-ci pipeline).\nhttps://zuul.teim.app/t/main/buildset/110d255a20f04427bb2a300a787aa126\n\n- teim-code-review https://zuul.teim.app/t/main/build/c847e46abcdf4d2eb48a3dc2b53d9d61 : SUCCESS in 9m 03s\n\nWarning:\n  The commit message opens with \u0027Configure tempest in the devstack plugin to add the cyborg_mdev section populated with the appropriate values to run the mdev tests.\u0027 However, the cyborg_mdev iniset block in cyborg_configure_tempest was added by the parent commit 35b6480 (\u0027Add mdev support to pci-sim DevStack plugin\u0027). This commit\u0027s devstack change is solely deletion of a blank line before the function\u0027s closing brace; its substantive content is the .zuul.yaml additions (PCI_SIM_NUM_PFS\u003d3 and explanatory comments) that size the fake topology for the mdev tests.\n\n**Severity**: WARNING | **Confidence**: 0.8\n\n**Impact**: If merged as-is onto master, the permanent history misattributes the tempest cyborg_mdev configuration: a reader landing on this commit will not find the plugin tempest changes it describes, and git log/blame of devstack/lib/cyborg will contradict the message. The actual content (CI topology sizing) is undocumented.\n\n**Suggestion**:\nReword the message to match this commit, e.g. \u0027Set PCI_SIM_NUM_PFS\u003d3 in the tempest Zuul jobs so the round-robin personalities yield two SR-IOV PFs (Nova and Cyborg) plus one mdev PF for the mdev lifecycle tests.\u0027 If the series is squashed into one change before merge the description may become accurate, but the current standalone commit message does not describe this commit.","accounts_in_message":[],"_revision_number":33},{"id":"ae7544ea70a9d3097b004c33fd49f81aa8254846","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-31 15:04:48.000000000","message":"Patch Set 33: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/ac5ccf766d434736b241e1468bfa19ab\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b28e94004e954a0daaeca32ae6522ae3 : SUCCESS in 4m 12s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/604c3b6b6e22474db226321af4897c75 : SUCCESS in 5m 23s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/87b811d238634a5b94f3e372527333a7 : SUCCESS in 5m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cace0cc62b29464fb03f46c0d958e03d : SUCCESS in 3m 12s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/b138eff14a784ed0ae3f81d3d485970c : SUCCESS in 3m 14s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/769bb29411904d938917ab861dfc2f47 : SUCCESS in 3m 20s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/e7d72201255e43068ee8560076cbcfd0 : SUCCESS in 4m 29s (non-voting)\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/b3f1e4e92f144238b746393a68fc28e6 : SUCCESS in 3m 03s\n- cyborg-tempest-py3 https://zuul.opendev.org/t/openstack/build/e5d56f16372f449383fc57fda78a75b1 : SUCCESS in 58m 06s\n- cyborg-tempest-ipv6 https://zuul.opendev.org/t/openstack/build/f050122206ab490493ebac9df2845a64 : SUCCESS in 25m 49s\n- cyborg-grenade https://zuul.opendev.org/t/openstack/build/b26b02e48d4b46bdba801a8b1b7da151 : SUCCESS in 45m 44s\n- cyborg-grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/5cae2c7790dc441f950e83523e405c72 : FAILURE in 27m 12s\n- cyborg-tempest-srbac https://zuul.opendev.org/t/openstack/build/29ddae6465d84f50a52b92a40a607968 : SUCCESS in 56m 31s","accounts_in_message":[],"_revision_number":33},{"id":"ba395697a47e9f61ba7695fcd68c5583c697fbea","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34452,"name":"Joan Gilabert","display_name":"jgilaber","email":"jgilaber@redhat.com","username":"jgilaber"},"date":"2026-08-31 15:06:57.000000000","message":"Uploaded patch set 34: Patch Set 33 was rebased.\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":34},{"id":"045371962fc6cea3dcb6420153f3280e4369fc20","tag":"autogenerated:zuul:automatic-ci","author":{"_account_id":28006,"name":"teim-ci","display_name":"teim-ci","email":"ci@seanmooney.info","username":"ci-sean-mooney","status":"this is a third-party ci account run by sean-k-mooney on irc\nhosted at zuul.teim.app"},"date":"2026-08-31 15:07:48.000000000","message":"Patch Set 34:\n\nStarting automatic-ci jobs.","accounts_in_message":[],"_revision_number":34},{"id":"2654a679252d39acd46483aaf85b1141e125dfe4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-31 16:16:35.000000000","message":"Patch Set 34: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/77c24e3343b0411c858e59a19f1aa012\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e8cc25c64efc457cab3a720794028496 : SUCCESS in 3m 34s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/cf96b5771b3f414faf250d9f6c6b4068 : SUCCESS in 3m 35s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ea0c6c351df34bf78bd9155d81e19e3f : SUCCESS in 9m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1216dc1c8a2443d9adc9112d94835bda : SUCCESS in 3m 20s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/3b5a1242079b411080d74afa59f7d8d4 : SUCCESS in 4m 40s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/b11ce1778b34460c9772aaabd248c028 : SUCCESS in 4m 30s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/b95e21415aba4746a939927429a6d319 : SUCCESS in 4m 10s (non-voting)\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/cc1cb5dcc4694461893a7fb80b942c34 : SUCCESS in 3m 18s\n- cyborg-tempest-py3 https://zuul.opendev.org/t/openstack/build/ab5520d49a2c43658ab66c0182735d30 : SUCCESS in 55m 25s\n- cyborg-tempest-ipv6 https://zuul.opendev.org/t/openstack/build/7c8c6727526945b2b1b41246218b4713 : SUCCESS in 30m 37s\n- cyborg-grenade https://zuul.opendev.org/t/openstack/build/5300d7de20d74df294bd4eee977da739 : SUCCESS in 1h 01m 55s\n- cyborg-grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/022b5411e6ae4cff8d692314a3e05ad1 : SUCCESS in 53m 23s\n- cyborg-tempest-srbac https://zuul.opendev.org/t/openstack/build/98d8fa4eb306444b9f59e03a2676cfcb : SUCCESS in 57m 29s","accounts_in_message":[],"_revision_number":34},{"id":"c2d06474056f20bf991e7f14546fa8647e25835b","tag":"autogenerated:zuul:automatic-ci","author":{"_account_id":28006,"name":"teim-ci","display_name":"teim-ci","email":"ci@seanmooney.info","username":"ci-sean-mooney","status":"this is a third-party ci account run by sean-k-mooney on irc\nhosted at zuul.teim.app"},"date":"2026-08-31 16:39:39.000000000","message":"Patch Set 34:\n\nBuild succeeded (automatic-ci pipeline).\nhttps://zuul.teim.app/t/main/buildset/fe3957aeb63945f0ad905e60f570e9b5\n\n- teim-code-review https://zuul.teim.app/t/main/build/ad1fa79d7ad64f7590c2389b38ae5016 : SUCCESS in 9m 43s","accounts_in_message":[],"_revision_number":34},{"id":"8775a73b007908c0faf9662a79254d3712468e79","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34452,"name":"Joan Gilabert","display_name":"jgilaber","email":"jgilaber@redhat.com","username":"jgilaber"},"date":"2026-09-01 09:52:29.000000000","message":"Uploaded patch set 35: Patch Set 34 was rebased.\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":35},{"id":"5ac9d3f7587fa3af5cbe052841276fa5b74b4191","tag":"autogenerated:zuul:automatic-ci","author":{"_account_id":28006,"name":"teim-ci","display_name":"teim-ci","email":"ci@seanmooney.info","username":"ci-sean-mooney","status":"this is a third-party ci account run by sean-k-mooney on irc\nhosted at zuul.teim.app"},"date":"2026-09-01 09:53:30.000000000","message":"Patch Set 35:\n\nStarting automatic-ci jobs.","accounts_in_message":[],"_revision_number":35},{"id":"1de2827129aa8f7a03e058006efb35816c1d20db","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-01 11:09:50.000000000","message":"Patch Set 35: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/406ac158d71d49d8a0c8220227ce19f5\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ff404dde60f741f788315f08fbde3ec9 : SUCCESS in 5m 46s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f1f6165ded5b45f2aecb391b1bee21ba : SUCCESS in 5m 35s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8dd9a842293b4d309cbf4f980114833c : SUCCESS in 9m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/231ac4e8a25645839b5fde4521a6a0cf : SUCCESS in 4m 30s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/dd073276bf73429d9f7c18edc28624ca : SUCCESS in 4m 42s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/3d73974684734363a4dc65b21a8e0bca : SUCCESS in 4m 38s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/09fbe64d4211444aa385904573639f2c : SUCCESS in 7m 18s (non-voting)\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/030e06518be646a1956054ff8dafc5ef : SUCCESS in 4m 14s\n- cyborg-tempest-py3 https://zuul.opendev.org/t/openstack/build/4f03dd13530345d998b5101e583e0932 : SUCCESS in 57m 45s\n- cyborg-tempest-ipv6 https://zuul.opendev.org/t/openstack/build/d926754d44e44240937a4c3998956e71 : SUCCESS in 52m 48s\n- cyborg-grenade https://zuul.opendev.org/t/openstack/build/1fdfc49f60c2460d830967a042b6e31e : SUCCESS in 1h 06m 39s\n- cyborg-grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/99a01f8a63d2474380a6b7baab0e9adc : SUCCESS in 26m 36s\n- cyborg-tempest-srbac https://zuul.opendev.org/t/openstack/build/2ff0b6b9a9e5420cbf106e50997ac7b3 : SUCCESS in 57m 14s","accounts_in_message":[],"_revision_number":35},{"id":"3d2ac4bbb7cb014e41e8dee560710e6c84d49584","tag":"autogenerated:zuul:automatic-ci","author":{"_account_id":28006,"name":"teim-ci","display_name":"teim-ci","email":"ci@seanmooney.info","username":"ci-sean-mooney","status":"this is a third-party ci account run by sean-k-mooney on irc\nhosted at zuul.teim.app"},"date":"2026-09-01 11:24:19.000000000","message":"Patch Set 35:\n\n(1 comment)\n\nBuild succeeded (automatic-ci pipeline).\nhttps://zuul.teim.app/t/main/buildset/783d0749283840d7869ec7a8e908067a\n\n- teim-code-review https://zuul.teim.app/t/main/build/b262531352744674beb92348fd7a12d5 : SUCCESS in 8m 19s\n\nWarning:\n  The commit message states this change configures tempest in the devstack plugin to add the cyborg_mdev section with appropriate values. However, the iniset block writing the [cyborg_mdev] section (device_profile_name, mdev_type, vendor_id, product_id) was already added to cyborg_configure_tempest in parent commit 370fd24 (\u0027Add mdev support to pci-sim DevStack plugin\u0027), which is a separate Gerrit change. The actual content of this commit is raising PCI_SIM_NUM_PFS from the default 2 to 3 in the cyborg tempest jobs (base vars and multinode subnode group-vars) so the mixed SR-IOV/mdev round-robin personality assignment yields two uart PFs (one for Nova, one for Cyborg) plus one mdev PF, plus an unrelated blank-line removal in devstack/lib/cyborg. Neither the NUM_PFS change nor its rationale is described anywhere in the message.\n\n**Severity**: WARNING | **Confidence**: 0.9\n\n**Impact**: The permanent change history claims functionality delivered by a different change in the series and does not explain why the CI PF count changed from 2 to 3. A reviewer or future archaeologist reading this message against this diff will not understand what the change actually does or why the mixed SR-IOV/mdev topology needs an extra PF, and may wrongly conclude the tempest config lives here.\n\n**Suggestion**:\nRewrite the message to describe what this commit actually does, e.g. \u0027Bump PCI_SIM_NUM_PFS to 3 in tempest CI jobs so the mixed SR-IOV/mdev topology keeps two uart PFs (one for Nova, one for Cyborg) alongside the mdev PF; with the default 2 PFs, the round-robin personality assignment leaves Cyborg with no SR-IOV VFs.\u0027","accounts_in_message":[],"_revision_number":35}],"current_revision_number":35,"current_revision":"9a648b211622e9685a8ec3a31a4dba2aa98a51a5","revisions":{"46f9392954e2c2beb5e0da40098d865daf5b9e2f":{"kind":"REWORK","_number":1,"created":"2026-07-15 10:41:05.000000000","uploader":{"_account_id":34452,"name":"Joan Gilabert","display_name":"jgilaber","email":"jgilaber@redhat.com","username":"jgilaber"},"ref":"refs/changes/64/997364/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cyborg","ref":"refs/changes/64/997364/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/64/997364/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/64/997364/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/64/997364/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cyborg refs/changes/64/997364/1"}}},"commit":{"parents":[{"commit":"06393d6d93bbd8369c3dc3901ed59e766cdc4d57","subject":"Add mdev support to pci-sim DevStack plugin","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/06393d6d93bbd8369c3dc3901ed59e766cdc4d57"}]}],"author":{"name":"jgilaber","email":"jgilaber@redhat.com","date":"2026-07-15 10:24:41.000000000","tz":120},"committer":{"name":"jgilaber","email":"jgilaber@redhat.com","date":"2026-07-15 10:40:30.000000000","tz":120},"subject":"[WIP] Configure mdev driver lifecycle tests","message":"[WIP] Configure mdev driver lifecycle tests\n\nChange-Id: I55eb22a68d3e1582cbcc337b85481ac4aabff971\nSigned-off-by: jgilaber \u003cjgilaber@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/46f9392954e2c2beb5e0da40098d865daf5b9e2f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/46f9392954e2c2beb5e0da40098d865daf5b9e2f"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"06393d6d93bbd8369c3dc3901ed59e766cdc4d57","is_merged_in_target_branch":false,"change_id":"I591e9ee7480ad602cea09c57d866be654e5238de","change_number":997055,"patch_set_number":2,"change_status":"NEW"}],"branch":"refs/heads/master"},"4aa53fad4aa90b42c6aaf3159b25059391ef0046":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2026-07-15 10:48:14.000000000","uploader":{"_account_id":34452,"name":"Joan Gilabert","display_name":"jgilaber","email":"jgilaber@redhat.com","username":"jgilaber"},"ref":"refs/changes/64/997364/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cyborg","ref":"refs/changes/64/997364/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/64/997364/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/64/997364/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/64/997364/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cyborg refs/changes/64/997364/2"}}},"commit":{"parents":[{"commit":"06393d6d93bbd8369c3dc3901ed59e766cdc4d57","subject":"Add mdev support to pci-sim DevStack plugin","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/06393d6d93bbd8369c3dc3901ed59e766cdc4d57"}]}],"author":{"name":"jgilaber","email":"jgilaber@redhat.com","date":"2026-07-15 10:24:41.000000000","tz":120},"committer":{"name":"jgilaber","email":"jgilaber@redhat.com","date":"2026-07-15 10:46:42.000000000","tz":120},"subject":"[WIP] Configure mdev driver lifecycle tests","message":"[WIP] Configure mdev driver lifecycle tests\n\nDepends-On: https://review.opendev.org/c/openstack/cyborg-tempest-plugin/+/997352\nChange-Id: I55eb22a68d3e1582cbcc337b85481ac4aabff971\nSigned-off-by: jgilaber \u003cjgilaber@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/4aa53fad4aa90b42c6aaf3159b25059391ef0046"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/4aa53fad4aa90b42c6aaf3159b25059391ef0046"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"06393d6d93bbd8369c3dc3901ed59e766cdc4d57","is_merged_in_target_branch":false,"change_id":"I591e9ee7480ad602cea09c57d866be654e5238de","change_number":997055,"patch_set_number":2,"change_status":"NEW"}],"branch":"refs/heads/master"},"a3c48fbef94f0e88ac8beaa6a34e907d25cddeaf":{"kind":"REWORK","_number":3,"created":"2026-07-15 11:21:01.000000000","uploader":{"_account_id":34452,"name":"Joan Gilabert","display_name":"jgilaber","email":"jgilaber@redhat.com","username":"jgilaber"},"ref":"refs/changes/64/997364/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cyborg","ref":"refs/changes/64/997364/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/64/997364/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/64/997364/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/64/997364/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cyborg refs/changes/64/997364/3"}}},"commit":{"parents":[{"commit":"06393d6d93bbd8369c3dc3901ed59e766cdc4d57","subject":"Add mdev support to pci-sim DevStack plugin","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/06393d6d93bbd8369c3dc3901ed59e766cdc4d57"}]}],"author":{"name":"jgilaber","email":"jgilaber@redhat.com","date":"2026-07-15 10:24:41.000000000","tz":120},"committer":{"name":"jgilaber","email":"jgilaber@redhat.com","date":"2026-07-15 11:20:41.000000000","tz":120},"subject":"[WIP] Configure mdev driver lifecycle tests","message":"[WIP] Configure mdev driver lifecycle tests\n\nDepends-On: https://review.opendev.org/c/openstack/cyborg-tempest-plugin/+/997352\nChange-Id: I55eb22a68d3e1582cbcc337b85481ac4aabff971\nSigned-off-by: jgilaber \u003cjgilaber@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/a3c48fbef94f0e88ac8beaa6a34e907d25cddeaf"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/a3c48fbef94f0e88ac8beaa6a34e907d25cddeaf"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"06393d6d93bbd8369c3dc3901ed59e766cdc4d57","is_merged_in_target_branch":false,"change_id":"I591e9ee7480ad602cea09c57d866be654e5238de","change_number":997055,"patch_set_number":2,"change_status":"NEW"}],"branch":"refs/heads/master"},"9bb8d97991bdff2fb785c54311b9e0e647177397":{"kind":"REWORK","_number":4,"created":"2026-07-15 12:02:11.000000000","uploader":{"_account_id":34452,"name":"Joan Gilabert","display_name":"jgilaber","email":"jgilaber@redhat.com","username":"jgilaber"},"ref":"refs/changes/64/997364/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cyborg","ref":"refs/changes/64/997364/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/64/997364/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/64/997364/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/64/997364/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cyborg refs/changes/64/997364/4"}}},"commit":{"parents":[{"commit":"06393d6d93bbd8369c3dc3901ed59e766cdc4d57","subject":"Add mdev support to pci-sim DevStack plugin","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/06393d6d93bbd8369c3dc3901ed59e766cdc4d57"}]}],"author":{"name":"jgilaber","email":"jgilaber@redhat.com","date":"2026-07-15 10:24:41.000000000","tz":120},"committer":{"name":"jgilaber","email":"jgilaber@redhat.com","date":"2026-07-15 12:02:01.000000000","tz":120},"subject":"[WIP] Configure mdev driver lifecycle tests","message":"[WIP] Configure mdev driver lifecycle tests\n\nDepends-On: https://review.opendev.org/c/openstack/cyborg-tempest-plugin/+/997352\nChange-Id: I55eb22a68d3e1582cbcc337b85481ac4aabff971\nSigned-off-by: jgilaber \u003cjgilaber@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/9bb8d97991bdff2fb785c54311b9e0e647177397"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/9bb8d97991bdff2fb785c54311b9e0e647177397"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"06393d6d93bbd8369c3dc3901ed59e766cdc4d57","is_merged_in_target_branch":false,"change_id":"I591e9ee7480ad602cea09c57d866be654e5238de","change_number":997055,"patch_set_number":2,"change_status":"NEW"}],"branch":"refs/heads/master"},"ba6d7e6e2d31aca669fabf65cb9389630ad24bd5":{"kind":"REWORK","_number":5,"created":"2026-07-15 13:12:26.000000000","uploader":{"_account_id":34452,"name":"Joan Gilabert","display_name":"jgilaber","email":"jgilaber@redhat.com","username":"jgilaber"},"ref":"refs/changes/64/997364/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cyborg","ref":"refs/changes/64/997364/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/64/997364/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/64/997364/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/64/997364/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cyborg refs/changes/64/997364/5"}}},"commit":{"parents":[{"commit":"06393d6d93bbd8369c3dc3901ed59e766cdc4d57","subject":"Add mdev support to pci-sim DevStack plugin","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/06393d6d93bbd8369c3dc3901ed59e766cdc4d57"}]}],"author":{"name":"jgilaber","email":"jgilaber@redhat.com","date":"2026-07-15 10:24:41.000000000","tz":120},"committer":{"name":"jgilaber","email":"jgilaber@redhat.com","date":"2026-07-15 13:10:35.000000000","tz":120},"subject":"[WIP] Configure mdev driver lifecycle tests","message":"[WIP] Configure mdev driver lifecycle tests\n\nDepends-On: https://review.opendev.org/c/openstack/cyborg-tempest-plugin/+/997352\nChange-Id: I55eb22a68d3e1582cbcc337b85481ac4aabff971\nSigned-off-by: jgilaber \u003cjgilaber@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/ba6d7e6e2d31aca669fabf65cb9389630ad24bd5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/ba6d7e6e2d31aca669fabf65cb9389630ad24bd5"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"06393d6d93bbd8369c3dc3901ed59e766cdc4d57","is_merged_in_target_branch":false,"change_id":"I591e9ee7480ad602cea09c57d866be654e5238de","change_number":997055,"patch_set_number":2,"change_status":"NEW"}],"branch":"refs/heads/master"},"d88fd1f9855cd7e08442c7795a75317cdd3decf3":{"kind":"NO_CODE_CHANGE","_number":6,"created":"2026-07-15 14:42:56.000000000","uploader":{"_account_id":34452,"name":"Joan Gilabert","display_name":"jgilaber","email":"jgilaber@redhat.com","username":"jgilaber"},"ref":"refs/changes/64/997364/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cyborg","ref":"refs/changes/64/997364/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/64/997364/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/64/997364/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/64/997364/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cyborg refs/changes/64/997364/6"}}},"commit":{"parents":[{"commit":"06393d6d93bbd8369c3dc3901ed59e766cdc4d57","subject":"Add mdev support to pci-sim DevStack plugin","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/06393d6d93bbd8369c3dc3901ed59e766cdc4d57"}]}],"author":{"name":"jgilaber","email":"jgilaber@redhat.com","date":"2026-07-15 10:24:41.000000000","tz":120},"committer":{"name":"jgilaber","email":"jgilaber@redhat.com","date":"2026-07-15 14:42:42.000000000","tz":120},"subject":"[WIP] Configure mdev driver lifecycle tests","message":"[WIP] Configure mdev driver lifecycle tests\n\nDepends-On: https://review.opendev.org/c/openstack/cyborg-tempest-plugin/+/997352\nDepends-On: https://review.opendev.org/c/openstack/nova/+/995840\nChange-Id: I55eb22a68d3e1582cbcc337b85481ac4aabff971\nSigned-off-by: jgilaber \u003cjgilaber@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/d88fd1f9855cd7e08442c7795a75317cdd3decf3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/d88fd1f9855cd7e08442c7795a75317cdd3decf3"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"06393d6d93bbd8369c3dc3901ed59e766cdc4d57","is_merged_in_target_branch":false,"change_id":"I591e9ee7480ad602cea09c57d866be654e5238de","change_number":997055,"patch_set_number":2,"change_status":"NEW"}],"branch":"refs/heads/master"},"7c2d14d87087373b341ecf9ef98a29f92db14ce8":{"kind":"REWORK","_number":7,"created":"2026-07-15 16:05:11.000000000","uploader":{"_account_id":34452,"name":"Joan Gilabert","display_name":"jgilaber","email":"jgilaber@redhat.com","username":"jgilaber"},"ref":"refs/changes/64/997364/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cyborg","ref":"refs/changes/64/997364/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/64/997364/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/64/997364/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/64/997364/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cyborg refs/changes/64/997364/7"}}},"commit":{"parents":[{"commit":"06393d6d93bbd8369c3dc3901ed59e766cdc4d57","subject":"Add mdev support to pci-sim DevStack plugin","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/06393d6d93bbd8369c3dc3901ed59e766cdc4d57"}]}],"author":{"name":"jgilaber","email":"jgilaber@redhat.com","date":"2026-07-15 10:24:41.000000000","tz":120},"committer":{"name":"jgilaber","email":"jgilaber@redhat.com","date":"2026-07-15 16:04:59.000000000","tz":120},"subject":"[WIP] Configure mdev driver lifecycle tests","message":"[WIP] Configure mdev driver lifecycle tests\n\nDepends-On: https://review.opendev.org/c/openstack/cyborg-tempest-plugin/+/997352\nDepends-On: https://review.opendev.org/c/openstack/nova/+/995840\nChange-Id: I55eb22a68d3e1582cbcc337b85481ac4aabff971\nSigned-off-by: jgilaber \u003cjgilaber@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/7c2d14d87087373b341ecf9ef98a29f92db14ce8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/7c2d14d87087373b341ecf9ef98a29f92db14ce8"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"06393d6d93bbd8369c3dc3901ed59e766cdc4d57","is_merged_in_target_branch":false,"change_id":"I591e9ee7480ad602cea09c57d866be654e5238de","change_number":997055,"patch_set_number":2,"change_status":"NEW"}],"branch":"refs/heads/master"},"20e8631f82f8fa87cf3c4ad0e569229656a9061d":{"kind":"REWORK","_number":8,"created":"2026-07-16 07:24:37.000000000","uploader":{"_account_id":34452,"name":"Joan Gilabert","display_name":"jgilaber","email":"jgilaber@redhat.com","username":"jgilaber"},"ref":"refs/changes/64/997364/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cyborg","ref":"refs/changes/64/997364/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/64/997364/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/64/997364/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/64/997364/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cyborg refs/changes/64/997364/8"}}},"commit":{"parents":[{"commit":"06393d6d93bbd8369c3dc3901ed59e766cdc4d57","subject":"Add mdev support to pci-sim DevStack plugin","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/06393d6d93bbd8369c3dc3901ed59e766cdc4d57"}]}],"author":{"name":"jgilaber","email":"jgilaber@redhat.com","date":"2026-07-15 10:24:41.000000000","tz":120},"committer":{"name":"jgilaber","email":"jgilaber@redhat.com","date":"2026-07-16 07:14:14.000000000","tz":120},"subject":"[WIP] Configure mdev driver lifecycle tests","message":"[WIP] Configure mdev driver lifecycle tests\n\nDepends-On: https://review.opendev.org/c/openstack/cyborg-tempest-plugin/+/997352\nDepends-On: https://review.opendev.org/c/openstack/nova/+/995840\nChange-Id: I55eb22a68d3e1582cbcc337b85481ac4aabff971\nSigned-off-by: jgilaber \u003cjgilaber@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/20e8631f82f8fa87cf3c4ad0e569229656a9061d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/20e8631f82f8fa87cf3c4ad0e569229656a9061d"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"06393d6d93bbd8369c3dc3901ed59e766cdc4d57","is_merged_in_target_branch":false,"change_id":"I591e9ee7480ad602cea09c57d866be654e5238de","change_number":997055,"patch_set_number":2,"change_status":"NEW"}],"branch":"refs/heads/master"},"61cb02d32532e567679ae84985f7cce6dd4c2619":{"kind":"REWORK","_number":9,"created":"2026-07-16 09:12:32.000000000","uploader":{"_account_id":34452,"name":"Joan Gilabert","display_name":"jgilaber","email":"jgilaber@redhat.com","username":"jgilaber"},"ref":"refs/changes/64/997364/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cyborg","ref":"refs/changes/64/997364/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/64/997364/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/64/997364/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/64/997364/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cyborg refs/changes/64/997364/9"}}},"commit":{"parents":[{"commit":"06393d6d93bbd8369c3dc3901ed59e766cdc4d57","subject":"Add mdev support to pci-sim DevStack plugin","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/06393d6d93bbd8369c3dc3901ed59e766cdc4d57"}]}],"author":{"name":"jgilaber","email":"jgilaber@redhat.com","date":"2026-07-15 10:24:41.000000000","tz":120},"committer":{"name":"jgilaber","email":"jgilaber@redhat.com","date":"2026-07-16 09:11:40.000000000","tz":120},"subject":"[WIP] Configure mdev driver lifecycle tests","message":"[WIP] Configure mdev driver lifecycle tests\n\nConfigure tempest in the devstack plugin to add the cyborg_mdev section\npopulated with the appropiate values to run the mdev tests.\n\nDepends-On: https://review.opendev.org/c/openstack/cyborg-tempest-plugin/+/997352\nDepends-On: https://review.opendev.org/c/openstack/nova/+/995840\nChange-Id: I55eb22a68d3e1582cbcc337b85481ac4aabff971\nSigned-off-by: jgilaber \u003cjgilaber@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/61cb02d32532e567679ae84985f7cce6dd4c2619"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/61cb02d32532e567679ae84985f7cce6dd4c2619"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"06393d6d93bbd8369c3dc3901ed59e766cdc4d57","is_merged_in_target_branch":false,"change_id":"I591e9ee7480ad602cea09c57d866be654e5238de","change_number":997055,"patch_set_number":2,"change_status":"NEW"}],"branch":"refs/heads/master"},"9f6cb12b774ec9b014bc25cd2e10959b300887dc":{"kind":"REWORK","_number":10,"created":"2026-07-16 12:13:34.000000000","uploader":{"_account_id":34452,"name":"Joan Gilabert","display_name":"jgilaber","email":"jgilaber@redhat.com","username":"jgilaber"},"ref":"refs/changes/64/997364/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cyborg","ref":"refs/changes/64/997364/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/64/997364/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/64/997364/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/64/997364/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cyborg refs/changes/64/997364/10"}}},"commit":{"parents":[{"commit":"06393d6d93bbd8369c3dc3901ed59e766cdc4d57","subject":"Add mdev support to pci-sim DevStack plugin","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/06393d6d93bbd8369c3dc3901ed59e766cdc4d57"}]}],"author":{"name":"jgilaber","email":"jgilaber@redhat.com","date":"2026-07-15 10:24:41.000000000","tz":120},"committer":{"name":"jgilaber","email":"jgilaber@redhat.com","date":"2026-07-16 12:06:02.000000000","tz":120},"subject":"[WIP] Configure mdev driver lifecycle tests","message":"[WIP] Configure mdev driver lifecycle tests\n\nConfigure tempest in the devstack plugin to add the cyborg_mdev section\npopulated with the appropiate values to run the mdev tests.\n\nDepends-On: https://review.opendev.org/c/openstack/cyborg-tempest-plugin/+/997352\nDepends-On: https://review.opendev.org/c/openstack/nova/+/995840\nChange-Id: I55eb22a68d3e1582cbcc337b85481ac4aabff971\nSigned-off-by: jgilaber \u003cjgilaber@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/9f6cb12b774ec9b014bc25cd2e10959b300887dc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/9f6cb12b774ec9b014bc25cd2e10959b300887dc"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"06393d6d93bbd8369c3dc3901ed59e766cdc4d57","is_merged_in_target_branch":false,"change_id":"I591e9ee7480ad602cea09c57d866be654e5238de","change_number":997055,"patch_set_number":2,"change_status":"NEW"}],"branch":"refs/heads/master"},"21a845edbd1600980bb7dcdd1bf591317bcb1b06":{"kind":"TRIVIAL_REBASE","_number":11,"created":"2026-07-16 16:37:26.000000000","uploader":{"_account_id":34452,"name":"Joan Gilabert","display_name":"jgilaber","email":"jgilaber@redhat.com","username":"jgilaber"},"ref":"refs/changes/64/997364/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cyborg","ref":"refs/changes/64/997364/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/64/997364/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/64/997364/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/64/997364/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cyborg refs/changes/64/997364/11"}}},"commit":{"parents":[{"commit":"ec0e9fbbff4b62ce15f5c6138b0c43624ae292c4","subject":"Add mdev support to pci-sim DevStack plugin","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/ec0e9fbbff4b62ce15f5c6138b0c43624ae292c4"}]}],"author":{"name":"jgilaber","email":"jgilaber@redhat.com","date":"2026-07-15 10:24:41.000000000","tz":120},"committer":{"name":"jgilaber","email":"jgilaber@redhat.com","date":"2026-07-16 16:02:09.000000000","tz":120},"subject":"[WIP] Configure mdev driver lifecycle tests","message":"[WIP] Configure mdev driver lifecycle tests\n\nConfigure tempest in the devstack plugin to add the cyborg_mdev section\npopulated with the appropiate values to run the mdev tests.\n\nDepends-On: https://review.opendev.org/c/openstack/cyborg-tempest-plugin/+/997352\nDepends-On: https://review.opendev.org/c/openstack/nova/+/995840\nChange-Id: I55eb22a68d3e1582cbcc337b85481ac4aabff971\nSigned-off-by: jgilaber \u003cjgilaber@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/21a845edbd1600980bb7dcdd1bf591317bcb1b06"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/21a845edbd1600980bb7dcdd1bf591317bcb1b06"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"ec0e9fbbff4b62ce15f5c6138b0c43624ae292c4","is_merged_in_target_branch":false,"change_id":"I591e9ee7480ad602cea09c57d866be654e5238de","change_number":997055,"patch_set_number":3,"change_status":"NEW"}],"branch":"refs/heads/master"},"6d1147b29353b371059668f20cd17001000f7785":{"kind":"TRIVIAL_REBASE","_number":12,"created":"2026-07-16 17:45:15.000000000","uploader":{"_account_id":34452,"name":"Joan Gilabert","display_name":"jgilaber","email":"jgilaber@redhat.com","username":"jgilaber"},"ref":"refs/changes/64/997364/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cyborg","ref":"refs/changes/64/997364/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/64/997364/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/64/997364/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/64/997364/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cyborg refs/changes/64/997364/12"}}},"commit":{"parents":[{"commit":"1260c771d0a6db4c15aa828a4617184625dfdacf","subject":"Add mdev support to pci-sim DevStack plugin","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/1260c771d0a6db4c15aa828a4617184625dfdacf"}]}],"author":{"name":"jgilaber","email":"jgilaber@redhat.com","date":"2026-07-15 10:24:41.000000000","tz":120},"committer":{"name":"jgilaber","email":"jgilaber@redhat.com","date":"2026-07-16 17:44:56.000000000","tz":120},"subject":"[WIP] Configure mdev driver lifecycle tests","message":"[WIP] Configure mdev driver lifecycle tests\n\nConfigure tempest in the devstack plugin to add the cyborg_mdev section\npopulated with the appropiate values to run the mdev tests.\n\nDepends-On: https://review.opendev.org/c/openstack/cyborg-tempest-plugin/+/997352\nDepends-On: https://review.opendev.org/c/openstack/nova/+/995840\nChange-Id: I55eb22a68d3e1582cbcc337b85481ac4aabff971\nSigned-off-by: jgilaber \u003cjgilaber@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/6d1147b29353b371059668f20cd17001000f7785"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/6d1147b29353b371059668f20cd17001000f7785"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"1260c771d0a6db4c15aa828a4617184625dfdacf","is_merged_in_target_branch":false,"change_id":"I591e9ee7480ad602cea09c57d866be654e5238de","change_number":997055,"patch_set_number":4,"change_status":"NEW"}],"branch":"refs/heads/master"},"8c81513b8fb871c10895791c0f833f7ab890946c":{"kind":"REWORK","_number":13,"created":"2026-07-20 10:11:47.000000000","uploader":{"_account_id":34452,"name":"Joan Gilabert","display_name":"jgilaber","email":"jgilaber@redhat.com","username":"jgilaber"},"ref":"refs/changes/64/997364/13","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cyborg","ref":"refs/changes/64/997364/13","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/64/997364/13 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/64/997364/13 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/64/997364/13 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cyborg refs/changes/64/997364/13"}}},"commit":{"parents":[{"commit":"9b651d33534fbc39bb37f15d24812954a15ff43c","subject":"Add mdev support to pci-sim DevStack plugin","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/9b651d33534fbc39bb37f15d24812954a15ff43c"}]}],"author":{"name":"jgilaber","email":"jgilaber@redhat.com","date":"2026-07-15 10:24:41.000000000","tz":120},"committer":{"name":"jgilaber","email":"jgilaber@redhat.com","date":"2026-07-20 10:11:38.000000000","tz":120},"subject":"[WIP] Configure mdev driver lifecycle tests","message":"[WIP] Configure mdev driver lifecycle tests\n\nConfigure tempest in the devstack plugin to add the cyborg_mdev section\npopulated with the appropiate values to run the mdev tests.\n\nDepends-On: https://review.opendev.org/c/openstack/cyborg-tempest-plugin/+/997352\nDepends-On: https://review.opendev.org/c/openstack/nova/+/995840\nChange-Id: I55eb22a68d3e1582cbcc337b85481ac4aabff971\nSigned-off-by: jgilaber \u003cjgilaber@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/8c81513b8fb871c10895791c0f833f7ab890946c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/8c81513b8fb871c10895791c0f833f7ab890946c"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"9b651d33534fbc39bb37f15d24812954a15ff43c","is_merged_in_target_branch":false,"change_id":"I591e9ee7480ad602cea09c57d866be654e5238de","change_number":997055,"patch_set_number":5,"change_status":"NEW"}],"branch":"refs/heads/master"},"7b3640d3ad73b82dd1017c69c8c0e07246b28637":{"kind":"REWORK","_number":14,"created":"2026-07-20 10:13:42.000000000","uploader":{"_account_id":34452,"name":"Joan Gilabert","display_name":"jgilaber","email":"jgilaber@redhat.com","username":"jgilaber"},"ref":"refs/changes/64/997364/14","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cyborg","ref":"refs/changes/64/997364/14","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/64/997364/14 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/64/997364/14 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/64/997364/14 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cyborg refs/changes/64/997364/14"}}},"commit":{"parents":[{"commit":"9b651d33534fbc39bb37f15d24812954a15ff43c","subject":"Add mdev support to pci-sim DevStack plugin","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/9b651d33534fbc39bb37f15d24812954a15ff43c"}]}],"author":{"name":"jgilaber","email":"jgilaber@redhat.com","date":"2026-07-15 10:24:41.000000000","tz":120},"committer":{"name":"jgilaber","email":"jgilaber@redhat.com","date":"2026-07-20 10:13:32.000000000","tz":120},"subject":"[WIP] Configure mdev driver lifecycle tests","message":"[WIP] Configure mdev driver lifecycle tests\n\nConfigure tempest in the devstack plugin to add the cyborg_mdev section\npopulated with the appropiate values to run the mdev tests.\n\nDepends-On: https://review.opendev.org/c/openstack/cyborg-tempest-plugin/+/997352\nDepends-On: https://review.opendev.org/c/openstack/nova/+/995840\nChange-Id: I55eb22a68d3e1582cbcc337b85481ac4aabff971\nSigned-off-by: jgilaber \u003cjgilaber@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/7b3640d3ad73b82dd1017c69c8c0e07246b28637"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/7b3640d3ad73b82dd1017c69c8c0e07246b28637"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"9b651d33534fbc39bb37f15d24812954a15ff43c","is_merged_in_target_branch":false,"change_id":"I591e9ee7480ad602cea09c57d866be654e5238de","change_number":997055,"patch_set_number":5,"change_status":"NEW"}],"branch":"refs/heads/master"},"08a8dc77821bbd5094fcb50ecc392b5c4d55cd94":{"kind":"TRIVIAL_REBASE","_number":15,"created":"2026-07-20 12:50:20.000000000","uploader":{"_account_id":34452,"name":"Joan Gilabert","display_name":"jgilaber","email":"jgilaber@redhat.com","username":"jgilaber"},"ref":"refs/changes/64/997364/15","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cyborg","ref":"refs/changes/64/997364/15","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/64/997364/15 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/64/997364/15 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/64/997364/15 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cyborg refs/changes/64/997364/15"}}},"commit":{"parents":[{"commit":"1a12d776bf47b3fac6b363bc9db62e52cda70e63","subject":"Add mdev support to pci-sim DevStack plugin","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/1a12d776bf47b3fac6b363bc9db62e52cda70e63"}]}],"author":{"name":"jgilaber","email":"jgilaber@redhat.com","date":"2026-07-15 10:24:41.000000000","tz":120},"committer":{"name":"jgilaber","email":"jgilaber@redhat.com","date":"2026-07-20 12:49:53.000000000","tz":120},"subject":"[WIP] Configure mdev driver lifecycle tests","message":"[WIP] Configure mdev driver lifecycle tests\n\nConfigure tempest in the devstack plugin to add the cyborg_mdev section\npopulated with the appropiate values to run the mdev tests.\n\nDepends-On: https://review.opendev.org/c/openstack/cyborg-tempest-plugin/+/997352\nDepends-On: https://review.opendev.org/c/openstack/nova/+/995840\nChange-Id: I55eb22a68d3e1582cbcc337b85481ac4aabff971\nSigned-off-by: jgilaber \u003cjgilaber@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/08a8dc77821bbd5094fcb50ecc392b5c4d55cd94"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/08a8dc77821bbd5094fcb50ecc392b5c4d55cd94"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"1a12d776bf47b3fac6b363bc9db62e52cda70e63","is_merged_in_target_branch":false,"change_id":"I591e9ee7480ad602cea09c57d866be654e5238de","change_number":997055,"patch_set_number":6,"change_status":"NEW"}],"branch":"refs/heads/master"},"1fe74b5d74865be41c5d63fba6b18d47d8cd0e5b":{"kind":"TRIVIAL_REBASE_WITH_MESSAGE_UPDATE","_number":16,"created":"2026-07-22 11:41:22.000000000","uploader":{"_account_id":34452,"name":"Joan Gilabert","display_name":"jgilaber","email":"jgilaber@redhat.com","username":"jgilaber"},"ref":"refs/changes/64/997364/16","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cyborg","ref":"refs/changes/64/997364/16","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/64/997364/16 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/64/997364/16 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/64/997364/16 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cyborg refs/changes/64/997364/16"}}},"commit":{"parents":[{"commit":"3540d1b6477083e71e6b29e6b6353fc70c087706","subject":"Add mdev support to pci-sim DevStack plugin","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/3540d1b6477083e71e6b29e6b6353fc70c087706"}]}],"author":{"name":"jgilaber","email":"jgilaber@redhat.com","date":"2026-07-15 10:24:41.000000000","tz":120},"committer":{"name":"jgilaber","email":"jgilaber@redhat.com","date":"2026-07-22 11:40:55.000000000","tz":120},"subject":"[WIP] Configure mdev driver lifecycle tests","message":"[WIP] Configure mdev driver lifecycle tests\n\nConfigure tempest in the devstack plugin to add the cyborg_mdev section\npopulated with the appropriate values to run the mdev tests.\n\nDepends-On: https://review.opendev.org/c/openstack/cyborg-tempest-plugin/+/997352\nDepends-On: https://review.opendev.org/c/openstack/nova/+/995840\nChange-Id: I55eb22a68d3e1582cbcc337b85481ac4aabff971\nSigned-off-by: jgilaber \u003cjgilaber@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/1fe74b5d74865be41c5d63fba6b18d47d8cd0e5b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/1fe74b5d74865be41c5d63fba6b18d47d8cd0e5b"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"3540d1b6477083e71e6b29e6b6353fc70c087706","is_merged_in_target_branch":false,"change_id":"I591e9ee7480ad602cea09c57d866be654e5238de","change_number":997055,"patch_set_number":7,"change_status":"NEW"}],"branch":"refs/heads/master"},"79436935ce16773318e7757d77dede46f529f10d":{"kind":"TRIVIAL_REBASE","_number":17,"created":"2026-07-22 14:57:35.000000000","uploader":{"_account_id":34452,"name":"Joan Gilabert","display_name":"jgilaber","email":"jgilaber@redhat.com","username":"jgilaber"},"ref":"refs/changes/64/997364/17","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cyborg","ref":"refs/changes/64/997364/17","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/64/997364/17 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/64/997364/17 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/64/997364/17 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cyborg refs/changes/64/997364/17"}}},"commit":{"parents":[{"commit":"a729b48fd26e2fab08802985187b4f6a476a8cca","subject":"Add mdev support to pci-sim DevStack plugin","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/a729b48fd26e2fab08802985187b4f6a476a8cca"}]}],"author":{"name":"jgilaber","email":"jgilaber@redhat.com","date":"2026-07-15 10:24:41.000000000","tz":120},"committer":{"name":"jgilaber","email":"jgilaber@redhat.com","date":"2026-07-22 14:57:17.000000000","tz":120},"subject":"[WIP] Configure mdev driver lifecycle tests","message":"[WIP] Configure mdev driver lifecycle tests\n\nConfigure tempest in the devstack plugin to add the cyborg_mdev section\npopulated with the appropriate values to run the mdev tests.\n\nDepends-On: https://review.opendev.org/c/openstack/cyborg-tempest-plugin/+/997352\nDepends-On: https://review.opendev.org/c/openstack/nova/+/995840\nChange-Id: I55eb22a68d3e1582cbcc337b85481ac4aabff971\nSigned-off-by: jgilaber \u003cjgilaber@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/79436935ce16773318e7757d77dede46f529f10d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/79436935ce16773318e7757d77dede46f529f10d"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"a729b48fd26e2fab08802985187b4f6a476a8cca","is_merged_in_target_branch":false,"change_id":"I591e9ee7480ad602cea09c57d866be654e5238de","change_number":997055,"patch_set_number":8,"change_status":"NEW"}],"branch":"refs/heads/master"},"555ab83065cd60c644ed0da12b7ea1544e697eb1":{"kind":"TRIVIAL_REBASE","_number":18,"created":"2026-07-23 10:16:02.000000000","uploader":{"_account_id":34452,"name":"Joan Gilabert","display_name":"jgilaber","email":"jgilaber@redhat.com","username":"jgilaber"},"ref":"refs/changes/64/997364/18","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cyborg","ref":"refs/changes/64/997364/18","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/64/997364/18 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/64/997364/18 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/64/997364/18 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cyborg refs/changes/64/997364/18"}}},"commit":{"parents":[{"commit":"52ad783a16344c6bc10837af6ce7cf745e58325b","subject":"Add mdev support to pci-sim DevStack plugin","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/52ad783a16344c6bc10837af6ce7cf745e58325b"}]}],"author":{"name":"jgilaber","email":"jgilaber@redhat.com","date":"2026-07-15 10:24:41.000000000","tz":120},"committer":{"name":"jgilaber","email":"jgilaber@redhat.com","date":"2026-07-23 10:15:54.000000000","tz":120},"subject":"[WIP] Configure mdev driver lifecycle tests","message":"[WIP] Configure mdev driver lifecycle tests\n\nConfigure tempest in the devstack plugin to add the cyborg_mdev section\npopulated with the appropriate values to run the mdev tests.\n\nDepends-On: https://review.opendev.org/c/openstack/cyborg-tempest-plugin/+/997352\nDepends-On: https://review.opendev.org/c/openstack/nova/+/995840\nChange-Id: I55eb22a68d3e1582cbcc337b85481ac4aabff971\nSigned-off-by: jgilaber \u003cjgilaber@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/555ab83065cd60c644ed0da12b7ea1544e697eb1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/555ab83065cd60c644ed0da12b7ea1544e697eb1"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"52ad783a16344c6bc10837af6ce7cf745e58325b","is_merged_in_target_branch":false,"change_id":"I591e9ee7480ad602cea09c57d866be654e5238de","change_number":997055,"patch_set_number":9,"change_status":"NEW"}],"branch":"refs/heads/master"},"9df16aaf6c26f17a0ef7e07b0710fc1a00d428b1":{"kind":"TRIVIAL_REBASE","_number":19,"created":"2026-07-23 13:01:51.000000000","uploader":{"_account_id":34452,"name":"Joan Gilabert","display_name":"jgilaber","email":"jgilaber@redhat.com","username":"jgilaber"},"ref":"refs/changes/64/997364/19","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cyborg","ref":"refs/changes/64/997364/19","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/64/997364/19 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/64/997364/19 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/64/997364/19 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cyborg refs/changes/64/997364/19"}}},"commit":{"parents":[{"commit":"bbff3d344e99538508b763749aa05ba774259991","subject":"Add mdev support to pci-sim DevStack plugin","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/bbff3d344e99538508b763749aa05ba774259991"}]}],"author":{"name":"jgilaber","email":"jgilaber@redhat.com","date":"2026-07-15 10:24:41.000000000","tz":120},"committer":{"name":"jgilaber","email":"jgilaber@redhat.com","date":"2026-07-23 13:01:44.000000000","tz":120},"subject":"[WIP] Configure mdev driver lifecycle tests","message":"[WIP] Configure mdev driver lifecycle tests\n\nConfigure tempest in the devstack plugin to add the cyborg_mdev section\npopulated with the appropriate values to run the mdev tests.\n\nDepends-On: https://review.opendev.org/c/openstack/cyborg-tempest-plugin/+/997352\nDepends-On: https://review.opendev.org/c/openstack/nova/+/995840\nChange-Id: I55eb22a68d3e1582cbcc337b85481ac4aabff971\nSigned-off-by: jgilaber \u003cjgilaber@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/9df16aaf6c26f17a0ef7e07b0710fc1a00d428b1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/9df16aaf6c26f17a0ef7e07b0710fc1a00d428b1"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"bbff3d344e99538508b763749aa05ba774259991","is_merged_in_target_branch":false,"change_id":"I591e9ee7480ad602cea09c57d866be654e5238de","change_number":997055,"patch_set_number":10,"change_status":"NEW"}],"branch":"refs/heads/master"},"2cf9632154036896f49df6fc18bbf03c5c3d8a58":{"kind":"TRIVIAL_REBASE","_number":20,"created":"2026-07-23 14:54:34.000000000","uploader":{"_account_id":34452,"name":"Joan Gilabert","display_name":"jgilaber","email":"jgilaber@redhat.com","username":"jgilaber"},"ref":"refs/changes/64/997364/20","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cyborg","ref":"refs/changes/64/997364/20","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/64/997364/20 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/64/997364/20 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/64/997364/20 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cyborg refs/changes/64/997364/20"}}},"commit":{"parents":[{"commit":"098c334fcd80e8099ca894ef96c05ff504cabba2","subject":"Add mdev support to pci-sim DevStack plugin","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/098c334fcd80e8099ca894ef96c05ff504cabba2"}]}],"author":{"name":"jgilaber","email":"jgilaber@redhat.com","date":"2026-07-15 10:24:41.000000000","tz":120},"committer":{"name":"jgilaber","email":"jgilaber@redhat.com","date":"2026-07-23 14:54:16.000000000","tz":120},"subject":"[WIP] Configure mdev driver lifecycle tests","message":"[WIP] Configure mdev driver lifecycle tests\n\nConfigure tempest in the devstack plugin to add the cyborg_mdev section\npopulated with the appropriate values to run the mdev tests.\n\nDepends-On: https://review.opendev.org/c/openstack/cyborg-tempest-plugin/+/997352\nDepends-On: https://review.opendev.org/c/openstack/nova/+/995840\nChange-Id: I55eb22a68d3e1582cbcc337b85481ac4aabff971\nSigned-off-by: jgilaber \u003cjgilaber@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/2cf9632154036896f49df6fc18bbf03c5c3d8a58"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/2cf9632154036896f49df6fc18bbf03c5c3d8a58"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"098c334fcd80e8099ca894ef96c05ff504cabba2","is_merged_in_target_branch":false,"change_id":"I591e9ee7480ad602cea09c57d866be654e5238de","change_number":997055,"patch_set_number":11,"change_status":"NEW"}],"branch":"refs/heads/master"},"63dfab8edfc7a11560c7666d70c43a5bd687d810":{"kind":"TRIVIAL_REBASE","_number":21,"created":"2026-07-23 16:19:30.000000000","uploader":{"_account_id":34452,"name":"Joan Gilabert","display_name":"jgilaber","email":"jgilaber@redhat.com","username":"jgilaber"},"ref":"refs/changes/64/997364/21","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cyborg","ref":"refs/changes/64/997364/21","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/64/997364/21 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/64/997364/21 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/64/997364/21 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cyborg refs/changes/64/997364/21"}}},"commit":{"parents":[{"commit":"fefca8858818f02606d6d2152ffaf750ed0dff42","subject":"Add mdev support to pci-sim DevStack plugin","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/fefca8858818f02606d6d2152ffaf750ed0dff42"}]}],"author":{"name":"jgilaber","email":"jgilaber@redhat.com","date":"2026-07-15 10:24:41.000000000","tz":120},"committer":{"name":"jgilaber","email":"jgilaber@redhat.com","date":"2026-07-23 16:19:09.000000000","tz":120},"subject":"[WIP] Configure mdev driver lifecycle tests","message":"[WIP] Configure mdev driver lifecycle tests\n\nConfigure tempest in the devstack plugin to add the cyborg_mdev section\npopulated with the appropriate values to run the mdev tests.\n\nDepends-On: https://review.opendev.org/c/openstack/cyborg-tempest-plugin/+/997352\nDepends-On: https://review.opendev.org/c/openstack/nova/+/995840\nChange-Id: I55eb22a68d3e1582cbcc337b85481ac4aabff971\nSigned-off-by: jgilaber \u003cjgilaber@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/63dfab8edfc7a11560c7666d70c43a5bd687d810"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/63dfab8edfc7a11560c7666d70c43a5bd687d810"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"fefca8858818f02606d6d2152ffaf750ed0dff42","is_merged_in_target_branch":false,"change_id":"I591e9ee7480ad602cea09c57d866be654e5238de","change_number":997055,"patch_set_number":12,"change_status":"NEW"}],"branch":"refs/heads/master"},"bff8b530e382e1b09dcd226aa48f7108fb5ec127":{"kind":"REWORK","_number":22,"created":"2026-07-23 17:37:07.000000000","uploader":{"_account_id":34452,"name":"Joan Gilabert","display_name":"jgilaber","email":"jgilaber@redhat.com","username":"jgilaber"},"ref":"refs/changes/64/997364/22","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cyborg","ref":"refs/changes/64/997364/22","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/64/997364/22 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/64/997364/22 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/64/997364/22 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cyborg refs/changes/64/997364/22"}}},"commit":{"parents":[{"commit":"c9898aab78a166bc6f63a8c744234341a5b59abb","subject":"Add mdev support to pci-sim DevStack plugin","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/c9898aab78a166bc6f63a8c744234341a5b59abb"}]}],"author":{"name":"jgilaber","email":"jgilaber@redhat.com","date":"2026-07-15 10:24:41.000000000","tz":120},"committer":{"name":"jgilaber","email":"jgilaber@redhat.com","date":"2026-07-23 17:36:47.000000000","tz":120},"subject":"[WIP] Configure mdev driver lifecycle tests","message":"[WIP] Configure mdev driver lifecycle tests\n\nConfigure tempest in the devstack plugin to add the cyborg_mdev section\npopulated with the appropriate values to run the mdev tests.\n\nDepends-On: https://review.opendev.org/c/openstack/cyborg-tempest-plugin/+/997352\nDepends-On: https://review.opendev.org/c/openstack/nova/+/995840\nChange-Id: I55eb22a68d3e1582cbcc337b85481ac4aabff971\nSigned-off-by: jgilaber \u003cjgilaber@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/bff8b530e382e1b09dcd226aa48f7108fb5ec127"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/bff8b530e382e1b09dcd226aa48f7108fb5ec127"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"c9898aab78a166bc6f63a8c744234341a5b59abb","is_merged_in_target_branch":false,"change_id":"I591e9ee7480ad602cea09c57d866be654e5238de","change_number":997055,"patch_set_number":13,"change_status":"NEW"}],"branch":"refs/heads/master"},"8294c08b8fcd857b468cbcd06dfc550d3711052c":{"kind":"TRIVIAL_REBASE","_number":23,"created":"2026-07-24 06:37:58.000000000","uploader":{"_account_id":34452,"name":"Joan Gilabert","display_name":"jgilaber","email":"jgilaber@redhat.com","username":"jgilaber"},"ref":"refs/changes/64/997364/23","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cyborg","ref":"refs/changes/64/997364/23","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/64/997364/23 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/64/997364/23 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/64/997364/23 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cyborg refs/changes/64/997364/23"}}},"commit":{"parents":[{"commit":"c75f253365dfadf0bd488f9bc55b418a1ddf0c42","subject":"Add mdev support to pci-sim DevStack plugin","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/c75f253365dfadf0bd488f9bc55b418a1ddf0c42"}]}],"author":{"name":"jgilaber","email":"jgilaber@redhat.com","date":"2026-07-15 10:24:41.000000000","tz":120},"committer":{"name":"jgilaber","email":"jgilaber@redhat.com","date":"2026-07-24 06:37:48.000000000","tz":120},"subject":"[WIP] Configure mdev driver lifecycle tests","message":"[WIP] Configure mdev driver lifecycle tests\n\nConfigure tempest in the devstack plugin to add the cyborg_mdev section\npopulated with the appropriate values to run the mdev tests.\n\nDepends-On: https://review.opendev.org/c/openstack/cyborg-tempest-plugin/+/997352\nDepends-On: https://review.opendev.org/c/openstack/nova/+/995840\nChange-Id: I55eb22a68d3e1582cbcc337b85481ac4aabff971\nSigned-off-by: jgilaber \u003cjgilaber@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/8294c08b8fcd857b468cbcd06dfc550d3711052c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/8294c08b8fcd857b468cbcd06dfc550d3711052c"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"c75f253365dfadf0bd488f9bc55b418a1ddf0c42","is_merged_in_target_branch":false,"change_id":"I591e9ee7480ad602cea09c57d866be654e5238de","change_number":997055,"patch_set_number":14,"change_status":"NEW"}],"branch":"refs/heads/master"},"67e1f36d6dab7891bbf1a659ce989911613c8dec":{"kind":"TRIVIAL_REBASE","_number":24,"created":"2026-07-24 07:19:49.000000000","uploader":{"_account_id":34452,"name":"Joan Gilabert","display_name":"jgilaber","email":"jgilaber@redhat.com","username":"jgilaber"},"ref":"refs/changes/64/997364/24","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cyborg","ref":"refs/changes/64/997364/24","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/64/997364/24 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/64/997364/24 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/64/997364/24 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cyborg refs/changes/64/997364/24"}}},"commit":{"parents":[{"commit":"125cf68259dfe26cb7e0d1616a0614134fdedfe6","subject":"Add mdev support to pci-sim DevStack plugin","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/125cf68259dfe26cb7e0d1616a0614134fdedfe6"}]}],"author":{"name":"jgilaber","email":"jgilaber@redhat.com","date":"2026-07-15 10:24:41.000000000","tz":120},"committer":{"name":"jgilaber","email":"jgilaber@redhat.com","date":"2026-07-24 07:19:42.000000000","tz":120},"subject":"[WIP] Configure mdev driver lifecycle tests","message":"[WIP] Configure mdev driver lifecycle tests\n\nConfigure tempest in the devstack plugin to add the cyborg_mdev section\npopulated with the appropriate values to run the mdev tests.\n\nDepends-On: https://review.opendev.org/c/openstack/cyborg-tempest-plugin/+/997352\nDepends-On: https://review.opendev.org/c/openstack/nova/+/995840\nChange-Id: I55eb22a68d3e1582cbcc337b85481ac4aabff971\nSigned-off-by: jgilaber \u003cjgilaber@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/67e1f36d6dab7891bbf1a659ce989911613c8dec"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/67e1f36d6dab7891bbf1a659ce989911613c8dec"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"125cf68259dfe26cb7e0d1616a0614134fdedfe6","is_merged_in_target_branch":false,"change_id":"I591e9ee7480ad602cea09c57d866be654e5238de","change_number":997055,"patch_set_number":15,"change_status":"NEW"}],"branch":"refs/heads/master"},"2048f735a70084a5cc3cacd19fe66b92b17fbf30":{"kind":"REWORK","_number":25,"created":"2026-07-24 08:31:59.000000000","uploader":{"_account_id":34452,"name":"Joan Gilabert","display_name":"jgilaber","email":"jgilaber@redhat.com","username":"jgilaber"},"ref":"refs/changes/64/997364/25","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cyborg","ref":"refs/changes/64/997364/25","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/64/997364/25 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/64/997364/25 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/64/997364/25 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cyborg refs/changes/64/997364/25"}}},"commit":{"parents":[{"commit":"125cf68259dfe26cb7e0d1616a0614134fdedfe6","subject":"Add mdev support to pci-sim DevStack plugin","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/125cf68259dfe26cb7e0d1616a0614134fdedfe6"}]}],"author":{"name":"jgilaber","email":"jgilaber@redhat.com","date":"2026-07-15 10:24:41.000000000","tz":120},"committer":{"name":"jgilaber","email":"jgilaber@redhat.com","date":"2026-07-24 08:31:31.000000000","tz":120},"subject":"[WIP] Configure mdev driver lifecycle tests","message":"[WIP] Configure mdev driver lifecycle tests\n\nConfigure tempest in the devstack plugin to add the cyborg_mdev section\npopulated with the appropriate values to run the mdev tests.\n\nDepends-On: https://review.opendev.org/c/openstack/cyborg-tempest-plugin/+/997352\nDepends-On: https://review.opendev.org/c/openstack/nova/+/995840\nChange-Id: I55eb22a68d3e1582cbcc337b85481ac4aabff971\nSigned-off-by: jgilaber \u003cjgilaber@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/2048f735a70084a5cc3cacd19fe66b92b17fbf30"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/2048f735a70084a5cc3cacd19fe66b92b17fbf30"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"125cf68259dfe26cb7e0d1616a0614134fdedfe6","is_merged_in_target_branch":false,"change_id":"I591e9ee7480ad602cea09c57d866be654e5238de","change_number":997055,"patch_set_number":15,"change_status":"NEW"}],"branch":"refs/heads/master"},"e8f51f7bddb2c809e0538fefc78aabbbcd5cf76b":{"kind":"TRIVIAL_REBASE","_number":26,"created":"2026-07-29 10:36:45.000000000","uploader":{"_account_id":34452,"name":"Joan Gilabert","display_name":"jgilaber","email":"jgilaber@redhat.com","username":"jgilaber"},"ref":"refs/changes/64/997364/26","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cyborg","ref":"refs/changes/64/997364/26","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/64/997364/26 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/64/997364/26 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/64/997364/26 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cyborg refs/changes/64/997364/26"}}},"commit":{"parents":[{"commit":"79d1739726d4277d8926210b2d844544ff7f4389","subject":"Add mdev support to pci-sim DevStack plugin","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/79d1739726d4277d8926210b2d844544ff7f4389"}]}],"author":{"name":"jgilaber","email":"jgilaber@redhat.com","date":"2026-07-15 10:24:41.000000000","tz":120},"committer":{"name":"jgilaber","email":"jgilaber@redhat.com","date":"2026-07-29 10:36:21.000000000","tz":120},"subject":"[WIP] Configure mdev driver lifecycle tests","message":"[WIP] Configure mdev driver lifecycle tests\n\nConfigure tempest in the devstack plugin to add the cyborg_mdev section\npopulated with the appropriate values to run the mdev tests.\n\nDepends-On: https://review.opendev.org/c/openstack/cyborg-tempest-plugin/+/997352\nDepends-On: https://review.opendev.org/c/openstack/nova/+/995840\nChange-Id: I55eb22a68d3e1582cbcc337b85481ac4aabff971\nSigned-off-by: jgilaber \u003cjgilaber@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/e8f51f7bddb2c809e0538fefc78aabbbcd5cf76b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/e8f51f7bddb2c809e0538fefc78aabbbcd5cf76b"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"79d1739726d4277d8926210b2d844544ff7f4389","is_merged_in_target_branch":false,"change_id":"I591e9ee7480ad602cea09c57d866be654e5238de","change_number":997055,"patch_set_number":16,"change_status":"NEW"}],"branch":"refs/heads/master"},"e068d541f0a4c748f42f71f1417795ef4ac5d7a4":{"kind":"TRIVIAL_REBASE","_number":27,"created":"2026-07-29 15:19:08.000000000","uploader":{"_account_id":34452,"name":"Joan Gilabert","display_name":"jgilaber","email":"jgilaber@redhat.com","username":"jgilaber"},"ref":"refs/changes/64/997364/27","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cyborg","ref":"refs/changes/64/997364/27","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/64/997364/27 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/64/997364/27 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/64/997364/27 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cyborg refs/changes/64/997364/27"}}},"commit":{"parents":[{"commit":"9482736763111d2d94cd3f5d87927b914b797430","subject":"Add mdev support to pci-sim DevStack plugin","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/9482736763111d2d94cd3f5d87927b914b797430"}]}],"author":{"name":"jgilaber","email":"jgilaber@redhat.com","date":"2026-07-15 10:24:41.000000000","tz":120},"committer":{"name":"jgilaber","email":"jgilaber@redhat.com","date":"2026-07-29 14:54:30.000000000","tz":120},"subject":"[WIP] Configure mdev driver lifecycle tests","message":"[WIP] Configure mdev driver lifecycle tests\n\nConfigure tempest in the devstack plugin to add the cyborg_mdev section\npopulated with the appropriate values to run the mdev tests.\n\nDepends-On: https://review.opendev.org/c/openstack/cyborg-tempest-plugin/+/997352\nDepends-On: https://review.opendev.org/c/openstack/nova/+/995840\nChange-Id: I55eb22a68d3e1582cbcc337b85481ac4aabff971\nSigned-off-by: jgilaber \u003cjgilaber@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/e068d541f0a4c748f42f71f1417795ef4ac5d7a4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/e068d541f0a4c748f42f71f1417795ef4ac5d7a4"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"9482736763111d2d94cd3f5d87927b914b797430","is_merged_in_target_branch":false,"change_id":"I591e9ee7480ad602cea09c57d866be654e5238de","change_number":997055,"patch_set_number":17,"change_status":"NEW"}],"branch":"refs/heads/master"},"9462b3bb382959fcd9f335129a241e26d52bd88e":{"kind":"TRIVIAL_REBASE","_number":28,"created":"2026-07-29 17:35:53.000000000","uploader":{"_account_id":34452,"name":"Joan Gilabert","display_name":"jgilaber","email":"jgilaber@redhat.com","username":"jgilaber"},"ref":"refs/changes/64/997364/28","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cyborg","ref":"refs/changes/64/997364/28","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/64/997364/28 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/64/997364/28 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/64/997364/28 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cyborg refs/changes/64/997364/28"}}},"commit":{"parents":[{"commit":"cb8522c04a82e78c9763724bfd09cf4e0cd9d0e6","subject":"Add mdev support to pci-sim DevStack plugin","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/cb8522c04a82e78c9763724bfd09cf4e0cd9d0e6"}]}],"author":{"name":"jgilaber","email":"jgilaber@redhat.com","date":"2026-07-15 10:24:41.000000000","tz":120},"committer":{"name":"jgilaber","email":"jgilaber@redhat.com","date":"2026-07-29 17:35:47.000000000","tz":120},"subject":"[WIP] Configure mdev driver lifecycle tests","message":"[WIP] Configure mdev driver lifecycle tests\n\nConfigure tempest in the devstack plugin to add the cyborg_mdev section\npopulated with the appropriate values to run the mdev tests.\n\nDepends-On: https://review.opendev.org/c/openstack/cyborg-tempest-plugin/+/997352\nDepends-On: https://review.opendev.org/c/openstack/nova/+/995840\nChange-Id: I55eb22a68d3e1582cbcc337b85481ac4aabff971\nSigned-off-by: jgilaber \u003cjgilaber@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/9462b3bb382959fcd9f335129a241e26d52bd88e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/9462b3bb382959fcd9f335129a241e26d52bd88e"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"cb8522c04a82e78c9763724bfd09cf4e0cd9d0e6","is_merged_in_target_branch":false,"change_id":"I591e9ee7480ad602cea09c57d866be654e5238de","change_number":997055,"patch_set_number":18,"change_status":"NEW"}],"branch":"refs/heads/master"},"120db330e4a7c8210bd073287708a0d3f8797ddb":{"kind":"TRIVIAL_REBASE","_number":29,"created":"2026-07-30 11:06:20.000000000","uploader":{"_account_id":34452,"name":"Joan Gilabert","display_name":"jgilaber","email":"jgilaber@redhat.com","username":"jgilaber"},"ref":"refs/changes/64/997364/29","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cyborg","ref":"refs/changes/64/997364/29","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/64/997364/29 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/64/997364/29 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/64/997364/29 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cyborg refs/changes/64/997364/29"}}},"commit":{"parents":[{"commit":"0318790264b237e392e0f17bbbfd1846c1e3bef7","subject":"Add mdev support to pci-sim DevStack plugin","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/0318790264b237e392e0f17bbbfd1846c1e3bef7"}]}],"author":{"name":"jgilaber","email":"jgilaber@redhat.com","date":"2026-07-15 10:24:41.000000000","tz":120},"committer":{"name":"jgilaber","email":"jgilaber@redhat.com","date":"2026-07-30 11:04:54.000000000","tz":120},"subject":"[WIP] Configure mdev driver lifecycle tests","message":"[WIP] Configure mdev driver lifecycle tests\n\nConfigure tempest in the devstack plugin to add the cyborg_mdev section\npopulated with the appropriate values to run the mdev tests.\n\nDepends-On: https://review.opendev.org/c/openstack/cyborg-tempest-plugin/+/997352\nDepends-On: https://review.opendev.org/c/openstack/nova/+/995840\nChange-Id: I55eb22a68d3e1582cbcc337b85481ac4aabff971\nSigned-off-by: jgilaber \u003cjgilaber@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/120db330e4a7c8210bd073287708a0d3f8797ddb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/120db330e4a7c8210bd073287708a0d3f8797ddb"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"0318790264b237e392e0f17bbbfd1846c1e3bef7","is_merged_in_target_branch":false,"change_id":"I591e9ee7480ad602cea09c57d866be654e5238de","change_number":997055,"patch_set_number":19,"change_status":"NEW"}],"branch":"refs/heads/master"},"b15efebd492a4c087bc9bb50c6d1897d542c5b18":{"kind":"REWORK","_number":30,"created":"2026-07-31 11:34:40.000000000","uploader":{"_account_id":34452,"name":"Joan Gilabert","display_name":"jgilaber","email":"jgilaber@redhat.com","username":"jgilaber"},"ref":"refs/changes/64/997364/30","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cyborg","ref":"refs/changes/64/997364/30","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/64/997364/30 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/64/997364/30 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/64/997364/30 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cyborg refs/changes/64/997364/30"}}},"commit":{"parents":[{"commit":"54fd30923bf1731436d06160d83a58f7db1cabc7","subject":"Add mdev support to pci-sim DevStack plugin","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/54fd30923bf1731436d06160d83a58f7db1cabc7"}]}],"author":{"name":"jgilaber","email":"jgilaber@redhat.com","date":"2026-07-15 10:24:41.000000000","tz":120},"committer":{"name":"jgilaber","email":"jgilaber@redhat.com","date":"2026-07-31 11:34:22.000000000","tz":120},"subject":"[WIP] Configure mdev driver lifecycle tests","message":"[WIP] Configure mdev driver lifecycle tests\n\nConfigure tempest in the devstack plugin to add the cyborg_mdev section\npopulated with the appropriate values to run the mdev tests.\n\nDepends-On: https://review.opendev.org/c/openstack/cyborg-tempest-plugin/+/997352\nDepends-On: https://review.opendev.org/c/openstack/nova/+/995840\nChange-Id: I55eb22a68d3e1582cbcc337b85481ac4aabff971\nSigned-off-by: jgilaber \u003cjgilaber@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/b15efebd492a4c087bc9bb50c6d1897d542c5b18"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/b15efebd492a4c087bc9bb50c6d1897d542c5b18"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"54fd30923bf1731436d06160d83a58f7db1cabc7","is_merged_in_target_branch":false,"change_id":"I591e9ee7480ad602cea09c57d866be654e5238de","change_number":997055,"patch_set_number":20,"change_status":"NEW"}],"branch":"refs/heads/master"},"2f4cbee9dc4847bcced60b0df65aecdb4f231b32":{"kind":"TRIVIAL_REBASE","_number":31,"created":"2026-08-03 08:02:03.000000000","uploader":{"_account_id":34452,"name":"Joan Gilabert","display_name":"jgilaber","email":"jgilaber@redhat.com","username":"jgilaber"},"ref":"refs/changes/64/997364/31","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cyborg","ref":"refs/changes/64/997364/31","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/64/997364/31 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/64/997364/31 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/64/997364/31 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cyborg refs/changes/64/997364/31"}}},"commit":{"parents":[{"commit":"8861d9eac085ad9cf31be922044da67603a1a7bd","subject":"Add mdev support to pci-sim DevStack plugin","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/8861d9eac085ad9cf31be922044da67603a1a7bd"}]}],"author":{"name":"jgilaber","email":"jgilaber@redhat.com","date":"2026-07-15 10:24:41.000000000","tz":120},"committer":{"name":"jgilaber","email":"jgilaber@redhat.com","date":"2026-08-03 08:01:55.000000000","tz":120},"subject":"[WIP] Configure mdev driver lifecycle tests","message":"[WIP] Configure mdev driver lifecycle tests\n\nConfigure tempest in the devstack plugin to add the cyborg_mdev section\npopulated with the appropriate values to run the mdev tests.\n\nDepends-On: https://review.opendev.org/c/openstack/cyborg-tempest-plugin/+/997352\nDepends-On: https://review.opendev.org/c/openstack/nova/+/995840\nChange-Id: I55eb22a68d3e1582cbcc337b85481ac4aabff971\nSigned-off-by: jgilaber \u003cjgilaber@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/2f4cbee9dc4847bcced60b0df65aecdb4f231b32"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/2f4cbee9dc4847bcced60b0df65aecdb4f231b32"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"8861d9eac085ad9cf31be922044da67603a1a7bd","is_merged_in_target_branch":false,"change_id":"I591e9ee7480ad602cea09c57d866be654e5238de","change_number":997055,"patch_set_number":21,"change_status":"NEW"}],"branch":"refs/heads/master"},"ff9e7f8baf4a90308fb70c42538047697058989f":{"kind":"NO_CHANGE","_number":32,"created":"2026-08-04 16:14:06.000000000","uploader":{"_account_id":34452,"name":"Joan Gilabert","display_name":"jgilaber","email":"jgilaber@redhat.com","username":"jgilaber"},"ref":"refs/changes/64/997364/32","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cyborg","ref":"refs/changes/64/997364/32","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/64/997364/32 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/64/997364/32 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/64/997364/32 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cyborg refs/changes/64/997364/32"}}},"commit":{"parents":[{"commit":"adcf71d16b358e1d5e8a7bc2805b8396965a62d4","subject":"Add mdev support to pci-sim DevStack plugin","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/adcf71d16b358e1d5e8a7bc2805b8396965a62d4"}]}],"author":{"name":"jgilaber","email":"jgilaber@redhat.com","date":"2026-07-15 10:24:41.000000000","tz":120},"committer":{"name":"jgilaber","email":"jgilaber@redhat.com","date":"2026-08-04 14:19:45.000000000","tz":120},"subject":"[WIP] Configure mdev driver lifecycle tests","message":"[WIP] Configure mdev driver lifecycle tests\n\nConfigure tempest in the devstack plugin to add the cyborg_mdev section\npopulated with the appropriate values to run the mdev tests.\n\nDepends-On: https://review.opendev.org/c/openstack/cyborg-tempest-plugin/+/997352\nDepends-On: https://review.opendev.org/c/openstack/nova/+/995840\nChange-Id: I55eb22a68d3e1582cbcc337b85481ac4aabff971\nSigned-off-by: jgilaber \u003cjgilaber@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/ff9e7f8baf4a90308fb70c42538047697058989f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/ff9e7f8baf4a90308fb70c42538047697058989f"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"adcf71d16b358e1d5e8a7bc2805b8396965a62d4","is_merged_in_target_branch":false,"change_id":"I591e9ee7480ad602cea09c57d866be654e5238de","change_number":997055,"patch_set_number":22,"change_status":"NEW"}],"branch":"refs/heads/master"},"f83c43f60410c81e25830d2f9fe885ea76b25055":{"kind":"TRIVIAL_REBASE","_number":33,"created":"2026-08-31 12:33:18.000000000","uploader":{"_account_id":34452,"name":"Joan Gilabert","display_name":"jgilaber","email":"jgilaber@redhat.com","username":"jgilaber"},"ref":"refs/changes/64/997364/33","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cyborg","ref":"refs/changes/64/997364/33","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/64/997364/33 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/64/997364/33 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/64/997364/33 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cyborg refs/changes/64/997364/33"}}},"commit":{"parents":[{"commit":"35b64808a231a0e74d86ecf46ef64af06c707b52","subject":"Add mdev support to pci-sim DevStack plugin","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/35b64808a231a0e74d86ecf46ef64af06c707b52"}]}],"author":{"name":"jgilaber","email":"jgilaber@redhat.com","date":"2026-07-15 10:24:41.000000000","tz":120},"committer":{"name":"jgilaber","email":"jgilaber@redhat.com","date":"2026-08-31 12:33:06.000000000","tz":120},"subject":"[WIP] Configure mdev driver lifecycle tests","message":"[WIP] Configure mdev driver lifecycle tests\n\nConfigure tempest in the devstack plugin to add the cyborg_mdev section\npopulated with the appropriate values to run the mdev tests.\n\nDepends-On: https://review.opendev.org/c/openstack/cyborg-tempest-plugin/+/997352\nDepends-On: https://review.opendev.org/c/openstack/nova/+/995840\nChange-Id: I55eb22a68d3e1582cbcc337b85481ac4aabff971\nSigned-off-by: jgilaber \u003cjgilaber@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/f83c43f60410c81e25830d2f9fe885ea76b25055"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/f83c43f60410c81e25830d2f9fe885ea76b25055"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"35b64808a231a0e74d86ecf46ef64af06c707b52","is_merged_in_target_branch":false,"change_id":"I591e9ee7480ad602cea09c57d866be654e5238de","change_number":997055,"patch_set_number":23,"change_status":"NEW"}],"branch":"refs/heads/master"},"3ec1a228de3719722ae39d8678395ddcaa17872e":{"kind":"TRIVIAL_REBASE","_number":34,"created":"2026-08-31 15:06:57.000000000","uploader":{"_account_id":34452,"name":"Joan Gilabert","display_name":"jgilaber","email":"jgilaber@redhat.com","username":"jgilaber"},"ref":"refs/changes/64/997364/34","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cyborg","ref":"refs/changes/64/997364/34","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/64/997364/34 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/64/997364/34 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/64/997364/34 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cyborg refs/changes/64/997364/34"}}},"commit":{"parents":[{"commit":"e088b3086ff4e3501692d111c8fc09dda7071916","subject":"Add mdev support to pci-sim DevStack plugin","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/e088b3086ff4e3501692d111c8fc09dda7071916"}]}],"author":{"name":"jgilaber","email":"jgilaber@redhat.com","date":"2026-07-15 10:24:41.000000000","tz":120},"committer":{"name":"jgilaber","email":"jgilaber@redhat.com","date":"2026-08-31 13:39:01.000000000","tz":120},"subject":"[WIP] Configure mdev driver lifecycle tests","message":"[WIP] Configure mdev driver lifecycle tests\n\nConfigure tempest in the devstack plugin to add the cyborg_mdev section\npopulated with the appropriate values to run the mdev tests.\n\nDepends-On: https://review.opendev.org/c/openstack/cyborg-tempest-plugin/+/997352\nDepends-On: https://review.opendev.org/c/openstack/nova/+/995840\nChange-Id: I55eb22a68d3e1582cbcc337b85481ac4aabff971\nSigned-off-by: jgilaber \u003cjgilaber@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/3ec1a228de3719722ae39d8678395ddcaa17872e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/3ec1a228de3719722ae39d8678395ddcaa17872e"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"e088b3086ff4e3501692d111c8fc09dda7071916","is_merged_in_target_branch":false,"change_id":"I591e9ee7480ad602cea09c57d866be654e5238de","change_number":997055,"patch_set_number":24,"change_status":"NEW"}],"branch":"refs/heads/master"},"9a648b211622e9685a8ec3a31a4dba2aa98a51a5":{"kind":"TRIVIAL_REBASE","_number":35,"created":"2026-09-01 09:52:29.000000000","uploader":{"_account_id":34452,"name":"Joan Gilabert","display_name":"jgilaber","email":"jgilaber@redhat.com","username":"jgilaber"},"ref":"refs/changes/64/997364/35","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cyborg","ref":"refs/changes/64/997364/35","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/64/997364/35 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/64/997364/35 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cyborg refs/changes/64/997364/35 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cyborg refs/changes/64/997364/35"}}},"commit":{"parents":[{"commit":"370fd24d01e9867a3dfe97b50eb7efcdd33475c1","subject":"Add mdev support to pci-sim DevStack plugin","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/370fd24d01e9867a3dfe97b50eb7efcdd33475c1"}]}],"author":{"name":"jgilaber","email":"jgilaber@redhat.com","date":"2026-07-15 10:24:41.000000000","tz":120},"committer":{"name":"jgilaber","email":"jgilaber@redhat.com","date":"2026-08-31 15:37:16.000000000","tz":120},"subject":"[WIP] Configure mdev driver lifecycle tests","message":"[WIP] Configure mdev driver lifecycle tests\n\nConfigure tempest in the devstack plugin to add the cyborg_mdev section\npopulated with the appropriate values to run the mdev tests.\n\nDepends-On: https://review.opendev.org/c/openstack/cyborg-tempest-plugin/+/997352\nDepends-On: https://review.opendev.org/c/openstack/nova/+/995840\nChange-Id: I55eb22a68d3e1582cbcc337b85481ac4aabff971\nSigned-off-by: jgilaber \u003cjgilaber@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/9a648b211622e9685a8ec3a31a4dba2aa98a51a5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg/commit/9a648b211622e9685a8ec3a31a4dba2aa98a51a5"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"370fd24d01e9867a3dfe97b50eb7efcdd33475c1","is_merged_in_target_branch":false,"change_id":"I591e9ee7480ad602cea09c57d866be654e5238de","change_number":997055,"patch_set_number":25,"change_status":"NEW"}],"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY"},{"label":"Workflow","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
