)]}'
{"id":"openstack%2Fdesignate-dashboard~664549","triplet_id":"openstack%2Fdesignate-dashboard~master~I943243704442ae78a0acf8985b1cebe993d97828","project":"openstack/designate-dashboard","branch":"master","hashtags":[],"change_id":"I943243704442ae78a0acf8985b1cebe993d97828","subject":"Sync Sphinx requirement","status":"MERGED","created":"2019-06-11 10:14:51.000000000","updated":"2019-07-16 10:09:01.000000000","submitted":"2019-07-16 10:09:01.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"664549-1563271741736-8d65fae4","meta_rev_id":"bcab82d43a6bc20d514c946b0947469e71819cba","_number":664549,"virtual_id_number":664549,"owner":{"_account_id":28743,"name":"huang liujie","email":"huang.liujie@99cloud.net","username":"ljhuang"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},{"value":2,"date":"2019-07-16 10:09:01.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},"all":[{"value":2,"date":"2019-07-16 09:56:09.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},"all":[{"value":1,"date":"2019-07-16 09:56:09.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-07-16 09:56:09.000000000","updated_by":{"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},"reviewer":{"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},"state":"REVIEWER"},{"updated":"2019-07-16 10:09:01.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"2b781c49faa734731db50fe74a0d9084700ea37c","author":{"_account_id":28743,"name":"huang liujie","email":"huang.liujie@99cloud.net","username":"ljhuang"},"date":"2019-06-11 10:14:51.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"321a5e9f192cdc42d2f766348c5f4137a69751be","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-11 11:19:50.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- test-release-openstack http://logs.openstack.org/49/664549/1/check/test-release-openstack/41b62db/ : SUCCESS in 3m 38s\n- requirements-check http://logs.openstack.org/49/664549/1/check/requirements-check/f9b639f/ : SUCCESS in 3m 21s\n- nodejs10-npm-run-lint http://logs.openstack.org/49/664549/1/check/nodejs10-npm-run-lint/42965ff/ : SUCCESS in 5m 59s (non-voting)\n- nodejs10-npm-run-test http://logs.openstack.org/49/664549/1/check/nodejs10-npm-run-test/94af4c0/ : SUCCESS in 6m 40s (non-voting)\n- designate-bind9 http://logs.openstack.org/49/664549/1/check/designate-bind9/84ab206/ : SUCCESS in 41m 30s\n- designate-pdns4 http://logs.openstack.org/49/664549/1/check/designate-pdns4/569d236/ : SUCCESS in 38m 02s\n- designate-bind9-manager-model http://logs.openstack.org/49/664549/1/check/designate-bind9-manager-model/e962b14/ : SUCCESS in 38m 12s\n- designate-bind9-py27 http://logs.openstack.org/49/664549/1/check/designate-bind9-py27/094b491/ : SUCCESS in 38m 41s\n- designate-pdns4-py27 http://logs.openstack.org/49/664549/1/check/designate-pdns4-py27/ea9b42b/ : SUCCESS in 37m 48s\n- designate-pdns4-postgres http://logs.openstack.org/49/664549/1/check/designate-pdns4-postgres/b3a59d4/ : SUCCESS in 39m 55s\n- designate-grenade-pdns4 http://logs.openstack.org/49/664549/1/check/designate-grenade-pdns4/2852ebb/ : SUCCESS in 58m 39s\n- openstack-tox-cover http://logs.openstack.org/49/664549/1/check/openstack-tox-cover/fe6a112/ : FAILURE in 5m 42s\n- openstack-tox-lower-constraints http://logs.openstack.org/49/664549/1/check/openstack-tox-lower-constraints/4da10b1/ : SUCCESS in 5m 00s\n- openstack-tox-pep8 http://logs.openstack.org/49/664549/1/check/openstack-tox-pep8/4b11f57/ : SUCCESS in 5m 28s\n- openstack-tox-py27 http://logs.openstack.org/49/664549/1/check/openstack-tox-py27/bcfb19b/ : FAILURE in 5m 15s\n- openstack-tox-py36 http://logs.openstack.org/49/664549/1/check/openstack-tox-py36/baadb15/ : FAILURE in 5m 33s\n- openstack-tox-py37 http://logs.openstack.org/49/664549/1/check/openstack-tox-py37/e6ed4f6/ : FAILURE in 6m 07s\n- openstack-tox-docs http://logs.openstack.org/49/664549/1/check/openstack-tox-docs/2c3e08b/html/ : SUCCESS in 4m 37s\n- horizon-openstack-tox-python3-django111 http://logs.openstack.org/49/664549/1/check/horizon-openstack-tox-python3-django111/6634de8/ : FAILURE in 8m 39s","accounts_in_message":[],"_revision_number":1},{"id":"c46278b74f709403195e54532e836511a958d5d4","author":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"date":"2019-07-16 05:13:25.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"91274dca9bfc157c943d38d97e7630182b24372d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-16 05:25:09.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- test-release-openstack http://logs.openstack.org/49/664549/2/check/test-release-openstack/e1750ef/ : SUCCESS in 2m 38s\n- requirements-check http://logs.openstack.org/49/664549/2/check/requirements-check/8a479e1/ : SUCCESS in 1m 58s\n- nodejs10-npm-run-lint http://logs.openstack.org/49/664549/2/check/nodejs10-npm-run-lint/1c5e541/ : SUCCESS in 4m 22s (non-voting)\n- nodejs10-npm-run-test http://logs.openstack.org/49/664549/2/check/nodejs10-npm-run-test/94eee4d/ : SUCCESS in 6m 05s (non-voting)\n- openstack-tox-cover http://logs.openstack.org/49/664549/2/check/openstack-tox-cover/ad28417/cover/ : SUCCESS in 4m 14s\n- openstack-tox-lower-constraints http://logs.openstack.org/49/664549/2/check/openstack-tox-lower-constraints/96a9564/ : SUCCESS in 3m 33s\n- openstack-tox-pep8 http://logs.openstack.org/49/664549/2/check/openstack-tox-pep8/d2e5c7f/ : SUCCESS in 5m 22s\n- openstack-tox-py27 http://logs.openstack.org/49/664549/2/check/openstack-tox-py27/c26dd5a/ : SUCCESS in 3m 41s\n- openstack-tox-py36 http://logs.openstack.org/49/664549/2/check/openstack-tox-py36/37406f9/ : SUCCESS in 4m 07s\n- openstack-tox-py37 http://logs.openstack.org/49/664549/2/check/openstack-tox-py37/d133437/ : SUCCESS in 5m 13s\n- openstack-tox-docs http://logs.openstack.org/49/664549/2/check/openstack-tox-docs/87bce75/html/ : SUCCESS in 2m 55s\n- horizon-openstack-tox-python3-django111 http://logs.openstack.org/49/664549/2/check/horizon-openstack-tox-python3-django111/5c38ddf/ : SUCCESS in 4m 29s","accounts_in_message":[],"_revision_number":2},{"id":"4364ef8931d584d4ed3461f61712fe891f1f6f6e","author":{"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},"date":"2019-07-16 09:56:09.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"405ba1b3aacf0f5bc48d52e7a2d40d9a2dfff98d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-16 09:56:20.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"4fcd2aee681e5f360b29497cbc41b389f8c88d8b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-16 10:09:01.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"bcab82d43a6bc20d514c946b0947469e71819cba","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-16 10:09:01.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- test-release-openstack http://logs.openstack.org/49/664549/2/gate/test-release-openstack/0b7baeb/ : SUCCESS in 1m 52s\n- requirements-check http://logs.openstack.org/49/664549/2/gate/requirements-check/5fd0005/ : SUCCESS in 2m 38s\n- openstack-tox-lower-constraints http://logs.openstack.org/49/664549/2/gate/openstack-tox-lower-constraints/7ca8c23/ : SUCCESS in 4m 23s\n- openstack-tox-pep8 http://logs.openstack.org/49/664549/2/gate/openstack-tox-pep8/65bb771/ : SUCCESS in 5m 20s\n- openstack-tox-py27 http://logs.openstack.org/49/664549/2/gate/openstack-tox-py27/2248c3a/ : SUCCESS in 4m 16s\n- openstack-tox-py36 http://logs.openstack.org/49/664549/2/gate/openstack-tox-py36/39d0bb2/ : SUCCESS in 4m 30s\n- openstack-tox-py37 http://logs.openstack.org/49/664549/2/gate/openstack-tox-py37/7c59d0b/ : SUCCESS in 6m 03s\n- openstack-tox-docs http://logs.openstack.org/49/664549/2/gate/openstack-tox-docs/03bd5bc/html/ : SUCCESS in 3m 40s\n- horizon-openstack-tox-python3-django111 http://logs.openstack.org/49/664549/2/gate/horizon-openstack-tox-python3-django111/7c4c30d/ : SUCCESS in 4m 38s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"9fecedf4dfd0faf52f56973bd631d75bb2d19ad3","revisions":{"bedcbb239e86e59af1ff5c497f8adf129c2d78d9":{"kind":"REWORK","_number":1,"created":"2019-06-11 10:14:51.000000000","uploader":{"_account_id":28743,"name":"huang liujie","email":"huang.liujie@99cloud.net","username":"ljhuang"},"ref":"refs/changes/49/664549/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate-dashboard","ref":"refs/changes/49/664549/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate-dashboard refs/changes/49/664549/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate-dashboard refs/changes/49/664549/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate-dashboard refs/changes/49/664549/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate-dashboard refs/changes/49/664549/1"}}},"commit":{"parents":[{"commit":"ff276852c6985910c4606e8eeb807328fd1f9126","subject":"Imported Translations from Zanata","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate-dashboard/commit/ff276852c6985910c4606e8eeb807328fd1f9126"}]}],"author":{"name":"ljhuang","email":"huang.liujie@99cloud.net","date":"2019-06-11 10:13:57.000000000","tz":0},"committer":{"name":"ljhuang","email":"huang.liujie@99cloud.net","date":"2019-06-11 10:14:42.000000000","tz":0},"subject":"Sync Sphinx requirement","message":"Sync Sphinx requirement\n\n1. Sync sphinx dependency with global requirements. It caps python 2 since\nsphinx 2.0 no longer supports Python 2.7.\n2. Remove unncessary \"\u003d\u003d\"\n\nChange-Id: I943243704442ae78a0acf8985b1cebe993d97828\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate-dashboard/commit/bedcbb239e86e59af1ff5c497f8adf129c2d78d9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate-dashboard/commit/bedcbb239e86e59af1ff5c497f8adf129c2d78d9"}]},"branch":"refs/heads/master"},"9fecedf4dfd0faf52f56973bd631d75bb2d19ad3":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2019-07-16 05:13:25.000000000","uploader":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"ref":"refs/changes/49/664549/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate-dashboard","ref":"refs/changes/49/664549/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate-dashboard refs/changes/49/664549/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate-dashboard refs/changes/49/664549/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate-dashboard refs/changes/49/664549/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate-dashboard refs/changes/49/664549/2"}}},"commit":{"parents":[{"commit":"d026fee8f72095d514a36288683e8a14b3d51882","subject":"Merge \"Imported Translations from Zanata\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate-dashboard/commit/d026fee8f72095d514a36288683e8a14b3d51882"}]}],"author":{"name":"ljhuang","email":"huang.liujie@99cloud.net","date":"2019-06-11 10:13:57.000000000","tz":0},"committer":{"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","date":"2019-07-16 05:13:25.000000000","tz":0},"subject":"Sync Sphinx requirement","message":"Sync Sphinx requirement\n\n1. Sync sphinx dependency with global requirements. It caps python 2 since\nsphinx 2.0 no longer supports Python 2.7.\n2. Remove unncessary \"\u003d\u003d\"\n\nChange-Id: I943243704442ae78a0acf8985b1cebe993d97828\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate-dashboard/commit/9fecedf4dfd0faf52f56973bd631d75bb2d19ad3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate-dashboard/commit/9fecedf4dfd0faf52f56973bd631d75bb2d19ad3"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
