)]}'
{"id":"openstack%2Fdesignate-dashboard~698908","triplet_id":"openstack%2Fdesignate-dashboard~stable%2Ftrain~Ib91f873e8f21b71a71b1684317db276b72ae8014","project":"openstack/designate-dashboard","branch":"stable/train","topic":"fix-list-zones-stable/train","hashtags":[],"change_id":"Ib91f873e8f21b71a71b1684317db276b72ae8014","subject":"Fix list zones updated at same time","status":"MERGED","created":"2019-12-13 14:02:43.000000000","updated":"2020-01-03 22:28:17.000000000","submitted":"2020-01-03 22:27:07.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"698908-1578090427997-b2b5e395","meta_rev_id":"7682b844ab38a7cbd0eaac66845e06bdbec1b50c","_number":698908,"virtual_id_number":698908,"owner":{"_account_id":14892,"name":"Keith Berger","email":"keith.berger@suse.com","username":"kberger"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":31074,"name":"Ingo Fischer","email":"i.fischer@syseleven.de","username":"i.fischer"},{"value":0,"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},{"value":0,"date":"2020-01-03 22:11:18.000000000","_account_id":14892,"name":"Keith Berger","email":"keith.berger@suse.com","username":"kberger"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-01-03 22:27:07.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},"all":[{"value":0,"_account_id":31074,"name":"Ingo Fischer","email":"i.fischer@syseleven.de","username":"i.fischer"},{"value":2,"date":"2019-12-16 12:27:16.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},{"value":0,"_account_id":14892,"name":"Keith Berger","email":"keith.berger@suse.com","username":"kberger"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2019-12-22 20:22:28.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"all":[{"value":0,"_account_id":31074,"name":"Ingo Fischer","email":"i.fischer@syseleven.de","username":"i.fischer"},{"value":0,"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},{"value":0,"_account_id":14892,"name":"Keith Berger","email":"keith.berger@suse.com","username":"kberger"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2019-12-22 20:22:28.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"date":"2019-12-13 14:02:43.000000000","_account_id":31074,"name":"Ingo Fischer","email":"i.fischer@syseleven.de","username":"i.fischer"},{"value":0,"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},{"value":0,"_account_id":14892,"name":"Keith Berger","email":"keith.berger@suse.com","username":"kberger"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},{"_account_id":14892,"name":"Keith Berger","email":"keith.berger@suse.com","username":"kberger"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},{"_account_id":31074,"name":"Ingo Fischer","email":"i.fischer@syseleven.de","username":"i.fischer"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-12-13 14:02:43.000000000","updated_by":{"_account_id":31074,"name":"Ingo Fischer","email":"i.fischer@syseleven.de","username":"i.fischer"},"reviewer":{"_account_id":31074,"name":"Ingo Fischer","email":"i.fischer@syseleven.de","username":"i.fischer"},"state":"REVIEWER"},{"updated":"2019-12-16 12:27:16.000000000","updated_by":{"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},"reviewer":{"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},"state":"REVIEWER"},{"updated":"2019-12-22 20:22:28.000000000","updated_by":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"reviewer":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"state":"REVIEWER"},{"updated":"2020-01-03 22:27:07.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"d4ade5e6250f4ed7996af7d4c945924f3be16c3b","author":{"_account_id":14892,"name":"Keith Berger","email":"keith.berger@suse.com","username":"kberger"},"date":"2019-12-13 14:02:43.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"25227ec9d2e49b30976c867c7042e756c4eb99f7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-13 14:20:42.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- nodejs10-npm-run-lint https://zuul.opendev.org/t/openstack/build/32645555cc9344708030b9480be8aa1a : SUCCESS in 5m 37s\n- nodejs10-npm-run-test https://zuul.opendev.org/t/openstack/build/9679669096f5434fb3a0496c9fbabc09 : SUCCESS in 6m 34s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/8e512c043f6546dd8e3730d180751b89 : SUCCESS in 6m 01s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/e1d19d4f8e09455a9bf80e43c6add28d : SUCCESS in 7m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2f80730e53e74d8282ee812d71c388cd : SUCCESS in 6m 19s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/4331d68c5c7f4ce6a7b08e15eda09282 : SUCCESS in 10m 21s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ea708e8d32e34bcb979cfd57dd1965ce : SUCCESS in 5m 23s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/7040360a0ba047f0b037128e690653db : SUCCESS in 6m 50s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bf561c342db84e2f8e5ef0dba9861709 : SUCCESS in 4m 05s\n- horizon-openstack-tox-python3-django111 https://zuul.opendev.org/t/openstack/build/3cc5237c526f4264a7cbc93000684c42 : SUCCESS in 4m 45s","accounts_in_message":[],"_revision_number":1},{"id":"0ea5c612a777c4f3e554d0f49771bdcd0f2740e5","author":{"_account_id":14892,"name":"Keith Berger","email":"keith.berger@suse.com","username":"kberger"},"date":"2019-12-13 14:27:05.000000000","message":"Patch Set 1: Cherry Picked\n\nThis patchset was cherry picked to branch stable/stein as commit a3a7a0e73b2640d3218c66da2fa19527fbbe0401","accounts_in_message":[],"_revision_number":1},{"id":"0cfa9e63c0a6ab9075058fb421ad82db67196db7","author":{"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},"date":"2019-12-16 12:27:16.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"1f0319802fd4a356688b1bc3300379e892e94ac9","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2019-12-22 20:22:28.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"906d4ab13e8e4a4a3e8016f00b922b7cd925ec3d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-22 20:22:39.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"acb55d9235a6df2e0325b232203f2780ec50eb86","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-22 20:41:48.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- nodejs10-npm-run-lint https://zuul.opendev.org/t/openstack/build/1feb9f04b1ca4b68a51af4dfe11454c8 : RETRY_LIMIT in 3m 57s\n- nodejs10-npm-run-test https://zuul.opendev.org/t/openstack/build/8c592a150969479d87fc5d2ec352def1 : RETRY_LIMIT in 4m 30s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/8cac4ae108a2416e89d2c3b7cb1574ff : SUCCESS in 3m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7f49e1a1599642238b87b20d09797525 : FAILURE in 3m 04s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/497b5d9ee721494b832ea7bb8f5e88c4 : FAILURE in 3m 02s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/07c1acef9f40422aa648323d8362f6df : FAILURE in 3m 01s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/872249fca64f482a86843513e63742bb : FAILURE in 3m 10s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7b35400d7b58481ba2cc96082785b736 : FAILURE in 3m 17s\n- horizon-openstack-tox-python3-django111 https://zuul.opendev.org/t/openstack/build/b625a42387c24a58a6c0ad21ce6a1df0 : FAILURE in 3m 03s","accounts_in_message":[],"_revision_number":1},{"id":"8e0968c2caaf64d57fe9bcd884846bd3d18a4269","author":{"_account_id":14892,"name":"Keith Berger","email":"keith.berger@suse.com","username":"kberger"},"date":"2019-12-23 15:07:55.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"5432ec1f4500249818a61875c5c849a248125bf2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-23 15:51:03.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- nodejs10-npm-run-lint https://zuul.opendev.org/t/openstack/build/271ec0ae87694d86ad02ede74966c7f9 : RETRY_LIMIT in 4m 12s\n- nodejs10-npm-run-test https://zuul.opendev.org/t/openstack/build/dba2d45b79e542879565c34486fc1bc6 : RETRY_LIMIT in 4m 21s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/975adb2620bc4a7b8568f0422f7d2e29 : FAILURE in 3m 38s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/7bab4988dd544c0697f5b8e9610b6f94 : SUCCESS in 3m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4d554f7077ec4ccfb46d6c6083e66c0b : FAILURE in 3m 10s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/f469962e44364db087165a77ecbaa75f : FAILURE in 3m 42s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4a5fef273d3c4655900452c53ad3171d : FAILURE in 4m 25s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/abde6b3006b444bb9a286138d0c7ec36 : FAILURE in 3m 40s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4b3927c628eb48b7852f53e552583ec7 : FAILURE in 4m 21s\n- horizon-openstack-tox-python3-django111 https://zuul.opendev.org/t/openstack/build/341ccd29388041e1a775b1ad9adee8c2 : FAILURE in 3m 20s","accounts_in_message":[],"_revision_number":1},{"id":"4aa7c14545207578ae8dee8d96039e9abc619ac4","author":{"_account_id":14892,"name":"Keith Berger","email":"keith.berger@suse.com","username":"kberger"},"date":"2020-01-03 14:36:45.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"bcbba0fb603641fb8f87d6b81699ba04960f62ec","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-03 14:59:59.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- nodejs10-npm-run-lint https://zuul.opendev.org/t/openstack/build/9d1e59d608284e309a858eff7bc63873 : RETRY_LIMIT in 4m 07s\n- nodejs10-npm-run-test https://zuul.opendev.org/t/openstack/build/c9a713ad03364683a3b284c7e5691a58 : RETRY_LIMIT in 4m 23s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ccee9697dd6b4672ab77addc921023e8 : FAILURE in 3m 04s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/182fa0cfcdb141cd8b0797658b9c2d32 : SUCCESS in 3m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/178e5890d70b42c8aeace5a448c2ab3a : FAILURE in 2m 58s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/b4d11d7b1b634b12869c9f3972f8bcb4 : FAILURE in 3m 48s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8010a1fbda9f4cb5b52d9b3bc58a94ef : FAILURE in 3m 04s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/4306dd65651444bdb0024bcdc9b8bd9e : FAILURE in 3m 54s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fbe1ab9829544f9c8026ed705b055aa6 : FAILURE in 3m 26s\n- horizon-openstack-tox-python3-django111 https://zuul.opendev.org/t/openstack/build/11d1b0e169b04375ba78a4a0a6b512b7 : FAILURE in 3m 42s","accounts_in_message":[],"_revision_number":1},{"id":"a95ff897ce4d0786fec94d1f00de8fd511664edf","author":{"_account_id":14892,"name":"Keith Berger","email":"keith.berger@suse.com","username":"kberger"},"date":"2020-01-03 22:11:18.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"36f76c438f752293934c73986cc08ff1903d3213","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-03 22:18:25.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- nodejs10-npm-run-lint https://zuul.opendev.org/t/openstack/build/b0dc1ecd614c423aa53350ec56669ece : SUCCESS in 4m 39s\n- nodejs10-npm-run-test https://zuul.opendev.org/t/openstack/build/b31824177def40ae830a18613fe2f3d0 : SUCCESS in 6m 17s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/85285e6c930f4f82b8342a3b082b4ef2 : SUCCESS in 4m 25s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/8f4dab7f34db404abcbdb5c94a888389 : SUCCESS in 3m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ed2008e9368141bd8af9f15774ccf574 : SUCCESS in 3m 54s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/075f897756f64c0b8b7b55ece330db58 : SUCCESS in 3m 44s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/47308e7ddd2a45c784cf580a74611eec : SUCCESS in 4m 12s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/dcd36f59ffa04790ba2764f0338ac9e5 : SUCCESS in 4m 45s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2193f10aec8b4650acc358fb09444e36 : SUCCESS in 3m 32s\n- horizon-openstack-tox-python3-django111 https://zuul.opendev.org/t/openstack/build/30b1d97a5344435e92798c2c2482d22d : SUCCESS in 4m 31s","accounts_in_message":[],"_revision_number":1},{"id":"837fad797c1476dc86c97545f04e9dab3a4fecb7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-03 22:18:36.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"1ec03315d48a7f9adf7751370d5f6ae1562b33b6","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-03 22:27:07.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"038c9c79e55a4445df60a511ed7ab4ad0991cc68","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-03 22:27:07.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- nodejs10-npm-run-lint https://zuul.opendev.org/t/openstack/build/d1b086053ee14e789ab979db35f232c5 : SUCCESS in 4m 10s\n- nodejs10-npm-run-test https://zuul.opendev.org/t/openstack/build/8aa16e233650490ea256e34e5c06e058 : SUCCESS in 5m 02s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/08681795d6b740e0ab9a1881eb8f87f6 : SUCCESS in 3m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/67afe7288dba44c3be9b04f759ce6e8a : SUCCESS in 3m 40s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/04acfc483d434b76bcd66237fd833185 : SUCCESS in 3m 59s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c6ae4f2cf0f44bd28403894998b25c32 : SUCCESS in 3m 31s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/5fb365c1baf84216b8a63387f5a98cec : SUCCESS in 4m 02s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/deb509d5313e4839b92b6c30407aedd9 : SUCCESS in 3m 20s\n- horizon-openstack-tox-python3-django111 https://zuul.opendev.org/t/openstack/build/9c8b078f19de49b0bd1e7467983a88b5 : SUCCESS in 4m 01s","accounts_in_message":[],"_revision_number":1},{"id":"60bed4ba8b4dfd9e742ea352dacd8066d0bf12de","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-03 22:28:17.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ddd3377121a04c7782daaca619c69ed6 : SUCCESS in 54s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"51a44589c2f59fca73ed8b1f3933c5d6cffae50c","revisions":{"51a44589c2f59fca73ed8b1f3933c5d6cffae50c":{"kind":"REWORK","_number":1,"created":"2019-12-13 14:02:43.000000000","uploader":{"_account_id":14892,"name":"Keith Berger","email":"keith.berger@suse.com","username":"kberger"},"ref":"refs/changes/08/698908/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate-dashboard","ref":"refs/changes/08/698908/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate-dashboard refs/changes/08/698908/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate-dashboard refs/changes/08/698908/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate-dashboard refs/changes/08/698908/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate-dashboard refs/changes/08/698908/1"}}},"commit":{"parents":[{"commit":"f0e33a7de4415ea91ea837d721960434a893145a","subject":"Update TOX/UPPER_CONSTRAINTS_FILE for stable/train","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate-dashboard/commit/f0e33a7de4415ea91ea837d721960434a893145a"}]}],"author":{"name":"Ingo Fischer","email":"i.fischer@syseleven.de","date":"2019-10-09 09:37:54.000000000","tz":120},"committer":{"name":"Keith Berger","email":"keith.berger@suse.com","date":"2019-12-13 14:02:43.000000000","tz":0},"subject":"Fix list zones updated at same time","message":"Fix list zones updated at same time\n\nIn case multiple zones have the same update timestamp (updated_at) the\nUI did not show any zone and a JS console listed an error\n\"Error: [ngRepeat:dupes] Duplicates in a repeater are not allowed.\"\nTo fix it addTimestampIds needs to get the correct parameter order,\ni.e. pass the name of the ID column before the timestamp column name.\n\nCloses-Bug: 1734601\n\nChange-Id: Ib91f873e8f21b71a71b1684317db276b72ae8014\n(cherry picked from commit c6d53e4f8a36046c55e08272bf3859d5cb4c7c44)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate-dashboard/commit/51a44589c2f59fca73ed8b1f3933c5d6cffae50c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate-dashboard/commit/51a44589c2f59fca73ed8b1f3933c5d6cffae50c"}]},"branch":"refs/heads/stable/train"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
