)]}'
{"id":"openstack%2Fdesignate-dashboard~699241","triplet_id":"openstack%2Fdesignate-dashboard~stable%2Frocky~I6b027d1d1f235cf35e0ebb6876b0962a024fa460","project":"openstack/designate-dashboard","branch":"stable/rocky","hashtags":[],"change_id":"I6b027d1d1f235cf35e0ebb6876b0962a024fa460","subject":"Limit npm jobs to nodeset: ubuntu-xenial","status":"ABANDONED","created":"2019-12-16 16:36:02.000000000","updated":"2021-11-03 19:04:07.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"3a7ed495a4b92833332db4eb58ff4ae236460af8","_number":699241,"virtual_id_number":699241,"owner":{"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2019-12-16 17:58:11.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"date":"2019-12-22 08:14:09.000000000","_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"date":"2019-12-23 07:12:06.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-12-16 17:58:11.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-12-22 08:14:09.000000000","updated_by":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"reviewer":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"state":"REVIEWER"},{"updated":"2019-12-23 07:12:06.000000000","updated_by":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"reviewer":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"state":"REVIEWER"}],"messages":[{"id":"9a0427a7cd620c321685fb85ed1de515af229a5d","author":{"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},"date":"2019-12-16 16:36:02.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"c59ac3c345c14f5b1dfdd411e457665caac490b7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-16 17:58:11.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/24d4669bacb143469a13a102610a7a0e : SUCCESS in 46m 42s\n- designate-bind9-manager-model https://zuul.opendev.org/t/openstack/build/61f760ce4bed4f3cb1246be13c3c6adb : SUCCESS in 1h 00m 17s\n- designate-bind9-py35 https://zuul.opendev.org/t/openstack/build/ff454df1fa144a19b8a8272d6aa1242d : SUCCESS in 44m 39s\n- designate-bind9-py36 https://zuul.opendev.org/t/openstack/build/a4a11398dd024f83ba523559f2a395d8 : SUCCESS in 43m 26s\n- designate-devstack-pdns4 https://zuul.opendev.org/t/openstack/build/e86881510b804764afbaeb66c42d1117 : FAILURE in 38m 33s\n- designate-devstack-pdns4-postgres https://zuul.opendev.org/t/openstack/build/cf6a630985d94a958a5c7027eb46a154 : SUCCESS in 48m 49s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/2503a31fc6aa4043ba5ff9cd4ee13a9b : SUCCESS in 1h 04m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a3a7eb6c2945471bb1045fef84bcacc9 : SUCCESS in 6m 39s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/e060f7b69e854f4ea907f0c55971b832 : SUCCESS in 6m 46s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/0988b7adb43f4598bcbe8a7d1df5516c : SUCCESS in 6m 41s\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/7467f34b0d8a45fcb4d1f6b6b36535ea : SUCCESS in 4m 50s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c14bf87765414622bffdec2f546946a0 : SUCCESS in 5m 20s\n- horizon-openstack-tox-python3-django111 https://zuul.opendev.org/t/openstack/build/c106eb29bf164f43a02ea32b8cfeb83e : SUCCESS in 6m 42s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/fe908c400b8d46438136acda5282cb02 : FAILURE in 7m 08s (non-voting)\n- nodejs-npm-run-lint https://zuul.opendev.org/t/openstack/build/4fea5f3aa4a149e38bae9fb817dbeb86 : SUCCESS in 7m 40s (non-voting)\n- nodejs-npm-run-test https://zuul.opendev.org/t/openstack/build/dc4d3c5a6bab4c57ab145de85a84fbe2 : SUCCESS in 10m 21s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"b3ccc6bc653b716fb6e75dd8ed8dca0aef67a45c","author":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"date":"2019-12-20 07:43:53.000000000","message":"Patch Set 1: Code-Review+2\n\nrecheck devstack-pdns4 job","accounts_in_message":[],"_revision_number":1},{"id":"a11b1a06d0b5c6e814e2ffafa3c92a13042d472e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-20 08:42:56.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/3e95a6f3223e48078d5e5463bb4be18f : SUCCESS in 38m 51s\n- designate-bind9-manager-model https://zuul.opendev.org/t/openstack/build/03af7e47bb3f4cc894943afcda423344 : SUCCESS in 36m 20s\n- designate-bind9-py35 https://zuul.opendev.org/t/openstack/build/4964d05895094025becb2a74e8ebc244 : SUCCESS in 39m 03s\n- designate-bind9-py36 https://zuul.opendev.org/t/openstack/build/2324d4adceb14332ad0c519263e2b6f2 : SUCCESS in 56m 01s\n- designate-devstack-pdns4 https://zuul.opendev.org/t/openstack/build/7ba394b9f2f346f7a26e9ae340d42949 : SUCCESS in 38m 59s\n- designate-devstack-pdns4-postgres https://zuul.opendev.org/t/openstack/build/786366c8577e4cf38538691e85420ad3 : SUCCESS in 52m 26s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/b70141d399b84e9eacc96fd222fe7086 : SUCCESS in 47m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b64064e2c303450a8e90a97717acbe4b : FAILURE in 4m 40s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/7bd5525f32ae43f3ae59dde548c9ce34 : FAILURE in 4m 54s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/b4d0cd756480417a8f84c2e1187abe8d : FAILURE in 3m 58s\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/0d9ff7393abf48ccb4a50a8105d9fc14 : SUCCESS in 5m 39s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/2e6049cbac8d444e8f5706dd73edf5e3 : SUCCESS in 4m 54s\n- horizon-openstack-tox-python3-django111 https://zuul.opendev.org/t/openstack/build/ad0777fe5c3742169fc046714cca6310 : FAILURE in 3m 57s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/cec0ba1523d24c8e85f55031724d27b8 : FAILURE in 5m 38s (non-voting)\n- nodejs-npm-run-lint https://zuul.opendev.org/t/openstack/build/89a13af41a0942548cca67434fc3df00 : RETRY_LIMIT in 5m 26s (non-voting)\n- nodejs-npm-run-test https://zuul.opendev.org/t/openstack/build/a427e95c535146ababa1dbe573dc7ff7 : RETRY_LIMIT in 4m 27s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"9aaf7a995dbe627123c6767c1f16fa79502679e2","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2019-12-22 08:14:09.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"d8efa6c6f4dd25193dbdb5710242bf376d2dc338","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-22 09:21:31.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/6a30c750b14c4c369ae8a9a855451fd8 : SUCCESS in 40m 48s\n- designate-bind9-manager-model https://zuul.opendev.org/t/openstack/build/1c5bc1b3b23848f3912eb2c262a04061 : SUCCESS in 45m 34s\n- designate-bind9-py35 https://zuul.opendev.org/t/openstack/build/1fe12b3497f54a48a4c1d663d298143f : SUCCESS in 40m 40s\n- designate-bind9-py36 https://zuul.opendev.org/t/openstack/build/29c804b0cdca4707a263c063cb0f6e41 : SUCCESS in 42m 10s\n- designate-devstack-pdns4 https://zuul.opendev.org/t/openstack/build/c07ce73515f84dc483f79fed8c1968c4 : SUCCESS in 35m 52s\n- designate-devstack-pdns4-postgres https://zuul.opendev.org/t/openstack/build/834c4476f48f4122b7bf4c35292bf82b : SUCCESS in 34m 35s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/b8a49d32206d4c7c92ee77b80b14358d : SUCCESS in 1h 00m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8ae32b177dc94fd7bce49eccef4850dc : FAILURE in 3m 11s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/db531a1d58e84e4ba53131d3512de8c0 : FAILURE in 3m 37s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/6b2cd8d3de434934a8988280665f0f17 : FAILURE in 3m 14s\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/f9b1b5907a5b47539b8bf495e252473c : SUCCESS in 4m 06s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/92d900c2df6b4937a09a8c87df815da1 : SUCCESS in 4m 06s\n- horizon-openstack-tox-python3-django111 https://zuul.opendev.org/t/openstack/build/68d81e59f2b74f20b3999fb19c390569 : FAILURE in 3m 45s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c27acab851df464a91766819aedd2c94 : FAILURE in 3m 53s (non-voting)\n- nodejs-npm-run-lint https://zuul.opendev.org/t/openstack/build/c329b1546c1c44f08f105eeaefc3bb1d : RETRY_LIMIT in 4m 43s (non-voting)\n- nodejs-npm-run-test https://zuul.opendev.org/t/openstack/build/e46ccfd9db9e47e5a6760ef04eafb43f : RETRY_LIMIT in 6m 25s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"660acca7048bf0114565c343ff11fe187b4de92a","author":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"date":"2019-12-22 14:42:45.000000000","message":"Patch Set 1: -Code-Review\n\nIt seems all failures happened because they try to install horizon ussuri.\nhorizon rocky is 14.0.z. Perhaps horizon\u003c15 should work.","accounts_in_message":[],"_revision_number":1},{"id":"0aafa990e1f7e965ae84cb085d1d37cfd96571f1","author":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"date":"2019-12-22 14:48:34.000000000","message":"Patch Set 1:\n\nhttps://zuul.opendev.org/t/openstack/build/6b2cd8d3de434934a8988280665f0f17/log/tox/py35-1.log#103","accounts_in_message":[],"_revision_number":1},{"id":"feb7fff82ca9b90a88fd0d134e2def809026e6b2","author":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"date":"2019-12-22 14:49:49.000000000","message":"Patch Set 1:\n\nhorizon is specified in required-projects in zuul.yaml, but the failure happened when creating a tox env before processing th required-projects. requirements in horizon ussuri and upper-constraints.txt conflict.","accounts_in_message":[],"_revision_number":1},{"id":"bab1abe91936263ce430a0f17d4b9f5648cc1806","author":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"date":"2019-12-23 07:12:06.000000000","message":"Patch Set 1:\n\nRegarding the failure in tox env creation, I proposed a patch to have horizon in upper-constraints.txt [1]. Let\u0027s see the discussion in the review.\n\n[1] https://review.opendev.org/700378","accounts_in_message":[],"_revision_number":1},{"id":"3a7ed495a4b92833332db4eb58ff4ae236460af8","tag":"autogenerated:gerrit:abandon","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2021-11-03 19:04:07.000000000","message":"Abandoned\n\nRocky is in EM now, abandoning this.","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"f2f95aafd865c236a10d435c3f47adfdc531158b","revisions":{"f2f95aafd865c236a10d435c3f47adfdc531158b":{"kind":"REWORK","_number":1,"created":"2019-12-16 16:36:02.000000000","uploader":{"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},"ref":"refs/changes/41/699241/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate-dashboard","ref":"refs/changes/41/699241/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate-dashboard refs/changes/41/699241/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate-dashboard refs/changes/41/699241/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate-dashboard refs/changes/41/699241/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate-dashboard refs/changes/41/699241/1"}}},"commit":{"parents":[{"commit":"d6dfda1b5b35d6425f825f0e5f55904225e8d48e","subject":"Imported Translations from Zanata","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate-dashboard/commit/d6dfda1b5b35d6425f825f0e5f55904225e8d48e"}]}],"author":{"name":"Graham Hayes","email":"gr@ham.ie","date":"2019-12-16 16:35:39.000000000","tz":0},"committer":{"name":"Graham Hayes","email":"gr@ham.ie","date":"2019-12-16 16:36:00.000000000","tz":0},"subject":"Limit npm jobs to nodeset: ubuntu-xenial","message":"Limit npm jobs to nodeset: ubuntu-xenial\n\nChange-Id: I6b027d1d1f235cf35e0ebb6876b0962a024fa460\nSigned-off-by: Graham Hayes \u003cgr@ham.ie\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate-dashboard/commit/f2f95aafd865c236a10d435c3f47adfdc531158b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate-dashboard/commit/f2f95aafd865c236a10d435c3f47adfdc531158b"}]},"branch":"refs/heads/stable/rocky"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
