)]}'
{"id":"openstack%2Fdesignate-dashboard~730185","triplet_id":"openstack%2Fdesignate-dashboard~master~I3996d8d10e9e840fe8305536cc47e00effd229cf","project":"openstack/designate-dashboard","branch":"master","hashtags":[],"change_id":"I3996d8d10e9e840fe8305536cc47e00effd229cf","subject":"Fix hacking min version to 3.0.1","status":"ABANDONED","created":"2020-05-22 03:46:45.000000000","updated":"2020-05-22 06:49:39.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"dec5cb47a31d8c001dded365ff397ba03280720e","_number":730185,"virtual_id_number":730185,"owner":{"_account_id":30384,"name":"zhangboye","email":"zhangby09@chinatelecom.cn","username":"zhangboye"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},{"tag":"autogenerated:zuul:check","value":1,"date":"2020-05-22 03:57:29.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"rejected":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"all":[{"value":-2,"date":"2020-05-22 06:47:44.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-05-22 03:57:29.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-05-22 06:47:44.000000000","updated_by":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"reviewer":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"state":"REVIEWER"}],"messages":[{"id":"9147a95b3a4debabccc799da3ca61448f47774d6","author":{"_account_id":30384,"name":"zhangboye","email":"zhangby09@chinatelecom.cn","username":"zhangboye"},"date":"2020-05-22 03:46:45.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"6b1ebbe3cccd857ee6cba5f4f93f88ee39028645","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-22 03:57:29.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/6417fb4256ba4ce2b5ee726d9b542663 : SUCCESS in 2m 19s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1dd2dc567c9b4856b7abb78f45638019 : SUCCESS in 6m 53s\n- horizon-nodejs10-run-lint https://zuul.opendev.org/t/openstack/build/fab792680dfb4741885bcde135dcc579 : SUCCESS in 4m 17s\n- horizon-nodejs10-run-test https://zuul.opendev.org/t/openstack/build/6487d4d4f4694e1d932b247a673df2f8 : SUCCESS in 6m 48s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/39cd0bf6d51945ad9107aad6c511005d : SUCCESS in 5m 08s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c82d85870c7549a484ae4f851eb22564 : SUCCESS in 4m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0b7a6182bebc4df7b686e57c7ff7d4dd : SUCCESS in 4m 56s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6f938a8deb83491b82c08482a3c98d2c : SUCCESS in 7m 48s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2eba19fac99f4388b4d80f74b664e213 : SUCCESS in 6m 04s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a7c0627127164e47bbeb0e9e6dc9e0b5 : SUCCESS in 4m 04s","accounts_in_message":[],"_revision_number":1},{"id":"4d0f880e4aa09eb3763dde18d7b541853cd9809d","author":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"date":"2020-05-22 06:47:44.000000000","message":"Patch Set 1: Code-Review-2\n\nIt is duplicate of https://review.opendev.org/#/c/728540/","accounts_in_message":[],"_revision_number":1},{"id":"dec5cb47a31d8c001dded365ff397ba03280720e","author":{"_account_id":30384,"name":"zhangboye","email":"zhangby09@chinatelecom.cn","username":"zhangboye"},"date":"2020-05-22 06:49:39.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"31a1b5faedcab9bf9d3e3ac9735fbfe565636919","revisions":{"31a1b5faedcab9bf9d3e3ac9735fbfe565636919":{"kind":"REWORK","_number":1,"created":"2020-05-22 03:46:45.000000000","uploader":{"_account_id":30384,"name":"zhangboye","email":"zhangby09@chinatelecom.cn","username":"zhangboye"},"ref":"refs/changes/85/730185/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate-dashboard","ref":"refs/changes/85/730185/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate-dashboard refs/changes/85/730185/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate-dashboard refs/changes/85/730185/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate-dashboard refs/changes/85/730185/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate-dashboard refs/changes/85/730185/1"}}},"commit":{"parents":[{"commit":"db12a8954eb1f081e403b8b004bc67f891eb669c","subject":"Switch to newer openstackdocstheme and reno versions","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate-dashboard/commit/db12a8954eb1f081e403b8b004bc67f891eb669c"}]}],"author":{"name":"zhangboye","email":"zhangboye@inspur.com","date":"2020-05-22 03:46:13.000000000","tz":480},"committer":{"name":"zhangboye","email":"zhangboye@inspur.com","date":"2020-05-22 03:46:38.000000000","tz":480},"subject":"Fix hacking min version to 3.0.1","message":"Fix hacking min version to 3.0.1\n\nChange-Id: I3996d8d10e9e840fe8305536cc47e00effd229cf\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate-dashboard/commit/31a1b5faedcab9bf9d3e3ac9735fbfe565636919"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate-dashboard/commit/31a1b5faedcab9bf9d3e3ac9735fbfe565636919"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
