)]}'
{"id":"openstack%2Fdesignate-dashboard~779124","triplet_id":"openstack%2Fdesignate-dashboard~stable%2Fvictoria~I6bd42b175e74c9015db803eaaff09d908fae6a10","project":"openstack/designate-dashboard","branch":"stable/victoria","topic":"temp-job-disable","hashtags":[],"change_id":"I6bd42b175e74c9015db803eaaff09d908fae6a10","subject":"Fix lower-constrains job","status":"ABANDONED","created":"2021-03-08 03:25:56.000000000","updated":"2021-03-08 04:49:37.000000000","total_comment_count":2,"unresolved_comment_count":2,"has_review_started":true,"meta_rev_id":"71610fe67821771cb0864161866c63a36d0d998f","_number":779124,"virtual_id_number":779124,"owner":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},{"tag":"autogenerated:zuul:check","value":1,"date":"2021-03-08 03:34:54.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"disliked":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"all":[{"value":-1,"date":"2021-03-08 03:43:25.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":-1,"default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-03-08 03:25:56.000000000","updated_by":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"reviewer":{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},"state":"REVIEWER"},{"updated":"2021-03-08 03:34:54.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-03-08 03:43:25.000000000","updated_by":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"reviewer":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"state":"REVIEWER"}],"messages":[{"id":"48e7b17b9e48f5b60687975b1f9311e44059f275","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"date":"2021-03-08 03:25:56.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"2d43538a9b1597a0a86bebf2b9feb5a6e404f834","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-08 03:34:54.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/dd8f73a510734d4b90fade7d88893bf9 : SUCCESS in 2m 14s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/fa876baa74b24b94b3680f9d7a564907 : SUCCESS in 4m 05s\n- horizon-nodejs10-run-lint https://zuul.opendev.org/t/openstack/build/41f06db3361c41f08ea64b42e3341a81 : SUCCESS in 5m 18s\n- horizon-nodejs10-run-test https://zuul.opendev.org/t/openstack/build/f9833dde31074ec3bcd44217c6e572ba : SUCCESS in 7m 45s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e72266d4b3da4b7d9768155720d83463 : SUCCESS in 4m 15s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/94822dcdca6748fe9e7fba4491f93043 : SUCCESS in 3m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/84893912654d4f42b8e04338b4420975 : SUCCESS in 4m 10s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4ab204c2ccf7411ebaf4e8f6e37f9a91 : SUCCESS in 4m 18s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1d76a419649843cf9a1bcc7caafe7317 : SUCCESS in 5m 14s","accounts_in_message":[],"_revision_number":1},{"id":"349f9e542a0aa9e078cb294448fae619107af3c7","author":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"date":"2021-03-08 03:43:25.000000000","message":"Patch Set 1: Code-Review-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":1},{"id":"71610fe67821771cb0864161866c63a36d0d998f","tag":"autogenerated:gerrit:abandon","author":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"date":"2021-03-08 04:49:37.000000000","message":"Abandoned\n\nIt turns out that designate stable/victoria branch dropped lower-constraints job.\nIt looks better to follow it. https://review.opendev.org/c/openstack/designate-dashboard/+/779122 will drop the lower-constraints job and this change is no longer needed.","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"d52c77bc3be3c6b2c196fe3c8d350f97f730263e","revisions":{"d52c77bc3be3c6b2c196fe3c8d350f97f730263e":{"kind":"REWORK","_number":1,"created":"2021-03-08 03:25:56.000000000","uploader":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"ref":"refs/changes/24/779124/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate-dashboard","ref":"refs/changes/24/779124/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate-dashboard refs/changes/24/779124/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate-dashboard refs/changes/24/779124/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate-dashboard refs/changes/24/779124/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate-dashboard refs/changes/24/779124/1"}}},"commit":{"parents":[{"commit":"a20cdbfc21bd3c9d9c2dc38cd92e23d08c6bb782","subject":"Run npm nodejs job with Firefox browser","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate-dashboard/commit/a20cdbfc21bd3c9d9c2dc38cd92e23d08c6bb782"}]}],"author":{"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","date":"2021-02-03 13:37:32.000000000","tz":-420},"committer":{"name":"Akihiro Motoki","email":"amotoki@gmail.com","date":"2021-03-08 03:24:13.000000000","tz":540},"subject":"Fix lower-constrains job","message":"Fix lower-constrains job\n\nApplied the same fix as was done for designate.\n\nhacking in lower-constraints.txt was too old and we cannot pass\nthe pip resolver. Linters are not used in lower-constraints job,\nso we can safely drop them from lower-constraints.txt.\n\nChange-Id: I6bd42b175e74c9015db803eaaff09d908fae6a10\nSigned-off-by: Nicolas Bock \u003cnicolas.bock@canonical.com\u003e\n(cherry picked from commit 9cc36c1ecc83f29b3984cead7dd35d2355765a27)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate-dashboard/commit/d52c77bc3be3c6b2c196fe3c8d350f97f730263e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate-dashboard/commit/d52c77bc3be3c6b2c196fe3c8d350f97f730263e"}]},"branch":"refs/heads/stable/victoria"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
