)]}'
{"id":"openstack%2Fdesignate-dashboard~795271","triplet_id":"openstack%2Fdesignate-dashboard~master~Ie737df9fe400fa0292f60b69c789231a46c73db7","project":"openstack/designate-dashboard","branch":"master","topic":"record-set-add-fail-master","hashtags":[],"change_id":"Ie737df9fe400fa0292f60b69c789231a46c73db7","subject":"Add validation for \"record\" field","status":"ABANDONED","created":"2021-06-08 10:20:44.000000000","updated":"2022-09-24 05:46:47.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"cef54abfa9060b1978b34ebe92f234f17486fc4b","_number":795271,"virtual_id_number":795271,"owner":{"_account_id":27838,"name":"Vadym Markov","email":"vmarkov@mirantis.com","username":"vmarkov"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2021-06-08 11:22:45.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-06-08 10:34:10.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"a21887e1f62728925552dbf12024b9f8fabe3130","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27838,"name":"Vadym Markov","email":"vmarkov@mirantis.com","username":"vmarkov"},"date":"2021-06-08 10:20:44.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"5581c6d7336e9f7d7414ab1efd9006941dcb7fac","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-08 10:34:10.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2b5d231d5d5743c89813f0e86df21301 : SUCCESS in 5m 15s\n- horizon-nodejs10-run-lint https://zuul.opendev.org/t/openstack/build/8a09914ffd434f8da938ce4ab4da2822 : SUCCESS in 5m 20s\n- horizon-nodejs10-run-test https://zuul.opendev.org/t/openstack/build/d3c611f1858144fdaffd848d4ee704b0 : SUCCESS in 8m 07s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/3dceeb312bd04f5b8ede6b8aa09ba781 : SUCCESS in 5m 31s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/0ba2e9a3b8a74c869ffbfc1200c79f68 : SUCCESS in 4m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4921397528ef4846b66cb294329edba3 : FAILURE in 3m 23s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/86973429b89943a89a4ea71b6c286cc1 : SUCCESS in 5m 52s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/095d6ad44eb8412daa4552124e437a9b : SUCCESS in 5m 18s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a22b03671960446fa9a6d616570462eb : SUCCESS in 6m 19s","accounts_in_message":[],"_revision_number":1},{"id":"0574846fde8fc6a65265eaef1970a48fcee25bbf","author":{"_account_id":27838,"name":"Vadym Markov","email":"vmarkov@mirantis.com","username":"vmarkov"},"date":"2021-06-08 11:12:38.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"d80cde034664a99908491509ea13365be95691b3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-08 11:22:45.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a04884de0ff142cfb2e5c8308d9ac7f3 : SUCCESS in 4m 51s\n- horizon-nodejs10-run-lint https://zuul.opendev.org/t/openstack/build/0eea80fe10a048a9bfec71ac1e1b9b60 : SUCCESS in 5m 50s\n- horizon-nodejs10-run-test https://zuul.opendev.org/t/openstack/build/c15b9b11c93148d5863a38dc6b90634c : SUCCESS in 7m 49s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/44f3268c219f404f83af7fcd8ac88cf8 : SUCCESS in 4m 47s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/a4b377e7143f4e3aa941d82e9a5463de : SUCCESS in 4m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8a26f9465478416a819a1ec77588fe37 : SUCCESS in 4m 20s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4904656e110141a8a229d15ce69de761 : SUCCESS in 6m 43s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/989e75a1c7024c3e8953820080ae73a2 : SUCCESS in 5m 28s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0fc10d1ab7dc4363882bfbe43d0be4d5 : SUCCESS in 5m 22s","accounts_in_message":[],"_revision_number":1},{"id":"cef54abfa9060b1978b34ebe92f234f17486fc4b","tag":"autogenerated:gerrit:abandon","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2022-09-24 05:46:47.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"c2c9839a0fc83309033964b4056d20300611957f","revisions":{"c2c9839a0fc83309033964b4056d20300611957f":{"kind":"REWORK","_number":1,"created":"2021-06-08 10:20:44.000000000","uploader":{"_account_id":27838,"name":"Vadym Markov","email":"vmarkov@mirantis.com","username":"vmarkov"},"ref":"refs/changes/71/795271/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate-dashboard","ref":"refs/changes/71/795271/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate-dashboard refs/changes/71/795271/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate-dashboard refs/changes/71/795271/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate-dashboard refs/changes/71/795271/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate-dashboard refs/changes/71/795271/1"}}},"commit":{"parents":[{"commit":"d819d5619638824806cc129e0cb8dbeebf4b40c3","subject":"Merge \"Use TOX_CONSTRAINTS_FILE\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate-dashboard/commit/d819d5619638824806cc129e0cb8dbeebf4b40c3"}]}],"author":{"name":"Vadym Markov","email":"vmarkov@mirantis.com","date":"2021-05-25 16:44:08.000000000","tz":180},"committer":{"name":"Vadym Markov","email":"vmarkov@mirantis.com","date":"2021-06-08 10:17:25.000000000","tz":180},"subject":"Add validation for \"record\" field","message":"Add validation for \"record\" field\n\nCurrently, \"Create Record Set\" modal window allows uploading of empty\nrecords. Proposed patch adds some validation for this input.\n\nChange-Id: Ie737df9fe400fa0292f60b69c789231a46c73db7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate-dashboard/commit/c2c9839a0fc83309033964b4056d20300611957f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate-dashboard/commit/c2c9839a0fc83309033964b4056d20300611957f"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
