)]}'
{"id":"openstack%2Fdesignate-specs~648949","triplet_id":"openstack%2Fdesignate-specs~master~I8f24fd878d034ed240ef5574afaafe1b4adb113b","project":"openstack/designate-specs","branch":"master","hashtags":[],"change_id":"I8f24fd878d034ed240ef5574afaafe1b4adb113b","subject":"Update to openstackdocstheme + sphinx 2.0","status":"MERGED","created":"2019-04-01 11:26:13.000000000","updated":"2019-05-20 03:29:27.000000000","submitted":"2019-05-20 03:29:27.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"648949-1558322967661-354c625b","meta_rev_id":"c8fee03868d2bce0267f54e027302a768dbfa32c","_number":648949,"virtual_id_number":648949,"owner":{"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},{"value":2,"date":"2019-05-20 03:29:27.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"all":[{"value":0,"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2019-05-20 03:22:06.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"all":[{"value":0,"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2019-05-20 03:22:06.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"approved":{"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},"all":[{"value":2,"date":"2019-04-02 10:04:47.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-05-20 03:22:06.000000000","updated_by":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"reviewer":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"state":"REVIEWER"},{"updated":"2019-05-20 03:29:27.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"6c9852615905f1080617dcc969350debc7855508","author":{"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},"date":"2019-04-01 11:26:13.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"b469ed1abef13ce8762b62794eb488630e783dd9","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-01 11:38:15.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/49/648949/1/check/openstack-tox-docs/1aee0a5/html/ : SUCCESS in 4m 06s\n- openstack-tox-pep8 http://logs.openstack.org/49/648949/1/check/openstack-tox-pep8/e7fe096/ : SUCCESS in 5m 39s","accounts_in_message":[],"_revision_number":1},{"id":"4f7d7ffcf0156c7f0a5bf6cf30df68a51ad1c5e3","author":{"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},"date":"2019-04-02 10:04:47.000000000","message":"Patch Set 1: Review-Priority+2\n\nthis means we can build specs again :)","accounts_in_message":[],"_revision_number":1},{"id":"29cfa10d9a69ff869143a7353e070b3cc7dc5b72","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2019-05-20 03:22:06.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"05898c9f910402833b6bc3ec3ec1a406323958f2","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-20 03:22:17.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"1efc1ce1d844f70ba8313aa1d758cf1781650e11","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-20 03:29:27.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"c8fee03868d2bce0267f54e027302a768dbfa32c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-20 03:29:27.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/49/648949/1/gate/openstack-tox-docs/47c669c/html/ : SUCCESS in 6m 18s\n- openstack-tox-pep8 http://logs.openstack.org/49/648949/1/gate/openstack-tox-pep8/21b4367/ : SUCCESS in 6m 33s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"99d49ceb6906fe11853c2c4f340d0190b90a1b0e","revisions":{"99d49ceb6906fe11853c2c4f340d0190b90a1b0e":{"kind":"REWORK","_number":1,"created":"2019-04-01 11:26:13.000000000","uploader":{"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},"ref":"refs/changes/49/648949/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate-specs","ref":"refs/changes/49/648949/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate-specs refs/changes/49/648949/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate-specs refs/changes/49/648949/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate-specs refs/changes/49/648949/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate-specs refs/changes/49/648949/1"}}},"commit":{"parents":[{"commit":"e78e492c894bd11a2c1d84c7e666e4b87090f697","subject":"Change openstack-dev to openstack-discuss","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate-specs/commit/e78e492c894bd11a2c1d84c7e666e4b87090f697"}]}],"author":{"name":"Graham Hayes","email":"gr@ham.ie","date":"2019-04-01 11:26:06.000000000","tz":60},"committer":{"name":"Graham Hayes","email":"gr@ham.ie","date":"2019-04-01 11:26:06.000000000","tz":60},"subject":"Update to openstackdocstheme + sphinx 2.0","message":"Update to openstackdocstheme + sphinx 2.0\n\nChange-Id: I8f24fd878d034ed240ef5574afaafe1b4adb113b\nSigned-off-by: Graham Hayes \u003cgr@ham.ie\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate-specs/commit/99d49ceb6906fe11853c2c4f340d0190b90a1b0e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate-specs/commit/99d49ceb6906fe11853c2c4f340d0190b90a1b0e"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
