)]}'
{"id":"openstack%2Fdesignate-specs~970698","triplet_id":"openstack%2Fdesignate-specs~master~Ic70ff15eda200d98dad75c71f372da75a8e5035b","project":"openstack/designate-specs","branch":"master","attention_set":{},"removed_from_attention_set":{"31664":{"account":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"last_update":"2026-05-07 11:34:40.000000000","reason":"\u003cGERRIT_ACCOUNT_31664\u003e replied on the change","reason_account":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"}},"37598":{"account":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"last_update":"2026-05-07 11:42:04.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ic70ff15eda200d98dad75c71f372da75a8e5035b","subject":"Bump minimum tox version to 3.18.0","status":"MERGED","created":"2025-12-11 20:44:43.000000000","updated":"2026-05-07 11:42:57.000000000","submitted":"2026-05-07 11:42:04.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":8,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"970698","meta_rev_id":"262d7051122083bcec76fa1aa82ac3005c8608d0","_number":970698,"virtual_id_number":970698,"owner":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-05-07 11:42:04.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"all":[{"value":2,"date":"2026-05-07 11:34:40.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"all":[{"value":1,"date":"2026-05-07 11:34:40.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-12-11 20:57:21.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-05-07 08:29:56.000000000","updated_by":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"reviewer":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"state":"REVIEWER"}],"messages":[{"id":"05eb5051ee19b27a739e1d71e2a17adc2dfc3787","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2025-12-11 20:44:43.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"97aa3bdad726fba67e8e8f17ac71b7e86c8e7015","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2025-12-11 20:49:29.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"eecc8ba2bb462c961b47b98ceaaa55884505bbf2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-11 20:57:21.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/50993f17ad5f45dd83e81989ab0bdc7b\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6d70101a8a0f47ce9f67b09caa470974 : FAILURE in 2m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b964364ee6e34617b360234493e52516 : SUCCESS in 2m 34s","accounts_in_message":[],"_revision_number":2},{"id":"b4b90dc95fa9df53e24b308c9ac9581a5eb1af6c","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2025-12-11 21:03:10.000000000","message":"Patch Set 3: Published edit on patch set 2.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"c0340ab5815e32619b60136bd37a4321e678c4a5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-11 21:10:42.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/51a0852b83be4fb6b6cc88ec4851731d\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/28f83f32129c40c8a8303160b9a9a7a2 : SUCCESS in 2m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/350a3f15960a46a9bc357426045f4868 : SUCCESS in 2m 23s","accounts_in_message":[],"_revision_number":3},{"id":"3029dce1cfb132a1f3ee8b012a93526d37addf46","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2025-12-11 21:14:11.000000000","message":"Patch Set 4: Published edit on patch set 3.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":4},{"id":"0e6e31a60dd2c1fea78ec87ec8521faeb77d140f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-11 21:23:06.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/cdb607c39120484ab3f8f57d4587e511\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ce04790f8aa84d48a46df025bd7e84e4 : SUCCESS in 3m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1e2157373ca04d8f9f1c2e1ae97d6b71 : SUCCESS in 2m 37s","accounts_in_message":[],"_revision_number":4},{"id":"c0a06a36003487617f3644a5523cbc2cc6c084d5","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2025-12-11 21:24:16.000000000","message":"Patch Set 5: Published edit on patch set 4.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":5},{"id":"877b19021aa85a012a9e555d6c4cb4b041cbc939","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-11 21:27:46.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/06db6303b13f452e809b7458b4dd5258\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/849e605b8be147c5a0aaf1d3f226225b : FAILURE in 2m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ad9e137ce7ed4c54b4d2f261353da746 : SUCCESS in 1m 53s","accounts_in_message":[],"_revision_number":5},{"id":"7100ea61b5298977c9632ac8e725bb8c89413b6a","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2025-12-11 21:30:52.000000000","message":"Patch Set 6: Published edit on patch set 5.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":6},{"id":"3ba8dbd71169507e1b5e07518bb738db042803b8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-11 21:33:40.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/fe730f841a5b425c8ec0552e12b09d05\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/da8544e8c0e24412b58a556ff109b629 : FAILURE in 1m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/79be6ba1165541c18a303d1d7860dcd3 : SUCCESS in 1m 59s","accounts_in_message":[],"_revision_number":6},{"id":"6a6712ab155fcb9e41a5fe4da64b60bdb1a0f657","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2025-12-11 21:39:41.000000000","message":"Patch Set 7: Published edit on patch set 6.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":7},{"id":"848f82b6c96d6b39088cd2bfbcdeb308ae75a6e8","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2025-12-11 21:41:09.000000000","message":"Patch Set 8: Commit message was updated.","accounts_in_message":[],"_revision_number":8},{"id":"a0935edbdb0c27fb3416d0669b16ca215d164c50","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2025-12-11 21:43:49.000000000","message":"Patch Set 9: Published edit on patch set 8.","accounts_in_message":[],"_revision_number":9},{"id":"3904fd9a7e759f71bb49647428e501436d69dd21","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2025-12-11 21:44:09.000000000","message":"Patch Set 10: Published edit on patch set 9.","accounts_in_message":[],"_revision_number":10},{"id":"f25dbb7775ac32ac9982d4373b099cdc9f6d5587","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2025-12-11 21:44:15.000000000","message":"Patch Set 11: Commit message was updated.","accounts_in_message":[],"_revision_number":11},{"id":"d8112f8d4e19063c05af4bc01208b69b16352803","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2025-12-11 21:50:09.000000000","message":"Patch Set 12: Commit message was updated.","accounts_in_message":[],"_revision_number":12},{"id":"0cab07f4f5157b480a4f757baecd4592a04e6a2f","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2025-12-11 21:51:31.000000000","message":"Patch Set 13: Published edit on patch set 12.","accounts_in_message":[],"_revision_number":13},{"id":"32d5684b35b23f5c2c798845e997dbdd31cbb314","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2025-12-11 21:51:36.000000000","message":"Patch Set 13:\n\nThis change is ready for review.","accounts_in_message":[],"_revision_number":13},{"id":"0dd41c5b601ba00e73c2eb5d0e5b0e20e67df82b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-11 21:58:21.000000000","message":"Patch Set 13: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6bac080421584af8a6b6d4de358fc388\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/efb793a99b76408cac28449f241c4037 : SUCCESS in 2m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cf1a634f5d4b4652a6ca27c7653bb903 : SUCCESS in 2m 15s","accounts_in_message":[],"_revision_number":13},{"id":"d909746aaaf199e958d9c1ee634b6ead3994b284","author":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"date":"2026-05-07 08:29:56.000000000","message":"Patch Set 13: Code-Review-1\n\n(4 comments)","accounts_in_message":[],"_revision_number":13},{"id":"21186695b8f959c67738bf6673ec20b9531fa9f4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-05-07 10:14:20.000000000","message":"Patch Set 14: Commit message was updated.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":14},{"id":"ee93e1e3e2891aae87557518b91faa19c4d54938","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-05-07 10:15:32.000000000","message":"Patch Set 15: Published edit on patch set 14.","accounts_in_message":[],"_revision_number":15},{"id":"10dd8f539a3f2f240aa5a01ba884d2f13e0e5979","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-05-07 10:15:37.000000000","message":"Patch Set 15:\n\n(3 comments)","accounts_in_message":[],"_revision_number":15},{"id":"4ae88b26cd5e4bb9f876e5bbddc837947dcdfb35","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-07 10:22:43.000000000","message":"Patch Set 15: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/20dc5a8e11aa431190b4180e388898c7\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1e11d144f8d146088f59e45a28ef95e7 : SUCCESS in 3m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8889354f8a6c4cec9678c7fe6db5f189 : SUCCESS in 2m 25s","accounts_in_message":[],"_revision_number":15},{"id":"d20b019f9484b604398d550026d29dd6a6e260aa","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"date":"2026-05-07 10:57:50.000000000","message":"Patch Set 16: Commit message was updated.\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":16},{"id":"2da0ed88dcbf81978f125af3a96207d26a833261","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-07 11:07:19.000000000","message":"Patch Set 16: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c214f3cc56fc4922ac8c9ad9457aa458\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/70747d43b4f24d6aa70253ed4ea811be : SUCCESS in 3m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/dfa6d31f9d294a00b8e6143f0f8a6730 : SUCCESS in 2m 11s","accounts_in_message":[],"_revision_number":16},{"id":"e55fea164958ee6b7fc53929cc8927120a7b2a90","author":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"date":"2026-05-07 11:34:40.000000000","message":"Patch Set 16: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":16},{"id":"50f7bf1fa843308441ae31b850baff1df0759e12","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-07 11:34:52.000000000","message":"Patch Set 16: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":16},{"id":"7e1fd88bd2a9ac486b6b878ac890cdde5ee37c68","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-07 11:42:04.000000000","message":"Patch Set 16: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/edd1a510552c4ca3bfb680ede2a18bc0\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b556650e13ae45a6b73dd402214abb8f : SUCCESS in 4m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/10a33d61f70b4cc5a97731606ca2ebac : SUCCESS in 2m 44s","accounts_in_message":[],"_revision_number":16},{"id":"5c76508889fbb4380e011cb75d87b48541999431","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-07 11:42:04.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":16},{"id":"262d7051122083bcec76fa1aa82ac3005c8608d0","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-07 11:42:57.000000000","message":"Patch Set 16:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/dba19bdd081c40ff8b4f5bd1fd032ea3\n\n- promote-openstack-specs https://zuul.opendev.org/t/openstack/build/80083b6b690848a29df42c568545c8d3 : SUCCESS in 43s","accounts_in_message":[],"_revision_number":16}],"current_revision_number":16,"current_revision":"3d72e1cc0430e993bfe00c9c01a1385e96902001","revisions":{"2f783acc7976752337d1862db6b885caf2b2bebf":{"kind":"REWORK","_number":1,"created":"2025-12-11 20:44:43.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/98/970698/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate-specs","ref":"refs/changes/98/970698/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate-specs refs/changes/98/970698/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate-specs refs/changes/98/970698/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate-specs refs/changes/98/970698/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate-specs refs/changes/98/970698/1"}}},"commit":{"parents":[{"commit":"cbf85d69b2a627dbe98aad971635456f325426aa","subject":"Adding a spec for providing catalog zones","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate-specs/commit/cbf85d69b2a627dbe98aad971635456f325426aa"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2025-12-11 20:44:43.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2025-12-11 20:44:43.000000000","tz":0},"subject":"wip","message":"wip\n\nChange-Id: Ic70ff15eda200d98dad75c71f372da75a8e5035b\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate-specs/commit/2f783acc7976752337d1862db6b885caf2b2bebf"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate-specs/commit/2f783acc7976752337d1862db6b885caf2b2bebf"}]},"branch":"refs/heads/master"},"37f67ccb6fff346c7fbcedea857b653a5a4b8c8b":{"kind":"REWORK","_number":2,"created":"2025-12-11 20:49:29.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/98/970698/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate-specs","ref":"refs/changes/98/970698/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate-specs refs/changes/98/970698/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate-specs refs/changes/98/970698/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate-specs refs/changes/98/970698/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate-specs refs/changes/98/970698/2"}}},"commit":{"parents":[{"commit":"cbf85d69b2a627dbe98aad971635456f325426aa","subject":"Adding a spec for providing catalog zones","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate-specs/commit/cbf85d69b2a627dbe98aad971635456f325426aa"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2025-12-11 20:44:43.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2025-12-11 20:49:28.000000000","tz":0},"subject":"wip","message":"wip\n\nChange-Id: Ic70ff15eda200d98dad75c71f372da75a8e5035b\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate-specs/commit/37f67ccb6fff346c7fbcedea857b653a5a4b8c8b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate-specs/commit/37f67ccb6fff346c7fbcedea857b653a5a4b8c8b"}]},"branch":"refs/heads/master"},"ea09285085a461a41dcfb1e1dab20607da71003d":{"kind":"REWORK","_number":3,"created":"2025-12-11 21:03:10.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/98/970698/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate-specs","ref":"refs/changes/98/970698/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate-specs refs/changes/98/970698/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate-specs refs/changes/98/970698/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate-specs refs/changes/98/970698/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate-specs refs/changes/98/970698/3"}}},"commit":{"parents":[{"commit":"cbf85d69b2a627dbe98aad971635456f325426aa","subject":"Adding a spec for providing catalog zones","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate-specs/commit/cbf85d69b2a627dbe98aad971635456f325426aa"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2025-12-11 20:44:43.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2025-12-11 21:03:09.000000000","tz":0},"subject":"wip","message":"wip\n\nChange-Id: Ic70ff15eda200d98dad75c71f372da75a8e5035b\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate-specs/commit/ea09285085a461a41dcfb1e1dab20607da71003d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate-specs/commit/ea09285085a461a41dcfb1e1dab20607da71003d"}]},"branch":"refs/heads/master"},"55c114e63bba874d9389395e25467030ac6747f1":{"kind":"REWORK","_number":4,"created":"2025-12-11 21:14:11.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/98/970698/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate-specs","ref":"refs/changes/98/970698/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate-specs refs/changes/98/970698/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate-specs refs/changes/98/970698/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate-specs refs/changes/98/970698/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate-specs refs/changes/98/970698/4"}}},"commit":{"parents":[{"commit":"cbf85d69b2a627dbe98aad971635456f325426aa","subject":"Adding a spec for providing catalog zones","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate-specs/commit/cbf85d69b2a627dbe98aad971635456f325426aa"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2025-12-11 20:44:43.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2025-12-11 21:14:08.000000000","tz":0},"subject":"wip","message":"wip\n\nChange-Id: Ic70ff15eda200d98dad75c71f372da75a8e5035b\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate-specs/commit/55c114e63bba874d9389395e25467030ac6747f1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate-specs/commit/55c114e63bba874d9389395e25467030ac6747f1"}]},"branch":"refs/heads/master"},"78ae9425a4282cc587231e400e74ca8e24108034":{"kind":"REWORK","_number":5,"created":"2025-12-11 21:24:16.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/98/970698/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate-specs","ref":"refs/changes/98/970698/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate-specs refs/changes/98/970698/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate-specs refs/changes/98/970698/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate-specs refs/changes/98/970698/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate-specs refs/changes/98/970698/5"}}},"commit":{"parents":[{"commit":"cbf85d69b2a627dbe98aad971635456f325426aa","subject":"Adding a spec for providing catalog zones","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate-specs/commit/cbf85d69b2a627dbe98aad971635456f325426aa"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2025-12-11 20:44:43.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2025-12-11 21:24:16.000000000","tz":0},"subject":"wip","message":"wip\n\nChange-Id: Ic70ff15eda200d98dad75c71f372da75a8e5035b\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate-specs/commit/78ae9425a4282cc587231e400e74ca8e24108034"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate-specs/commit/78ae9425a4282cc587231e400e74ca8e24108034"}]},"branch":"refs/heads/master"},"6d2fb93e6c221a19b3c59c1d03feb34933b30a57":{"kind":"REWORK","_number":6,"created":"2025-12-11 21:30:52.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/98/970698/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate-specs","ref":"refs/changes/98/970698/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate-specs refs/changes/98/970698/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate-specs refs/changes/98/970698/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate-specs refs/changes/98/970698/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate-specs refs/changes/98/970698/6"}}},"commit":{"parents":[{"commit":"cbf85d69b2a627dbe98aad971635456f325426aa","subject":"Adding a spec for providing catalog zones","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate-specs/commit/cbf85d69b2a627dbe98aad971635456f325426aa"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2025-12-11 20:44:43.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2025-12-11 21:30:52.000000000","tz":0},"subject":"wip","message":"wip\n\nChange-Id: Ic70ff15eda200d98dad75c71f372da75a8e5035b\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate-specs/commit/6d2fb93e6c221a19b3c59c1d03feb34933b30a57"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate-specs/commit/6d2fb93e6c221a19b3c59c1d03feb34933b30a57"}]},"branch":"refs/heads/master"},"9e07fd63381a3b8c481369985a2ba54e3dad055d":{"kind":"REWORK","_number":7,"created":"2025-12-11 21:39:41.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/98/970698/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate-specs","ref":"refs/changes/98/970698/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate-specs refs/changes/98/970698/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate-specs refs/changes/98/970698/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate-specs refs/changes/98/970698/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate-specs refs/changes/98/970698/7"}}},"commit":{"parents":[{"commit":"cbf85d69b2a627dbe98aad971635456f325426aa","subject":"Adding a spec for providing catalog zones","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate-specs/commit/cbf85d69b2a627dbe98aad971635456f325426aa"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2025-12-11 20:44:43.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2025-12-11 21:39:40.000000000","tz":0},"subject":"wip","message":"wip\n\nChange-Id: Ic70ff15eda200d98dad75c71f372da75a8e5035b\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate-specs/commit/9e07fd63381a3b8c481369985a2ba54e3dad055d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate-specs/commit/9e07fd63381a3b8c481369985a2ba54e3dad055d"}]},"branch":"refs/heads/master"},"7efa39e05bb7d13bdb02488bea568a8b7fcb762e":{"kind":"NO_CODE_CHANGE","_number":8,"created":"2025-12-11 21:41:09.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/98/970698/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate-specs","ref":"refs/changes/98/970698/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate-specs refs/changes/98/970698/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate-specs refs/changes/98/970698/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate-specs refs/changes/98/970698/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate-specs refs/changes/98/970698/8"}}},"commit":{"parents":[{"commit":"cbf85d69b2a627dbe98aad971635456f325426aa","subject":"Adding a spec for providing catalog zones","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate-specs/commit/cbf85d69b2a627dbe98aad971635456f325426aa"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2025-12-11 20:44:43.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2025-12-11 21:41:09.000000000","tz":0},"subject":"Bump minimum tox version to 3.18.0","message":"Bump minimum tox version to 3.18.0\n\nfor fix\nhttps://zuul.opendev.org/t/openstack/build/e8db7eb566ed479db709b416db346dbb\n\nChange-Id: Ic70ff15eda200d98dad75c71f372da75a8e5035b\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate-specs/commit/7efa39e05bb7d13bdb02488bea568a8b7fcb762e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate-specs/commit/7efa39e05bb7d13bdb02488bea568a8b7fcb762e"}]},"branch":"refs/heads/master","description":"Edit commit message"},"8f67b125cc67c61835822026612a5bcb09d69e68":{"kind":"REWORK","_number":9,"created":"2025-12-11 21:43:49.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/98/970698/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate-specs","ref":"refs/changes/98/970698/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate-specs refs/changes/98/970698/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate-specs refs/changes/98/970698/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate-specs refs/changes/98/970698/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate-specs refs/changes/98/970698/9"}}},"commit":{"parents":[{"commit":"cbf85d69b2a627dbe98aad971635456f325426aa","subject":"Adding a spec for providing catalog zones","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate-specs/commit/cbf85d69b2a627dbe98aad971635456f325426aa"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2025-12-11 20:44:43.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2025-12-11 21:43:49.000000000","tz":0},"subject":"Bump minimum tox version to 3.18.0","message":"Bump minimum tox version to 3.18.0\n\nfor fix\nhttps://zuul.opendev.org/t/openstack/build/e8db7eb566ed479db709b416db346dbb\n\nChange-Id: Ic70ff15eda200d98dad75c71f372da75a8e5035b\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate-specs/commit/8f67b125cc67c61835822026612a5bcb09d69e68"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate-specs/commit/8f67b125cc67c61835822026612a5bcb09d69e68"}]},"branch":"refs/heads/master"},"3403e67db9e5bb8d3d637445a3bccc8eb4bbe823":{"kind":"REWORK","_number":10,"created":"2025-12-11 21:44:09.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/98/970698/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate-specs","ref":"refs/changes/98/970698/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate-specs refs/changes/98/970698/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate-specs refs/changes/98/970698/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate-specs refs/changes/98/970698/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate-specs refs/changes/98/970698/10"}}},"commit":{"parents":[{"commit":"cbf85d69b2a627dbe98aad971635456f325426aa","subject":"Adding a spec for providing catalog zones","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate-specs/commit/cbf85d69b2a627dbe98aad971635456f325426aa"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2025-12-11 20:44:43.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2025-12-11 21:44:07.000000000","tz":0},"subject":"Bump minimum tox version to 3.18.0","message":"Bump minimum tox version to 3.18.0\n\nfor fix\nhttps://zuul.opendev.org/t/openstack/build/e8db7eb566ed479db709b416db346dbb\n\nChange-Id: Ic70ff15eda200d98dad75c71f372da75a8e5035b\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate-specs/commit/3403e67db9e5bb8d3d637445a3bccc8eb4bbe823"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate-specs/commit/3403e67db9e5bb8d3d637445a3bccc8eb4bbe823"}]},"branch":"refs/heads/master"},"76b719fe3b0a4fb283d9046277af5bcd34fc31c7":{"kind":"NO_CODE_CHANGE","_number":11,"created":"2025-12-11 21:44:15.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/98/970698/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate-specs","ref":"refs/changes/98/970698/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate-specs refs/changes/98/970698/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate-specs refs/changes/98/970698/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate-specs refs/changes/98/970698/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate-specs refs/changes/98/970698/11"}}},"commit":{"parents":[{"commit":"cbf85d69b2a627dbe98aad971635456f325426aa","subject":"Adding a spec for providing catalog zones","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate-specs/commit/cbf85d69b2a627dbe98aad971635456f325426aa"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2025-12-11 20:44:43.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2025-12-11 21:44:15.000000000","tz":0},"subject":"wip","message":"wip\n\nChange-Id: Ic70ff15eda200d98dad75c71f372da75a8e5035b\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate-specs/commit/76b719fe3b0a4fb283d9046277af5bcd34fc31c7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate-specs/commit/76b719fe3b0a4fb283d9046277af5bcd34fc31c7"}]},"branch":"refs/heads/master","description":"Edit commit message"},"cc95345fda6315ad91d5b6d73a0ee421ec174d35":{"kind":"NO_CODE_CHANGE","_number":12,"created":"2025-12-11 21:50:09.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/98/970698/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate-specs","ref":"refs/changes/98/970698/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate-specs refs/changes/98/970698/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate-specs refs/changes/98/970698/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate-specs refs/changes/98/970698/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate-specs refs/changes/98/970698/12"}}},"commit":{"parents":[{"commit":"cbf85d69b2a627dbe98aad971635456f325426aa","subject":"Adding a spec for providing catalog zones","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate-specs/commit/cbf85d69b2a627dbe98aad971635456f325426aa"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2025-12-11 20:44:43.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2025-12-11 21:50:09.000000000","tz":0},"subject":"Bump minimum tox version to 3.18.0","message":"Bump minimum tox version to 3.18.0\n\nfor fix\nhttps://zuul.opendev.org/t/openstack/build/e8db7eb566ed479db709b416db346dbb\n\nChange-Id: Ic70ff15eda200d98dad75c71f372da75a8e5035b\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate-specs/commit/cc95345fda6315ad91d5b6d73a0ee421ec174d35"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate-specs/commit/cc95345fda6315ad91d5b6d73a0ee421ec174d35"}]},"branch":"refs/heads/master","description":"Edit commit message"},"7dd7538a63d5dd70acca7943187ecca6b3747f53":{"kind":"REWORK","_number":13,"created":"2025-12-11 21:51:31.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/98/970698/13","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate-specs","ref":"refs/changes/98/970698/13","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate-specs refs/changes/98/970698/13 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate-specs refs/changes/98/970698/13 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate-specs refs/changes/98/970698/13 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate-specs refs/changes/98/970698/13"}}},"commit":{"parents":[{"commit":"cbf85d69b2a627dbe98aad971635456f325426aa","subject":"Adding a spec for providing catalog zones","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate-specs/commit/cbf85d69b2a627dbe98aad971635456f325426aa"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2025-12-11 20:44:43.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2025-12-11 21:51:31.000000000","tz":0},"subject":"Bump minimum tox version to 3.18.0","message":"Bump minimum tox version to 3.18.0\n\nfor fix\nhttps://zuul.opendev.org/t/openstack/build/e8db7eb566ed479db709b416db346dbb\n\nChange-Id: Ic70ff15eda200d98dad75c71f372da75a8e5035b\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate-specs/commit/7dd7538a63d5dd70acca7943187ecca6b3747f53"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate-specs/commit/7dd7538a63d5dd70acca7943187ecca6b3747f53"}]},"branch":"refs/heads/master"},"9723980f4c713169b5e61cc40d34ca58e7107b41":{"kind":"NO_CODE_CHANGE","_number":14,"created":"2026-05-07 10:14:20.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/98/970698/14","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate-specs","ref":"refs/changes/98/970698/14","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate-specs refs/changes/98/970698/14 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate-specs refs/changes/98/970698/14 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate-specs refs/changes/98/970698/14 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate-specs refs/changes/98/970698/14"}}},"commit":{"parents":[{"commit":"cbf85d69b2a627dbe98aad971635456f325426aa","subject":"Adding a spec for providing catalog zones","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate-specs/commit/cbf85d69b2a627dbe98aad971635456f325426aa"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2025-12-11 20:44:43.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-05-07 10:14:20.000000000","tz":0},"subject":"Bump minimum tox version to 3.18.0","message":"Bump minimum tox version to 3.18.0\n\nChange-Id: Ic70ff15eda200d98dad75c71f372da75a8e5035b\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate-specs/commit/9723980f4c713169b5e61cc40d34ca58e7107b41"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate-specs/commit/9723980f4c713169b5e61cc40d34ca58e7107b41"}]},"branch":"refs/heads/master","description":"Edit commit message"},"43e5118e361c5bc3d825a5c5a07cd881f021dee4":{"kind":"REWORK","_number":15,"created":"2026-05-07 10:15:32.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/98/970698/15","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate-specs","ref":"refs/changes/98/970698/15","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate-specs refs/changes/98/970698/15 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate-specs refs/changes/98/970698/15 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate-specs refs/changes/98/970698/15 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate-specs refs/changes/98/970698/15"}}},"commit":{"parents":[{"commit":"cbf85d69b2a627dbe98aad971635456f325426aa","subject":"Adding a spec for providing catalog zones","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate-specs/commit/cbf85d69b2a627dbe98aad971635456f325426aa"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2025-12-11 20:44:43.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-05-07 10:15:20.000000000","tz":0},"subject":"Bump minimum tox version to 3.18.0","message":"Bump minimum tox version to 3.18.0\n\nChange-Id: Ic70ff15eda200d98dad75c71f372da75a8e5035b\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate-specs/commit/43e5118e361c5bc3d825a5c5a07cd881f021dee4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate-specs/commit/43e5118e361c5bc3d825a5c5a07cd881f021dee4"}]},"branch":"refs/heads/master"},"3d72e1cc0430e993bfe00c9c01a1385e96902001":{"kind":"NO_CODE_CHANGE","_number":16,"created":"2026-05-07 10:57:50.000000000","uploader":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"ref":"refs/changes/98/970698/16","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate-specs","ref":"refs/changes/98/970698/16","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate-specs refs/changes/98/970698/16 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate-specs refs/changes/98/970698/16 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate-specs refs/changes/98/970698/16 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate-specs refs/changes/98/970698/16"}}},"commit":{"parents":[{"commit":"cbf85d69b2a627dbe98aad971635456f325426aa","subject":"Adding a spec for providing catalog zones","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate-specs/commit/cbf85d69b2a627dbe98aad971635456f325426aa"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2025-12-11 20:44:43.000000000","tz":0},"committer":{"name":"Omer Schwartz","email":"oschwart@redhat.com","date":"2026-05-07 10:57:50.000000000","tz":0},"subject":"Bump minimum tox version to 3.18.0","message":"Bump minimum tox version to 3.18.0\n\nBump tox minversion from 2.0 to 3.18.0 to match the convention\nused across OpenStack projects.\n\nRemove deprecated tox3 options that are unnecessary or have become\ndefaults in tox4: skipsdist, basepython, and the VIRTUAL_ENV setenv.\n\nPin dependencies against the OpenStack upper-constraints file.\n\nRework the docs build to clean the output directory before running\nsphinx-build, making the -a flag redundant.\n\nChange-Id: Ic70ff15eda200d98dad75c71f372da75a8e5035b\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate-specs/commit/3d72e1cc0430e993bfe00c9c01a1385e96902001"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate-specs/commit/3d72e1cc0430e993bfe00c9c01a1385e96902001"}]},"branch":"refs/heads/master","description":"Edit commit message"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{"label:Review-Priority\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
