)]}'
{"id":"openstack%2Fdesignate-specs~982357","triplet_id":"openstack%2Fdesignate-specs~master~If20010b9bbceca3450a6245e0b7bfcc5ac057b23","project":"openstack/designate-specs","branch":"master","topic":"shared-pools","attention_set":{"33685":{"account":{"_account_id":33685,"name":"Mikhail","display_name":"Mikhail Samoylov","email":"mikhailsamoiloff@gmail.com","username":"mikhails"},"last_update":"2026-04-14 09:39:16.000000000","reason":"\u003cGERRIT_ACCOUNT_31664\u003e replied on the change","reason_account":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"}},"11628":{"account":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"last_update":"2026-03-29 12:40:04.000000000","reason":"\u003cGERRIT_ACCOUNT_33685\u003e replied on the change","reason_account":{"_account_id":33685,"name":"Mikhail","display_name":"Mikhail Samoylov","email":"mikhailsamoiloff@gmail.com","username":"mikhails"}}},"removed_from_attention_set":{"31664":{"account":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"last_update":"2026-04-14 09:39:16.000000000","reason":"\u003cGERRIT_ACCOUNT_31664\u003e replied on the change","reason_account":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"}}},"hashtags":[],"change_id":"If20010b9bbceca3450a6245e0b7bfcc5ac057b23","subject":"Add spec for share pools.","status":"NEW","created":"2026-03-27 05:44:27.000000000","updated":"2026-04-14 09:47:58.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":false,"submittable":false,"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"df7ac3107563c57e9f43755b99cdedb705f374a9","_number":982357,"virtual_id_number":982357,"owner":{"_account_id":33685,"name":"Mikhail","display_name":"Mikhail Samoylov","email":"mikhailsamoiloff@gmail.com","username":"mikhails"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"tag":"autogenerated:zuul:check","value":1,"date":"2026-04-14 09:47:58.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-03-27 05:51:37.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-03-29 12:40:04.000000000","updated_by":{"_account_id":33685,"name":"Mikhail","display_name":"Mikhail Samoylov","email":"mikhailsamoiloff@gmail.com","username":"mikhails"},"reviewer":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"state":"REVIEWER"},{"updated":"2026-03-29 12:40:04.000000000","updated_by":{"_account_id":33685,"name":"Mikhail","display_name":"Mikhail Samoylov","email":"mikhailsamoiloff@gmail.com","username":"mikhails"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"}],"messages":[{"id":"aee3c0ecf7d1d6fc68d8126f32cc0e27bcdd6d69","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33685,"name":"Mikhail","display_name":"Mikhail Samoylov","email":"mikhailsamoiloff@gmail.com","username":"mikhails"},"date":"2026-03-27 05:44:27.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"38b3be6bbf50142fa92905c99f259b29d8bd24ba","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-27 05:51:37.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/b0f9ae37dab04966a76c06a6a4656a67\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3bf7d7e1332244baa3ef51fbe30f318c : FAILURE in 3m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fa380bf91e1d408f81488bec481576c8 : SUCCESS in 2m 41s","accounts_in_message":[],"_revision_number":1},{"id":"a89579d0586e82fbeb4d573def602ab8a17599a2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33685,"name":"Mikhail","display_name":"Mikhail Samoylov","email":"mikhailsamoiloff@gmail.com","username":"mikhails"},"date":"2026-03-27 05:55:59.000000000","message":"Uploaded patch set 2: Commit message was updated.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"24be09d463dc52ad541770e0cc46eec1897db241","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-27 05:58:57.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/ddef8efaed8b42afa87edc1b18b8944e\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d3f279542bc942fe982f3237c9e8f354 : FAILURE in 2m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6f21a0d0872145369fd11bc10a0e12db : SUCCESS in 2m 14s","accounts_in_message":[],"_revision_number":2},{"id":"03dc72afd16a490e2bc62ab81be27ad0c450c2d4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33685,"name":"Mikhail","display_name":"Mikhail Samoylov","email":"mikhailsamoiloff@gmail.com","username":"mikhails"},"date":"2026-03-27 07:06:06.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"d986f03393b782a9f32a5cd209c3f36032f60e70","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-27 07:11:52.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/f40e865b703f4d32af99769791e78463\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4c618c29fd964f2ab7d105713db19323 : FAILURE in 4m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c394440ffecc4fd8959046d4552f2f8c : SUCCESS in 2m 20s","accounts_in_message":[],"_revision_number":3},{"id":"1e4c7e494b7f6d8cbdd1d7fb3a0f95a3320e3b47","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33685,"name":"Mikhail","display_name":"Mikhail Samoylov","email":"mikhailsamoiloff@gmail.com","username":"mikhails"},"date":"2026-03-27 08:03:17.000000000","message":"Uploaded patch set 4.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":4},{"id":"55310bcd3617afb6a95e23523a5582dbe05b655f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-27 08:07:57.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/0a5a21c09d8040d08c5ea5c8949fdaea\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/56f2b49dbc61448ea5e99c5fd3cb354b : FAILURE in 3m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3f471726cca04c508513b97fedbbde49 : SUCCESS in 3m 07s","accounts_in_message":[],"_revision_number":4},{"id":"cbfb0588a44e141da318956202e434fd1046dd46","author":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"date":"2026-04-14 09:39:16.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"df7ac3107563c57e9f43755b99cdedb705f374a9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-14 09:47:58.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/79539a0f6f3b41918c49db8b1f678afe\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a293c14b3cc14b51bdefa1adde89e42c : SUCCESS in 2m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2518c2d09e7b427c831c6c8087cff9a0 : SUCCESS in 2m 33s","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"4e162eb6d2ab77942dfeb8f339bc99eea12f36cb","revisions":{"145d86742730c3b49df689e1af59c2fe4552c401":{"kind":"REWORK","_number":1,"created":"2026-03-27 05:44:27.000000000","uploader":{"_account_id":33685,"name":"Mikhail","display_name":"Mikhail Samoylov","email":"mikhailsamoiloff@gmail.com","username":"mikhails"},"ref":"refs/changes/57/982357/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate-specs","ref":"refs/changes/57/982357/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate-specs refs/changes/57/982357/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate-specs refs/changes/57/982357/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate-specs refs/changes/57/982357/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate-specs refs/changes/57/982357/1"}}},"commit":{"parents":[{"commit":"cbf85d69b2a627dbe98aad971635456f325426aa","subject":"Adding a spec for providing catalog zones","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate-specs/commit/cbf85d69b2a627dbe98aad971635456f325426aa"}]}],"author":{"name":"Mikhail Samoylov","email":"mikhail.samoylov@sap.com","date":"2026-03-27 05:42:16.000000000","tz":420},"committer":{"name":"Mikhail Samoylov","email":"mikhailsamoiloff@gmail.com","date":"2026-03-27 05:43:52.000000000","tz":420},"subject":"Add spec for share pools.","message":"Add spec for share pools.\n\nChange-Id: If20010b9bbceca3450a6245e0b7bfcc5ac057b23\nSigned-off-by: Mikhail Samoylov \u003cmikhailsamoiloff@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate-specs/commit/145d86742730c3b49df689e1af59c2fe4552c401"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate-specs/commit/145d86742730c3b49df689e1af59c2fe4552c401"}]},"branch":"refs/heads/master"},"6a8eae9e67652f8e244aab11bd6bf680115c743b":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2026-03-27 05:55:59.000000000","uploader":{"_account_id":33685,"name":"Mikhail","display_name":"Mikhail Samoylov","email":"mikhailsamoiloff@gmail.com","username":"mikhails"},"ref":"refs/changes/57/982357/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate-specs","ref":"refs/changes/57/982357/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate-specs refs/changes/57/982357/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate-specs refs/changes/57/982357/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate-specs refs/changes/57/982357/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate-specs refs/changes/57/982357/2"}}},"commit":{"parents":[{"commit":"cbf85d69b2a627dbe98aad971635456f325426aa","subject":"Adding a spec for providing catalog zones","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate-specs/commit/cbf85d69b2a627dbe98aad971635456f325426aa"}]}],"author":{"name":"Mikhail Samoylov","email":"mikhailsamoiloff@gmail.com","date":"2026-03-27 05:42:16.000000000","tz":420},"committer":{"name":"Mikhail Samoylov","email":"mikhailsamoiloff@gmail.com","date":"2026-03-27 05:55:34.000000000","tz":420},"subject":"Add spec for share pools.","message":"Add spec for share pools.\n\nImplements: blueprint share-pools\nChange-Id: If20010b9bbceca3450a6245e0b7bfcc5ac057b23\nSigned-off-by: Mikhail Samoylov \u003cmikhailsamoiloff@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate-specs/commit/6a8eae9e67652f8e244aab11bd6bf680115c743b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate-specs/commit/6a8eae9e67652f8e244aab11bd6bf680115c743b"}]},"branch":"refs/heads/master"},"cfb046e152db0571540d2c9925369e074f70b2ec":{"kind":"REWORK","_number":3,"created":"2026-03-27 07:06:06.000000000","uploader":{"_account_id":33685,"name":"Mikhail","display_name":"Mikhail Samoylov","email":"mikhailsamoiloff@gmail.com","username":"mikhails"},"ref":"refs/changes/57/982357/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate-specs","ref":"refs/changes/57/982357/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate-specs refs/changes/57/982357/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate-specs refs/changes/57/982357/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate-specs refs/changes/57/982357/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate-specs refs/changes/57/982357/3"}}},"commit":{"parents":[{"commit":"cbf85d69b2a627dbe98aad971635456f325426aa","subject":"Adding a spec for providing catalog zones","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate-specs/commit/cbf85d69b2a627dbe98aad971635456f325426aa"}]}],"author":{"name":"Mikhail Samoylov","email":"mikhailsamoiloff@gmail.com","date":"2026-03-27 05:42:16.000000000","tz":420},"committer":{"name":"Mikhail Samoylov","email":"mikhailsamoiloff@gmail.com","date":"2026-03-27 07:05:48.000000000","tz":420},"subject":"Add spec for share pools.","message":"Add spec for share pools.\n\nImplements: blueprint share-pools\nChange-Id: If20010b9bbceca3450a6245e0b7bfcc5ac057b23\nSigned-off-by: Mikhail Samoylov \u003cmikhailsamoiloff@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate-specs/commit/cfb046e152db0571540d2c9925369e074f70b2ec"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate-specs/commit/cfb046e152db0571540d2c9925369e074f70b2ec"}]},"branch":"refs/heads/master"},"4e162eb6d2ab77942dfeb8f339bc99eea12f36cb":{"kind":"REWORK","_number":4,"created":"2026-03-27 08:03:17.000000000","uploader":{"_account_id":33685,"name":"Mikhail","display_name":"Mikhail Samoylov","email":"mikhailsamoiloff@gmail.com","username":"mikhails"},"ref":"refs/changes/57/982357/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate-specs","ref":"refs/changes/57/982357/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate-specs refs/changes/57/982357/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate-specs refs/changes/57/982357/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate-specs refs/changes/57/982357/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate-specs refs/changes/57/982357/4"}}},"commit":{"parents":[{"commit":"cbf85d69b2a627dbe98aad971635456f325426aa","subject":"Adding a spec for providing catalog zones","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate-specs/commit/cbf85d69b2a627dbe98aad971635456f325426aa"}]}],"author":{"name":"Mikhail Samoylov","email":"mikhailsamoiloff@gmail.com","date":"2026-03-27 05:42:16.000000000","tz":420},"committer":{"name":"Mikhail Samoylov","email":"mikhailsamoiloff@gmail.com","date":"2026-03-27 08:03:01.000000000","tz":420},"subject":"Add spec for share pools.","message":"Add spec for share pools.\n\nImplements: blueprint share-pools\nChange-Id: If20010b9bbceca3450a6245e0b7bfcc5ac057b23\nSigned-off-by: Mikhail Samoylov \u003cmikhailsamoiloff@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate-specs/commit/4e162eb6d2ab77942dfeb8f339bc99eea12f36cb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate-specs/commit/4e162eb6d2ab77942dfeb8f339bc99eea12f36cb"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY"},{"label":"Workflow","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{"label:Review-Priority\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
