)]}'
{"id":"openstack%2Fdesignate-tempest-plugin~1004219","triplet_id":"openstack%2Fdesignate-tempest-plugin~master~Ib33c1b1ded37673f412ac20c3573a5fe900a4128","project":"openstack/designate-tempest-plugin","branch":"master","attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2026-09-07 06:38:09.000000000","reason":"\u003cGERRIT_ACCOUNT_13252\u003e replied on the change","reason_account":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}}},"removed_from_attention_set":{},"hashtags":[],"change_id":"Ib33c1b1ded37673f412ac20c3573a5fe900a4128","subject":"tox: Drop redundant passenv","status":"NEW","created":"2026-09-05 18:07:12.000000000","updated":"2026-09-07 06:38:09.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":true,"submittable":false,"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"0420d0736aca71afa8750d853a04ef6282a7b83e","_number":1004219,"virtual_id_number":1004219,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"tag":"autogenerated:zuul:check","value":1,"date":"2026-09-06 14:01:05.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":2,"date":"2026-09-07 06:38:09.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"recommended":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":1,"date":"2026-09-07 06:38:09.000000000","permitted_voting_range":{"min":-1,"max":2},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","value":1,"default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-09-05 19:37:52.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-09-07 06:38:09.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"REVIEWER"}],"messages":[{"id":"2d66d172f28a12d003dec13a19ea2eb7e5a0ebb9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-09-05 18:07:12.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"1ec6f8ca8222616c39ec09782dc71cb28a8d19fc","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-09-05 18:13:06.000000000","message":"Patch Set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"6e514f6c5a5a2045489be899c8f54e8cea07f638","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-09-05 18:16:37.000000000","message":"Patch Set 3: Commit message was updated.","accounts_in_message":[],"_revision_number":3},{"id":"13661f11a769353abb2d979dddbda42fcd5af007","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-05 19:37:52.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/59dd2ad4fc51415ba3c5500560d5f80a\n\n- designate-bind9-core https://zuul.opendev.org/t/openstack/build/9de0614d31b2437aa881de885b292b68 : SUCCESS in 34m 49s\n- designate-bind9-with-keystone-default-roles https://zuul.opendev.org/t/openstack/build/00eebf10d5c147fabb145bdfc01d228c : SUCCESS in 43m 51s\n- designate-pdns4-core https://zuul.opendev.org/t/openstack/build/c7cf3276138f460eb1d66a76f3b28a8c : SUCCESS in 43m 16s\n- designate-bind9-multipool https://zuul.opendev.org/t/openstack/build/60a6558736734f739d3529a79d1aa9f3 : SUCCESS in 33m 22s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/6d02b9c2970f43688e364f3104347ab4 : SUCCESS in 1h 11m 12s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/6549b6d3bab748fe9dcbc5756baf490c : SUCCESS in 59m 21s\n- designate-grenade-bind9-skip-level https://zuul.opendev.org/t/openstack/build/62b0604c70f046bf8d24e771ff011e31 : SUCCESS in 57m 40s\n- designate-grenade-pdns4-skip-level https://zuul.opendev.org/t/openstack/build/632b1d11c4ae41b49ea7227dab79ba38 : SUCCESS in 1h 04m 57s\n- designate-bind9-core-catalog-zones https://zuul.opendev.org/t/openstack/build/5f242653bc034acdba79fe49f924ced2 : SUCCESS in 38m 03s (non-voting)\n- designate-only-ipv6-pdns4 https://zuul.opendev.org/t/openstack/build/0f19bdba8b29404ab69c9f04b8fa18f6 : SUCCESS in 46m 05s\n- designate-only-ipv6-bind9 https://zuul.opendev.org/t/openstack/build/b2d528a7ed904515b9f268fc79814e93 : SUCCESS in 33m 06s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0702a48529fd44ccb48033eda85105c7 : SUCCESS in 3m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/316233eb10514121a21c9c261f8c1183 : SUCCESS in 2m 09s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/526353a3ac034b5d8c11900d36ffe716 : SUCCESS in 3m 14s\n- designate-bind9-stable-2026-1 https://zuul.opendev.org/t/openstack/build/7f5d74e5302845ebb92ca9d728a0bc0e : SUCCESS in 39m 50s\n- designate-bind9-stable-2025-2 https://zuul.opendev.org/t/openstack/build/5050596125864f9e96c75bc52e26a497 : SUCCESS in 54m 52s\n- designate-bind9-stable-2025-1 https://zuul.opendev.org/t/openstack/build/c880a3e6f32e4eaeb60e61c9310adf9a : SUCCESS in 55m 03s\n- designate-bind9-multipool-stable-2026-1 https://zuul.opendev.org/t/openstack/build/6f2b2c625f67480e9eb37304966dc626 : SUCCESS in 37m 53s\n- designate-bind9-multipool-stable-2025-2 https://zuul.opendev.org/t/openstack/build/a221ab491ef54366adf6c1cd1d0caeed : SUCCESS in 17m 57s\n- designate-bind9-multipool-stable-2025-1 https://zuul.opendev.org/t/openstack/build/67d1ded59b9745c186a4fce4da992869 : SUCCESS in 16m 18s\n- designate-bind9-split-horizon https://zuul.opendev.org/t/openstack/build/cc7814572fa94633b02c3b287747fd01 : SUCCESS in 33m 20s","accounts_in_message":[],"_revision_number":3},{"id":"da0ba1a900aedc61a172a7af21c8631fb5fc7a1d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-09-06 12:47:40.000000000","message":"Uploaded patch set 4: Patch Set 3 was rebased. Commit message was updated.\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":4},{"id":"761af070fc9400679733b5a2e3db24217d7626af","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-06 14:01:05.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/228c4e68215444c3a5128f57134257c7\n\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/13f8373058d44fe782d88eec3c6de49e : SUCCESS in 49m 00s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/5ad0c22ebdc2480dbe8816b5900ca729 : SUCCESS in 1h 07m 18s\n- designate-grenade-bind9-skip-level https://zuul.opendev.org/t/openstack/build/65e0ca6c97f6451ba91360c4109e49f4 : SUCCESS in 50m 11s\n- designate-grenade-pdns4-skip-level https://zuul.opendev.org/t/openstack/build/6c1cf3fc57c548a088dc3760ebdd9542 : SUCCESS in 48m 49s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2d03517ff7484709b0c655cbf1812705 : SUCCESS in 2m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a4889116fc624f8984348038e0c935ce : SUCCESS in 2m 10s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d403cbc86d594dc5a626d2bbcc62692a : SUCCESS in 3m 57s","accounts_in_message":[],"_revision_number":4},{"id":"0420d0736aca71afa8750d853a04ef6282a7b83e","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2026-09-07 06:38:09.000000000","message":"Patch Set 4: Code-Review+2 Review-Priority+1","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"28cefe7bf460ae19e27aeb3889d5d87c822af96a","revisions":{"c97ef6297940afec600c961048225821dcfa7c17":{"kind":"REWORK","_number":1,"created":"2026-09-05 18:07:12.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/19/1004219/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate-tempest-plugin","ref":"refs/changes/19/1004219/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate-tempest-plugin refs/changes/19/1004219/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate-tempest-plugin refs/changes/19/1004219/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate-tempest-plugin refs/changes/19/1004219/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate-tempest-plugin refs/changes/19/1004219/1"}}},"commit":{"parents":[{"commit":"4c69b6f219de7d6ad3b07be1f572edba64aab0c7","subject":"Merge \"Multi Pool - new scenario covering \u0027Move Zone API\u0027\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate-tempest-plugin/commit/4c69b6f219de7d6ad3b07be1f572edba64aab0c7"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-09-05 18:07:03.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-09-05 18:07:03.000000000","tz":540},"subject":"tox: Drop redundant passenv","message":"tox: Drop redundant passenv\n\nSince 4.0.0[1], tox passes basic environment variables automatically\nand we don\u0027t have add these explicitly.\n\ncf. https: //tox.wiki/en/latest/reference/config.html#passenv\n\nSome variables are always passed through to ensure the basic\nfunctionality of standard library functions or tooling like pip.\nThis is also not case sensitive on all platforms except Windows.\n\n[1] https://github.com/tox-dev/tox/commit/afb3b395dd\n\nChange-Id: Ib33c1b1ded37673f412ac20c3573a5fe900a4128\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate-tempest-plugin/commit/c97ef6297940afec600c961048225821dcfa7c17"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate-tempest-plugin/commit/c97ef6297940afec600c961048225821dcfa7c17"}]},"branch":"refs/heads/master"},"65870beb825dc93d43431026f1e9c35dfab20964":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2026-09-05 18:13:06.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/19/1004219/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate-tempest-plugin","ref":"refs/changes/19/1004219/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate-tempest-plugin refs/changes/19/1004219/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate-tempest-plugin refs/changes/19/1004219/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate-tempest-plugin refs/changes/19/1004219/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate-tempest-plugin refs/changes/19/1004219/2"}}},"commit":{"parents":[{"commit":"4c69b6f219de7d6ad3b07be1f572edba64aab0c7","subject":"Merge \"Multi Pool - new scenario covering \u0027Move Zone API\u0027\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate-tempest-plugin/commit/4c69b6f219de7d6ad3b07be1f572edba64aab0c7"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-09-05 18:07:03.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-09-05 18:13:06.000000000","tz":0},"subject":"tox: Drop redundant passenv","message":"tox: Drop redundant passenv\n\nSince 4.0.0[1], tox passes basic environment variables automatically\nand we don\u0027t have add these explicitly.\n\ncf. https://tox.wiki/en/latest/reference/config.html#passenv\n\nSome variables are always passed through to ensure the basic\nfunctionality of standard library functions or tooling like pip.\nThis is also not case sensitive on all platforms except Windows.\n\n[1] https://github.com/tox-dev/tox/commit/afb3b395dd\n\nChange-Id: Ib33c1b1ded37673f412ac20c3573a5fe900a4128\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate-tempest-plugin/commit/65870beb825dc93d43431026f1e9c35dfab20964"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate-tempest-plugin/commit/65870beb825dc93d43431026f1e9c35dfab20964"}]},"branch":"refs/heads/master","description":"Edit commit message"},"a1802e3c4bf3d0fe3f7b58f99a36902adfe9f2f4":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2026-09-05 18:16:37.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/19/1004219/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate-tempest-plugin","ref":"refs/changes/19/1004219/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate-tempest-plugin refs/changes/19/1004219/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate-tempest-plugin refs/changes/19/1004219/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate-tempest-plugin refs/changes/19/1004219/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate-tempest-plugin refs/changes/19/1004219/3"}}},"commit":{"parents":[{"commit":"4c69b6f219de7d6ad3b07be1f572edba64aab0c7","subject":"Merge \"Multi Pool - new scenario covering \u0027Move Zone API\u0027\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate-tempest-plugin/commit/4c69b6f219de7d6ad3b07be1f572edba64aab0c7"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-09-05 18:07:03.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-09-05 18:16:37.000000000","tz":0},"subject":"tox: Drop redundant passenv","message":"tox: Drop redundant passenv\n\nSince 4.0.0[1], tox passes basic environment variables automatically\nand we don\u0027t have to add these explicitly.\n\ncf. https://tox.wiki/en/latest/reference/config.html#passenv\n\nSome variables are always passed through to ensure the basic\nfunctionality of standard library functions or tooling like pip.\nThis is also not case sensitive on all platforms except Windows.\n\n[1] https://github.com/tox-dev/tox/commit/afb3b395dd\n\nChange-Id: Ib33c1b1ded37673f412ac20c3573a5fe900a4128\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate-tempest-plugin/commit/a1802e3c4bf3d0fe3f7b58f99a36902adfe9f2f4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate-tempest-plugin/commit/a1802e3c4bf3d0fe3f7b58f99a36902adfe9f2f4"}]},"branch":"refs/heads/master","description":"Edit commit message"},"28cefe7bf460ae19e27aeb3889d5d87c822af96a":{"kind":"TRIVIAL_REBASE_WITH_MESSAGE_UPDATE","_number":4,"created":"2026-09-06 12:47:40.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/19/1004219/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate-tempest-plugin","ref":"refs/changes/19/1004219/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate-tempest-plugin refs/changes/19/1004219/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate-tempest-plugin refs/changes/19/1004219/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate-tempest-plugin refs/changes/19/1004219/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate-tempest-plugin refs/changes/19/1004219/4"}}},"commit":{"parents":[{"commit":"e3814605a274f49e607e081c491459c624c80b4e","subject":"Skip tempest tests for tox.ini update","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate-tempest-plugin/commit/e3814605a274f49e607e081c491459c624c80b4e"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-09-05 18:07:03.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-09-06 12:47:30.000000000","tz":540},"subject":"tox: Drop redundant passenv","message":"tox: Drop redundant passenv\n\nSince 4.0.0[1], tox passes basic environment variables automatically\nand we don\u0027t have add these explicitly.\n\ncf. https: //tox.wiki/en/latest/reference/config.html#passenv\n\nSome variables are always passed through to ensure the basic\nfunctionality of standard library functions or tooling like pip.\nThis is also not case sensitive on all platforms except Windows.\n\n[1] https://github.com/tox-dev/tox/commit/afb3b395dd\n\nChange-Id: Ib33c1b1ded37673f412ac20c3573a5fe900a4128\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate-tempest-plugin/commit/28cefe7bf460ae19e27aeb3889d5d87c822af96a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate-tempest-plugin/commit/28cefe7bf460ae19e27aeb3889d5d87c822af96a"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}},{"label":"Workflow","status":"MAY"},{"label":"Review-Priority","status":"MAY","applied_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{"label:Review-Priority\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
