)]}'
{"id":"openstack%2Fdesignate-tempest-plugin~1004308","triplet_id":"openstack%2Fdesignate-tempest-plugin~master~I3da781ded24912d8dfe6ee05de2e74adb19f6bcb","project":"openstack/designate-tempest-plugin","branch":"master","attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2026-09-07 06:35:24.000000000","reason":"\u003cGERRIT_ACCOUNT_13252\u003e replied on the change","reason_account":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}}},"removed_from_attention_set":{},"hashtags":[],"change_id":"I3da781ded24912d8dfe6ee05de2e74adb19f6bcb","subject":"Skip tempest tests for tox.ini update","status":"NEW","created":"2026-09-06 12:47:40.000000000","updated":"2026-09-07 06:35:24.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":true,"submittable":false,"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"feff01c1ceef949d70aa7d2d3150ad44c8d19aa1","_number":1004308,"virtual_id_number":1004308,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"tag":"autogenerated:zuul:check","value":1,"date":"2026-09-06 14:03:58.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":2,"date":"2026-09-07 06:35:24.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"recommended":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":1,"date":"2026-09-07 06:35:24.000000000","permitted_voting_range":{"min":-1,"max":2},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","value":1,"default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-09-06 14:03:58.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-09-07 06:35:24.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"REVIEWER"}],"messages":[{"id":"be9bb2ec9551ff3facb9e6bc4dd5e37e132896e1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-09-06 12:47:40.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"8e89973efce40aea62c1ad931c6e07432a9157b7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-06 14:03:58.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d40d5d7522bf4b62acd324e1a51a1639\n\n- designate-bind9-core https://zuul.opendev.org/t/openstack/build/0f486c1eadfc497080127d402518f83e : SUCCESS in 45m 13s\n- designate-bind9-with-keystone-default-roles https://zuul.opendev.org/t/openstack/build/bfd4952c2f704932bd042ecf3e3d5f84 : SUCCESS in 42m 36s\n- designate-pdns4-core https://zuul.opendev.org/t/openstack/build/b9c5e3b6a23d4d9eac2a55c096bddd1c : SUCCESS in 37m 18s\n- designate-bind9-multipool https://zuul.opendev.org/t/openstack/build/a4403138bb264e0c8467e52de8cc988a : SUCCESS in 20m 10s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/137824a1283245cba3ebcc1c1176a95b : SUCCESS in 51m 11s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/00fa2e1ee1b249ce8bc0a59264016867 : SUCCESS in 48m 42s\n- designate-grenade-bind9-skip-level https://zuul.opendev.org/t/openstack/build/1632d8dc45b94567935fe81f92cdc67a : SUCCESS in 1h 10m 14s\n- designate-grenade-pdns4-skip-level https://zuul.opendev.org/t/openstack/build/819845a2bcea4e87b64462d185257638 : SUCCESS in 47m 00s\n- designate-bind9-core-catalog-zones https://zuul.opendev.org/t/openstack/build/74fea13dfe934355949855b1984cbff9 : SUCCESS in 39m 12s (non-voting)\n- designate-only-ipv6-pdns4 https://zuul.opendev.org/t/openstack/build/4ec7144540764d65b1615a51db708690 : SUCCESS in 37m 38s\n- designate-only-ipv6-bind9 https://zuul.opendev.org/t/openstack/build/9fd9b76f5ad94f208c03c3a6875556b3 : SUCCESS in 33m 13s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/414499cfced347b2b53ae4ed3cb2480a : SUCCESS in 2m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9b507579b2be46d4a7e6053c662ce2e2 : SUCCESS in 3m 47s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2368151bbe4d40588d6e97cb5e902db0 : SUCCESS in 3m 47s\n- designate-bind9-stable-2026-1 https://zuul.opendev.org/t/openstack/build/5d9cff9dbd98432fab71a7a9979609e4 : SUCCESS in 48m 29s\n- designate-bind9-stable-2025-2 https://zuul.opendev.org/t/openstack/build/1f7b601faa6f4bfda6fdf9081ac5409a : SUCCESS in 47m 48s\n- designate-bind9-stable-2025-1 https://zuul.opendev.org/t/openstack/build/d707c09ea7f84484ba81b2b0b391e54d : SUCCESS in 50m 10s\n- designate-bind9-multipool-stable-2026-1 https://zuul.opendev.org/t/openstack/build/74696a66772143a1a1ca61d7816ff45c : SUCCESS in 34m 50s\n- designate-bind9-multipool-stable-2025-2 https://zuul.opendev.org/t/openstack/build/39ec11da1b8748cfa24fd201ee367d57 : SUCCESS in 31m 02s\n- designate-bind9-multipool-stable-2025-1 https://zuul.opendev.org/t/openstack/build/2a67f3556871433abbe57c541b398b0e : SUCCESS in 17m 20s\n- designate-bind9-split-horizon https://zuul.opendev.org/t/openstack/build/810e1c753ca04932b261154d6e82f102 : SUCCESS in 34m 47s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/5a2bb4e7c1bb4a3abdeb858e5a89a9bc : SUCCESS in 30m 06s","accounts_in_message":[],"_revision_number":1},{"id":"feff01c1ceef949d70aa7d2d3150ad44c8d19aa1","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2026-09-07 06:35:24.000000000","message":"Patch Set 1: Code-Review+2 Review-Priority+1","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"e3814605a274f49e607e081c491459c624c80b4e","revisions":{"e3814605a274f49e607e081c491459c624c80b4e":{"kind":"REWORK","_number":1,"created":"2026-09-06 12:47:40.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/08/1004308/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate-tempest-plugin","ref":"refs/changes/08/1004308/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate-tempest-plugin refs/changes/08/1004308/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate-tempest-plugin refs/changes/08/1004308/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate-tempest-plugin refs/changes/08/1004308/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate-tempest-plugin refs/changes/08/1004308/1"}}},"commit":{"parents":[{"commit":"4c69b6f219de7d6ad3b07be1f572edba64aab0c7","subject":"Merge \"Multi Pool - new scenario covering \u0027Move Zone API\u0027\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate-tempest-plugin/commit/4c69b6f219de7d6ad3b07be1f572edba64aab0c7"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-09-06 12:46:55.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-09-06 12:47:30.000000000","tz":540},"subject":"Skip tempest tests for tox.ini update","message":"Skip tempest tests for tox.ini update\n\nThe tox.ini file is not used in tempest tests. Skip tempest jobs to\navoid consuming unnecessary CI resources.\n\nChange-Id: I3da781ded24912d8dfe6ee05de2e74adb19f6bcb\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate-tempest-plugin/commit/e3814605a274f49e607e081c491459c624c80b4e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate-tempest-plugin/commit/e3814605a274f49e607e081c491459c624c80b4e"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}},{"label":"Workflow","status":"MAY"},{"label":"Review-Priority","status":"MAY","applied_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{"label:Review-Priority\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
