)]}'
{"id":"openstack%2Fdesignate-tempest-plugin~1004310","triplet_id":"openstack%2Fdesignate-tempest-plugin~master~I52239eab046ea594b99d851c338d9c9a0f14e0f8","project":"openstack/designate-tempest-plugin","branch":"master","hashtags":[],"change_id":"I52239eab046ea594b99d851c338d9c9a0f14e0f8","subject":"Skip tempest jobs for locale/test updates","status":"NEW","created":"2026-09-06 14:25:29.000000000","updated":"2026-09-06 15:41:29.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":true,"submittable":false,"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"a1ce8fa1b74777618063c510f1f8646e06dbf7dc","_number":1004310,"virtual_id_number":1004310,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2026-09-06 15:41:29.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-09-06 15:41:29.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"a074669a31b19a36462386a32dad82a704d648d6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-09-06 14:25:29.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"a1ce8fa1b74777618063c510f1f8646e06dbf7dc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-06 15:41:29.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/992888b3c6d24cfcb1253c6cbc60bafe\n\n- designate-bind9-core https://zuul.opendev.org/t/openstack/build/aa0e75b0768a4c1082caaacd0d44f927 : SUCCESS in 44m 28s\n- designate-bind9-with-keystone-default-roles https://zuul.opendev.org/t/openstack/build/5593e3f4a41b4d40833082b87fba81b4 : SUCCESS in 45m 01s\n- designate-pdns4-core https://zuul.opendev.org/t/openstack/build/8631774f9f9f44c2841510d9b01c7908 : SUCCESS in 30m 40s\n- designate-bind9-multipool https://zuul.opendev.org/t/openstack/build/821f999ca6ca4256acf86e2912c26439 : SUCCESS in 20m 25s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/e0c2ad1d77be4472b1e6e76dc52b3706 : SUCCESS in 56m 42s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/ee02150e4cf844cabeb9278b90ea8653 : SUCCESS in 1h 05m 04s\n- designate-grenade-bind9-skip-level https://zuul.opendev.org/t/openstack/build/52ceb04e26da40e2b8129fd69eb6b7c5 : SUCCESS in 1h 08m 49s\n- designate-grenade-pdns4-skip-level https://zuul.opendev.org/t/openstack/build/69d1e50df5d44d32a9187d5375347bd3 : SUCCESS in 1h 07m 29s\n- designate-bind9-core-catalog-zones https://zuul.opendev.org/t/openstack/build/fc4e65ed8e2246a2bd9c7a9373c24817 : SUCCESS in 43m 28s (non-voting)\n- designate-only-ipv6-pdns4 https://zuul.opendev.org/t/openstack/build/b9d428405c6b44e3b925862bcccb84bc : SUCCESS in 45m 51s\n- designate-only-ipv6-bind9 https://zuul.opendev.org/t/openstack/build/7641cd152f3c4cc29a30a5d145a44d12 : SUCCESS in 45m 59s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7d21a28a973a4878b1ac70c44d568221 : SUCCESS in 2m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6d10131888a049d5b16ebe92bce54f8c : SUCCESS in 2m 09s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e224c76aba2547c6b75af84eb2616df8 : SUCCESS in 3m 56s\n- designate-bind9-stable-2026-1 https://zuul.opendev.org/t/openstack/build/8128a922fc6343bdbeddadf8b82e99b8 : SUCCESS in 47m 39s\n- designate-bind9-stable-2025-2 https://zuul.opendev.org/t/openstack/build/582abebe7bb944e78ee7d8ab8ba18a70 : SUCCESS in 49m 40s\n- designate-bind9-stable-2025-1 https://zuul.opendev.org/t/openstack/build/3f8407409adc4094be9cf35f5f29bcea : SUCCESS in 49m 41s\n- designate-bind9-multipool-stable-2026-1 https://zuul.opendev.org/t/openstack/build/c8ce53a1df0448a39fb8cf80d0c290e5 : SUCCESS in 36m 16s\n- designate-bind9-multipool-stable-2025-2 https://zuul.opendev.org/t/openstack/build/653001ebe7664de8a9068a073222263e : SUCCESS in 35m 34s\n- designate-bind9-multipool-stable-2025-1 https://zuul.opendev.org/t/openstack/build/6723ff0a9e354316afcb0f78604fc33b : SUCCESS in 34m 49s\n- designate-bind9-split-horizon https://zuul.opendev.org/t/openstack/build/32f87bc782b140eca39ef343f30c147d : SUCCESS in 29m 02s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/27dee903ec764436bd3bce690ed5fcb3 : SUCCESS in 31m 28s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"81754cbecdfeaa86940ee606e320622de3912886","revisions":{"81754cbecdfeaa86940ee606e320622de3912886":{"kind":"REWORK","_number":1,"created":"2026-09-06 14:25:29.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/10/1004310/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate-tempest-plugin","ref":"refs/changes/10/1004310/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate-tempest-plugin refs/changes/10/1004310/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate-tempest-plugin refs/changes/10/1004310/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate-tempest-plugin refs/changes/10/1004310/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate-tempest-plugin refs/changes/10/1004310/1"}}},"commit":{"parents":[{"commit":"4c69b6f219de7d6ad3b07be1f572edba64aab0c7","subject":"Merge \"Multi Pool - new scenario covering \u0027Move Zone API\u0027\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate-tempest-plugin/commit/4c69b6f219de7d6ad3b07be1f572edba64aab0c7"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-09-06 14:24:02.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-09-06 14:24:42.000000000","tz":540},"subject":"Skip tempest jobs for locale/test updates","message":"Skip tempest jobs for locale/test updates\n\nThese files do not affect functionalities tested in these tests.\n\nChange-Id: I52239eab046ea594b99d851c338d9c9a0f14e0f8\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate-tempest-plugin/commit/81754cbecdfeaa86940ee606e320622de3912886"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate-tempest-plugin/commit/81754cbecdfeaa86940ee606e320622de3912886"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY"},{"label":"Workflow","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{"label:Review-Priority\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
