)]}'
{"id":"openstack%2Fdesignate-tempest-plugin~1005311","triplet_id":"openstack%2Fdesignate-tempest-plugin~master~I70f0e9c23d2878c00b51004f60ce5d1c9a988ab3","project":"openstack/designate-tempest-plugin","branch":"master","topic":"2026.2-job","hashtags":[],"change_id":"I70f0e9c23d2878c00b51004f60ce5d1c9a988ab3","subject":"Add stable/2026.2 job","status":"NEW","created":"2026-09-11 17:08:42.000000000","updated":"2026-09-11 18:34:57.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":true,"submittable":false,"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"f97336d7db3412168c75423a5db914ef7af5fa0f","_number":1005311,"virtual_id_number":1005311,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2026-09-11 18:34:57.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-09-11 18:34:57.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"5f25e15d148f5d612161cfa32c3271fad7b30984","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-09-11 17:08:42.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"f97336d7db3412168c75423a5db914ef7af5fa0f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-11 18:34:57.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2ec632d1384b4c6ca62fe701d0665db2\n\n- designate-bind9-core https://zuul.opendev.org/t/openstack/build/a2949b1c98644bcab2847384a2ab3211 : SUCCESS in 47m 52s\n- designate-bind9-with-keystone-default-roles https://zuul.opendev.org/t/openstack/build/2f42eab8ceaa4093a59a687fa4d9ed82 : SUCCESS in 35m 10s\n- designate-pdns4-core https://zuul.opendev.org/t/openstack/build/c2840ebaec464b28bd611e620d7db934 : SUCCESS in 41m 10s\n- designate-bind9-multipool https://zuul.opendev.org/t/openstack/build/4ea94222696b47acb8107362bba1dd5b : SUCCESS in 38m 58s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/d6adcde02ce94a79ba7f0cde6db0df5f : SUCCESS in 1h 10m 41s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/33bb57f1e1644bf4842e83327b19709c : SUCCESS in 1h 07m 17s\n- designate-grenade-bind9-skip-level https://zuul.opendev.org/t/openstack/build/ec16959c5f104d549c3122bfcba6419e : SUCCESS in 1h 05m 37s\n- designate-grenade-pdns4-skip-level https://zuul.opendev.org/t/openstack/build/33bb8642389548d8931cc41ebbd4dca6 : SUCCESS in 1h 17m 00s\n- designate-bind9-core-catalog-zones https://zuul.opendev.org/t/openstack/build/726d61fc527d43f8a1fcebfeb130d05d : SUCCESS in 46m 23s (non-voting)\n- designate-only-ipv6-pdns4 https://zuul.opendev.org/t/openstack/build/c1ab118ae83e42dc9b58e8967d1df7c1 : SUCCESS in 43m 54s\n- designate-only-ipv6-bind9 https://zuul.opendev.org/t/openstack/build/436f51141ab045bb93b52f3293180c65 : SUCCESS in 49m 09s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a4fc678127d7436e99c0285a8590c272 : SUCCESS in 2m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3a3b9a98158243dc869175259e1f138c : SUCCESS in 3m 19s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/38b21345427146a9acacead06541a2cb : SUCCESS in 2m 55s\n- designate-bind9-stable-2026-2 https://zuul.opendev.org/t/openstack/build/a7082bd823d847389be5abfad60eddd9 : SUCCESS in 50m 33s\n- designate-bind9-stable-2026-1 https://zuul.opendev.org/t/openstack/build/276640452d424839b414b8eb2798286a : SUCCESS in 44m 09s\n- designate-bind9-stable-2025-2 https://zuul.opendev.org/t/openstack/build/b1f5d3917de145f784dcec41b68336ab : SUCCESS in 1h 02m 21s\n- designate-bind9-stable-2025-1 https://zuul.opendev.org/t/openstack/build/d9ae18fbc0b74dafbc57b8bf952d346d : SUCCESS in 53m 08s\n- designate-bind9-multipool-stable-2026-2 https://zuul.opendev.org/t/openstack/build/ce51e36d94da4d3a9adae09a9520e328 : SUCCESS in 35m 24s\n- designate-bind9-multipool-stable-2026-1 https://zuul.opendev.org/t/openstack/build/90ab5aba0d1a4c78a1915d8a2f61c55e : SUCCESS in 32m 09s\n- designate-bind9-multipool-stable-2025-2 https://zuul.opendev.org/t/openstack/build/22931909813441d8b13f5c33bf84ce6d : SUCCESS in 34m 07s\n- designate-bind9-multipool-stable-2025-1 https://zuul.opendev.org/t/openstack/build/5b2ca013195047c9ad83f3a0ecc7af9d : SUCCESS in 34m 25s\n- designate-bind9-split-horizon https://zuul.opendev.org/t/openstack/build/dbe57132f71b4567b5156da0f0b2b02d : SUCCESS in 32m 34s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/bf9a7b1a48934786803cab425c7e3e12 : SUCCESS in 30m 10s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"a34f83a38b45babc2b4766bd2e9e0eb332a31650","revisions":{"a34f83a38b45babc2b4766bd2e9e0eb332a31650":{"kind":"REWORK","_number":1,"created":"2026-09-11 17:08:42.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/11/1005311/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate-tempest-plugin","ref":"refs/changes/11/1005311/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate-tempest-plugin refs/changes/11/1005311/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate-tempest-plugin refs/changes/11/1005311/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate-tempest-plugin refs/changes/11/1005311/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate-tempest-plugin refs/changes/11/1005311/1"}}},"commit":{"parents":[{"commit":"4c69b6f219de7d6ad3b07be1f572edba64aab0c7","subject":"Merge \"Multi Pool - new scenario covering \u0027Move Zone API\u0027\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate-tempest-plugin/commit/4c69b6f219de7d6ad3b07be1f572edba64aab0c7"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-09-11 17:08:19.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-09-11 17:08:31.000000000","tz":540},"subject":"Add stable/2026.2 job","message":"Add stable/2026.2 job\n\nThe branch was created so we should start testing it.\n\nChange-Id: I70f0e9c23d2878c00b51004f60ce5d1c9a988ab3\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate-tempest-plugin/commit/a34f83a38b45babc2b4766bd2e9e0eb332a31650"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate-tempest-plugin/commit/a34f83a38b45babc2b4766bd2e9e0eb332a31650"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY"},{"label":"Workflow","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{"label:Review-Priority\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
