)]}'
{"id":"openstack%2Fdesignate-tempest-plugin~321807","triplet_id":"openstack%2Fdesignate-tempest-plugin~master~I71d406efc5f2c9c35eae50761446c00a328ec5a2","project":"openstack/designate-tempest-plugin","branch":"master","topic":"zone-import-tests","hashtags":[],"change_id":"I71d406efc5f2c9c35eae50761446c00a328ec5a2","subject":"Add a zone import scenario test","status":"MERGED","created":"2016-05-26 18:35:26.000000000","updated":"2016-06-07 22:15:43.000000000","submitted":"2016-06-07 22:15:43.000000000","submitter":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"1249592001bacda00a5fb3f988417617ecbe55a1","_number":321807,"virtual_id_number":321807,"owner":{"_account_id":11662,"name":"Paul Glass","email":"paul.glass@rackspace.com","username":"pglass"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"all":[{"value":0,"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},{"value":2,"date":"2016-06-07 22:15:43.000000000","post_submit":true,"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":0,"_account_id":741,"name":"Kiall Mac Innes","email":"kiall@macinnes.ie","username":"kiall"},{"value":0,"_account_id":11662,"name":"Paul Glass","email":"paul.glass@rackspace.com","username":"pglass"},{"value":0,"_account_id":8174,"name":"Tim Simmons","email":"tim.r.simmons@gmail.com","username":"TimSimmons"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},"all":[{"value":2,"date":"2016-06-07 21:30:11.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},{"value":0,"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":2,"date":"2016-06-07 21:26:09.000000000","_account_id":741,"name":"Kiall Mac Innes","email":"kiall@macinnes.ie","username":"kiall"},{"value":0,"_account_id":11662,"name":"Paul Glass","email":"paul.glass@rackspace.com","username":"pglass"},{"value":0,"_account_id":8174,"name":"Tim Simmons","email":"tim.r.simmons@gmail.com","username":"TimSimmons"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},"all":[{"value":1,"date":"2016-06-07 21:30:11.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},{"value":0,"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":0,"_account_id":741,"name":"Kiall Mac Innes","email":"kiall@macinnes.ie","username":"kiall"},{"value":0,"_account_id":11662,"name":"Paul Glass","email":"paul.glass@rackspace.com","username":"pglass"},{"value":0,"_account_id":8174,"name":"Tim Simmons","email":"tim.r.simmons@gmail.com","username":"TimSimmons"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},{"value":0,"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":0,"_account_id":741,"name":"Kiall Mac Innes","email":"kiall@macinnes.ie","username":"kiall"},{"value":0,"_account_id":11662,"name":"Paul Glass","email":"paul.glass@rackspace.com","username":"pglass"},{"value":0,"_account_id":8174,"name":"Tim Simmons","email":"tim.r.simmons@gmail.com","username":"TimSimmons"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":3,"name":"Jenkins","username":"jenkins"},{"_account_id":741,"name":"Kiall Mac Innes","email":"kiall@macinnes.ie","username":"kiall"},{"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},{"_account_id":8174,"name":"Tim Simmons","email":"tim.r.simmons@gmail.com","username":"TimSimmons"},{"_account_id":11662,"name":"Paul Glass","email":"paul.glass@rackspace.com","username":"pglass"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2016-06-03 19:54:21.000000000","updated_by":{"_account_id":8174,"name":"Tim Simmons","email":"tim.r.simmons@gmail.com","username":"TimSimmons"},"reviewer":{"_account_id":8174,"name":"Tim Simmons","email":"tim.r.simmons@gmail.com","username":"TimSimmons"},"state":"REVIEWER"},{"updated":"2016-06-07 21:26:09.000000000","updated_by":{"_account_id":741,"name":"Kiall Mac Innes","email":"kiall@macinnes.ie","username":"kiall"},"reviewer":{"_account_id":741,"name":"Kiall Mac Innes","email":"kiall@macinnes.ie","username":"kiall"},"state":"REVIEWER"},{"updated":"2016-06-07 21:30:11.000000000","updated_by":{"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},"reviewer":{"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},"state":"REVIEWER"},{"updated":"2016-06-07 22:15:43.000000000","updated_by":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"reviewer":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"state":"REVIEWER"}],"messages":[{"id":"5800bb5024dc57a64ae49f9aee83eead0e5dbc49","author":{"_account_id":11662,"name":"Paul Glass","email":"paul.glass@rackspace.com","username":"pglass"},"date":"2016-05-26 18:35:26.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"1b371a9b2bab885dacf26469d0b0b9fe8597b9fb","author":{"_account_id":11662,"name":"Paul Glass","email":"paul.glass@rackspace.com","username":"pglass"},"date":"2016-05-26 19:52:18.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"2e0fb7b1ce6fcb983fe448ca3fab98ed99242b42","author":{"_account_id":11662,"name":"Paul Glass","email":"paul.glass@rackspace.com","username":"pglass"},"date":"2016-05-26 20:01:32.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"976d1349b1fab025d2a2ba5949e28e97b679ad39","author":{"_account_id":11662,"name":"Paul Glass","email":"paul.glass@rackspace.com","username":"pglass"},"date":"2016-05-26 20:02:38.000000000","message":"Uploaded patch set 4: Commit message was updated.","accounts_in_message":[],"_revision_number":4},{"id":"242cd04e9fd4a7149bee7002dd53d0410df26501","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-05-26 21:33:51.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- gate-designate-dsvm-powerdns http://logs.openstack.org/07/321807/4/check/gate-designate-dsvm-powerdns/9fd015d/ : SUCCESS in 22m 53s\n- gate-designate-dsvm-bind9 http://logs.openstack.org/07/321807/4/check/gate-designate-dsvm-bind9/664dbb0/ : SUCCESS in 31m 09s\n- gate-tempest-dsvm-designate-powerdns http://logs.openstack.org/07/321807/4/check/gate-tempest-dsvm-designate-powerdns/b0f95f3/ : SUCCESS in 14m 28s\n- gate-tempest-dsvm-designate-bind9 http://logs.openstack.org/07/321807/4/check/gate-tempest-dsvm-designate-bind9/c784e79/ : SUCCESS in 13m 45s\n- gate-rally-dsvm-designate-designate http://logs.openstack.org/07/321807/4/check/gate-rally-dsvm-designate-designate/3a0be8f/ : FAILURE in 13m 43s (non-voting)\n- gate-designate-tempest-plugin-pep8 http://logs.openstack.org/07/321807/4/check/gate-designate-tempest-plugin-pep8/974a85b/ : SUCCESS in 1m 13s\n- gate-tempest-dsvm-designate-powerdns-mitaka http://logs.openstack.org/07/321807/4/check/gate-tempest-dsvm-designate-powerdns-mitaka/371d4c0/ : SUCCESS in 14m 31s\n- gate-tempest-dsvm-designate-bind9-mitaka http://logs.openstack.org/07/321807/4/check/gate-tempest-dsvm-designate-bind9-mitaka/3d9f9d4/ : SUCCESS in 13m 35s","accounts_in_message":[],"_revision_number":4},{"id":"4517d1b41945948b60577477b1c57c5afff7b7b9","author":{"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},"date":"2016-05-31 19:04:34.000000000","message":"Patch Set 4: Code-Review+2","accounts_in_message":[],"_revision_number":4},{"id":"3f09744c6cddc13cd39046613a4bc2299cc3511d","author":{"_account_id":741,"name":"Kiall Mac Innes","email":"kiall@macinnes.ie","username":"kiall"},"date":"2016-06-01 17:20:16.000000000","message":"Patch Set 4: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"1ef3f7fbd9e57cb0f7ea4308eb05436a66743dec","author":{"_account_id":11662,"name":"Paul Glass","email":"paul.glass@rackspace.com","username":"pglass"},"date":"2016-06-01 17:29:33.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"410c7bf2a2d003931c5e16542edf61c1142c6cd6","author":{"_account_id":8174,"name":"Tim Simmons","email":"tim.r.simmons@gmail.com","username":"TimSimmons"},"date":"2016-06-03 19:54:21.000000000","message":"Patch Set 4: Code-Review+2\n\nWill wait for Kiall\u0027s +A ;)","accounts_in_message":[],"_revision_number":4},{"id":"b74551e10ae2f6ec1e5ba0182f3f4d89ecf55f21","author":{"_account_id":741,"name":"Kiall Mac Innes","email":"kiall@macinnes.ie","username":"kiall"},"date":"2016-06-07 14:46:54.000000000","message":"Patch Set 4: Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"b4542f18511727a2d8d6006e6641bf4c4251bebd","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-06-07 14:48:43.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.\nhttp://status.openstack.org/zuul/","accounts_in_message":[],"_revision_number":4},{"id":"19a207bc635056b681998a5c1777135265c809f5","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-06-07 16:55:15.000000000","message":"Patch Set 4: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see http://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n- gate-designate-dsvm-powerdns http://logs.openstack.org/07/321807/4/gate/gate-designate-dsvm-powerdns/bcd24e5/ : SUCCESS in 34m 11s\n- gate-designate-dsvm-bind9 http://logs.openstack.org/07/321807/4/gate/gate-designate-dsvm-bind9/43dd411/ : SUCCESS in 31m 55s\n- gate-tempest-dsvm-designate-powerdns http://logs.openstack.org/07/321807/4/gate/gate-tempest-dsvm-designate-powerdns/7a33c9a/ : FAILURE in 32m 55s\n- gate-tempest-dsvm-designate-bind9 http://logs.openstack.org/07/321807/4/gate/gate-tempest-dsvm-designate-bind9/d3093b1/ : FAILURE in 25m 26s\n- gate-designate-tempest-plugin-pep8 http://logs.openstack.org/07/321807/4/gate/gate-designate-tempest-plugin-pep8/17d246b/ : SUCCESS in 3m 12s\n- gate-tempest-dsvm-designate-powerdns-mitaka http://logs.openstack.org/07/321807/4/gate/gate-tempest-dsvm-designate-powerdns-mitaka/682fdb2/ : FAILURE in 15m 49s\n- gate-tempest-dsvm-designate-bind9-mitaka http://logs.openstack.org/07/321807/4/gate/gate-tempest-dsvm-designate-bind9-mitaka/05c1cef/ : FAILURE in 53m 09s","accounts_in_message":[],"_revision_number":4},{"id":"f10c601b4d362a5bcc81c649ae66834a0bf8aac2","author":{"_account_id":11662,"name":"Paul Glass","email":"paul.glass@rackspace.com","username":"pglass"},"date":"2016-06-07 20:32:38.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"eb441476562572194c4ece9dc5ffc8c59cf25d43","author":{"_account_id":11662,"name":"Paul Glass","email":"paul.glass@rackspace.com","username":"pglass"},"date":"2016-06-07 20:35:03.000000000","message":"Patch Set 5:\n\nSomething merged between the time the verify and gate jobs ran that conflicted with the change I made to the base client. So rebased and fixed that.","accounts_in_message":[],"_revision_number":5},{"id":"4e95a9c67eed3263bf92711c341ab62f0cf00d57","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-06-07 21:23:47.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- gate-designate-dsvm-powerdns http://logs.openstack.org/07/321807/5/check/gate-designate-dsvm-powerdns/8f5ce6e/ : SUCCESS in 28m 11s\n- gate-designate-dsvm-bind9 http://logs.openstack.org/07/321807/5/check/gate-designate-dsvm-bind9/bde7c9a/ : SUCCESS in 36m 10s\n- gate-tempest-dsvm-designate-powerdns http://logs.openstack.org/07/321807/5/check/gate-tempest-dsvm-designate-powerdns/6640131/ : SUCCESS in 31m 26s\n- gate-tempest-dsvm-designate-bind9 http://logs.openstack.org/07/321807/5/check/gate-tempest-dsvm-designate-bind9/96b6fd1/ : SUCCESS in 20m 32s\n- gate-rally-dsvm-designate-designate http://logs.openstack.org/07/321807/5/check/gate-rally-dsvm-designate-designate/b38b8fd/ : FAILURE in 14m 55s (non-voting)\n- gate-designate-tempest-plugin-pep8 http://logs.openstack.org/07/321807/5/check/gate-designate-tempest-plugin-pep8/108a7cd/ : SUCCESS in 4m 03s\n- gate-tempest-dsvm-designate-powerdns-mitaka http://logs.openstack.org/07/321807/5/check/gate-tempest-dsvm-designate-powerdns-mitaka/983280f/ : SUCCESS in 29m 50s\n- gate-tempest-dsvm-designate-bind9-mitaka http://logs.openstack.org/07/321807/5/check/gate-tempest-dsvm-designate-bind9-mitaka/2146205/ : SUCCESS in 27m 04s","accounts_in_message":[],"_revision_number":5},{"id":"168b2f548ebc42823b39b185cba5223893ca01d4","author":{"_account_id":741,"name":"Kiall Mac Innes","email":"kiall@macinnes.ie","username":"kiall"},"date":"2016-06-07 21:26:09.000000000","message":"Patch Set 5: Code-Review+2","accounts_in_message":[],"_revision_number":5},{"id":"362712232979ad1ad75c06696a489d3c810b3deb","author":{"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},"date":"2016-06-07 21:30:11.000000000","message":"Patch Set 5: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":5},{"id":"a6ccb1f541ff0aceb2196bf9abf172d14050f8ae","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-06-07 21:30:40.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.\nhttp://status.openstack.org/zuul/","accounts_in_message":[],"_revision_number":5},{"id":"215beecd65f6b07595015683f8da5f9667b8dca6","date":"2016-06-07 22:15:43.000000000","message":"Change has been successfully merged into the git repository by Jenkins","accounts_in_message":[],"_revision_number":5},{"id":"bff4327cf5ab5718fcd061c1f7869ad5c5b74957","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-06-07 22:15:43.000000000","message":"Patch Set 5: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- gate-designate-dsvm-powerdns http://logs.openstack.org/07/321807/5/gate/gate-designate-dsvm-powerdns/4e75076/ : SUCCESS in 33m 21s\n- gate-designate-dsvm-bind9 http://logs.openstack.org/07/321807/5/gate/gate-designate-dsvm-bind9/8834ab4/ : SUCCESS in 34m 00s\n- gate-tempest-dsvm-designate-powerdns http://logs.openstack.org/07/321807/5/gate/gate-tempest-dsvm-designate-powerdns/469baf3/ : SUCCESS in 29m 58s\n- gate-tempest-dsvm-designate-bind9 http://logs.openstack.org/07/321807/5/gate/gate-tempest-dsvm-designate-bind9/48049a1/ : SUCCESS in 26m 18s\n- gate-designate-tempest-plugin-pep8 http://logs.openstack.org/07/321807/5/gate/gate-designate-tempest-plugin-pep8/3b062e1/ : SUCCESS in 3m 33s\n- gate-tempest-dsvm-designate-powerdns-mitaka http://logs.openstack.org/07/321807/5/gate/gate-tempest-dsvm-designate-powerdns-mitaka/2de9882/ : SUCCESS in 27m 24s\n- gate-tempest-dsvm-designate-bind9-mitaka http://logs.openstack.org/07/321807/5/gate/gate-tempest-dsvm-designate-bind9-mitaka/9aec05d/ : SUCCESS in 25m 43s","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"6e0e80ec148af484ae63fa9e86be958c3f26bf05","revisions":{"49899e166e81d252c957e7d510c36235970a6e5e":{"kind":"REWORK","_number":1,"created":"2016-05-26 18:35:26.000000000","uploader":{"_account_id":11662,"name":"Paul Glass","email":"paul.glass@rackspace.com","username":"pglass"},"ref":"refs/changes/07/321807/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate-tempest-plugin","ref":"refs/changes/07/321807/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate-tempest-plugin refs/changes/07/321807/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate-tempest-plugin refs/changes/07/321807/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate-tempest-plugin refs/changes/07/321807/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate-tempest-plugin refs/changes/07/321807/1"}}},"commit":{"parents":[{"commit":"74948a57e251c32194e2a50cb49b2f97c47c9959","subject":"Merge \"Replace assertEqual(a\u003eb) with  assertGreater(a, b)\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate-tempest-plugin/commit/74948a57e251c32194e2a50cb49b2f97c47c9959"}]}],"author":{"name":"Paul Glass","email":"paul.glass@rackspace.com","date":"2016-05-26 18:27:49.000000000","tz":0},"committer":{"name":"Paul Glass","email":"paul.glass@rackspace.com","date":"2016-05-26 18:32:30.000000000","tz":0},"subject":"WIP: Add zone import scenario tests","message":"WIP: Add zone import scenario tests\n\nChange-Id: I71d406efc5f2c9c35eae50761446c00a328ec5a2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate-tempest-plugin/commit/49899e166e81d252c957e7d510c36235970a6e5e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate-tempest-plugin/commit/49899e166e81d252c957e7d510c36235970a6e5e"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"74948a57e251c32194e2a50cb49b2f97c47c9959","is_merged_in_target_branch":true}],"branch":"refs/heads/master"},"52d105936448e4c22210c0def22602e2d0e8b8fc":{"kind":"REWORK","_number":2,"created":"2016-05-26 19:52:18.000000000","uploader":{"_account_id":11662,"name":"Paul Glass","email":"paul.glass@rackspace.com","username":"pglass"},"ref":"refs/changes/07/321807/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate-tempest-plugin","ref":"refs/changes/07/321807/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate-tempest-plugin refs/changes/07/321807/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate-tempest-plugin refs/changes/07/321807/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate-tempest-plugin refs/changes/07/321807/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate-tempest-plugin refs/changes/07/321807/2"}}},"commit":{"parents":[{"commit":"74948a57e251c32194e2a50cb49b2f97c47c9959","subject":"Merge \"Replace assertEqual(a\u003eb) with  assertGreater(a, b)\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate-tempest-plugin/commit/74948a57e251c32194e2a50cb49b2f97c47c9959"}]}],"author":{"name":"Paul Glass","email":"paul.glass@rackspace.com","date":"2016-05-26 18:27:49.000000000","tz":0},"committer":{"name":"Paul Glass","email":"paul.glass@rackspace.com","date":"2016-05-26 19:48:51.000000000","tz":0},"subject":"Add a zone import scenario test","message":"Add a zone import scenario test\n\nThis adds a test that imports a zone and checks the zone is created and\ngoes to active.\n\nThe base client\u0027s create/update calls now accept either a dict or string\nas request data. Serialization is skipped if the data is a string. This\nfixes an bug in the tests causing all zone import\u0027s to go to error.\n\nChange-Id: I71d406efc5f2c9c35eae50761446c00a328ec5a2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate-tempest-plugin/commit/52d105936448e4c22210c0def22602e2d0e8b8fc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate-tempest-plugin/commit/52d105936448e4c22210c0def22602e2d0e8b8fc"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"74948a57e251c32194e2a50cb49b2f97c47c9959","is_merged_in_target_branch":true}],"branch":"refs/heads/master"},"ac29d24fb9171fc615d496db9d1bffd90735025c":{"kind":"REWORK","_number":3,"created":"2016-05-26 20:01:32.000000000","uploader":{"_account_id":11662,"name":"Paul Glass","email":"paul.glass@rackspace.com","username":"pglass"},"ref":"refs/changes/07/321807/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate-tempest-plugin","ref":"refs/changes/07/321807/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate-tempest-plugin refs/changes/07/321807/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate-tempest-plugin refs/changes/07/321807/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate-tempest-plugin refs/changes/07/321807/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate-tempest-plugin refs/changes/07/321807/3"}}},"commit":{"parents":[{"commit":"74948a57e251c32194e2a50cb49b2f97c47c9959","subject":"Merge \"Replace assertEqual(a\u003eb) with  assertGreater(a, b)\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate-tempest-plugin/commit/74948a57e251c32194e2a50cb49b2f97c47c9959"}]}],"author":{"name":"Paul Glass","email":"paul.glass@rackspace.com","date":"2016-05-26 18:27:49.000000000","tz":0},"committer":{"name":"Paul Glass","email":"paul.glass@rackspace.com","date":"2016-05-26 20:01:13.000000000","tz":0},"subject":"Add a zone import scenario test","message":"Add a zone import scenario test\n\nThis adds a test that imports a zone and checks the zone is created and\ngoes to active.\n\nThe base client\u0027s create/update calls now accept either a dict or string\nas request data. Serialization is skipped if the data is a string. This\nfixes an bug in the tests causing all zone import\u0027s to go to error.\n\nChange-Id: I71d406efc5f2c9c35eae50761446c00a328ec5a2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate-tempest-plugin/commit/ac29d24fb9171fc615d496db9d1bffd90735025c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate-tempest-plugin/commit/ac29d24fb9171fc615d496db9d1bffd90735025c"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"74948a57e251c32194e2a50cb49b2f97c47c9959","is_merged_in_target_branch":true}],"branch":"refs/heads/master"},"b51bfd75c19d46798a205c0753d7ac1b57e82cc1":{"kind":"NO_CODE_CHANGE","_number":4,"created":"2016-05-26 20:02:38.000000000","uploader":{"_account_id":11662,"name":"Paul Glass","email":"paul.glass@rackspace.com","username":"pglass"},"ref":"refs/changes/07/321807/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate-tempest-plugin","ref":"refs/changes/07/321807/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate-tempest-plugin refs/changes/07/321807/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate-tempest-plugin refs/changes/07/321807/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate-tempest-plugin refs/changes/07/321807/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate-tempest-plugin refs/changes/07/321807/4"}}},"commit":{"parents":[{"commit":"74948a57e251c32194e2a50cb49b2f97c47c9959","subject":"Merge \"Replace assertEqual(a\u003eb) with  assertGreater(a, b)\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate-tempest-plugin/commit/74948a57e251c32194e2a50cb49b2f97c47c9959"}]}],"author":{"name":"Paul Glass","email":"paul.glass@rackspace.com","date":"2016-05-26 18:27:49.000000000","tz":0},"committer":{"name":"Paul Glass","email":"paul.glass@rackspace.com","date":"2016-05-26 20:02:37.000000000","tz":0},"subject":"Add a zone import scenario test","message":"Add a zone import scenario test\n\nThis adds a test that imports a zone and checks the zone is created and\ngoes to active.\n\nThe base client\u0027s create/update calls now accept either a dict or string\nas request data. Serialization is skipped if the data is a string. This\nfixes a bug in the tests causing all zone import\u0027s to go to error.\n\nChange-Id: I71d406efc5f2c9c35eae50761446c00a328ec5a2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate-tempest-plugin/commit/b51bfd75c19d46798a205c0753d7ac1b57e82cc1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate-tempest-plugin/commit/b51bfd75c19d46798a205c0753d7ac1b57e82cc1"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"74948a57e251c32194e2a50cb49b2f97c47c9959","is_merged_in_target_branch":true}],"branch":"refs/heads/master"},"6e0e80ec148af484ae63fa9e86be958c3f26bf05":{"kind":"REWORK","_number":5,"created":"2016-06-07 20:32:38.000000000","uploader":{"_account_id":11662,"name":"Paul Glass","email":"paul.glass@rackspace.com","username":"pglass"},"ref":"refs/changes/07/321807/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate-tempest-plugin","ref":"refs/changes/07/321807/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate-tempest-plugin refs/changes/07/321807/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate-tempest-plugin refs/changes/07/321807/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate-tempest-plugin refs/changes/07/321807/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate-tempest-plugin refs/changes/07/321807/5"}}},"commit":{"parents":[{"commit":"84f0aa501eb544b6ab56ea2bbde2a44a7d9f461a","subject":"Merge \"Add a zone export scenario test\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate-tempest-plugin/commit/84f0aa501eb544b6ab56ea2bbde2a44a7d9f461a"}]}],"author":{"name":"Paul Glass","email":"paul.glass@rackspace.com","date":"2016-05-26 18:27:49.000000000","tz":0},"committer":{"name":"Paul Glass","email":"paul.glass@rackspace.com","date":"2016-06-07 20:32:25.000000000","tz":0},"subject":"Add a zone import scenario test","message":"Add a zone import scenario test\n\nThis adds a test that imports a zone and checks the zone is created and\ngoes to active.\n\nThe base client\u0027s create/update calls now accept either a dict or string\nas request data. Serialization is skipped if the data is a string. This\nfixes a bug in the tests causing all zone import\u0027s to go to error.\n\nChange-Id: I71d406efc5f2c9c35eae50761446c00a328ec5a2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate-tempest-plugin/commit/6e0e80ec148af484ae63fa9e86be958c3f26bf05"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate-tempest-plugin/commit/6e0e80ec148af484ae63fa9e86be958c3f26bf05"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"84f0aa501eb544b6ab56ea2bbde2a44a7d9f461a","is_merged_in_target_branch":true}],"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
