)]}'
{"id":"openstack%2Fdesignate-tempest-plugin~329187","triplet_id":"openstack%2Fdesignate-tempest-plugin~master~I9752685afccee5dcd009f76b5242fd6f12f0fc37","project":"openstack/designate-tempest-plugin","branch":"master","topic":"migrate-unauthed-tests","hashtags":[],"change_id":"I9752685afccee5dcd009f76b5242fd6f12f0fc37","subject":"Migrate the \"unauthed\" tests","status":"MERGED","created":"2016-06-13 21:04:27.000000000","updated":"2016-06-14 14:20:38.000000000","submitted":"2016-06-14 14:20:38.000000000","submitter":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"a99434f05b419ba358529d7ea580750f8d2e4e9f","_number":329187,"virtual_id_number":329187,"owner":{"_account_id":11662,"name":"Paul Glass","email":"paul.glass@rackspace.com","username":"pglass"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"all":[{"value":0,"_account_id":15699,"name":"sonu","email":"sonu.bhumca11@gmail.com","username":"sonu.kumar"},{"value":0,"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},{"value":2,"date":"2016-06-14 14:20:38.000000000","post_submit":true,"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":0,"_account_id":741,"name":"Kiall Mac Innes","email":"kiall@macinnes.ie","username":"kiall"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},"recommended":{"_account_id":15699,"name":"sonu","email":"sonu.bhumca11@gmail.com","username":"sonu.kumar"},"all":[{"value":1,"date":"2016-06-14 11:10:21.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15699,"name":"sonu","email":"sonu.bhumca11@gmail.com","username":"sonu.kumar"},{"value":2,"date":"2016-06-14 12:31:05.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},{"value":0,"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":0,"_account_id":741,"name":"Kiall Mac Innes","email":"kiall@macinnes.ie","username":"kiall"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":741,"name":"Kiall Mac Innes","email":"kiall@macinnes.ie","username":"kiall"},"all":[{"value":0,"_account_id":15699,"name":"sonu","email":"sonu.bhumca11@gmail.com","username":"sonu.kumar"},{"value":0,"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},{"value":0,"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":1,"date":"2016-06-14 13:37:04.000000000","_account_id":741,"name":"Kiall Mac Innes","email":"kiall@macinnes.ie","username":"kiall"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":15699,"name":"sonu","email":"sonu.bhumca11@gmail.com","username":"sonu.kumar"},{"value":0,"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},{"value":0,"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":0,"_account_id":741,"name":"Kiall Mac Innes","email":"kiall@macinnes.ie","username":"kiall"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":3,"name":"Jenkins","username":"jenkins"},{"_account_id":741,"name":"Kiall Mac Innes","email":"kiall@macinnes.ie","username":"kiall"},{"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},{"_account_id":15699,"name":"sonu","email":"sonu.bhumca11@gmail.com","username":"sonu.kumar"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2016-06-14 11:10:21.000000000","updated_by":{"_account_id":15699,"name":"sonu","email":"sonu.bhumca11@gmail.com","username":"sonu.kumar"},"reviewer":{"_account_id":15699,"name":"sonu","email":"sonu.bhumca11@gmail.com","username":"sonu.kumar"},"state":"REVIEWER"},{"updated":"2016-06-14 12:31:05.000000000","updated_by":{"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},"reviewer":{"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},"state":"REVIEWER"},{"updated":"2016-06-14 13:37:04.000000000","updated_by":{"_account_id":741,"name":"Kiall Mac Innes","email":"kiall@macinnes.ie","username":"kiall"},"reviewer":{"_account_id":741,"name":"Kiall Mac Innes","email":"kiall@macinnes.ie","username":"kiall"},"state":"REVIEWER"},{"updated":"2016-06-14 14:20:38.000000000","updated_by":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"reviewer":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"state":"REVIEWER"}],"messages":[{"id":"ba30b51567f7e5733422664436acb89becb7bdaa","author":{"_account_id":11662,"name":"Paul Glass","email":"paul.glass@rackspace.com","username":"pglass"},"date":"2016-06-13 21:04:27.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"e5e953d8cdc73d62f2b77efa42a3fbad14d1f730","author":{"_account_id":11662,"name":"Paul Glass","email":"paul.glass@rackspace.com","username":"pglass"},"date":"2016-06-13 21:13:38.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"21319dc7079a64a26e025903eee5ae058648e779","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-06-13 21:57:36.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see http://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n- gate-designate-dsvm-powerdns http://logs.openstack.org/87/329187/2/check/gate-designate-dsvm-powerdns/7dfa794/ : SUCCESS in 34m 48s\n- gate-designate-dsvm-bind9 http://logs.openstack.org/87/329187/2/check/gate-designate-dsvm-bind9/825a45d/ : SUCCESS in 32m 47s\n- gate-tempest-dsvm-designate-powerdns http://logs.openstack.org/87/329187/2/check/gate-tempest-dsvm-designate-powerdns/de2a4b6/ : FAILURE in 32m 17s\n- gate-tempest-dsvm-designate-bind9 http://logs.openstack.org/87/329187/2/check/gate-tempest-dsvm-designate-bind9/3d84d5f/ : FAILURE in 27m 17s\n- gate-rally-dsvm-designate-designate http://logs.openstack.org/87/329187/2/check/gate-rally-dsvm-designate-designate/fb1f722/ : FAILURE in 26m 46s (non-voting)\n- gate-designate-tempest-plugin-pep8 http://logs.openstack.org/87/329187/2/check/gate-designate-tempest-plugin-pep8/d5cf407/ : SUCCESS in 1m 30s\n- gate-tempest-dsvm-designate-powerdns-mitaka http://logs.openstack.org/87/329187/2/check/gate-tempest-dsvm-designate-powerdns-mitaka/d4af2cc/ : FAILURE in 14m 35s\n- gate-tempest-dsvm-designate-bind9-mitaka http://logs.openstack.org/87/329187/2/check/gate-tempest-dsvm-designate-bind9-mitaka/1ddf20c/ : FAILURE in 13m 46s","accounts_in_message":[],"_revision_number":2},{"id":"e03e3731f269947865736aef9569839a7d52d917","author":{"_account_id":11662,"name":"Paul Glass","email":"paul.glass@rackspace.com","username":"pglass"},"date":"2016-06-13 22:15:27.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"38a4a9578a699a44cec8c95cab3ae05320b18972","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-06-13 22:45:08.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- gate-designate-dsvm-powerdns http://logs.openstack.org/87/329187/3/check/gate-designate-dsvm-powerdns/5168f8d/ : SUCCESS in 21m 49s\n- gate-designate-dsvm-bind9 http://logs.openstack.org/87/329187/3/check/gate-designate-dsvm-bind9/7321e97/ : SUCCESS in 26m 25s\n- gate-tempest-dsvm-designate-powerdns http://logs.openstack.org/87/329187/3/check/gate-tempest-dsvm-designate-powerdns/da00a5d/ : SUCCESS in 20m 49s\n- gate-tempest-dsvm-designate-bind9 http://logs.openstack.org/87/329187/3/check/gate-tempest-dsvm-designate-bind9/1862b9a/ : SUCCESS in 19m 48s\n- gate-rally-dsvm-designate-designate http://logs.openstack.org/87/329187/3/check/gate-rally-dsvm-designate-designate/764084a/ : FAILURE in 24m 57s (non-voting)\n- gate-designate-tempest-plugin-pep8 http://logs.openstack.org/87/329187/3/check/gate-designate-tempest-plugin-pep8/3417b18/ : SUCCESS in 1m 58s\n- gate-tempest-dsvm-designate-powerdns-mitaka http://logs.openstack.org/87/329187/3/check/gate-tempest-dsvm-designate-powerdns-mitaka/866ebb8/ : SUCCESS in 25m 27s\n- gate-tempest-dsvm-designate-bind9-mitaka http://logs.openstack.org/87/329187/3/check/gate-tempest-dsvm-designate-bind9-mitaka/d88eb57/ : SUCCESS in 25m 01s","accounts_in_message":[],"_revision_number":3},{"id":"f90b8145efec8e96ed6712af68dacf504784a335","author":{"_account_id":15699,"name":"sonu","email":"sonu.bhumca11@gmail.com","username":"sonu.kumar"},"date":"2016-06-14 11:10:21.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"4a59a6a77089b062586b4bdff7f3cb326ebdda73","author":{"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},"date":"2016-06-14 12:31:05.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"b1deb62f3de57cebc2fdb5a890d04a8c1d504d72","author":{"_account_id":741,"name":"Kiall Mac Innes","email":"kiall@macinnes.ie","username":"kiall"},"date":"2016-06-14 13:37:04.000000000","message":"Patch Set 3: Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"9a462d20f430309ab81734b9315365690e4a761a","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-06-14 13:43:36.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.\nhttp://status.openstack.org/zuul/","accounts_in_message":[],"_revision_number":3},{"id":"b8f68628dbd981b1c30da88b5f6fced98fda8200","date":"2016-06-14 14:20:38.000000000","message":"Change has been successfully merged into the git repository by Jenkins","accounts_in_message":[],"_revision_number":3},{"id":"c898202d0472cf751bb740711085a988e9465a85","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-06-14 14:20:38.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- gate-designate-dsvm-powerdns http://logs.openstack.org/87/329187/3/gate/gate-designate-dsvm-powerdns/041b54b/ : SUCCESS in 34m 20s\n- gate-designate-dsvm-bind9 http://logs.openstack.org/87/329187/3/gate/gate-designate-dsvm-bind9/b440e1c/ : SUCCESS in 22m 33s\n- gate-tempest-dsvm-designate-powerdns http://logs.openstack.org/87/329187/3/gate/gate-tempest-dsvm-designate-powerdns/4187769/ : SUCCESS in 28m 28s\n- gate-tempest-dsvm-designate-bind9 http://logs.openstack.org/87/329187/3/gate/gate-tempest-dsvm-designate-bind9/985c861/ : SUCCESS in 32m 55s\n- gate-designate-tempest-plugin-pep8 http://logs.openstack.org/87/329187/3/gate/gate-designate-tempest-plugin-pep8/c3d62cb/ : SUCCESS in 1m 35s\n- gate-tempest-dsvm-designate-powerdns-mitaka http://logs.openstack.org/87/329187/3/gate/gate-tempest-dsvm-designate-powerdns-mitaka/e4a5e6a/ : SUCCESS in 14m 53s\n- gate-tempest-dsvm-designate-bind9-mitaka http://logs.openstack.org/87/329187/3/gate/gate-tempest-dsvm-designate-bind9-mitaka/02c1e2a/ : SUCCESS in 14m 06s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"d56330ab97edb49faa1488a15ebbc8a751946ed3","revisions":{"47a75e27f601965b4e3f5799f49459797659ae2e":{"kind":"REWORK","_number":1,"created":"2016-06-13 21:04:27.000000000","uploader":{"_account_id":11662,"name":"Paul Glass","email":"paul.glass@rackspace.com","username":"pglass"},"ref":"refs/changes/87/329187/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate-tempest-plugin","ref":"refs/changes/87/329187/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate-tempest-plugin refs/changes/87/329187/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate-tempest-plugin refs/changes/87/329187/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate-tempest-plugin refs/changes/87/329187/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate-tempest-plugin refs/changes/87/329187/1"}}},"commit":{"parents":[{"commit":"24df485be70f8a1d96e10764b1000d0e134c6ce2","subject":"Replace idempotent_id decorator with tempest.lib version","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate-tempest-plugin/commit/24df485be70f8a1d96e10764b1000d0e134c6ce2"}]}],"author":{"name":"Paul Glass","email":"paul.glass@rackspace.com","date":"2016-06-13 21:03:33.000000000","tz":0},"committer":{"name":"Paul Glass","email":"paul.glass@rackspace.com","date":"2016-06-13 21:04:00.000000000","tz":0},"subject":"Migrate the \"unauthed\" tests","message":"Migrate the \"unauthed\" tests\n\nThis migrates the tests that check for a 401 if a token is not provided\nin the request headers.\n\nChange-Id: I9752685afccee5dcd009f76b5242fd6f12f0fc37\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate-tempest-plugin/commit/47a75e27f601965b4e3f5799f49459797659ae2e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate-tempest-plugin/commit/47a75e27f601965b4e3f5799f49459797659ae2e"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"24df485be70f8a1d96e10764b1000d0e134c6ce2","is_merged_in_target_branch":true,"change_id":"Ida2efd9fdd9ce095aaa33ce41c90f93efbcf0970","change_number":320017,"patch_set_number":3,"change_status":"MERGED"}],"branch":"refs/heads/master"},"1ad4c9caccb35671fc0dbf38ae9a4385f7913d8f":{"kind":"REWORK","_number":2,"created":"2016-06-13 21:13:38.000000000","uploader":{"_account_id":11662,"name":"Paul Glass","email":"paul.glass@rackspace.com","username":"pglass"},"ref":"refs/changes/87/329187/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate-tempest-plugin","ref":"refs/changes/87/329187/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate-tempest-plugin refs/changes/87/329187/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate-tempest-plugin refs/changes/87/329187/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate-tempest-plugin refs/changes/87/329187/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate-tempest-plugin refs/changes/87/329187/2"}}},"commit":{"parents":[{"commit":"24df485be70f8a1d96e10764b1000d0e134c6ce2","subject":"Replace idempotent_id decorator with tempest.lib version","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate-tempest-plugin/commit/24df485be70f8a1d96e10764b1000d0e134c6ce2"}]}],"author":{"name":"Paul Glass","email":"paul.glass@rackspace.com","date":"2016-06-13 21:03:33.000000000","tz":0},"committer":{"name":"Paul Glass","email":"paul.glass@rackspace.com","date":"2016-06-13 21:13:47.000000000","tz":0},"subject":"Migrate the \"unauthed\" tests","message":"Migrate the \"unauthed\" tests\n\nThis migrates the tests that check for a 401 if a token is not provided\nin the request headers.\n\nChange-Id: I9752685afccee5dcd009f76b5242fd6f12f0fc37\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate-tempest-plugin/commit/1ad4c9caccb35671fc0dbf38ae9a4385f7913d8f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate-tempest-plugin/commit/1ad4c9caccb35671fc0dbf38ae9a4385f7913d8f"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"24df485be70f8a1d96e10764b1000d0e134c6ce2","is_merged_in_target_branch":true,"change_id":"Ida2efd9fdd9ce095aaa33ce41c90f93efbcf0970","change_number":320017,"patch_set_number":3,"change_status":"MERGED"}],"branch":"refs/heads/master"},"d56330ab97edb49faa1488a15ebbc8a751946ed3":{"kind":"REWORK","_number":3,"created":"2016-06-13 22:15:27.000000000","uploader":{"_account_id":11662,"name":"Paul Glass","email":"paul.glass@rackspace.com","username":"pglass"},"ref":"refs/changes/87/329187/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate-tempest-plugin","ref":"refs/changes/87/329187/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate-tempest-plugin refs/changes/87/329187/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate-tempest-plugin refs/changes/87/329187/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate-tempest-plugin refs/changes/87/329187/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate-tempest-plugin refs/changes/87/329187/3"}}},"commit":{"parents":[{"commit":"24df485be70f8a1d96e10764b1000d0e134c6ce2","subject":"Replace idempotent_id decorator with tempest.lib version","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate-tempest-plugin/commit/24df485be70f8a1d96e10764b1000d0e134c6ce2"}]}],"author":{"name":"Paul Glass","email":"paul.glass@rackspace.com","date":"2016-06-13 21:03:33.000000000","tz":0},"committer":{"name":"Paul Glass","email":"paul.glass@rackspace.com","date":"2016-06-13 22:15:38.000000000","tz":0},"subject":"Migrate the \"unauthed\" tests","message":"Migrate the \"unauthed\" tests\n\nThis migrates the tests that check for a 401 if a token is not provided\nin the request headers.\n\nChange-Id: I9752685afccee5dcd009f76b5242fd6f12f0fc37\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate-tempest-plugin/commit/d56330ab97edb49faa1488a15ebbc8a751946ed3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate-tempest-plugin/commit/d56330ab97edb49faa1488a15ebbc8a751946ed3"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"24df485be70f8a1d96e10764b1000d0e134c6ce2","is_merged_in_target_branch":true,"change_id":"Ida2efd9fdd9ce095aaa33ce41c90f93efbcf0970","change_number":320017,"patch_set_number":3,"change_status":"MERGED"}],"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
