)]}'
{"id":"openstack%2Fdesignate-tempest-plugin~509147","triplet_id":"openstack%2Fdesignate-tempest-plugin~master~I496d02c1be165eeaf721dea2eafe5f6a70279e68","project":"openstack/designate-tempest-plugin","branch":"master","topic":"rm-v1-tests","hashtags":[],"change_id":"I496d02c1be165eeaf721dea2eafe5f6a70279e68","subject":"Remove v1 API tests","status":"MERGED","created":"2017-10-03 11:00:52.000000000","updated":"2021-09-24 05:16:30.000000000","submitted":"2021-09-24 05:15:16.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"509147-rm-v1-tests","meta_rev_id":"c94214fb6b89732b5d8aa6ea25e696b9a4ad4364","_number":509147,"virtual_id_number":509147,"owner":{"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},{"value":0,"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":0,"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-09-24 05:15:16.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},"all":[{"value":2,"date":"2021-09-23 18:43:09.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},{"value":0,"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":0,"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},{"value":2,"date":"2021-09-23 18:39:29.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2021-08-06 15:43:28.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"all":[{"value":0,"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},{"value":0,"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":0,"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":1,"date":"2021-09-23 20:55:05.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"recommended":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":0,"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},{"value":0,"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":0,"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},{"value":1,"date":"2021-09-23 18:39:29.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","value":1,"default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":3,"name":"Jenkins","username":"jenkins"},{"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2017-10-04 12:10:31.000000000","updated_by":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"reviewer":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"state":"REVIEWER"},{"updated":"2017-11-17 18:06:54.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2017-11-26 10:21:05.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"REVIEWER"},{"updated":"2021-07-06 22:33:57.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"},{"updated":"2021-08-06 15:43:28.000000000","updated_by":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"reviewer":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"state":"REVIEWER"},{"updated":"2021-09-23 18:43:09.000000000","updated_by":{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},"reviewer":{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},"state":"REVIEWER"}],"messages":[{"id":"d891200f2ba8ea540f6d3914702d1e838ef93c37","author":{"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},"date":"2017-10-03 11:00:52.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"63bf3f4e93c5a2712f61c117f81d3a27a313d0de","author":{"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},"date":"2017-10-04 12:05:35.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"d30751f1ac06a3d389dba589b47d6c1516baadf6","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2017-10-04 12:10:31.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- gate-designate-tempest-plugin-pep8-ubuntu-xenial http://logs.openstack.org/47/509147/1/check/gate-designate-tempest-plugin-pep8-ubuntu-xenial/75d5b13/ : SUCCESS in 3m 24s\n- gate-designate-tempest-plugin-docs-ubuntu-xenial http://docs-draft.openstack.org/47/509147/1/check/gate-designate-tempest-plugin-docs-ubuntu-xenial/67bee0a//doc/build/html/ : SUCCESS in 2m 33s","accounts_in_message":[],"_revision_number":1},{"id":"f624884c9c45bfa66cbd6167e37f77e3d3183161","author":{"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},"date":"2017-11-17 12:38:15.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"954fabf44863898f2c90347f20cb338f50bc0f36","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2017-11-17 13:57:34.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- legacy-tempest-dsvm-designate-pdns4 http://logs.openstack.org/47/509147/1/check/legacy-tempest-dsvm-designate-pdns4/0a29eef/ : SUCCESS in 30m 25s\n- legacy-tempest-dsvm-designate-bind9 http://logs.openstack.org/47/509147/1/check/legacy-tempest-dsvm-designate-bind9/f8df00c/ : SUCCESS in 29m 11s\n- legacy-tempest-dsvm-designate-pdns4-postgres http://logs.openstack.org/47/509147/1/check/legacy-tempest-dsvm-designate-pdns4-postgres/5af4c67/ : SUCCESS in 32m 28s\n- legacy-tempest-dsvm-designate-pdns4-worker-model http://logs.openstack.org/47/509147/1/check/legacy-tempest-dsvm-designate-pdns4-worker-model/894a46a/ : SUCCESS in 29m 28s\n- legacy-rally-dsvm-designate-designate-pdns4 http://logs.openstack.org/47/509147/1/check/legacy-rally-dsvm-designate-designate-pdns4/ff731c9/ : FAILURE in 25m 52s (non-voting)\n- legacy-grenade-dsvm-designate-pdns4 http://logs.openstack.org/47/509147/1/check/legacy-grenade-dsvm-designate-pdns4/d5f8971/ : SUCCESS in 1h 17m 30s\n- legacy-tempest-dsvm-designate-pdns4-identity-v3-only http://logs.openstack.org/47/509147/1/check/legacy-tempest-dsvm-designate-pdns4-identity-v3-only/30b06ad/ : SUCCESS in 38m 25s\n- legacy-tempest-dsvm-designate-pdns4-py35 http://logs.openstack.org/47/509147/1/check/legacy-tempest-dsvm-designate-pdns4-py35/2baa0eb/ : SUCCESS in 32m 59s (non-voting)\n- build-openstack-sphinx-docs http://logs.openstack.org/47/509147/1/check/build-openstack-sphinx-docs/7d26a05/html/ : SUCCESS in 4m 12s\n- openstack-tox-pep8 http://logs.openstack.org/47/509147/1/check/openstack-tox-pep8/837824a/ : SUCCESS in 4m 17s","accounts_in_message":[],"_revision_number":1},{"id":"071464e57eb48ace8fda8740d0f12635c2924933","author":{"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},"date":"2017-11-17 16:58:24.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"39475fc63ed3c98c8c48f6c134288cfaf2bce751","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2017-11-17 18:06:54.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- legacy-tempest-dsvm-designate-pdns4 http://logs.openstack.org/47/509147/2/check/legacy-tempest-dsvm-designate-pdns4/30ec7de/ : SUCCESS in 32m 00s\n- legacy-tempest-dsvm-designate-bind9 http://logs.openstack.org/47/509147/2/check/legacy-tempest-dsvm-designate-bind9/eaa7f98/ : SUCCESS in 27m 27s\n- legacy-tempest-dsvm-designate-pdns4-postgres http://logs.openstack.org/47/509147/2/check/legacy-tempest-dsvm-designate-pdns4-postgres/3203abc/ : SUCCESS in 32m 27s\n- legacy-tempest-dsvm-designate-pdns4-worker-model http://logs.openstack.org/47/509147/2/check/legacy-tempest-dsvm-designate-pdns4-worker-model/d7d7ab3/ : SUCCESS in 28m 17s\n- legacy-rally-dsvm-designate-designate-pdns4 http://logs.openstack.org/47/509147/2/check/legacy-rally-dsvm-designate-designate-pdns4/b2f3350/ : FAILURE in 26m 23s (non-voting)\n- legacy-grenade-dsvm-designate-pdns4 http://logs.openstack.org/47/509147/2/check/legacy-grenade-dsvm-designate-pdns4/d209a11/ : SUCCESS in 59m 52s\n- legacy-tempest-dsvm-designate-pdns4-identity-v3-only http://logs.openstack.org/47/509147/2/check/legacy-tempest-dsvm-designate-pdns4-identity-v3-only/4dc410c/ : SUCCESS in 27m 37s\n- legacy-tempest-dsvm-designate-pdns4-py35 http://logs.openstack.org/47/509147/2/check/legacy-tempest-dsvm-designate-pdns4-py35/3344f4f/ : SUCCESS in 34m 06s (non-voting)\n- build-openstack-sphinx-docs http://logs.openstack.org/47/509147/2/check/build-openstack-sphinx-docs/53d83c7/html/ : SUCCESS in 4m 01s\n- openstack-tox-pep8 http://logs.openstack.org/47/509147/2/check/openstack-tox-pep8/c61d70f/ : SUCCESS in 4m 51s","accounts_in_message":[],"_revision_number":2},{"id":"d050befe22ec39eb5dbe5751af3a4ee71ebbfb9c","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2017-11-26 10:21:05.000000000","message":"Patch Set 2: Code-Review-1\n\nThe plugin is branchless, so I\u0027m not sure we can remove this while we still have to test stable versions that include the v1 API.","accounts_in_message":[],"_revision_number":2},{"id":"dda80ada6f7e992f9ea1323faf4121386b075aab","author":{"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},"date":"2017-11-27 16:09:42.000000000","message":"Patch Set 2:\n\ntrue - I need to check, but I think we are not actually testing the V1 API anymore (as it has been off for the last 3 cycles)","accounts_in_message":[],"_revision_number":2},{"id":"9ed4cd4049cd6f5ced26cf0659d4f5c1562196ec","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2021-07-06 22:33:57.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"17fecf945f5dacfa23500e357626471a07ac3810","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-06 23:47:21.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/760f9e5b50e84616b878550cac830dcc : SUCCESS in 41m 18s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/5b020856a37f41f4a3a10e6dd7b19252 : SUCCESS in 41m 49s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/10862d07b88c44eda0d32cb8a29bc8e2 : SUCCESS in 1h 12m 57s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/20821ebb27544e9487ba38e2c6096bb4 : SUCCESS in 44m 06s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/9df0f0292fd042f3bec836036d6c0ccf : SUCCESS in 40m 39s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7f8eb2f9404543af81d766d48f7c719f : SUCCESS in 3m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a4327f06d41b46498ac26deca03b74f3 : SUCCESS in 3m 20s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f0f3067ab7ef42c6b77b7a05e663eab9 : SUCCESS in 3m 25s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/3eea1a021ecc45c79f11912dc16aba14 : SUCCESS in 34m 26s","accounts_in_message":[],"_revision_number":3},{"id":"6fd13d06b9c1a9f77e458faf71e0958b3f2b3476","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2021-08-06 15:43:28.000000000","message":"Patch Set 3: Code-Review+1\n\nwe are good to remove the v1 tests or service clients.\nTest removal checks:\n---------------------\n- v1 API are removed in queens. stable/queens use old tempest[1] and designate-tempest-plugin can also be used as old tag released in queens or so. Tempest and plugins master needs to support only supported stable branches not EM branch. which are \u003c\u003d Train and for \u003c Train we need to use old tag of Tempest or any plugins. basically we cannot run master Tempest or plugins for \u003c\u003d stable/queens testing.\n\n- interop does not use the v1 API test so we are good from interop side too[2]\n\nService clients removal checks:\n-------------------------------\n\n- I quickly checked if these service clients are used anywhere or not and found no one use and they are good to remove[3].\n\n[1] https://opendev.org/openstack/devstack/src/branch/stable/queens/stackrc#L315\n[2] https://opendev.org/osf/interop/src/branch/master/add-ons/guidelines/dns.2020.11.json\n[3] https://codesearch.opendev.org/?q\u003dfrom%20designate_tempest_plugin.services.dns.v1\u0026i\u003dnope\u0026files\u003d\u0026excludeFiles\u003d\u0026repos\u003d","accounts_in_message":[],"_revision_number":3},{"id":"61bbf60aebcde6ceab0889ca7d0ad29b09d81721","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2021-09-23 18:39:29.000000000","message":"Patch Set 3: Code-Review+2 Review-Priority+1","accounts_in_message":[],"_revision_number":3},{"id":"866beed9d6dcd868804dc67f64c52745991f2a3f","author":{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},"date":"2021-09-23 18:43:09.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"bbe450770ec3112adf9466a407985bcd9be63d64","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2021-09-23 20:55:05.000000000","message":"Patch Set 3: Workflow+1\n\nLGTM","accounts_in_message":[],"_revision_number":3},{"id":"206316b6820648ec3ed6b940f8fb1ed942c44855","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-23 20:55:38.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"8ba4091490afae69cec0f6856e547a4ea798dba1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-23 22:02:22.000000000","message":"Patch Set 3: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/947c2f8852f141809061b4cba8480d6b : SUCCESS in 41m 41s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/6900f6c24acd4a288dda36e503f80650 : FAILURE in 42m 07s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/e2ca3bb460d741cdb758f506ab2e29e6 : SUCCESS in 1h 03m 00s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/d134bde354b443b59dcaff2884664c40 : SUCCESS in 40m 58s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/e1665ff0d9054970bba375eacee63a79 : SUCCESS in 41m 43s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7d54030b032b40e3b0f7277e5ee1ae4f : SUCCESS in 3m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/02118618fa164a658d1ed88fcbeee195 : SUCCESS in 3m 33s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0bcee6247bd04587b87b3dda834c73f0 : SUCCESS in 3m 37s","accounts_in_message":[],"_revision_number":3},{"id":"bdeda4c041a9e418cdf53400de63ec90dd509529","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2021-09-23 22:03:11.000000000","message":"Patch Set 3:\n\nrecheck\nfip ptr issue","accounts_in_message":[],"_revision_number":3},{"id":"0ecdb0822e2b3064af4bae68ea8669034983d2ae","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-23 23:03:07.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/7cd634fa0fdf427b9f27388292004f26 : SUCCESS in 31m 01s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/26f369b7d43f4b2abd7ea3f35ba8895e : SUCCESS in 46m 25s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/d3ed3790890245a98a45bfeb8998bf00 : SUCCESS in 56m 19s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/21c1504dc2724405ae8878154c8f742e : SUCCESS in 32m 37s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/13ac783621f5410a8ef642e4d2e1ae78 : SUCCESS in 38m 26s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e849f02156a14c559656d6757f57435f : SUCCESS in 4m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/60f43c38930b4a70908328a2d03b6115 : SUCCESS in 4m 23s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/52592a2a315d4f40aa82a21619d1fe2a : SUCCESS in 4m 45s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/2fd643ac68254af58d3cebf0d09b56dc : SUCCESS in 37m 24s","accounts_in_message":[],"_revision_number":3},{"id":"89ee21333f721ec7c3b18d2f45193106533bc639","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-23 23:03:39.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"5319523ae18ae16560a89b62e6dc67a4ed25d4dc","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-24 00:09:29.000000000","message":"Patch Set 3: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/7d58944aa0bd46d2acb01d9bf082cfce : SUCCESS in 39m 42s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/80cec348ac4d41d4b9622018134c7ae1 : FAILURE in 36m 34s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/a92b92c11ca64a95979a2a57c9a04a24 : SUCCESS in 1h 05m 03s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/9e0013079ec349bca674e47cefd581ae : SUCCESS in 40m 48s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/8fe7bbfb1f944bc0b766b2aa031632b2 : SUCCESS in 35m 24s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/90d17bed64cc4dcea91d1b935ac66d05 : SUCCESS in 4m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f17d5cf2858e4b238b5cc7dfca8b18f4 : SUCCESS in 4m 00s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1a21667ee6ab41eea4965fd196993785 : SUCCESS in 3m 58s","accounts_in_message":[],"_revision_number":3},{"id":"bef3ff9084324585d81a72f6419e8bc6fa385770","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2021-09-24 03:12:19.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"aad0affb76ee90b7ff01259a4ecf67a5194571c0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-24 04:08:06.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/70780837820f42bc8232dce9ec2374eb : SUCCESS in 33m 49s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/d705ae8d0f1c4d9c9a34b40d81d1c0ed : SUCCESS in 34m 31s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/768a0c949c57483b89e9d856d9532d0f : SUCCESS in 54m 11s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/423c1d7c353445c3a30ef500b742241f : SUCCESS in 40m 49s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/9654c84b33e74247b310cb86df2dcf01 : SUCCESS in 37m 20s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c600754573604461bafaeb71d6cd01c2 : SUCCESS in 4m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6224fe9e54ae4160bc6f10692faf97f1 : SUCCESS in 3m 43s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/74b7d21f490849759586f314e36a4296 : SUCCESS in 3m 44s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/9b4698e8bf3c4cb0b4b212ba82aa1802 : SUCCESS in 28m 14s","accounts_in_message":[],"_revision_number":3},{"id":"efb6675b56e2637bbcbf5d9e135f83263d3f320b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-24 04:08:39.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"34bfe8c0bb2e7a33f9cd56652d7853350b141c1d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-24 05:15:16.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/0822fe2d7ead46329c56955b3c4d4a7e : SUCCESS in 37m 54s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/605696a594b141f79050ea8ecdd3308a : SUCCESS in 50m 58s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/e2ef1e0d47884f1cad8b16a58e25e1a3 : SUCCESS in 1h 06m 26s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/2b87df15c0fd488a89cff4047e41f738 : SUCCESS in 41m 30s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/ffd4664fff044b4791286c1ec52af711 : SUCCESS in 41m 14s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8512b9b2c56f45c6bb9261f5b3813e65 : SUCCESS in 3m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6f12fe7e9a824f36bc24cf1493a9e937 : SUCCESS in 3m 35s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/bf80dbe0e6be41cda96e504f39bc9283 : SUCCESS in 3m 32s","accounts_in_message":[],"_revision_number":3},{"id":"b100daa7efe262492c3ff6c4c099ac542b8eadb3","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-24 05:15:16.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3},{"id":"c94214fb6b89732b5d8aa6ea25e696b9a4ad4364","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-24 05:16:30.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/eb6b8491b04544edbbb143f3333e04f5 : SUCCESS in 53s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9957d05093664cffb5c80b49cf7a36b9 : SUCCESS in 54s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"dcfa7c0372e41dd7a7c208ce04209f978de6539a","revisions":{"6828c7b68f3cd9400dde15a7c58d6c63f53ffe13":{"kind":"REWORK","_number":1,"created":"2017-10-03 11:00:52.000000000","uploader":{"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},"ref":"refs/changes/47/509147/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate-tempest-plugin","ref":"refs/changes/47/509147/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate-tempest-plugin refs/changes/47/509147/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate-tempest-plugin refs/changes/47/509147/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate-tempest-plugin refs/changes/47/509147/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate-tempest-plugin refs/changes/47/509147/1"}}},"commit":{"parents":[{"commit":"ba1288637e40f938b39f6ff10127217cdc9300e6","subject":"Updated from global requirements","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate-tempest-plugin/commit/ba1288637e40f938b39f6ff10127217cdc9300e6"}]}],"author":{"name":"Graham Hayes","email":"gr@ham.ie","date":"2017-10-03 11:00:33.000000000","tz":60},"committer":{"name":"Graham Hayes","email":"gr@ham.ie","date":"2017-10-03 11:00:33.000000000","tz":60},"subject":"Remove v1 API tests","message":"Remove v1 API tests\n\nChange-Id: I496d02c1be165eeaf721dea2eafe5f6a70279e68\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate-tempest-plugin/commit/6828c7b68f3cd9400dde15a7c58d6c63f53ffe13"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate-tempest-plugin/commit/6828c7b68f3cd9400dde15a7c58d6c63f53ffe13"}]},"branch":"refs/heads/master"},"3c25b80fed9fe15b7f35ad7e7e10acab031956aa":{"kind":"REWORK","_number":2,"created":"2017-11-17 16:58:24.000000000","uploader":{"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},"ref":"refs/changes/47/509147/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate-tempest-plugin","ref":"refs/changes/47/509147/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate-tempest-plugin refs/changes/47/509147/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate-tempest-plugin refs/changes/47/509147/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate-tempest-plugin refs/changes/47/509147/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate-tempest-plugin refs/changes/47/509147/2"}}},"commit":{"parents":[{"commit":"ba1288637e40f938b39f6ff10127217cdc9300e6","subject":"Updated from global requirements","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate-tempest-plugin/commit/ba1288637e40f938b39f6ff10127217cdc9300e6"}]}],"author":{"name":"Graham Hayes","email":"gr@ham.ie","date":"2017-10-03 11:00:33.000000000","tz":60},"committer":{"name":"Graham Hayes","email":"gr@ham.ie","date":"2017-11-17 16:58:08.000000000","tz":0},"subject":"Remove v1 API tests","message":"Remove v1 API tests\n\nChange-Id: I496d02c1be165eeaf721dea2eafe5f6a70279e68\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate-tempest-plugin/commit/3c25b80fed9fe15b7f35ad7e7e10acab031956aa"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate-tempest-plugin/commit/3c25b80fed9fe15b7f35ad7e7e10acab031956aa"}]},"branch":"refs/heads/master"},"dcfa7c0372e41dd7a7c208ce04209f978de6539a":{"kind":"REWORK","_number":3,"created":"2021-07-06 22:33:57.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/47/509147/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate-tempest-plugin","ref":"refs/changes/47/509147/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate-tempest-plugin refs/changes/47/509147/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate-tempest-plugin refs/changes/47/509147/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate-tempest-plugin refs/changes/47/509147/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate-tempest-plugin refs/changes/47/509147/3"}}},"commit":{"parents":[{"commit":"da27a70ae2b39695ef6f03bbefb55afeacf1cdf3","subject":"Merge \"Re-enable zone export list filter test\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate-tempest-plugin/commit/da27a70ae2b39695ef6f03bbefb55afeacf1cdf3"}]}],"author":{"name":"Graham Hayes","email":"gr@ham.ie","date":"2017-10-03 11:00:33.000000000","tz":60},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2021-07-06 22:33:56.000000000","tz":0},"subject":"Remove v1 API tests","message":"Remove v1 API tests\n\nThe V1 API was removed from Designate in the Queens release. It has now been\neight releases since then, so it is time to remove the tests for it. If these\ntests are still needed, please use git tag version 0.11.0 (Wallaby) or older.\n\nChange-Id: I496d02c1be165eeaf721dea2eafe5f6a70279e68\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate-tempest-plugin/commit/dcfa7c0372e41dd7a7c208ce04209f978de6539a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate-tempest-plugin/commit/dcfa7c0372e41dd7a7c208ce04209f978de6539a"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}},{"label":"Review-Priority","status":"MAY","applied_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
