)]}'
{"id":"openstack%2Fdesignate-tempest-plugin~710243","triplet_id":"openstack%2Fdesignate-tempest-plugin~master~I8e42ddd73c4c4f4a2d5cc348b75ddf342c7d04f2","project":"openstack/designate-tempest-plugin","branch":"master","hashtags":[],"change_id":"I8e42ddd73c4c4f4a2d5cc348b75ddf342c7d04f2","subject":"Limit where dnspython3 is installed","status":"MERGED","created":"2020-02-27 12:47:52.000000000","updated":"2020-03-05 19:02:02.000000000","submitted":"2020-03-05 18:58:26.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"710243-1583434706006-20d59637","meta_rev_id":"8be08c3bb0af6bce7c4f1632f0dc1b79aba6b072","_number":710243,"virtual_id_number":710243,"owner":{"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},{"value":0,"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-03-05 18:58:25.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},"recommended":{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},"all":[{"value":1,"date":"2020-02-27 12:49:16.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},{"value":2,"date":"2020-03-05 15:33:27.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},{"value":2,"date":"2020-03-05 15:15:17.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},"all":[{"value":0,"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},{"value":1,"date":"2020-03-05 15:33:27.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"approved":{"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},"all":[{"value":0,"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},{"value":2,"date":"2020-02-27 13:45:52.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-02-27 12:49:16.000000000","updated_by":{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},"reviewer":{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},"state":"REVIEWER"},{"updated":"2020-03-05 15:15:17.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"REVIEWER"},{"updated":"2020-03-05 18:58:25.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"334caf031390ef890f1d2a633c25c08575f478df","author":{"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},"date":"2020-02-27 12:47:52.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"72e4e5ab244be16f01c6928a9a957328f44efda6","author":{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},"date":"2020-02-27 12:49:16.000000000","message":"Patch Set 1: Code-Review+1\n\nLGTM","accounts_in_message":[],"_revision_number":1},{"id":"77655ec716fc32e3499223ff760a10cd8411650e","author":{"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},"date":"2020-02-27 13:45:52.000000000","message":"Patch Set 1: Review-Priority+2","accounts_in_message":[],"_revision_number":1},{"id":"8f9e8da0a63a9947dc927cf815ce011b1852ffc6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-27 15:22:35.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/f71c1452fa874e7cbecba3ea229bfd54 : SUCCESS in 36m 24s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/8c3139e931d942e8af3d04a53338211f : SUCCESS in 38m 04s\n- designate-pdns4-postgres https://zuul.opendev.org/t/openstack/build/a639641dd6874143b2745b19138f3859 : FAILURE in 27m 46s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/7de6220d624a4e939bdf7c3fdd63b90c : RETRY_LIMIT in 23m 27s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/f9989b91771a41f19764fdc82c9180e2 : SUCCESS in 38m 37s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/fcb2454942d44fea9bfa192a6828099e : SUCCESS in 44m 54s\n- requirements-check https://zuul.opendev.org/t/openstack/build/43672323bd7640dfa99f5e579b56a7bb : SUCCESS in 3m 53s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/67ac76586cba483194dc9c82c17222ca : SUCCESS in 5m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b0e6033f82b44385a7d09f51f70ace34 : SUCCESS in 5m 19s","accounts_in_message":[],"_revision_number":1},{"id":"03fcb91595ba74e274b3f8b22ad603ad62ee2084","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2020-03-05 13:37:48.000000000","message":"Patch Set 1:\n\nrecheck holding a node to debug grenade","accounts_in_message":[],"_revision_number":1},{"id":"082e3a0fc2e6a397cff81aebaadcaf0f7e54e9b3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-05 14:43:19.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/7b7894d0522c44738499fa8065369125 : SUCCESS in 39m 34s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/84bdd3b9da13491eabf4ac675a31fd48 : SUCCESS in 41m 43s\n- designate-pdns4-postgres https://zuul.opendev.org/t/openstack/build/9a97f2ad5a9a433c84cc350534581550 : FAILURE in 29m 38s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/c416f1877fd249d3b21f9d1f0f4d0adb : SUCCESS in 1h 02m 45s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/983ab7b0e97b4f9fad442e818faa85ab : SUCCESS in 52m 02s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/9edfcb73178c4b5ab805b935c8b919fd : SUCCESS in 52m 19s\n- requirements-check https://zuul.opendev.org/t/openstack/build/1d0261456d9f472e8b008ddf4c1aefa6 : SUCCESS in 2m 46s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a7a2b14a98da4d56bfddffe31099ac86 : SUCCESS in 6m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a059c62be95f48b89bf6543f647a1665 : SUCCESS in 3m 51s","accounts_in_message":[],"_revision_number":1},{"id":"b5f4b16a29de2e6a948d3c711e2bdb737260eb42","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2020-03-05 15:15:17.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"b5e635be106aed4f7de9b42c8b6880089fdefc12","author":{"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},"date":"2020-03-05 15:33:27.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\nninja\u0027ing in as it blocking a rocky release, and is a non code change, also taking Nicholas\u0027s +1 as a +2 :)","accounts_in_message":[],"_revision_number":1},{"id":"0098d9bd2167e7dafadaa5c0b9cecfcc21860482","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-05 15:33:38.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"de4097ee9c6063270170d69e9178bb48b908b455","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-05 16:38:31.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/b7f3ec730a3d472ebad0d16c4d285c8c : SUCCESS in 43m 05s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/3c46f6bb8dff4f7793ced0d3b73dab72 : SUCCESS in 39m 53s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/227b0ec5dae44f05ab87ee1b33a91aae : SUCCESS in 58m 50s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/dd2e76c669124fae92b4aaa03a60d6bb : SUCCESS in 40m 12s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/fea2c019a64641c9a9227b55a7b07b47 : FAILURE in 36m 27s\n- requirements-check https://zuul.opendev.org/t/openstack/build/f05b4f951b36495c835ee11a156bc643 : SUCCESS in 2m 57s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/38baa759936b47548e72f41639ef80a7 : SUCCESS in 4m 56s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b524964674f443139aa6a0f547bb1304 : SUCCESS in 4m 09s","accounts_in_message":[],"_revision_number":1},{"id":"436bfe710800622bdac7a18e6f0769524eec8d6d","author":{"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},"date":"2020-03-05 16:42:26.000000000","message":"Patch Set 1:\n\nrecheck bind timeout","accounts_in_message":[],"_revision_number":1},{"id":"013f51119fd7c77abbfe7c52eb302a220b9ff1fe","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-05 17:55:15.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/4e2fc1c337c04321981c54aec04ba7c5 : SUCCESS in 34m 15s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/5a4ec785cc2040dc82066e9f6329ce94 : SUCCESS in 37m 47s\n- designate-pdns4-postgres https://zuul.opendev.org/t/openstack/build/ed546e98ed404f38a499229ac45e17bf : FAILURE in 28m 36s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/f21655e12aa64935b4df35c77cc4fe17 : SUCCESS in 1h 09m 26s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/a452a56cdede432897909a76bf0cb28f : SUCCESS in 38m 37s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/40723b641a7349638aff851b66fcbb13 : SUCCESS in 36m 27s\n- requirements-check https://zuul.opendev.org/t/openstack/build/829da68b83694a85b9f2e3a081b595f3 : SUCCESS in 3m 27s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ffa1963c13274642a84786dc3d4f7e4f : SUCCESS in 3m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b6b4995fa684497ebc7a945e1fdca91e : SUCCESS in 3m 53s","accounts_in_message":[],"_revision_number":1},{"id":"966c4ff0ab60150ffe8c0419c53a6b98d49f637a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-05 17:55:27.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"1582dd6026e3162f85f86e7bfb2888b0df50b987","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-05 18:58:25.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/aabd4d6c78364bf8816015b1760cd686 : SUCCESS in 39m 15s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/788f3d99c19b403e8aba00233b5feb4f : SUCCESS in 40m 07s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/8dbcf20b3a23447d8d3ff98dbefa6f34 : SUCCESS in 58m 29s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/f041069d5b0b48d98885822bc7cd742d : SUCCESS in 38m 44s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/800cbb8ce7a248f9aaa4669a46fd45a5 : SUCCESS in 39m 55s\n- requirements-check https://zuul.opendev.org/t/openstack/build/b45ac331f4f540f2ba09ec79fdcd7a43 : SUCCESS in 2m 31s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ee20aa1aa50b4da0b7ccf3fb6511cff8 : SUCCESS in 3m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ec82e31b239a453ebc3976a8c6fb7109 : SUCCESS in 4m 01s","accounts_in_message":[],"_revision_number":1},{"id":"b9e993a4f997a1fd0744a206bea8ebedc02f5cad","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-05 18:58:26.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"8be08c3bb0af6bce7c4f1632f0dc1b79aba6b072","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-05 19:02:02.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/640f7e9bcf574d6cb7c6bdd3e3b70d3f : SUCCESS in 1m 18s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"0fb8c493a256d076b6023a1213569326c051fc42","revisions":{"0fb8c493a256d076b6023a1213569326c051fc42":{"kind":"REWORK","_number":1,"created":"2020-02-27 12:47:52.000000000","uploader":{"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},"ref":"refs/changes/43/710243/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate-tempest-plugin","ref":"refs/changes/43/710243/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate-tempest-plugin refs/changes/43/710243/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate-tempest-plugin refs/changes/43/710243/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate-tempest-plugin refs/changes/43/710243/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate-tempest-plugin refs/changes/43/710243/1"}}},"commit":{"parents":[{"commit":"3862c0add2550705f9e9cf8e61215d5f528329df","subject":"[ussuri][goal] Drop python 2.7 support and testing","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate-tempest-plugin/commit/3862c0add2550705f9e9cf8e61215d5f528329df"}]}],"author":{"name":"Graham Hayes","email":"gr@ham.ie","date":"2020-02-27 12:46:36.000000000","tz":0},"committer":{"name":"Graham Hayes","email":"gr@ham.ie","date":"2020-02-27 12:46:43.000000000","tz":0},"subject":"Limit where dnspython3 is installed","message":"Limit where dnspython3 is installed\n\ntempest is still testing tools that have python2 environments, and due to\nthe branchless setup, we need to make sure dnspython3 is not installed on\npython2.7\n\nChange-Id: I8e42ddd73c4c4f4a2d5cc348b75ddf342c7d04f2\nSigned-off-by: Graham Hayes \u003cgr@ham.ie\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate-tempest-plugin/commit/0fb8c493a256d076b6023a1213569326c051fc42"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate-tempest-plugin/commit/0fb8c493a256d076b6023a1213569326c051fc42"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
