)]}'
{"id":"openstack%2Fdesignate-tempest-plugin~731958","triplet_id":"openstack%2Fdesignate-tempest-plugin~master~I87335b8494a7b1b3eddd3b6b46b5b5d04227bb6d","project":"openstack/designate-tempest-plugin","branch":"master","topic":"reno-openstackdocstheme","hashtags":[],"change_id":"I87335b8494a7b1b3eddd3b6b46b5b5d04227bb6d","subject":"Switch to newer openstackdocstheme version","status":"MERGED","created":"2020-05-30 14:43:21.000000000","updated":"2020-06-04 13:47:24.000000000","submitted":"2020-06-04 13:44:16.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"731958-1591278256962-0a49205c","meta_rev_id":"ea44e68f9c01e349754d95b3d46a81e254e51450","_number":731958,"virtual_id_number":731958,"owner":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},{"value":0,"date":"2020-06-04 11:37:01.000000000","_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-06-04 13:44:16.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},"all":[{"value":2,"date":"2020-06-01 12:21:50.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},{"value":0,"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":1,"date":"2020-06-02 09:22:53.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},"all":[{"value":1,"date":"2020-06-03 12:52:54.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},{"value":0,"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},{"value":0,"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-06-02 09:22:53.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2020-06-03 12:52:54.000000000","updated_by":{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},"reviewer":{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},"state":"REVIEWER"},{"updated":"2020-06-04 13:44:16.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"414cb02514945995fec5e54739ac9d1c26a2d79f","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2020-05-30 14:43:21.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"22960885a74677fd609ab5fc8a2e33f84f7690c2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-30 15:51:58.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/b311c44ed6824a4b86094722622894ac : SUCCESS in 42m 56s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/a4b07b1ff21f4f30af08caf43dff1f69 : SUCCESS in 35m 15s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/9233befc107c49d3ba405e03256e3f89 : FAILURE in 1h 07m 17s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/bed51b832d45413f8a1d154e32a3bd1f : SUCCESS in 35m 59s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/c0bd7c4217ec430aa3463ff1a9792bee : SUCCESS in 36m 35s\n- requirements-check https://zuul.opendev.org/t/openstack/build/966df9da8f5249dcb87ae6e9121ecc8f : SUCCESS in 1m 52s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f30dccb1e8e640238502ac8a6b2b50de : SUCCESS in 4m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/de71ca2a04974ff691f163393666febc : SUCCESS in 2m 57s","accounts_in_message":[],"_revision_number":1},{"id":"0d8885a282617a98e02272c5c40e43b3af8a9aef","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2020-05-30 16:07:33.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"b78a0edec3ed5fe0d9d074f0f0a9e640e77a139a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-30 17:05:46.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/84385e897ea441a1a799b6d6e46db996 : SUCCESS in 35m 23s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/b533ca0eb42f4cec841572ad7d935854 : SUCCESS in 29m 03s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/95d7667577514914ba13908759a35d18 : SUCCESS in 46m 54s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/af1fd5a8e0f44c2d9dc429f740a4998b : SUCCESS in 36m 16s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/f6d83bb04b634fb9ba4caa42aae0f06c : SUCCESS in 38m 45s\n- requirements-check https://zuul.opendev.org/t/openstack/build/44e96bff89ea4de7984c1f358c8534cd : SUCCESS in 2m 03s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4813d56c443240f98bcf77110da6723c : SUCCESS in 3m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c4a318b9b44e478386c08a25b89354aa : SUCCESS in 3m 07s","accounts_in_message":[],"_revision_number":1},{"id":"c628ca4d678eaaa961efaab94dfd0814ab7dd868","author":{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},"date":"2020-06-01 12:21:50.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"d0c0dc29300d017c9ec83ca795213c900bd20a9b","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-06-02 09:22:53.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"e9ebda329eb401b2ef4304ac887579541f896145","author":{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},"date":"2020-06-03 12:52:54.000000000","message":"Patch Set 1: Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"90e599472f661f3b465ee58960c8af1668163e5b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-03 12:53:05.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"395145858796e1a1fe327a9fffeb7f219dcbd909","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-03 13:56:49.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/0070d20b94034e36910981d31263715d : SUCCESS in 46m 59s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/c68ee924ad864617926b5f7e7f46bbda : FAILURE in 43m 31s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/c644cff1e4514702973baf9bb30ffec5 : SUCCESS in 55m 34s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/a433dc670bcb4a458b7d26d199257ba8 : SUCCESS in 37m 40s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/91d83c254451424fa020fadc621acc5f : SUCCESS in 38m 11s\n- requirements-check https://zuul.opendev.org/t/openstack/build/c828558265114a5290514d4250274eb2 : SUCCESS in 3m 44s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c9c0d034238b4c4e91703619844ed842 : SUCCESS in 3m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d8b4b9ecd7a74113a7bfd93617c90840 : SUCCESS in 3m 28s","accounts_in_message":[],"_revision_number":1},{"id":"d6f739ba31533d253233ef0d99c50b3e879fd56b","author":{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},"date":"2020-06-03 14:08:34.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"d3cf0445fd10131d232b45cd00246eb8a8163423","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2020-06-03 15:05:32.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"e28d100af6db1f2893e0d281f29e2bd6609b69aa","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-03 15:13:50.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/7bb75c63d8764c6fa257f7f9f51ae8f7 : SUCCESS in 35m 54s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/a50f5a96761f4e5dbc3968a3f5fee0e1 : SUCCESS in 33m 18s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/c9c437e40b674ee7baf99a12c8c9d2d4 : SUCCESS in 55m 12s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/75eeb49890744ff8a15b1675a811d609 : SUCCESS in 43m 06s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/151891002b6f4f63921a056f7a2010b0 : SUCCESS in 29m 04s\n- requirements-check https://zuul.opendev.org/t/openstack/build/96c4cd92d0ba405ebfb9d57251c6b156 : SUCCESS in 3m 04s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ad76c6e4d4a54e4b9242f031b99152d9 : SUCCESS in 4m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c6435f7f4cfc4401aeffcdc26780bb43 : SUCCESS in 3m 42s","accounts_in_message":[],"_revision_number":1},{"id":"2f852ec53a11ae3c501bf90e51320e541122c2b5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-03 15:14:05.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"8fc2e751f1d50237cfa2a22c781247ee1f4e1cd4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-03 16:14:19.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/caad4f3d656d471fa70a1c5f3151451b : SUCCESS in 37m 20s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/a8f089132f3849b8bc9cd2e46d44971f : FAILURE in 46m 20s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/9024836ba5d245bd89c39645733e0f5a : SUCCESS in 58m 53s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/8e6aa8ac3343487daa4eaa298703c62c : SUCCESS in 38m 04s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/ee39fbedc0d04d58b6bb4219828b811c : SUCCESS in 37m 14s\n- requirements-check https://zuul.opendev.org/t/openstack/build/d193bf6967ca47af92deb0a852c0a322 : SUCCESS in 2m 54s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b9b8c116ee22431daa0c0f21a7cb9349 : SUCCESS in 3m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/095eb7f86157472eac03dde58c9b907a : SUCCESS in 3m 44s","accounts_in_message":[],"_revision_number":1},{"id":"7bfdbe66c8f2a7e5372733147b403989e8e4c392","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2020-06-03 17:19:05.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"4402eea530a7267e7b45fe22c44a3f3a05a2f132","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-03 18:26:19.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/f88c63a122f84921b3a52a23bce12e31 : SUCCESS in 39m 52s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/e6091e59ea5c4a6c881d5b1a0264b460 : SUCCESS in 38m 09s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/f8e0c394dbcb4843876c6d5f806cef67 : SUCCESS in 1h 01m 00s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/e245706513f44cc99c2c4261e5462398 : SUCCESS in 39m 44s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/3816d55055ef48b18e7a7b2e339d68a0 : SUCCESS in 44m 06s\n- requirements-check https://zuul.opendev.org/t/openstack/build/dafe19d01f5a4f4a8db9236ab7f50ffe : SUCCESS in 2m 26s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1782ccaa7b4f417eacb27029f79ec635 : SUCCESS in 5m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3d41edade79e4d3783848a7b6acb405e : SUCCESS in 4m 16s","accounts_in_message":[],"_revision_number":1},{"id":"6868083a90dfc5876c61ea4f89d6ad8ffb60b29f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-03 18:26:30.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"74cc09b1005f6399e05bdfb3b0232b024ac49fa4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-03 19:07:41.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/90f7228e30cd4ea085ae4320a527b219 : SUCCESS in 26m 52s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/c6c394ea5b694043a8565617a70131da : SUCCESS in 34m 29s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/a4ab514228df496d9a3a5a81535ab5d8 : SUCCESS in 39m 44s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/e47f8b399ec14ad383888a85fcaaa301 : FAILURE in 29m 39s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/d56ec1e6cac54f8a98e57106bb69a49b : SUCCESS in 26m 32s\n- requirements-check https://zuul.opendev.org/t/openstack/build/0daaed4f57d94feebcacf44f67d48fe7 : SUCCESS in 2m 35s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/064486baa6ec4ea6834be374c95d22d8 : SUCCESS in 3m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d9060fd6409747849ede3dfbe9d72f6f : SUCCESS in 3m 13s","accounts_in_message":[],"_revision_number":1},{"id":"90581d051df19cb2dc29943d536ae186107337b5","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2020-06-03 19:25:00.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"8d08e514a43cf92e9db21c3ffabdf7ec2ed3d7d4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-03 20:20:16.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/4d61d89325ca41b0b6c3eedf88c1f00e : SUCCESS in 38m 38s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/ace797b21a0c4bcaad35ca7803d37465 : SUCCESS in 38m 42s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/d34c7b0b5cb74f14919ff60d7510240c : FAILURE in 53m 45s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/5ecf2fece0a6468c932e5b85933b7a9a : SUCCESS in 36m 40s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/0288ca5094df442c8b948915177ea481 : SUCCESS in 39m 07s\n- requirements-check https://zuul.opendev.org/t/openstack/build/e9c93eafdc614642bd7ece6701023914 : SUCCESS in 2m 18s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/68fdeb6b2569445ab3f79e49928c04ce : SUCCESS in 4m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0f44d9b6cbd944879a8b61dc4f9fe271 : SUCCESS in 3m 41s","accounts_in_message":[],"_revision_number":1},{"id":"b46d437c7c449cf2d1a2a8c1eccf102d97bf01fe","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2020-06-03 20:24:56.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"41898847111629bf4f8485f240c8e7eb6ad79209","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-03 22:00:09.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/142cf66396fa413e82db11bc1db8c6ea : FAILURE in 44m 26s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/d885dbc1bab545d5bf6b5152c8de3fbd : SUCCESS in 39m 30s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/588e41fdc420454c86822d23dbd4a2c2 : SUCCESS in 54m 00s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/7e6f82992f444000a1defd027f018a4b : SUCCESS in 39m 10s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/78edcee7bbfb4898bb71d58eb631a1ab : SUCCESS in 38m 18s\n- requirements-check https://zuul.opendev.org/t/openstack/build/6eec5c6fb7a34ea3b9eabb7ed7b8b6d2 : SUCCESS in 2m 28s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6be6832702064272b5419e001fdde370 : SUCCESS in 4m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0a1e9dd7e1734d2197cc5901a4c88fd4 : SUCCESS in 4m 16s","accounts_in_message":[],"_revision_number":1},{"id":"479fa794238563330fe254962224433dfd4c077f","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2020-06-04 04:41:43.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"c81783af59fb9711e5a95da6ed002517ed8c3156","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-04 05:41:47.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/4b19fe19e75949a4bd88497b5f9bb2dc : SUCCESS in 36m 16s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/8fffa03144f2401fbab00d2e6695f563 : SUCCESS in 41m 02s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/0430991b3b034c04b49a669d43f93769 : SUCCESS in 55m 00s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/4032df847fe6454a8a7864ebc50fb26c : SUCCESS in 36m 18s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/2fc4fac8964c4aaeba79c59c9935a94e : SUCCESS in 39m 28s\n- requirements-check https://zuul.opendev.org/t/openstack/build/7e7a1f4103ae4e57970a250c4ef20a23 : SUCCESS in 2m 06s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1713de94cbbd4820b598d9ff9962b3b7 : SUCCESS in 3m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f863068d167f40219376ed2dbbe0d63a : SUCCESS in 3m 12s","accounts_in_message":[],"_revision_number":1},{"id":"0b827a6326e75e488f1412b9cc1ce4fa56a57447","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-04 05:41:58.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"6268b6ff98484f2e4e1bf4fd62d090577cff31fa","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-04 06:40:43.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/1c094f7fc865492ca51e412ecf792f10 : SUCCESS in 41m 46s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/8f34fbf92afe45b297b95a5e729e2368 : FAILURE in 41m 20s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/68cff42c11f3458a8b233b62e3913129 : SUCCESS in 57m 38s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/6a3e45db54a44faeacbf528c4c670a3e : SUCCESS in 44m 43s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/39abe6458bfa441e94731002953adbe3 : SUCCESS in 41m 12s\n- requirements-check https://zuul.opendev.org/t/openstack/build/ba63246caec4404fa3ed6749e977b987 : SUCCESS in 2m 03s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/eae2afb1ea0145afb7cb62772325cf90 : SUCCESS in 3m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2a6bb36a6e4f49bdaa0e949faf366bcf : SUCCESS in 3m 04s","accounts_in_message":[],"_revision_number":1},{"id":"b491ee58c06b2d2f0c51bf2a55d0dbad39b959a6","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2020-06-04 07:10:20.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"6a0a72f8261471e7d40036bb9818fbf699fafaf0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-04 08:10:12.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/a1f055276e184ae888c477db22115d08 : FAILURE in 37m 53s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/833e4e67af5441c4bdfe7b7a038196b5 : SUCCESS in 46m 31s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/204aa3dd4cae4c60aabefb564f8490b4 : SUCCESS in 57m 54s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/c0e52e089b26493d85aa4d80a0ed8329 : SUCCESS in 46m 15s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/ee6b72bc0f2b485d8942b23be881eb89 : SUCCESS in 39m 11s\n- requirements-check https://zuul.opendev.org/t/openstack/build/0521d0f974e145438d534ba868ec04d2 : SUCCESS in 3m 01s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ad9d0482e9da44068f77b1e93d9f0860 : SUCCESS in 4m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7c6bdce5d06b4078a861c1c57af49039 : SUCCESS in 4m 05s","accounts_in_message":[],"_revision_number":1},{"id":"e86e6708c38adcd94422072a306141db7e24ffe5","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2020-06-04 08:16:10.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"1e8d92bb042e3d9d643bde6b20fde2796516f94a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-04 08:56:14.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/bf15a52510b243a9af2644539a706c9c : SUCCESS in 28m 04s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/4502385814034c1cadce2c4bb316f99c : SUCCESS in 36m 24s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/0c5ba7bb25184907ac34c51b9a3c3708 : SUCCESS in 37m 54s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/dac327b5503e40a7bef6126838157fee : SUCCESS in 35m 51s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/c394f494f77d41619a24b1510ea18019 : SUCCESS in 26m 49s\n- requirements-check https://zuul.opendev.org/t/openstack/build/98bde059988b46d080c6e034b13576a2 : SUCCESS in 2m 12s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0d3a81c4967748b1923ddf3564348973 : SUCCESS in 3m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/811ba7152412429691e7fd486c05aee8 : SUCCESS in 3m 19s","accounts_in_message":[],"_revision_number":1},{"id":"f55e8f8254cd063d45fff5ad5b1c6b2b99bec4d7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-04 08:56:25.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"9e7b6215c5a6b3edc42a138edc800da0223f8e34","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-04 09:59:22.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/10fa9c5229d943e68b87b4acc14e4970 : FAILURE in 37m 33s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/9f8fcdd7748d4712bb5e15ca01f24c67 : SUCCESS in 50m 46s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/050bc1e8ec514a67b348eb62d5018a44 : SUCCESS in 57m 28s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/2843b82408a94be0b64ab6eb64da8091 : FAILURE in 29m 14s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/22c0e317b61446a3903f3c98af0cc0e5 : FAILURE in 39m 12s\n- requirements-check https://zuul.opendev.org/t/openstack/build/b17b022ec6f741fda156b82729f54c8c : SUCCESS in 3m 11s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/675dc12450b64a738266f60a30db464f : SUCCESS in 6m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0e8be08e09e640b5a279adcf2521493d : SUCCESS in 5m 26s","accounts_in_message":[],"_revision_number":1},{"id":"8e1410d801f2930e68dc4e9962d22f451d756b9e","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2020-06-04 11:37:01.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"1664f29aef54d2a238b7bc4c5e02cb2e027fbb65","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-04 12:41:16.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/45a813ec32f0478d985b20387fbd4993 : SUCCESS in 42m 06s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/a9d686a066164b39b4ac56ae133c5f9b : SUCCESS in 32m 57s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/e443782874304eb8a6b44887a01c4ebb : SUCCESS in 58m 04s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/769373373c5745ad9dcc78b71a7720d4 : SUCCESS in 43m 38s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/c6506c58147841dc8a4299a7e9f8609e : SUCCESS in 42m 38s\n- requirements-check https://zuul.opendev.org/t/openstack/build/467ec194eac64e5bac2abecbd1e5f974 : SUCCESS in 2m 35s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f3cf62bace2342d3abb843137a7214c5 : SUCCESS in 3m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/22509321085d4aaba75bb1b52cee8265 : SUCCESS in 4m 36s","accounts_in_message":[],"_revision_number":1},{"id":"3a1938c36e5f742e6984803d69ca4b49f0748f21","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-04 12:41:27.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"1f5d587bf2fce0d93f8090f77e65512fd55d7346","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-04 13:44:16.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"1c17074ca4e650103a7047157bb972ba3b5182f3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-04 13:44:16.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/cdbb2b66b4c1415b8fcb88851ad8d605 : SUCCESS in 36m 23s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/04bca9a4de0b409092bf530c0224588b : SUCCESS in 36m 39s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/ad332db0fe5b43aaa505465dd5789226 : SUCCESS in 56m 10s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/68df19f8cd9046d5a15bb2f28a9b6fdf : SUCCESS in 38m 04s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/967e60475693479c99ea6583e3a542f6 : SUCCESS in 41m 07s\n- requirements-check https://zuul.opendev.org/t/openstack/build/f247da017b19439794d1cdb723565cdc : SUCCESS in 3m 10s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f55f019cef9341b7a7c3cc606db5794d : SUCCESS in 4m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fdb2637b22ef49619bdcb136fa85542a : SUCCESS in 5m 08s","accounts_in_message":[],"_revision_number":1},{"id":"e6c7152db26701ce41f4fd50e26545daab0fa3ab","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-04 13:47:24.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ff5436582143473eafdafbd6c7ce41f4 : SUCCESS in 2m 47s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"88c40dbc1e4d0d38d9f53f76b2ced0b76e275737","revisions":{"88c40dbc1e4d0d38d9f53f76b2ced0b76e275737":{"kind":"REWORK","_number":1,"created":"2020-05-30 14:43:21.000000000","uploader":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"ref":"refs/changes/58/731958/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate-tempest-plugin","ref":"refs/changes/58/731958/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate-tempest-plugin refs/changes/58/731958/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate-tempest-plugin refs/changes/58/731958/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate-tempest-plugin refs/changes/58/731958/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate-tempest-plugin refs/changes/58/731958/1"}}},"commit":{"parents":[{"commit":"5fb81faee08cbf9a9a1c72503886e94d25ef5f10","subject":"Update hacking for Python3","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate-tempest-plugin/commit/5fb81faee08cbf9a9a1c72503886e94d25ef5f10"}]}],"author":{"name":"Andreas Jaeger","email":"aj@suse.com","date":"2020-05-30 14:40:23.000000000","tz":120},"committer":{"name":"Andreas Jaeger","email":"aj@suse.com","date":"2020-05-30 14:43:12.000000000","tz":120},"subject":"Switch to newer openstackdocstheme version","message":"Switch to newer openstackdocstheme version\n\nSwitch to openstackdocstheme 2.2.1 version. Using\nthis version will allow especially:\n* Linking from HTML to PDF document\n* Allow parallel building of documents\n* Fix some rendering problems\n\nUpdate Sphinx version as well.\n\nChange pygments_style to \u0027native\u0027 since old theme version always used\n\u0027native\u0027 and the theme now respects the setting and using \u0027sphinx\u0027 can\nlead to some strange rendering.\n\nopenstackdocstheme renames some variables, so follow the renames\nbefore the next release removes them. A couple of variables are also\nnot needed anymore, remove them.\n\nSee also\nhttp://lists.openstack.org/pipermail/openstack-discuss/2020-May/014971.html\n\nChange-Id: I87335b8494a7b1b3eddd3b6b46b5b5d04227bb6d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate-tempest-plugin/commit/88c40dbc1e4d0d38d9f53f76b2ced0b76e275737"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate-tempest-plugin/commit/88c40dbc1e4d0d38d9f53f76b2ced0b76e275737"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
