)]}'
{"id":"openstack%2Fdesignate-tempest-plugin~781442","triplet_id":"openstack%2Fdesignate-tempest-plugin~master~Ib7a3b829129f2534f67c66c5f58e7c7c3c2f93c9","project":"openstack/designate-tempest-plugin","branch":"master","topic":"ptr_records","hashtags":[],"change_id":"Ib7a3b829129f2534f67c66c5f58e7c7c3c2f93c9","subject":"Test cases for: \"FloatingIPs PTR\" resords.","status":"MERGED","created":"2021-03-18 12:42:27.000000000","updated":"2021-06-04 06:14:07.000000000","submitted":"2021-05-21 06:33:53.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":10,"unresolved_comment_count":3,"has_review_started":true,"submission_id":"781442-ptr_records","meta_rev_id":"dfd27d1f1027101015daf808486c542cf709b8d9","_number":781442,"virtual_id_number":781442,"owner":{"_account_id":28609,"name":"Arkady Shtempler","email":"ashtempl@redhat.com","username":"ashtempl"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28609,"name":"Arkady Shtempler","email":"ashtempl@redhat.com","username":"ashtempl"},{"value":0,"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-05-21 06:33:52.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},"all":[{"value":0,"_account_id":28609,"name":"Arkady Shtempler","email":"ashtempl@redhat.com","username":"ashtempl"},{"value":2,"date":"2021-05-14 19:33:15.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2021-05-20 18:35:00.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"all":[{"value":0,"_account_id":28609,"name":"Arkady Shtempler","email":"ashtempl@redhat.com","username":"ashtempl"},{"value":0,"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2021-05-20 18:35:00.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":28609,"name":"Arkady Shtempler","email":"ashtempl@redhat.com","username":"ashtempl"},{"value":0,"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":15025,"name":"Michael Johnson","email":"Michael.Johnson@sas.com","username":"mikjo"},{"_account_id":29681,"name":"Bruna Bonguardo","display_name":"Bruna Bonguardo","email":"bbonguar@redhat.com","username":"bonguardo"},{"_account_id":6994,"name":"Michael Chapman","email":"woppin@gmail.com","username":"michaeltchapman"},{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":18894,"name":"Toni Freger","email":"tfreger@redhat.com"},{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"reviewers":{"REVIEWER":[{"_account_id":6994,"name":"Michael Chapman","email":"woppin@gmail.com","username":"michaeltchapman"},{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":15025,"name":"Michael Johnson","email":"Michael.Johnson@sas.com","username":"mikjo"},{"_account_id":18894,"name":"Toni Freger","email":"tfreger@redhat.com"},{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},{"_account_id":28609,"name":"Arkady Shtempler","email":"ashtempl@redhat.com","username":"ashtempl"},{"_account_id":29681,"name":"Bruna Bonguardo","display_name":"Bruna Bonguardo","email":"bbonguar@redhat.com","username":"bonguardo"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-03-18 14:02:18.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-03-19 12:56:15.000000000","updated_by":{"_account_id":28609,"name":"Arkady Shtempler","email":"ashtempl@redhat.com","username":"ashtempl"},"reviewer":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"state":"REVIEWER"},{"updated":"2021-03-19 12:56:15.000000000","updated_by":{"_account_id":28609,"name":"Arkady Shtempler","email":"ashtempl@redhat.com","username":"ashtempl"},"reviewer":{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},"state":"REVIEWER"},{"updated":"2021-03-19 12:56:15.000000000","updated_by":{"_account_id":28609,"name":"Arkady Shtempler","email":"ashtempl@redhat.com","username":"ashtempl"},"reviewer":{"_account_id":15025,"name":"Michael Johnson","email":"Michael.Johnson@sas.com","username":"mikjo"},"state":"REVIEWER"},{"updated":"2021-03-19 12:56:15.000000000","updated_by":{"_account_id":28609,"name":"Arkady Shtempler","email":"ashtempl@redhat.com","username":"ashtempl"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"REVIEWER"},{"updated":"2021-04-01 08:43:51.000000000","updated_by":{"_account_id":28609,"name":"Arkady Shtempler","email":"ashtempl@redhat.com","username":"ashtempl"},"reviewer":{"_account_id":29681,"name":"Bruna Bonguardo","display_name":"Bruna Bonguardo","email":"bbonguar@redhat.com","username":"bonguardo"},"state":"REVIEWER"},{"updated":"2021-04-01 08:43:51.000000000","updated_by":{"_account_id":28609,"name":"Arkady Shtempler","email":"ashtempl@redhat.com","username":"ashtempl"},"reviewer":{"_account_id":18894,"name":"Toni Freger","email":"tfreger@redhat.com"},"state":"REVIEWER"},{"updated":"2021-04-12 09:18:13.000000000","updated_by":{"_account_id":28609,"name":"Arkady Shtempler","email":"ashtempl@redhat.com","username":"ashtempl"},"reviewer":{"_account_id":6994,"name":"Michael Chapman","email":"woppin@gmail.com","username":"michaeltchapman"},"state":"REVIEWER"},{"updated":"2021-04-27 17:47:46.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"}],"messages":[{"id":"e2ff41cba6ebebd5f166d97193461a5ea632ad54","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28609,"name":"Arkady Shtempler","email":"ashtempl@redhat.com","username":"ashtempl"},"date":"2021-03-18 12:42:27.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"c4de7a43078af605512d71c8acde2b08da5e53bf","author":{"_account_id":28609,"name":"Arkady Shtempler","email":"ashtempl@redhat.com","username":"ashtempl"},"date":"2021-03-18 12:44:40.000000000","message":"Patch Set 1: Workflow-1\n\ntest_unset_floatingip_ptr - fails, maybe Designate bug or DOC bug.\nStill under investigation.","accounts_in_message":[],"_revision_number":1},{"id":"553182419c60d77103d25c63bcdcfbb65f82a0b3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-18 14:02:18.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/82e9091f1dfe4933ae026adc06e9a59b : FAILURE in 39m 16s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/ea7bde8191344c729c796a09889171d2 : FAILURE in 40m 12s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/f2fbcbf8a3234c1e808be12ad5467a55 : FAILURE in 1h 00m 58s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/5baaee04cae246829257fc8fbdbde191 : FAILURE in 46m 26s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/8aeffee55d644c19a6d5e93534321a63 : FAILURE in 48m 59s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c11239f6d1fb4f08b4a330035acb3f1d : SUCCESS in 4m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b3825514ddac4f12ad6e2228439459c5 : SUCCESS in 6m 04s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/f15602801d7f4e4a9b261c4c1b2ea9ed : SUCCESS in 29m 30s","accounts_in_message":[],"_revision_number":1},{"id":"2fea3fa0096ba92bd479fbac88fcbe82339f7a78","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28609,"name":"Arkady Shtempler","email":"ashtempl@redhat.com","username":"ashtempl"},"date":"2021-03-18 14:59:40.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"0559406d7e7f769f553b0fad7a7dc29d35fc035f","author":{"_account_id":28609,"name":"Arkady Shtempler","email":"ashtempl@redhat.com","username":"ashtempl"},"date":"2021-03-18 15:01:17.000000000","message":"Patch Set 2: Workflow-1\n\nCleanup?","accounts_in_message":[],"_revision_number":2},{"id":"973bd44d1aa33d1135c0631f629c0c3d637acada","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28609,"name":"Arkady Shtempler","email":"ashtempl@redhat.com","username":"ashtempl"},"date":"2021-03-18 15:05:28.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"ebd6357b63255502a04580992b1b79278b5a507a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-18 16:08:18.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/9573a6a3f29948b5968b6104341e0078 : FAILURE in 35m 25s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/0be8a856f96f48e4b8be4cf21990dc6a : FAILURE in 32m 58s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/4dc7bc23f3284c2792cab7e9c7a552d5 : FAILURE in 50m 37s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/2269963a305746e3afb4306525a277f4 : FAILURE in 34m 00s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/ec7bd8e442fd4899a84f75ff8286d3c4 : FAILURE in 41m 40s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/eb08e47839bb4be3ade959fd8a542930 : SUCCESS in 5m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/927346acb65c473c89fcccf690ebeead : SUCCESS in 4m 18s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/db00c4f8e4054fceb2a1c4513c5057e4 : SUCCESS in 35m 28s","accounts_in_message":[],"_revision_number":3},{"id":"5599761f59d040aa3bc4259b444b8c3b9a46dd9f","author":{"_account_id":28609,"name":"Arkady Shtempler","email":"ashtempl@redhat.com","username":"ashtempl"},"date":"2021-03-18 16:11:56.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"f7a61a58e285688153f6ae2f2aa730509d823aed","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-18 17:32:18.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/dfb65566e76749d68cedcc9809c3b00d : FAILURE in 39m 05s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/87cd55ee460d44bbb06f08f871e8993d : FAILURE in 37m 40s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/f2640362598e4e3aa544486a4c171410 : FAILURE in 1h 01m 57s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/058db715416c426b89ea58a88a0066eb : FAILURE in 44m 59s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/4b04d447142f4c699934098431489473 : FAILURE in 44m 24s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/814800ddb1f8490381bfdd8e9e7f25d6 : SUCCESS in 4m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c888f2bc674a4f7995812bc52a5b2449 : SUCCESS in 4m 06s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/70e30f6491ed4cae8f33b6a528f17f88 : SUCCESS in 36m 24s","accounts_in_message":[],"_revision_number":3},{"id":"ce0067887bc32c16e6348eaae84316bcfb1e0978","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28609,"name":"Arkady Shtempler","email":"ashtempl@redhat.com","username":"ashtempl"},"date":"2021-03-18 17:57:52.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"9c822fcf9c029f2590c9c7d52476ae2ad4102377","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-18 19:13:29.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/b130ca3e44ca42b68cffc5cb095ad990 : SUCCESS in 42m 21s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/65033c7f114e4705be54d990e1953095 : SUCCESS in 41m 29s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/0efdc3d282fd4e65ac55c24269e6043f : SUCCESS in 1h 03m 56s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/07ddc6cc17c64e50b1478273107b6ba5 : SUCCESS in 48m 33s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/f4f849d6c98c42b58e8aed2e1967e5ec : SUCCESS in 43m 26s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c7039b0db61e4797a5b5681c8964a820 : SUCCESS in 7m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/624b117e756d4fc99576c7f79eeee715 : SUCCESS in 6m 07s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/7a1a10e788184a07988059e26eb9cf4d : SUCCESS in 38m 42s","accounts_in_message":[],"_revision_number":4},{"id":"c4913397b2d603b80cd8c3d0577a2793448c01af","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28609,"name":"Arkady Shtempler","email":"ashtempl@redhat.com","username":"ashtempl"},"date":"2021-03-18 20:14:18.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"e8e665a8284913ba5a6fb0c4157d5931ae555420","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-18 21:15:25.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/fe8f8fe778e047ac8882ead361f4007d : SUCCESS in 40m 54s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/d74fa3079e0a44a987f43a244824cb07 : SUCCESS in 40m 05s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/32ae0840b00a40ee82da88ee8f139547 : FAILURE in 58m 52s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/485f25e3a3c74746a185a20bc1beb54b : SUCCESS in 36m 20s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/2f71c52e87d646dfb77a449ff4107524 : SUCCESS in 37m 14s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2fea9364b0fb4fb78bb3a4cc6ea63b97 : SUCCESS in 4m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2090f0d9e91c4eb887fda162c45c470f : SUCCESS in 3m 57s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/306b1f52bfa9486fbe0bb2198df1d7a2 : SUCCESS in 34m 59s","accounts_in_message":[],"_revision_number":5},{"id":"7b7a746d085813ebc9dbed7aaf92ad59459af566","author":{"_account_id":28609,"name":"Arkady Shtempler","email":"ashtempl@redhat.com","username":"ashtempl"},"date":"2021-03-18 21:24:31.000000000","message":"Patch Set 5:\n\nrecheck","accounts_in_message":[],"_revision_number":5},{"id":"aab569a0d61db58d7a950049c2ed71705ed2b264","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-18 22:32:03.000000000","message":"Patch Set 5:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/038285c26f324107899740392457d207 : FAILURE in 40m 34s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/13e92fe0e0ca4cd881e544c05f727200 : FAILURE in 49m 05s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/5bdbf3ba3b644ca9b2e3a6c9b0f4011b : SUCCESS in 1h 06m 35s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/009fe9de24b34609a30a8714f43ac77e : SUCCESS in 40m 48s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/eccc3fde6c7d4b289fdd348b3effcfca : SUCCESS in 38m 00s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/970b5a8bd7e44398b8ba8403c9374427 : SUCCESS in 3m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/84ca152e84b640fa8002e7d87f91eeab : SUCCESS in 3m 38s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/397b381ce96549938e5f84a3f1e31b4a : SUCCESS in 33m 10s","accounts_in_message":[],"_revision_number":5},{"id":"6d0558e6d165c208c9d96c6f449fe07e96902744","author":{"_account_id":28609,"name":"Arkady Shtempler","email":"ashtempl@redhat.com","username":"ashtempl"},"date":"2021-03-19 05:41:08.000000000","message":"Patch Set 5:\n\nrecheck","accounts_in_message":[],"_revision_number":5},{"id":"6ff96114766c53b23270bcbd602ceed90063e81f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-19 06:48:01.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/fa26413fa5264eaf8a5f8ffbd5512a89 : SUCCESS in 46m 52s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/aaf91154bdde4a899906c814dbc698c6 : SUCCESS in 45m 47s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/d1256fa7cfa34801a606871ac3746492 : SUCCESS in 1h 04m 46s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/5235e76fa4c743c9a73574208b352ec3 : SUCCESS in 48m 40s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/8053adbf83a24cf3b7f189fa73bc3db3 : SUCCESS in 59m 24s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1cb8a30d58c4420c9ce8185a29d5b3ca : SUCCESS in 4m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0267006ca4b34c83b68be3416b089f6f : SUCCESS in 3m 51s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/777657eca5c0447eadefc79880e7de13 : SUCCESS in 37m 25s","accounts_in_message":[],"_revision_number":5},{"id":"e591c5cf7ea1afb06c726812dfd74980361bacac","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2021-04-27 17:47:46.000000000","message":"Patch Set 5: Code-Review-1\n\n(4 comments)\n\nJust a few minor comments.","accounts_in_message":[],"_revision_number":5},{"id":"4ca3637d8016c42b6d7d8336ab73a4adea033a3b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28609,"name":"Arkady Shtempler","email":"ashtempl@redhat.com","username":"ashtempl"},"date":"2021-04-27 19:52:22.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"311280ddae5dda426eb8468736ba6beea7bca4a0","author":{"_account_id":28609,"name":"Arkady Shtempler","email":"ashtempl@redhat.com","username":"ashtempl"},"date":"2021-04-27 19:57:08.000000000","message":"Patch Set 6:\n\n(4 comments)","accounts_in_message":[],"_revision_number":6},{"id":"fb5c358b929d758b21bc742dd8f4805159f72e80","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-27 21:50:07.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/d6e763e42b7f41f6a5dadcc87fab9b54 : SUCCESS in 36m 05s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/9a64fcd0ada0443fab55c99272e0bf96 : SUCCESS in 32m 17s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/359625a62511431eabfe4f1247096ea9 : SUCCESS in 53m 12s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/67ca1610b2b14d368d9fd54d35ccd47e : SUCCESS in 43m 59s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/8fd284fc934944be83fd54f354e4cad0 : SUCCESS in 38m 10s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/27d41dbf049a462ea7ac25d47a3498af : SUCCESS in 6m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c56875507c4640019861393a27ecc4dd : SUCCESS in 4m 22s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/ca0b9d2d346f45e5b874108539484bd8 : FAILURE in 32m 13s","accounts_in_message":[],"_revision_number":6},{"id":"ff11fe1626a00040b762108faa08c74e990cefb0","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2021-04-28 21:42:47.000000000","message":"Patch Set 6: Code-Review+2\n\nLGTM","accounts_in_message":[],"_revision_number":6},{"id":"3f3e2d4dcdc8ac579dd6ce0f723057f7f3f53b7e","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2021-04-28 21:42:53.000000000","message":"Patch Set 6:\n\nrecheck","accounts_in_message":[],"_revision_number":6},{"id":"5bf98927fd3044d97c570a24c5ebb6f385d98de8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-28 22:51:15.000000000","message":"Patch Set 6:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/6c8dfd7b42994938a89248e97437f857 : SUCCESS in 38m 55s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/01bc32d1084946e78c427b3c7b7ffb0a : SUCCESS in 39m 31s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/636f833e298245e298e70f9872c8aa72 : SUCCESS in 1h 03m 20s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/000e77b8ad7f42a58bb42c0e46389d93 : SUCCESS in 41m 50s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/cc5050283ca841569a82f9e38f96fc49 : SUCCESS in 39m 42s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/da30fa9dc71f436d955ba7ef381ba8f1 : SUCCESS in 3m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f41dc09e912d4048a9955634efd196d1 : SUCCESS in 3m 50s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/97bcd1212bd14f13b5ca01c442f94473 : FAILURE in 35m 58s","accounts_in_message":[],"_revision_number":6},{"id":"eccf989e66c904ae4aa415901d3bd274cec3d717","author":{"_account_id":28609,"name":"Arkady Shtempler","email":"ashtempl@redhat.com","username":"ashtempl"},"date":"2021-05-02 07:27:30.000000000","message":"Patch Set 6:\n\nrecheck","accounts_in_message":[],"_revision_number":6},{"id":"59b022e1f218013a4bcc009d7e7f5518b313325d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-02 08:38:45.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/c90e037d44a0489794bb71951c241de1 : SUCCESS in 39m 20s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/cd7daae4894045a083554c1c9efb7287 : SUCCESS in 39m 37s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/8e744accaf4b439f8e2dbfa8d862bf8a : SUCCESS in 1h 05m 39s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/5ca920d1fb5640ed942fda5ba6e4dd39 : SUCCESS in 39m 24s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/71ed071e641a43b98b3f4868b84cd6e2 : SUCCESS in 38m 52s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c47b682599084483af7b6d6444ccac52 : SUCCESS in 3m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6d4a4974e3dd4bc9866a03550626c107 : SUCCESS in 3m 30s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/4fc75b4cfead42399032708562b0ed74 : SUCCESS in 31m 49s","accounts_in_message":[],"_revision_number":6},{"id":"46e56ea8cc548356ef7f21427ddf124e5e730213","author":{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},"date":"2021-05-03 17:19:26.000000000","message":"Patch Set 6: Code-Review+2","accounts_in_message":[],"_revision_number":6},{"id":"813bced074720ecc7b2f56042b2aeca64f4f13aa","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2021-05-03 18:22:48.000000000","message":"Patch Set 6: Workflow+1","accounts_in_message":[],"_revision_number":6},{"id":"9705e5169b0b9aa00d99969c961c4b3010274413","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-03 18:23:37.000000000","message":"Patch Set 6: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":6},{"id":"a0b0848c8a8e65050f323cdc98ad6b3d86182b0e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-03 20:49:48.000000000","message":"Patch Set 6: Verified-2\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":6},{"id":"b87e849aabe450c619cca462efc1c4a227891d31","author":{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},"date":"2021-05-03 21:00:31.000000000","message":"Patch Set 6:\n\nrecheck","accounts_in_message":[],"_revision_number":6},{"id":"7a7a86631222f89a5da14b4dbc670f863f849ae7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-03 21:00:44.000000000","message":"Patch Set 6: Verified-1\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":6},{"id":"4153c36478ba39b981b89d65d3ca97418a54c8d4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28609,"name":"Arkady Shtempler","email":"ashtempl@redhat.com","username":"ashtempl"},"date":"2021-05-04 07:48:13.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"0e2751d16e8a330d7dc86ab9b966d02d5d51e23c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-04 08:31:54.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/d111f7f837d546aa8d746a9518ac5f76 : FAILURE in 24m 47s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/36071fb7c23a4aebb3f587f5a626f31e : FAILURE in 21m 31s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/cfc696cc247d4080b18e9b17105fe3d1 : FAILURE in 42m 18s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/0399ec7846f7472dacd1c265a17b57e0 : FAILURE in 28m 10s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/5ad34ee1b20b4602b55544468e56fb38 : FAILURE in 24m 15s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0f954a20c3da4217a3ea29e7839e0a60 : SUCCESS in 3m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e2e1276db9604589b69ffb00a9bac720 : FAILURE in 3m 36s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/df48b75cac6b4839901daa592ab24f9e : FAILURE in 19m 58s","accounts_in_message":[],"_revision_number":7},{"id":"e7a03f833a50ff7f717011164dc80caa8d0f4383","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28609,"name":"Arkady Shtempler","email":"ashtempl@redhat.com","username":"ashtempl"},"date":"2021-05-04 08:49:22.000000000","message":"Uploaded patch set 8.","accounts_in_message":[],"_revision_number":8},{"id":"54af369a5a9a051605296a1d6c6fbf90b81e89a2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-04 09:56:13.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/94b456c881d146998cfeff77f68e72df : SUCCESS in 41m 05s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/eba2d42e41d54f10b7bcb1bccfc80970 : SUCCESS in 33m 15s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/ba805a19d2d5456789e1d6f76607087a : SUCCESS in 1h 05m 44s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/3659df467b6b4081bcc9c0b2d1e73719 : SUCCESS in 40m 40s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/661f836a62444c4bac074e3f0eb178d8 : SUCCESS in 45m 25s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6c7ace08fcb4496d8df5ed242ce64b3c : SUCCESS in 4m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/74f6b41edc4248b19bbecc739bf8786e : SUCCESS in 3m 33s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/68bafc7930444df7aaae1e02e55b7c58 : SUCCESS in 29m 18s","accounts_in_message":[],"_revision_number":8},{"id":"2f1b6b654946715f09d779b634f7fe812bcac0ef","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2021-05-06 19:03:03.000000000","message":"Patch Set 8: Code-Review+2\n\n(2 comments)\n\nI will +2 this even though it will likely have another merge conflict simply because the conflict will depend on the order in which the patches merge.","accounts_in_message":[],"_revision_number":8},{"id":"0e9e260a32d02c2d37ea7d62ae3486d9a693535a","author":{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},"date":"2021-05-06 19:24:46.000000000","message":"Patch Set 8: Code-Review+2","accounts_in_message":[],"_revision_number":8},{"id":"8c9eb0543c20534fd6c437363413bfbb00d2b6f2","author":{"_account_id":28609,"name":"Arkady Shtempler","email":"ashtempl@redhat.com","username":"ashtempl"},"date":"2021-05-11 12:22:17.000000000","message":"Patch Set 8:\n\nNicolas, Michael - we are missing workflow set to +1 to start merging.","accounts_in_message":[],"_revision_number":8},{"id":"ea9ad8171940b363ca085bb367e2ed4af10bbb00","author":{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},"date":"2021-05-11 12:47:50.000000000","message":"Patch Set 8: Workflow+1","accounts_in_message":[],"_revision_number":8},{"id":"6b0820434b9c395ad238b8d12c778da062a61631","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-11 13:57:56.000000000","message":"Patch Set 8: Verified-2\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":8},{"id":"a08aac08bc63744db70dde6baddf328dfd8597f9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28609,"name":"Arkady Shtempler","email":"ashtempl@redhat.com","username":"ashtempl"},"date":"2021-05-12 10:01:43.000000000","message":"Uploaded patch set 9.","accounts_in_message":[],"_revision_number":9},{"id":"3e28975f3fdb46773df708f6a43a9b4f3b92aaa4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-12 11:10:24.000000000","message":"Patch Set 9: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/e72f660283734d37a686b07ee71f4f05 : FAILURE in 38m 25s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/0d0fbb85e6fe404abf9a11b7152cbd7b : FAILURE in 41m 30s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/49ccc8f1874b49d784e1275d21879238 : FAILURE in 1h 02m 40s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/db3b2245d04d47ba856669663eac0b1a : FAILURE in 35m 32s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/765f16d38156476b86f7347d2da6d47b : FAILURE in 37m 09s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/935f259da7ce48a590942c002f7913de : SUCCESS in 3m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3a849848e23c40dba7e2b93a785e7aaf : FAILURE in 3m 43s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/c5b32fd91fd7434594304c97d9505e79 : SUCCESS in 29m 53s","accounts_in_message":[],"_revision_number":9},{"id":"0cefdcbdecb091aad4088224692346b8ad7771c0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28609,"name":"Arkady Shtempler","email":"ashtempl@redhat.com","username":"ashtempl"},"date":"2021-05-12 11:37:33.000000000","message":"Uploaded patch set 10.","accounts_in_message":[],"_revision_number":10},{"id":"4164a15458106a49812131c0222ac48255d5b03b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-12 12:40:18.000000000","message":"Patch Set 10: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/a41f79d1063848fcaeac84ce5cf4381f : FAILURE in 37m 26s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/17038d16023e43bb8f88e1a26a2bb075 : FAILURE in 39m 14s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/3535691ac3da4f0d9bdacc61b9f50792 : FAILURE in 1h 01m 02s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/b5932bd2a7d647729c2b0a90c558c1ff : FAILURE in 39m 47s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/6810a6bc6c9149378f948d1a56f3d181 : POST_FAILURE in 40m 21s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/276656670b1a4ad799f5691b1e5a90df : SUCCESS in 3m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7105b7a3b15e4724b2837f852250e9fb : SUCCESS in 3m 46s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/43ba7d66900049b4a3f2b1cb6cda5c37 : SUCCESS in 28m 45s","accounts_in_message":[],"_revision_number":10},{"id":"c37ea9982e8c62142eb9ff03afbec28d38329999","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28609,"name":"Arkady Shtempler","email":"ashtempl@redhat.com","username":"ashtempl"},"date":"2021-05-12 13:47:33.000000000","message":"Uploaded patch set 11.","accounts_in_message":[],"_revision_number":11},{"id":"53bef7f285a06304d44271036007ac463f6914cc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-12 14:42:05.000000000","message":"Patch Set 11: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/2840565a60984a7398699dbd85148a00 : FAILURE in 31m 26s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/7d019a7854df43958360fc2edcf31b50 : FAILURE in 33m 59s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/14279adfaf814b028289a82713506bbe : FAILURE in 52m 41s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/30217400b7ac4ed88368b7f2e6b1e32d : FAILURE in 39m 28s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/ab446b87f8424f9cbedc490086a54c48 : FAILURE in 39m 19s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/743c0ec3efff498ca36fa2a7c0e2c7dc : SUCCESS in 4m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/84d627c8f9ea4e4da3f14f6b3e4d3a1e : SUCCESS in 4m 24s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/a101c72e8d3542dda1b4ae11d58d9d70 : SUCCESS in 32m 44s","accounts_in_message":[],"_revision_number":11},{"id":"27f19f47d98675b5309f03e626483948a13f2eec","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28609,"name":"Arkady Shtempler","email":"ashtempl@redhat.com","username":"ashtempl"},"date":"2021-05-13 10:40:36.000000000","message":"Uploaded patch set 12.","accounts_in_message":[],"_revision_number":12},{"id":"779f61163da4722ace9b025e49c5f4aef35337e2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-13 11:37:29.000000000","message":"Patch Set 12: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/3e379415df1f42fe92347f2436586181 : SUCCESS in 37m 02s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/9711d9c16a3b4136af338fc34d196707 : SUCCESS in 30m 01s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/1d2c475dae5e4c43aad63ddfb15e7fcb : FAILURE in 55m 47s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/b67bf979e50b4f47a5bcb2484e382827 : SUCCESS in 36m 21s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/e4db455c6de34ef7a00c9d175ba187fb : FAILURE in 33m 40s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0dc4c051af534ae7bc39c70c9107fff1 : SUCCESS in 3m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/519b7837c7224375b7cb2ac9f2ee7db1 : FAILURE in 3m 22s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/60d41645d8b248b5af098521579c31f3 : SUCCESS in 35m 40s","accounts_in_message":[],"_revision_number":12},{"id":"5ec55203f582b6a96375d35edc8d8b8849f1488e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28609,"name":"Arkady Shtempler","email":"ashtempl@redhat.com","username":"ashtempl"},"date":"2021-05-13 12:04:26.000000000","message":"Uploaded patch set 13.","accounts_in_message":[],"_revision_number":13},{"id":"3ac2210f0a8d326351b012af5b37d3062b8fc900","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-13 13:09:57.000000000","message":"Patch Set 13: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/5b054fbee33443a18a593b83589b7943 : SUCCESS in 32m 22s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/b863f3a9c95945c5bee8ba710e00003f : FAILURE in 25m 24s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/533d69f5ece1460b93761ab539d2ee30 : FAILURE in 1h 04m 54s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/ed6fd9d1181d4ee18c55eb9d5ed17902 : SUCCESS in 32m 56s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/5b1836716f1e41d6976c541e04f0ec92 : SUCCESS in 37m 10s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/837231ee46ab408ba1e1f89ff5803a4f : SUCCESS in 3m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fc5f8d03138a4a38b95b484e28e1cf1e : SUCCESS in 3m 42s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/9e9b2e2163aa4aa0b23095ce6f710e6c : SUCCESS in 33m 43s","accounts_in_message":[],"_revision_number":13},{"id":"b22197ba7706bf4c1fd493ba3b022a699f85268e","author":{"_account_id":28609,"name":"Arkady Shtempler","email":"ashtempl@redhat.com","username":"ashtempl"},"date":"2021-05-13 14:28:53.000000000","message":"Patch Set 13:\n\nrecheck","accounts_in_message":[],"_revision_number":13},{"id":"9d07325283597ea616dc66f30e2d6c30f9a50f53","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-13 15:26:04.000000000","message":"Patch Set 13:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/fd03d04ac95d4ded88fd1e73e34a86e5 : SUCCESS in 42m 47s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/a653f4d084234332a19b6a08059a726b : SUCCESS in 39m 44s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/c16f55ca9ced45398a255aa5f677242e : FAILURE in 55m 15s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/e001fc3a1e0d4e5b9acda638e19b63aa : FAILURE in 38m 08s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/5b65e16133ed43638949c28a5e141ae0 : FAILURE in 36m 56s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0c48ea0028364178832522375d21c09b : SUCCESS in 3m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f1f22e53d5994a8795a6d7b413d2edf5 : SUCCESS in 3m 56s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/e1273ee4f1344576a4eb3f411563d7bb : SUCCESS in 29m 12s","accounts_in_message":[],"_revision_number":13},{"id":"5a23c2ddbcd317b9fc03674e7db78099e3ad8df8","author":{"_account_id":28609,"name":"Arkady Shtempler","email":"ashtempl@redhat.com","username":"ashtempl"},"date":"2021-05-14 07:12:26.000000000","message":"Patch Set 13:\n\nrecheck","accounts_in_message":[],"_revision_number":13},{"id":"1941e11eff7acd48c118db2ec28d39274bc9f843","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-14 08:18:02.000000000","message":"Patch Set 13:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/ab43cf0765024a96a0044713345d90f4 : SUCCESS in 27m 31s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/61d7fed1f770483a942252c918b1e51f : SUCCESS in 31m 41s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/fd0d525c15a940fba82c2eaebced76dd : SUCCESS in 1h 04m 18s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/783f9da6b49b47769daaf62d90333c22 : FAILURE in 35m 25s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/3af41af907964241b8cafdd382d3739d : SUCCESS in 41m 19s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8546b180a92c4e449cf3092a229e49a4 : SUCCESS in 4m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/832d07136e68449cbc9ee44187c396c4 : SUCCESS in 3m 33s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/d37c6a1e02624b068d0c6f79714ee6b9 : SUCCESS in 33m 17s","accounts_in_message":[],"_revision_number":13},{"id":"98632e50a3a705f65212e3a682ef68afe86e6581","author":{"_account_id":28609,"name":"Arkady Shtempler","email":"ashtempl@redhat.com","username":"ashtempl"},"date":"2021-05-14 08:34:03.000000000","message":"Patch Set 13:\n\nrecheck","accounts_in_message":[],"_revision_number":13},{"id":"900ca430fe7e5871f9f402973f17149d31989582","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-14 09:35:33.000000000","message":"Patch Set 13: Verified+1\n\nBuild succeeded (check pipeline).\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/2d0011b9ddde4ea381ded8455fab12d4 : SUCCESS in 35m 40s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/3743830137ea4ee3a245b20db751c317 : SUCCESS in 35m 09s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/c4f65f9fcb214ec1ad1f90a0c970aa54 : SUCCESS in 57m 32s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/e9b9fa10fc4d4fa5bd7de279915b7cc6 : SUCCESS in 36m 32s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/8e686bd2f2044d8eaa47f01564ddafeb : SUCCESS in 37m 34s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4eff824514994993803b2572ba96c5c3 : SUCCESS in 3m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/85df211db7104861a797af184b9ba521 : SUCCESS in 3m 58s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/9c51c9c0d0e54e00b7ed251b2af5d70f : SUCCESS in 31m 04s","accounts_in_message":[],"_revision_number":13},{"id":"7b932fc16dfed2bd7798836df9b2a782abb152bb","author":{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},"date":"2021-05-14 19:33:15.000000000","message":"Patch Set 13: Code-Review+2\n\nThanks!","accounts_in_message":[],"_revision_number":13},{"id":"e6f888b4efe8cc89b4c9ab0ec7ec1544fd474b02","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2021-05-20 18:35:00.000000000","message":"Patch Set 13: Code-Review+2 Workflow+1\n\nLGTM","accounts_in_message":[],"_revision_number":13},{"id":"0a7148587c7c411be9de24b0cc3a6cf8a856dd35","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-20 18:35:28.000000000","message":"Patch Set 13: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":13},{"id":"f5b3dbf809e941725fcda51b23a6713004262f73","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-20 19:40:22.000000000","message":"Patch Set 13: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/8aa668525b2941408240fe44154559ec : SUCCESS in 38m 15s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/6c5b635fa1304b1ca8cbd08a4bd3d9b7 : FAILURE in 36m 41s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/c9763cfa19944a7b913fc5685edd1a7e : SUCCESS in 1h 00m 56s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/9ea0d511d053495d908be0479ed24489 : SUCCESS in 37m 12s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/2a1ab73f541e4e53b74060a533ce83a1 : FAILURE in 39m 38s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1b3d22554d274704bed780d59d572b1f : SUCCESS in 3m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9b658ce914a94473b1d7d08ec1271833 : SUCCESS in 3m 37s","accounts_in_message":[],"_revision_number":13},{"id":"72ab40aee628b4c19aafe6d64ba262f088ede056","author":{"_account_id":28609,"name":"Arkady Shtempler","email":"ashtempl@redhat.com","username":"ashtempl"},"date":"2021-05-20 19:49:02.000000000","message":"Patch Set 13:\n\nrecheck","accounts_in_message":[],"_revision_number":13},{"id":"f75fe1e4346b5ffea9b570180cfac23bc38d7da0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-20 20:52:59.000000000","message":"Patch Set 13: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/36b4b53e5bf841ad9fd2567f3c2c44a3 : SUCCESS in 40m 25s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/5a9b91d1311a421ca30836f1fcf008c9 : SUCCESS in 34m 36s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/c9f7df6d9f294fc7b64ab9419ae75f71 : SUCCESS in 59m 03s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/a92faaf33ff24400bba2bfb881614e6f : FAILURE in 36m 34s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/2aa2cd5cff6549028d56ec9975550f3a : SUCCESS in 37m 12s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/db0e1fa90d9c4c2caa8475f90243a3c6 : SUCCESS in 3m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9ba1aca39c58465e81562c86bf03ac65 : SUCCESS in 4m 04s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/0ae24670273942f49a154d814ce10da2 : SUCCESS in 24m 18s","accounts_in_message":[],"_revision_number":13},{"id":"3218d66ad644cdd775c0cfbaa09f9e0d1eb2ccd5","author":{"_account_id":28609,"name":"Arkady Shtempler","email":"ashtempl@redhat.com","username":"ashtempl"},"date":"2021-05-21 04:16:29.000000000","message":"Patch Set 13:\n\nrecheck","accounts_in_message":[],"_revision_number":13},{"id":"96da049697ee52fd98063d4dde27844e4fa970dc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-21 05:27:58.000000000","message":"Patch Set 13: Verified+1\n\nBuild succeeded (check pipeline).\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/8cc317009c3c4038ae52fd006cb271be : SUCCESS in 39m 48s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/0727db9b6bcb468faf7b18550be35d81 : SUCCESS in 36m 46s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/67505593b59e4107b2121068b9b6cd37 : SUCCESS in 1h 05m 47s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/0a399271e46c4d0cad2fa8e4a642fb25 : SUCCESS in 39m 26s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/21314a5979c64a84b652f1d454f8c2ce : SUCCESS in 36m 57s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1cd7d0131b104801892d0909b0c3b87e : SUCCESS in 3m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5d834c3e416040ff9ddc0bded2e06e0b : SUCCESS in 3m 21s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/fb586cb290994071be43142a56217a9c : SUCCESS in 30m 12s","accounts_in_message":[],"_revision_number":13},{"id":"29a3ae8102cb83f29a16ec49accae066ade9013c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-21 05:28:17.000000000","message":"Patch Set 13: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":13},{"id":"66eeb1fb508283b786155edda8f0fef940b19686","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-21 06:33:52.000000000","message":"Patch Set 13: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/b7c785b2aff94017a30935b9d38b0015 : SUCCESS in 34m 35s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/cde6613423254b9ea526a7295b46de21 : SUCCESS in 39m 21s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/e16163bc38fb4c3bb157ac6000bf5d82 : SUCCESS in 1h 01m 42s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/b2e64d8e36664247a6c35bce28a40b64 : SUCCESS in 36m 40s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/0addde55e363468b8deb5d11f1bfd9f2 : SUCCESS in 38m 42s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4f68b2255e2e4175ba7b6730eeb22dc4 : SUCCESS in 4m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8a8d4bc22e1243f4aa0b2348a66520c9 : SUCCESS in 3m 48s","accounts_in_message":[],"_revision_number":13},{"id":"99d1696feeee79f8e226f9b32ea41a9cb7b96f3c","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-21 06:33:53.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":13},{"id":"7bd5b386781007f6465c7f6cd33b11419950eb69","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-21 06:36:01.000000000","message":"Patch Set 13:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3c74a14a7a2d4216927891b77d50dcd4 : SUCCESS in 1m 46s","accounts_in_message":[],"_revision_number":13},{"id":"8f6c777e1cb09c231c38c541430c7c08a0537249","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2021-06-03 18:12:25.000000000","message":"Patch Set 13:\n\nThis test has a pre-condition bug as shown here: https://storage.gra.cloud.ovh.net/v1/AUTH_dcaab5e32b234d56b626f72581e3644c/zuul_opendev_logs_a92/781442/13/check/designate-ipv6-only-pdns4/a92faaf/testr_results.html\n\nWe need to fix this or revert this test patch. It is causing job failures for other patches.","accounts_in_message":[],"_revision_number":13},{"id":"dfd27d1f1027101015daf808486c542cf709b8d9","author":{"_account_id":28609,"name":"Arkady Shtempler","email":"ashtempl@redhat.com","username":"ashtempl"},"date":"2021-06-04 06:14:07.000000000","message":"Patch Set 13:\n\n\u003e Patch Set 13:\n\u003e \n\u003e This test has a pre-condition bug as shown here: https://storage.gra.cloud.ovh.net/v1/AUTH_dcaab5e32b234d56b626f72581e3644c/zuul_opendev_logs_a92/781442/13/check/designate-ipv6-only-pdns4/a92faaf/testr_results.html\n\u003e \n\u003e We need to fix this or revert this test patch. It is causing job failures for other patches.\n\nHi all!\n\nUnfortunately I wasn\u0027t able to reproduce this issue and it\u0027s not so clear from API response what exactly went wrong (Designate bug if the Error message is not clear to you as well, maybe instead of \"invalid_zone_name\" it should be \"invalid_ptrdname or \"invalid_domain\"?):\n\nBody: {\"ptrdname\": \"779206967.com.\", \"ttl\": 28984, \"description\": \"tempest-test-ptr-459118540\"}\nResponse - Headers: {\u0027date\u0027: \u0027Thu, 20 May 2021 20:20:46 GMT\u0027, \u0027server\u0027: \u0027Apache/2.4.41 (Ubuntu)\u0027, \u0027content-length\u0027: \u0027126\u0027, \u0027x-openstack-request-id\u0027: \u0027req-5041cd6b-5122-4aa8-8ab9-844dda6d8b94\u0027, \u0027content-type\u0027: \u0027application/json\u0027, \u0027connection\u0027: \u0027close\u0027, \u0027status\u0027: \u0027400\u0027, \u0027content-location\u0027: \u0027https://[2001:4801:7824:104:be76:4eff:fe10:1ebf]/dns/v2/reverse/floatingips/RegionOne:4365ad6e-b143-4d72-88bc-d86c052d85a1\u0027}\n        Body: b\u0027{\"code\": 400, \"type\": \"invalid_zone_name\", \"message\": \"Invalid TLD\", \"request_id\": \"req-5041cd6b-5122-4aa8-8ab9-844dda6d8b94\"}\u0027\n\n\nI\u0027ve uploaded this patch:\nhttps://review.opendev.org/c/openstack/designate-tempest-plugin/+/794708\nthat is now adds a \"test\" prefix by default for creating a random \"ptrdname\", so maybe that used \"779206967.com.\" could have caused failure? Is it not a valid value? \n\nThanks!","accounts_in_message":[],"_revision_number":13}],"current_revision_number":13,"current_revision":"7986b469bee80d3a8923ec5a7dfae05f5113e032","revisions":{"fc24cf77b108e7ebb3500b866d2acb398e126974":{"kind":"REWORK","_number":1,"created":"2021-03-18 12:42:27.000000000","uploader":{"_account_id":28609,"name":"Arkady Shtempler","email":"ashtempl@redhat.com","username":"ashtempl"},"ref":"refs/changes/42/781442/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate-tempest-plugin","ref":"refs/changes/42/781442/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate-tempest-plugin refs/changes/42/781442/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate-tempest-plugin refs/changes/42/781442/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate-tempest-plugin refs/changes/42/781442/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate-tempest-plugin refs/changes/42/781442/1"}}},"commit":{"parents":[{"commit":"e98a05ac3520a8264ce1a31b1e9a0faef45bebc1","subject":"Testing \"Lists all recordsets owned by a project in Designate\" API","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate-tempest-plugin/commit/e98a05ac3520a8264ce1a31b1e9a0faef45bebc1"}]}],"author":{"name":"Arkady Shtempler","email":"ashtempl@redhat.com","date":"2021-03-18 12:38:00.000000000","tz":120},"committer":{"name":"Arkady Shtempler","email":"ashtempl@redhat.com","date":"2021-03-18 12:38:07.000000000","tz":120},"subject":"Test cases for: \"FloatingIPs PTR\" resords\"","message":"Test cases for: \"FloatingIPs PTR\" resords\"\n\n1) test_set_floatingip_ptr\n  Set a PTR record for the given FloatingIP.\n2) test_show_floatingip_ptr\n  Shows a particular FloatingIP PTR\n3) test_list_floatingip_ptr_records\n  List FloatingIP PTR records\n4) test_unset_floatingip_ptr\n  Unset the PTR record for a FloatingIP\n\nChange-Id: Ib7a3b829129f2534f67c66c5f58e7c7c3c2f93c9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate-tempest-plugin/commit/fc24cf77b108e7ebb3500b866d2acb398e126974"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate-tempest-plugin/commit/fc24cf77b108e7ebb3500b866d2acb398e126974"}]},"branch":"refs/heads/master"},"63ca4805e7daa8c516e3af7b39ca82f688437005":{"kind":"REWORK","_number":2,"created":"2021-03-18 14:59:40.000000000","uploader":{"_account_id":28609,"name":"Arkady Shtempler","email":"ashtempl@redhat.com","username":"ashtempl"},"ref":"refs/changes/42/781442/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate-tempest-plugin","ref":"refs/changes/42/781442/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate-tempest-plugin refs/changes/42/781442/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate-tempest-plugin refs/changes/42/781442/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate-tempest-plugin refs/changes/42/781442/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate-tempest-plugin refs/changes/42/781442/2"}}},"commit":{"parents":[{"commit":"e98a05ac3520a8264ce1a31b1e9a0faef45bebc1","subject":"Testing \"Lists all recordsets owned by a project in Designate\" API","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate-tempest-plugin/commit/e98a05ac3520a8264ce1a31b1e9a0faef45bebc1"}]}],"author":{"name":"Arkady Shtempler","email":"ashtempl@redhat.com","date":"2021-03-18 12:38:00.000000000","tz":120},"committer":{"name":"Arkady Shtempler","email":"ashtempl@redhat.com","date":"2021-03-18 14:59:16.000000000","tz":120},"subject":"Test cases for: \"FloatingIPs PTR\" resords\"","message":"Test cases for: \"FloatingIPs PTR\" resords\"\n\n1) test_set_floatingip_ptr\n  Set a PTR record for the given FloatingIP.\n2) test_show_floatingip_ptr\n  Shows a particular FloatingIP PTR\n3) test_list_floatingip_ptr_records\n  List FloatingIP PTR records\n4) test_unset_floatingip_ptr\n  Unset the PTR record for a FloatingIP\n\nChange-Id: Ib7a3b829129f2534f67c66c5f58e7c7c3c2f93c9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate-tempest-plugin/commit/63ca4805e7daa8c516e3af7b39ca82f688437005"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate-tempest-plugin/commit/63ca4805e7daa8c516e3af7b39ca82f688437005"}]},"branch":"refs/heads/master"},"4e250fa716aa6c1fa4eb08654242fc6a8802ebe5":{"kind":"REWORK","_number":3,"created":"2021-03-18 15:05:28.000000000","uploader":{"_account_id":28609,"name":"Arkady Shtempler","email":"ashtempl@redhat.com","username":"ashtempl"},"ref":"refs/changes/42/781442/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate-tempest-plugin","ref":"refs/changes/42/781442/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate-tempest-plugin refs/changes/42/781442/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate-tempest-plugin refs/changes/42/781442/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate-tempest-plugin refs/changes/42/781442/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate-tempest-plugin refs/changes/42/781442/3"}}},"commit":{"parents":[{"commit":"e98a05ac3520a8264ce1a31b1e9a0faef45bebc1","subject":"Testing \"Lists all recordsets owned by a project in Designate\" API","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate-tempest-plugin/commit/e98a05ac3520a8264ce1a31b1e9a0faef45bebc1"}]}],"author":{"name":"Arkady Shtempler","email":"ashtempl@redhat.com","date":"2021-03-18 12:38:00.000000000","tz":120},"committer":{"name":"Arkady Shtempler","email":"ashtempl@redhat.com","date":"2021-03-18 15:05:11.000000000","tz":120},"subject":"Test cases for: \"FloatingIPs PTR\" resords\"","message":"Test cases for: \"FloatingIPs PTR\" resords\"\n\n1) test_set_floatingip_ptr\n  Set a PTR record for the given FloatingIP.\n2) test_show_floatingip_ptr\n  Shows a particular FloatingIP PTR\n3) test_list_floatingip_ptr_records\n  List FloatingIP PTR records\n4) test_unset_floatingip_ptr\n  Unset the PTR record for a FloatingIP\n\nChange-Id: Ib7a3b829129f2534f67c66c5f58e7c7c3c2f93c9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate-tempest-plugin/commit/4e250fa716aa6c1fa4eb08654242fc6a8802ebe5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate-tempest-plugin/commit/4e250fa716aa6c1fa4eb08654242fc6a8802ebe5"}]},"branch":"refs/heads/master"},"77fae4da12b4ddc5427039dc39f60e8ec22629cb":{"kind":"REWORK","_number":4,"created":"2021-03-18 17:57:52.000000000","uploader":{"_account_id":28609,"name":"Arkady Shtempler","email":"ashtempl@redhat.com","username":"ashtempl"},"ref":"refs/changes/42/781442/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate-tempest-plugin","ref":"refs/changes/42/781442/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate-tempest-plugin refs/changes/42/781442/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate-tempest-plugin refs/changes/42/781442/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate-tempest-plugin refs/changes/42/781442/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate-tempest-plugin refs/changes/42/781442/4"}}},"commit":{"parents":[{"commit":"e98a05ac3520a8264ce1a31b1e9a0faef45bebc1","subject":"Testing \"Lists all recordsets owned by a project in Designate\" API","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate-tempest-plugin/commit/e98a05ac3520a8264ce1a31b1e9a0faef45bebc1"}]}],"author":{"name":"Arkady Shtempler","email":"ashtempl@redhat.com","date":"2021-03-18 12:38:00.000000000","tz":120},"committer":{"name":"Arkady Shtempler","email":"ashtempl@redhat.com","date":"2021-03-18 17:48:06.000000000","tz":120},"subject":"Test cases for: \"FloatingIPs PTR\" resords\"","message":"Test cases for: \"FloatingIPs PTR\" resords\"\n\n1) test_set_floatingip_ptr\n  Set a PTR record for the given FloatingIP.\n2) test_show_floatingip_ptr\n  Shows a particular FloatingIP PTR\n3) test_list_floatingip_ptr_records\n  List FloatingIP PTR records\n4) test_unset_floatingip_ptr\n  Unset the PTR record for a FloatingIP\n\nChange-Id: Ib7a3b829129f2534f67c66c5f58e7c7c3c2f93c9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate-tempest-plugin/commit/77fae4da12b4ddc5427039dc39f60e8ec22629cb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate-tempest-plugin/commit/77fae4da12b4ddc5427039dc39f60e8ec22629cb"}]},"branch":"refs/heads/master"},"fe2e02c420013466c105ac0f22c30bb7e965e0b3":{"kind":"REWORK","_number":5,"created":"2021-03-18 20:14:18.000000000","uploader":{"_account_id":28609,"name":"Arkady Shtempler","email":"ashtempl@redhat.com","username":"ashtempl"},"ref":"refs/changes/42/781442/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate-tempest-plugin","ref":"refs/changes/42/781442/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate-tempest-plugin refs/changes/42/781442/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate-tempest-plugin refs/changes/42/781442/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate-tempest-plugin refs/changes/42/781442/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate-tempest-plugin refs/changes/42/781442/5"}}},"commit":{"parents":[{"commit":"e98a05ac3520a8264ce1a31b1e9a0faef45bebc1","subject":"Testing \"Lists all recordsets owned by a project in Designate\" API","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate-tempest-plugin/commit/e98a05ac3520a8264ce1a31b1e9a0faef45bebc1"}]}],"author":{"name":"Arkady Shtempler","email":"ashtempl@redhat.com","date":"2021-03-18 12:38:00.000000000","tz":120},"committer":{"name":"Arkady Shtempler","email":"ashtempl@redhat.com","date":"2021-03-18 20:13:38.000000000","tz":120},"subject":"Test cases for: \"FloatingIPs PTR\" resords.","message":"Test cases for: \"FloatingIPs PTR\" resords.\n\n1) test_set_floatingip_ptr\n   Set a PTR record for the given FloatingIP.\n2) test_show_floatingip_ptr\n   Shows a particular FloatingIP PTR\n3) test_list_floatingip_ptr_records\n   List FloatingIP PTR records\n4) test_unset_floatingip_ptr\n   Unset the PTR record for a FloatingIP\n\nChange-Id: Ib7a3b829129f2534f67c66c5f58e7c7c3c2f93c9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate-tempest-plugin/commit/fe2e02c420013466c105ac0f22c30bb7e965e0b3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate-tempest-plugin/commit/fe2e02c420013466c105ac0f22c30bb7e965e0b3"}]},"branch":"refs/heads/master"},"f951d770830cdb81c12eadd4de3a7dbeded6ce17":{"kind":"REWORK","_number":6,"created":"2021-04-27 19:52:22.000000000","uploader":{"_account_id":28609,"name":"Arkady Shtempler","email":"ashtempl@redhat.com","username":"ashtempl"},"ref":"refs/changes/42/781442/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate-tempest-plugin","ref":"refs/changes/42/781442/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate-tempest-plugin refs/changes/42/781442/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate-tempest-plugin refs/changes/42/781442/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate-tempest-plugin refs/changes/42/781442/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate-tempest-plugin refs/changes/42/781442/6"}}},"commit":{"parents":[{"commit":"e98a05ac3520a8264ce1a31b1e9a0faef45bebc1","subject":"Testing \"Lists all recordsets owned by a project in Designate\" API","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate-tempest-plugin/commit/e98a05ac3520a8264ce1a31b1e9a0faef45bebc1"}]}],"author":{"name":"Arkady Shtempler","email":"ashtempl@redhat.com","date":"2021-03-18 12:38:00.000000000","tz":120},"committer":{"name":"Arkady Shtempler","email":"ashtempl@redhat.com","date":"2021-04-27 19:50:41.000000000","tz":180},"subject":"Test cases for: \"FloatingIPs PTR\" resords.","message":"Test cases for: \"FloatingIPs PTR\" resords.\n\n1) test_set_floatingip_ptr\n   Set a PTR record for the given FloatingIP.\n2) test_show_floatingip_ptr\n   Shows a particular FloatingIP PTR\n3) test_list_floatingip_ptr_records\n   List FloatingIP PTR records\n4) test_unset_floatingip_ptr\n   Unset the PTR record for a FloatingIP\n5) test_set_floatingip_ptr_invalid_ttl\n   Try to use invalid TTL to set PTR record.\n   Expected 400 BadRequest\n\nChange-Id: Ib7a3b829129f2534f67c66c5f58e7c7c3c2f93c9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate-tempest-plugin/commit/f951d770830cdb81c12eadd4de3a7dbeded6ce17"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate-tempest-plugin/commit/f951d770830cdb81c12eadd4de3a7dbeded6ce17"}]},"branch":"refs/heads/master"},"5bf03fe38c90fd15b307da10d2a593a537fb3686":{"kind":"REWORK","_number":7,"created":"2021-05-04 07:48:13.000000000","uploader":{"_account_id":28609,"name":"Arkady Shtempler","email":"ashtempl@redhat.com","username":"ashtempl"},"ref":"refs/changes/42/781442/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate-tempest-plugin","ref":"refs/changes/42/781442/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate-tempest-plugin refs/changes/42/781442/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate-tempest-plugin refs/changes/42/781442/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate-tempest-plugin refs/changes/42/781442/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate-tempest-plugin refs/changes/42/781442/7"}}},"commit":{"parents":[{"commit":"6d3ef7847d77456c4ce359fef372abfae7159608","subject":"Merge \"New test cases for \"zone recordsets\" test suite\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate-tempest-plugin/commit/6d3ef7847d77456c4ce359fef372abfae7159608"}]}],"author":{"name":"Arkady Shtempler","email":"ashtempl@redhat.com","date":"2021-03-18 12:38:00.000000000","tz":120},"committer":{"name":"Arkady Shtempler","email":"ashtempl@redhat.com","date":"2021-05-04 07:48:02.000000000","tz":180},"subject":"Test cases for: \"FloatingIPs PTR\" resords.","message":"Test cases for: \"FloatingIPs PTR\" resords.\n\n1) test_set_floatingip_ptr\n   Set a PTR record for the given FloatingIP.\n2) test_show_floatingip_ptr\n   Shows a particular FloatingIP PTR\n3) test_list_floatingip_ptr_records\n   List FloatingIP PTR records\n4) test_unset_floatingip_ptr\n   Unset the PTR record for a FloatingIP\n5) test_set_floatingip_ptr_invalid_ttl\n   Try to use invalid TTL to set PTR record.\n   Expected 400 BadRequest\n\nChange-Id: Ib7a3b829129f2534f67c66c5f58e7c7c3c2f93c9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate-tempest-plugin/commit/5bf03fe38c90fd15b307da10d2a593a537fb3686"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate-tempest-plugin/commit/5bf03fe38c90fd15b307da10d2a593a537fb3686"}]},"branch":"refs/heads/master"},"d346e648a83ab9aad9a1797cd226179f08230206":{"kind":"REWORK","_number":8,"created":"2021-05-04 08:49:22.000000000","uploader":{"_account_id":28609,"name":"Arkady Shtempler","email":"ashtempl@redhat.com","username":"ashtempl"},"ref":"refs/changes/42/781442/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate-tempest-plugin","ref":"refs/changes/42/781442/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate-tempest-plugin refs/changes/42/781442/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate-tempest-plugin refs/changes/42/781442/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate-tempest-plugin refs/changes/42/781442/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate-tempest-plugin refs/changes/42/781442/8"}}},"commit":{"parents":[{"commit":"6d3ef7847d77456c4ce359fef372abfae7159608","subject":"Merge \"New test cases for \"zone recordsets\" test suite\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate-tempest-plugin/commit/6d3ef7847d77456c4ce359fef372abfae7159608"}]}],"author":{"name":"Arkady Shtempler","email":"ashtempl@redhat.com","date":"2021-03-18 12:38:00.000000000","tz":120},"committer":{"name":"Arkady Shtempler","email":"ashtempl@redhat.com","date":"2021-05-04 08:49:05.000000000","tz":180},"subject":"Test cases for: \"FloatingIPs PTR\" resords.","message":"Test cases for: \"FloatingIPs PTR\" resords.\n\n1) test_set_floatingip_ptr\n   Set a PTR record for the given FloatingIP.\n2) test_show_floatingip_ptr\n   Shows a particular FloatingIP PTR\n3) test_list_floatingip_ptr_records\n   List FloatingIP PTR records\n4) test_unset_floatingip_ptr\n   Unset the PTR record for a FloatingIP\n5) test_set_floatingip_ptr_invalid_ttl\n   Try to use invalid TTL to set PTR record.\n   Expected 400 BadRequest\n\nChange-Id: Ib7a3b829129f2534f67c66c5f58e7c7c3c2f93c9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate-tempest-plugin/commit/d346e648a83ab9aad9a1797cd226179f08230206"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate-tempest-plugin/commit/d346e648a83ab9aad9a1797cd226179f08230206"}]},"branch":"refs/heads/master"},"3c13160bf34afe93ab6c2ea97da0d7133ee5aafd":{"kind":"REWORK","_number":9,"created":"2021-05-12 10:01:43.000000000","uploader":{"_account_id":28609,"name":"Arkady Shtempler","email":"ashtempl@redhat.com","username":"ashtempl"},"ref":"refs/changes/42/781442/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate-tempest-plugin","ref":"refs/changes/42/781442/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate-tempest-plugin refs/changes/42/781442/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate-tempest-plugin refs/changes/42/781442/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate-tempest-plugin refs/changes/42/781442/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate-tempest-plugin refs/changes/42/781442/9"}}},"commit":{"parents":[{"commit":"199d6bc74da8398fffe9fc62b8b09770f3fd521d","subject":"Merge \"New test case for a \"list export zone\", filter listed results.\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate-tempest-plugin/commit/199d6bc74da8398fffe9fc62b8b09770f3fd521d"}]}],"author":{"name":"Arkady Shtempler","email":"ashtempl@redhat.com","date":"2021-03-18 12:38:00.000000000","tz":120},"committer":{"name":"Arkady Shtempler","email":"ashtempl@redhat.com","date":"2021-05-12 10:01:19.000000000","tz":180},"subject":"Test cases for: \"FloatingIPs PTR\" resords.","message":"Test cases for: \"FloatingIPs PTR\" resords.\n\n1) test_set_floatingip_ptr\n   Set a PTR record for the given FloatingIP.\n2) test_show_floatingip_ptr\n   Shows a particular FloatingIP PTR\n3) test_list_floatingip_ptr_records\n   List FloatingIP PTR records\n4) test_unset_floatingip_ptr\n   Unset the PTR record for a FloatingIP\n5) test_set_floatingip_ptr_invalid_ttl\n   Try to use invalid TTL to set PTR record.\n   Expected 400 BadRequest\n\nChange-Id: Ib7a3b829129f2534f67c66c5f58e7c7c3c2f93c9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate-tempest-plugin/commit/3c13160bf34afe93ab6c2ea97da0d7133ee5aafd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate-tempest-plugin/commit/3c13160bf34afe93ab6c2ea97da0d7133ee5aafd"}]},"branch":"refs/heads/master"},"a7f3928d21a47f9da1e18fc3e50aab117f5c257e":{"kind":"REWORK","_number":10,"created":"2021-05-12 11:37:33.000000000","uploader":{"_account_id":28609,"name":"Arkady Shtempler","email":"ashtempl@redhat.com","username":"ashtempl"},"ref":"refs/changes/42/781442/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate-tempest-plugin","ref":"refs/changes/42/781442/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate-tempest-plugin refs/changes/42/781442/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate-tempest-plugin refs/changes/42/781442/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate-tempest-plugin refs/changes/42/781442/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate-tempest-plugin refs/changes/42/781442/10"}}},"commit":{"parents":[{"commit":"199d6bc74da8398fffe9fc62b8b09770f3fd521d","subject":"Merge \"New test case for a \"list export zone\", filter listed results.\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate-tempest-plugin/commit/199d6bc74da8398fffe9fc62b8b09770f3fd521d"}]}],"author":{"name":"Arkady Shtempler","email":"ashtempl@redhat.com","date":"2021-03-18 12:38:00.000000000","tz":120},"committer":{"name":"Arkady Shtempler","email":"ashtempl@redhat.com","date":"2021-05-12 11:37:18.000000000","tz":180},"subject":"Test cases for: \"FloatingIPs PTR\" resords.","message":"Test cases for: \"FloatingIPs PTR\" resords.\n\n1) test_set_floatingip_ptr\n   Set a PTR record for the given FloatingIP.\n2) test_show_floatingip_ptr\n   Shows a particular FloatingIP PTR\n3) test_list_floatingip_ptr_records\n   List FloatingIP PTR records\n4) test_unset_floatingip_ptr\n   Unset the PTR record for a FloatingIP\n5) test_set_floatingip_ptr_invalid_ttl\n   Try to use invalid TTL to set PTR record.\n   Expected 400 BadRequest\n\nChange-Id: Ib7a3b829129f2534f67c66c5f58e7c7c3c2f93c9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate-tempest-plugin/commit/a7f3928d21a47f9da1e18fc3e50aab117f5c257e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate-tempest-plugin/commit/a7f3928d21a47f9da1e18fc3e50aab117f5c257e"}]},"branch":"refs/heads/master"},"ed10bfd01a926afdb2aacdbe31f4799bbe259ac1":{"kind":"REWORK","_number":11,"created":"2021-05-12 13:47:33.000000000","uploader":{"_account_id":28609,"name":"Arkady Shtempler","email":"ashtempl@redhat.com","username":"ashtempl"},"ref":"refs/changes/42/781442/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate-tempest-plugin","ref":"refs/changes/42/781442/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate-tempest-plugin refs/changes/42/781442/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate-tempest-plugin refs/changes/42/781442/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate-tempest-plugin refs/changes/42/781442/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate-tempest-plugin refs/changes/42/781442/11"}}},"commit":{"parents":[{"commit":"199d6bc74da8398fffe9fc62b8b09770f3fd521d","subject":"Merge \"New test case for a \"list export zone\", filter listed results.\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate-tempest-plugin/commit/199d6bc74da8398fffe9fc62b8b09770f3fd521d"}]}],"author":{"name":"Arkady Shtempler","email":"ashtempl@redhat.com","date":"2021-03-18 12:38:00.000000000","tz":120},"committer":{"name":"Arkady Shtempler","email":"ashtempl@redhat.com","date":"2021-05-12 13:47:20.000000000","tz":180},"subject":"Test cases for: \"FloatingIPs PTR\" resords.","message":"Test cases for: \"FloatingIPs PTR\" resords.\n\n1) test_set_floatingip_ptr\n   Set a PTR record for the given FloatingIP.\n2) test_show_floatingip_ptr\n   Shows a particular FloatingIP PTR\n3) test_list_floatingip_ptr_records\n   List FloatingIP PTR records\n4) test_unset_floatingip_ptr\n   Unset the PTR record for a FloatingIP\n5) test_set_floatingip_ptr_invalid_ttl\n   Try to use invalid TTL to set PTR record.\n   Expected 400 BadRequest\n\nChange-Id: Ib7a3b829129f2534f67c66c5f58e7c7c3c2f93c9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate-tempest-plugin/commit/ed10bfd01a926afdb2aacdbe31f4799bbe259ac1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate-tempest-plugin/commit/ed10bfd01a926afdb2aacdbe31f4799bbe259ac1"}]},"branch":"refs/heads/master"},"f36b1e87a9a7f07ada7b913d61848b56e20e8375":{"kind":"REWORK","_number":12,"created":"2021-05-13 10:40:36.000000000","uploader":{"_account_id":28609,"name":"Arkady Shtempler","email":"ashtempl@redhat.com","username":"ashtempl"},"ref":"refs/changes/42/781442/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate-tempest-plugin","ref":"refs/changes/42/781442/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate-tempest-plugin refs/changes/42/781442/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate-tempest-plugin refs/changes/42/781442/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate-tempest-plugin refs/changes/42/781442/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate-tempest-plugin refs/changes/42/781442/12"}}},"commit":{"parents":[{"commit":"199d6bc74da8398fffe9fc62b8b09770f3fd521d","subject":"Merge \"New test case for a \"list export zone\", filter listed results.\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate-tempest-plugin/commit/199d6bc74da8398fffe9fc62b8b09770f3fd521d"}]}],"author":{"name":"Arkady Shtempler","email":"ashtempl@redhat.com","date":"2021-03-18 12:38:00.000000000","tz":120},"committer":{"name":"Arkady Shtempler","email":"ashtempl@redhat.com","date":"2021-05-13 10:40:25.000000000","tz":180},"subject":"Test cases for: \"FloatingIPs PTR\" resords.","message":"Test cases for: \"FloatingIPs PTR\" resords.\n\n1) test_set_floatingip_ptr\n   Set a PTR record for the given FloatingIP.\n2) test_show_floatingip_ptr\n   Shows a particular FloatingIP PTR\n3) test_list_floatingip_ptr_records\n   List FloatingIP PTR records\n4) test_unset_floatingip_ptr\n   Unset the PTR record for a FloatingIP\n5) test_set_floatingip_ptr_invalid_ttl\n   Try to use invalid TTL to set PTR record.\n   Expected 400 BadRequest\n\nChange-Id: Ib7a3b829129f2534f67c66c5f58e7c7c3c2f93c9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate-tempest-plugin/commit/f36b1e87a9a7f07ada7b913d61848b56e20e8375"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate-tempest-plugin/commit/f36b1e87a9a7f07ada7b913d61848b56e20e8375"}]},"branch":"refs/heads/master"},"7986b469bee80d3a8923ec5a7dfae05f5113e032":{"kind":"REWORK","_number":13,"created":"2021-05-13 12:04:26.000000000","uploader":{"_account_id":28609,"name":"Arkady Shtempler","email":"ashtempl@redhat.com","username":"ashtempl"},"ref":"refs/changes/42/781442/13","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate-tempest-plugin","ref":"refs/changes/42/781442/13","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate-tempest-plugin refs/changes/42/781442/13 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate-tempest-plugin refs/changes/42/781442/13 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate-tempest-plugin refs/changes/42/781442/13 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate-tempest-plugin refs/changes/42/781442/13"}}},"commit":{"parents":[{"commit":"199d6bc74da8398fffe9fc62b8b09770f3fd521d","subject":"Merge \"New test case for a \"list export zone\", filter listed results.\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate-tempest-plugin/commit/199d6bc74da8398fffe9fc62b8b09770f3fd521d"}]}],"author":{"name":"Arkady Shtempler","email":"ashtempl@redhat.com","date":"2021-03-18 12:38:00.000000000","tz":120},"committer":{"name":"Arkady Shtempler","email":"ashtempl@redhat.com","date":"2021-05-13 12:04:16.000000000","tz":180},"subject":"Test cases for: \"FloatingIPs PTR\" resords.","message":"Test cases for: \"FloatingIPs PTR\" resords.\n\n1) test_set_floatingip_ptr\n   Set a PTR record for the given FloatingIP.\n2) test_show_floatingip_ptr\n   Shows a particular FloatingIP PTR\n3) test_list_floatingip_ptr_records\n   List FloatingIP PTR records\n4) test_unset_floatingip_ptr\n   Unset the PTR record for a FloatingIP\n5) test_set_floatingip_ptr_invalid_ttl\n   Try to use invalid TTL to set PTR record.\n   Expected 400 BadRequest\n\nChange-Id: Ib7a3b829129f2534f67c66c5f58e7c7c3c2f93c9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate-tempest-plugin/commit/7986b469bee80d3a8923ec5a7dfae05f5113e032"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate-tempest-plugin/commit/7986b469bee80d3a8923ec5a7dfae05f5113e032"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
