)]}'
{"id":"openstack%2Fdesignate-tempest-plugin~797904","triplet_id":"openstack%2Fdesignate-tempest-plugin~master~Id01a1fc5eaf9d08039696795c6121e20fbe42bc7","project":"openstack/designate-tempest-plugin","branch":"master","topic":"new_tld_test_cases","attention_set":{},"removed_from_attention_set":{"28609":{"account":{"_account_id":28609,"name":"Arkady Shtempler","email":"ashtempl@redhat.com","username":"ashtempl"},"last_update":"2021-10-20 09:27:01.000000000","reason":"Arkady Shtempler replied on the change"},"11628":{"account":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"last_update":"2021-10-20 11:16:20.000000000","reason":"Change was submitted"},"22623":{"account":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"last_update":"2021-10-20 02:53:40.000000000","reason":"Erik Olof Gunnar Andersson replied on the change"}},"hashtags":[],"change_id":"Id01a1fc5eaf9d08039696795c6121e20fbe42bc7","subject":"New test cases for TLD tests suite","status":"MERGED","created":"2021-06-24 13:36:55.000000000","updated":"2021-10-20 11:18:05.000000000","submitted":"2021-10-20 11:16:20.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":20,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"797904-new_tld_test_cases","meta_rev_id":"7b4546e707c7cb22db2e2d2620b2fe31586bb7da","_number":797904,"virtual_id_number":797904,"owner":{"_account_id":28609,"name":"Arkady Shtempler","email":"ashtempl@redhat.com","username":"ashtempl"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-10-20 11:16:19.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"all":[{"value":2,"date":"2021-10-06 21:40:27.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2021-10-20 02:53:40.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"all":[{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2021-10-20 02:55:25.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":15025,"name":"Michael Johnson","email":"Michael.Johnson@sas.com","username":"mikjo"},{"_account_id":29681,"name":"Bruna Bonguardo","display_name":"Bruna Bonguardo","email":"bbonguar@redhat.com","username":"bonguardo"},{"_account_id":6994,"name":"Michael Chapman","email":"woppin@gmail.com","username":"michaeltchapman"},{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":18894,"name":"Toni Freger","email":"tfreger@redhat.com"}],"reviewers":{"REVIEWER":[{"_account_id":6994,"name":"Michael Chapman","email":"woppin@gmail.com","username":"michaeltchapman"},{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":15025,"name":"Michael Johnson","email":"Michael.Johnson@sas.com","username":"mikjo"},{"_account_id":18894,"name":"Toni Freger","email":"tfreger@redhat.com"},{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},{"_account_id":29681,"name":"Bruna Bonguardo","display_name":"Bruna Bonguardo","email":"bbonguar@redhat.com","username":"bonguardo"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-06-24 13:40:25.000000000","updated_by":{"_account_id":28609,"name":"Arkady Shtempler","email":"ashtempl@redhat.com","username":"ashtempl"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"REVIEWER"},{"updated":"2021-06-24 13:40:25.000000000","updated_by":{"_account_id":28609,"name":"Arkady Shtempler","email":"ashtempl@redhat.com","username":"ashtempl"},"reviewer":{"_account_id":6994,"name":"Michael Chapman","email":"woppin@gmail.com","username":"michaeltchapman"},"state":"REVIEWER"},{"updated":"2021-06-24 13:40:25.000000000","updated_by":{"_account_id":28609,"name":"Arkady Shtempler","email":"ashtempl@redhat.com","username":"ashtempl"},"reviewer":{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},"state":"REVIEWER"},{"updated":"2021-06-24 13:40:25.000000000","updated_by":{"_account_id":28609,"name":"Arkady Shtempler","email":"ashtempl@redhat.com","username":"ashtempl"},"reviewer":{"_account_id":29681,"name":"Bruna Bonguardo","display_name":"Bruna Bonguardo","email":"bbonguar@redhat.com","username":"bonguardo"},"state":"REVIEWER"},{"updated":"2021-06-24 13:40:25.000000000","updated_by":{"_account_id":28609,"name":"Arkady Shtempler","email":"ashtempl@redhat.com","username":"ashtempl"},"reviewer":{"_account_id":15025,"name":"Michael Johnson","email":"Michael.Johnson@sas.com","username":"mikjo"},"state":"REVIEWER"},{"updated":"2021-06-24 13:40:25.000000000","updated_by":{"_account_id":28609,"name":"Arkady Shtempler","email":"ashtempl@redhat.com","username":"ashtempl"},"reviewer":{"_account_id":18894,"name":"Toni Freger","email":"tfreger@redhat.com"},"state":"REVIEWER"},{"updated":"2021-06-24 19:00:25.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-10-05 23:30:31.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"},{"updated":"2021-10-19 13:15:00.000000000","updated_by":{"_account_id":28609,"name":"Arkady Shtempler","email":"ashtempl@redhat.com","username":"ashtempl"},"reviewer":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"state":"REVIEWER"}],"messages":[{"id":"b538aca0d5097942abb52be72a055147ac6502f5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28609,"name":"Arkady Shtempler","email":"ashtempl@redhat.com","username":"ashtempl"},"date":"2021-06-24 13:36:55.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"ea6b73fff4a30fb3555fe18afe9677846e3390e8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-24 19:00:25.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/b576912c21af4d209e3c0970d1d69d24 : SUCCESS in 40m 06s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/08539d8b31d94994bb272dcd34214d70 : SUCCESS in 47m 57s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/4845a364cdf44a50976131d86fc5be81 : SUCCESS in 1h 10m 29s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/a6dd7d4de0df47569d5f3207af814939 : SUCCESS in 46m 35s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/1476dfca29fe4f0e82eb7f5f6fd8420b : SUCCESS in 42m 08s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1f402d2b49854cdba6a47132d4655326 : SUCCESS in 5m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/eadb3880b9ee4906b3621640aff6e833 : SUCCESS in 5m 45s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/2ad7967afe734ce5b82a8833e6b81998 : SUCCESS in 42m 30s","accounts_in_message":[],"_revision_number":1},{"id":"770ff32cc381f9c14548c8696ce5494f07487275","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2021-10-05 23:30:31.000000000","message":"Patch Set 1: Code-Review-1\n\n(9 comments)\n\nThere are a couple of updates needed here, plus a rebase onto the secure RBAC chain.","accounts_in_message":[],"_revision_number":1},{"id":"cc2db675dccc79a3aab0d822927c3ec5c22a53cb","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28609,"name":"Arkady Shtempler","email":"ashtempl@redhat.com","username":"ashtempl"},"date":"2021-10-06 10:29:03.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"1aeddfa3fb462a63185322fc4ee7f7f8ae9c3daa","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-06 11:40:48.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/a48a6c9a06d540f6bd17e146a84f4a9a : SUCCESS in 41m 36s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/7b414f95fbe44f7b9e1e46db16406228 : SUCCESS in 43m 47s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/4958df9f908c404aa11dfcc06ed00914 : SUCCESS in 1h 02m 44s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/70b104a7ce104e2196f6865e502e6620 : SUCCESS in 53m 52s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/2b602c5b7b7e4838b351af6813f89edd : SUCCESS in 39m 01s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9b53d646865e4bb29edbf5032f309634 : SUCCESS in 6m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e2236d77554a4149862ed5b6fda856ef : SUCCESS in 4m 13s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/166dda0343f14712b28e7e3e40ac2185 : SUCCESS in 38m 39s","accounts_in_message":[],"_revision_number":2},{"id":"83e99825906176f93cb441a30a9585d4889b9f38","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28609,"name":"Arkady Shtempler","email":"ashtempl@redhat.com","username":"ashtempl"},"date":"2021-10-06 13:39:53.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"94f87cde17056e346e94755bea6575c85ccb7e44","author":{"_account_id":28609,"name":"Arkady Shtempler","email":"ashtempl@redhat.com","username":"ashtempl"},"date":"2021-10-06 13:40:51.000000000","message":"Patch Set 3:\n\n(9 comments)","accounts_in_message":[],"_revision_number":3},{"id":"606343ea2c3044c953004cec1bcc80be65342002","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-06 14:43:15.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/0fefbafe7c7345d78d07953598f30a9d : SUCCESS in 42m 12s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/1a68de3d46974239a88fe1d2116058ab : SUCCESS in 36m 54s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/f8b5d24485a549b28a0f8ba1a014c8b6 : SUCCESS in 55m 39s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/eb5046fcf4474a23902f0da5fc8a30e2 : SUCCESS in 40m 33s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/6a0be15f6f0e460296b88c7720d5b55c : SUCCESS in 40m 35s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d1fce4b3357f4c1b99884b8a4b5c64b0 : SUCCESS in 3m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f5473b58814d4bd28b3e3b3add21dd78 : SUCCESS in 4m 13s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/954e3e1d30f0493c812f0eea46682102 : FAILURE in 31m 37s","accounts_in_message":[],"_revision_number":3},{"id":"e7032dbdf2bbc24ddf0fa2c579dc67d03267db58","author":{"_account_id":28609,"name":"Arkady Shtempler","email":"ashtempl@redhat.com","username":"ashtempl"},"date":"2021-10-06 15:28:53.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"5901a9c28eae949939d331147f118c58500888c4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-06 16:35:02.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/82c40561ae5045c0a006dea15f399b36 : SUCCESS in 47m 48s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/f565cc5b9d2242ca9384dee4030968e8 : SUCCESS in 38m 41s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/e59dd7f4a23549c7a4a5af09913f3e7d : SUCCESS in 1h 00m 44s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/d49e90973d954f309095aaca323dbda5 : SUCCESS in 39m 39s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/47727c5e10d1436ebb924ce7e2a69c66 : SUCCESS in 41m 45s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fa3ff5c0257e4e01bf342ee6a4d2fb70 : SUCCESS in 4m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b36ddf1ed2f14d0a853fc4d14fd023c1 : SUCCESS in 4m 08s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/45b609ccc83641ce89516fc8df5e00b1 : SUCCESS in 33m 49s","accounts_in_message":[],"_revision_number":3},{"id":"ee40c36f3e7e6d452c91281be71aa91ec3b4ca82","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2021-10-06 21:39:54.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"7cdd72e5e76882bd133075620fbcc5005f396d31","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2021-10-06 21:40:27.000000000","message":"Patch Set 4: Code-Review+2\n\nLGTM","accounts_in_message":[],"_revision_number":4},{"id":"fc684a351dfc1978ded9d1201f33157559a9f947","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-06 22:46:39.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/575dc2a4d74749c89e77986a4802fc48 : SUCCESS in 48m 03s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/0ce30c4878b141cc8df88cb1756ab690 : SUCCESS in 46m 24s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/745e3ca8c7254c539d2520310273ee76 : SUCCESS in 1h 00m 13s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/8b349e13f5c54bc29c09a26164a8bf2d : SUCCESS in 38m 56s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/a8318da72d124ea3af5b1fdc56824572 : SUCCESS in 44m 23s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5f70906bf79244ed9266ffb9efa5410a : SUCCESS in 4m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0c0a25061b5f4685a03f759f4ce91ece : SUCCESS in 3m 56s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/87cc6097877f46a5978bca0e8d555faf : SUCCESS in 34m 39s","accounts_in_message":[],"_revision_number":4},{"id":"d3dc78d32eb731df9ada8e4d931702a68e072bde","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2021-10-20 02:53:40.000000000","message":"Patch Set 4: Code-Review+2","accounts_in_message":[],"_revision_number":4},{"id":"489beb796fdce129bf738ff45697eee54d431d60","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2021-10-20 02:55:25.000000000","message":"Patch Set 4: Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"8c0486b18c149fd4848860fd4098aaf361790fb4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-20 05:16:13.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"3ffad3f9ee926bb74903aaeadceb009f9fd3cd8e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-20 06:37:23.000000000","message":"Patch Set 4: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/ef92213b7ab749b79a792a0385b9cb72 : TIMED_OUT in 1h 13m 47s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/288dcd0474ef422f848298f3ac9ea4f4 : SUCCESS in 37m 12s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/55e8f1e1748940aa82f9cf5bfe2b8779 : SUCCESS in 1h 00m 21s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/7f60b6eeeedf4fb396d920af83a63b1c : SUCCESS in 46m 39s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/e5166f81bf8d47648d82efe3c8e8e8cb : SUCCESS in 37m 12s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/47ce135137fe4a05be658e5e3f1f2825 : SUCCESS in 4m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4433d944d2b741ef90fe625dafc11cfa : SUCCESS in 4m 43s","accounts_in_message":[],"_revision_number":4},{"id":"9717103bc1be17cf42a055893237430a8c9f7afc","author":{"_account_id":28609,"name":"Arkady Shtempler","email":"ashtempl@redhat.com","username":"ashtempl"},"date":"2021-10-20 07:02:26.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"a0299eabf458b4b5dc07c43ae8f5c335a83c645c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-20 08:06:41.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/dde334129f1a4cf6b65e5c02008d922c : SUCCESS in 45m 48s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/ac52d8af34ab4331b399a5a405ae5388 : SUCCESS in 51m 40s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/68de0fd13320479f93478997478cc398 : SUCCESS in 1h 01m 23s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/afee52b34e164ba38899cb5526be19fb : SUCCESS in 41m 50s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/ee37b69bb6a14c648c86e3dbb90f0f8f : SUCCESS in 38m 32s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/816d89cb0a674de9ab72a6cfffd3aeb5 : SUCCESS in 5m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c10f628cde87416795b7e8dbff529552 : SUCCESS in 5m 08s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/7fc3614c05d748ef84705b069c3c0153 : SUCCESS in 39m 20s","accounts_in_message":[],"_revision_number":4},{"id":"e64a20c5e28260640b233431e46f04598d1aa1ee","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-20 08:07:21.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"9dd90318dca9f72dfd2d0a82bed30fa48e03a4bf","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-20 09:21:11.000000000","message":"Patch Set 4: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/23a442075e8e4788be54f5bd98b4392f : FAILURE in 1h 13m 40s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/56ffce2a64be47e38aa2f045d8391a5d : SUCCESS in 38m 27s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/31abae4bc2d044e5bb48938e031cfa67 : SUCCESS in 49m 47s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/06e00ee81efd4ed58a2463969172a224 : SUCCESS in 36m 44s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/1fe31f90231a4e95a4afc7b7c5bf62d1 : SUCCESS in 35m 16s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2565212f526542908bdb8d037305a537 : SUCCESS in 4m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1b5f1a62ce59444aa69a50b4f021cfcd : SUCCESS in 4m 16s","accounts_in_message":[],"_revision_number":4},{"id":"f26f78db52f8f653b86c1c45bad168f6018fce95","author":{"_account_id":28609,"name":"Arkady Shtempler","email":"ashtempl@redhat.com","username":"ashtempl"},"date":"2021-10-20 09:27:01.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"d0cc8a1146a7c34214819a30eec54d7a67bd6f48","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-20 10:25:11.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/9f15cb76ee694e218bd9551ca54c6228 : SUCCESS in 35m 33s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/1a1a2c00cd184a97813b3dadccf38974 : SUCCESS in 51m 07s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/5e00d08cde9c47f287c61701d8175d72 : SUCCESS in 40m 04s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/a3bdebb3113a4f62b358ef30283fc46d : SUCCESS in 37m 27s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/7bd7c02053434557937b0f01fc9e1352 : SUCCESS in 38m 21s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9d03bb3b036346ce8aa4de82d6c3cf7a : SUCCESS in 4m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/702d5040da864aeda75ff4d7ea58d082 : SUCCESS in 5m 08s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/477e661306b04da4a5d88f5f1f673ab8 : SUCCESS in 45m 20s","accounts_in_message":[],"_revision_number":4},{"id":"19bad18fb2d620fb775887787f6dc3891b177c89","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-20 10:25:48.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"18504a90a0b7922dc4c0e9bfaf2b2023bd4043e2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-20 11:16:19.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/99d3dbebb22949a0a53b2f2216d0a993 : SUCCESS in 40m 28s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/4e6b4439d85d4071aaab3995414b4859 : SUCCESS in 41m 58s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/13145c04c215481289270e3630222578 : SUCCESS in 49m 27s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/9d81956389944aa5b288edd93d7c85ec : SUCCESS in 34m 48s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/f10fae4b86044d41a5b86d754d5a88c5 : SUCCESS in 42m 09s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ec0c4b88a3c14da4abbbf74d1601715b : SUCCESS in 4m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/140b13f144b14595831326bef4970331 : SUCCESS in 3m 48s","accounts_in_message":[],"_revision_number":4},{"id":"6f38e662f81738ea08f291bfd7da330d17a69034","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-20 11:16:20.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":4},{"id":"7b4546e707c7cb22db2e2d2620b2fe31586bb7da","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-20 11:18:05.000000000","message":"Patch Set 4:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e01709a2e2f04053b9a74990977d807d : SUCCESS in 1m 09s","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"d6a6de5da84b34646badcef23f2c79bd7c6550f9","revisions":{"a35d48f04c2bd46e9c5819834286ced9023e967d":{"kind":"REWORK","_number":1,"created":"2021-06-24 13:36:55.000000000","uploader":{"_account_id":28609,"name":"Arkady Shtempler","email":"ashtempl@redhat.com","username":"ashtempl"},"ref":"refs/changes/04/797904/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate-tempest-plugin","ref":"refs/changes/04/797904/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate-tempest-plugin refs/changes/04/797904/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate-tempest-plugin refs/changes/04/797904/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate-tempest-plugin refs/changes/04/797904/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate-tempest-plugin refs/changes/04/797904/1"}}},"commit":{"parents":[{"commit":"da27a70ae2b39695ef6f03bbefb55afeacf1cdf3","subject":"Merge \"Re-enable zone export list filter test\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate-tempest-plugin/commit/da27a70ae2b39695ef6f03bbefb55afeacf1cdf3"}]}],"author":{"name":"zahlabut","email":"ashtempl@redhat.com","date":"2021-06-24 13:26:24.000000000","tz":180},"committer":{"name":"zahlabut","email":"ashtempl@redhat.com","date":"2021-06-24 13:36:36.000000000","tz":180},"subject":"New test cases for TLD tests suite","message":"New test cases for TLD tests suite\n\n1) test_create_duplicated_tlds\n   Try to create the same TLD twice, first attempt should PASS\n   while second fails with 409 Conflict.\n2) test_create_multiply_tlds\n   Create 3 diferent TLDs, expected should PASS.\n3) test_create_invalid_tld\n   Attempts to create TLD using invalid values and\n   huge size description. Expected 400 BadRequest\n4) test_create_zone_for_not_existing_tld\n   Create a zone with existing TLD, should PASS.\n   Create a zone with non existing TLD, should fail with\n   400 BadRequest.\n5) test_create_tld_as_primary_user\n   Try to create a TLD usig Primary tenant.\n   Expected: should fail with 403 Forbidden\n\nChange-Id: Id01a1fc5eaf9d08039696795c6121e20fbe42bc7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate-tempest-plugin/commit/a35d48f04c2bd46e9c5819834286ced9023e967d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate-tempest-plugin/commit/a35d48f04c2bd46e9c5819834286ced9023e967d"}]},"branch":"refs/heads/master"},"bbf22cf6493f2a5682c191e4e5b9728fa4bca114":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2021-10-06 10:29:03.000000000","uploader":{"_account_id":28609,"name":"Arkady Shtempler","email":"ashtempl@redhat.com","username":"ashtempl"},"ref":"refs/changes/04/797904/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate-tempest-plugin","ref":"refs/changes/04/797904/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate-tempest-plugin refs/changes/04/797904/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate-tempest-plugin refs/changes/04/797904/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate-tempest-plugin refs/changes/04/797904/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate-tempest-plugin refs/changes/04/797904/2"}}},"commit":{"parents":[{"commit":"dcfa7c0372e41dd7a7c208ce04209f978de6539a","subject":"Remove v1 API tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate-tempest-plugin/commit/dcfa7c0372e41dd7a7c208ce04209f978de6539a"}]}],"author":{"name":"zahlabut","email":"ashtempl@redhat.com","date":"2021-06-24 13:26:24.000000000","tz":180},"committer":{"name":"Arkady Shtempler","email":"ashtempl@redhat.com","date":"2021-10-06 10:29:03.000000000","tz":0},"subject":"New test cases for TLD tests suite","message":"New test cases for TLD tests suite\n\n1) test_create_duplicated_tlds\n   Try to create the same TLD twice, first attempt should PASS\n   while second fails with 409 Conflict.\n2) test_create_multiply_tlds\n   Create 3 diferent TLDs, expected should PASS.\n3) test_create_invalid_tld\n   Attempts to create TLD using invalid values and\n   huge size description. Expected 400 BadRequest\n4) test_create_zone_for_not_existing_tld\n   Create a zone with existing TLD, should PASS.\n   Create a zone with non existing TLD, should fail with\n   400 BadRequest.\n5) test_create_tld_as_primary_user\n   Try to create a TLD usig Primary tenant.\n   Expected: should fail with 403 Forbidden\n\nChange-Id: Id01a1fc5eaf9d08039696795c6121e20fbe42bc7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate-tempest-plugin/commit/bbf22cf6493f2a5682c191e4e5b9728fa4bca114"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate-tempest-plugin/commit/bbf22cf6493f2a5682c191e4e5b9728fa4bca114"}]},"branch":"refs/heads/master","description":"Rebase"},"9c4d94a24186eb36e867f822ee0e966e39149718":{"kind":"REWORK","_number":3,"created":"2021-10-06 13:39:53.000000000","uploader":{"_account_id":28609,"name":"Arkady Shtempler","email":"ashtempl@redhat.com","username":"ashtempl"},"ref":"refs/changes/04/797904/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate-tempest-plugin","ref":"refs/changes/04/797904/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate-tempest-plugin refs/changes/04/797904/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate-tempest-plugin refs/changes/04/797904/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate-tempest-plugin refs/changes/04/797904/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate-tempest-plugin refs/changes/04/797904/3"}}},"commit":{"parents":[{"commit":"da27a70ae2b39695ef6f03bbefb55afeacf1cdf3","subject":"Merge \"Re-enable zone export list filter test\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate-tempest-plugin/commit/da27a70ae2b39695ef6f03bbefb55afeacf1cdf3"}]}],"author":{"name":"zahlabut","email":"ashtempl@redhat.com","date":"2021-06-24 13:26:24.000000000","tz":180},"committer":{"name":"zahlabut","email":"ashtempl@redhat.com","date":"2021-10-06 13:39:02.000000000","tz":180},"subject":"New test cases for TLD tests suite","message":"New test cases for TLD tests suite\n\n1) test_create_duplicated_tlds\n   Try to create the same TLD twice, first attempt should PASS\n   while second fails with 409 Conflict.\n2) test_create_multiply_tlds\n   Create 3 diferent TLDs, expected should PASS.\n3) test_create_invalid_tld\n   Attempts to create TLD using invalid values and\n   huge size description. Expected 400 BadRequest\n4) test_create_zone_for_not_existing_tld\n   Create a zone with non existing TLD, should fail with\n   400 BadRequest.\n5) test_create_tld_as_primary_user\n   Try to create a TLD usig Primary tenant.\n   Expected: should fail with 403 Forbidden\n\nChange-Id: Id01a1fc5eaf9d08039696795c6121e20fbe42bc7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate-tempest-plugin/commit/9c4d94a24186eb36e867f822ee0e966e39149718"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate-tempest-plugin/commit/9c4d94a24186eb36e867f822ee0e966e39149718"}]},"branch":"refs/heads/master"},"d6a6de5da84b34646badcef23f2c79bd7c6550f9":{"kind":"REWORK","_number":4,"created":"2021-10-06 21:39:54.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/04/797904/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate-tempest-plugin","ref":"refs/changes/04/797904/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate-tempest-plugin refs/changes/04/797904/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate-tempest-plugin refs/changes/04/797904/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate-tempest-plugin refs/changes/04/797904/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate-tempest-plugin refs/changes/04/797904/4"}}},"commit":{"parents":[{"commit":"a3a236331c95b1a4cfb5975a0ef7f216feef15b4","subject":"Make the API tests compatible with scoped tokens","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate-tempest-plugin/commit/a3a236331c95b1a4cfb5975a0ef7f216feef15b4"}]}],"author":{"name":"zahlabut","email":"ashtempl@redhat.com","date":"2021-06-24 13:26:24.000000000","tz":180},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2021-10-06 21:38:49.000000000","tz":0},"subject":"New test cases for TLD tests suite","message":"New test cases for TLD tests suite\n\n1) test_create_duplicated_tlds\n   Try to create the same TLD twice, first attempt should PASS\n   while second fails with 409 Conflict.\n2) test_create_multiply_tlds\n   Create 3 diferent TLDs, expected should PASS.\n3) test_create_invalid_tld\n   Attempts to create TLD using invalid values and\n   huge size description. Expected 400 BadRequest\n4) test_create_zone_for_not_existing_tld\n   Create a zone with non existing TLD, should fail with\n   400 BadRequest.\n5) test_create_tld_as_primary_user\n   Try to create a TLD usig Primary tenant.\n   Expected: should fail with 403 Forbidden\n\nChange-Id: Id01a1fc5eaf9d08039696795c6121e20fbe42bc7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate-tempest-plugin/commit/d6a6de5da84b34646badcef23f2c79bd7c6550f9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate-tempest-plugin/commit/d6a6de5da84b34646badcef23f2c79bd7c6550f9"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
