)]}'
{"id":"openstack%2Fdesignate-tempest-plugin~814497","triplet_id":"openstack%2Fdesignate-tempest-plugin~master~I6d7c9358a336eedb313157ebad666d37d18f03d3","project":"openstack/designate-tempest-plugin","branch":"master","topic":"bug/1943115","attention_set":{},"removed_from_attention_set":{"13252":{"account":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"last_update":"2021-10-28 19:19:01.000000000","reason":"Dr. Jens Harbott (frickler) replied on the change"},"11628":{"account":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"last_update":"2021-11-14 01:23:49.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I6d7c9358a336eedb313157ebad666d37d18f03d3","subject":"Unroll DDT API tests used for refstack/interop","status":"MERGED","created":"2021-10-18 22:51:40.000000000","updated":"2021-11-14 01:25:17.000000000","submitted":"2021-11-14 01:23:49.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":11,"unresolved_comment_count":2,"has_review_started":true,"submission_id":"814497-bug/1943115","meta_rev_id":"becece2a2d0c1ef24bcd5f8590da0241255c27cc","_number":814497,"virtual_id_number":814497,"owner":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-11-14 01:23:49.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":2,"date":"2021-10-28 19:19:01.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":2,"date":"2021-11-14 00:17:00.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":1,"date":"2021-11-14 00:17:00.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-10-18 23:58:00.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-10-23 11:43:53.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"CC"},{"updated":"2021-10-26 07:15:00.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"REVIEWER"},{"updated":"2021-11-14 00:17:00.000000000","updated_by":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"reviewer":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"state":"REVIEWER"}],"messages":[{"id":"3240adb6754f16eedb532917cf44badd7ffd9f8e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2021-10-18 22:51:40.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"a15ad1816b5d97d983066579b7ac0635face6011","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2021-10-18 22:57:08.000000000","message":"Patch Set 1: Workflow-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"91ecc3cd4354f69644336453d20b724ef44a621c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-18 23:58:00.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/cb9aa356ef1242c3917335e86e5fb839 : SUCCESS in 41m 47s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/1cdeb84e17864415af3c5c8c15c3f340 : SUCCESS in 48m 43s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/4b52470832414646a1aed4e9b2ba1b32 : SUCCESS in 1h 00m 07s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/53bed438fd3949ae862819ba8e00a922 : SUCCESS in 49m 40s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/e84808667a5745009f33cd265e76acc8 : SUCCESS in 45m 45s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/aad7d6f3901d4c60a754e9df5fb82c46 : SUCCESS in 3m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7e951f4996fa4c219729e6cdf2891583 : SUCCESS in 3m 49s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/05faa789daec42a2aec45fee0c3a56f8 : SUCCESS in 50m 25s","accounts_in_message":[],"_revision_number":1},{"id":"279c0af0d2069542e2dc98861bba391e64cc3f25","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2021-10-21 13:38:21.000000000","message":"Patch Set 1: -Workflow Review-Priority+1","accounts_in_message":[],"_revision_number":1},{"id":"09b21d7c50d0a9334514d02650a22a1e9d60f33d","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2021-10-23 11:43:53.000000000","message":"Patch Set 1:\n\n(4 comments)","accounts_in_message":[],"_revision_number":1},{"id":"5a6b13225ea2804b6db0b74f2e4d7662388e6158","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2021-10-25 18:23:39.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"42d3cf9c6bd868ae0c18543e86bf8c4f0449ceec","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2021-10-25 19:22:19.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"ed218d2a4528a454acc32486fb26d09a2032c061","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2021-10-25 19:22:26.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"4f36d361cf614bc403f2126931c0d77b242b3a04","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2021-10-25 19:23:08.000000000","message":"Patch Set 3: Published edit on patch set 2.","accounts_in_message":[],"_revision_number":3},{"id":"b5d129c008d7eb97f773feda99fe1da66340423c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-25 20:09:52.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/da4f840c3e4f41b781e00cda06bbd52c : SUCCESS in 35m 14s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/4c798f0ec6584e889b5220980e871f7f : SUCCESS in 31m 26s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/7d2955f101ce488297aaaf6c35108392 : SUCCESS in 44m 21s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/8c7c903ddcb2455185742c28ac73d41a : SUCCESS in 34m 00s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/738a244d1cc5484393f860358bb7b36a : SUCCESS in 32m 51s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1bbb4955d88a42b89ef8886f710758c5 : SUCCESS in 3m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fbcee77f19724604ba9bd620daf05103 : SUCCESS in 3m 23s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/50c1bfed0a574a3089a030f612039dfe : SUCCESS in 25m 20s","accounts_in_message":[],"_revision_number":3},{"id":"45de71196f953c5c41b536f4540658f8d527209e","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2021-10-26 07:15:00.000000000","message":"Patch Set 3: Code-Review+2 Review-Priority+1\n\n(2 comments)","accounts_in_message":[],"_revision_number":3},{"id":"9ea547e24b4768439d2f31143d79f5e7e8b9f4a0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2021-10-26 20:46:23.000000000","message":"Patch Set 4: Published edit on patch set 3.","accounts_in_message":[],"_revision_number":4},{"id":"8084d137d42e407a0a7b562c515debc89ead657b","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2021-10-26 20:47:33.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"8f73fb5e8e823306158b2535f2f284d6e4017c4c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-26 22:02:02.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/e01f6be56f44409a876c4698a432e15d : SUCCESS in 48m 07s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/429491e5225f48d5ada4cce0a9071635 : SUCCESS in 40m 07s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/3826a8c24599465ab3f33257d4111213 : SUCCESS in 1h 07m 25s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/fa70c0a8deb24c638403adb5dcf39b11 : SUCCESS in 42m 20s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/eadc7c0b32c845d38f0e0599b38d3ad5 : FAILURE in 46m 46s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fc19250fff974a899a110599c7907cc8 : SUCCESS in 5m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/dd6090a5e3d8405bb120de12916a2bfc : SUCCESS in 3m 49s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/5790cdcb1d584ffdbf7f740dfa74cad4 : SUCCESS in 41m 25s","accounts_in_message":[],"_revision_number":4},{"id":"98847dfb38d9443186c2c7163ed1744e5ff3dbe4","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2021-10-27 19:39:39.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"0c73e21fe88b325d2dc1a50a5ea62fcabaede4a6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-27 20:43:17.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/a336bdc1c5a94e6db22a00bf67d01f54 : SUCCESS in 39m 45s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/022cf9168b1c4d1e96f5fbaf20f4b99e : SUCCESS in 40m 23s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/080655d3df9e4ba08990404f57179b9a : SUCCESS in 56m 06s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/a3fccaace63c488ab78a8969445fea3d : SUCCESS in 38m 42s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/a159067f5117426ca963d421022251e7 : SUCCESS in 42m 00s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/94b8c265511f4d569253c57601813f87 : SUCCESS in 5m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ff5c8324d4794b91b1d1d5583406a9c1 : SUCCESS in 4m 25s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/643143bcc85c4faab428532e66ebdcb7 : SUCCESS in 53m 22s","accounts_in_message":[],"_revision_number":4},{"id":"f49b1b74a831dff5330f9bed50a3846d4cc75992","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2021-10-28 19:19:01.000000000","message":"Patch Set 4: Code-Review+2","accounts_in_message":[],"_revision_number":4},{"id":"7a08706d2f9d0f7dcc3cf4c2e12fabbc5825a8c8","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2021-11-14 00:17:00.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"d43a1c0e810f4319e8723c030feeae3438f16044","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-14 00:17:38.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"5bab4c3086108d78d61a87fa4a43e5cc0118d45d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-14 01:23:49.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/7ef5c999d5f5488e88426c3e8f165f70 : SUCCESS in 43m 52s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/8b50fd82289843b686c41c86153cb5d6 : SUCCESS in 40m 38s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/bd7417c1e4b447bc8f643c519e901952 : SUCCESS in 1h 01m 33s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/c254bc4b3f694d24b02fe2e0bc77a330 : SUCCESS in 37m 38s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/4f0823a8b79942699b23d30419a8dca6 : SUCCESS in 40m 14s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/db287997cc8840418bf094a55dfac2c6 : SUCCESS in 3m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/52510aac2f5f4a529769fa930b8e5307 : SUCCESS in 3m 33s","accounts_in_message":[],"_revision_number":4},{"id":"a8e64a2a47fa46c5ae75fdbd9da07801e3bc660f","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-14 01:23:49.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":4},{"id":"becece2a2d0c1ef24bcd5f8590da0241255c27cc","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-14 01:25:17.000000000","message":"Patch Set 4:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4ab1e6881f424e829badf6070a95fbd9 : SUCCESS in 1m 01s","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"f871a0e74f91e417adae5d7dd2a6cc17a62ba347","revisions":{"89dc6679503bc446c31a510ec5d08a1887efab55":{"kind":"REWORK","_number":1,"created":"2021-10-18 22:51:40.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/97/814497/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate-tempest-plugin","ref":"refs/changes/97/814497/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate-tempest-plugin refs/changes/97/814497/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate-tempest-plugin refs/changes/97/814497/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate-tempest-plugin refs/changes/97/814497/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate-tempest-plugin refs/changes/97/814497/1"}}},"commit":{"parents":[{"commit":"a9804e184a8f1214edbfca6d791626a756a3fa59","subject":"Merge \"Remove temporary client\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate-tempest-plugin/commit/a9804e184a8f1214edbfca6d791626a756a3fa59"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2021-10-18 22:46:01.000000000","tz":0},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2021-10-18 22:46:01.000000000","tz":0},"subject":"Unroll DDT API tests used for refstack/interop","message":"Unroll DDT API tests used for refstack/interop\n\nThis patch unrolls the test_create_all_recordset_types API tests from\nusing DDT. DDT is not OpenStack aware, so the idempotent_id\u0027s end up\nbeing duplicated and the test names can be inconsistent due to value\npadding.\nThis test has a limited number of combinations, so it is easy to unroll\nthese into stub tests that call a common, private, test method.\n\nCloses-Bug: 1943115\nChange-Id: I6d7c9358a336eedb313157ebad666d37d18f03d3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate-tempest-plugin/commit/89dc6679503bc446c31a510ec5d08a1887efab55"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate-tempest-plugin/commit/89dc6679503bc446c31a510ec5d08a1887efab55"}]},"branch":"refs/heads/master"},"847565b4c0c712e6e61275e2f84e7d3a07e82179":{"kind":"REWORK","_number":2,"created":"2021-10-25 19:22:19.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/97/814497/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate-tempest-plugin","ref":"refs/changes/97/814497/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate-tempest-plugin refs/changes/97/814497/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate-tempest-plugin refs/changes/97/814497/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate-tempest-plugin refs/changes/97/814497/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate-tempest-plugin refs/changes/97/814497/2"}}},"commit":{"parents":[{"commit":"a9804e184a8f1214edbfca6d791626a756a3fa59","subject":"Merge \"Remove temporary client\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate-tempest-plugin/commit/a9804e184a8f1214edbfca6d791626a756a3fa59"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2021-10-18 22:46:01.000000000","tz":0},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2021-10-25 19:22:15.000000000","tz":0},"subject":"Unroll DDT API tests used for refstack/interop","message":"Unroll DDT API tests used for refstack/interop\n\nThis patch unrolls the test_create_all_recordset_types API tests from\nusing DDT. DDT is not OpenStack aware, so the idempotent_id\u0027s end up\nbeing duplicated and the test names can be inconsistent due to value\npadding.\nThis test has a limited number of combinations, so it is easy to unroll\nthese into stub tests that call a common, private, test method.\n\nCloses-Bug: 1943115\nChange-Id: I6d7c9358a336eedb313157ebad666d37d18f03d3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate-tempest-plugin/commit/847565b4c0c712e6e61275e2f84e7d3a07e82179"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate-tempest-plugin/commit/847565b4c0c712e6e61275e2f84e7d3a07e82179"}]},"branch":"refs/heads/master"},"4b61ce03ddb4fc860b366313c1a7c4e8398e5b53":{"kind":"REWORK","_number":3,"created":"2021-10-25 19:23:08.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/97/814497/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate-tempest-plugin","ref":"refs/changes/97/814497/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate-tempest-plugin refs/changes/97/814497/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate-tempest-plugin refs/changes/97/814497/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate-tempest-plugin refs/changes/97/814497/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate-tempest-plugin refs/changes/97/814497/3"}}},"commit":{"parents":[{"commit":"a9804e184a8f1214edbfca6d791626a756a3fa59","subject":"Merge \"Remove temporary client\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate-tempest-plugin/commit/a9804e184a8f1214edbfca6d791626a756a3fa59"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2021-10-18 22:46:01.000000000","tz":0},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2021-10-25 19:23:08.000000000","tz":0},"subject":"Unroll DDT API tests used for refstack/interop","message":"Unroll DDT API tests used for refstack/interop\n\nThis patch unrolls the test_create_all_recordset_types API tests from\nusing DDT. DDT is not OpenStack aware, so the idempotent_id\u0027s end up\nbeing duplicated and the test names can be inconsistent due to value\npadding.\nThis test has a limited number of combinations, so it is easy to unroll\nthese into stub tests that call a common, private, test method.\n\nCloses-Bug: 1943115\nChange-Id: I6d7c9358a336eedb313157ebad666d37d18f03d3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate-tempest-plugin/commit/4b61ce03ddb4fc860b366313c1a7c4e8398e5b53"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate-tempest-plugin/commit/4b61ce03ddb4fc860b366313c1a7c4e8398e5b53"}]},"branch":"refs/heads/master"},"f871a0e74f91e417adae5d7dd2a6cc17a62ba347":{"kind":"REWORK","_number":4,"created":"2021-10-26 20:46:23.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/97/814497/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate-tempest-plugin","ref":"refs/changes/97/814497/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate-tempest-plugin refs/changes/97/814497/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate-tempest-plugin refs/changes/97/814497/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate-tempest-plugin refs/changes/97/814497/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate-tempest-plugin refs/changes/97/814497/4"}}},"commit":{"parents":[{"commit":"a9804e184a8f1214edbfca6d791626a756a3fa59","subject":"Merge \"Remove temporary client\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate-tempest-plugin/commit/a9804e184a8f1214edbfca6d791626a756a3fa59"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2021-10-18 22:46:01.000000000","tz":0},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2021-10-26 20:46:23.000000000","tz":0},"subject":"Unroll DDT API tests used for refstack/interop","message":"Unroll DDT API tests used for refstack/interop\n\nThis patch unrolls the test_create_all_recordset_types API tests from\nusing DDT. DDT is not OpenStack aware, so the idempotent_id\u0027s end up\nbeing duplicated and the test names can be inconsistent due to value\npadding.\nThis test has a limited number of combinations, so it is easy to unroll\nthese into stub tests that call a common, private, test method.\n\nCloses-Bug: 1943115\nChange-Id: I6d7c9358a336eedb313157ebad666d37d18f03d3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate-tempest-plugin/commit/f871a0e74f91e417adae5d7dd2a6cc17a62ba347"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate-tempest-plugin/commit/f871a0e74f91e417adae5d7dd2a6cc17a62ba347"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
