)]}'
{"id":"openstack%2Fdesignate-tempest-plugin~838298","triplet_id":"openstack%2Fdesignate-tempest-plugin~master~Ie9fca2598a3eab66f9cbd718e52e575744def160","project":"openstack/designate-tempest-plugin","branch":"master","topic":"tld_scenario","attention_set":{},"removed_from_attention_set":{"31664":{"account":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"last_update":"2022-05-11 15:36:24.000000000","reason":"Change was submitted"},"28609":{"account":{"_account_id":28609,"name":"Arkady Shtempler","email":"ashtempl@redhat.com","username":"ashtempl"},"last_update":"2022-05-11 15:36:24.000000000","reason":"Change was submitted"},"15025":{"account":{"_account_id":15025,"name":"Michael Johnson","email":"Michael.Johnson@sas.com","username":"mikjo"},"last_update":"2022-05-11 15:36:24.000000000","reason":"Change was submitted"},"29681":{"account":{"_account_id":29681,"name":"Bruna Bonguardo","display_name":"Bruna Bonguardo","email":"bbonguar@redhat.com","username":"bonguardo"},"last_update":"2022-05-11 15:36:24.000000000","reason":"Change was submitted"},"19298":{"account":{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},"last_update":"2022-05-11 14:27:15.000000000","reason":"Nicolas Bock replied on the change"},"34067":{"account":{"_account_id":34067,"name":"Lilach Avraham","email":"lavraham@redhat.com","username":"lavraham"},"last_update":"2022-05-11 15:36:24.000000000","reason":"Change was submitted"},"5572":{"account":{"_account_id":5572,"name":"Don Kehn","display_name":"DEKehn","email":"dekehn@gmail.com","username":"dekehn"},"last_update":"2022-04-27 21:28:44.000000000","reason":"DEKehn replied on the change"},"11628":{"account":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"last_update":"2022-05-10 22:04:57.000000000","reason":"Michael Johnson replied on the change"},"18894":{"account":{"_account_id":18894,"name":"Toni Freger","email":"tfreger@redhat.com"},"last_update":"2022-05-11 15:36:24.000000000","reason":"Change was submitted"},"22623":{"account":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"last_update":"2022-05-11 15:36:24.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ie9fca2598a3eab66f9cbd718e52e575744def160","subject":"New scenario test case for TLD + re-factoring","status":"MERGED","created":"2022-04-18 12:32:53.000000000","updated":"2022-05-11 15:37:46.000000000","submitted":"2022-05-11 15:36:24.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":9,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"838298-tld_scenario","meta_rev_id":"e7d88bc737d17923366322e224587423510677a7","_number":838298,"virtual_id_number":838298,"owner":{"_account_id":28609,"name":"Arkady Shtempler","email":"ashtempl@redhat.com","username":"ashtempl"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28609,"name":"Arkady Shtempler","email":"ashtempl@redhat.com","username":"ashtempl"},{"value":0,"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},{"value":0,"_account_id":5572,"name":"Don Kehn","display_name":"DEKehn","email":"dekehn@gmail.com","username":"dekehn"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-05-11 15:36:24.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},"all":[{"value":0,"_account_id":28609,"name":"Arkady Shtempler","email":"ashtempl@redhat.com","username":"ashtempl"},{"value":2,"date":"2022-05-11 14:27:15.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},{"value":1,"date":"2022-04-27 21:28:44.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":5572,"name":"Don Kehn","display_name":"DEKehn","email":"dekehn@gmail.com","username":"dekehn"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2022-05-10 22:04:57.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},"all":[{"value":0,"_account_id":28609,"name":"Arkady Shtempler","email":"ashtempl@redhat.com","username":"ashtempl"},{"value":1,"date":"2022-05-11 14:31:29.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},{"value":0,"_account_id":5572,"name":"Don Kehn","display_name":"DEKehn","email":"dekehn@gmail.com","username":"dekehn"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":28609,"name":"Arkady Shtempler","email":"ashtempl@redhat.com","username":"ashtempl"},{"value":0,"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},{"value":0,"_account_id":5572,"name":"Don Kehn","display_name":"DEKehn","email":"dekehn@gmail.com","username":"dekehn"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},{"_account_id":15025,"name":"Michael Johnson","email":"Michael.Johnson@sas.com","username":"mikjo"},{"_account_id":29681,"name":"Bruna Bonguardo","display_name":"Bruna Bonguardo","email":"bbonguar@redhat.com","username":"bonguardo"},{"_account_id":34067,"name":"Lilach Avraham","email":"lavraham@redhat.com","username":"lavraham"},{"_account_id":18894,"name":"Toni Freger","email":"tfreger@redhat.com"},{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"reviewers":{"REVIEWER":[{"_account_id":5572,"name":"Don Kehn","display_name":"DEKehn","email":"dekehn@gmail.com","username":"dekehn"},{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":15025,"name":"Michael Johnson","email":"Michael.Johnson@sas.com","username":"mikjo"},{"_account_id":18894,"name":"Toni Freger","email":"tfreger@redhat.com"},{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},{"_account_id":28609,"name":"Arkady Shtempler","email":"ashtempl@redhat.com","username":"ashtempl"},{"_account_id":29681,"name":"Bruna Bonguardo","display_name":"Bruna Bonguardo","email":"bbonguar@redhat.com","username":"bonguardo"},{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},{"_account_id":34067,"name":"Lilach Avraham","email":"lavraham@redhat.com","username":"lavraham"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-04-18 12:36:42.000000000","updated_by":{"_account_id":28609,"name":"Arkady Shtempler","email":"ashtempl@redhat.com","username":"ashtempl"},"reviewer":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"state":"REVIEWER"},{"updated":"2022-04-18 12:36:42.000000000","updated_by":{"_account_id":28609,"name":"Arkady Shtempler","email":"ashtempl@redhat.com","username":"ashtempl"},"reviewer":{"_account_id":34067,"name":"Lilach Avraham","email":"lavraham@redhat.com","username":"lavraham"},"state":"REVIEWER"},{"updated":"2022-04-18 12:36:42.000000000","updated_by":{"_account_id":28609,"name":"Arkady Shtempler","email":"ashtempl@redhat.com","username":"ashtempl"},"reviewer":{"_account_id":29681,"name":"Bruna Bonguardo","display_name":"Bruna Bonguardo","email":"bbonguar@redhat.com","username":"bonguardo"},"state":"REVIEWER"},{"updated":"2022-04-18 12:36:42.000000000","updated_by":{"_account_id":28609,"name":"Arkady Shtempler","email":"ashtempl@redhat.com","username":"ashtempl"},"reviewer":{"_account_id":18894,"name":"Toni Freger","email":"tfreger@redhat.com"},"state":"REVIEWER"},{"updated":"2022-04-18 12:36:42.000000000","updated_by":{"_account_id":28609,"name":"Arkady Shtempler","email":"ashtempl@redhat.com","username":"ashtempl"},"reviewer":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"state":"REVIEWER"},{"updated":"2022-04-18 12:36:42.000000000","updated_by":{"_account_id":28609,"name":"Arkady Shtempler","email":"ashtempl@redhat.com","username":"ashtempl"},"reviewer":{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},"state":"REVIEWER"},{"updated":"2022-04-18 12:36:42.000000000","updated_by":{"_account_id":28609,"name":"Arkady Shtempler","email":"ashtempl@redhat.com","username":"ashtempl"},"reviewer":{"_account_id":5572,"name":"Don Kehn","display_name":"DEKehn","email":"dekehn@gmail.com","username":"dekehn"},"state":"REVIEWER"},{"updated":"2022-04-18 12:36:42.000000000","updated_by":{"_account_id":28609,"name":"Arkady Shtempler","email":"ashtempl@redhat.com","username":"ashtempl"},"reviewer":{"_account_id":15025,"name":"Michael Johnson","email":"Michael.Johnson@sas.com","username":"mikjo"},"state":"REVIEWER"},{"updated":"2022-04-18 13:38:05.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-04-19 18:39:20.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"}],"messages":[{"id":"629ce8e7ff2f50872c689e312cd2488b93d489ba","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28609,"name":"Arkady Shtempler","email":"ashtempl@redhat.com","username":"ashtempl"},"date":"2022-04-18 12:32:53.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"5b0728a7631d60de603db85a4fe71c1006092d66","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-18 13:38:05.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/978ba4a7696b47d688332c355243580a : FAILURE in 32m 57s\n- designate-bind9-centos8stream-fips https://zuul.opendev.org/t/openstack/build/fc04360d13234bf79bb35824ad04152b : FAILURE in 39m 04s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/9bf1ecca20064959b6c6efdc4c33fbf0 : FAILURE in 38m 02s (non-voting)\n- designate-bind9-scoped-tokens https://zuul.opendev.org/t/openstack/build/98d4482657a0453884b0fc89e45475b9 : FAILURE in 34m 49s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/f7a39eebd46a4cdca246e71024efb442 : FAILURE in 32m 24s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/678c4cf5387c47b1bfda66f9cb33b5c4 : FAILURE in 49m 58s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/fdc7fdbe20714389aa61a4283f9ab467 : FAILURE in 34m 01s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/010639e740074d17a2e2a131a46c75d4 : FAILURE in 33m 30s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/83986318dcb04de285b0319d09e6453e : SUCCESS in 3m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f5dc12e1d0074af6a114a648e9bf49ce : SUCCESS in 4m 53s\n- designate-bind9-stable-xena https://zuul.opendev.org/t/openstack/build/f8b0ab722a0e434dabf9c398bc91d047 : FAILURE in 28m 40s\n- designate-bind9-stable-wallaby https://zuul.opendev.org/t/openstack/build/a3e8f9083b834a48affc658168b2cf4e : FAILURE in 33m 01s\n- designate-bind9-stable-victoria https://zuul.opendev.org/t/openstack/build/15c485054a3c49f5bdb10aa4e1e2b54e : FAILURE in 28m 54s\n- designate-bind9-stable-ussuri https://zuul.opendev.org/t/openstack/build/c688579c91474edea931bbf0928c9182 : FAILURE in 9m 11s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/9e37fe6706854dbaa8ebaef548ca9972 : SUCCESS in 26m 30s","accounts_in_message":[],"_revision_number":1},{"id":"91585aea15d262f00cc6eef823f33a4f54e115e0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28609,"name":"Arkady Shtempler","email":"ashtempl@redhat.com","username":"ashtempl"},"date":"2022-04-19 08:28:54.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"00f859d7f13284d3af2fa79c794a0b13feaf388e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-19 09:15:03.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/052beb2d28ce40a0904152a8027fe8b7 : FAILURE in 28m 27s\n- designate-bind9-centos8stream-fips https://zuul.opendev.org/t/openstack/build/8fb3cf52ac8741378ff873586094a940 : FAILURE in 34m 50s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/0f827aae76c447beb9d15e0ff625464f : FAILURE in 30m 27s (non-voting)\n- designate-bind9-scoped-tokens https://zuul.opendev.org/t/openstack/build/f383b03507f24c0d918e32f1dabce0a1 : FAILURE in 31m 16s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/96a0a6e7353b4b958e1750453415ff86 : FAILURE in 27m 51s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/05232a09684a4dd09ef1d5b8ff2f79ec : FAILURE in 43m 14s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/4dfa55b780d54c41a4a7ed208d41067c : FAILURE in 36m 43s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/32d8a22ea6cd47e2bfbdc0526be11a25 : FAILURE in 42m 25s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3196cd3989dd447a9a2f3fc7e0a9c992 : SUCCESS in 3m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7a3e1017fe1844549de83e5ff1884540 : FAILURE in 3m 25s\n- designate-bind9-stable-xena https://zuul.opendev.org/t/openstack/build/665aa18b95684fd5a2274ca02bc4703d : FAILURE in 25m 58s\n- designate-bind9-stable-wallaby https://zuul.opendev.org/t/openstack/build/af3b29bc295948aeb46e99cd9313b7b2 : FAILURE in 29m 46s\n- designate-bind9-stable-victoria https://zuul.opendev.org/t/openstack/build/ca18737b83de4b529e1748eca0e96829 : FAILURE in 30m 50s\n- designate-bind9-stable-ussuri https://zuul.opendev.org/t/openstack/build/c9cc8d6b38b04556a9a24e640545a4d0 : FAILURE in 12m 10s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/08e22dc376574a42ae10f2f19188c503 : FAILURE in 31m 26s","accounts_in_message":[],"_revision_number":2},{"id":"ecaa9e14e71f6a577695be1a26e0863dca6122df","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28609,"name":"Arkady Shtempler","email":"ashtempl@redhat.com","username":"ashtempl"},"date":"2022-04-19 09:50:51.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"e9213ef5c38385054ed09edecfbd104c1a4ed85d","author":{"_account_id":28609,"name":"Arkady Shtempler","email":"ashtempl@redhat.com","username":"ashtempl"},"date":"2022-04-19 09:51:50.000000000","message":"Patch Set 3: Workflow-1","accounts_in_message":[],"_revision_number":3},{"id":"e7248981d5aef3459138a618789520a72fa05672","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-19 10:35:39.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/95e39ec9d0b34d148cadc9d66a06eac4 : FAILURE in 32m 14s\n- designate-bind9-centos8stream-fips https://zuul.opendev.org/t/openstack/build/27bc1ec585584835903f720b3aa8730d : FAILURE in 40m 12s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/ce20e134bb4d4efeabfca03db7225863 : FAILURE in 36m 49s (non-voting)\n- designate-bind9-scoped-tokens https://zuul.opendev.org/t/openstack/build/945f77594dd443f994a51f3eb00b6413 : FAILURE in 32m 29s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/ab3fd3ebae8b4961ad76d609e2423b29 : FAILURE in 31m 49s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/092c8ed6bae447369106fbc0d1e190e7 : FAILURE in 24m 21s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/4b94cc08eeab42cd9532a1a5290341ff : FAILURE in 25m 25s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/8c732db2958d4967aab3cb43a2e718ef : FAILURE in 31m 55s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c5911329358f4ce5bdd2256f4bbd960c : SUCCESS in 3m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/de69eb2c94534ca2b201a3dd3185ce7f : FAILURE in 3m 12s\n- designate-bind9-stable-xena https://zuul.opendev.org/t/openstack/build/ca09682cfe7644048546e884db815e72 : FAILURE in 28m 29s\n- designate-bind9-stable-wallaby https://zuul.opendev.org/t/openstack/build/62793c3e19b94c31980bbe5b45c297e9 : FAILURE in 24m 57s\n- designate-bind9-stable-victoria https://zuul.opendev.org/t/openstack/build/88565bf862384ed5b6413dbfa3ef4573 : FAILURE in 32m 20s\n- designate-bind9-stable-ussuri https://zuul.opendev.org/t/openstack/build/5b2e619aa0264b549c280363ee20c2b9 : FAILURE in 15m 32s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/3c15afbf6e814b9f84998305c1434081 : FAILURE in 41m 47s","accounts_in_message":[],"_revision_number":3},{"id":"7597e05806a0a6b6e02dc3794b8d720d4ce2d4eb","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28609,"name":"Arkady Shtempler","email":"ashtempl@redhat.com","username":"ashtempl"},"date":"2022-04-19 10:42:45.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"b9514465e4f42d28f6caefcb719a15f1e3b5dc78","author":{"_account_id":28609,"name":"Arkady Shtempler","email":"ashtempl@redhat.com","username":"ashtempl"},"date":"2022-04-19 10:43:03.000000000","message":"Patch Set 4: Workflow-1","accounts_in_message":[],"_revision_number":4},{"id":"5d22f80040c5bb14c7267642aff59f42f8d7cf58","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-19 11:26:48.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/1e0ce8f841e64fd9a7a20e8f9a5a1b6e : FAILURE in 36m 18s\n- designate-bind9-centos8stream-fips https://zuul.opendev.org/t/openstack/build/5ec0d7c61fd94210a83436eb6e8bbaf6 : FAILURE in 42m 20s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/8a8a78be1bce44c9a54fd706deaefe34 : FAILURE in 27m 38s (non-voting)\n- designate-bind9-scoped-tokens https://zuul.opendev.org/t/openstack/build/0039dd27d3cd4fac9945969982e08514 : FAILURE in 34m 51s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/602e73eadf8848c2aa03349e4aaec5d4 : FAILURE in 36m 42s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/6473670f6f1d473db983b04dee44c759 : FAILURE in 34m 58s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/2d1a450528614905adaee5eb3ca12099 : FAILURE in 33m 42s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/760548f297a242048ddd322071a00437 : FAILURE in 35m 15s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ee7229440d294697996cf149b6a3409b : SUCCESS in 3m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/be470df32ac0425ba02bfebf13c8c3d5 : FAILURE in 3m 24s\n- designate-bind9-stable-xena https://zuul.opendev.org/t/openstack/build/4d676757de034158a39ce70557839ec0 : FAILURE in 32m 21s\n- designate-bind9-stable-wallaby https://zuul.opendev.org/t/openstack/build/ef2944fe67054324b0879c2739e0574b : FAILURE in 39m 04s\n- designate-bind9-stable-victoria https://zuul.opendev.org/t/openstack/build/52e64d95de174a8d8b3543dd37f63188 : FAILURE in 38m 55s\n- designate-bind9-stable-ussuri https://zuul.opendev.org/t/openstack/build/1a42b16791f84ad4b4e1a70dce603a95 : FAILURE in 12m 52s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/e693319d8cbe48ee90ce9241dd5feb81 : FAILURE in 31m 57s","accounts_in_message":[],"_revision_number":4},{"id":"7b678e0ea76e1a3d7fb010d477c8c0f7b57c5e17","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28609,"name":"Arkady Shtempler","email":"ashtempl@redhat.com","username":"ashtempl"},"date":"2022-04-19 13:09:21.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"2631b1a6e3e8ba375685f4f701320970abb82163","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-19 13:52:51.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/1ecad6ac56974607a9e9c3663e1aecc1 : FAILURE in 31m 12s\n- designate-bind9-centos8stream-fips https://zuul.opendev.org/t/openstack/build/54b0fc3a55dd4c97b77dc42ebea2dc40 : FAILURE in 39m 00s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/a9a9b91809154a89aa79f38815b6caf4 : FAILURE in 29m 35s (non-voting)\n- designate-bind9-scoped-tokens https://zuul.opendev.org/t/openstack/build/40906095193f47118edb44c0185a2ca0 : FAILURE in 32m 47s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/7fc64f183796423caaa4d09b61dda6bb : FAILURE in 31m 28s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/d3340aad80174ec3a9127a52c03b969f : FAILURE in 34m 56s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/db4aed4f1d3940408714064fd87d15ac : FAILURE in 38m 02s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/2eaf968eedf841ebb517a50f798b08f1 : FAILURE in 36m 39s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5f3e78dce9b541278cd7441c5af2e396 : SUCCESS in 3m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8861a10ec719483fb9cdb28981a7fa31 : SUCCESS in 3m 24s\n- designate-bind9-stable-xena https://zuul.opendev.org/t/openstack/build/3dedfc5dfb1b4221ab7608c13b0534da : FAILURE in 30m 23s\n- designate-bind9-stable-wallaby https://zuul.opendev.org/t/openstack/build/0adbbdd3a91846a5ba3a49a340ccd9e2 : FAILURE in 34m 12s\n- designate-bind9-stable-victoria https://zuul.opendev.org/t/openstack/build/f9a1228229a7400cb2401f38ec8592ed : FAILURE in 33m 10s\n- designate-bind9-stable-ussuri https://zuul.opendev.org/t/openstack/build/7ae2ca66c33d4a6781e10f2d9770af14 : FAILURE in 18m 24s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/0d51a26eaa8e4e0aa8e0ee9f6bccd858 : FAILURE in 31m 14s","accounts_in_message":[],"_revision_number":5},{"id":"6a61fbe128b71abcedcefc9de605f8a5b82fcd25","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2022-04-19 18:39:20.000000000","message":"Patch Set 5: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"a401fd87179901fbb13a01db48ef2000fea69bb0","author":{"_account_id":5572,"name":"Don Kehn","display_name":"DEKehn","email":"dekehn@gmail.com","username":"dekehn"},"date":"2022-04-19 18:42:56.000000000","message":"Patch Set 5: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"75046e3baccfa4277fd1878967ac988c0fd5ec8b","author":{"_account_id":28609,"name":"Arkady Shtempler","email":"ashtempl@redhat.com","username":"ashtempl"},"date":"2022-04-20 08:13:09.000000000","message":"Patch Set 5:\n\n(2 comments)","accounts_in_message":[],"_revision_number":5},{"id":"29431dddf7c9ca28f5720a9f0f462350d5d9caee","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28609,"name":"Arkady Shtempler","email":"ashtempl@redhat.com","username":"ashtempl"},"date":"2022-04-20 08:20:14.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"6ff12a0e960854cb6b7606920a091e2f89f5aba2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-20 09:20:14.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/d12e8cad484742768f17d1ba3704de91 : SUCCESS in 41m 22s\n- designate-bind9-centos8stream-fips https://zuul.opendev.org/t/openstack/build/bf3e1f5cca0e43b98f7224b1f81d3b0e : SUCCESS in 43m 46s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/4ca1568d4e6c4c0c847be98197d1fe47 : FAILURE in 8m 51s (non-voting)\n- designate-bind9-scoped-tokens https://zuul.opendev.org/t/openstack/build/f1a9db7d2b4348d4ab1e940225f627b5 : SUCCESS in 49m 46s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/c782cf4679cc46c993f7602aee03f4d2 : SUCCESS in 41m 03s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/343e3b8104584684a1c4f712afa9d9c4 : SUCCESS in 57m 56s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/80019603e56d47169c5b9065189cf452 : SUCCESS in 39m 42s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/df21ea1e479347cd81cb69d0e0aafeae : SUCCESS in 41m 55s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/63418f02d56e443983d135ed50f532b8 : SUCCESS in 3m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/65eb2f026a8b49baa83226d24023f004 : SUCCESS in 3m 11s\n- designate-bind9-stable-xena https://zuul.opendev.org/t/openstack/build/a7e6539c379c4371931e33c628177383 : SUCCESS in 42m 03s\n- designate-bind9-stable-wallaby https://zuul.opendev.org/t/openstack/build/1a05ea3eb62b4ffc89ef5e86d69441cc : SUCCESS in 46m 16s\n- designate-bind9-stable-victoria https://zuul.opendev.org/t/openstack/build/32d007c1d8594d96b88d9a56c1e3b88b : SUCCESS in 50m 42s\n- designate-bind9-stable-ussuri https://zuul.opendev.org/t/openstack/build/3251c932a36f467eb8d757667729924a : SUCCESS in 45m 13s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/4486a3e108fe47f4aaeceb4c9af42ff8 : SUCCESS in 31m 03s","accounts_in_message":[],"_revision_number":6},{"id":"9e061fda8352544af2c77e2976c3072c7ddd1795","author":{"_account_id":5572,"name":"Don Kehn","display_name":"DEKehn","email":"dekehn@gmail.com","username":"dekehn"},"date":"2022-04-25 21:19:38.000000000","message":"Patch Set 6: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"67fac273069972d4b5a2d45c9be76a10a7a5fb1a","author":{"_account_id":28609,"name":"Arkady Shtempler","email":"ashtempl@redhat.com","username":"ashtempl"},"date":"2022-04-26 08:27:35.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"1c1719d5308b2145e6f323d5013d18696a03b66a","author":{"_account_id":5572,"name":"Don Kehn","display_name":"DEKehn","email":"dekehn@gmail.com","username":"dekehn"},"date":"2022-04-27 21:28:44.000000000","message":"Patch Set 6: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"05be2c995beab15a8e161baee4924fa869369bea","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2022-05-10 22:04:57.000000000","message":"Patch Set 6: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"d944f588e9ea76572174691fa4480754a655b2d3","author":{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},"date":"2022-05-11 14:27:15.000000000","message":"Patch Set 6: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"01f02aeb335d571c1435f32f157cd212dadf91e4","author":{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},"date":"2022-05-11 14:31:29.000000000","message":"Patch Set 6: Workflow+1","accounts_in_message":[],"_revision_number":6},{"id":"a9bfc781aee8f90c2562344238e003246308fb4b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-11 14:32:11.000000000","message":"Patch Set 6: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":6},{"id":"db9a8712627ed9e23d5e3a2c337b38c6ee433472","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-11 15:36:24.000000000","message":"Patch Set 6: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/6ee8a28220764c6fb1e8df71a43df797 : SUCCESS in 59m 14s\n- designate-bind9-scoped-tokens https://zuul.opendev.org/t/openstack/build/8279006698eb4ead9230f6d13984e829 : SUCCESS in 37m 17s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/93208d41d6ec4d72a65ce8e3f7c99543 : SUCCESS in 39m 43s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/950f0d3656a348919191010558be0872 : SUCCESS in 1h 01m 26s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/9582e32703e84cddb859d4f1896dbf7c : SUCCESS in 39m 03s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/8e578e9bbc054761b6da99071063108d : SUCCESS in 48m 49s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0ac8a7a218774c5d91ab841b884447ec : SUCCESS in 5m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4f8fc0f656214a1094cffc0b5c04dd60 : SUCCESS in 2m 51s","accounts_in_message":[],"_revision_number":6},{"id":"4aa00603822194d8bdd2fb3eabd33fb58a3aa5f5","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-11 15:36:24.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":6},{"id":"e7d88bc737d17923366322e224587423510677a7","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-11 15:37:46.000000000","message":"Patch Set 6:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/54ed6e950aa24bd2b69b9284e3af2ea7 : SUCCESS in 52s","accounts_in_message":[],"_revision_number":6}],"current_revision_number":6,"current_revision":"6d6b27d53764e885dd030dc5b6296043808508e6","revisions":{"f6566262c8e0d15bce56b598070fe499ae523e49":{"kind":"REWORK","_number":1,"created":"2022-04-18 12:32:53.000000000","uploader":{"_account_id":28609,"name":"Arkady Shtempler","email":"ashtempl@redhat.com","username":"ashtempl"},"ref":"refs/changes/98/838298/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate-tempest-plugin","ref":"refs/changes/98/838298/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate-tempest-plugin refs/changes/98/838298/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate-tempest-plugin refs/changes/98/838298/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate-tempest-plugin refs/changes/98/838298/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate-tempest-plugin refs/changes/98/838298/1"}}},"commit":{"parents":[{"commit":"1cbfc88d6742aac850a108b9d37223dde8123a43","subject":"Merge \"Fix waiters to raise on resources going to ERROR\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate-tempest-plugin/commit/1cbfc88d6742aac850a108b9d37223dde8123a43"}]}],"author":{"name":"Arkady Shtempler","email":"arkadysh@gmail.com","date":"2022-04-18 12:25:00.000000000","tz":180},"committer":{"name":"Arkady Shtempler","email":"arkadysh@gmail.com","date":"2022-04-18 12:32:01.000000000","tz":180},"subject":"New scenario test case for TLD + re-factoring","message":"New scenario test case for TLD + re-factoring\n\n1) Moving \"test_create_zone_for_not_existing_tld\" from API\n   to scenario: \"test_create_zone_using_not_existing_tld\"\n   Create a Zone using not existing TLD is supposed to FAIL\n2) New test case: \"test_create_zone_using_existing_tld\"\n   Create a Zone using existing TLD is supposed to PASS\n3) Using indexes instead of underscores variables\n\nChange-Id: Ie9fca2598a3eab66f9cbd718e52e575744def160\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate-tempest-plugin/commit/f6566262c8e0d15bce56b598070fe499ae523e49"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate-tempest-plugin/commit/f6566262c8e0d15bce56b598070fe499ae523e49"}]},"branch":"refs/heads/master"},"082f5967b0715efc3809380e15c08c78c4199aa6":{"kind":"REWORK","_number":2,"created":"2022-04-19 08:28:54.000000000","uploader":{"_account_id":28609,"name":"Arkady Shtempler","email":"ashtempl@redhat.com","username":"ashtempl"},"ref":"refs/changes/98/838298/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate-tempest-plugin","ref":"refs/changes/98/838298/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate-tempest-plugin refs/changes/98/838298/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate-tempest-plugin refs/changes/98/838298/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate-tempest-plugin refs/changes/98/838298/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate-tempest-plugin refs/changes/98/838298/2"}}},"commit":{"parents":[{"commit":"1cbfc88d6742aac850a108b9d37223dde8123a43","subject":"Merge \"Fix waiters to raise on resources going to ERROR\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate-tempest-plugin/commit/1cbfc88d6742aac850a108b9d37223dde8123a43"}]}],"author":{"name":"Arkady Shtempler","email":"arkadysh@gmail.com","date":"2022-04-18 12:25:00.000000000","tz":180},"committer":{"name":"Arkady Shtempler","email":"arkadysh@gmail.com","date":"2022-04-19 08:24:31.000000000","tz":180},"subject":"New scenario test case for TLD + re-factoring","message":"New scenario test case for TLD + re-factoring\n\n1) Moving \"test_create_zone_for_not_existing_tld\" from API\n   to scenario: \"test_create_zone_using_not_existing_tld\"\n   Create a Zone using not existing TLD is supposed to FAIL\n2) New test case: \"test_create_zone_using_existing_tld\"\n   Create a Zone using existing TLD is supposed to PASS\n3) Using indexes instead of underscores variables\n4) Adding \"tld_suffix\" parameter into configuration (default is: \u0027com\u0027)\n   to avoid (Non existing TLD) issues. All tests will be using\n   the same(common) TLD suffix.\n\nChange-Id: Ie9fca2598a3eab66f9cbd718e52e575744def160\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate-tempest-plugin/commit/082f5967b0715efc3809380e15c08c78c4199aa6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate-tempest-plugin/commit/082f5967b0715efc3809380e15c08c78c4199aa6"}]},"branch":"refs/heads/master"},"1962e1307bce151a259c6a4d960b4f7e7243f54c":{"kind":"REWORK","_number":3,"created":"2022-04-19 09:50:51.000000000","uploader":{"_account_id":28609,"name":"Arkady Shtempler","email":"ashtempl@redhat.com","username":"ashtempl"},"ref":"refs/changes/98/838298/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate-tempest-plugin","ref":"refs/changes/98/838298/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate-tempest-plugin refs/changes/98/838298/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate-tempest-plugin refs/changes/98/838298/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate-tempest-plugin refs/changes/98/838298/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate-tempest-plugin refs/changes/98/838298/3"}}},"commit":{"parents":[{"commit":"1cbfc88d6742aac850a108b9d37223dde8123a43","subject":"Merge \"Fix waiters to raise on resources going to ERROR\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate-tempest-plugin/commit/1cbfc88d6742aac850a108b9d37223dde8123a43"}]}],"author":{"name":"Arkady Shtempler","email":"arkadysh@gmail.com","date":"2022-04-18 12:25:00.000000000","tz":180},"committer":{"name":"Arkady Shtempler","email":"arkadysh@gmail.com","date":"2022-04-19 09:50:32.000000000","tz":180},"subject":"New scenario test case for TLD + re-factoring","message":"New scenario test case for TLD + re-factoring\n\n1) Moving \"test_create_zone_for_not_existing_tld\" from API\n   to scenario: \"test_create_zone_using_not_existing_tld\"\n   Create a Zone using not existing TLD is supposed to FAIL\n2) New test case: \"test_create_zone_using_existing_tld\"\n   Create a Zone using existing TLD is supposed to PASS\n3) Using indexes instead of underscores variables\n4) Adding \"tld_suffix\" parameter into configuration (default is: \u0027com\u0027)\n   to avoid (Non existing TLD) issues. All tests will be using\n   the same(common) TLD suffix.\n\nChange-Id: Ie9fca2598a3eab66f9cbd718e52e575744def160\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate-tempest-plugin/commit/1962e1307bce151a259c6a4d960b4f7e7243f54c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate-tempest-plugin/commit/1962e1307bce151a259c6a4d960b4f7e7243f54c"}]},"branch":"refs/heads/master"},"770c7a87ad2aa4a072b1d9ee3d7a4ce51ae1b3c2":{"kind":"REWORK","_number":4,"created":"2022-04-19 10:42:45.000000000","uploader":{"_account_id":28609,"name":"Arkady Shtempler","email":"ashtempl@redhat.com","username":"ashtempl"},"ref":"refs/changes/98/838298/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate-tempest-plugin","ref":"refs/changes/98/838298/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate-tempest-plugin refs/changes/98/838298/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate-tempest-plugin refs/changes/98/838298/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate-tempest-plugin refs/changes/98/838298/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate-tempest-plugin refs/changes/98/838298/4"}}},"commit":{"parents":[{"commit":"1cbfc88d6742aac850a108b9d37223dde8123a43","subject":"Merge \"Fix waiters to raise on resources going to ERROR\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate-tempest-plugin/commit/1cbfc88d6742aac850a108b9d37223dde8123a43"}]}],"author":{"name":"Arkady Shtempler","email":"arkadysh@gmail.com","date":"2022-04-18 12:25:00.000000000","tz":180},"committer":{"name":"Arkady Shtempler","email":"arkadysh@gmail.com","date":"2022-04-19 10:42:30.000000000","tz":180},"subject":"New scenario test case for TLD + re-factoring","message":"New scenario test case for TLD + re-factoring\n\n1) Moving \"test_create_zone_for_not_existing_tld\" from API\n   to scenario: \"test_create_zone_using_not_existing_tld\"\n   Create a Zone using not existing TLD is supposed to FAIL\n2) New test case: \"test_create_zone_using_existing_tld\"\n   Create a Zone using existing TLD is supposed to PASS\n3) Using indexes instead of underscores variables\n4) Adding \"tld_suffix\" parameter into configuration (default is: \u0027com\u0027)\n   to avoid (Non existing TLD) issues. All tests will be using\n   the same(common) TLD suffix.\n\nChange-Id: Ie9fca2598a3eab66f9cbd718e52e575744def160\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate-tempest-plugin/commit/770c7a87ad2aa4a072b1d9ee3d7a4ce51ae1b3c2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate-tempest-plugin/commit/770c7a87ad2aa4a072b1d9ee3d7a4ce51ae1b3c2"}]},"branch":"refs/heads/master"},"8b3e7efdeb401f81c6a95412dee6d8e100eff7d3":{"kind":"REWORK","_number":5,"created":"2022-04-19 13:09:21.000000000","uploader":{"_account_id":28609,"name":"Arkady Shtempler","email":"ashtempl@redhat.com","username":"ashtempl"},"ref":"refs/changes/98/838298/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate-tempest-plugin","ref":"refs/changes/98/838298/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate-tempest-plugin refs/changes/98/838298/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate-tempest-plugin refs/changes/98/838298/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate-tempest-plugin refs/changes/98/838298/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate-tempest-plugin refs/changes/98/838298/5"}}},"commit":{"parents":[{"commit":"1cbfc88d6742aac850a108b9d37223dde8123a43","subject":"Merge \"Fix waiters to raise on resources going to ERROR\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate-tempest-plugin/commit/1cbfc88d6742aac850a108b9d37223dde8123a43"}]}],"author":{"name":"Arkady Shtempler","email":"arkadysh@gmail.com","date":"2022-04-18 12:25:00.000000000","tz":180},"committer":{"name":"Arkady Shtempler","email":"arkadysh@gmail.com","date":"2022-04-19 13:09:11.000000000","tz":180},"subject":"New scenario test case for TLD + re-factoring","message":"New scenario test case for TLD + re-factoring\n\n1) Moving \"test_create_zone_for_not_existing_tld\" from API\n   to scenario: \"test_create_zone_using_not_existing_tld\"\n   Create a Zone using not existing TLD is supposed to FAIL\n2) New test case: \"test_create_zone_using_existing_tld\"\n   Create a Zone using existing TLD is supposed to PASS\n3) Using indexes instead of underscores variables\n4) Adding \"tld_suffix\" parameter into configuration (default is: \u0027com\u0027)\n   to avoid (Non existing TLD) issues. All tests will be using\n   the same(common) TLD suffix.\n\nChange-Id: Ie9fca2598a3eab66f9cbd718e52e575744def160\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate-tempest-plugin/commit/8b3e7efdeb401f81c6a95412dee6d8e100eff7d3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate-tempest-plugin/commit/8b3e7efdeb401f81c6a95412dee6d8e100eff7d3"}]},"branch":"refs/heads/master"},"6d6b27d53764e885dd030dc5b6296043808508e6":{"kind":"REWORK","_number":6,"created":"2022-04-20 08:20:14.000000000","uploader":{"_account_id":28609,"name":"Arkady Shtempler","email":"ashtempl@redhat.com","username":"ashtempl"},"ref":"refs/changes/98/838298/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate-tempest-plugin","ref":"refs/changes/98/838298/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate-tempest-plugin refs/changes/98/838298/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate-tempest-plugin refs/changes/98/838298/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate-tempest-plugin refs/changes/98/838298/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate-tempest-plugin refs/changes/98/838298/6"}}},"commit":{"parents":[{"commit":"1cbfc88d6742aac850a108b9d37223dde8123a43","subject":"Merge \"Fix waiters to raise on resources going to ERROR\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate-tempest-plugin/commit/1cbfc88d6742aac850a108b9d37223dde8123a43"}]}],"author":{"name":"Arkady Shtempler","email":"arkadysh@gmail.com","date":"2022-04-18 12:25:00.000000000","tz":180},"committer":{"name":"Arkady Shtempler","email":"arkadysh@gmail.com","date":"2022-04-20 08:18:47.000000000","tz":180},"subject":"New scenario test case for TLD + re-factoring","message":"New scenario test case for TLD + re-factoring\n\n1) Moving \"test_create_zone_for_not_existing_tld\" from API\n   to scenario: \"test_create_zone_using_not_existing_tld\"\n   Create a Zone using not existing TLD is supposed to FAIL\n2) New test case: \"test_create_zone_using_existing_tld\"\n   Create a Zone using existing TLD is supposed to PASS\n3) Using indexes instead of underscores variables\n4) Adding \"tld_suffix\" parameter into configuration (default is: \u0027com\u0027)\n   to avoid (Non existing TLD) issues. All tests will be using\n   the same(common) TLD suffix.\n\nChange-Id: Ie9fca2598a3eab66f9cbd718e52e575744def160\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate-tempest-plugin/commit/6d6b27d53764e885dd030dc5b6296043808508e6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate-tempest-plugin/commit/6d6b27d53764e885dd030dc5b6296043808508e6"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
