)]}'
{"id":"openstack%2Fdesignate-tempest-plugin~880681","triplet_id":"openstack%2Fdesignate-tempest-plugin~master~I85d871350fcacfb6e293092ff6c16a6dba7f952e","project":"openstack/designate-tempest-plugin","branch":"master","attention_set":{},"removed_from_attention_set":{"11628":{"account":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"last_update":"2023-05-24 21:02:49.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I85d871350fcacfb6e293092ff6c16a6dba7f952e","subject":"Unroll test_update_records_propagated_to_backends","status":"MERGED","created":"2023-04-17 19:23:31.000000000","updated":"2023-05-24 21:03:48.000000000","submitted":"2023-05-24 21:02:49.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"880681","meta_rev_id":"7f54e31f453d366a98c39dd2f70d12eeeb6d2d15","_number":880681,"virtual_id_number":880681,"owner":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28609,"name":"Arkady Shtempler","email":"ashtempl@redhat.com","username":"ashtempl"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-05-24 21:02:49.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"recommended":{"_account_id":28609,"name":"Arkady Shtempler","email":"ashtempl@redhat.com","username":"ashtempl"},"all":[{"value":1,"date":"2023-04-18 14:56:56.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28609,"name":"Arkady Shtempler","email":"ashtempl@redhat.com","username":"ashtempl"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2023-05-24 19:37:58.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"all":[{"value":0,"_account_id":28609,"name":"Arkady Shtempler","email":"ashtempl@redhat.com","username":"ashtempl"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2023-05-24 19:49:48.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":28609,"name":"Arkady Shtempler","email":"ashtempl@redhat.com","username":"ashtempl"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},{"_account_id":28609,"name":"Arkady Shtempler","email":"ashtempl@redhat.com","username":"ashtempl"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-04-17 21:08:26.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-04-18 10:43:01.000000000","updated_by":{"_account_id":28609,"name":"Arkady Shtempler","email":"ashtempl@redhat.com","username":"ashtempl"},"reviewer":{"_account_id":28609,"name":"Arkady Shtempler","email":"ashtempl@redhat.com","username":"ashtempl"},"state":"REVIEWER"},{"updated":"2023-05-24 19:37:58.000000000","updated_by":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"reviewer":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"state":"REVIEWER"}],"messages":[{"id":"c117518d6dc4f7f1beec8d4b569f7e4bf515003a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2023-04-17 19:23:31.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"5e805d3402785073a64fecc146193d054aaae248","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2023-04-17 20:07:17.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"b19898df9c9d9e9308e5835d44c77240e88cd0c5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-17 21:08:26.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8d5ae9525791495eaeb5ce4920c0c075\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/a949a90ce28c460dbdf6ca4a1d63fc33 : SUCCESS in 42m 20s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/4481730bce9544a4ab36cac0e5db9667 : SUCCESS in 40m 09s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/ca9ea4438a3f4b9a82b16f14a9f4028f : SUCCESS in 38m 54s (non-voting)\n- designate-bind9-keystone-default-roles https://zuul.opendev.org/t/openstack/build/a0d544b07e6848058f5706d2309746b9 : SUCCESS in 39m 51s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/1c74732f1eb3468884d2b9a92f206f66 : SUCCESS in 42m 13s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/437943d8392c4ad99ae03154d5301db3 : SUCCESS in 58m 23s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/45035a40dcc54c6a88361dfe1c007cf3 : SUCCESS in 58m 45s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/2359a7e2787b479d8434780425346cd0 : SUCCESS in 37m 23s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/4ac52b24e6574ff0955a772d8a66973b : SUCCESS in 41m 03s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e2d15effac2247db9d61bf74f2f54dbf : SUCCESS in 3m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a577ff89f30646059f108bd08d0cf4e1 : SUCCESS in 3m 10s\n- designate-bind9-stable-antelope https://zuul.opendev.org/t/openstack/build/40f91101a2624c41a1071e77a44b5b29 : SUCCESS in 36m 46s\n- designate-bind9-stable-zed https://zuul.opendev.org/t/openstack/build/c1cc3ea64d6d49aaae7b09873b61a85d : SUCCESS in 36m 08s\n- designate-bind9-stable-yoga https://zuul.opendev.org/t/openstack/build/2ff2b6f703194448a48bba1b5b85bd12 : SUCCESS in 42m 25s\n- designate-bind9-stable-xena https://zuul.opendev.org/t/openstack/build/ae62cccf2b414367b53108c7b85fd1c8 : SUCCESS in 44m 14s\n- designate-bind9-stable-wallaby https://zuul.opendev.org/t/openstack/build/064072f1e2534b4488c4e214374b93a7 : SUCCESS in 39m 25s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/95e82e2385534fc8b380f86b9d080879 : SUCCESS in 39m 37s","accounts_in_message":[],"_revision_number":2},{"id":"b4a41884cffdc493d567c053d19fb102ee537d71","author":{"_account_id":28609,"name":"Arkady Shtempler","email":"ashtempl@redhat.com","username":"ashtempl"},"date":"2023-04-18 10:43:01.000000000","message":"Patch Set 2: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"0c1d060b49ed460ab4b8cb1f2cf18b8482b73034","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2023-04-18 14:34:40.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"64fe7f7238373499bbd4803401dc0c666e986061","author":{"_account_id":28609,"name":"Arkady Shtempler","email":"ashtempl@redhat.com","username":"ashtempl"},"date":"2023-04-18 14:56:56.000000000","message":"Patch Set 2: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"57aac41789721ffe2cd34ee2d4efb7e754b32ce6","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2023-05-24 19:37:58.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"4c34731db0db7180366055033518bcb84a24c6de","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2023-05-24 19:49:48.000000000","message":"Patch Set 2: Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"6cbb6e080bab65461fc31f4ff6c6547559b9d283","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-24 19:50:24.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"74a2eb0684253999d815c9e2f7653b99b7f9bc1c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-24 21:02:49.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/dd97f38ef765459baace1718112d2e2c\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/6972cc46d05d4d00a10f3e37facc5c6f : SUCCESS in 44m 33s\n- designate-bind9-keystone-default-roles https://zuul.opendev.org/t/openstack/build/136ec7c4c0024c279d8808f26b8452d5 : SUCCESS in 40m 22s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/c4387d1dda8e491b816cb3086bb995dc : SUCCESS in 42m 35s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/a45937f9c802497a8d036be998a3e138 : SUCCESS in 1h 05m 45s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/2485ae3660874b7f91f37244013e0068 : SUCCESS in 44m 31s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/2665f08a8ec24f5ebd7041ca704764fd : SUCCESS in 48m 25s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/93ce2e8fd09b48d7b792daa3dfe7e297 : SUCCESS in 3m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ae811b71d44143369dba5f932ed45fc1 : SUCCESS in 2m 30s","accounts_in_message":[],"_revision_number":2},{"id":"18a9a7ea20ac2f77388d66476c59b147897bacc1","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-24 21:02:49.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"7f54e31f453d366a98c39dd2f70d12eeeb6d2d15","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-24 21:03:48.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ae5df6cc3a9c4c9491d54dd1ab59650d\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/442f4a61962c4170977b2889ce37e7c9 : SUCCESS in 39s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"a5ba980ef4840a13032098bc583d556bbdef2b94","revisions":{"c060ef0c9a3308eb3efebe49e58b67417904c7fe":{"kind":"REWORK","_number":1,"created":"2023-04-17 19:23:31.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/81/880681/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate-tempest-plugin","ref":"refs/changes/81/880681/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate-tempest-plugin refs/changes/81/880681/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate-tempest-plugin refs/changes/81/880681/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate-tempest-plugin refs/changes/81/880681/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate-tempest-plugin refs/changes/81/880681/1"}}},"commit":{"parents":[{"commit":"3b98486527503c96caca56268353bd09d83d32b5","subject":"Merge \"Extending \"Shared Zones\" test coverage\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate-tempest-plugin/commit/3b98486527503c96caca56268353bd09d83d32b5"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2023-04-17 19:20:01.000000000","tz":0},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2023-04-17 19:23:29.000000000","tz":0},"subject":"Unroll test_update_records_propagated_to_backends","message":"Unroll test_update_records_propagated_to_backends\n\nThis patch unrolls the test_update_records_propagated_to_backends tests from\nDDT to ensure each test gets a proper idempotent ID. Previously fourteen tests\nhad the same ID.\nThis patch also fixes a scenario where if no nameservers are provided in the\ntempest.conf, the test would pass instead of being skipped.\n\nChange-Id: I85d871350fcacfb6e293092ff6c16a6dba7f952e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate-tempest-plugin/commit/c060ef0c9a3308eb3efebe49e58b67417904c7fe"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate-tempest-plugin/commit/c060ef0c9a3308eb3efebe49e58b67417904c7fe"}]},"branch":"refs/heads/master"},"a5ba980ef4840a13032098bc583d556bbdef2b94":{"kind":"REWORK","_number":2,"created":"2023-04-17 20:07:17.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/81/880681/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate-tempest-plugin","ref":"refs/changes/81/880681/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate-tempest-plugin refs/changes/81/880681/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate-tempest-plugin refs/changes/81/880681/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate-tempest-plugin refs/changes/81/880681/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate-tempest-plugin refs/changes/81/880681/2"}}},"commit":{"parents":[{"commit":"3b98486527503c96caca56268353bd09d83d32b5","subject":"Merge \"Extending \"Shared Zones\" test coverage\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate-tempest-plugin/commit/3b98486527503c96caca56268353bd09d83d32b5"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2023-04-17 19:20:01.000000000","tz":0},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2023-04-17 20:07:11.000000000","tz":0},"subject":"Unroll test_update_records_propagated_to_backends","message":"Unroll test_update_records_propagated_to_backends\n\nThis patch unrolls the test_update_records_propagated_to_backends tests from\nDDT to ensure each test gets a proper idempotent ID. Previously fourteen tests\nhad the same ID.\nThis patch also fixes a scenario where if no nameservers are provided in the\ntempest.conf, the test would pass instead of being skipped.\n\nChange-Id: I85d871350fcacfb6e293092ff6c16a6dba7f952e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate-tempest-plugin/commit/a5ba980ef4840a13032098bc583d556bbdef2b94"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate-tempest-plugin/commit/a5ba980ef4840a13032098bc583d556bbdef2b94"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":28609,"name":"Arkady Shtempler","email":"ashtempl@redhat.com","username":"ashtempl"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
