)]}'
{"id":"openstack%2Fdesignate-tempest-plugin~900870","triplet_id":"openstack%2Fdesignate-tempest-plugin~master~I67e709c9d61af3c5b75b79f27c0c2e9dc8fa08b9","project":"openstack/designate-tempest-plugin","branch":"master","topic":"adapt-to-catalog-zones","attention_set":{},"removed_from_attention_set":{"33634":{"account":{"_account_id":33634,"name":"Jan Hartkopf","email":"j@hartkopf.io","username":"jhartkopf"},"last_update":"2024-01-16 15:56:33.000000000","reason":"Change was submitted"},"11628":{"account":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"last_update":"2024-01-16 15:56:33.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I67e709c9d61af3c5b75b79f27c0c2e9dc8fa08b9","subject":"Fix test_show_zone_impersonate_another_project","status":"MERGED","created":"2023-11-14 10:34:08.000000000","updated":"2024-01-16 15:57:29.000000000","submitted":"2024-01-16 15:56:33.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":4,"unresolved_comment_count":2,"has_review_started":true,"submission_id":"900870-adapt-to-catalog-zones","meta_rev_id":"822114335a5050df876d5be9cec758e2268e8d01","_number":900870,"virtual_id_number":900870,"owner":{"_account_id":33634,"name":"Jan Hartkopf","email":"j@hartkopf.io","username":"jhartkopf"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28609,"name":"Arkady Shtempler","email":"ashtempl@redhat.com","username":"ashtempl"},{"value":0,"_account_id":33634,"name":"Jan Hartkopf","email":"j@hartkopf.io","username":"jhartkopf"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-01-16 15:56:33.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"recommended":{"_account_id":33634,"name":"Jan Hartkopf","email":"j@hartkopf.io","username":"jhartkopf"},"all":[{"value":1,"date":"2023-12-04 08:38:21.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28609,"name":"Arkady Shtempler","email":"ashtempl@redhat.com","username":"ashtempl"},{"value":1,"date":"2023-11-15 09:26:40.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":33634,"name":"Jan Hartkopf","email":"j@hartkopf.io","username":"jhartkopf"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2023-11-14 23:37:48.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":2,"date":"2024-01-16 14:35:34.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"all":[{"value":0,"_account_id":28609,"name":"Arkady Shtempler","email":"ashtempl@redhat.com","username":"ashtempl"},{"value":0,"_account_id":33634,"name":"Jan Hartkopf","email":"j@hartkopf.io","username":"jhartkopf"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":1,"date":"2024-01-16 14:35:34.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":28609,"name":"Arkady Shtempler","email":"ashtempl@redhat.com","username":"ashtempl"},{"value":0,"_account_id":33634,"name":"Jan Hartkopf","email":"j@hartkopf.io","username":"jhartkopf"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},{"_account_id":28609,"name":"Arkady Shtempler","email":"ashtempl@redhat.com","username":"ashtempl"},{"_account_id":33634,"name":"Jan Hartkopf","email":"j@hartkopf.io","username":"jhartkopf"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-11-14 12:02:10.000000000","updated_by":{"_account_id":33634,"name":"Jan Hartkopf","email":"j@hartkopf.io","username":"jhartkopf"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"},{"updated":"2023-11-14 12:07:08.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-12-04 08:38:21.000000000","updated_by":{"_account_id":28609,"name":"Arkady Shtempler","email":"ashtempl@redhat.com","username":"ashtempl"},"reviewer":{"_account_id":28609,"name":"Arkady Shtempler","email":"ashtempl@redhat.com","username":"ashtempl"},"state":"REVIEWER"},{"updated":"2024-01-16 14:35:34.000000000","updated_by":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"reviewer":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"state":"REVIEWER"}],"messages":[{"id":"7c09b8defa88f9daa6e8ffd6851f47d30f2c4f9f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33634,"name":"Jan Hartkopf","email":"j@hartkopf.io","username":"jhartkopf"},"date":"2023-11-14 10:34:08.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"be095766ad138cd51e60747487f62b82c0f0fe14","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-14 12:07:08.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/99b2351e5d4247d2a003496b59a792e9\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/085c445973cc4085a9a8aa584e8b4f6d : SUCCESS in 49m 47s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/9e621ce36be043f88c5bd44aefad1f30 : SUCCESS in 1h 08m 14s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/cda27fdea117486390fec1108904b30c : SUCCESS in 58m 47s (non-voting)\n- designate-bind9-keystone-default-roles https://zuul.opendev.org/t/openstack/build/8e4c61b18df041eb81d7422f6cfb6e2b : SUCCESS in 51m 11s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/8b9d35bb012d4cedba88e1d5273c31c4 : SUCCESS in 47m 44s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/4a033c06b2254a5c8222ab0c33c98989 : SUCCESS in 1h 08m 42s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/7540460308c84dfaa269c679765ef9d6 : SUCCESS in 1h 25m 57s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/a58513baf9674432991f5532d7fe7bcf : SUCCESS in 48m 42s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/44a1b134539f4d00ae2896cb6b14fff9 : SUCCESS in 51m 05s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bac95e6f88884623bbf025ba2befdcd5 : SUCCESS in 2m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2efa7597e29a400f87ac271087892866 : SUCCESS in 2m 22s\n- designate-bind9-stable-bobcat https://zuul.opendev.org/t/openstack/build/202353be4046483c9a2816dd1439fa98 : SUCCESS in 47m 20s\n- designate-bind9-stable-antelope https://zuul.opendev.org/t/openstack/build/717e5e86299a44df89520fc8c43eee8e : SUCCESS in 32m 13s\n- designate-bind9-stable-zed https://zuul.opendev.org/t/openstack/build/8248941e7f6d4fc48b0b33085855cbef : SUCCESS in 45m 07s\n- designate-bind9-stable-yoga https://zuul.opendev.org/t/openstack/build/f244f694bf8640b8a09c580f2d097229 : SUCCESS in 29m 52s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/bb62fd0e04874095b46b882ab10b0d93 : SUCCESS in 34m 15s","accounts_in_message":[],"_revision_number":1},{"id":"83a484923e9900e4f474ee719a615a2af81a9763","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2023-11-14 22:39:34.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"dc0c4d728fbcd46d55827fe86fd11db7e0986dc5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2023-11-14 22:44:52.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"924dfbea19860a308b940e24b414974184a07b2c","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2023-11-14 22:49:22.000000000","message":"Patch Set 2: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"4b0976754b199bd552da22264279e78f70194374","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2023-11-14 22:49:28.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n","accounts_in_message":[],"_revision_number":3},{"id":"a779a13bd7143805b3f62c7f50223e37487e1255","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2023-11-14 23:37:48.000000000","message":"Patch Set 3: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"d014ac2a9416bf99b16a0ba36a41ad152d07319b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-15 00:07:28.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/76dbbe5249394c0186f27e21241121a4\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/32ab1a9dd05b4f9ca302cc78a5e6b142 : SUCCESS in 39m 02s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/e37786e2734a45eea3b7f0c269c8500f : SUCCESS in 42m 46s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/960834912ae44b0c8b1b9f58aae5def8 : SUCCESS in 39m 42s (non-voting)\n- designate-bind9-keystone-default-roles https://zuul.opendev.org/t/openstack/build/6539e35129244aa6904038d61115e93c : SUCCESS in 31m 08s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/2d399fa5db424d8ab847f2811997e868 : SUCCESS in 46m 35s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/6339757104cc430b92dc85db919051da : SUCCESS in 1h 07m 44s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/ecdb75dadbb94973b7ede0918f54c3b9 : SUCCESS in 1h 12m 19s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/21dd22509ac940378fc7410cd944608d : SUCCESS in 46m 46s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/c228f7a4899c42f0a99fa89693533351 : SUCCESS in 46m 52s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/634774b5bd5542879aa5219837eaef1a : SUCCESS in 2m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/de9100f84e3142938d2984ff2e43625b : SUCCESS in 2m 31s\n- designate-bind9-stable-bobcat https://zuul.opendev.org/t/openstack/build/d831e53d76854c499fa5ce8bd53422fe : SUCCESS in 47m 28s\n- designate-bind9-stable-antelope https://zuul.opendev.org/t/openstack/build/146ef7d5b6604c8a9d9c86d2031deb54 : SUCCESS in 38m 45s\n- designate-bind9-stable-zed https://zuul.opendev.org/t/openstack/build/201df43c8247411b80911863d2671233 : SUCCESS in 44m 46s\n- designate-bind9-stable-yoga https://zuul.opendev.org/t/openstack/build/162ac6d04be4433ab33e14223b052ede : SUCCESS in 51m 14s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/9dde50d376cf4dde9f0a8dc68529bf93 : SUCCESS in 34m 23s","accounts_in_message":[],"_revision_number":3},{"id":"b38b7997ed96fe77f0717c004675eb77b3ad97b9","author":{"_account_id":33634,"name":"Jan Hartkopf","email":"j@hartkopf.io","username":"jhartkopf"},"date":"2023-11-15 09:26:40.000000000","message":"Patch Set 3: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"a7b8284c9faaadda8c9833e865a8b9ace579ac34","author":{"_account_id":28609,"name":"Arkady Shtempler","email":"ashtempl@redhat.com","username":"ashtempl"},"date":"2023-12-04 08:38:21.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"e6aaf6c6bed28d71dc79199aaa8e73a8295faf74","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2024-01-16 14:35:34.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"c6e8e96515c4e7b4db715f51b6beef5e34b802c5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-16 14:36:09.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"f078fbcc8cbe41e9daf98d994a30a3965f61de0b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-16 15:56:33.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b9cdaa9a6be64675816168caba03a131\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/08a38bffdfdb46648394b934a244debd : SUCCESS in 56m 29s\n- designate-bind9-keystone-default-roles https://zuul.opendev.org/t/openstack/build/6b6066a9bef9432cbd634b5e922a7956 : SUCCESS in 52m 34s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/52ce8319cfb34647864f23141150b416 : SUCCESS in 54m 47s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/ec1ac9674b0145f49c76c339592ad70d : SUCCESS in 1h 14m 53s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/fcddc783494544ac93a2a7099b4e2971 : SUCCESS in 49m 22s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/70e17cca231444f4b96c4a9d773a82f7 : SUCCESS in 55m 51s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d2859f7735fd4ceab4d09ba8262382c6 : SUCCESS in 2m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/71372c73c4a846b9ac62f5d607c49d12 : SUCCESS in 2m 14s","accounts_in_message":[],"_revision_number":3},{"id":"2aecfb4439c635ed27cec4eabf11b8807f8a58f5","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-16 15:56:33.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"822114335a5050df876d5be9cec758e2268e8d01","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-16 15:57:29.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/68392771787d49b688425e56adee897b\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0d05c1c27426487081193f8cf2ebcfc9 : SUCCESS in 37s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"d1c54e39a3ad2f832386129ee2d6530db01c7576","revisions":{"6c045e31c704391ffd8f86f1b621c86119048df3":{"kind":"REWORK","_number":1,"created":"2023-11-14 10:34:08.000000000","uploader":{"_account_id":33634,"name":"Jan Hartkopf","email":"j@hartkopf.io","username":"jhartkopf"},"ref":"refs/changes/70/900870/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate-tempest-plugin","ref":"refs/changes/70/900870/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate-tempest-plugin refs/changes/70/900870/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate-tempest-plugin refs/changes/70/900870/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate-tempest-plugin refs/changes/70/900870/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate-tempest-plugin refs/changes/70/900870/1"}}},"commit":{"parents":[{"commit":"1e733017424590f8e5c994b5ecd7b128618756ad","subject":"Increase dns lookup timeout","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate-tempest-plugin/commit/1e733017424590f8e5c994b5ecd7b128618756ad"}]}],"author":{"name":"Jan Hartkopf","email":"jhartkopf@inovex.de","date":"2023-11-14 10:29:06.000000000","tz":60},"committer":{"name":"Jan Hartkopf","email":"jhartkopf@inovex.de","date":"2023-11-14 10:29:06.000000000","tz":60},"subject":"Adapt Tempest test to catalog zones","message":"Adapt Tempest test to catalog zones\n\nTest test_show_zone_impersonate_another_project expected the tested\nzone to be the first one. However, with catalog zones enabled, the\nfirst zone returned is now the catalog zone and not the expected one.\nAdapt the test to catalog zones by checking the last zone instead.\n\nChange-Id: I67e709c9d61af3c5b75b79f27c0c2e9dc8fa08b9\nSigned-off-by: Jan Hartkopf \u003cjhartkopf@inovex.de\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate-tempest-plugin/commit/6c045e31c704391ffd8f86f1b621c86119048df3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate-tempest-plugin/commit/6c045e31c704391ffd8f86f1b621c86119048df3"}]},"branch":"refs/heads/master"},"207fcaa9af48751a1b6164e2e52a5a4b76731dd7":{"kind":"REWORK","_number":2,"created":"2023-11-14 22:44:52.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/70/900870/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate-tempest-plugin","ref":"refs/changes/70/900870/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate-tempest-plugin refs/changes/70/900870/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate-tempest-plugin refs/changes/70/900870/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate-tempest-plugin refs/changes/70/900870/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate-tempest-plugin refs/changes/70/900870/2"}}},"commit":{"parents":[{"commit":"1e733017424590f8e5c994b5ecd7b128618756ad","subject":"Increase dns lookup timeout","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate-tempest-plugin/commit/1e733017424590f8e5c994b5ecd7b128618756ad"}]}],"author":{"name":"Jan Hartkopf","email":"jhartkopf@inovex.de","date":"2023-11-14 10:29:06.000000000","tz":60},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2023-11-14 22:44:50.000000000","tz":0},"subject":"Fix test_show_zone_impersonate_another_project","message":"Fix test_show_zone_impersonate_another_project\n\nTest test_show_zone_impersonate_another_project was testing by getting a list\nof zones from the API instead of requesting a specific zone to show.\nThis patch corrects that, such that the test is using the show zone API code\npath.\n\nChange-Id: I67e709c9d61af3c5b75b79f27c0c2e9dc8fa08b9\nSigned-off-by: Jan Hartkopf \u003cjhartkopf@inovex.de\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate-tempest-plugin/commit/207fcaa9af48751a1b6164e2e52a5a4b76731dd7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate-tempest-plugin/commit/207fcaa9af48751a1b6164e2e52a5a4b76731dd7"}]},"branch":"refs/heads/master"},"d1c54e39a3ad2f832386129ee2d6530db01c7576":{"kind":"REWORK","_number":3,"created":"2023-11-14 22:49:28.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/70/900870/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate-tempest-plugin","ref":"refs/changes/70/900870/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate-tempest-plugin refs/changes/70/900870/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate-tempest-plugin refs/changes/70/900870/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate-tempest-plugin refs/changes/70/900870/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate-tempest-plugin refs/changes/70/900870/3"}}},"commit":{"parents":[{"commit":"1e733017424590f8e5c994b5ecd7b128618756ad","subject":"Increase dns lookup timeout","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate-tempest-plugin/commit/1e733017424590f8e5c994b5ecd7b128618756ad"}]}],"author":{"name":"Jan Hartkopf","email":"jhartkopf@inovex.de","date":"2023-11-14 10:29:06.000000000","tz":60},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2023-11-14 22:48:36.000000000","tz":0},"subject":"Fix test_show_zone_impersonate_another_project","message":"Fix test_show_zone_impersonate_another_project\n\nTest test_show_zone_impersonate_another_project was testing by getting a list\nof zones from the API instead of requesting a specific zone to show.\nThis patch corrects that, such that the test is using the show zone API code\npath.\n\nChange-Id: I67e709c9d61af3c5b75b79f27c0c2e9dc8fa08b9\nSigned-off-by: Jan Hartkopf \u003cjhartkopf@inovex.de\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate-tempest-plugin/commit/d1c54e39a3ad2f832386129ee2d6530db01c7576"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate-tempest-plugin/commit/d1c54e39a3ad2f832386129ee2d6530db01c7576"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":28609,"name":"Arkady Shtempler","email":"ashtempl@redhat.com","username":"ashtempl"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
