)]}'
{"id":"openstack%2Fdesignate-tempest-plugin~918726","triplet_id":"openstack%2Fdesignate-tempest-plugin~master~I6579b4018333848ab206cca275dc7f6aa7578ef7","project":"openstack/designate-tempest-plugin","branch":"master","topic":"test_designate_manage","attention_set":{},"removed_from_attention_set":{"31664":{"account":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"last_update":"2024-05-28 10:17:02.000000000","reason":"\u003cGERRIT_ACCOUNT_31664\u003e replied on the change","reason_account":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"}}},"hashtags":[],"change_id":"I6579b4018333848ab206cca275dc7f6aa7578ef7","subject":"WIP Add designate-manage pool scenario tests","status":"ABANDONED","created":"2024-05-09 11:50:49.000000000","updated":"2025-03-05 16:01:15.000000000","total_comment_count":4,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"e6ef44e4f0ff4239d0c2b03b4ade4a4dcd11b344","_number":918726,"virtual_id_number":918726,"owner":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"actions":{},"labels":{"Verified":{"all":[{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-05-09 13:03:37.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"7a9247f3ec4b04893885a57c5723649ab2f55c7d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"date":"2024-05-09 11:50:49.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"0987bf32778c95397ecd3b05002ef17bcb765d73","author":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"date":"2024-05-09 11:51:49.000000000","message":"Patch Set 1: Workflow-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"3dfa1f7ba084b37ec2a4c69d2a7a544276d28cb7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-09 13:03:37.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/37962c4cc7b6455888bc447358ae7dc1\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/9aff7797cb05452bbd578e1af70ebe73 : FAILURE in 21m 11s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/cb0d1f1dac494208a59a8b0b9f8cc7cc : FAILURE in 29m 11s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/9b4c73fc268545ad8ac57cfc4ad722b0 : FAILURE in 40m 37s (non-voting)\n- designate-bind9-keystone-default-roles https://zuul.opendev.org/t/openstack/build/e2f854ecc6e14db987895a04df2d6440 : FAILURE in 19m 23s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/a3cf20725b714d40abeb96f54d4eafcc : FAILURE in 32m 38s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/dda5698513614a748d332272c3bfc58e : FAILURE in 49m 08s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/87c988ea18084c90b59213ab2a30a72c : FAILURE in 57m 11s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/94466b9cd23f4eaa95d9823960b61f2a : FAILURE in 21m 30s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/dcad82cb12314091b9fc856f4293091d : FAILURE in 32m 49s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d739271558864552b878252ca134462f : SUCCESS in 2m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/df5915eeef564f1a97d1bd7fd808c9c3 : SUCCESS in 2m 20s\n- designate-bind9-stable-bobcat https://zuul.opendev.org/t/openstack/build/0e79c821c0f040e9a1a23d11103b42f3 : FAILURE in 34m 18s\n- designate-bind9-stable-antelope https://zuul.opendev.org/t/openstack/build/86683cecb3e44cd8a1f1eb2a68deaef7 : FAILURE in 24m 07s\n- designate-bind9-stable-zed https://zuul.opendev.org/t/openstack/build/d86ba0c2c29e48acb3ae65a71dab6111 : FAILURE in 6m 10s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/6e4b4544592944e4948dbeb4a65890f9 : FAILURE in 29m 36s","accounts_in_message":[],"_revision_number":1},{"id":"701fe1e51e4833ea9f44580cfb4a974fd7d18c0c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"date":"2024-05-16 17:39:33.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1\n* Workflow-1\n","accounts_in_message":[],"_revision_number":2},{"id":"1a3c7cb103d63621bc6f24348018791a88bbb9b2","author":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"date":"2024-05-16 17:40:15.000000000","message":"Patch Set 2: Workflow-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"6420dce80bad78a1c7ab674d58211954683c0575","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-16 18:54:54.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/967578d64f2847bf93c345a80f041cc0\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/6045e35416124d4ea0ea745e4683596b : FAILURE in 55m 59s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/e859302dda3546b485461abc3edd1601 : FAILURE in 58m 12s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/5dce3b39574c4020a4e0181a512d9d90 : FAILURE in 43m 44s (non-voting)\n- designate-bind9-keystone-default-roles https://zuul.opendev.org/t/openstack/build/bdb13d6c7e8243869fd69c16ee22ad2d : FAILURE in 43m 36s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/566fa67f847d4521b4337c53951a4ca1 : FAILURE in 49m 29s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/e320cac8612f4a5a83a99cea84999435 : FAILURE in 1h 00m 17s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/30462b82382b41a6810aa5aa669e7160 : FAILURE in 1h 07m 24s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/5a05630aec58467a86acf09260d078c3 : FAILURE in 47m 07s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/73787bb3c57d4996a9229aa8a46a40d7 : FAILURE in 42m 49s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/29ce60b292bc41bd8b3e41440b66bd4b : SUCCESS in 2m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5304746d9b5c489b8048dd0b29f21b59 : SUCCESS in 2m 31s\n- designate-bind9-stable-bobcat https://zuul.opendev.org/t/openstack/build/ea25966188be457089e2f2839ba86c1b : FAILURE in 33m 46s\n- designate-bind9-stable-antelope https://zuul.opendev.org/t/openstack/build/65db1190f8e14f5284789bc2fe50ae8a : FAILURE in 29m 23s\n- designate-bind9-stable-zed https://zuul.opendev.org/t/openstack/build/f47082015c974222ad091be51bdb5996 : FAILURE in 14m 52s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/5b1b3f73e1354eb1b377a648d9958ac2 : SUCCESS in 32m 40s","accounts_in_message":[],"_revision_number":2},{"id":"e85041baf47936bb888d1bbfa2d780182e1ddbc3","author":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"date":"2024-05-20 11:46:06.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"7dcd22c75ac7879cd9339c4c4331b115a710453d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"date":"2024-05-20 11:46:51.000000000","message":"Patch Set 3: Published edit on patch set 2.\n\nOutdated Votes:\n* Verified-1\n* Workflow-1\n","accounts_in_message":[],"_revision_number":3},{"id":"afb26c83ae6c0f8c9535503894329908b05c7cf7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-20 13:03:24.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/03bc712b31074b65bad6888b9ba4d3db\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/58e85a4544034dbb803d2dd7a76bc6e9 : FAILURE in 34m 40s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/781e53d730b24334b4947b220263ec26 : FAILURE in 48m 56s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/4a956263eefb4055bf20ecbf9dfe25f4 : FAILURE in 38m 12s (non-voting)\n- designate-bind9-keystone-default-roles https://zuul.opendev.org/t/openstack/build/31a8f123bf3f47a6b45a260e14d6f1cb : FAILURE in 39m 58s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/ec8a414b896845839cbbfde4e5afa3af : FAILURE in 33m 33s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/c81d306c5eab45709534242a442c3e19 : FAILURE in 48m 43s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/99ec3a37e5a44b99ba9cde2dd16295d5 : FAILURE in 1h 06m 08s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/ccc3e76ec6044fcabaa64a4e8bf0da1d : FAILURE in 34m 22s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/2d625628dcca46118859a7284918d583 : FAILURE in 53m 17s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/070d676b2ff74001a35ed9c97bccc585 : SUCCESS in 2m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0211c00ec32b4397b5e4b261f7861f3c : SUCCESS in 3m 20s\n- designate-bind9-stable-bobcat https://zuul.opendev.org/t/openstack/build/b24c9ffc775f41b5bb5d1fc822e364a7 : FAILURE in 46m 25s\n- designate-bind9-stable-antelope https://zuul.opendev.org/t/openstack/build/002c5cd11d484f218d317b17d2e68f91 : FAILURE in 42m 28s\n- designate-bind9-stable-zed https://zuul.opendev.org/t/openstack/build/886a5c44ca634e09a39f3003e79d265f : FAILURE in 15m 04s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/dd9aa414e6df43bfbf82d2aa08fc80c2 : SUCCESS in 31m 35s","accounts_in_message":[],"_revision_number":3},{"id":"dcc9d2d99e9352ba5271d7a76b33fd4bb9e1ae02","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"date":"2024-05-28 09:50:08.000000000","message":"Uploaded patch set 4.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":4},{"id":"0acd23533d630850c256209bd79bf9cc4843ba77","author":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"date":"2024-05-28 10:17:02.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"3e09f7459b794c8db7eeed99cf9184fa6d0f86a7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"date":"2024-05-28 10:19:30.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"e6ef44e4f0ff4239d0c2b03b4ade4a4dcd11b344","tag":"autogenerated:gerrit:abandon","author":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"date":"2025-03-05 16:01:15.000000000","message":"Abandoned\n\nduplicate of https://review.opendev.org/c/openstack/designate-tempest-plugin/+/920653","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"61f033c6be7bc3d98599f9f1deef04bae4957324","revisions":{"72487bcf89e67399b5199af3690673cdebdaee29":{"kind":"REWORK","_number":1,"created":"2024-05-09 11:50:49.000000000","uploader":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"ref":"refs/changes/26/918726/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate-tempest-plugin","ref":"refs/changes/26/918726/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate-tempest-plugin refs/changes/26/918726/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate-tempest-plugin refs/changes/26/918726/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate-tempest-plugin refs/changes/26/918726/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate-tempest-plugin refs/changes/26/918726/1"}}},"commit":{"parents":[{"commit":"347fdbc9b4595a10b726526b3c0b5928e5b7fcf2","subject":"Using a dedicated External network for PTR test suite","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate-tempest-plugin/commit/347fdbc9b4595a10b726526b3c0b5928e5b7fcf2"}]}],"author":{"name":"Omer","email":"oschwart@redhat.com","date":"2024-05-08 22:21:51.000000000","tz":120},"committer":{"name":"Omer","email":"oschwart@redhat.com","date":"2024-05-09 11:50:48.000000000","tz":120},"subject":"WIP Add designate-manage pool scenario tests","message":"WIP Add designate-manage pool scenario tests\n\nSo far we did not test designate-manage pool commands with real end-user\nscenarios.\n\nThis patch adds designate-manage pool scenario tests.\n\nChange-Id: I6579b4018333848ab206cca275dc7f6aa7578ef7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate-tempest-plugin/commit/72487bcf89e67399b5199af3690673cdebdaee29"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate-tempest-plugin/commit/72487bcf89e67399b5199af3690673cdebdaee29"}]},"branch":"refs/heads/master"},"b21a9d0386448422c63bf2ad02675fd242961094":{"kind":"REWORK","_number":2,"created":"2024-05-16 17:39:33.000000000","uploader":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"ref":"refs/changes/26/918726/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate-tempest-plugin","ref":"refs/changes/26/918726/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate-tempest-plugin refs/changes/26/918726/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate-tempest-plugin refs/changes/26/918726/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate-tempest-plugin refs/changes/26/918726/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate-tempest-plugin refs/changes/26/918726/2"}}},"commit":{"parents":[{"commit":"347fdbc9b4595a10b726526b3c0b5928e5b7fcf2","subject":"Using a dedicated External network for PTR test suite","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate-tempest-plugin/commit/347fdbc9b4595a10b726526b3c0b5928e5b7fcf2"}]}],"author":{"name":"Omer","email":"oschwart@redhat.com","date":"2024-05-08 22:21:51.000000000","tz":120},"committer":{"name":"Omer","email":"oschwart@redhat.com","date":"2024-05-16 17:39:24.000000000","tz":120},"subject":"WIP Add designate-manage pool scenario tests","message":"WIP Add designate-manage pool scenario tests\n\nSo far we did not test designate-manage pool commands with real end-user\nscenarios.\n\nThis patch adds designate-manage pool scenario tests.\n\nDepends-On: Icf73e730b31ab26b8be65347239636c0137ab4bd\nChange-Id: I6579b4018333848ab206cca275dc7f6aa7578ef7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate-tempest-plugin/commit/b21a9d0386448422c63bf2ad02675fd242961094"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate-tempest-plugin/commit/b21a9d0386448422c63bf2ad02675fd242961094"}]},"branch":"refs/heads/master"},"f2c6ffc54cad6684d21f9719e8cc89d335ce872f":{"kind":"REWORK","_number":3,"created":"2024-05-20 11:46:51.000000000","uploader":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"ref":"refs/changes/26/918726/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate-tempest-plugin","ref":"refs/changes/26/918726/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate-tempest-plugin refs/changes/26/918726/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate-tempest-plugin refs/changes/26/918726/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate-tempest-plugin refs/changes/26/918726/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate-tempest-plugin refs/changes/26/918726/3"}}},"commit":{"parents":[{"commit":"347fdbc9b4595a10b726526b3c0b5928e5b7fcf2","subject":"Using a dedicated External network for PTR test suite","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate-tempest-plugin/commit/347fdbc9b4595a10b726526b3c0b5928e5b7fcf2"}]}],"author":{"name":"Omer","email":"oschwart@redhat.com","date":"2024-05-08 22:21:51.000000000","tz":120},"committer":{"name":"Omer Schwartz","email":"oschwart@redhat.com","date":"2024-05-20 11:46:51.000000000","tz":0},"subject":"WIP Add designate-manage pool scenario tests","message":"WIP Add designate-manage pool scenario tests\n\nSo far we did not test designate-manage pool commands with real end-user\nscenarios.\n\nThis patch adds designate-manage pool scenario tests.\n\nDepends-On: Icf73e730b31ab26b8be65347239636c0137ab4bd\nChange-Id: I6579b4018333848ab206cca275dc7f6aa7578ef7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate-tempest-plugin/commit/f2c6ffc54cad6684d21f9719e8cc89d335ce872f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate-tempest-plugin/commit/f2c6ffc54cad6684d21f9719e8cc89d335ce872f"}]},"branch":"refs/heads/master"},"5569cc45bef14f0f780608d038338f206fd02784":{"kind":"REWORK","_number":4,"created":"2024-05-28 09:50:08.000000000","uploader":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"ref":"refs/changes/26/918726/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate-tempest-plugin","ref":"refs/changes/26/918726/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate-tempest-plugin refs/changes/26/918726/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate-tempest-plugin refs/changes/26/918726/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate-tempest-plugin refs/changes/26/918726/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate-tempest-plugin refs/changes/26/918726/4"}}},"commit":{"parents":[{"commit":"347fdbc9b4595a10b726526b3c0b5928e5b7fcf2","subject":"Using a dedicated External network for PTR test suite","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate-tempest-plugin/commit/347fdbc9b4595a10b726526b3c0b5928e5b7fcf2"}]}],"author":{"name":"Omer","email":"oschwart@redhat.com","date":"2024-05-08 22:21:51.000000000","tz":120},"committer":{"name":"Omer","email":"oschwart@redhat.com","date":"2024-05-28 09:49:58.000000000","tz":120},"subject":"WIP Add designate-manage pool scenario tests","message":"WIP Add designate-manage pool scenario tests\n\nSo far we did not test designate-manage pool commands with real end-user\nscenarios.\n\nThis patch adds designate-manage pool scenario tests.\n\nDepends-On: Icf73e730b31ab26b8be65347239636c0137ab4bd\nChange-Id: I6579b4018333848ab206cca275dc7f6aa7578ef7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate-tempest-plugin/commit/5569cc45bef14f0f780608d038338f206fd02784"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate-tempest-plugin/commit/5569cc45bef14f0f780608d038338f206fd02784"}]},"branch":"refs/heads/master"},"61f033c6be7bc3d98599f9f1deef04bae4957324":{"kind":"REWORK","_number":5,"created":"2024-05-28 10:19:30.000000000","uploader":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"ref":"refs/changes/26/918726/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate-tempest-plugin","ref":"refs/changes/26/918726/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate-tempest-plugin refs/changes/26/918726/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate-tempest-plugin refs/changes/26/918726/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate-tempest-plugin refs/changes/26/918726/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate-tempest-plugin refs/changes/26/918726/5"}}},"commit":{"parents":[{"commit":"347fdbc9b4595a10b726526b3c0b5928e5b7fcf2","subject":"Using a dedicated External network for PTR test suite","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate-tempest-plugin/commit/347fdbc9b4595a10b726526b3c0b5928e5b7fcf2"}]}],"author":{"name":"Omer","email":"oschwart@redhat.com","date":"2024-05-08 22:21:51.000000000","tz":120},"committer":{"name":"Omer","email":"oschwart@redhat.com","date":"2024-05-28 10:18:54.000000000","tz":120},"subject":"WIP Add designate-manage pool scenario tests","message":"WIP Add designate-manage pool scenario tests\n\nSo far we did not test designate-manage pool commands with real end-user\nscenarios.\n\nThis patch adds designate-manage pool scenario tests.\n\nDepends-On: Icf73e730b31ab26b8be65347239636c0137ab4bd\nChange-Id: I6579b4018333848ab206cca275dc7f6aa7578ef7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate-tempest-plugin/commit/61f033c6be7bc3d98599f9f1deef04bae4957324"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate-tempest-plugin/commit/61f033c6be7bc3d98599f9f1deef04bae4957324"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
