)]}'
{"id":"openstack%2Fdesignate-tempest-plugin~923061","triplet_id":"openstack%2Fdesignate-tempest-plugin~master~Ie543305bfc526654e32b0f876833fc58871f8418","project":"openstack/designate-tempest-plugin","branch":"master","attention_set":{},"removed_from_attention_set":{"11628":{"account":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"last_update":"2026-06-23 18:01:39.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ie543305bfc526654e32b0f876833fc58871f8418","subject":"Increase the default query timeout setting","status":"MERGED","created":"2024-06-28 16:34:17.000000000","updated":"2026-06-23 18:03:51.000000000","submitted":"2026-06-23 18:01:39.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"923061","meta_rev_id":"96530a24533e9dc4309a268c3effa654103709d7","_number":923061,"virtual_id_number":923061,"owner":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-06-23 18:01:39.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"all":[{"value":2,"date":"2026-06-23 16:25:11.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"all":[{"value":1,"date":"2026-06-23 16:25:11.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-06-28 17:49:59.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-07-11 08:18:22.000000000","updated_by":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"reviewer":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"state":"REVIEWER"}],"messages":[{"id":"d29b7a8a9dffd6bb87cc5fed941de177b7282ae0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2024-06-28 16:34:17.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"a7eefb888ff4c161f9bcee811e29b31444c0c409","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-28 17:49:59.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/31fbf107ebd44d7481a2525b73b81060\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/1a0bbcfc355f4458ae04377a84c76176 : SUCCESS in 39m 36s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/3b808203521f41929459e55ebcd09554 : SUCCESS in 56m 52s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/dbb7333249e9412ab761acf35a12314b : SUCCESS in 52m 53s (non-voting)\n- designate-bind9-keystone-default-roles https://zuul.opendev.org/t/openstack/build/e2fe448e31434ff79fbcb62cdfc0e818 : SUCCESS in 44m 00s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/6793aefacc254c4ea5433b85e67f58d3 : SUCCESS in 44m 22s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/a9d669a5f697467c803106e6b31bdbaa : SUCCESS in 1h 15m 01s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/cda9825a7c304bdd80c231bf4f65190b : SUCCESS in 1h 11m 59s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/7f1d390cdde0492ca4255b1c05128c21 : SUCCESS in 45m 01s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/a52bc639fac64d82a6e0530a905540c1 : FAILURE in 54m 01s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b98d703ace2a4cd983acd1a0bcba486a : SUCCESS in 3m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/81f4faf5a465459893abbe99e738819d : SUCCESS in 2m 25s\n- designate-bind9-stable-bobcat https://zuul.opendev.org/t/openstack/build/6219944df3004492bb302e946f09819f : SUCCESS in 48m 09s\n- designate-bind9-stable-antelope https://zuul.opendev.org/t/openstack/build/e0abd4ba2f2d4a36a6ebf3c19838c3dd : SUCCESS in 40m 26s\n- designate-bind9-stable-zed https://zuul.opendev.org/t/openstack/build/3bfe3245d7084bc1bef6b4af88064cae : FAILURE in 16m 43s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/d992134e751543fe8b6b884f22c76b5c : SUCCESS in 38m 37s","accounts_in_message":[],"_revision_number":1},{"id":"f41ebbb302821fa1d0856ec220017636918479b8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2024-06-29 00:19:11.000000000","message":"Patch Set 2: Patch Set 1 was rebased\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"6d62f384f73bc36d8d867684ae6ad44fbcbc1fae","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-29 01:35:26.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d45fe06c13394f35b76ffa618b9f66ee\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/9fc049aba14c4fe78dc1b53cf5a970f3 : SUCCESS in 43m 40s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/f0480d79e50343b5be0305584de1d5ab : SUCCESS in 45m 42s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/f533013e34d7484abb2b6a3004d6cef2 : SUCCESS in 38m 35s (non-voting)\n- designate-bind9-keystone-default-roles https://zuul.opendev.org/t/openstack/build/2e50860845304df1afaf561f49002a78 : SUCCESS in 45m 20s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/d56e0a05fb6d4b34bec06c140a8f339f : SUCCESS in 36m 29s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/be86968433e44ed0be36010832d17472 : SUCCESS in 1h 10m 53s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/2ce30ab278224e0993f7e97fd4f86c64 : SUCCESS in 1h 03m 15s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/fb1c944a9d70478ea8b738a513884269 : SUCCESS in 37m 59s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/8a07d6cf388b472f995460fd00577a0b : SUCCESS in 42m 22s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e1799082215443a389f8ef217b03419b : SUCCESS in 3m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/01d6808b2efa4b2aaeedc6dad9b32bbd : SUCCESS in 2m 28s\n- designate-bind9-stable-2024-1 https://zuul.opendev.org/t/openstack/build/be090d86cd0a4608acfcba61a8ad4b55 : SUCCESS in 44m 55s\n- designate-bind9-stable-bobcat https://zuul.opendev.org/t/openstack/build/52400f0bafe349b5b034f73fa5337678 : SUCCESS in 44m 31s\n- designate-bind9-stable-antelope https://zuul.opendev.org/t/openstack/build/6911e035e19741568b64e3b78bfeb284 : SUCCESS in 46m 25s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/a65eef48004e40feb42c5a6a7306cfbf : SUCCESS in 31m 03s","accounts_in_message":[],"_revision_number":2},{"id":"9c8a6e193bf199ffab132b6ba29cdc4c975a73c5","author":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"date":"2024-07-11 08:18:22.000000000","message":"Patch Set 2: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"a57cdcbf9979982fcf32400d4dc986a28b2da732","author":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"date":"2026-06-23 16:25:11.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"969d86cadfd24c1b9529b92f7f8dbc2491f0b8d2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-23 16:26:02.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"36a76621ded5b449d4b3fdaf0901641795f19438","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-23 18:01:39.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3e812703b1074d8c815ad356a58eeec2\n\n- designate-bind9-core https://zuul.opendev.org/t/openstack/build/5a749a1c67534579b7b3c426d9c3c5bf : SUCCESS in 49m 04s\n- designate-bind9-with-keystone-default-roles https://zuul.opendev.org/t/openstack/build/5126b7749cbb4589af2a6253181fc465 : SUCCESS in 43m 01s\n- designate-pdns4-core https://zuul.opendev.org/t/openstack/build/2a2fa3309f884032a3f45073bbe9b053 : SUCCESS in 53m 51s\n- designate-bind9-multipool https://zuul.opendev.org/t/openstack/build/f3eed25af14748108ab9136b9a723d8f : SUCCESS in 18m 22s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/ccc846755264439299d4b1738001cdf3 : SUCCESS in 1h 18m 53s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/40617dc1cc864180a6d3ab5f8b6b23e4 : SUCCESS in 1h 21m 19s\n- designate-grenade-pdns4-skip-level https://zuul.opendev.org/t/openstack/build/66d7b43d291a4f8d90abfd1676c07987 : SUCCESS in 1h 07m 06s\n- designate-grenade-bind9-skip-level https://zuul.opendev.org/t/openstack/build/9694a533409841e388a4e2905becee8b : SUCCESS in 1h 06m 29s\n- designate-only-ipv6-pdns4 https://zuul.opendev.org/t/openstack/build/4f11ac3cb96e429db8445802c04853a1 : SUCCESS in 51m 55s\n- designate-only-ipv6-bind9 https://zuul.opendev.org/t/openstack/build/941353af1dd348c78ab8901d5b8bd2fe : SUCCESS in 51m 50s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c6fb9b65f94c4d37ac5e7226390f2509 : SUCCESS in 2m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/16ff03ecc5af45b9af2b1543c2d6f126 : SUCCESS in 1m 57s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e5b448679dab4bd880ac7967f9f2f189 : SUCCESS in 5m 02s\n- designate-bind9-stable-2026-1 https://zuul.opendev.org/t/openstack/build/fa606d1d613d48b5b46906051b265cb5 : SUCCESS in 1h 00m 02s\n- designate-bind9-stable-2025-2 https://zuul.opendev.org/t/openstack/build/f744886d5fc2435f96699670010982c1 : SUCCESS in 52m 42s\n- designate-bind9-stable-2025-1 https://zuul.opendev.org/t/openstack/build/bff6dfd39f234296b28c4d19d45601e6 : SUCCESS in 49m 13s\n- designate-bind9-multipool-stable-2026-1 https://zuul.opendev.org/t/openstack/build/3567d7675c4e40559a44ed516e58435e : SUCCESS in 18m 14s\n- designate-bind9-multipool-stable-2025-2 https://zuul.opendev.org/t/openstack/build/5603fde436db43ba9c40d8b7bdbfee8b : SUCCESS in 33m 06s\n- designate-bind9-multipool-stable-2025-1 https://zuul.opendev.org/t/openstack/build/752404b001d6435bb016a3371f2a5a38 : SUCCESS in 31m 47s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/8abe30cd01144a31aa9a8014822157af : SUCCESS in 33m 54s","accounts_in_message":[],"_revision_number":2},{"id":"707b42826635545fe12ebffb540d31658ce80521","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-23 18:01:39.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"96530a24533e9dc4309a268c3effa654103709d7","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-23 18:03:51.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/543102af586d4b5db0824d5e4b50fb1e\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c2e6e87f8adf47e782aa5b7cc2cbe4fd : SUCCESS in 40s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"c264ee2f1fe0e61b3b93ff429f4cad4ebd7bdd45","revisions":{"f65b40ec8afe2d9446572f0a6b70e61600f7f340":{"kind":"REWORK","_number":1,"created":"2024-06-28 16:34:17.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/61/923061/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate-tempest-plugin","ref":"refs/changes/61/923061/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate-tempest-plugin refs/changes/61/923061/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate-tempest-plugin refs/changes/61/923061/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate-tempest-plugin refs/changes/61/923061/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate-tempest-plugin refs/changes/61/923061/1"}}},"commit":{"parents":[{"commit":"347fdbc9b4595a10b726526b3c0b5928e5b7fcf2","subject":"Using a dedicated External network for PTR test suite","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate-tempest-plugin/commit/347fdbc9b4595a10b726526b3c0b5928e5b7fcf2"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2024-06-28 16:32:33.000000000","tz":0},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2024-06-28 16:34:16.000000000","tz":0},"subject":"Increase the default query timeout setting","message":"Increase the default query timeout setting\n\nWe are seeing more dnspython query timeouts across the backends during\ntest runs. This patch increases the default timeout to 4.\nThere are similar changes in the dnspython tests recently as well[1].\n\n[1] https://github.com/rthalley/dnspython/commit/51db0f61e45ce0f1a3f14400f8729e7bb33695cc\n\nChange-Id: Ie543305bfc526654e32b0f876833fc58871f8418\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate-tempest-plugin/commit/f65b40ec8afe2d9446572f0a6b70e61600f7f340"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate-tempest-plugin/commit/f65b40ec8afe2d9446572f0a6b70e61600f7f340"}]},"branch":"refs/heads/master"},"c264ee2f1fe0e61b3b93ff429f4cad4ebd7bdd45":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2024-06-29 00:19:11.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/61/923061/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate-tempest-plugin","ref":"refs/changes/61/923061/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate-tempest-plugin refs/changes/61/923061/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate-tempest-plugin refs/changes/61/923061/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate-tempest-plugin refs/changes/61/923061/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate-tempest-plugin refs/changes/61/923061/2"}}},"commit":{"parents":[{"commit":"7767d31665fec70c088e6aa2d2757eece21dfda7","subject":"Merge \"Updating python testing classifier as per Yoga testing runtime\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate-tempest-plugin/commit/7767d31665fec70c088e6aa2d2757eece21dfda7"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2024-06-28 16:32:33.000000000","tz":0},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2024-06-29 00:19:11.000000000","tz":0},"subject":"Increase the default query timeout setting","message":"Increase the default query timeout setting\n\nWe are seeing more dnspython query timeouts across the backends during\ntest runs. This patch increases the default timeout to 4.\nThere are similar changes in the dnspython tests recently as well[1].\n\n[1] https://github.com/rthalley/dnspython/commit/51db0f61e45ce0f1a3f14400f8729e7bb33695cc\n\nChange-Id: Ie543305bfc526654e32b0f876833fc58871f8418\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate-tempest-plugin/commit/c264ee2f1fe0e61b3b93ff429f4cad4ebd7bdd45"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate-tempest-plugin/commit/c264ee2f1fe0e61b3b93ff429f4cad4ebd7bdd45"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{"label:Review-Priority\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
