)]}'
{"id":"openstack%2Fdesignate-tempest-plugin~927814","triplet_id":"openstack%2Fdesignate-tempest-plugin~master~Ic32b4b4d2f9cff45de94f55246572257fd7dea99","project":"openstack/designate-tempest-plugin","branch":"master","topic":"remove_ptr_record_tests","attention_set":{},"removed_from_attention_set":{"31664":{"account":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"last_update":"2025-04-07 12:59:34.000000000","reason":"\u003cGERRIT_ACCOUNT_31664\u003e replied on the change","reason_account":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"}}},"hashtags":[],"change_id":"Ic32b4b4d2f9cff45de94f55246572257fd7dea99","subject":"Adapt ptr record api tests","status":"NEW","created":"2024-09-03 13:06:41.000000000","updated":"2026-06-24 12:26:26.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":true,"submittable":false,"total_comment_count":4,"unresolved_comment_count":1,"has_review_started":true,"meta_rev_id":"f70de9ae13abb654b9f8ad61e1880842345a469e","_number":927814,"virtual_id_number":927814,"owner":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},{"tag":"autogenerated:zuul:check","value":1,"date":"2025-04-07 14:36:23.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"rejected":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"all":[{"value":-1,"date":"2026-06-24 12:26:26.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-09-03 14:13:14.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"01c20c10cfa86e24d841287d59a19888571e28d3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"date":"2024-09-03 13:06:41.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"d72cecd5488906918c75b020d4b9628982cecd3d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-03 14:13:14.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ba36912aecc34a75b60386e5a4eb91f8\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/04ae9c009285400b896c7e1099991585 : SUCCESS in 50m 25s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/81875087ec9145d78e921bf34cd8c3ac : FAILURE in 31m 28s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/95cbf8ca8a394d5a95031b078b5bfec6 : FAILURE in 25m 43s (non-voting)\n- designate-bind9-keystone-default-roles https://zuul.opendev.org/t/openstack/build/4ce3ced448924c42a2e1077b1b6a5e94 : SUCCESS in 43m 08s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/787e8df151644c13a010f176b689cb35 : SUCCESS in 52m 02s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/feb699407d77426c8d7b9bb9243bd4ae : SUCCESS in 1h 04m 35s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/bed8bf383c6340f4b4da6a3c61cd649e : SUCCESS in 1h 04m 24s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/87d2947cd13c47fa8364af9f8d583e9b : SUCCESS in 42m 34s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/b62e914a1d0f4efd868ba6d106927114 : SUCCESS in 45m 22s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ccdd2f3d71364f739bd527b8e2ff0af4 : SUCCESS in 2m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7e4d012efd28405ab7107f2b182b9dcf : SUCCESS in 2m 43s\n- designate-bind9-stable-2024-1 https://zuul.opendev.org/t/openstack/build/d677f9b0d40a4a6faea5057a1ed62490 : SUCCESS in 36m 19s\n- designate-bind9-stable-bobcat https://zuul.opendev.org/t/openstack/build/6afbbc58f5f64b6e8979f013d2d38510 : SUCCESS in 41m 33s\n- designate-bind9-stable-antelope https://zuul.opendev.org/t/openstack/build/8c4e6e5b45ef4a76a2e784165bde331d : SUCCESS in 39m 09s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/125f6f9b50224aef9b6bb794faa7b462 : SUCCESS in 45m 27s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"640cb15ead350791f31f1d3a61bd92f64188af27","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"date":"2024-09-10 17:02:11.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"96a79c97aa431428903bbbe77171e3fe488b2633","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-10 18:29:02.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/9aa3b3feff784320b3901f65d4d4092b\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/75ebfbe9b45a4a229411b1c2f99bf615 : SUCCESS in 42m 30s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/3683f4dc802249e4bc0df2b71219320d : SUCCESS in 59m 42s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/5e741efec15a42da8a880e76ccf46cd3 : SUCCESS in 49m 42s (non-voting)\n- designate-bind9-keystone-default-roles https://zuul.opendev.org/t/openstack/build/4e783b0b879540caa4d0388ff99827f4 : SUCCESS in 50m 00s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/86443a8ff46b47c295d07016d064d413 : SUCCESS in 40m 45s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/5d3ad27e66ff43afb48efc1c96f0b1df : SUCCESS in 1h 18m 46s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/e9f4e591d8fb42189ba9967eca09e2b5 : SUCCESS in 1h 03m 52s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/414a213bc9f34fb2ac380e2e13a50293 : SUCCESS in 52m 07s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/5d4abcb59d2342969b0f22ffd28496c5 : SUCCESS in 45m 19s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5f95e50633ea4b0996e4af339eafb3e6 : SUCCESS in 2m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/dd03d161fa4f4ef5a3bc80ba8332cd68 : FAILURE in 2m 20s\n- designate-bind9-stable-2024-1 https://zuul.opendev.org/t/openstack/build/03bfe77f0a4e4cc4b91511921a0a8230 : SUCCESS in 42m 41s\n- designate-bind9-stable-bobcat https://zuul.opendev.org/t/openstack/build/b870d4054eaa42b79c5ec1f1a2720b60 : SUCCESS in 37m 38s\n- designate-bind9-stable-antelope https://zuul.opendev.org/t/openstack/build/c8da731f9c4c49b4837df35bdfbc5561 : SUCCESS in 42m 42s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/5af5ec30f1784358a12f8457ad52a4e5 : SUCCESS in 38m 27s","accounts_in_message":[],"_revision_number":2},{"id":"153934964fe9db9c66e7be2d9a370aecd3fbac6f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"date":"2024-09-10 23:20:35.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"800b4c1474acfe3ec07924839afa0815638c8687","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-11 00:32:34.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/565dab2bc4374ed6b7ff125f10cd7195\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/caade58c6b664192a4c4939d0863933d : SUCCESS in 41m 13s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/e194546d91364d7a8c238f6b445d4068 : SUCCESS in 48m 20s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/40110a5c190341788b66eec135365917 : SUCCESS in 46m 07s (non-voting)\n- designate-bind9-keystone-default-roles https://zuul.opendev.org/t/openstack/build/2d2f62455ec74b4683b7f000bab8b034 : SUCCESS in 50m 21s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/d35c5732505c44a5b4b384bb7d17b8a5 : SUCCESS in 45m 39s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/ba2d11ffbc204c9ab8a69b7cc2eb79a1 : SUCCESS in 51m 44s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/2f14e437c084469bb540bbdaa087bc14 : SUCCESS in 52m 46s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/ef3269c6dc3c4f448cd798769feb701f : SUCCESS in 44m 34s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/ca939a905a6b4a9cb3fc11f79ddf5dd5 : SUCCESS in 37m 57s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/49975ff2bf9240b29eb335e8a488f550 : SUCCESS in 2m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6f8de10719914fe8abe9121510beef1f : SUCCESS in 1m 59s\n- designate-bind9-stable-2024-1 https://zuul.opendev.org/t/openstack/build/78090dee635c43a0a98fbd192fa03a62 : SUCCESS in 42m 19s\n- designate-bind9-stable-bobcat https://zuul.opendev.org/t/openstack/build/93792292d5f1401090287e78bb3bbb86 : SUCCESS in 44m 43s\n- designate-bind9-stable-antelope https://zuul.opendev.org/t/openstack/build/da4d2dd67a1440e290613f93982221d2 : SUCCESS in 33m 46s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/cb174d2142e349a3a4dd0020ace07077 : SUCCESS in 31m 19s","accounts_in_message":[],"_revision_number":3},{"id":"7474332ce907424a227bd5d9b8d2a9285ad5498c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"date":"2024-09-12 10:00:41.000000000","message":"Uploaded patch set 4.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":4},{"id":"ab3a597d975f8c05706f142206ceb3dba2187f3e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-12 11:15:57.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/c402325d48ec40de8205fecc67fe7958\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/7e14c284cc58455d9387f38c2bc6d0af : FAILURE in 45m 31s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/ddf6df9bba8f4ed484cdad7b61cf4139 : FAILURE in 46m 44s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/2665b4ea4bef4bb9a04b3c277b7e6c2d : FAILURE in 53m 13s (non-voting)\n- designate-bind9-keystone-default-roles https://zuul.opendev.org/t/openstack/build/35810b286fee4559b2d992bdceb685e7 : FAILURE in 46m 43s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/deb97e85f262416482476ad48d011c3b : FAILURE in 36m 09s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/f9051d64d54946ac8f59eb0d7760d3e3 : FAILURE in 1h 11m 52s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/53d726eab1ec4f30a5d7098d4fd98974 : FAILURE in 56m 20s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/2b75196e8ff1468cb84b1cafddf3d32e : FAILURE in 37m 00s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/b81106f1f78c45c6bd9ded91e6a09809 : FAILURE in 47m 02s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c5b8417bcc1849169e38ca620de96476 : SUCCESS in 2m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/244e790044d6406c9e97f956f77d68f0 : FAILURE in 3m 52s\n- designate-bind9-stable-2024-1 https://zuul.opendev.org/t/openstack/build/e1c220e2d1db4bb3a0fcfd82b7db0c27 : FAILURE in 38m 58s\n- designate-bind9-stable-bobcat https://zuul.opendev.org/t/openstack/build/8b214d6e77be46c38b0cdbef5af6dc95 : FAILURE in 50m 32s\n- designate-bind9-stable-antelope https://zuul.opendev.org/t/openstack/build/d810ddc944064647ab9591e72940f1a5 : FAILURE in 42m 45s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/81b44cee096d4f85b3256bb6d05cf005 : SUCCESS in 48m 19s","accounts_in_message":[],"_revision_number":4},{"id":"05a2e60c051b8b759e48ae89998f47571f812ba1","author":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"date":"2024-09-13 09:50:58.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"e671d2c143dc5cb808baa437c39437a4b5310ba3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"date":"2024-10-01 15:35:03.000000000","message":"Patch Set 5: Commit message was updated.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":5},{"id":"c9ffedebac1d83600045cdc34cb6d50db3d5bc42","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-01 15:35:15.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/cf3217dbc46b41d6982e7192006eb81d\n\n\nWarning:\n  Dependency cycle detected and project openstack/designate doesn\u0027t allow circular dependencies","accounts_in_message":[],"_revision_number":5},{"id":"49ebea8927597b8ada1930eca863c42380cc3ba4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"date":"2025-01-07 12:53:21.000000000","message":"Uploaded patch set 6.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":6},{"id":"38a61b565d2303ad346fe669cb78d0b221fa5dec","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-07 14:05:40.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/4e8ebf2c5b8b48b59a7e6c2651f72bce\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/7f3fc1ade250432baf0299b0bca88433 : SUCCESS in 41m 29s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/1e5b5abeb8394f29a2ef96abb1eed5a8 : SUCCESS in 48m 36s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/5f62c843736a4f7eafeb29b8fd02275c : SUCCESS in 57m 34s (non-voting)\n- designate-bind9-keystone-default-roles https://zuul.opendev.org/t/openstack/build/ceb303e525cc4462b52904c035c8643e : SUCCESS in 36m 55s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/5c62d5b2482346e7ba9e8798ff186d04 : SUCCESS in 49m 43s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/b1dd1e985808436c9f4b9129f276f1a8 : SUCCESS in 59m 28s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/624f88eb520840ba84bc5c2bfc29d230 : SUCCESS in 59m 16s\n- designate-grenade-bind9-skip-level https://zuul.opendev.org/t/openstack/build/9c5ce058247147e8beef9c46026402da : SUCCESS in 1h 06m 52s (non-voting)\n- designate-grenade-pdns4-skip-level https://zuul.opendev.org/t/openstack/build/f98579c6598241cdb043db5e1ca652d0 : SUCCESS in 58m 45s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/c3faa990a11d47668f4e88643a6a83b1 : SUCCESS in 49m 09s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/86fc0fc4ffdf40dd8917aae7b8f3c337 : SUCCESS in 47m 00s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/aaa20e5ad95e4d7b849284c8f7dd3d3f : SUCCESS in 2m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/510602c3711e47368c841c9dd5572353 : FAILURE in 2m 14s\n- designate-bind9-stable-2024-2 https://zuul.opendev.org/t/openstack/build/a414416eabed447fa8c946bb3780f256 : SUCCESS in 46m 57s\n- designate-bind9-stable-2024-1 https://zuul.opendev.org/t/openstack/build/8723287d151a4320869c39656a55e981 : SUCCESS in 41m 44s\n- designate-bind9-stable-2023-2 https://zuul.opendev.org/t/openstack/build/8a62c71d511c4602aa773b0f60b3e6a4 : SUCCESS in 46m 19s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/9bb1c45e4bec4e36ab9c6ff991ff9a03 : SUCCESS in 32m 46s","accounts_in_message":[],"_revision_number":6},{"id":"f75d197b9d8889062ab177e100ba0e2e7a469a6e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"date":"2025-01-07 14:19:05.000000000","message":"Uploaded patch set 7.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":7},{"id":"994adbeccf2c287b317589bf26d589ca45cf30b0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-07 15:31:11.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1b04b68f214c49d5b5940c397211c3ad\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/b5e69c0ccd0f485692879a5789be659e : SUCCESS in 50m 47s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/f2401567cef64ff2ab4a7b4aea16511f : SUCCESS in 1h 06m 05s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/ab7398efd3554e19a2101969d4af6a14 : SUCCESS in 55m 48s (non-voting)\n- designate-bind9-keystone-default-roles https://zuul.opendev.org/t/openstack/build/bf8cb7c5b4f1411f86411a1d1eca352b : SUCCESS in 52m 03s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/e5165a95b61849dda1f188f0eaf73949 : SUCCESS in 51m 17s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/477a3aa8c35344d5bdc7e0f4f76c16b2 : SUCCESS in 1h 02m 40s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/26e3001e4bbb4e63be8172dc4a3b316d : SUCCESS in 52m 34s\n- designate-grenade-bind9-skip-level https://zuul.opendev.org/t/openstack/build/192439c5e3df408daf70ff158d90689b : SUCCESS in 1h 01m 08s (non-voting)\n- designate-grenade-pdns4-skip-level https://zuul.opendev.org/t/openstack/build/471a43b998ec4adc815ba55473898eef : SUCCESS in 58m 15s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/77e68f6745e0487fa0f78f6b95922e92 : SUCCESS in 39m 18s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/41a8d22d46b84c6c9fe8c0928a6aa23e : SUCCESS in 51m 25s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9d32fe7851394856a0a0bf4dae5eb8f2 : SUCCESS in 2m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5ea0451ff8fd4bd68cfa7faa832016a7 : SUCCESS in 2m 23s\n- designate-bind9-stable-2024-2 https://zuul.opendev.org/t/openstack/build/b0813d58d253425682738f0572f04b32 : SUCCESS in 54m 44s\n- designate-bind9-stable-2024-1 https://zuul.opendev.org/t/openstack/build/146d0382dd394cf8abbc105aa548bbb6 : SUCCESS in 36m 05s\n- designate-bind9-stable-2023-2 https://zuul.opendev.org/t/openstack/build/1e84d7f155cd4043898185e53e54b196 : SUCCESS in 29m 34s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/70a13c5a2ce04b8eb66b3d305cdc7b19 : SUCCESS in 43m 34s","accounts_in_message":[],"_revision_number":7},{"id":"fe32a9a2480bb55c257cb4a79a2851639bfa6d1c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"date":"2025-04-07 11:30:40.000000000","message":"Uploaded patch set 8.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":8},{"id":"eb2f9be9b290ecf6e98364c5a90a70f09c8033b9","author":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"date":"2025-04-07 11:44:42.000000000","message":"Patch Set 8:\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"ac0ba040508f66b4aebc18758f39b7830c2651f6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"date":"2025-04-07 11:44:50.000000000","message":"Uploaded patch set 9.","accounts_in_message":[],"_revision_number":9},{"id":"2e20478a0f6728c4acbfb85a57db77e4a48c2024","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-07 12:52:15.000000000","message":"Patch Set 9: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/f997f428ae1449f0ad034b989853e474\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/addf378b641b4401aab16b530708be13 : FAILURE in 46m 50s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/eb5e5b13b7394d188c68a2d07d35380b : FAILURE in 52m 02s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/686fc6de088b4dbd8075391818d4df5c : FAILURE in 45m 59s (non-voting)\n- designate-bind9-keystone-old-default-roles https://zuul.opendev.org/t/openstack/build/759f9855cdf147c59588f6b781b84c93 : FAILURE in 44m 40s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/6dd4ea9ab36d473bb0b1f8498aead6a8 : FAILURE in 49m 02s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/a72204d5aede4dba91754067ffa82d6b : FAILURE in 1h 02m 14s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/ec71336f2770489f884bac894f1306e7 : FAILURE in 50m 40s\n- designate-grenade-bind9-skip-level https://zuul.opendev.org/t/openstack/build/05b0e8eb0c7c4b00bd0a6f179b19cc72 : FAILURE in 44m 28s (non-voting)\n- designate-grenade-pdns4-skip-level https://zuul.opendev.org/t/openstack/build/c1899ab9e3ac406e8a9a5ea7683a5c67 : FAILURE in 47m 34s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/c599550435e0442f82ad12e76579360d : FAILURE in 46m 55s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/bd8db3e7e27041848f91fa30df739925 : FAILURE in 44m 45s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7b825704540349029eba6924d41c5911 : SUCCESS in 2m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9283b8bc9a73450986ecc5dcc6e9dde3 : SUCCESS in 2m 18s\n- designate-bind9-stable-2024-2 https://zuul.opendev.org/t/openstack/build/9b779b6ffce041f48da219dbf68428db : FAILURE in 36m 30s\n- designate-bind9-stable-2024-1 https://zuul.opendev.org/t/openstack/build/cda04048c2ae4a70bc6d3cdd0e82a933 : FAILURE in 30m 11s\n- designate-bind9-stable-2023-2 https://zuul.opendev.org/t/openstack/build/2b3f68c685cf4ba38f333b772e162c67 : FAILURE in 32m 36s\n- designate-bind9-multipool https://zuul.opendev.org/t/openstack/build/3e85f98deb384cf1a724bf4da2ff9019 : SUCCESS in 25m 49s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/2eb0e2d0e3794bb5bef9bd8a2e4863a4 : SUCCESS in 45m 41s","accounts_in_message":[],"_revision_number":9},{"id":"9323c6ed31bd46815f898a8308521daa8dd06872","author":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"date":"2025-04-07 12:59:34.000000000","message":"Patch Set 9:\n\n(1 comment)","accounts_in_message":[],"_revision_number":9},{"id":"919ae95122fe93175ce5669cc06ee393dc9c6a46","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"date":"2025-04-07 13:02:03.000000000","message":"Uploaded patch set 10.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":10},{"id":"cca8866762261f3d4a60ed04d56bf1e299a74a64","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"date":"2025-04-07 13:13:13.000000000","message":"Uploaded patch set 11.","accounts_in_message":[],"_revision_number":11},{"id":"9554cea959917dbb7661469ce72432dc084ceaaf","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-07 14:36:23.000000000","message":"Patch Set 11: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/33c88b484eca4489b6b5670a5e72df43\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/9a69b9f6474e475b9b782e310c413313 : SUCCESS in 35m 55s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/60f07cd6967f4fbcb4dea0ff945c9866 : SUCCESS in 1h 15m 44s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/a5be9e96a18c4a129775d45f58d4fdee : SUCCESS in 55m 26s (non-voting)\n- designate-bind9-keystone-old-default-roles https://zuul.opendev.org/t/openstack/build/c702b3d0665143489cffdf1e9a7de7e9 : SUCCESS in 48m 30s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/06b5bf3232c24ae9b67dd241948363d8 : SUCCESS in 42m 53s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/b366be43167840daae80a897b934a8fe : SUCCESS in 1h 04m 00s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/26ae92acbbd64aeda7a53e8e3be2ea8d : SUCCESS in 53m 32s\n- designate-grenade-bind9-skip-level https://zuul.opendev.org/t/openstack/build/1222fa1b0d9f421e9cfb07ba2d0f53e8 : SUCCESS in 1h 01m 41s (non-voting)\n- designate-grenade-pdns4-skip-level https://zuul.opendev.org/t/openstack/build/4a29647098174d02a98105f332b65cd6 : SUCCESS in 1h 04m 51s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/77ef4d723fc149629a094418155f1828 : SUCCESS in 47m 54s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/20a83447b7104c5c8b7c1e3d4e9765a3 : SUCCESS in 47m 20s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7c130a8c245d4f95981a9b2f13225838 : SUCCESS in 3m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/59582226407246ea8ad308fab460c942 : SUCCESS in 4m 49s\n- designate-bind9-stable-2024-2 https://zuul.opendev.org/t/openstack/build/e80334089ac44ddb8488e972605c4f61 : SUCCESS in 44m 12s\n- designate-bind9-stable-2024-1 https://zuul.opendev.org/t/openstack/build/3e45a89cc9e8419891303e89b5da49e3 : SUCCESS in 37m 10s\n- designate-bind9-stable-2023-2 https://zuul.opendev.org/t/openstack/build/b23fad66950147f0b318d61e33bd2ef9 : SUCCESS in 37m 40s\n- designate-bind9-multipool https://zuul.opendev.org/t/openstack/build/a1c12eb8cc034fadabe125f19f8c91a3 : SUCCESS in 24m 59s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/e220ddff30014c71843fe534f99df53f : SUCCESS in 46m 22s","accounts_in_message":[],"_revision_number":11},{"id":"f70de9ae13abb654b9f8ad61e1880842345a469e","author":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"date":"2026-06-24 12:26:26.000000000","message":"Patch Set 11: Workflow-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":11}],"current_revision_number":11,"current_revision":"71e4c9616da041d0d3a124fa17d885678706fa2f","revisions":{"7c484b9618c46f52c3da9510099af6fe4f015bc2":{"kind":"REWORK","_number":1,"created":"2024-09-03 13:06:41.000000000","uploader":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"ref":"refs/changes/14/927814/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate-tempest-plugin","ref":"refs/changes/14/927814/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate-tempest-plugin refs/changes/14/927814/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate-tempest-plugin refs/changes/14/927814/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate-tempest-plugin refs/changes/14/927814/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate-tempest-plugin refs/changes/14/927814/1"}}},"commit":{"parents":[{"commit":"bb1996d7a87c7a68262506b8d5185b81b436e4c4","subject":"Remove system scope token usage","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate-tempest-plugin/commit/bb1996d7a87c7a68262506b8d5185b81b436e4c4"}]}],"author":{"name":"Omer","email":"omersch381@gmail.com","date":"2024-09-03 12:57:46.000000000","tz":0},"committer":{"name":"Omer Schwartz","email":"oschwart@redhat.com","date":"2024-09-03 13:06:37.000000000","tz":0},"subject":"Remove obsolete ptr record api tests","message":"Remove obsolete ptr record api tests\n\nAfter https://review.opendev.org/c/openstack/designate/+/927373 is\nmerged, some api tests won\u0027y be relevant anymore.\n\nThis patch deletes them.\n\nChange-Id: Ic32b4b4d2f9cff45de94f55246572257fd7dea99\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate-tempest-plugin/commit/7c484b9618c46f52c3da9510099af6fe4f015bc2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate-tempest-plugin/commit/7c484b9618c46f52c3da9510099af6fe4f015bc2"}]},"branch":"refs/heads/master"},"692aac6687210e6ea055970207aa7e77824422da":{"kind":"REWORK","_number":2,"created":"2024-09-10 17:02:11.000000000","uploader":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"ref":"refs/changes/14/927814/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate-tempest-plugin","ref":"refs/changes/14/927814/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate-tempest-plugin refs/changes/14/927814/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate-tempest-plugin refs/changes/14/927814/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate-tempest-plugin refs/changes/14/927814/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate-tempest-plugin refs/changes/14/927814/2"}}},"commit":{"parents":[{"commit":"bb1996d7a87c7a68262506b8d5185b81b436e4c4","subject":"Remove system scope token usage","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate-tempest-plugin/commit/bb1996d7a87c7a68262506b8d5185b81b436e4c4"}]}],"author":{"name":"Omer","email":"oschwart@redhat.com","date":"2024-09-03 12:57:46.000000000","tz":0},"committer":{"name":"Omer","email":"oschwart@redhat.com","date":"2024-09-10 17:01:30.000000000","tz":120},"subject":"Remove obsolete ptr record api tests","message":"Remove obsolete ptr record api tests\n\nAfter https://review.opendev.org/c/openstack/designate/+/927373 is\nmerged, some api tests won\u0027y be relevant anymore.\n\nThis patch deletes them.\n\nChange-Id: Ic32b4b4d2f9cff45de94f55246572257fd7dea99\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate-tempest-plugin/commit/692aac6687210e6ea055970207aa7e77824422da"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate-tempest-plugin/commit/692aac6687210e6ea055970207aa7e77824422da"}]},"branch":"refs/heads/master"},"00d57e0b90659850ec18faccfcb3e686213afb89":{"kind":"REWORK","_number":3,"created":"2024-09-10 23:20:35.000000000","uploader":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"ref":"refs/changes/14/927814/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate-tempest-plugin","ref":"refs/changes/14/927814/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate-tempest-plugin refs/changes/14/927814/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate-tempest-plugin refs/changes/14/927814/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate-tempest-plugin refs/changes/14/927814/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate-tempest-plugin refs/changes/14/927814/3"}}},"commit":{"parents":[{"commit":"bb1996d7a87c7a68262506b8d5185b81b436e4c4","subject":"Remove system scope token usage","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate-tempest-plugin/commit/bb1996d7a87c7a68262506b8d5185b81b436e4c4"}]}],"author":{"name":"Omer","email":"oschwart@redhat.com","date":"2024-09-03 12:57:46.000000000","tz":0},"committer":{"name":"Omer","email":"oschwart@redhat.com","date":"2024-09-10 23:20:28.000000000","tz":120},"subject":"Remove obsolete ptr record api tests","message":"Remove obsolete ptr record api tests\n\nAfter https://review.opendev.org/c/openstack/designate/+/927373 is\nmerged, some api tests won\u0027y be relevant anymore.\n\nThis patch deletes them.\n\nChange-Id: Ic32b4b4d2f9cff45de94f55246572257fd7dea99\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate-tempest-plugin/commit/00d57e0b90659850ec18faccfcb3e686213afb89"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate-tempest-plugin/commit/00d57e0b90659850ec18faccfcb3e686213afb89"}]},"branch":"refs/heads/master"},"269abe4d6281e58fac810ba3138ade839ac462ab":{"kind":"REWORK","_number":4,"created":"2024-09-12 10:00:41.000000000","uploader":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"ref":"refs/changes/14/927814/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate-tempest-plugin","ref":"refs/changes/14/927814/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate-tempest-plugin refs/changes/14/927814/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate-tempest-plugin refs/changes/14/927814/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate-tempest-plugin refs/changes/14/927814/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate-tempest-plugin refs/changes/14/927814/4"}}},"commit":{"parents":[{"commit":"bb1996d7a87c7a68262506b8d5185b81b436e4c4","subject":"Remove system scope token usage","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate-tempest-plugin/commit/bb1996d7a87c7a68262506b8d5185b81b436e4c4"}]}],"author":{"name":"Omer","email":"oschwart@redhat.com","date":"2024-09-03 12:57:46.000000000","tz":0},"committer":{"name":"Omer","email":"oschwart@redhat.com","date":"2024-09-12 09:59:06.000000000","tz":120},"subject":"Adapt ptr record api tests","message":"Adapt ptr record api tests\n\nAfter https://review.opendev.org/c/openstack/designate/+/927373 is\nmerged, some API tests will have to be adapted.\n\nThis patch adapts those API tests.\n\nChange-Id: Ic32b4b4d2f9cff45de94f55246572257fd7dea99\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate-tempest-plugin/commit/269abe4d6281e58fac810ba3138ade839ac462ab"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate-tempest-plugin/commit/269abe4d6281e58fac810ba3138ade839ac462ab"}]},"branch":"refs/heads/master"},"15a50c18ae3fd61c38fad6987f0aafe74b404477":{"kind":"NO_CODE_CHANGE","_number":5,"created":"2024-10-01 15:35:03.000000000","uploader":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"ref":"refs/changes/14/927814/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate-tempest-plugin","ref":"refs/changes/14/927814/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate-tempest-plugin refs/changes/14/927814/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate-tempest-plugin refs/changes/14/927814/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate-tempest-plugin refs/changes/14/927814/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate-tempest-plugin refs/changes/14/927814/5"}}},"commit":{"parents":[{"commit":"bb1996d7a87c7a68262506b8d5185b81b436e4c4","subject":"Remove system scope token usage","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate-tempest-plugin/commit/bb1996d7a87c7a68262506b8d5185b81b436e4c4"}]}],"author":{"name":"Omer","email":"oschwart@redhat.com","date":"2024-09-03 12:57:46.000000000","tz":0},"committer":{"name":"Omer Schwartz","email":"oschwart@redhat.com","date":"2024-10-01 15:35:03.000000000","tz":0},"subject":"Adapt ptr record api tests","message":"Adapt ptr record api tests\n\nAfter https://review.opendev.org/c/openstack/designate/+/927373 is\nmerged, some API tests will have to be adapted.\n\nThis patch adapts those API tests.\n\nDepends-On: Icc591a41abe3393dc3ac41e9b8216770d404a29d\nChange-Id: Ic32b4b4d2f9cff45de94f55246572257fd7dea99\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate-tempest-plugin/commit/15a50c18ae3fd61c38fad6987f0aafe74b404477"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate-tempest-plugin/commit/15a50c18ae3fd61c38fad6987f0aafe74b404477"}]},"branch":"refs/heads/master","description":"Edit commit message"},"8841fd70b7547af06ddbea037203b6b490aacc4e":{"kind":"REWORK","_number":6,"created":"2025-01-07 12:53:21.000000000","uploader":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"ref":"refs/changes/14/927814/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate-tempest-plugin","ref":"refs/changes/14/927814/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate-tempest-plugin refs/changes/14/927814/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate-tempest-plugin refs/changes/14/927814/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate-tempest-plugin refs/changes/14/927814/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate-tempest-plugin refs/changes/14/927814/6"}}},"commit":{"parents":[{"commit":"070e4cb6bf2768e6e12a967d6674fd3707901853","subject":"Add designate-manage path to tempest via a devstack script","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate-tempest-plugin/commit/070e4cb6bf2768e6e12a967d6674fd3707901853"}]}],"author":{"name":"Omer","email":"oschwart@redhat.com","date":"2024-09-03 12:57:46.000000000","tz":0},"committer":{"name":"Omer","email":"oschwart@redhat.com","date":"2024-12-26 16:50:53.000000000","tz":60},"subject":"Adapt ptr record api tests","message":"Adapt ptr record api tests\n\nAfter https://review.opendev.org/c/openstack/designate/+/927373 is\nmerged, some API tests will have to be adapted.\n\nThis patch adapts those API tests.\n\nChange-Id: Ic32b4b4d2f9cff45de94f55246572257fd7dea99\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate-tempest-plugin/commit/8841fd70b7547af06ddbea037203b6b490aacc4e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate-tempest-plugin/commit/8841fd70b7547af06ddbea037203b6b490aacc4e"}]},"branch":"refs/heads/master"},"4a538faf0f434256c5ff4b9a63b2e9317e75d1d6":{"kind":"REWORK","_number":7,"created":"2025-01-07 14:19:05.000000000","uploader":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"ref":"refs/changes/14/927814/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate-tempest-plugin","ref":"refs/changes/14/927814/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate-tempest-plugin refs/changes/14/927814/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate-tempest-plugin refs/changes/14/927814/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate-tempest-plugin refs/changes/14/927814/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate-tempest-plugin refs/changes/14/927814/7"}}},"commit":{"parents":[{"commit":"070e4cb6bf2768e6e12a967d6674fd3707901853","subject":"Add designate-manage path to tempest via a devstack script","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate-tempest-plugin/commit/070e4cb6bf2768e6e12a967d6674fd3707901853"}]}],"author":{"name":"Omer","email":"oschwart@redhat.com","date":"2024-09-03 12:57:46.000000000","tz":0},"committer":{"name":"Omer","email":"oschwart@redhat.com","date":"2025-01-07 14:17:42.000000000","tz":60},"subject":"Adapt ptr record api tests","message":"Adapt ptr record api tests\n\nAfter https://review.opendev.org/c/openstack/designate/+/927373 is\nmerged, some API tests will have to be adapted.\n\nThis patch adapts those API tests.\n\nChange-Id: Ic32b4b4d2f9cff45de94f55246572257fd7dea99\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate-tempest-plugin/commit/4a538faf0f434256c5ff4b9a63b2e9317e75d1d6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate-tempest-plugin/commit/4a538faf0f434256c5ff4b9a63b2e9317e75d1d6"}]},"branch":"refs/heads/master"},"26346ba94602f18ab087255f3b39d6b79751bb20":{"kind":"REWORK","_number":8,"created":"2025-04-07 11:30:40.000000000","uploader":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"ref":"refs/changes/14/927814/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate-tempest-plugin","ref":"refs/changes/14/927814/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate-tempest-plugin refs/changes/14/927814/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate-tempest-plugin refs/changes/14/927814/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate-tempest-plugin refs/changes/14/927814/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate-tempest-plugin refs/changes/14/927814/8"}}},"commit":{"parents":[{"commit":"f36fcee44adaf33e9280231c6b0b3a40c9cacd32","subject":"Add multipool scenario tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate-tempest-plugin/commit/f36fcee44adaf33e9280231c6b0b3a40c9cacd32"}]}],"author":{"name":"Omer","email":"oschwart@redhat.com","date":"2024-09-03 12:57:46.000000000","tz":0},"committer":{"name":"Omer","email":"oschwart@redhat.com","date":"2025-04-07 11:28:33.000000000","tz":120},"subject":"Adapt ptr record api tests","message":"Adapt ptr record api tests\n\nAfter https://review.opendev.org/c/openstack/designate/+/927373 is\nmerged, some API tests will have to be adapted.\n\nThis patch adapts those API tests.\n\nChange-Id: Ic32b4b4d2f9cff45de94f55246572257fd7dea99\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate-tempest-plugin/commit/26346ba94602f18ab087255f3b39d6b79751bb20"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate-tempest-plugin/commit/26346ba94602f18ab087255f3b39d6b79751bb20"}]},"branch":"refs/heads/master"},"cd7bb852ddc4f29bc1088fed491b5307f586d331":{"kind":"REWORK","_number":9,"created":"2025-04-07 11:44:50.000000000","uploader":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"ref":"refs/changes/14/927814/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate-tempest-plugin","ref":"refs/changes/14/927814/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate-tempest-plugin refs/changes/14/927814/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate-tempest-plugin refs/changes/14/927814/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate-tempest-plugin refs/changes/14/927814/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate-tempest-plugin refs/changes/14/927814/9"}}},"commit":{"parents":[{"commit":"f36fcee44adaf33e9280231c6b0b3a40c9cacd32","subject":"Add multipool scenario tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate-tempest-plugin/commit/f36fcee44adaf33e9280231c6b0b3a40c9cacd32"}]}],"author":{"name":"Omer","email":"oschwart@redhat.com","date":"2024-09-03 12:57:46.000000000","tz":0},"committer":{"name":"Omer","email":"oschwart@redhat.com","date":"2025-04-07 11:44:34.000000000","tz":120},"subject":"Adapt ptr record api tests","message":"Adapt ptr record api tests\n\nAfter https://review.opendev.org/c/openstack/designate/+/927373 is\nmerged, some API tests will have to be adapted.\n\nThis patch adapts those API tests.\n\nChange-Id: Ic32b4b4d2f9cff45de94f55246572257fd7dea99\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate-tempest-plugin/commit/cd7bb852ddc4f29bc1088fed491b5307f586d331"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate-tempest-plugin/commit/cd7bb852ddc4f29bc1088fed491b5307f586d331"}]},"branch":"refs/heads/master"},"b84f7964d2a22b8cff3fe89063bb3b5fcf63c8cd":{"kind":"REWORK","_number":10,"created":"2025-04-07 13:02:03.000000000","uploader":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"ref":"refs/changes/14/927814/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate-tempest-plugin","ref":"refs/changes/14/927814/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate-tempest-plugin refs/changes/14/927814/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate-tempest-plugin refs/changes/14/927814/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate-tempest-plugin refs/changes/14/927814/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate-tempest-plugin refs/changes/14/927814/10"}}},"commit":{"parents":[{"commit":"f36fcee44adaf33e9280231c6b0b3a40c9cacd32","subject":"Add multipool scenario tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate-tempest-plugin/commit/f36fcee44adaf33e9280231c6b0b3a40c9cacd32"}]}],"author":{"name":"Omer","email":"oschwart@redhat.com","date":"2024-09-03 12:57:46.000000000","tz":0},"committer":{"name":"Omer","email":"oschwart@redhat.com","date":"2025-04-07 13:01:50.000000000","tz":120},"subject":"Adapt ptr record api tests","message":"Adapt ptr record api tests\n\nAfter https://review.opendev.org/c/openstack/designate/+/927373 is\nmerged, some API tests will have to be adapted.\n\nThis patch adapts those API tests.\n\nChange-Id: Ic32b4b4d2f9cff45de94f55246572257fd7dea99\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate-tempest-plugin/commit/b84f7964d2a22b8cff3fe89063bb3b5fcf63c8cd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate-tempest-plugin/commit/b84f7964d2a22b8cff3fe89063bb3b5fcf63c8cd"}]},"branch":"refs/heads/master"},"71e4c9616da041d0d3a124fa17d885678706fa2f":{"kind":"REWORK","_number":11,"created":"2025-04-07 13:13:13.000000000","uploader":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"ref":"refs/changes/14/927814/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate-tempest-plugin","ref":"refs/changes/14/927814/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate-tempest-plugin refs/changes/14/927814/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate-tempest-plugin refs/changes/14/927814/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate-tempest-plugin refs/changes/14/927814/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate-tempest-plugin refs/changes/14/927814/11"}}},"commit":{"parents":[{"commit":"f36fcee44adaf33e9280231c6b0b3a40c9cacd32","subject":"Add multipool scenario tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate-tempest-plugin/commit/f36fcee44adaf33e9280231c6b0b3a40c9cacd32"}]}],"author":{"name":"Omer","email":"oschwart@redhat.com","date":"2024-09-03 12:57:46.000000000","tz":0},"committer":{"name":"Omer","email":"oschwart@redhat.com","date":"2025-04-07 13:12:38.000000000","tz":120},"subject":"Adapt ptr record api tests","message":"Adapt ptr record api tests\n\nAfter https://review.opendev.org/c/openstack/designate/+/927373 is\nmerged, some API tests will have to be adapted.\n\nThis patch adapts those API tests.\n\nChange-Id: Ic32b4b4d2f9cff45de94f55246572257fd7dea99\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate-tempest-plugin/commit/71e4c9616da041d0d3a124fa17d885678706fa2f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate-tempest-plugin/commit/71e4c9616da041d0d3a124fa17d885678706fa2f"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY"},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{"label:Review-Priority\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":["label:Workflow\u003dMIN"],"failing_atoms":["label:Workflow\u003dMAX"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
