)]}'
{"id":"openstack%2Fdesignate-tempest-plugin~946144","triplet_id":"openstack%2Fdesignate-tempest-plugin~master~I168590cd2fe0a0520adae817e3a5871479d27117","project":"openstack/designate-tempest-plugin","branch":"master","topic":"fix_multipool_test_query","hashtags":[],"change_id":"I168590cd2fe0a0520adae817e3a5871479d27117","subject":"Add a method to get nameservers in use","status":"MERGED","created":"2025-04-02 12:49:01.000000000","updated":"2026-04-07 10:37:57.000000000","submitted":"2026-04-07 10:37:06.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"946144-fix_multipool_test_query","meta_rev_id":"859aff7c1e473ac5276b7d6455c06131d017507d","_number":946144,"virtual_id_number":946144,"owner":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-04-07 10:37:06.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"all":[{"value":2,"date":"2026-04-07 09:19:47.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"all":[{"value":1,"date":"2026-04-07 09:19:47.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-04-02 14:06:21.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"7e5dd87b46c4e66457017135bb1f66894bd57826","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"date":"2025-04-02 12:49:01.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"4c5e33ce153227d85076d439fe108f9334ff8bd7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-02 14:06:21.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/cc4ab53f63314eadba784318b847f992\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/7a29d1f2ff544796be1a890f38b7e5a2 : SUCCESS in 54m 34s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/5fa42ba912ea4497b092dc880bc73a85 : SUCCESS in 53m 32s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/e49afa3fb17446649562ab82c33e12e1 : SUCCESS in 38m 28s (non-voting)\n- designate-bind9-keystone-old-default-roles https://zuul.opendev.org/t/openstack/build/05772c916fc145438f6c2f300ada58c0 : SUCCESS in 36m 35s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/d0e1ec6ae8a44c8eaa6e252dfb1d407f : SUCCESS in 50m 23s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/d4133597fefc492fb21f326f9424b45a : SUCCESS in 53m 18s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/b127a55ea0494f47a6a5153dd3b06ed2 : SUCCESS in 1h 05m 40s\n- designate-grenade-bind9-skip-level https://zuul.opendev.org/t/openstack/build/5c2323baa526428cb10ab1fa782bd278 : SUCCESS in 58m 49s (non-voting)\n- designate-grenade-pdns4-skip-level https://zuul.opendev.org/t/openstack/build/49a54229fe6c4c89ae904afa3d280a6c : SUCCESS in 57m 55s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/efe86ef00dee4d739fd9b6e872ad74de : SUCCESS in 35m 21s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/860a6888cf0c47518f39f280eff0ec9f : SUCCESS in 41m 32s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f8c05ad1594c43bb950d8be2dc892432 : SUCCESS in 3m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ec461c70c1c046c9820016b5e1e42893 : SUCCESS in 2m 18s\n- designate-bind9-stable-2024-2 https://zuul.opendev.org/t/openstack/build/21a1a107d0c84affa35de38be79fa72f : SUCCESS in 36m 46s\n- designate-bind9-stable-2024-1 https://zuul.opendev.org/t/openstack/build/181e71f8044a41e4bedaf9c4d4c0f5fb : SUCCESS in 39m 35s\n- designate-bind9-stable-2023-2 https://zuul.opendev.org/t/openstack/build/d11e1cc183964755941e598195a8cbc2 : SUCCESS in 41m 43s\n- designate-bind9-multipool https://zuul.opendev.org/t/openstack/build/d7d71b88e7774bd6b5752535e09f5b19 : SUCCESS in 30m 26s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/d25e79f675e8477d97a0e7b3d9ca9ca0 : SUCCESS in 30m 28s","accounts_in_message":[],"_revision_number":1},{"id":"26cc41e740bd446b84ce956fae1885cb1b6a45ba","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"date":"2025-11-21 15:29:03.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"ba5227473b90e78c1143039a586578104811da5c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-21 16:39:35.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/23db0fa0fe594d4698cde2b7744749eb\n\n- designate-bind9-core https://zuul.opendev.org/t/openstack/build/47ad20b9f8d4464ea9cdc516264a0a0a : SUCCESS in 46m 25s\n- designate-bind9-with-keystone-default-roles https://zuul.opendev.org/t/openstack/build/20428b8ae87c49b8bdb761059c68c252 : SUCCESS in 47m 14s\n- designate-pdns4-core https://zuul.opendev.org/t/openstack/build/bb2c15348fe844dcbce12b5821b00209 : SUCCESS in 47m 15s\n- designate-bind9-multipool https://zuul.opendev.org/t/openstack/build/7bb5822ffc454063b731e482a7140f31 : SUCCESS in 24m 04s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/1438688ff10f4d23a8838b39275f21c2 : SUCCESS in 1h 03m 04s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/8bdafb2816ff42bfa420ebb6860caa0a : SUCCESS in 1h 02m 43s\n- designate-grenade-bind9-skip-level https://zuul.opendev.org/t/openstack/build/ced476194739453bb4acd866b55866ab : SUCCESS in 52m 24s\n- designate-grenade-pdns4-skip-level https://zuul.opendev.org/t/openstack/build/a34ea8d1631e4dd398a2a6b69025b2a6 : SUCCESS in 54m 26s\n- designate-bind9-core-catalog-zones https://zuul.opendev.org/t/openstack/build/4f6369205ad14a1b9ebe4e1f43c2f9b0 : FAILURE in 44m 47s (non-voting)\n- designate-only-ipv6-pdns4 https://zuul.opendev.org/t/openstack/build/32fd349d4f0d471989a6c40849eebbae : SUCCESS in 45m 23s\n- designate-only-ipv6-bind9 https://zuul.opendev.org/t/openstack/build/ea25baefe18440dbae09d7f9ea401e39 : SUCCESS in 51m 40s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6d07e69f955347e38f1a0a65c93697ad : SUCCESS in 3m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/229bcbb2762e4b4c956f6531b31035d2 : SUCCESS in 1m 39s\n- designate-bind9-stable-2025-2 https://zuul.opendev.org/t/openstack/build/3c33ce8c612041f4a5a97e9faca10e2f : SUCCESS in 51m 50s\n- designate-bind9-stable-2025-1 https://zuul.opendev.org/t/openstack/build/2ff4f94a8d0543709d43117bd82d675f : SUCCESS in 48m 58s\n- designate-bind9-stable-2024-2 https://zuul.opendev.org/t/openstack/build/a7462f8e6cea461693ffe1bc3a9e67bd : SUCCESS in 45m 32s\n- designate-bind9-multipool-stable-2025-2 https://zuul.opendev.org/t/openstack/build/f47c63b2ab034c4e947ee8e646ab6ad4 : SUCCESS in 31m 56s\n- designate-bind9-multipool-stable-2025-1 https://zuul.opendev.org/t/openstack/build/8e62de10dfb443d7b73a8485a809ae50 : SUCCESS in 15m 09s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/a2fc6c1873c64287a769fa55cc06256f : SUCCESS in 33m 02s","accounts_in_message":[],"_revision_number":2},{"id":"9c33bff2e6741c1d030cffea7365d42433d08188","author":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"date":"2026-04-07 09:19:47.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"bc44e77425a6055b4bdb0631e56c72c4dd99518f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-07 09:20:19.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"5a243e5cfeb776e300bd7962704b6f0a17c5694f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-07 10:37:06.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ad4c2ccaad574f218e3946ce17a03fa0\n\n- designate-bind9-core https://zuul.opendev.org/t/openstack/build/a24278563d35460aa2b4846e9c55c7d2 : SUCCESS in 55m 09s\n- designate-bind9-with-keystone-default-roles https://zuul.opendev.org/t/openstack/build/35f29118df41448f860809264aa991b3 : SUCCESS in 49m 52s\n- designate-pdns4-core https://zuul.opendev.org/t/openstack/build/bbb031f04f464bbaafd9eddfffd9e20f : SUCCESS in 50m 16s\n- designate-bind9-multipool https://zuul.opendev.org/t/openstack/build/9cc3102e618644429488c5e1f7e4ce16 : SUCCESS in 16m 38s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/7afd4c80bb1f4288b11b3553dbf0fdbd : SUCCESS in 1h 00m 50s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/78ae8b2f930a4ff6a25fbbe6bbb10d4b : SUCCESS in 1h 12m 07s\n- designate-grenade-pdns4-skip-level https://zuul.opendev.org/t/openstack/build/7fac4ef32f864b638ccbb2ff1ca2e085 : SUCCESS in 1h 13m 28s\n- designate-grenade-bind9-skip-level https://zuul.opendev.org/t/openstack/build/31ea864edd8f41ba90790ef5f3dac475 : SUCCESS in 55m 54s\n- designate-only-ipv6-pdns4 https://zuul.opendev.org/t/openstack/build/db4c7944ff4c4c85b512e94303d19efd : SUCCESS in 40m 30s\n- designate-only-ipv6-bind9 https://zuul.opendev.org/t/openstack/build/c1a2dfc5ed8544d2a582c5e4ecfb4fd7 : SUCCESS in 54m 06s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e1fa3f8883e14ee0942e850039c4812e : SUCCESS in 2m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e910907ec2b3405ea3303d153177598e : SUCCESS in 2m 43s\n- designate-bind9-stable-2026-1 https://zuul.opendev.org/t/openstack/build/204edfd75aeb4f1c96f86318ff1b5493 : SUCCESS in 54m 13s\n- designate-bind9-stable-2025-2 https://zuul.opendev.org/t/openstack/build/c2bd3b078e73425ea61633259debb63b : SUCCESS in 47m 14s\n- designate-bind9-stable-2025-1 https://zuul.opendev.org/t/openstack/build/31cd592e06724524a11771e5547ea7a9 : SUCCESS in 51m 25s\n- designate-bind9-stable-2024-2 https://zuul.opendev.org/t/openstack/build/d10c5099ab7d43298796b1270afa6710 : SUCCESS in 44m 27s\n- designate-bind9-multipool-stable-2026-1 https://zuul.opendev.org/t/openstack/build/1fb0f40e3ad04b0582e9333a21b6e78c : SUCCESS in 32m 23s\n- designate-bind9-multipool-stable-2025-2 https://zuul.opendev.org/t/openstack/build/3a2058e647a54b2389195c0a3fc38fa0 : SUCCESS in 31m 58s\n- designate-bind9-multipool-stable-2025-1 https://zuul.opendev.org/t/openstack/build/1fc263a311a8450dab9fa3f6972f9c96 : SUCCESS in 38m 25s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/64f8206903bb484291005230a75d7c8f : SUCCESS in 33m 04s","accounts_in_message":[],"_revision_number":2},{"id":"4bb9485cb04866a2739d0d90123fa9952c0a579f","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-07 10:37:06.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"859aff7c1e473ac5276b7d6455c06131d017507d","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-07 10:37:57.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e26e225b5c1b4c60950c4203d2da0a12\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7ba5ca3f186e4304bd5ac8f49240276b : SUCCESS in 41s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"0fbfca98e799813f18194a79e037a8475f669438","revisions":{"7ab33fa54347831c3b5c7b8bd4f16d0df1d0215c":{"kind":"REWORK","_number":1,"created":"2025-04-02 12:49:01.000000000","uploader":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"ref":"refs/changes/44/946144/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate-tempest-plugin","ref":"refs/changes/44/946144/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate-tempest-plugin refs/changes/44/946144/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate-tempest-plugin refs/changes/44/946144/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate-tempest-plugin refs/changes/44/946144/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate-tempest-plugin refs/changes/44/946144/1"}}},"commit":{"parents":[{"commit":"f36fcee44adaf33e9280231c6b0b3a40c9cacd32","subject":"Add multipool scenario tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate-tempest-plugin/commit/f36fcee44adaf33e9280231c6b0b3a40c9cacd32"}]}],"author":{"name":"Omer","email":"oschwart@redhat.com","date":"2025-04-02 12:30:50.000000000","tz":120},"committer":{"name":"Omer","email":"oschwart@redhat.com","date":"2025-04-02 12:46:06.000000000","tz":120},"subject":"Add a method to get nameservers in use","message":"Add a method to get nameservers in use\n\nPreviously, we processed designate-manage pool show command output\nthrough string manipulations, which was difficult to maintain.\n\nThis patch adds a dedicated method for the nameservers.\n\nThe method was originally presented in the following patch, which is\ncurrently blocked:\nhttps://review.opendev.org/c/openstack/designate-tempest-plugin/+/944760\n\nAuthored-by: Arkady Shtempler \u003cashtempl@redhat.com\u003e\nChange-Id: I168590cd2fe0a0520adae817e3a5871479d27117\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate-tempest-plugin/commit/7ab33fa54347831c3b5c7b8bd4f16d0df1d0215c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate-tempest-plugin/commit/7ab33fa54347831c3b5c7b8bd4f16d0df1d0215c"}]},"branch":"refs/heads/master"},"0fbfca98e799813f18194a79e037a8475f669438":{"kind":"TRIVIAL_REBASE_WITH_MESSAGE_UPDATE","_number":2,"created":"2025-11-21 15:29:03.000000000","uploader":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"ref":"refs/changes/44/946144/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate-tempest-plugin","ref":"refs/changes/44/946144/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate-tempest-plugin refs/changes/44/946144/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate-tempest-plugin refs/changes/44/946144/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate-tempest-plugin refs/changes/44/946144/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate-tempest-plugin refs/changes/44/946144/2"}}},"commit":{"parents":[{"commit":"88fc51f6327a4b8ac50078380d0fc91eca7eb449","subject":"Merge \"Update stable branch jobs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate-tempest-plugin/commit/88fc51f6327a4b8ac50078380d0fc91eca7eb449"}]}],"author":{"name":"Omer","email":"oschwart@redhat.com","date":"2025-04-02 12:30:50.000000000","tz":120},"committer":{"name":"Omer","email":"oschwart@redhat.com","date":"2025-11-21 15:28:52.000000000","tz":60},"subject":"Add a method to get nameservers in use","message":"Add a method to get nameservers in use\n\nPreviously, we processed designate-manage pool show command output\nthrough string manipulations, which was difficult to maintain.\n\nThis patch adds a dedicated method for the nameservers.\n\nThe method was originally presented in the following patch, which is\ncurrently blocked:\nhttps://review.opendev.org/c/openstack/designate-tempest-plugin/+/944760\n\nAuthored-by: Arkady Shtempler \u003cashtempl@redhat.com\u003e\nChange-Id: I168590cd2fe0a0520adae817e3a5871479d27117\nSigned-off-by: Omer \u003coschwart@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate-tempest-plugin/commit/0fbfca98e799813f18194a79e037a8475f669438"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate-tempest-plugin/commit/0fbfca98e799813f18194a79e037a8475f669438"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
