)]}'
{"id":"openstack%2Fdesignate-tempest-plugin~946157","triplet_id":"openstack%2Fdesignate-tempest-plugin~master~I08757cf37cd870fc216e1ed31d1de89eba7fee1c","project":"openstack/designate-tempest-plugin","branch":"master","topic":"potential_fix_futurist","attention_set":{},"removed_from_attention_set":{"31664":{"account":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"last_update":"2025-04-07 11:51:09.000000000","reason":"\u003cGERRIT_ACCOUNT_31664\u003e replied on the change","reason_account":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"}}},"hashtags":[],"change_id":"I08757cf37cd870fc216e1ed31d1de89eba7fee1c","subject":"DNM: testing futurist removal BlockingIOError fix","status":"ABANDONED","created":"2025-04-02 14:12:22.000000000","updated":"2025-08-27 10:05:01.000000000","total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"4055ff8a042d48568084f6b1649482594eae864c","_number":946157,"virtual_id_number":946157,"owner":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2025-04-07 12:55:37.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-04-02 15:40:17.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"fb1d6e6bc0636a49de6e9fa9219621d7e723ac07","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"date":"2025-04-02 14:12:22.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"22c0efb066453f870b198b0994be4a2dc3fbd78b","author":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"date":"2025-04-02 14:16:03.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"1d84aa27921f26021e9fdaf9f298d5ef6a78212e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-02 15:40:17.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b6756af8802c4a589f956b71eb38c753\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/10ccb608dc244d53bebc3e15ef515c2b : SUCCESS in 47m 12s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/7bf93e1e002f4518b2b77436cd25c73e : SUCCESS in 56m 17s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/e18f027bbcfa44a5bb97b5023b661770 : SUCCESS in 49m 08s (non-voting)\n- designate-bind9-keystone-old-default-roles https://zuul.opendev.org/t/openstack/build/926f0640c8cf4107bdee4b2d162d0707 : SUCCESS in 39m 25s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/7f41028ff6854d58bc60d19450d55ae2 : SUCCESS in 40m 55s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/cb96f075f1734f43a8faa65c0d4cc074 : SUCCESS in 1h 03m 10s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/18fcc999099248028d383c1eff345509 : SUCCESS in 1h 01m 12s\n- designate-grenade-bind9-skip-level https://zuul.opendev.org/t/openstack/build/7f827d98aad14c9aa1a3b66c1d67f198 : SUCCESS in 58m 18s (non-voting)\n- designate-grenade-pdns4-skip-level https://zuul.opendev.org/t/openstack/build/4e2896bc6f7847b2b8a97e06480251d6 : SUCCESS in 1h 04m 25s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/27f89eced1fe40199d27b965364851d8 : SUCCESS in 38m 24s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/f5c1818a001a477abfdde8447248aa50 : SUCCESS in 43m 39s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/595cf2d4ccf447019788a7584c22c15c : SUCCESS in 2m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e21cc6a67f77400ca06728c6d297c3c3 : SUCCESS in 3m 41s\n- designate-bind9-stable-2024-2 https://zuul.opendev.org/t/openstack/build/a2efa3eccd00462394cfd755f480f28b : SUCCESS in 33m 03s\n- designate-bind9-stable-2024-1 https://zuul.opendev.org/t/openstack/build/0428945c47004a1c85fdafeba61dd5d6 : SUCCESS in 36m 50s\n- designate-bind9-stable-2023-2 https://zuul.opendev.org/t/openstack/build/85fe59f0b23e4b888d875dd6e6a87332 : SUCCESS in 37m 45s\n- designate-bind9-multipool https://zuul.opendev.org/t/openstack/build/e9d8580123fb402e9df487ec9dd95ac9 : SUCCESS in 27m 42s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/c1abb5744afb4e1fb9e997144dbe1318 : SUCCESS in 30m 25s","accounts_in_message":[],"_revision_number":1},{"id":"fbd52bea900288056153eb43f04f41779a0f7b14","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"date":"2025-04-03 09:22:14.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"7718f1c9288b00e645832a0b3b5ff0c358622818","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-03 10:34:20.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d009a99402044538ba97572e7f2659cd\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/ec1a26b7f6da416485f266c9d113433b : SUCCESS in 46m 03s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/3b082be420c1469681458668732d2630 : SUCCESS in 54m 28s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/2a877e3c0d7c4851a2fe5c21103a6d8c : SUCCESS in 44m 04s (non-voting)\n- designate-bind9-keystone-old-default-roles https://zuul.opendev.org/t/openstack/build/8b9cb941b65e4bb184938899634f36af : SUCCESS in 45m 02s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/fdfc771e1b164b2491387276f4f571f6 : SUCCESS in 46m 57s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/6c6c2235dbda441288580bad1d98f900 : SUCCESS in 1h 05m 06s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/c2509819b2ce4e49b94bcc54f2677283 : SUCCESS in 56m 38s\n- designate-grenade-bind9-skip-level https://zuul.opendev.org/t/openstack/build/aedd9c49775647a28d060942d9cf557e : SUCCESS in 50m 38s (non-voting)\n- designate-grenade-pdns4-skip-level https://zuul.opendev.org/t/openstack/build/ca541454f09748fab3520671488fa125 : SUCCESS in 1h 00m 32s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/05917755883d4124ab508670f467cf89 : SUCCESS in 37m 55s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/2afdeaa1b4314e5d9d9dd9d047424b4b : SUCCESS in 45m 04s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1563682a59104ecd8da7af866a9ebb00 : SUCCESS in 2m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ba288936ad1740ea8b4fe4cc993b8d6f : SUCCESS in 2m 00s\n- designate-bind9-stable-2024-2 https://zuul.opendev.org/t/openstack/build/769a974d539946f6b78b54a01752ebb9 : SUCCESS in 42m 02s\n- designate-bind9-stable-2024-1 https://zuul.opendev.org/t/openstack/build/0e38171b3d6f480bb019669765aea854 : SUCCESS in 31m 18s\n- designate-bind9-stable-2023-2 https://zuul.opendev.org/t/openstack/build/7629103ab4b647ccace2c589334b0bcf : SUCCESS in 37m 30s\n- designate-bind9-multipool https://zuul.opendev.org/t/openstack/build/bd7f9713b3a64c18ba00912c5b167f14 : SUCCESS in 23m 08s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/58e099c608cf45f3a9b15cd2cffdd3ae : SUCCESS in 34m 55s","accounts_in_message":[],"_revision_number":2},{"id":"722cfcfa1c50972d46446f8c93414ea9f36b2ba1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"date":"2025-04-03 17:06:50.000000000","message":"Patch Set 3: Published edit on patch set 2.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":3},{"id":"f1875428cb547f17c00601f126e92219e01b132f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-03 18:13:53.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e2e9432744fa490d8bc03790d03f3dcd\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/96654300edae4a70a8cf0ea2ba00aee4 : SUCCESS in 40m 48s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/021824b160984234aaf1f1881a1a55d2 : SUCCESS in 50m 12s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/2a6587aa781246c2a1216b5e383ac140 : SUCCESS in 47m 16s (non-voting)\n- designate-bind9-keystone-old-default-roles https://zuul.opendev.org/t/openstack/build/d04cbc4cf01f4c33b54d488fe7c511cc : SUCCESS in 45m 53s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/0f121fe371eb4defa6717e100c425584 : SUCCESS in 48m 55s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/2aa525e057754a74b059493651da6960 : FAILURE in 1h 04m 06s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/0409abeac3454c5b8e47c7984209dd36 : SUCCESS in 56m 49s\n- designate-grenade-bind9-skip-level https://zuul.opendev.org/t/openstack/build/b846d3ce95b94237a91a6c2d91eccd5d : SUCCESS in 1h 04m 07s (non-voting)\n- designate-grenade-pdns4-skip-level https://zuul.opendev.org/t/openstack/build/ffa2f4ea59664136be2c748b1a390912 : SUCCESS in 1h 04m 36s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/a25b74b897534c2790f6573bfeec4948 : SUCCESS in 43m 11s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/8d1e1451bfbd44aba01fe20e6c8a0eb5 : SUCCESS in 40m 31s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/656821371103436b85772c20c4142a85 : SUCCESS in 2m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/80e353a1416c478190c6119d9a98305b : SUCCESS in 2m 18s\n- designate-bind9-stable-2024-2 https://zuul.opendev.org/t/openstack/build/826d585218204316bc007a23234f63ba : SUCCESS in 37m 12s\n- designate-bind9-stable-2024-1 https://zuul.opendev.org/t/openstack/build/778a920570404ebc8462b9d17918bd66 : SUCCESS in 37m 46s\n- designate-bind9-stable-2023-2 https://zuul.opendev.org/t/openstack/build/6fada2c9e12b406c94cb81a4f2907f88 : SUCCESS in 33m 23s\n- designate-bind9-multipool https://zuul.opendev.org/t/openstack/build/b39487d4396d426fabae814e213449a5 : SUCCESS in 14m 34s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/aeedb4db46a54d3ca3b8c7eda5267431 : SUCCESS in 40m 02s","accounts_in_message":[],"_revision_number":3},{"id":"9111b328556227a98340e9a0ceeb1d94341f32fc","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"date":"2025-04-04 16:07:36.000000000","message":"Patch Set 4: Published edit on patch set 3.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":4},{"id":"412e46cef2e33575abb6fbe4dd1cf8df7dfe944f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-04 17:31:39.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/2f4cf4d5125e45849831920fad102ee1\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/b267aedd03a143f3bcffac41101b920d : SUCCESS in 40m 03s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/2c71f70d9ac24492afe1b9354554a070 : SUCCESS in 46m 54s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/9a0da814c4fa440083ab3626ad01883d : SUCCESS in 38m 47s (non-voting)\n- designate-bind9-keystone-old-default-roles https://zuul.opendev.org/t/openstack/build/af47369312474eabbb7abcc3b215a51d : SUCCESS in 39m 59s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/ca33c03f26bd4b09808edef719dc37e3 : SUCCESS in 39m 19s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/9b04c257971248b19d71f46161fba189 : SUCCESS in 56m 41s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/1f4636828bd44085b39932b606427f21 : FAILURE in 49m 10s\n- designate-grenade-bind9-skip-level https://zuul.opendev.org/t/openstack/build/99b36826a0f04b8a80d73bb65c2cf006 : SUCCESS in 1h 07m 44s (non-voting)\n- designate-grenade-pdns4-skip-level https://zuul.opendev.org/t/openstack/build/1d7ccc83b26740ddbce58ed38abfd51c : SUCCESS in 1h 07m 43s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/517d9765d97f404ba81765935d7b331d : SUCCESS in 33m 20s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/e8bcfccee509488da85a924b68995cf9 : SUCCESS in 36m 41s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/364b3e347e574cca8e54a82d31955de8 : SUCCESS in 3m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4aedb0aa641747838f847752ee408d0b : SUCCESS in 2m 48s\n- designate-bind9-stable-2024-2 https://zuul.opendev.org/t/openstack/build/256c4d320f4446a49c3f8a6f33a01c86 : SUCCESS in 42m 39s\n- designate-bind9-stable-2024-1 https://zuul.opendev.org/t/openstack/build/1a02de060fc3495ab26b05897a6d3626 : SUCCESS in 36m 13s\n- designate-bind9-stable-2023-2 https://zuul.opendev.org/t/openstack/build/9fce17d89b0f422bb9f0ec4065fe5292 : SUCCESS in 31m 22s\n- designate-bind9-multipool https://zuul.opendev.org/t/openstack/build/737194762e8b40b8a00888994eb4b87e : SUCCESS in 19m 29s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/809e56870169479c9b68ae76f6be7689 : SUCCESS in 32m 36s","accounts_in_message":[],"_revision_number":4},{"id":"0bfcd3b780348f10160ff3b765b6c05dc26854db","author":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"date":"2025-04-07 11:51:09.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"c976085f5cda7d79999f5eddb333be5106637dad","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-07 12:55:37.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/81b82d92ac9842d88a86aab106209b5b\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/0ba118960734431babf775944b043f5d : SUCCESS in 43m 20s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/09ef5ba77b7d45e5ad244edace8af2f5 : SUCCESS in 42m 15s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/e4694e0b4a2a4be6b3f4d922b0ccb763 : SUCCESS in 43m 53s (non-voting)\n- designate-bind9-keystone-old-default-roles https://zuul.opendev.org/t/openstack/build/cb57af1624a14437a61d2c5bad9938e3 : SUCCESS in 34m 44s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/6f5d54549ff349519a6f5f46017c7534 : SUCCESS in 35m 03s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/56d7b52bddb04b949c286001e2fb65f1 : SUCCESS in 59m 53s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/30b8aa334aaa47db891ba505ab028783 : SUCCESS in 59m 24s\n- designate-grenade-bind9-skip-level https://zuul.opendev.org/t/openstack/build/3118293e80e94f6d849b3eedc6b818ae : SUCCESS in 45m 29s (non-voting)\n- designate-grenade-pdns4-skip-level https://zuul.opendev.org/t/openstack/build/4255fbbd705040dcb73721b5bfc0413b : SUCCESS in 51m 14s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/ab8f0951493b4c3ea274b80f46f75509 : SUCCESS in 42m 08s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/f03bd78211974ce583380415b5e4a73f : SUCCESS in 47m 20s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cd33f183f22a453ba28a9ca0742001a5 : SUCCESS in 2m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a8c4cfbeb5e84b238dc8de23d822fff8 : SUCCESS in 1m 51s\n- designate-bind9-stable-2024-2 https://zuul.opendev.org/t/openstack/build/7860b557ab774eeb8f99c93faa0eb7bf : SUCCESS in 42m 46s\n- designate-bind9-stable-2024-1 https://zuul.opendev.org/t/openstack/build/f2bce65df9a34722bb533200ab50425e : SUCCESS in 32m 50s\n- designate-bind9-stable-2023-2 https://zuul.opendev.org/t/openstack/build/a50bfa7601644590be3eac933e680b14 : SUCCESS in 39m 40s\n- designate-bind9-multipool https://zuul.opendev.org/t/openstack/build/62cfa51c2323452ab79e9b2b2d4ce285 : SUCCESS in 20m 22s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/96fcf67b084f40efaa45d8ddeee50dde : SUCCESS in 37m 28s","accounts_in_message":[],"_revision_number":4},{"id":"4055ff8a042d48568084f6b1649482594eae864c","tag":"autogenerated:gerrit:abandon","author":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"date":"2025-08-27 10:05:01.000000000","message":"Abandoned\n\ntesting is done","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"ec3bee211d5d9880380ad7a33cea25fc70449645","revisions":{"2265a81513577a3b206abff28ca31976a855cccf":{"kind":"REWORK","_number":1,"created":"2025-04-02 14:12:22.000000000","uploader":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"ref":"refs/changes/57/946157/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate-tempest-plugin","ref":"refs/changes/57/946157/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate-tempest-plugin refs/changes/57/946157/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate-tempest-plugin refs/changes/57/946157/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate-tempest-plugin refs/changes/57/946157/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate-tempest-plugin refs/changes/57/946157/1"}}},"commit":{"parents":[{"commit":"f36fcee44adaf33e9280231c6b0b3a40c9cacd32","subject":"Add multipool scenario tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate-tempest-plugin/commit/f36fcee44adaf33e9280231c6b0b3a40c9cacd32"}]}],"author":{"name":"Omer","email":"oschwart@redhat.com","date":"2025-04-02 14:11:44.000000000","tz":120},"committer":{"name":"Omer","email":"oschwart@redhat.com","date":"2025-04-02 14:12:02.000000000","tz":120},"subject":"DNM: testing if locking the query solves racing conditions issues","message":"DNM: testing if locking the query solves racing conditions issues\n\nChange-Id: I08757cf37cd870fc216e1ed31d1de89eba7fee1c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate-tempest-plugin/commit/2265a81513577a3b206abff28ca31976a855cccf"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate-tempest-plugin/commit/2265a81513577a3b206abff28ca31976a855cccf"}]},"branch":"refs/heads/master"},"bad7c5406d882c813fd2d30fcc89c595448b5f73":{"kind":"REWORK","_number":2,"created":"2025-04-03 09:22:14.000000000","uploader":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"ref":"refs/changes/57/946157/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate-tempest-plugin","ref":"refs/changes/57/946157/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate-tempest-plugin refs/changes/57/946157/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate-tempest-plugin refs/changes/57/946157/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate-tempest-plugin refs/changes/57/946157/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate-tempest-plugin refs/changes/57/946157/2"}}},"commit":{"parents":[{"commit":"f36fcee44adaf33e9280231c6b0b3a40c9cacd32","subject":"Add multipool scenario tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate-tempest-plugin/commit/f36fcee44adaf33e9280231c6b0b3a40c9cacd32"}]}],"author":{"name":"Omer","email":"oschwart@redhat.com","date":"2025-04-02 14:11:44.000000000","tz":120},"committer":{"name":"Omer","email":"oschwart@redhat.com","date":"2025-04-03 09:22:05.000000000","tz":120},"subject":"DNM: testing futurist removal BlockingIOError fix","message":"DNM: testing futurist removal BlockingIOError fix\n\nChange-Id: I08757cf37cd870fc216e1ed31d1de89eba7fee1c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate-tempest-plugin/commit/bad7c5406d882c813fd2d30fcc89c595448b5f73"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate-tempest-plugin/commit/bad7c5406d882c813fd2d30fcc89c595448b5f73"}]},"branch":"refs/heads/master"},"be56e1752b08ee29253c5af60e4fd96382d0618b":{"kind":"REWORK","_number":3,"created":"2025-04-03 17:06:50.000000000","uploader":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"ref":"refs/changes/57/946157/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate-tempest-plugin","ref":"refs/changes/57/946157/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate-tempest-plugin refs/changes/57/946157/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate-tempest-plugin refs/changes/57/946157/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate-tempest-plugin refs/changes/57/946157/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate-tempest-plugin refs/changes/57/946157/3"}}},"commit":{"parents":[{"commit":"f36fcee44adaf33e9280231c6b0b3a40c9cacd32","subject":"Add multipool scenario tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate-tempest-plugin/commit/f36fcee44adaf33e9280231c6b0b3a40c9cacd32"}]}],"author":{"name":"Omer","email":"oschwart@redhat.com","date":"2025-04-02 14:11:44.000000000","tz":120},"committer":{"name":"Omer Schwartz","email":"oschwart@redhat.com","date":"2025-04-03 17:06:50.000000000","tz":0},"subject":"DNM: testing futurist removal BlockingIOError fix","message":"DNM: testing futurist removal BlockingIOError fix\n\nChange-Id: I08757cf37cd870fc216e1ed31d1de89eba7fee1c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate-tempest-plugin/commit/be56e1752b08ee29253c5af60e4fd96382d0618b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate-tempest-plugin/commit/be56e1752b08ee29253c5af60e4fd96382d0618b"}]},"branch":"refs/heads/master"},"ec3bee211d5d9880380ad7a33cea25fc70449645":{"kind":"REWORK","_number":4,"created":"2025-04-04 16:07:36.000000000","uploader":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"ref":"refs/changes/57/946157/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate-tempest-plugin","ref":"refs/changes/57/946157/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate-tempest-plugin refs/changes/57/946157/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate-tempest-plugin refs/changes/57/946157/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate-tempest-plugin refs/changes/57/946157/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate-tempest-plugin refs/changes/57/946157/4"}}},"commit":{"parents":[{"commit":"f36fcee44adaf33e9280231c6b0b3a40c9cacd32","subject":"Add multipool scenario tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate-tempest-plugin/commit/f36fcee44adaf33e9280231c6b0b3a40c9cacd32"}]}],"author":{"name":"Omer","email":"oschwart@redhat.com","date":"2025-04-02 14:11:44.000000000","tz":120},"committer":{"name":"Omer Schwartz","email":"oschwart@redhat.com","date":"2025-04-04 16:07:35.000000000","tz":0},"subject":"DNM: testing futurist removal BlockingIOError fix","message":"DNM: testing futurist removal BlockingIOError fix\n\nChange-Id: I08757cf37cd870fc216e1ed31d1de89eba7fee1c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate-tempest-plugin/commit/ec3bee211d5d9880380ad7a33cea25fc70449645"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate-tempest-plugin/commit/ec3bee211d5d9880380ad7a33cea25fc70449645"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
