)]}'
{"id":"openstack%2Fdesignate-tempest-plugin~946555","triplet_id":"openstack%2Fdesignate-tempest-plugin~master~Ib95e94e7ec265c666763f69f861cb4518760979d","project":"openstack/designate-tempest-plugin","branch":"master","topic":"remove_ptr_record_tests","attention_set":{},"removed_from_attention_set":{"31664":{"account":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"last_update":"2026-06-24 12:26:32.000000000","reason":"\u003cGERRIT_ACCOUNT_31664\u003e replied on the change","reason_account":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"}}},"hashtags":[],"change_id":"Ib95e94e7ec265c666763f69f861cb4518760979d","subject":"Uncomment test_show_floatingip_ptr_impersonate_another_project","status":"NEW","created":"2025-04-07 13:04:37.000000000","updated":"2026-06-24 12:26:32.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":true,"submittable":false,"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"448c7aca601808010a5dc038190e868b6cdaf376","_number":946555,"virtual_id_number":946555,"owner":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2025-04-08 14:06:10.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"rejected":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"all":[{"value":-1,"date":"2026-06-24 12:26:32.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-04-07 13:13:31.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"f61af76d04a59f9cc006be83b1ffa63d95c22619","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"date":"2025-04-07 13:04:37.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"0c950ab9082c5cc201eaf2d6789e06d8d255227b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"date":"2025-04-07 13:13:13.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"d20d6353ce05627d250ccd828d510c58864d173b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-07 13:13:31.000000000","message":"Patch Set 1: Verified-1\n\nThis change depends on a change that failed to merge.\n\nChange https://review.opendev.org/927373 is needed.","accounts_in_message":[],"_revision_number":1},{"id":"04e8e88fe074c4d44de37c835424ce0a014adc69","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-07 14:22:28.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/dd2cd5b9ff6444199d9eab0605dcbda0\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/90956229d33242e59a711c78f3fe48e9 : SUCCESS in 42m 44s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/d68ce17f78ef41ce9d59c2b59576b161 : SUCCESS in 51m 22s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/25a11de9eb3c454baca3800ea8b5a80a : SUCCESS in 45m 41s (non-voting)\n- designate-bind9-keystone-old-default-roles https://zuul.opendev.org/t/openstack/build/80e30714e13e45718688e0f122357035 : SUCCESS in 40m 46s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/90fc0149355341759be5476faac8257b : SUCCESS in 44m 42s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/f9d015230a6a4b7db33707918fa34fa8 : SUCCESS in 59m 58s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/bcf03438d92840f49f3c887c3b7897c3 : SUCCESS in 56m 11s\n- designate-grenade-bind9-skip-level https://zuul.opendev.org/t/openstack/build/8ee564485e05461a8edfe2c1a9f0d7a6 : SUCCESS in 58m 39s (non-voting)\n- designate-grenade-pdns4-skip-level https://zuul.opendev.org/t/openstack/build/6473bd1f363e4f2993994cc03596e462 : SUCCESS in 59m 43s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/32dc22cb99164905b2a1dabefd7ef897 : SUCCESS in 40m 08s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/51c3af39aeee40acb510454de90a317a : SUCCESS in 43m 27s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8cc9b29c747b474ba786ca83b8415b32 : SUCCESS in 1m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b9a9accfc3d447fd914b8fb8605b82e2 : SUCCESS in 2m 29s\n- designate-bind9-stable-2024-2 https://zuul.opendev.org/t/openstack/build/622d3d0298cc414b9fe7cda63222ad78 : FAILURE in 48m 13s\n- designate-bind9-stable-2024-1 https://zuul.opendev.org/t/openstack/build/bd78a6f6fb01466fb94d659aaedf7153 : FAILURE in 43m 11s\n- designate-bind9-stable-2023-2 https://zuul.opendev.org/t/openstack/build/d096f928b1a3483a9c9e6be7a2505af6 : FAILURE in 32m 45s\n- designate-bind9-multipool https://zuul.opendev.org/t/openstack/build/b77b0f1922b04f36885d1c63732fddb4 : SUCCESS in 24m 18s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/9ece9d25799a441f9ac3ba16d208880e : SUCCESS in 17m 55s","accounts_in_message":[],"_revision_number":2},{"id":"23ee2910eff46fd320bf0dcfe4332018ea0d91dd","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"date":"2025-04-08 11:01:24.000000000","message":"Patch Set 3: Commit message was updated.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"c18e97a4fa6f44123934d2106497088f2fd71e83","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-08 12:02:52.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/66ca5a9691b44ddc9a125508ce258289\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/df28070de36d485fa70be90f9cd3b5b5 : FAILURE in 49m 44s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/5183184325664bd28a0f2e57fe9794ce : FAILURE in 42m 25s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/332db9d460464c41bb6eafd73c99d811 : FAILURE in 40m 15s (non-voting)\n- designate-bind9-keystone-old-default-roles https://zuul.opendev.org/t/openstack/build/85ac5bc5a9794266bbc0ebb1900e827e : FAILURE in 43m 20s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/6dcd1260b450400b80e696586a308eec : FAILURE in 35m 46s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/7126b37c7b744f5fa3f1b3559f34cdc7 : FAILURE in 56m 19s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/e1869f4730d9491497b9214ff6f213af : FAILURE in 1h 00m 46s\n- designate-grenade-bind9-skip-level https://zuul.opendev.org/t/openstack/build/c91fd75d1c09417d992ca85c43feca50 : FAILURE in 58m 29s (non-voting)\n- designate-grenade-pdns4-skip-level https://zuul.opendev.org/t/openstack/build/b06e9101bcd74791b000d1b12357ee1d : FAILURE in 50m 05s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/9f3b4394022d4534b12573aae3c58d2f : FAILURE in 36m 34s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/e148b09302d24291ba0af42f9e22bf65 : FAILURE in 35m 56s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/812f7956e2bb4529961db71d5fa469cc : SUCCESS in 2m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7aa83dc70e804aea9ad5645f142938fe : SUCCESS in 2m 17s\n- designate-bind9-stable-2024-2 https://zuul.opendev.org/t/openstack/build/735ae96108124b5a87148f74edf4d6be : FAILURE in 46m 52s\n- designate-bind9-stable-2024-1 https://zuul.opendev.org/t/openstack/build/9ae8624665dd4b618354718d0e207cdb : FAILURE in 36m 35s\n- designate-bind9-stable-2023-2 https://zuul.opendev.org/t/openstack/build/895d2d43d64843f3a5eb591963f8ff2f : FAILURE in 37m 52s\n- designate-bind9-multipool https://zuul.opendev.org/t/openstack/build/d7f6bf3c08894ff78890d4b5c082ad53 : SUCCESS in 28m 24s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/6976455046414af494b4607bcca31c9f : SUCCESS in 46m 05s","accounts_in_message":[],"_revision_number":3},{"id":"12cf456040e681cac6ffd89f2282881f080f0e00","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"date":"2025-04-08 12:56:04.000000000","message":"Patch Set 4: Commit message was updated.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":4},{"id":"a247f797fcb4d160a8c1984857596b6d1ecf274c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-08 14:06:10.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/837b49b311ba45fab92c19c44809e737\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/be6ddee9d05b42939163886d27f24524 : SUCCESS in 41m 10s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/482aba4fa1a74412927f799c5eae2b69 : SUCCESS in 57m 27s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/00dc9c49baba45f4816422001cd51a5d : SUCCESS in 50m 34s (non-voting)\n- designate-bind9-keystone-old-default-roles https://zuul.opendev.org/t/openstack/build/7983ceaa9e644e8ea7c12d2ea2408809 : SUCCESS in 31m 58s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/09b7e45912734cee9d580545cd61696e : SUCCESS in 39m 02s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/f668a466dead4e2e9040ab25621f9023 : SUCCESS in 49m 34s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/32a3254daa5f48a7b689996118850570 : SUCCESS in 59m 29s\n- designate-grenade-bind9-skip-level https://zuul.opendev.org/t/openstack/build/0b596115cace428997d3c7782cc353b1 : SUCCESS in 1h 03m 34s (non-voting)\n- designate-grenade-pdns4-skip-level https://zuul.opendev.org/t/openstack/build/70be7c019f4f48318b6401c0c2bd3566 : SUCCESS in 45m 12s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/df682fdefd794487ab9893ecfaad73ef : SUCCESS in 38m 07s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/05b83c232dab4a46948a59e192c39e45 : SUCCESS in 45m 44s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cffdea2b409b4f4ebbdd16f32fc8429c : SUCCESS in 3m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d6763834667641e98068b23a213cec8c : SUCCESS in 3m 37s\n- designate-bind9-stable-2024-2 https://zuul.opendev.org/t/openstack/build/ab8447d569ca4743b495e3041269c2e4 : FAILURE in 35m 03s\n- designate-bind9-stable-2024-1 https://zuul.opendev.org/t/openstack/build/f36abbc2d5ec4360b3fd69467cf13f31 : FAILURE in 44m 00s\n- designate-bind9-stable-2023-2 https://zuul.opendev.org/t/openstack/build/10788ff817bf4e88a23d63dd7deba755 : FAILURE in 29m 05s\n- designate-bind9-multipool https://zuul.opendev.org/t/openstack/build/79b42aea67534610a5a8f496b4cf5a96 : SUCCESS in 25m 37s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/b7341fff55bc4f7f9d6d768fdff1e43d : SUCCESS in 32m 28s","accounts_in_message":[],"_revision_number":4},{"id":"448c7aca601808010a5dc038190e868b6cdaf376","author":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"date":"2026-06-24 12:26:32.000000000","message":"Patch Set 4: Workflow-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"14afec8c8e117ce353e0195d8f52330fd255856c","revisions":{"d3ca3c0a2b2adf41840de888230a4a555fe4751a":{"kind":"REWORK","_number":1,"created":"2025-04-07 13:04:37.000000000","uploader":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"ref":"refs/changes/55/946555/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate-tempest-plugin","ref":"refs/changes/55/946555/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate-tempest-plugin refs/changes/55/946555/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate-tempest-plugin refs/changes/55/946555/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate-tempest-plugin refs/changes/55/946555/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate-tempest-plugin refs/changes/55/946555/1"}}},"commit":{"parents":[{"commit":"b84f7964d2a22b8cff3fe89063bb3b5fcf63c8cd","subject":"Adapt ptr record api tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate-tempest-plugin/commit/b84f7964d2a22b8cff3fe89063bb3b5fcf63c8cd"}]}],"author":{"name":"Omer","email":"oschwart@redhat.com","date":"2025-04-07 13:02:16.000000000","tz":120},"committer":{"name":"Omer","email":"oschwart@redhat.com","date":"2025-04-07 13:03:50.000000000","tz":120},"subject":"Uncomment test_show_floatingip_ptr_impersonate_another_project","message":"Uncomment test_show_floatingip_ptr_impersonate_another_project\n\nAfter https://review.opendev.org/c/openstack/designate/+/927373 is\nmerged, we can uncomment this test whose behavior is changed by it.\n\nDepends-On: https://review.opendev.org/c/openstack/designate/+/927373\nChange-Id: Ib95e94e7ec265c666763f69f861cb4518760979d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate-tempest-plugin/commit/d3ca3c0a2b2adf41840de888230a4a555fe4751a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate-tempest-plugin/commit/d3ca3c0a2b2adf41840de888230a4a555fe4751a"}]},"branch":"refs/heads/master"},"c9241dfa1c4d75227924f0a7fda360c6abeff089":{"kind":"REWORK","_number":2,"created":"2025-04-07 13:13:13.000000000","uploader":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"ref":"refs/changes/55/946555/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate-tempest-plugin","ref":"refs/changes/55/946555/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate-tempest-plugin refs/changes/55/946555/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate-tempest-plugin refs/changes/55/946555/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate-tempest-plugin refs/changes/55/946555/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate-tempest-plugin refs/changes/55/946555/2"}}},"commit":{"parents":[{"commit":"71e4c9616da041d0d3a124fa17d885678706fa2f","subject":"Adapt ptr record api tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate-tempest-plugin/commit/71e4c9616da041d0d3a124fa17d885678706fa2f"}]}],"author":{"name":"Omer","email":"oschwart@redhat.com","date":"2025-04-07 13:02:16.000000000","tz":120},"committer":{"name":"Omer","email":"oschwart@redhat.com","date":"2025-04-07 13:13:05.000000000","tz":120},"subject":"Uncomment test_show_floatingip_ptr_impersonate_another_project","message":"Uncomment test_show_floatingip_ptr_impersonate_another_project\n\nAfter https://review.opendev.org/c/openstack/designate/+/927373 is\nmerged, we can uncomment this test whose behavior is changed by it.\n\nDepends-On: https://review.opendev.org/c/openstack/designate/+/927373\nChange-Id: Ib95e94e7ec265c666763f69f861cb4518760979d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate-tempest-plugin/commit/c9241dfa1c4d75227924f0a7fda360c6abeff089"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate-tempest-plugin/commit/c9241dfa1c4d75227924f0a7fda360c6abeff089"}]},"branch":"refs/heads/master"},"e8592122f9fb3c4b0efacb64eae4883e0864c699":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2025-04-08 11:01:24.000000000","uploader":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"ref":"refs/changes/55/946555/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate-tempest-plugin","ref":"refs/changes/55/946555/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate-tempest-plugin refs/changes/55/946555/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate-tempest-plugin refs/changes/55/946555/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate-tempest-plugin refs/changes/55/946555/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate-tempest-plugin refs/changes/55/946555/3"}}},"commit":{"parents":[{"commit":"71e4c9616da041d0d3a124fa17d885678706fa2f","subject":"Adapt ptr record api tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate-tempest-plugin/commit/71e4c9616da041d0d3a124fa17d885678706fa2f"}]}],"author":{"name":"Omer","email":"oschwart@redhat.com","date":"2025-04-07 13:02:16.000000000","tz":120},"committer":{"name":"Omer Schwartz","email":"oschwart@redhat.com","date":"2025-04-08 11:01:24.000000000","tz":0},"subject":"Uncomment test_show_floatingip_ptr_impersonate_another_project","message":"Uncomment test_show_floatingip_ptr_impersonate_another_project\n\nAfter https://review.opendev.org/c/openstack/designate/+/927373 is\nmerged and backported, we could uncomment this test whose behavior is changed by it.\n\nChange-Id: Ib95e94e7ec265c666763f69f861cb4518760979d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate-tempest-plugin/commit/e8592122f9fb3c4b0efacb64eae4883e0864c699"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate-tempest-plugin/commit/e8592122f9fb3c4b0efacb64eae4883e0864c699"}]},"branch":"refs/heads/master","description":"Edit commit message"},"14afec8c8e117ce353e0195d8f52330fd255856c":{"kind":"NO_CODE_CHANGE","_number":4,"created":"2025-04-08 12:56:04.000000000","uploader":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"ref":"refs/changes/55/946555/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate-tempest-plugin","ref":"refs/changes/55/946555/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate-tempest-plugin refs/changes/55/946555/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate-tempest-plugin refs/changes/55/946555/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate-tempest-plugin refs/changes/55/946555/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate-tempest-plugin refs/changes/55/946555/4"}}},"commit":{"parents":[{"commit":"71e4c9616da041d0d3a124fa17d885678706fa2f","subject":"Adapt ptr record api tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate-tempest-plugin/commit/71e4c9616da041d0d3a124fa17d885678706fa2f"}]}],"author":{"name":"Omer","email":"oschwart@redhat.com","date":"2025-04-07 13:02:16.000000000","tz":120},"committer":{"name":"Omer Schwartz","email":"oschwart@redhat.com","date":"2025-04-08 12:56:04.000000000","tz":0},"subject":"Uncomment test_show_floatingip_ptr_impersonate_another_project","message":"Uncomment test_show_floatingip_ptr_impersonate_another_project\n\nAfter https://review.opendev.org/c/openstack/designate/+/927373 is\nmerged and backported, we could uncomment this test whose behavior is changed by it.\n\nDepends-On: https://review.opendev.org/c/openstack/designate/+/927373\nChange-Id: Ib95e94e7ec265c666763f69f861cb4518760979d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate-tempest-plugin/commit/14afec8c8e117ce353e0195d8f52330fd255856c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate-tempest-plugin/commit/14afec8c8e117ce353e0195d8f52330fd255856c"}]},"branch":"refs/heads/master","description":"Edit commit message"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY"},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{"label:Review-Priority\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":["label:Workflow\u003dMIN"],"failing_atoms":["label:Workflow\u003dMAX"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
