)]}'
{"id":"openstack%2Fdesignate~1004213","triplet_id":"openstack%2Fdesignate~master~I7ab254613d002753c35e3cb147f67cd8e24802b1","project":"openstack/designate","branch":"master","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2026-09-18 16:47:40.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I7ab254613d002753c35e3cb147f67cd8e24802b1","subject":"tox: Drop redundant passenv","status":"MERGED","created":"2026-09-05 18:05:04.000000000","updated":"2026-09-18 16:48:59.000000000","submitted":"2026-09-18 16:47:40.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"1004213","meta_rev_id":"3950866944edd878744b7a483933ac141e04c871","_number":1004213,"virtual_id_number":1004213,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-09-18 16:47:40.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"all":[{"value":2,"date":"2026-09-18 15:41:25.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"all":[{"value":1,"date":"2026-09-18 15:41:25.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-09-05 19:33:42.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-09-18 15:41:25.000000000","updated_by":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"reviewer":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"state":"REVIEWER"}],"messages":[{"id":"2e4741527a4e9dff3b8027353138a8bc9085c378","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-09-05 18:05:04.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"0aae6d32f4ced6b16ef98bdb9fbc385ecb6d84e1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-09-05 18:12:26.000000000","message":"Patch Set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"7b04679203ad6aa6b5da11f00dc9d83be63eeaaf","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-09-05 18:16:06.000000000","message":"Patch Set 3: Commit message was updated.","accounts_in_message":[],"_revision_number":3},{"id":"293a64d158dab594aeba0637cda55de63c37b2d9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-05 19:33:42.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/8f3657f1afa74206ab9f48cd9118d511\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/d8a8beb4a3f84d07b543c0d4ddf47cde : SUCCESS in 4m 55s\n- designate-bind9-core https://zuul.opendev.org/t/openstack/build/f504348772df44269125019cc422198f : FAILURE in 41m 15s\n- designate-bind9-with-keystone-default-roles https://zuul.opendev.org/t/openstack/build/1606a2a29c6a4d739dd0b2faef5a301e : SUCCESS in 42m 15s\n- designate-pdns4-core https://zuul.opendev.org/t/openstack/build/6c86b44c8fa145f3881a49f258ec01bc : SUCCESS in 44m 39s\n- designate-bind9-multipool https://zuul.opendev.org/t/openstack/build/197c67a6c72249dda77a738eb3a90e8c : SUCCESS in 23m 36s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/e3f7217b428d4797b16b6bb5be221359 : SUCCESS in 55m 20s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/fee74f2d54dd4b38ae0aa5db64ef73f5 : SUCCESS in 1h 02m 51s\n- designate-grenade-bind9-skip-level https://zuul.opendev.org/t/openstack/build/bbd6cbaa2f054147b3e92995b4b44b9b : SUCCESS in 1h 06m 17s\n- designate-grenade-pdns4-skip-level https://zuul.opendev.org/t/openstack/build/86f37ec93983403284fe7007b94678b0 : SUCCESS in 1h 05m 52s\n- designate-bind9-core-catalog-zones https://zuul.opendev.org/t/openstack/build/a9032dab8e274bff9455e058fa96745b : SUCCESS in 53m 47s (non-voting)\n- designate-only-ipv6-pdns4 https://zuul.opendev.org/t/openstack/build/0a6f34d3a8274e42920575a56b6f4998 : SUCCESS in 47m 41s\n- designate-only-ipv6-bind9 https://zuul.opendev.org/t/openstack/build/2042057b95de4580a7eab4523586029d : SUCCESS in 41m 42s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1508ef6ae3b0473eb3f993b73d68aecc : SUCCESS in 3m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a91aaf0f88f84cb1aad6078a2cabc944 : SUCCESS in 2m 34s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/0e1b60c33ab14e4aa10e09a63241e1ca : SUCCESS in 5m 42s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/d3f62c3331dd450c972f9f79100f2ce3 : SUCCESS in 4m 33s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/c93bb36352dd4381b47691bdc28d00e6 : SUCCESS in 9m 06s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a29acb9b237d4b53b8031fa68089f798 : SUCCESS in 5m 03s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/26d33003bdde4d069025e98bf99cc058 : SUCCESS in 7m 15s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/2649185def0b40beb0305fe6f2ee544e : SUCCESS in 3m 24s","accounts_in_message":[],"_revision_number":3},{"id":"53a11b602c6d0dc81ac7089f8d0e569ec68ca5e4","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-09-06 03:37:24.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"655f93182834d64f071585f833c421ff301f2f1f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-06 04:51:19.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/6a82ab6f75594e2d80152efb0ab6eaeb\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/4b8ab7e9b9384df088b3df465b828831 : SUCCESS in 3m 19s\n- designate-bind9-core https://zuul.opendev.org/t/openstack/build/1063e0770a9845dead6f454d7cb1d2df : SUCCESS in 42m 23s\n- designate-bind9-with-keystone-default-roles https://zuul.opendev.org/t/openstack/build/e0ff46c8fdc3497187b6e402ee50e6ea : SUCCESS in 36m 15s\n- designate-pdns4-core https://zuul.opendev.org/t/openstack/build/56487a10dfc44ffab5d813687dfaefb8 : FAILURE in 50m 10s\n- designate-bind9-multipool https://zuul.opendev.org/t/openstack/build/d0384dd8caaf4b9599e134c06f3e08cb : SUCCESS in 17m 55s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/b024bc1496d642c4b24d19ecb12dfe7b : SUCCESS in 44m 01s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/170d8494aa0f4b85b14f1cd3f6f6744a : SUCCESS in 1h 11m 23s\n- designate-grenade-bind9-skip-level https://zuul.opendev.org/t/openstack/build/82a0229bb23242c69420d77caeceba51 : SUCCESS in 55m 00s\n- designate-grenade-pdns4-skip-level https://zuul.opendev.org/t/openstack/build/e22ef6cd24b14e5f98d148ca29314d4e : SUCCESS in 1h 10m 14s\n- designate-bind9-core-catalog-zones https://zuul.opendev.org/t/openstack/build/181d7a1ce0d64ab0a86e7af6e3e2e55a : SUCCESS in 42m 39s (non-voting)\n- designate-only-ipv6-pdns4 https://zuul.opendev.org/t/openstack/build/15b1718416a445cabf303c7ae3fa88d1 : SUCCESS in 46m 24s\n- designate-only-ipv6-bind9 https://zuul.opendev.org/t/openstack/build/01cd915bda4c4d0184c61a6cb978a933 : SUCCESS in 50m 06s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/3a58189b7d9b41179903baccaa78901b : SUCCESS in 3m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8eee92d683e94f1e91b27446c079bf91 : SUCCESS in 5m 19s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/bc979afd8d1a4b418968730c21786885 : SUCCESS in 3m 14s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/ef45da1d82744a57a9b912945d6bd18d : SUCCESS in 3m 08s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/8f01eb46c0a74d3bb03df10fc7765f8f : SUCCESS in 9m 47s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c2ac27b98e6c48b98ef4469e69c98c23 : SUCCESS in 8m 41s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/80c33023ce2b489bb01c7c11be031d2e : SUCCESS in 6m 34s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/74490147b3394ee0b2f6316cf32f9b93 : SUCCESS in 3m 09s","accounts_in_message":[],"_revision_number":3},{"id":"9161be8d9b308e55722ba0fa37ad791077d21aaa","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-09-06 12:46:20.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"a98d753f75c30799e2be4e3639a2c602492d8880","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-06 14:08:21.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3936b3698ee94a3cbeacec63f4732011\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/a8a7967d2f5a4757997a56875f287ed0 : SUCCESS in 3m 20s\n- designate-bind9-core https://zuul.opendev.org/t/openstack/build/1950c2cbfa874fbb910cd3b120b01c20 : SUCCESS in 35m 58s\n- designate-bind9-with-keystone-default-roles https://zuul.opendev.org/t/openstack/build/0021b803e61b48b795231e62a6ef852a : SUCCESS in 42m 25s\n- designate-pdns4-core https://zuul.opendev.org/t/openstack/build/0cee1ca9d4d84adea24e6d069394098a : SUCCESS in 39m 10s\n- designate-bind9-multipool https://zuul.opendev.org/t/openstack/build/7f7eff46cdaf4d4ba3aa363cb2774c8e : SUCCESS in 31m 46s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/70f13ddee0ba41bfb5ea65dd2a561a3d : SUCCESS in 47m 09s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/bfcef4c1c2e34713925eacf25bdab07a : SUCCESS in 1h 07m 37s\n- designate-grenade-bind9-skip-level https://zuul.opendev.org/t/openstack/build/4522f2cdddef4b00857ad20793794548 : SUCCESS in 1h 06m 33s\n- designate-grenade-pdns4-skip-level https://zuul.opendev.org/t/openstack/build/edf1b16ffe8b4b1ab035ee858ba6a071 : SUCCESS in 1h 12m 31s\n- designate-bind9-core-catalog-zones https://zuul.opendev.org/t/openstack/build/f854d6ccf66a4fb6b39a4d56ab5e7d54 : FAILURE in 47m 15s (non-voting)\n- designate-only-ipv6-pdns4 https://zuul.opendev.org/t/openstack/build/aea351900bf340d6bb87f6f39602dd7a : SUCCESS in 41m 55s\n- designate-only-ipv6-bind9 https://zuul.opendev.org/t/openstack/build/946be703dd0545d592a5c1e619e4f0ed : SUCCESS in 41m 50s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f76dbbecf69948da8bea4ee36825a805 : SUCCESS in 5m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/66eeceb0789241fcbc75e0fdddbf77af : SUCCESS in 2m 38s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/31fce07067dd4d958bd437d95cc94639 : SUCCESS in 4m 22s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/5d69210cbcab4bfbb0d4723329883336 : SUCCESS in 5m 00s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/146ca73fb6a540a6951150de9a0e7d61 : SUCCESS in 5m 33s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/656c7f1cf2e1481483eb4779cf502d56 : SUCCESS in 9m 22s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/65d267bf93914f8691154bdc4451e915 : SUCCESS in 6m 14s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/2fd7812ded72450eb2a66a6fe2baa71c : SUCCESS in 3m 31s","accounts_in_message":[],"_revision_number":3},{"id":"fd6da38fad4786db553974c5868e2582416361e9","author":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"date":"2026-09-18 15:41:25.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"981ce11b4b9b67ed65c6482b49fb6e98212774c6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-18 15:42:36.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"d3cf3c8a1e915cd1b67069c4bec63c37add9eebd","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-18 16:47:40.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9ce281699311483c925cc19a60098dac\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/019b7a729fcf4247a80bcfb572b9fdb8 : SUCCESS in 4m 21s\n- designate-bind9-core https://zuul.opendev.org/t/openstack/build/8024ae38e8ce49f1b333cd736183c4ea : SUCCESS in 39m 34s\n- designate-bind9-with-keystone-default-roles https://zuul.opendev.org/t/openstack/build/e38dbccce2a240fdb950eedeef7989aa : SUCCESS in 46m 44s\n- designate-pdns4-core https://zuul.opendev.org/t/openstack/build/b0dbf81ae3fc4966862b10daa24bbd5e : SUCCESS in 42m 32s\n- designate-bind9-multipool https://zuul.opendev.org/t/openstack/build/9572dab395ad4b79a22df4ebcf9fe934 : SUCCESS in 31m 51s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/ae7c2eebed0c4249b221f2e0e69c5f01 : SUCCESS in 58m 57s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/185ef8ada824435f9e3c4f491d21f339 : SUCCESS in 56m 42s\n- designate-grenade-pdns4-skip-level https://zuul.opendev.org/t/openstack/build/524a3926e44a44c2bb117e1e7f511277 : SUCCESS in 45m 29s\n- designate-grenade-bind9-skip-level https://zuul.opendev.org/t/openstack/build/81cc2c9a29aa47448a1663460b7332ae : SUCCESS in 50m 21s\n- designate-only-ipv6-pdns4 https://zuul.opendev.org/t/openstack/build/0826e1f470d64bd9853af0a9652f9f53 : SUCCESS in 43m 31s\n- designate-only-ipv6-bind9 https://zuul.opendev.org/t/openstack/build/505e0c9cbc5445bfafa08981fa0bf112 : SUCCESS in 46m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5f4468428ea741c69390688ea3e9ea37 : SUCCESS in 3m 14s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/fb652eba6493419cb6f71a280493dc0f : SUCCESS in 3m 13s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/8c29bbf28da44a9382bb35b92417b2bd : SUCCESS in 4m 47s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bc5cb2aceaa043a693cb02c2bd2dfc1a : SUCCESS in 9m 39s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f05d75337e6d49a6973fe4f808324323 : SUCCESS in 3m 12s","accounts_in_message":[],"_revision_number":3},{"id":"5f38947fd2089b5db3f42f5baf9ecdec9222f0a4","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-18 16:47:40.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"3950866944edd878744b7a483933ac141e04c871","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-18 16:48:59.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f5506e66cd02423aba679a2d40138bb2\n\n- promote-api-ref https://zuul.opendev.org/t/openstack/build/0a49704f9bb848efb95a81c05c1c605a : SUCCESS in 53s\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c1c11328ec4a469e89e723ff693d78ab : SUCCESS in 1m 00s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/010fe913c46348a39a15235235215feb : SUCCESS in 51s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"07ea6600f0df0066126dd181d4538fc6387ceb3a","revisions":{"d482df5946c811f6ef701f1d3d0812ecc7e7caa2":{"kind":"REWORK","_number":1,"created":"2026-09-05 18:05:04.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/13/1004213/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/13/1004213/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/13/1004213/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/13/1004213/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/13/1004213/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/13/1004213/1"}}},"commit":{"parents":[{"commit":"960b3b7f2bcda207100c16868f3272b77b0e9675","subject":"Merge \"tox: Drop redundant override of install_command\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/960b3b7f2bcda207100c16868f3272b77b0e9675"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-09-05 18:04:21.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-09-05 18:04:21.000000000","tz":540},"subject":"tox: Drop redundant passenv","message":"tox: Drop redundant passenv\n\nSince 4.0.0[1], tox passes basic environment variables automatically\nand we don\u0027t have add these explicitly.\n\ncf. https: //tox.wiki/en/latest/reference/config.html#passenv\n\nSome variables are always passed through to ensure the basic\nfunctionality of standard library functions or tooling like pip.\nThis is also not case sensitive on all platforms except Windows.\n\n[1] https://github.com/tox-dev/tox/commit/afb3b395dd\n\nChange-Id: I7ab254613d002753c35e3cb147f67cd8e24802b1\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/d482df5946c811f6ef701f1d3d0812ecc7e7caa2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/d482df5946c811f6ef701f1d3d0812ecc7e7caa2"}]},"branch":"refs/heads/master"},"de20bfb514360335935fb96b87485f51da2f0e10":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2026-09-05 18:12:26.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/13/1004213/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/13/1004213/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/13/1004213/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/13/1004213/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/13/1004213/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/13/1004213/2"}}},"commit":{"parents":[{"commit":"960b3b7f2bcda207100c16868f3272b77b0e9675","subject":"Merge \"tox: Drop redundant override of install_command\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/960b3b7f2bcda207100c16868f3272b77b0e9675"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-09-05 18:04:21.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-09-05 18:12:26.000000000","tz":0},"subject":"tox: Drop redundant passenv","message":"tox: Drop redundant passenv\n\nSince 4.0.0[1], tox passes basic environment variables automatically\nand we don\u0027t have add these explicitly.\n\ncf. https://tox.wiki/en/latest/reference/config.html#passenv\n\nSome variables are always passed through to ensure the basic\nfunctionality of standard library functions or tooling like pip.\nThis is also not case sensitive on all platforms except Windows.\n\n[1] https://github.com/tox-dev/tox/commit/afb3b395dd\n\nChange-Id: I7ab254613d002753c35e3cb147f67cd8e24802b1\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/de20bfb514360335935fb96b87485f51da2f0e10"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/de20bfb514360335935fb96b87485f51da2f0e10"}]},"branch":"refs/heads/master","description":"Edit commit message"},"07ea6600f0df0066126dd181d4538fc6387ceb3a":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2026-09-05 18:16:06.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/13/1004213/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/13/1004213/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/13/1004213/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/13/1004213/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/13/1004213/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/13/1004213/3"}}},"commit":{"parents":[{"commit":"960b3b7f2bcda207100c16868f3272b77b0e9675","subject":"Merge \"tox: Drop redundant override of install_command\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/960b3b7f2bcda207100c16868f3272b77b0e9675"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-09-05 18:04:21.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-09-05 18:16:06.000000000","tz":0},"subject":"tox: Drop redundant passenv","message":"tox: Drop redundant passenv\n\nSince 4.0.0[1], tox passes basic environment variables automatically\nand we don\u0027t have to add these explicitly.\n\ncf. https://tox.wiki/en/latest/reference/config.html#passenv\n\nSome variables are always passed through to ensure the basic\nfunctionality of standard library functions or tooling like pip.\nThis is also not case sensitive on all platforms except Windows.\n\n[1] https://github.com/tox-dev/tox/commit/afb3b395dd\n\nChange-Id: I7ab254613d002753c35e3cb147f67cd8e24802b1\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/07ea6600f0df0066126dd181d4538fc6387ceb3a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/07ea6600f0df0066126dd181d4538fc6387ceb3a"}]},"branch":"refs/heads/master","description":"Edit commit message"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{"label:Review-Priority\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
