)]}'
{"id":"openstack%2Fdesignate~1004309","triplet_id":"openstack%2Fdesignate~master~I6427665146146a3ba08155a7e68af6a67e54789e","project":"openstack/designate","branch":"master","attention_set":{"31664":{"account":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"last_update":"2026-09-18 16:48:32.000000000","reason":"Vote got outdated and was removed: Code-Review-1"},"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2026-09-18 18:06:06.000000000","reason":"A robot voted negatively on a label"}},"removed_from_attention_set":{},"hashtags":[],"change_id":"I6427665146146a3ba08155a7e68af6a67e54789e","subject":"Extend irrelevant-files for grenade jobs","status":"NEW","created":"2026-09-06 14:23:55.000000000","updated":"2026-09-18 18:06:06.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":true,"submittable":false,"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"bdf4b34198f0b520cdd2874414cd556e26244192","_number":1004309,"virtual_id_number":1004309,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2026-09-18 18:06:06.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-09-06 15:44:04.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-09-18 15:45:42.000000000","updated_by":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"reviewer":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"state":"REVIEWER"}],"messages":[{"id":"e1127bf87418dfcee481c4023893f1e40e038f96","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-09-06 14:23:55.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"0216d122df75a64ad3662efb91233b72fc6f59b6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-06 15:44:04.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/57a259908e874235b8faa6271cd39e04\n\n- designate-bind9-core https://zuul.opendev.org/t/openstack/build/ff9122482c1d46a48f865e0a956c29f4 : SUCCESS in 43m 22s\n- designate-bind9-with-keystone-default-roles https://zuul.opendev.org/t/openstack/build/809da49198e74dc29634f415a340ff4c : SUCCESS in 31m 44s\n- designate-pdns4-core https://zuul.opendev.org/t/openstack/build/7b9ed41c3d354edaaae6b760ccd7c8dd : SUCCESS in 45m 25s\n- designate-bind9-multipool https://zuul.opendev.org/t/openstack/build/14c85c6574d14a129de8cb0cb82b1d02 : SUCCESS in 17m 47s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/31af67893ec94c4d91442e02db8f8d6a : SUCCESS in 58m 32s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/769ea8402a8b4af690d29f97db675180 : SUCCESS in 1h 08m 04s\n- designate-grenade-bind9-skip-level https://zuul.opendev.org/t/openstack/build/0f61ad7251e14aeb99de29b59fd00ee4 : SUCCESS in 1h 03m 04s\n- designate-grenade-pdns4-skip-level https://zuul.opendev.org/t/openstack/build/9b88555c77df4884a802c9b6c07c84ce : SUCCESS in 57m 12s\n- designate-bind9-core-catalog-zones https://zuul.opendev.org/t/openstack/build/f40dc8576fb045d090554c3e9cdc7917 : FAILURE in 34m 47s (non-voting)\n- designate-only-ipv6-pdns4 https://zuul.opendev.org/t/openstack/build/05315c390f40440cb910f02c3ebe64ef : SUCCESS in 42m 33s\n- designate-only-ipv6-bind9 https://zuul.opendev.org/t/openstack/build/638434e3593341ff897076f1aef50ac8 : SUCCESS in 49m 05s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/6defc887f08f4a228d4fd6bd4e5fd1f4 : SUCCESS in 4m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/57b0d617e419450aafbcebc406092485 : SUCCESS in 1m 58s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/5f7014892a2e44e5b1d0bd20431136ba : SUCCESS in 4m 48s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/aab6ff7098ec476596da97392b16c094 : SUCCESS in 4m 54s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/2f72b1f3bde64b8e8379853dd9ba9054 : SUCCESS in 7m 35s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/48ea3fd1ce674b5aa6fe5e8a496a6984 : SUCCESS in 9m 47s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e3e7c3a642e84a2ba6c74a551f9246fe : SUCCESS in 9m 45s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/69f6014b93e344e2b94bb029da2c43a4 : SUCCESS in 39m 56s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/62a3842e215f4c2188f41d59c0df8cbf : SUCCESS in 4m 33s","accounts_in_message":[],"_revision_number":1},{"id":"2d427d5f7e728ae87f838e42ad7c1890a0b44869","author":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"date":"2026-09-18 15:45:42.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"3fd363403a1cbf7f18c2e61a0bb6d3990df3452a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-09-18 16:48:32.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":2},{"id":"9d4d4d91fbd8e12df63a3402354ad6d6b930483c","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-09-18 16:48:51.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"bdf4b34198f0b520cdd2874414cd556e26244192","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-18 18:06:06.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/da7cbef4c8a94853a508543e2bab082e\n\n- designate-bind9-core https://zuul.opendev.org/t/openstack/build/93bfca8ccb0e4c39949f6174c7895bb7 : FAILURE in 49m 37s\n- designate-bind9-with-keystone-default-roles https://zuul.opendev.org/t/openstack/build/c560ffaa9c5f4b478b0588c750e26d41 : SUCCESS in 41m 50s\n- designate-pdns4-core https://zuul.opendev.org/t/openstack/build/e636a2c610224a4a81203577076f4960 : SUCCESS in 46m 41s\n- designate-bind9-multipool https://zuul.opendev.org/t/openstack/build/05814309bebc44d1ad279293c0658323 : SUCCESS in 36m 31s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/5fd0e221c4364963a473f6f933f18de0 : SUCCESS in 49m 05s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/df6ada114efe44ae89caa74730cabb90 : SUCCESS in 1h 02m 10s\n- designate-grenade-bind9-skip-level https://zuul.opendev.org/t/openstack/build/94a3472b27aa442d8fd37e1cd2310da9 : SUCCESS in 1h 09m 56s\n- designate-grenade-pdns4-skip-level https://zuul.opendev.org/t/openstack/build/e797588271bc476da5cf1a4eb80020b0 : SUCCESS in 59m 28s\n- designate-bind9-core-catalog-zones https://zuul.opendev.org/t/openstack/build/5ec01f9060f1407bb89cc83890568349 : SUCCESS in 44m 15s (non-voting)\n- designate-only-ipv6-pdns4 https://zuul.opendev.org/t/openstack/build/fb13f7a1bfe64c3a903067e93497da1c : SUCCESS in 50m 45s\n- designate-only-ipv6-bind9 https://zuul.opendev.org/t/openstack/build/6f74ba84673849c88de3282bc2822408 : SUCCESS in 41m 53s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/48ea5522ba3e495487f00d7639a97ab8 : SUCCESS in 6m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/85e7f242e6014667b03c082939029bff : SUCCESS in 2m 20s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/4dc15964c8254eacb4f4c4c86f0cac53 : SUCCESS in 5m 12s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/10ef71928b21408996fd87868223d11a : SUCCESS in 5m 27s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/0a2a2845a26b461a999c3107176e173b : SUCCESS in 8m 02s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/773631720d8f443d8f5af3ae2021c828 : SUCCESS in 9m 52s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2a698d99952f4439ac09dc08d8bcf93a : SUCCESS in 6m 59s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/b1b7d75dc5a749c5a6d7a9abdb1fac0f : SUCCESS in 31m 19s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/30853de6934741169f99ac163563797d : SUCCESS in 7m 44s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"6d11a63c1fc768e1e3f1f01b03f86bf4c4af3648","revisions":{"e88535d63e1d6f58fd11a7d15f3ab552540f160f":{"kind":"REWORK","_number":1,"created":"2026-09-06 14:23:55.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/09/1004309/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/09/1004309/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/09/1004309/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/09/1004309/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/09/1004309/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/09/1004309/1"}}},"commit":{"parents":[{"commit":"960b3b7f2bcda207100c16868f3272b77b0e9675","subject":"Merge \"tox: Drop redundant override of install_command\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/960b3b7f2bcda207100c16868f3272b77b0e9675"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-09-06 14:23:07.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-09-06 14:23:47.000000000","tz":540},"subject":"Extend irrelevant-files for grenade jobs","message":"Extend irrelevant-files for grenade jobs\n\n... to reduce unnecessary triggers.\n\nChange-Id: I6427665146146a3ba08155a7e68af6a67e54789e\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/e88535d63e1d6f58fd11a7d15f3ab552540f160f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/e88535d63e1d6f58fd11a7d15f3ab552540f160f"}]},"branch":"refs/heads/master"},"6d11a63c1fc768e1e3f1f01b03f86bf4c4af3648":{"kind":"REWORK","_number":2,"created":"2026-09-18 16:48:32.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/09/1004309/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/09/1004309/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/09/1004309/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/09/1004309/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/09/1004309/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/09/1004309/2"}}},"commit":{"parents":[{"commit":"960b3b7f2bcda207100c16868f3272b77b0e9675","subject":"Merge \"tox: Drop redundant override of install_command\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/960b3b7f2bcda207100c16868f3272b77b0e9675"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-09-06 14:23:07.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-09-18 16:48:20.000000000","tz":540},"subject":"Extend irrelevant-files for grenade jobs","message":"Extend irrelevant-files for grenade jobs\n\n... to reduce unnecessary triggers.\n\nChange-Id: I6427665146146a3ba08155a7e68af6a67e54789e\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/6d11a63c1fc768e1e3f1f01b03f86bf4c4af3648"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/6d11a63c1fc768e1e3f1f01b03f86bf4c4af3648"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY"},{"label":"Workflow","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{"label:Review-Priority\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
