)]}'
{"id":"openstack%2Fdesignate~108391","triplet_id":"openstack%2Fdesignate~master~I7a923dc14969e6ed01790f12b5c5a18c62c08f2f","project":"openstack/designate","branch":"master","topic":"enable_h306","hashtags":[],"change_id":"I7a923dc14969e6ed01790f12b5c5a18c62c08f2f","subject":"Enable hacking check H306","status":"ABANDONED","created":"2014-07-21 14:47:29.000000000","updated":"2014-07-30 06:00:19.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"4cd48fbcd923c1dac7f1a5df744f40c6e7cca6d7","_number":108391,"virtual_id_number":108391,"owner":{"_account_id":167,"name":"Christian Berendt","email":"berendt@osism.tech","username":"berendt"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"all":[{"_account_id":8130,"name":"Betsy Luzader","email":"betsy.luzader@rackspace.com","username":"bluzader"},{"value":-1,"date":"2014-07-29 18:43:51.000000000","_account_id":3,"name":"Jenkins","username":"jenkins"},{"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},{"date":"2014-07-23 19:40:04.000000000","_account_id":167,"name":"Christian Berendt","email":"berendt@osism.tech","username":"berendt"},{"_account_id":11626,"name":"Jordan Cazamias","email":"jordan.cazamias@rackspace.com","username":"jord7642"},{"_account_id":8094,"name":"Vinod Mangalpally","email":"mrvinod@outlook.com","username":"vinodmr"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"rejected":{"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":8130,"name":"Betsy Luzader","email":"betsy.luzader@rackspace.com","username":"bluzader"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":-2,"date":"2014-07-29 18:43:49.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":167,"name":"Christian Berendt","email":"berendt@osism.tech","username":"berendt"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":11626,"name":"Jordan Cazamias","email":"jordan.cazamias@rackspace.com","username":"jord7642"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":8094,"name":"Vinod Mangalpally","email":"mrvinod@outlook.com","username":"vinodmr"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":8130,"name":"Betsy Luzader","email":"betsy.luzader@rackspace.com","username":"bluzader"},{"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},{"value":0,"permitted_voting_range":{"min":-1,"max":0},"_account_id":167,"name":"Christian Berendt","email":"berendt@osism.tech","username":"berendt"},{"_account_id":11626,"name":"Jordan Cazamias","email":"jordan.cazamias@rackspace.com","username":"jord7642"},{"_account_id":8094,"name":"Vinod Mangalpally","email":"mrvinod@outlook.com","username":"vinodmr"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":8130,"name":"Betsy Luzader","email":"betsy.luzader@rackspace.com","username":"bluzader"},{"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},{"_account_id":167,"name":"Christian Berendt","email":"berendt@osism.tech","username":"berendt"},{"_account_id":11626,"name":"Jordan Cazamias","email":"jordan.cazamias@rackspace.com","username":"jord7642"},{"_account_id":8094,"name":"Vinod Mangalpally","email":"mrvinod@outlook.com","username":"vinodmr"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":3,"name":"Jenkins","username":"jenkins"},{"_account_id":167,"name":"Christian Berendt","email":"berendt@osism.tech","username":"berendt"},{"_account_id":8094,"name":"Vinod Mangalpally","email":"mrvinod@outlook.com","username":"vinodmr"},{"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},{"_account_id":8130,"name":"Betsy Luzader","email":"betsy.luzader@rackspace.com","username":"bluzader"},{"_account_id":11626,"name":"Jordan Cazamias","email":"jordan.cazamias@rackspace.com","username":"jord7642"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2014-07-21 15:39:52.000000000","updated_by":{"_account_id":11626,"name":"Jordan Cazamias","email":"jordan.cazamias@rackspace.com","username":"jord7642"},"reviewer":{"_account_id":11626,"name":"Jordan Cazamias","email":"jordan.cazamias@rackspace.com","username":"jord7642"},"state":"REVIEWER"},{"updated":"2014-07-23 17:35:43.000000000","updated_by":{"_account_id":8094,"name":"Vinod Mangalpally","email":"mrvinod@outlook.com","username":"vinodmr"},"reviewer":{"_account_id":8094,"name":"Vinod Mangalpally","email":"mrvinod@outlook.com","username":"vinodmr"},"state":"REVIEWER"},{"updated":"2014-07-23 18:56:33.000000000","updated_by":{"_account_id":8130,"name":"Betsy Luzader","email":"betsy.luzader@rackspace.com","username":"bluzader"},"reviewer":{"_account_id":8130,"name":"Betsy Luzader","email":"betsy.luzader@rackspace.com","username":"bluzader"},"state":"REVIEWER"},{"updated":"2014-07-29 18:43:49.000000000","updated_by":{"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},"reviewer":{"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},"state":"REVIEWER"},{"updated":"2014-07-29 18:43:51.000000000","updated_by":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"reviewer":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"state":"REVIEWER"}],"messages":[{"id":"4198baf20bb6aff27f73025408158a55a2f730d1","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2014-07-21 14:47:29.000000000","message":"Patch Set 1:\n\nStarting check jobs.\nhttp://status.openstack.org/zuul/","accounts_in_message":[],"_revision_number":1},{"id":"61330db1477d9f7e14d1e664d0d510d3f8c1b182","author":{"_account_id":167,"name":"Christian Berendt","email":"berendt@osism.tech","username":"berendt"},"date":"2014-07-21 14:47:29.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"4b7beeeef69772f3153687cb7b68df7e61626c5b","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2014-07-21 15:29:36.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded.\n\n- gate-designate-pep8 http://logs.openstack.org/91/108391/1/check/gate-designate-pep8/2d87f61 : SUCCESS in 2m 40s\n- gate-designate-docs http://docs-draft.openstack.org/91/108391/1/check/gate-designate-docs/4d19385/doc/build/html/ : SUCCESS in 2m 07s\n- gate-designate-python26 http://logs.openstack.org/91/108391/1/check/gate-designate-python26/ac3381e : SUCCESS in 2m 46s\n- gate-designate-python27 http://logs.openstack.org/91/108391/1/check/gate-designate-python27/b1b8acf : SUCCESS in 1m 37s\n- gate-designate-devstack-dsvm http://logs.openstack.org/91/108391/1/check/gate-designate-devstack-dsvm/b3deb23 : SUCCESS in 14m 50s","accounts_in_message":[],"_revision_number":1},{"id":"e975c61f034fabf66e7bee4b53c2a9d1efed4c4b","author":{"_account_id":11626,"name":"Jordan Cazamias","email":"jordan.cazamias@rackspace.com","username":"jord7642"},"date":"2014-07-21 15:39:52.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"657541e0fcc0014626b343f7534a54bad0332c73","author":{"_account_id":8094,"name":"Vinod Mangalpally","email":"mrvinod@outlook.com","username":"vinodmr"},"date":"2014-07-23 17:35:43.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"bee07990aba84cb7d8f153cd3a4f1b18547db49a","author":{"_account_id":8130,"name":"Betsy Luzader","email":"betsy.luzader@rackspace.com","username":"bluzader"},"date":"2014-07-23 18:56:33.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"22e322e966cc40327411f8bffd583a3afedc6fab","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2014-07-23 19:05:50.000000000","message":"Patch Set 1: -Verified\n\nStarting check jobs.\nhttp://status.openstack.org/zuul/","accounts_in_message":[],"_revision_number":1},{"id":"036ffe7d0d28bdbed194d80c62beadb4b4efd711","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2014-07-23 19:06:42.000000000","message":"Patch Set 1: Verified-1\n\nMerge Failed.\n\nThis change was unable to be automatically merged with the current state of the repository. Please rebase your change and upload a new patchset.","accounts_in_message":[],"_revision_number":1},{"id":"196500d5c7faa73fe9e602a1d80c2a7a2c7c8f23","author":{"_account_id":167,"name":"Christian Berendt","email":"berendt@osism.tech","username":"berendt"},"date":"2014-07-23 19:23:32.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"def10b902335160292d81097ee9d84963fe2ce70","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2014-07-23 19:23:33.000000000","message":"Patch Set 2:\n\nStarting check jobs.\nhttp://status.openstack.org/zuul/","accounts_in_message":[],"_revision_number":2},{"id":"d192a29994a1fa91846144ffc5baacbdc53818ea","author":{"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},"date":"2014-07-23 19:29:52.000000000","message":"Patch Set 2: Code-Review-1\n\nNot sure how I feel about this check - I think we should discuss it before we merge","accounts_in_message":[],"_revision_number":2},{"id":"a68898e6e5749f8d072aab28b8a43116b5eadbed","author":{"_account_id":167,"name":"Christian Berendt","email":"berendt@osism.tech","username":"berendt"},"date":"2014-07-23 19:40:04.000000000","message":"Patch Set 2:\n\nGraham, what is wrong with this check? Most of the other projects are using it.","accounts_in_message":[],"_revision_number":2},{"id":"2cfac688f4f04c0fdb48a6b7f8bd007e76c0ef76","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2014-07-23 23:57:24.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded.\n\n- gate-designate-pep8 http://logs.openstack.org/91/108391/2/check/gate-designate-pep8/9cde793 : SUCCESS in 2m 35s\n- gate-designate-docs http://docs-draft.openstack.org/91/108391/2/check/gate-designate-docs/7a35233/doc/build/html/ : SUCCESS in 3m 16s\n- gate-designate-python26 http://logs.openstack.org/91/108391/2/check/gate-designate-python26/1b1cb46 : SUCCESS in 3m 40s\n- gate-designate-python27 http://logs.openstack.org/91/108391/2/check/gate-designate-python27/57fb3dd : SUCCESS in 2m 33s\n- gate-designate-devstack-dsvm http://logs.openstack.org/91/108391/2/check/gate-designate-devstack-dsvm/3f9fafb : SUCCESS in 22m 17s","accounts_in_message":[],"_revision_number":2},{"id":"be57618567d999a40202e227028a2d1970a9fd48","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2014-07-29 18:43:49.000000000","message":"Patch Set 2: -Verified\n\nStarting check jobs.\nhttp://status.openstack.org/zuul/","accounts_in_message":[],"_revision_number":2},{"id":"4e84d6525777476232ee767cf8510829ad91028f","author":{"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},"date":"2014-07-29 18:43:49.000000000","message":"Patch Set 2: Code-Review-2\n\nAfter discussion with some of the core group at the mid cycle - we have decided that we are OK not having this check.\n\nOne of the main issues is that we disagree with the sorting - \n\n\"openstack\" should be before\n\"opentstack.common\" ,\n\nbut with H306, it is not always like that.","accounts_in_message":[],"_revision_number":2},{"id":"0974738b4bf9ada890da852ccae4c30770469af3","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2014-07-29 18:43:51.000000000","message":"Patch Set 2: Verified-1\n\nMerge Failed.\n\nThis change was unable to be automatically merged with the current state of the repository. Please rebase your change and upload a new patchset.","accounts_in_message":[],"_revision_number":2},{"id":"e89887bc4daaddfc89d731c0bb506c0e659c0da6","author":{"_account_id":167,"name":"Christian Berendt","email":"berendt@osism.tech","username":"berendt"},"date":"2014-07-30 06:00:19.000000000","message":"Abandoned\n\nH306 should not be enabled. Maybe we should add a note into the tox.ini about that.","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"6e42231eb1ac2f75323e9ecb48426e3b8b39a9d9","revisions":{"03a8af9728d9eef23dd85d4a213f5f6b37195269":{"kind":"REWORK","_number":1,"created":"2014-07-21 14:47:29.000000000","uploader":{"_account_id":167,"name":"Christian Berendt","email":"berendt@osism.tech","username":"berendt"},"ref":"refs/changes/91/108391/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/91/108391/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/91/108391/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/91/108391/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/91/108391/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/91/108391/1"}}},"commit":{"parents":[{"commit":"0e2293d1073926d6ef011a967e99498a0f8aecda","subject":"Merge \"Cleaning up index.rst file\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/0e2293d1073926d6ef011a967e99498a0f8aecda"}]}],"author":{"name":"Christian Berendt","email":"berendt@b1-systems.de","date":"2014-07-21 14:47:08.000000000","tz":120},"committer":{"name":"Christian Berendt","email":"berendt@b1-systems.de","date":"2014-07-21 14:47:08.000000000","tz":120},"subject":"Enable hacking check H306","message":"Enable hacking check H306\n\nChange-Id: I7a923dc14969e6ed01790f12b5c5a18c62c08f2f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/03a8af9728d9eef23dd85d4a213f5f6b37195269"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/03a8af9728d9eef23dd85d4a213f5f6b37195269"}]},"branch":"refs/heads/master"},"6e42231eb1ac2f75323e9ecb48426e3b8b39a9d9":{"kind":"REWORK","_number":2,"created":"2014-07-23 19:23:32.000000000","uploader":{"_account_id":167,"name":"Christian Berendt","email":"berendt@osism.tech","username":"berendt"},"ref":"refs/changes/91/108391/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/91/108391/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/91/108391/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/91/108391/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/91/108391/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/91/108391/2"}}},"commit":{"parents":[{"commit":"b26f35dc4f1f21a2b8a51b80538143bfaf742285","subject":"Imported Translations from Transifex","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/b26f35dc4f1f21a2b8a51b80538143bfaf742285"}]}],"author":{"name":"Christian Berendt","email":"berendt@b1-systems.de","date":"2014-07-21 14:47:08.000000000","tz":120},"committer":{"name":"Christian Berendt","email":"berendt@b1-systems.de","date":"2014-07-23 19:23:41.000000000","tz":120},"subject":"Enable hacking check H306","message":"Enable hacking check H306\n\nChange-Id: I7a923dc14969e6ed01790f12b5c5a18c62c08f2f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/6e42231eb1ac2f75323e9ecb48426e3b8b39a9d9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/6e42231eb1ac2f75323e9ecb48426e3b8b39a9d9"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
