)]}'
{"id":"openstack%2Fdesignate~196261","triplet_id":"openstack%2Fdesignate~master~I8e5a4b518d9ff1159720a18711052005305806ae","project":"openstack/designate","branch":"master","topic":"openstack/requirements","hashtags":[],"change_id":"I8e5a4b518d9ff1159720a18711052005305806ae","subject":"Updated from global requirements","status":"MERGED","created":"2015-06-26 21:48:36.000000000","updated":"2015-06-29 13:30:57.000000000","submitted":"2015-06-29 13:30:54.000000000","submitter":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"411540016d91e2338bfc4f389832b8d1e296e97c","_number":196261,"virtual_id_number":196261,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"all":[{"value":0,"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},{"value":2,"date":"2015-06-29 13:30:54.000000000","_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":0,"_account_id":741,"name":"Kiall Mac Innes","email":"kiall@macinnes.ie","username":"kiall"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},"all":[{"value":2,"date":"2015-06-29 13:07:09.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},{"value":0,"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":2,"date":"2015-06-27 09:18:10.000000000","_account_id":741,"name":"Kiall Mac Innes","email":"kiall@macinnes.ie","username":"kiall"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},"all":[{"value":1,"date":"2015-06-29 13:07:09.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},{"value":0,"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":0,"_account_id":741,"name":"Kiall Mac Innes","email":"kiall@macinnes.ie","username":"kiall"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},{"value":0,"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":0,"_account_id":741,"name":"Kiall Mac Innes","email":"kiall@macinnes.ie","username":"kiall"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":3,"name":"Jenkins","username":"jenkins"},{"_account_id":741,"name":"Kiall Mac Innes","email":"kiall@macinnes.ie","username":"kiall"},{"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2015-06-27 09:18:10.000000000","updated_by":{"_account_id":741,"name":"Kiall Mac Innes","email":"kiall@macinnes.ie","username":"kiall"},"reviewer":{"_account_id":741,"name":"Kiall Mac Innes","email":"kiall@macinnes.ie","username":"kiall"},"state":"REVIEWER"},{"updated":"2015-06-29 13:07:09.000000000","updated_by":{"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},"reviewer":{"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},"state":"REVIEWER"},{"updated":"2015-06-29 13:30:54.000000000","updated_by":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"reviewer":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"state":"REVIEWER"}],"messages":[{"id":"603123c3a0eb8fbdc93da1e5df3458a82c02ef43","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2015-06-26 21:48:36.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"11f0f3b3c9ed224e011bf2dca9ff84cf3f4f5d6e","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2015-06-26 22:30:49.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- gate-designate-pep8 http://logs.openstack.org/61/196261/1/check/gate-designate-pep8/7929654/ : SUCCESS in 2m 29s\n- gate-designate-docs http://docs-draft.openstack.org/61/196261/1/check/gate-designate-docs/75d70c6//doc/build/html/ : SUCCESS in 3m 03s\n- gate-designate-python27 http://logs.openstack.org/61/196261/1/check/gate-designate-python27/dd1cbde/ : SUCCESS in 3m 16s\n- gate-designate-requirements http://logs.openstack.org/61/196261/1/check/gate-designate-requirements/aefc220/ : SUCCESS in 19s\n- gate-designate-dsvm-powerdns http://logs.openstack.org/61/196261/1/check/gate-designate-dsvm-powerdns/a608457/ : SUCCESS in 20m 59s\n- gate-designate-dsvm-bind9 http://logs.openstack.org/61/196261/1/check/gate-designate-dsvm-bind9/19b626b/ : SUCCESS in 22m 14s\n- gate-rally-dsvm-designate-designate http://logs.openstack.org/61/196261/1/check/gate-rally-dsvm-designate-designate/a3c1fd1/ : SUCCESS in 18m 10s (non-voting)\n- designate-coverage http://logs.openstack.org/61/196261/1/check/designate-coverage/4c54b63/ : SUCCESS in 3m 12s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"29435d9e63600563296a0200c3a03d9840da75b3","author":{"_account_id":741,"name":"Kiall Mac Innes","email":"kiall@macinnes.ie","username":"kiall"},"date":"2015-06-27 09:18:10.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"95dd48ab15c5a26aa62397d09dd3bd8e012a6ea4","author":{"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},"date":"2015-06-29 13:07:09.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"5766a63dcc06c57ec080992fce15b2aacdf93733","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2015-06-29 13:07:15.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.\nhttp://status.openstack.org/zuul/","accounts_in_message":[],"_revision_number":1},{"id":"bf2413ae131bf4e4ed2af75ec5ed51bf929d67c4","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2015-06-29 13:30:54.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- gate-designate-docs http://docs-draft.openstack.org/61/196261/1/gate/gate-designate-docs/da41fe2//doc/build/html/ : SUCCESS in 6m 32s\n- gate-designate-pep8 http://logs.openstack.org/61/196261/1/gate/gate-designate-pep8/78a8039/ : SUCCESS in 2m 06s\n- gate-designate-python27 http://logs.openstack.org/61/196261/1/gate/gate-designate-python27/68e1af5/ : SUCCESS in 2m 27s\n- gate-designate-requirements http://logs.openstack.org/61/196261/1/gate/gate-designate-requirements/f2eeafb/ : SUCCESS in 15s\n- gate-designate-dsvm-powerdns http://logs.openstack.org/61/196261/1/gate/gate-designate-dsvm-powerdns/8391210/ : SUCCESS in 22m 32s\n- gate-designate-dsvm-bind9 http://logs.openstack.org/61/196261/1/gate/gate-designate-dsvm-bind9/a9c270e/ : SUCCESS in 22m 51s","accounts_in_message":[],"_revision_number":1},{"id":"6a849916162e959c2974936521d1bc7897ecc554","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2015-06-29 13:30:57.000000000","message":"Change has been successfully merged into the git repository.","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"f6d4333eb01fdc8af8749fc08506d84c9db705ec","revisions":{"f6d4333eb01fdc8af8749fc08506d84c9db705ec":{"kind":"REWORK","_number":1,"created":"2015-06-26 21:48:36.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/61/196261/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/61/196261/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/61/196261/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/61/196261/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/61/196261/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/61/196261/1"}}},"commit":{"parents":[{"commit":"e674ec8e6d62d851d784fac35b70e4d80fd8c0f7","subject":"Merge \"port to oslo.service\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/e674ec8e6d62d851d784fac35b70e4d80fd8c0f7"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2015-06-26 21:48:35.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2015-06-26 21:48:35.000000000","tz":0},"subject":"Updated from global requirements","message":"Updated from global requirements\n\nChange-Id: I8e5a4b518d9ff1159720a18711052005305806ae\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/f6d4333eb01fdc8af8749fc08506d84c9db705ec"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/f6d4333eb01fdc8af8749fc08506d84c9db705ec"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"e674ec8e6d62d851d784fac35b70e4d80fd8c0f7","is_merged_in_target_branch":true}],"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
