)]}'
{"id":"openstack%2Fdesignate~220813","triplet_id":"openstack%2Fdesignate~master~Ia6b1e855708254ab2d17ffd92eed3c140b2b9836","project":"openstack/designate","branch":"master","topic":"bug/1492749","hashtags":[],"change_id":"Ia6b1e855708254ab2d17ffd92eed3c140b2b9836","subject":"Fix AXFR / Secondary zone functionality","status":"MERGED","created":"2015-09-06 12:29:49.000000000","updated":"2015-09-07 13:01:24.000000000","submitted":"2015-09-07 13:01:23.000000000","submitter":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"a798637fd3bb5c2097d8f7f8dd34c01e47ba5ad5","_number":220813,"virtual_id_number":220813,"owner":{"_account_id":395,"name":"Endre Karlson","email":"endre.karlson@gmail.com","username":"endre-karlson"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"all":[{"value":0,"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},{"value":2,"date":"2015-09-07 13:01:23.000000000","_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":0,"_account_id":741,"name":"Kiall Mac Innes","email":"kiall@macinnes.ie","username":"kiall"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},"all":[{"value":2,"date":"2015-09-07 12:27:59.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},{"value":0,"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":2,"date":"2015-09-07 08:58:13.000000000","_account_id":741,"name":"Kiall Mac Innes","email":"kiall@macinnes.ie","username":"kiall"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},"all":[{"value":1,"date":"2015-09-07 12:27:59.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},{"value":0,"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":0,"_account_id":741,"name":"Kiall Mac Innes","email":"kiall@macinnes.ie","username":"kiall"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},{"value":0,"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":0,"_account_id":741,"name":"Kiall Mac Innes","email":"kiall@macinnes.ie","username":"kiall"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":3,"name":"Jenkins","username":"jenkins"},{"_account_id":741,"name":"Kiall Mac Innes","email":"kiall@macinnes.ie","username":"kiall"},{"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2015-09-07 08:58:13.000000000","updated_by":{"_account_id":741,"name":"Kiall Mac Innes","email":"kiall@macinnes.ie","username":"kiall"},"reviewer":{"_account_id":741,"name":"Kiall Mac Innes","email":"kiall@macinnes.ie","username":"kiall"},"state":"REVIEWER"},{"updated":"2015-09-07 12:27:59.000000000","updated_by":{"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},"reviewer":{"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},"state":"REVIEWER"},{"updated":"2015-09-07 13:01:23.000000000","updated_by":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"reviewer":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"state":"REVIEWER"}],"messages":[{"id":"e0f9e73371170a290f9fa5a76f173838da2a27b1","author":{"_account_id":395,"name":"Endre Karlson","email":"endre.karlson@gmail.com","username":"endre-karlson"},"date":"2015-09-06 12:29:49.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"aae5bb0e3f0d229ef299fb20a17681a9c61294b2","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2015-09-06 12:55:43.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- gate-designate-pep8 http://logs.openstack.org/13/220813/1/check/gate-designate-pep8/38f16ef/ : SUCCESS in 1m 49s\n- gate-designate-docs http://docs-draft.openstack.org/13/220813/1/check/gate-designate-docs/23090d3//doc/build/html/ : SUCCESS in 1m 41s\n- gate-designate-python27 http://logs.openstack.org/13/220813/1/check/gate-designate-python27/a7c63df/ : SUCCESS in 2m 20s\n- gate-designate-dsvm-powerdns http://logs.openstack.org/13/220813/1/check/gate-designate-dsvm-powerdns/9e71d2b/ : SUCCESS in 25m 38s\n- gate-designate-dsvm-bind9 http://logs.openstack.org/13/220813/1/check/gate-designate-dsvm-bind9/2f28c4f/ : SUCCESS in 24m 16s\n- gate-rally-dsvm-designate-designate http://logs.openstack.org/13/220813/1/check/gate-rally-dsvm-designate-designate/eb4b116/ : SUCCESS in 22m 37s (non-voting)\n- designate-coverage http://logs.openstack.org/13/220813/1/check/designate-coverage/d86e3f9/ : SUCCESS in 3m 41s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"f6d120b1e0618f928aa26281523205dc65896d90","author":{"_account_id":741,"name":"Kiall Mac Innes","email":"kiall@macinnes.ie","username":"kiall"},"date":"2015-09-07 08:58:13.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"e9207fb9f46fdef1b2ada59f7e8269b355ba38e7","author":{"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},"date":"2015-09-07 12:27:59.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"a0b7dde4b41c82558f645f6090da7be8be89c156","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2015-09-07 12:28:05.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.\nhttp://status.openstack.org/zuul/","accounts_in_message":[],"_revision_number":1},{"id":"df213f807c069462e6f6b6e2a11d323c4a7d3d10","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2015-09-07 13:01:23.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- gate-designate-docs http://docs-draft.openstack.org/13/220813/1/gate/gate-designate-docs/a134f92//doc/build/html/ : SUCCESS in 2m 17s\n- gate-designate-pep8 http://logs.openstack.org/13/220813/1/gate/gate-designate-pep8/7cc0e89/ : SUCCESS in 2m 02s\n- gate-designate-python27 http://logs.openstack.org/13/220813/1/gate/gate-designate-python27/9a66693/ : SUCCESS in 3m 08s\n- gate-designate-dsvm-powerdns http://logs.openstack.org/13/220813/1/gate/gate-designate-dsvm-powerdns/3df164b/ : SUCCESS in 29m 00s\n- gate-designate-dsvm-bind9 http://logs.openstack.org/13/220813/1/gate/gate-designate-dsvm-bind9/4293283/ : SUCCESS in 25m 19s","accounts_in_message":[],"_revision_number":1},{"id":"ef2e684b28c1f679df99b9c4d06ef4551decd239","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2015-09-07 13:01:24.000000000","message":"Change has been successfully merged into the git repository.","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"9350ea7ad0d2ea2085a0dbc05ee6a93d8b08c4c6","revisions":{"9350ea7ad0d2ea2085a0dbc05ee6a93d8b08c4c6":{"kind":"REWORK","_number":1,"created":"2015-09-06 12:29:49.000000000","uploader":{"_account_id":395,"name":"Endre Karlson","email":"endre.karlson@gmail.com","username":"endre-karlson"},"ref":"refs/changes/13/220813/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/13/220813/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/13/220813/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/13/220813/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/13/220813/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/13/220813/1"}}},"commit":{"parents":[{"commit":"dbd58dfe00883d8e45db905aa8287f845821f1da","subject":"Updated from global requirements","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/dbd58dfe00883d8e45db905aa8287f845821f1da"}]}],"author":{"name":"Endre Karlson","email":"endre.karlson@hp.com","date":"2015-09-06 12:02:22.000000000","tz":120},"committer":{"name":"Endre Karlson","email":"endre.karlson@hp.com","date":"2015-09-06 12:29:38.000000000","tz":120},"subject":"Fix AXFR / Secondary zone functionality","message":"Fix AXFR / Secondary zone functionality\n\n* Dont load masters when loading attribute relations as this causes\n  update_domain() to explode after the zone is AXFR\u0027d\n* Change srv.ip \u003e srv.host\n\nCloses-Bug: #1492749\nCloses-Bug: #1492750\nChange-Id: Ia6b1e855708254ab2d17ffd92eed3c140b2b9836\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/9350ea7ad0d2ea2085a0dbc05ee6a93d8b08c4c6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/9350ea7ad0d2ea2085a0dbc05ee6a93d8b08c4c6"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"dbd58dfe00883d8e45db905aa8287f845821f1da","is_merged_in_target_branch":true,"change_id":"I88ca5f963c906eee602c57c99b4a33ad5eacc252","change_number":219491,"patch_set_number":2,"change_status":"MERGED"}],"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
