)]}'
{"id":"openstack%2Fdesignate~445019","triplet_id":"openstack%2Fdesignate~stable%2Fmitaka~I0af3b9a2121674cce3dfa9d666b6fad819e3a175","project":"openstack/designate","branch":"stable/mitaka","hashtags":[],"change_id":"I0af3b9a2121674cce3dfa9d666b6fad819e3a175","subject":"Fix broken doc link","status":"ABANDONED","created":"2017-03-13 17:13:05.000000000","updated":"2017-07-05 07:59:18.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"1d32178b11b411b2841656b0fb402de09bfee32f","_number":445019,"virtual_id_number":445019,"owner":{"_account_id":23185,"name":"Samantha Blanco","email":"samantha.blanco@att.com","username":"sblanco"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"all":[{"date":"2017-03-24 16:08:13.000000000","_account_id":23185,"name":"Samantha Blanco","email":"samantha.blanco@att.com","username":"sblanco"},{"value":-1,"date":"2017-03-13 17:23:22.000000000","_account_id":3,"name":"Jenkins","username":"jenkins"},{"_account_id":25254,"name":"Nguyen Van Trung","email":"trungnvfet@outlook.com","username":"trungnv"},{"_account_id":24441,"name":"Cuong Nguyen","email":"cuongnv@vn.fujitsu.com","username":"cuongnv"},{"date":"2017-06-01 16:39:42.000000000","_account_id":19930,"name":"Dai Dang Van","email":"daidv@omzcloud.vn","username":"daidv"},{"_account_id":13340,"name":"Konstantin Baikov","email":"konstantin.baikov@gmail.com","username":"kbaikov"},{"date":"2017-07-05 07:59:18.000000000","_account_id":7069,"name":"Joshua Hesketh","email":"josh@nitrotech.org","username":"jhesketh"},{"_account_id":19741,"name":"Hieu LE","email":"hieulq2@viettel.com.vn","username":"hieulq"},{"date":"2017-03-13 19:01:10.000000000","_account_id":8174,"name":"Tim Simmons","email":"tim.r.simmons@gmail.com","username":"TimSimmons"},{"_account_id":25903,"name":"Vu Cong Tuan","email":"tuanvc@vn.fujitsu.com","username":"tuanvu"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"recommended":{"_account_id":13340,"name":"Konstantin Baikov","email":"konstantin.baikov@gmail.com","username":"kbaikov"},"disliked":{"_account_id":24441,"name":"Cuong Nguyen","email":"cuongnv@vn.fujitsu.com","username":"cuongnv"},"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":23185,"name":"Samantha Blanco","email":"samantha.blanco@att.com","username":"sblanco"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":1,"date":"2017-06-15 05:34:10.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":25254,"name":"Nguyen Van Trung","email":"trungnvfet@outlook.com","username":"trungnv"},{"value":-1,"date":"2017-07-04 02:14:01.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":24441,"name":"Cuong Nguyen","email":"cuongnv@vn.fujitsu.com","username":"cuongnv"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":19930,"name":"Dai Dang Van","email":"daidv@omzcloud.vn","username":"daidv"},{"value":1,"date":"2017-05-30 14:55:37.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":13340,"name":"Konstantin Baikov","email":"konstantin.baikov@gmail.com","username":"kbaikov"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":7069,"name":"Joshua Hesketh","email":"josh@nitrotech.org","username":"jhesketh"},{"value":1,"date":"2017-05-19 00:59:31.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":19741,"name":"Hieu LE","email":"hieulq2@viettel.com.vn","username":"hieulq"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":8174,"name":"Tim Simmons","email":"tim.r.simmons@gmail.com","username":"TimSimmons"},{"value":1,"date":"2017-05-22 08:02:46.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":25903,"name":"Vu Cong Tuan","email":"tuanvc@vn.fujitsu.com","username":"tuanvu"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":-1,"default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":0},"_account_id":23185,"name":"Samantha Blanco","email":"samantha.blanco@att.com","username":"sblanco"},{"_account_id":3,"name":"Jenkins","username":"jenkins"},{"_account_id":25254,"name":"Nguyen Van Trung","email":"trungnvfet@outlook.com","username":"trungnv"},{"_account_id":24441,"name":"Cuong Nguyen","email":"cuongnv@vn.fujitsu.com","username":"cuongnv"},{"_account_id":19930,"name":"Dai Dang Van","email":"daidv@omzcloud.vn","username":"daidv"},{"_account_id":13340,"name":"Konstantin Baikov","email":"konstantin.baikov@gmail.com","username":"kbaikov"},{"_account_id":7069,"name":"Joshua Hesketh","email":"josh@nitrotech.org","username":"jhesketh"},{"_account_id":19741,"name":"Hieu LE","email":"hieulq2@viettel.com.vn","username":"hieulq"},{"_account_id":8174,"name":"Tim Simmons","email":"tim.r.simmons@gmail.com","username":"TimSimmons"},{"_account_id":25903,"name":"Vu Cong Tuan","email":"tuanvc@vn.fujitsu.com","username":"tuanvu"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":23185,"name":"Samantha Blanco","email":"samantha.blanco@att.com","username":"sblanco"},{"_account_id":3,"name":"Jenkins","username":"jenkins"},{"_account_id":25254,"name":"Nguyen Van Trung","email":"trungnvfet@outlook.com","username":"trungnv"},{"_account_id":24441,"name":"Cuong Nguyen","email":"cuongnv@vn.fujitsu.com","username":"cuongnv"},{"_account_id":19930,"name":"Dai Dang Van","email":"daidv@omzcloud.vn","username":"daidv"},{"_account_id":13340,"name":"Konstantin Baikov","email":"konstantin.baikov@gmail.com","username":"kbaikov"},{"_account_id":7069,"name":"Joshua Hesketh","email":"josh@nitrotech.org","username":"jhesketh"},{"_account_id":19741,"name":"Hieu LE","email":"hieulq2@viettel.com.vn","username":"hieulq"},{"_account_id":8174,"name":"Tim Simmons","email":"tim.r.simmons@gmail.com","username":"TimSimmons"},{"_account_id":25903,"name":"Vu Cong Tuan","email":"tuanvc@vn.fujitsu.com","username":"tuanvu"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":3,"name":"Jenkins","username":"jenkins"},{"_account_id":7069,"name":"Joshua Hesketh","email":"josh@nitrotech.org","username":"jhesketh"},{"_account_id":8174,"name":"Tim Simmons","email":"tim.r.simmons@gmail.com","username":"TimSimmons"},{"_account_id":13340,"name":"Konstantin Baikov","email":"konstantin.baikov@gmail.com","username":"kbaikov"},{"_account_id":19741,"name":"Hieu LE","email":"hieulq2@viettel.com.vn","username":"hieulq"},{"_account_id":19930,"name":"Dai Dang Van","email":"daidv@omzcloud.vn","username":"daidv"},{"_account_id":23185,"name":"Samantha Blanco","email":"samantha.blanco@att.com","username":"sblanco"},{"_account_id":24441,"name":"Cuong Nguyen","email":"cuongnv@vn.fujitsu.com","username":"cuongnv"},{"_account_id":25254,"name":"Nguyen Van Trung","email":"trungnvfet@outlook.com","username":"trungnv"},{"_account_id":25903,"name":"Vu Cong Tuan","email":"tuanvc@vn.fujitsu.com","username":"tuanvu"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2017-03-13 17:23:22.000000000","updated_by":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"reviewer":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"state":"REVIEWER"},{"updated":"2017-03-13 19:01:10.000000000","updated_by":{"_account_id":8174,"name":"Tim Simmons","email":"tim.r.simmons@gmail.com","username":"TimSimmons"},"reviewer":{"_account_id":8174,"name":"Tim Simmons","email":"tim.r.simmons@gmail.com","username":"TimSimmons"},"state":"REVIEWER"},{"updated":"2017-05-19 00:59:31.000000000","updated_by":{"_account_id":19741,"name":"Hieu LE","email":"hieulq2@viettel.com.vn","username":"hieulq"},"reviewer":{"_account_id":19741,"name":"Hieu LE","email":"hieulq2@viettel.com.vn","username":"hieulq"},"state":"REVIEWER"},{"updated":"2017-05-22 08:02:46.000000000","updated_by":{"_account_id":25903,"name":"Vu Cong Tuan","email":"tuanvc@vn.fujitsu.com","username":"tuanvu"},"reviewer":{"_account_id":25903,"name":"Vu Cong Tuan","email":"tuanvc@vn.fujitsu.com","username":"tuanvu"},"state":"REVIEWER"},{"updated":"2017-05-30 14:55:37.000000000","updated_by":{"_account_id":13340,"name":"Konstantin Baikov","email":"konstantin.baikov@gmail.com","username":"kbaikov"},"reviewer":{"_account_id":13340,"name":"Konstantin Baikov","email":"konstantin.baikov@gmail.com","username":"kbaikov"},"state":"REVIEWER"},{"updated":"2017-06-01 16:39:42.000000000","updated_by":{"_account_id":19930,"name":"Dai Dang Van","email":"daidv@omzcloud.vn","username":"daidv"},"reviewer":{"_account_id":19930,"name":"Dai Dang Van","email":"daidv@omzcloud.vn","username":"daidv"},"state":"REVIEWER"},{"updated":"2017-06-15 05:34:10.000000000","updated_by":{"_account_id":25254,"name":"Nguyen Van Trung","email":"trungnvfet@outlook.com","username":"trungnv"},"reviewer":{"_account_id":25254,"name":"Nguyen Van Trung","email":"trungnvfet@outlook.com","username":"trungnv"},"state":"REVIEWER"},{"updated":"2017-07-04 02:14:01.000000000","updated_by":{"_account_id":24441,"name":"Cuong Nguyen","email":"cuongnv@vn.fujitsu.com","username":"cuongnv"},"reviewer":{"_account_id":24441,"name":"Cuong Nguyen","email":"cuongnv@vn.fujitsu.com","username":"cuongnv"},"state":"REVIEWER"},{"updated":"2017-07-05 07:59:18.000000000","updated_by":{"_account_id":7069,"name":"Joshua Hesketh","email":"josh@nitrotech.org","username":"jhesketh"},"reviewer":{"_account_id":7069,"name":"Joshua Hesketh","email":"josh@nitrotech.org","username":"jhesketh"},"state":"REVIEWER"}],"messages":[{"id":"4f4c3adeaaab056a4efb94c31cdf0b02464cdd31","author":{"_account_id":23185,"name":"Samantha Blanco","email":"samantha.blanco@att.com","username":"sblanco"},"date":"2017-03-13 17:13:05.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"de525785d4fd545f84458c17133eb5282964e71b","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2017-03-13 17:23:22.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see http://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n- gate-designate-docs-ubuntu-trusty http://docs-draft.openstack.org/19/445019/1/check/gate-designate-docs-ubuntu-trusty/ddcf53d//doc/build/html/ : SUCCESS in 3m 01s\n- gate-designate-pep8-ubuntu-trusty http://logs.openstack.org/19/445019/1/check/gate-designate-pep8-ubuntu-trusty/57b22cb/ : SUCCESS in 3m 17s\n- gate-designate-python27-ubuntu-trusty http://logs.openstack.org/19/445019/1/check/gate-designate-python27-ubuntu-trusty/ec91ad0/ : SUCCESS in 3m 23s\n- gate-designate-python34 http://logs.openstack.org/19/445019/1/check/gate-designate-python34/e2a355e/ : SUCCESS in 4m 16s\n- gate-designate-dsvm-powerdns-ubuntu-trusty http://logs.openstack.org/19/445019/1/check/gate-designate-dsvm-powerdns-ubuntu-trusty/0cdc6dc/ : FAILURE in 3m 34s\n- gate-designate-dsvm-bind9-ubuntu-trusty http://logs.openstack.org/19/445019/1/check/gate-designate-dsvm-bind9-ubuntu-trusty/245afab/ : FAILURE in 6m 27s\n- designate-coverage-ubuntu-trusty-nv http://logs.openstack.org/19/445019/1/check/designate-coverage-ubuntu-trusty-nv/f722ac6/ : SUCCESS in 4m 52s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"bf4760e28d956ec4cb884483af87ddd54e9c9962","author":{"_account_id":8174,"name":"Tim Simmons","email":"tim.r.simmons@gmail.com","username":"TimSimmons"},"date":"2017-03-13 19:01:10.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"5e4549088d3ae9ddc01cbac80badb2df43428be6","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2017-03-13 19:07:50.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see http://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n- gate-designate-docs-ubuntu-trusty http://docs-draft.openstack.org/19/445019/1/check/gate-designate-docs-ubuntu-trusty/f18db1a//doc/build/html/ : SUCCESS in 2m 30s\n- gate-designate-pep8-ubuntu-trusty http://logs.openstack.org/19/445019/1/check/gate-designate-pep8-ubuntu-trusty/bdb630c/ : SUCCESS in 2m 56s\n- gate-designate-python27-ubuntu-trusty http://logs.openstack.org/19/445019/1/check/gate-designate-python27-ubuntu-trusty/c95444b/ : SUCCESS in 3m 59s\n- gate-designate-python34 http://logs.openstack.org/19/445019/1/check/gate-designate-python34/b7bb13b/ : SUCCESS in 3m 02s\n- gate-designate-dsvm-powerdns-ubuntu-trusty http://logs.openstack.org/19/445019/1/check/gate-designate-dsvm-powerdns-ubuntu-trusty/ae4a8b1/ : FAILURE in 4m 17s\n- gate-designate-dsvm-bind9-ubuntu-trusty http://logs.openstack.org/19/445019/1/check/gate-designate-dsvm-bind9-ubuntu-trusty/49891bf/ : FAILURE in 3m 45s\n- designate-coverage-ubuntu-trusty-nv http://logs.openstack.org/19/445019/1/check/designate-coverage-ubuntu-trusty-nv/5c7b1cc/ : SUCCESS in 5m 10s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"2825828b99d17a0e7cce7365ab8511cfd44c78ec","author":{"_account_id":23185,"name":"Samantha Blanco","email":"samantha.blanco@att.com","username":"sblanco"},"date":"2017-03-24 16:08:13.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"acf6b820cedfb784b56263da4f76005d770dbce0","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2017-03-24 16:17:41.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see http://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n- gate-designate-docs-ubuntu-trusty http://docs-draft.openstack.org/19/445019/1/check/gate-designate-docs-ubuntu-trusty/cbb2af3//doc/build/html/ : SUCCESS in 4m 08s\n- gate-designate-pep8-ubuntu-trusty http://logs.openstack.org/19/445019/1/check/gate-designate-pep8-ubuntu-trusty/e721a8c/ : SUCCESS in 3m 27s\n- gate-designate-python27-ubuntu-trusty http://logs.openstack.org/19/445019/1/check/gate-designate-python27-ubuntu-trusty/6fa85d7/ : SUCCESS in 3m 36s\n- gate-designate-python34 http://logs.openstack.org/19/445019/1/check/gate-designate-python34/96773bb/ : SUCCESS in 2m 58s\n- gate-designate-dsvm-powerdns-ubuntu-trusty http://logs.openstack.org/19/445019/1/check/gate-designate-dsvm-powerdns-ubuntu-trusty/506e1dd/ : FAILURE in 3m 58s\n- gate-designate-dsvm-bind9-ubuntu-trusty http://logs.openstack.org/19/445019/1/check/gate-designate-dsvm-bind9-ubuntu-trusty/9845d34/ : FAILURE in 5m 55s\n- designate-coverage-ubuntu-trusty-nv http://logs.openstack.org/19/445019/1/check/designate-coverage-ubuntu-trusty-nv/1bb15d8/ : SUCCESS in 5m 53s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"9a69ff68e6dd83f680981fb447a65e90d73b6045","author":{"_account_id":25254,"name":"Nguyen Van Trung","email":"trungnvfet@outlook.com","username":"trungnv"},"date":"2017-05-18 08:49:12.000000000","message":"Patch Set 1: Code-Review+1\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"14c4eda448bf8553a7c65ff21ad7f2ce1d4cf7fb","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2017-05-18 08:56:31.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see http://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n- gate-designate-docs-ubuntu-trusty http://docs-draft.openstack.org/19/445019/1/check/gate-designate-docs-ubuntu-trusty/5910f1d//doc/build/html/ : SUCCESS in 2m 46s\n- gate-designate-pep8-ubuntu-trusty http://logs.openstack.org/19/445019/1/check/gate-designate-pep8-ubuntu-trusty/ab379a4/ : SUCCESS in 3m 43s\n- gate-designate-python27-ubuntu-trusty http://logs.openstack.org/19/445019/1/check/gate-designate-python27-ubuntu-trusty/bc0c805/ : SUCCESS in 3m 39s\n- gate-designate-python34 http://logs.openstack.org/19/445019/1/check/gate-designate-python34/110fb61/ : SUCCESS in 2m 43s\n- gate-designate-dsvm-powerdns-ubuntu-trusty http://logs.openstack.org/19/445019/1/check/gate-designate-dsvm-powerdns-ubuntu-trusty/13f0f89/ : FAILURE in 3m 06s\n- gate-designate-dsvm-bind9-ubuntu-trusty http://logs.openstack.org/19/445019/1/check/gate-designate-dsvm-bind9-ubuntu-trusty/510b664/ : FAILURE in 3m 25s\n- designate-coverage-ubuntu-trusty-nv http://logs.openstack.org/19/445019/1/check/designate-coverage-ubuntu-trusty-nv/2bf5337/ : FAILURE in 2m 31s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"067bf13aae50805667714c12fd026cd359f26ce6","author":{"_account_id":25254,"name":"Nguyen Van Trung","email":"trungnvfet@outlook.com","username":"trungnv"},"date":"2017-05-18 17:14:24.000000000","message":"Patch Set 1: -Code-Review\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"8385c20d9896f4d320d82ad1aad3560479eb122e","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2017-05-18 17:24:12.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see http://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n- gate-designate-docs-ubuntu-trusty http://docs-draft.openstack.org/19/445019/1/check/gate-designate-docs-ubuntu-trusty/b46dd84//doc/build/html/ : SUCCESS in 2m 38s\n- gate-designate-pep8-ubuntu-trusty http://logs.openstack.org/19/445019/1/check/gate-designate-pep8-ubuntu-trusty/5ec5ee3/ : SUCCESS in 3m 27s\n- gate-designate-python27-ubuntu-trusty http://logs.openstack.org/19/445019/1/check/gate-designate-python27-ubuntu-trusty/4387569/ : SUCCESS in 3m 30s\n- gate-designate-python34 http://logs.openstack.org/19/445019/1/check/gate-designate-python34/a8b31f5/ : SUCCESS in 2m 50s\n- gate-designate-dsvm-powerdns-ubuntu-trusty http://logs.openstack.org/19/445019/1/check/gate-designate-dsvm-powerdns-ubuntu-trusty/28bb8df/ : FAILURE in 2m 38s\n- gate-designate-dsvm-bind9-ubuntu-trusty http://logs.openstack.org/19/445019/1/check/gate-designate-dsvm-bind9-ubuntu-trusty/34ea723/ : FAILURE in 5m 26s\n- designate-coverage-ubuntu-trusty-nv http://logs.openstack.org/19/445019/1/check/designate-coverage-ubuntu-trusty-nv/027cee0/ : FAILURE in 2m 29s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"fa26152c39842559ce42ae3ac179976550406d38","author":{"_account_id":19741,"name":"Hieu LE","email":"hieulq2@viettel.com.vn","username":"hieulq"},"date":"2017-05-19 00:59:31.000000000","message":"Patch Set 1: Code-Review+1\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"072837a1441a8f25b96c890f13f97cf944c2d190","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2017-05-19 01:05:56.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see http://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n- gate-designate-docs-ubuntu-trusty http://docs-draft.openstack.org/19/445019/1/check/gate-designate-docs-ubuntu-trusty/9741eaf//doc/build/html/ : SUCCESS in 2m 27s\n- gate-designate-pep8-ubuntu-trusty http://logs.openstack.org/19/445019/1/check/gate-designate-pep8-ubuntu-trusty/d2d325c/ : SUCCESS in 2m 48s\n- gate-designate-python27-ubuntu-trusty http://logs.openstack.org/19/445019/1/check/gate-designate-python27-ubuntu-trusty/ae5955f/ : SUCCESS in 3m 14s\n- gate-designate-python34 http://logs.openstack.org/19/445019/1/check/gate-designate-python34/5446258/ : SUCCESS in 2m 36s\n- gate-designate-dsvm-powerdns-ubuntu-trusty http://logs.openstack.org/19/445019/1/check/gate-designate-dsvm-powerdns-ubuntu-trusty/1e3a9e5/ : FAILURE in 2m 39s\n- gate-designate-dsvm-bind9-ubuntu-trusty http://logs.openstack.org/19/445019/1/check/gate-designate-dsvm-bind9-ubuntu-trusty/0dae901/ : FAILURE in 4m 58s\n- designate-coverage-ubuntu-trusty-nv http://logs.openstack.org/19/445019/1/check/designate-coverage-ubuntu-trusty-nv/8d10d29/ : FAILURE in 1m 34s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"9093ee2dac01a1f5ace99870a72526f750f5e072","author":{"_account_id":25903,"name":"Vu Cong Tuan","email":"tuanvc@vn.fujitsu.com","username":"tuanvu"},"date":"2017-05-22 08:02:46.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"da6107ecbbf2bafbfa57c881a750d279969c1bda","author":{"_account_id":13340,"name":"Konstantin Baikov","email":"konstantin.baikov@gmail.com","username":"kbaikov"},"date":"2017-05-30 14:55:37.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"e1b340babd86aec53eafe852de04bad91322ce74","author":{"_account_id":19930,"name":"Dai Dang Van","email":"daidv@omzcloud.vn","username":"daidv"},"date":"2017-06-01 16:39:42.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"2939e9b625c5486807b979c2e8f41cb63ec166f7","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2017-06-01 16:50:06.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see http://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n- gate-designate-docs-ubuntu-trusty http://docs-draft.openstack.org/19/445019/1/check/gate-designate-docs-ubuntu-trusty/7632731//doc/build/html/ : SUCCESS in 2m 29s\n- gate-designate-pep8-ubuntu-trusty http://logs.openstack.org/19/445019/1/check/gate-designate-pep8-ubuntu-trusty/5456d13/ : SUCCESS in 3m 01s\n- gate-designate-python27-ubuntu-trusty http://logs.openstack.org/19/445019/1/check/gate-designate-python27-ubuntu-trusty/88bd1f5/ : SUCCESS in 3m 31s\n- gate-designate-python34 http://logs.openstack.org/19/445019/1/check/gate-designate-python34/6e1f79d/ : SUCCESS in 3m 01s\n- gate-designate-dsvm-powerdns-ubuntu-trusty http://logs.openstack.org/19/445019/1/check/gate-designate-dsvm-powerdns-ubuntu-trusty/670f062/ : FAILURE in 2m 26s\n- gate-designate-dsvm-bind9-ubuntu-trusty http://logs.openstack.org/19/445019/1/check/gate-designate-dsvm-bind9-ubuntu-trusty/3a1807a/ : FAILURE in 7m 11s\n- designate-coverage-ubuntu-trusty-nv http://logs.openstack.org/19/445019/1/check/designate-coverage-ubuntu-trusty-nv/7e21b35/ : FAILURE in 3m 01s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"215b794c0fa462fadded037cae3508e4f8add6d0","author":{"_account_id":25254,"name":"Nguyen Van Trung","email":"trungnvfet@outlook.com","username":"trungnv"},"date":"2017-06-15 05:34:10.000000000","message":"Patch Set 1: Code-Review+1\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"c50b266f94ff998c3208700841b8faefed51edb2","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2017-06-15 05:47:12.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see http://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n- gate-designate-docs-ubuntu-trusty http://docs-draft.openstack.org/19/445019/1/check/gate-designate-docs-ubuntu-trusty/d93f88c//doc/build/html/ : SUCCESS in 2m 25s\n- gate-designate-pep8-ubuntu-trusty http://logs.openstack.org/19/445019/1/check/gate-designate-pep8-ubuntu-trusty/72c6745/ : SUCCESS in 2m 51s\n- gate-designate-python27-ubuntu-trusty http://logs.openstack.org/19/445019/1/check/gate-designate-python27-ubuntu-trusty/9697e7c/ : SUCCESS in 3m 18s\n- gate-designate-python34 http://logs.openstack.org/19/445019/1/check/gate-designate-python34/92f5dfa/ : SUCCESS in 4m 46s\n- gate-designate-dsvm-powerdns-ubuntu-trusty http://logs.openstack.org/19/445019/1/check/gate-designate-dsvm-powerdns-ubuntu-trusty/5cace22/ : FAILURE in 7m 43s\n- gate-designate-dsvm-bind9-ubuntu-trusty http://logs.openstack.org/19/445019/1/check/gate-designate-dsvm-bind9-ubuntu-trusty/4028b33/ : FAILURE in 2m 48s\n- designate-coverage-ubuntu-trusty-nv http://logs.openstack.org/19/445019/1/check/designate-coverage-ubuntu-trusty-nv/7a82c3d/ : FAILURE in 2m 21s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"687015584dfe0874a44251510ad81a98496a71ce","author":{"_account_id":24441,"name":"Cuong Nguyen","email":"cuongnv@vn.fujitsu.com","username":"cuongnv"},"date":"2017-07-04 02:14:01.000000000","message":"Patch Set 1: Code-Review-1\n\nnew link is still broken, pls update","accounts_in_message":[],"_revision_number":1},{"id":"538ae1fbddec97baa69c215c828917986f7c3081","author":{"_account_id":7069,"name":"Joshua Hesketh","email":"josh@nitrotech.org","username":"jhesketh"},"date":"2017-07-05 07:59:18.000000000","message":"Abandoned\n\nThis branch (stable/mitaka) is at End Of Life","accounts_in_message":[],"_revision_number":1},{"id":"1d32178b11b411b2841656b0fb402de09bfee32f","author":{"_account_id":7069,"name":"Joshua Hesketh","email":"josh@nitrotech.org","username":"jhesketh"},"date":"2017-07-05 07:59:18.000000000","message":"Patch Set 1:\n\nThis branch (stable/mitaka) is at End Of Life","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"39f622293dc64191ae5c20439b794c358671db0f","revisions":{"39f622293dc64191ae5c20439b794c358671db0f":{"kind":"REWORK","_number":1,"created":"2017-03-13 17:13:05.000000000","uploader":{"_account_id":23185,"name":"Samantha Blanco","email":"samantha.blanco@att.com","username":"sblanco"},"ref":"refs/changes/19/445019/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/19/445019/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/19/445019/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/19/445019/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/19/445019/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/19/445019/1"}}},"commit":{"parents":[{"commit":"5203c5cb6c8bdd59792890d815c355461d46aed6","subject":"zone_name is null in zone trasfer request response","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/5203c5cb6c8bdd59792890d815c355461d46aed6"}]}],"author":{"name":"Samantha Blanco","email":"samantha.blanco@att.com","date":"2017-03-13 16:46:32.000000000","tz":-240},"committer":{"name":"Samantha Blanco","email":"samantha.blanco@att.com","date":"2017-03-13 17:12:50.000000000","tz":-240},"subject":"Fix broken doc link","message":"Fix broken doc link\n\nCorrects broken doc link in Mitaka branch how to.\n\nChange-Id: I0af3b9a2121674cce3dfa9d666b6fad819e3a175\nCloses-Bug: #1662174\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/39f622293dc64191ae5c20439b794c358671db0f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/39f622293dc64191ae5c20439b794c358671db0f"}]},"parents_data":[{"branch_name":"refs/heads/stable/mitaka","commit_id":"5203c5cb6c8bdd59792890d815c355461d46aed6","is_merged_in_target_branch":true,"change_id":"Id2d093891953efcbb125560ea1113b8a9e613a9c","change_number":391460,"patch_set_number":1,"change_status":"MERGED"}],"branch":"refs/heads/stable/mitaka"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
