)]}'
{"id":"openstack%2Fdesignate~482161","triplet_id":"openstack%2Fdesignate~master~Iac78b8668bcdca34bf354d6feab3e9547ba2c71f","project":"openstack/designate","branch":"master","topic":"482161","hashtags":[],"change_id":"Iac78b8668bcdca34bf354d6feab3e9547ba2c71f","subject":"Update Driver statuses","status":"MERGED","created":"2017-07-10 14:38:35.000000000","updated":"2019-05-20 04:18:21.000000000","submitted":"2019-05-20 04:18:21.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"482161-1558325901335-16b933f0","meta_rev_id":"7ce035abc641baaa0756c33f4b1785b742e6b98e","_number":482161,"virtual_id_number":482161,"owner":{"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"date":"2019-04-02 13:38:07.000000000","_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},{"value":0,"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":0,"date":"2019-04-02 11:59:24.000000000","_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":25254,"name":"Nguyen Van Trung","email":"trungnvfet@outlook.com","username":"trungnv"},{"value":0,"_account_id":24441,"name":"Cuong Nguyen","email":"cuongnv@vn.fujitsu.com","username":"cuongnv"},{"value":0,"_account_id":19930,"name":"Dai Dang Van","email":"daidv@omzcloud.vn","username":"daidv"},{"value":2,"date":"2019-05-20 04:18:21.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"all":[{"value":0,"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},{"value":0,"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":25254,"name":"Nguyen Van Trung","email":"trungnvfet@outlook.com","username":"trungnv"},{"value":0,"_account_id":24441,"name":"Cuong Nguyen","email":"cuongnv@vn.fujitsu.com","username":"cuongnv"},{"value":0,"_account_id":19930,"name":"Dai Dang Van","email":"daidv@omzcloud.vn","username":"daidv"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2019-04-01 17:54:53.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"all":[{"value":0,"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},{"value":0,"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":25254,"name":"Nguyen Van Trung","email":"trungnvfet@outlook.com","username":"trungnv"},{"value":0,"_account_id":24441,"name":"Cuong Nguyen","email":"cuongnv@vn.fujitsu.com","username":"cuongnv"},{"value":0,"_account_id":19930,"name":"Dai Dang Van","email":"daidv@omzcloud.vn","username":"daidv"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2019-05-20 03:18:51.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},{"value":0,"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":25254,"name":"Nguyen Van Trung","email":"trungnvfet@outlook.com","username":"trungnv"},{"value":0,"_account_id":24441,"name":"Cuong Nguyen","email":"cuongnv@vn.fujitsu.com","username":"cuongnv"},{"value":0,"_account_id":19930,"name":"Dai Dang Van","email":"daidv@omzcloud.vn","username":"daidv"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":3,"name":"Jenkins","username":"jenkins"},{"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":19930,"name":"Dai Dang Van","email":"daidv@omzcloud.vn","username":"daidv"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},{"_account_id":24441,"name":"Cuong Nguyen","email":"cuongnv@vn.fujitsu.com","username":"cuongnv"},{"_account_id":25254,"name":"Nguyen Van Trung","email":"trungnvfet@outlook.com","username":"trungnv"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2017-07-10 18:50:07.000000000","updated_by":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"reviewer":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"state":"REVIEWER"},{"updated":"2017-07-12 04:21:33.000000000","updated_by":{"_account_id":19930,"name":"Dai Dang Van","email":"daidv@omzcloud.vn","username":"daidv"},"reviewer":{"_account_id":19930,"name":"Dai Dang Van","email":"daidv@omzcloud.vn","username":"daidv"},"state":"REVIEWER"},{"updated":"2017-07-19 09:46:26.000000000","updated_by":{"_account_id":24441,"name":"Cuong Nguyen","email":"cuongnv@vn.fujitsu.com","username":"cuongnv"},"reviewer":{"_account_id":24441,"name":"Cuong Nguyen","email":"cuongnv@vn.fujitsu.com","username":"cuongnv"},"state":"REVIEWER"},{"updated":"2017-07-19 15:03:19.000000000","updated_by":{"_account_id":25254,"name":"Nguyen Van Trung","email":"trungnvfet@outlook.com","username":"trungnv"},"reviewer":{"_account_id":25254,"name":"Nguyen Van Trung","email":"trungnvfet@outlook.com","username":"trungnv"},"state":"REVIEWER"},{"updated":"2019-04-02 11:59:24.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"REVIEWER"},{"updated":"2019-05-20 03:18:51.000000000","updated_by":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"reviewer":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"state":"REVIEWER"},{"updated":"2019-05-20 04:18:21.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"26edc1ea20722c33a3664b9dd3446b61e17bf471","author":{"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},"date":"2017-07-10 14:38:35.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"55ab722702809aab5ae7b2ec293ead80f6222ed6","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2017-07-10 14:53:14.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- gate-designate-docs-ubuntu-xenial http://docs-draft.openstack.org/61/482161/1/check/gate-designate-docs-ubuntu-xenial/ab6d610//doc/build/html/ : SUCCESS in 11m 39s\n- gate-designate-pep8-ubuntu-xenial http://logs.openstack.org/61/482161/1/check/gate-designate-pep8-ubuntu-xenial/292e0d6/ : SUCCESS in 11m 41s\n- gate-designate-python27-ubuntu-xenial http://logs.openstack.org/61/482161/1/check/gate-designate-python27-ubuntu-xenial/0eb4245/ : SUCCESS in 11m 39s\n- gate-designate-python35 http://logs.openstack.org/61/482161/1/check/gate-designate-python35/453407f/ : SUCCESS in 11m 39s\n- designate-coverage-ubuntu-xenial-nv http://logs.openstack.org/61/482161/1/check/designate-coverage-ubuntu-xenial-nv/28074da/ : SUCCESS in 11m 28s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"7924ba2eb510f36d3229e25217f881369b5b4427","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2017-07-10 14:57:13.000000000","message":"Patch Set 1: Verified-1\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":1},{"id":"20aa4e9bb3a9c42dc8c022c3804a582e388542b3","author":{"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},"date":"2017-07-10 15:11:49.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"8aea6d610134aa9e402625a8fbf18098854b8ba9","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2017-07-10 15:27:35.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- gate-designate-docs-ubuntu-xenial http://docs-draft.openstack.org/61/482161/2/check/gate-designate-docs-ubuntu-xenial/1377c33//doc/build/html/ : SUCCESS in 12m 52s\n- gate-designate-pep8-ubuntu-xenial http://logs.openstack.org/61/482161/2/check/gate-designate-pep8-ubuntu-xenial/8cf3627/ : SUCCESS in 13m 26s\n- gate-designate-python27-ubuntu-xenial http://logs.openstack.org/61/482161/2/check/gate-designate-python27-ubuntu-xenial/3988b40/ : SUCCESS in 7s\n- gate-designate-python35 http://logs.openstack.org/61/482161/2/check/gate-designate-python35/8aa62d2/ : SUCCESS in 6s\n- designate-coverage-ubuntu-xenial-nv http://logs.openstack.org/61/482161/2/check/designate-coverage-ubuntu-xenial-nv/3997ecf/ : SUCCESS in 2s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"f83adc120318f460d7e15faf36e18a6597922c9d","author":{"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},"date":"2017-07-10 16:24:28.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"159d3fdd21e03cec6d290f731c630ece43b13d0f","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2017-07-10 18:50:07.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- gate-designate-docs-ubuntu-xenial http://docs-draft.openstack.org/61/482161/3/check/gate-designate-docs-ubuntu-xenial/1ff92e8//doc/build/html/ : SUCCESS in 0s\n- gate-designate-pep8-ubuntu-xenial http://logs.openstack.org/61/482161/3/check/gate-designate-pep8-ubuntu-xenial/d59a0b4/ : SUCCESS in 0s\n- gate-designate-python27-ubuntu-xenial http://logs.openstack.org/61/482161/3/check/gate-designate-python27-ubuntu-xenial/3347d98/ : SUCCESS in 0s\n- gate-designate-python35 http://logs.openstack.org/61/482161/3/check/gate-designate-python35/d844e67/ : SUCCESS in 0s\n- gate-tempest-dsvm-designate-pdns4-ubuntu-xenial http://logs.openstack.org/61/482161/3/check/gate-tempest-dsvm-designate-pdns4-ubuntu-xenial/8309709/ : SUCCESS in 32m 19s\n- gate-tempest-dsvm-designate-bind9-ubuntu-xenial http://logs.openstack.org/61/482161/3/check/gate-tempest-dsvm-designate-bind9-ubuntu-xenial/c7ae142/ : SUCCESS in 17m 16s\n- gate-tempest-dsvm-designate-pdns4-postgres-ubuntu-xenial http://logs.openstack.org/61/482161/3/check/gate-tempest-dsvm-designate-pdns4-postgres-ubuntu-xenial/e375579/ : SUCCESS in 32m 21s\n- gate-tempest-dsvm-designate-pdns4-worker-model-ubuntu-xenial http://logs.openstack.org/61/482161/3/check/gate-tempest-dsvm-designate-pdns4-worker-model-ubuntu-xenial/a45cbf5/ : SUCCESS in 32m 20s\n- gate-rally-dsvm-designate-designate-ubuntu-xenial-nv http://logs.openstack.org/61/482161/3/check/gate-rally-dsvm-designate-designate-ubuntu-xenial-nv/8039e62/ : FAILURE in 5m 08s (non-voting)\n- gate-grenade-dsvm-designate-pdns4-ubuntu-xenial http://logs.openstack.org/61/482161/3/check/gate-grenade-dsvm-designate-pdns4-ubuntu-xenial/3217fb9/ : SUCCESS in 1h 02m 17s\n- gate-tempest-dsvm-designate-pdns4-identity-v3-only-ubuntu-xenial http://logs.openstack.org/61/482161/3/check/gate-tempest-dsvm-designate-pdns4-identity-v3-only-ubuntu-xenial/26e261d/ : SUCCESS in 16m 45s\n- designate-coverage-ubuntu-xenial-nv http://logs.openstack.org/61/482161/3/check/designate-coverage-ubuntu-xenial-nv/7985ae4/ : SUCCESS in 0s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"f1620c6f1039125ecf21e1dc2730f369dc014358","author":{"_account_id":19930,"name":"Dai Dang Van","email":"daidv@omzcloud.vn","username":"daidv"},"date":"2017-07-12 04:21:33.000000000","message":"Patch Set 3: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"d0cd99f4debc19bb1deff0129455e6e74f656965","author":{"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},"date":"2017-07-12 13:53:53.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"f87332e20c687b377d16cb8bdeec001b5ca13714","author":{"_account_id":24441,"name":"Cuong Nguyen","email":"cuongnv@vn.fujitsu.com","username":"cuongnv"},"date":"2017-07-19 09:46:26.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"a87e28d4f00ce2065d56b51b9ce4436d0a20fd39","author":{"_account_id":25254,"name":"Nguyen Van Trung","email":"trungnvfet@outlook.com","username":"trungnv"},"date":"2017-07-19 15:03:19.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"d918bd0939ac3e23f7ecd1a6b00c9e98e723f5a1","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2017-07-20 18:49:12.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"9258223796364bc98af5642bcec73ba03b78db2c","author":{"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},"date":"2019-04-01 14:03:25.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"2ddbc68a9e6e1bffd318586619baefba6ee7bc41","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-01 15:03:41.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- designate-bind9 http://logs.openstack.org/61/482161/4/check/designate-bind9/a6e0f19/ : SUCCESS in 39m 26s\n- designate-bind9-manager-model http://logs.openstack.org/61/482161/4/check/designate-bind9-manager-model/247e618/ : SUCCESS in 41m 37s\n- designate-bind9-py36 http://logs.openstack.org/61/482161/4/check/designate-bind9-py36/0c75122/ : SUCCESS in 37m 54s\n- designate-devstack-pdns4 http://logs.openstack.org/61/482161/4/check/designate-devstack-pdns4/f20c990/ : SUCCESS in 33m 38s\n- designate-devstack-pdns4-postgres http://logs.openstack.org/61/482161/4/check/designate-devstack-pdns4-postgres/bb4db25/ : SUCCESS in 38m 12s\n- designate-grenade-pdns4 http://logs.openstack.org/61/482161/4/check/designate-grenade-pdns4/5d02bb4/ : SUCCESS in 53m 41s\n- openstack-tox-cover http://logs.openstack.org/61/482161/4/check/openstack-tox-cover/8283d77/cover/ : SUCCESS in 7m 41s\n- openstack-tox-lower-constraints http://logs.openstack.org/61/482161/4/check/openstack-tox-lower-constraints/50e9216/ : SUCCESS in 6m 08s\n- openstack-tox-pep8 http://logs.openstack.org/61/482161/4/check/openstack-tox-pep8/7ca050e/ : SUCCESS in 7m 00s\n- openstack-tox-py27 http://logs.openstack.org/61/482161/4/check/openstack-tox-py27/07944a8/ : SUCCESS in 6m 33s\n- openstack-tox-py35 http://logs.openstack.org/61/482161/4/check/openstack-tox-py35/8dd5530/ : SUCCESS in 7m 18s\n- openstack-tox-py36 http://logs.openstack.org/61/482161/4/check/openstack-tox-py36/6931bd0/ : SUCCESS in 6m 42s\n- openstack-tox-py37 http://logs.openstack.org/61/482161/4/check/openstack-tox-py37/52d226e/ : SUCCESS in 8m 19s\n- openstack-tox-docs http://logs.openstack.org/61/482161/4/check/openstack-tox-docs/5e1a353/html/ : SUCCESS in 6m 00s\n- neutron-tempest-plugin-designate-scenario http://logs.openstack.org/61/482161/4/check/neutron-tempest-plugin-designate-scenario/fddef0b/ : FAILURE in 15m 12s","accounts_in_message":[],"_revision_number":4},{"id":"c854c423dbaee73a497510084bc0f004593c1b8f","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2019-04-01 17:54:53.000000000","message":"Patch Set 4: Code-Review+2","accounts_in_message":[],"_revision_number":4},{"id":"da59f5e14410028433bbf6625cfc6bab8a67cba5","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2019-04-02 11:59:24.000000000","message":"Patch Set 4:\n\nrecheck","accounts_in_message":[],"_revision_number":4},{"id":"abb1f1aa7803fed9915c332cddb3bfa55f8d4237","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-02 13:10:48.000000000","message":"Patch Set 4:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- designate-bind9 http://logs.openstack.org/61/482161/4/check/designate-bind9/7cba70d/ : SUCCESS in 37m 26s\n- designate-pdns4 designate-pdns4 : ERROR Unable to find playbook /var/lib/zuul/builds/56ecea086d7f4a2fa1d7dd6d3ae67895/untrusted/project_6/git.openstack.org/openstack/designate/playbooks/designate-pdns4/post.yaml\n- designate-bind9-manager-model http://logs.openstack.org/61/482161/4/check/designate-bind9-manager-model/bb842fc/ : SUCCESS in 45m 33s\n- designate-bind9-py27 designate-bind9-py27 : ERROR Unable to find playbook /var/lib/zuul/builds/fae3f348cf9940e09de0c537124e3ea1/untrusted/project_6/git.openstack.org/openstack/designate/playbooks/designate-bind9/post.yaml\n- designate-pdns4-py27 designate-pdns4-py27 : ERROR Unable to find playbook /var/lib/zuul/builds/cc7c84ff6a58422aa10ddd2ffd8e67eb/untrusted/project_6/git.openstack.org/openstack/designate/playbooks/designate-pdns4/post.yaml\n- designate-pdns4-postgres designate-pdns4-postgres : ERROR Unable to find playbook /var/lib/zuul/builds/22b1c75257074c65bc32d5fe20d96d15/untrusted/project_6/git.openstack.org/openstack/designate/playbooks/designate-pdns4/post.yaml\n- designate-grenade-pdns4 http://logs.openstack.org/61/482161/4/check/designate-grenade-pdns4/ac1133a/ : FAILURE in 1h 05m 29s\n- openstack-tox-cover http://logs.openstack.org/61/482161/4/check/openstack-tox-cover/7c77674/cover/ : SUCCESS in 7m 17s\n- openstack-tox-lower-constraints http://logs.openstack.org/61/482161/4/check/openstack-tox-lower-constraints/b3d3197/ : SUCCESS in 4m 56s\n- openstack-tox-pep8 http://logs.openstack.org/61/482161/4/check/openstack-tox-pep8/97daf61/ : SUCCESS in 11m 27s\n- openstack-tox-py27 http://logs.openstack.org/61/482161/4/check/openstack-tox-py27/fe0f7ee/ : SUCCESS in 6m 40s\n- openstack-tox-py35 http://logs.openstack.org/61/482161/4/check/openstack-tox-py35/26a406b/ : SUCCESS in 8m 19s\n- openstack-tox-py36 http://logs.openstack.org/61/482161/4/check/openstack-tox-py36/775a3a1/ : SUCCESS in 7m 34s\n- openstack-tox-py37 http://logs.openstack.org/61/482161/4/check/openstack-tox-py37/cbb1137/ : SUCCESS in 7m 59s\n- openstack-tox-docs http://logs.openstack.org/61/482161/4/check/openstack-tox-docs/6d46938/html/ : SUCCESS in 5m 42s\n- neutron-tempest-plugin-designate-scenario http://logs.openstack.org/61/482161/4/check/neutron-tempest-plugin-designate-scenario/c20f134/ : SUCCESS in 31m 40s","accounts_in_message":[],"_revision_number":4},{"id":"22d28e9d0c5cf5b7262a39b076fd64a96e380567","author":{"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},"date":"2019-04-02 13:38:07.000000000","message":"Patch Set 4:\n\nrecheck","accounts_in_message":[],"_revision_number":4},{"id":"e672f52f89472ebce366e597c6a8233e4066baca","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-02 14:56:16.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- designate-bind9 http://logs.openstack.org/61/482161/4/check/designate-bind9/89955e7/ : SUCCESS in 38m 59s\n- designate-pdns4 http://logs.openstack.org/61/482161/4/check/designate-pdns4/2a7b902/ : SUCCESS in 39m 40s\n- designate-bind9-manager-model http://logs.openstack.org/61/482161/4/check/designate-bind9-manager-model/2b36468/ : SUCCESS in 40m 47s\n- designate-bind9-py27 http://logs.openstack.org/61/482161/4/check/designate-bind9-py27/30f8720/ : SUCCESS in 1h 08m 13s\n- designate-pdns4-py27 http://logs.openstack.org/61/482161/4/check/designate-pdns4-py27/8ce5bc1/ : SUCCESS in 43m 50s\n- designate-pdns4-postgres http://logs.openstack.org/61/482161/4/check/designate-pdns4-postgres/1fc6d08/ : SUCCESS in 44m 23s\n- designate-grenade-pdns4 http://logs.openstack.org/61/482161/4/check/designate-grenade-pdns4/bb5a5e6/ : SUCCESS in 1h 00m 33s\n- openstack-tox-cover http://logs.openstack.org/61/482161/4/check/openstack-tox-cover/bd78ff9/cover/ : SUCCESS in 7m 40s\n- openstack-tox-lower-constraints http://logs.openstack.org/61/482161/4/check/openstack-tox-lower-constraints/6fbd227/ : SUCCESS in 6m 19s\n- openstack-tox-pep8 http://logs.openstack.org/61/482161/4/check/openstack-tox-pep8/d80a84b/ : SUCCESS in 5m 55s\n- openstack-tox-py27 http://logs.openstack.org/61/482161/4/check/openstack-tox-py27/d268156/ : SUCCESS in 7m 03s\n- openstack-tox-py35 http://logs.openstack.org/61/482161/4/check/openstack-tox-py35/9f03e19/ : SUCCESS in 7m 43s\n- openstack-tox-py36 http://logs.openstack.org/61/482161/4/check/openstack-tox-py36/9ad3834/ : SUCCESS in 7m 37s\n- openstack-tox-py37 http://logs.openstack.org/61/482161/4/check/openstack-tox-py37/0adedd2/ : SUCCESS in 7m 36s\n- openstack-tox-docs http://logs.openstack.org/61/482161/4/check/openstack-tox-docs/1446c8e/html/ : SUCCESS in 5m 58s\n- neutron-tempest-plugin-designate-scenario http://logs.openstack.org/61/482161/4/check/neutron-tempest-plugin-designate-scenario/c3b681c/ : SUCCESS in 38m 48s","accounts_in_message":[],"_revision_number":4},{"id":"0e3ecebf93242f41df57f3a45bd92af83e617e73","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2019-05-20 03:18:51.000000000","message":"Patch Set 4: Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"953dfcd507de5ce0a1ed3cd7ba2535512afdd571","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-20 03:19:03.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"2f5ccea11f28156c122e652105f4a3809a83773c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-20 04:18:21.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":4},{"id":"6b0ef118cdd85d6610bb735a92fbf49b4f91b102","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-20 04:18:21.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- designate-bind9 http://logs.openstack.org/61/482161/4/gate/designate-bind9/4918ca6/ : SUCCESS in 35m 43s\n- designate-pdns4 http://logs.openstack.org/61/482161/4/gate/designate-pdns4/f9b00a8/ : SUCCESS in 28m 52s\n- designate-bind9-manager-model http://logs.openstack.org/61/482161/4/gate/designate-bind9-manager-model/b5cda1d/ : SUCCESS in 37m 26s\n- designate-bind9-py27 http://logs.openstack.org/61/482161/4/gate/designate-bind9-py27/8541c6f/ : SUCCESS in 35m 55s\n- designate-pdns4-py27 http://logs.openstack.org/61/482161/4/gate/designate-pdns4-py27/962e108/ : SUCCESS in 40m 03s\n- designate-pdns4-postgres http://logs.openstack.org/61/482161/4/gate/designate-pdns4-postgres/3c64b0b/ : SUCCESS in 29m 30s\n- designate-grenade-pdns4 http://logs.openstack.org/61/482161/4/gate/designate-grenade-pdns4/8dc0bd0/ : SUCCESS in 56m 05s\n- openstack-tox-lower-constraints http://logs.openstack.org/61/482161/4/gate/openstack-tox-lower-constraints/1a502b8/ : SUCCESS in 4m 47s\n- openstack-tox-pep8 http://logs.openstack.org/61/482161/4/gate/openstack-tox-pep8/8031e29/ : SUCCESS in 4m 59s\n- openstack-tox-py27 http://logs.openstack.org/61/482161/4/gate/openstack-tox-py27/dcf05e8/ : SUCCESS in 5m 07s\n- openstack-tox-py36 http://logs.openstack.org/61/482161/4/gate/openstack-tox-py36/5ff577c/ : SUCCESS in 5m 11s\n- openstack-tox-py37 http://logs.openstack.org/61/482161/4/gate/openstack-tox-py37/7116eec/ : SUCCESS in 6m 03s\n- openstack-tox-docs http://logs.openstack.org/61/482161/4/gate/openstack-tox-docs/4518e71/html/ : SUCCESS in 4m 51s","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"3fbd9dea9dfc1acbf045438459e267c927f70f11","revisions":{"b01ed9b55aea1089e7d3e29b6d9bec2e0818d323":{"kind":"REWORK","_number":1,"created":"2017-07-10 14:38:35.000000000","uploader":{"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},"ref":"refs/changes/61/482161/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/61/482161/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/61/482161/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/61/482161/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/61/482161/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/61/482161/1"}}},"commit":{"parents":[{"commit":"3c323d5eab31920733fd55cc1a73076ddf08dcd5","subject":"Merge \"Make the Zone object py3 compatible\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/3c323d5eab31920733fd55cc1a73076ddf08dcd5"}]}],"author":{"name":"Graham Hayes","email":"graham.hayes@suse.com","date":"2017-07-10 14:38:07.000000000","tz":60},"committer":{"name":"Graham Hayes","email":"graham.hayes@suse.com","date":"2017-07-10 14:38:15.000000000","tz":60},"subject":"Update Driver statuses","message":"Update Driver statuses\n\nChange-Id: Iac78b8668bcdca34bf354d6feab3e9547ba2c71f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/b01ed9b55aea1089e7d3e29b6d9bec2e0818d323"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/b01ed9b55aea1089e7d3e29b6d9bec2e0818d323"}]},"branch":"refs/heads/master"},"656c00b5e774aaacac2872ca3652c983a79ee9da":{"kind":"REWORK","_number":2,"created":"2017-07-10 15:11:49.000000000","uploader":{"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},"ref":"refs/changes/61/482161/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/61/482161/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/61/482161/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/61/482161/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/61/482161/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/61/482161/2"}}},"commit":{"parents":[{"commit":"84040e07edf6a5aa36ce4a12933e508c62ee2b73","subject":"rearrange existing documentation according to the new standard layout","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/84040e07edf6a5aa36ce4a12933e508c62ee2b73"}]}],"author":{"name":"Graham Hayes","email":"graham.hayes@suse.com","date":"2017-07-10 14:38:07.000000000","tz":60},"committer":{"name":"Graham Hayes","email":"graham.hayes@suse.com","date":"2017-07-10 15:11:28.000000000","tz":60},"subject":"Update Driver statuses","message":"Update Driver statuses\n\nChange-Id: Iac78b8668bcdca34bf354d6feab3e9547ba2c71f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/656c00b5e774aaacac2872ca3652c983a79ee9da"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/656c00b5e774aaacac2872ca3652c983a79ee9da"}]},"branch":"refs/heads/master"},"f882d96761a54dbc70b6794950d39a9d15e95e6c":{"kind":"REWORK","_number":3,"created":"2017-07-10 16:24:28.000000000","uploader":{"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},"ref":"refs/changes/61/482161/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/61/482161/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/61/482161/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/61/482161/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/61/482161/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/61/482161/3"}}},"commit":{"parents":[{"commit":"84040e07edf6a5aa36ce4a12933e508c62ee2b73","subject":"rearrange existing documentation according to the new standard layout","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/84040e07edf6a5aa36ce4a12933e508c62ee2b73"}]}],"author":{"name":"Graham Hayes","email":"graham.hayes@suse.com","date":"2017-07-10 14:38:07.000000000","tz":60},"committer":{"name":"Graham Hayes","email":"graham.hayes@suse.com","date":"2017-07-10 16:24:08.000000000","tz":60},"subject":"Update Driver statuses","message":"Update Driver statuses\n\nChange-Id: Iac78b8668bcdca34bf354d6feab3e9547ba2c71f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/f882d96761a54dbc70b6794950d39a9d15e95e6c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/f882d96761a54dbc70b6794950d39a9d15e95e6c"}]},"branch":"refs/heads/master"},"3fbd9dea9dfc1acbf045438459e267c927f70f11":{"kind":"REWORK","_number":4,"created":"2019-04-01 14:03:25.000000000","uploader":{"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},"ref":"refs/changes/61/482161/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/61/482161/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/61/482161/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/61/482161/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/61/482161/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/61/482161/4"}}},"commit":{"parents":[{"commit":"33ca0b87891228d86e440e929feeb324353b39d3","subject":"Merge \"Allow pdns4 backend to select tsigkey\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/33ca0b87891228d86e440e929feeb324353b39d3"}]}],"author":{"name":"Graham Hayes","email":"graham.hayes@suse.com","date":"2017-07-10 14:38:07.000000000","tz":60},"committer":{"name":"Graham Hayes","email":"gr@ham.ie","date":"2019-04-01 14:03:09.000000000","tz":60},"subject":"Update Driver statuses","message":"Update Driver statuses\n\nChange-Id: Iac78b8668bcdca34bf354d6feab3e9547ba2c71f\nSigned-off-by: Graham Hayes \u003cgr@ham.ie\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/3fbd9dea9dfc1acbf045438459e267c927f70f11"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/3fbd9dea9dfc1acbf045438459e267c927f70f11"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
