)]}'
{"id":"openstack%2Fdesignate~555398","triplet_id":"openstack%2Fdesignate~master~I159d0732688ddf1337ab3602a84a43fd043dcaa2","project":"openstack/designate","branch":"master","topic":"bug/1755788","hashtags":[],"change_id":"I159d0732688ddf1337ab3602a84a43fd043dcaa2","subject":"Check TXT/SPF records for RFC1035 sec. 5.1","status":"MERGED","created":"2018-03-22 18:05:31.000000000","updated":"2019-05-20 04:18:20.000000000","submitted":"2019-05-20 04:18:20.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"555398-1558325900332-172dd170","meta_rev_id":"7f870ef8915e57cb36e4ac74cc03bcdc4134432f","_number":555398,"virtual_id_number":555398,"owner":{"_account_id":20104,"name":"Dmitrii Galkin","display_name":"Dmitry Galkin","email":"galkindmitrii@gmail.com","username":"galkindmitrii"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":18418,"name":"Andy Botting","email":"andy@andybotting.com","username":"andybotting"},{"value":0,"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":20104,"name":"Dmitrii Galkin","display_name":"Dmitry Galkin","email":"galkindmitrii@gmail.com","username":"galkindmitrii"},{"value":0,"_account_id":15736,"name":"Ritesh Anand","email":"tellrit@gmail.com","username":"ritesh"},{"value":2,"date":"2019-05-20 04:18:20.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},"all":[{"value":0,"_account_id":18418,"name":"Andy Botting","email":"andy@andybotting.com","username":"andybotting"},{"value":2,"date":"2019-03-25 16:04:00.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":20104,"name":"Dmitrii Galkin","display_name":"Dmitry Galkin","email":"galkindmitrii@gmail.com","username":"galkindmitrii"},{"value":0,"_account_id":15736,"name":"Ritesh Anand","email":"tellrit@gmail.com","username":"ritesh"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2019-05-20 03:18:03.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"all":[{"value":0,"_account_id":18418,"name":"Andy Botting","email":"andy@andybotting.com","username":"andybotting"},{"value":0,"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":20104,"name":"Dmitrii Galkin","display_name":"Dmitry Galkin","email":"galkindmitrii@gmail.com","username":"galkindmitrii"},{"value":0,"_account_id":15736,"name":"Ritesh Anand","email":"tellrit@gmail.com","username":"ritesh"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2019-05-20 03:18:03.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"recommended":{"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},"all":[{"value":0,"_account_id":18418,"name":"Andy Botting","email":"andy@andybotting.com","username":"andybotting"},{"value":1,"date":"2019-03-25 16:04:00.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":20104,"name":"Dmitrii Galkin","display_name":"Dmitry Galkin","email":"galkindmitrii@gmail.com","username":"galkindmitrii"},{"value":0,"_account_id":15736,"name":"Ritesh Anand","email":"tellrit@gmail.com","username":"ritesh"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","value":1,"default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":15736,"name":"Ritesh Anand","email":"tellrit@gmail.com","username":"ritesh"},{"_account_id":18418,"name":"Andy Botting","email":"andy@andybotting.com","username":"andybotting"},{"_account_id":20104,"name":"Dmitrii Galkin","display_name":"Dmitry Galkin","email":"galkindmitrii@gmail.com","username":"galkindmitrii"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2018-06-18 02:30:12.000000000","updated_by":{"_account_id":18418,"name":"Andy Botting","email":"andy@andybotting.com","username":"andybotting"},"reviewer":{"_account_id":18418,"name":"Andy Botting","email":"andy@andybotting.com","username":"andybotting"},"state":"REVIEWER"},{"updated":"2019-02-12 22:53:01.000000000","updated_by":{"_account_id":15736,"name":"Ritesh Anand","email":"tellrit@gmail.com","username":"ritesh"},"reviewer":{"_account_id":15736,"name":"Ritesh Anand","email":"tellrit@gmail.com","username":"ritesh"},"state":"REVIEWER"},{"updated":"2019-03-07 12:50:58.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"REVIEWER"},{"updated":"2019-03-25 16:04:00.000000000","updated_by":{"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},"reviewer":{"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},"state":"REVIEWER"},{"updated":"2019-05-20 03:18:03.000000000","updated_by":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"reviewer":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"state":"REVIEWER"},{"updated":"2019-05-20 04:18:20.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"d4c80aaf2db2be37081baa76003da5e4207b96f5","author":{"_account_id":20104,"name":"Dmitrii Galkin","display_name":"Dmitry Galkin","email":"galkindmitrii@gmail.com","username":"galkindmitrii"},"date":"2018-03-22 18:05:31.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"7e282fad8017824b9e457b20d9130cb609dca6f5","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-03-22 18:56:19.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 http://logs.openstack.org/98/555398/1/check/openstack-tox-pep8/de74850/ : FAILURE in 4m 59s\n- openstack-tox-py27 http://logs.openstack.org/98/555398/1/check/openstack-tox-py27/fa958f6/ : SUCCESS in 5m 06s\n- openstack-tox-py35 http://logs.openstack.org/98/555398/1/check/openstack-tox-py35/fea9112/ : SUCCESS in 5m 10s\n- build-openstack-sphinx-docs http://logs.openstack.org/98/555398/1/check/build-openstack-sphinx-docs/3ee8002/html/ : SUCCESS in 4m 53s\n- openstack-tox-cover http://logs.openstack.org/98/555398/1/check/openstack-tox-cover/e796641/cover/ : SUCCESS in 5m 39s (non-voting)\n- designate-devstack-bind9 http://logs.openstack.org/98/555398/1/check/designate-devstack-bind9/a8ab453/ : FAILURE in 23m 48s\n- designate-devstack-pdns4 http://logs.openstack.org/98/555398/1/check/designate-devstack-pdns4/c285917/ : FAILURE in 28m 10s\n- designate-devstack-pdns4-postgres http://logs.openstack.org/98/555398/1/check/designate-devstack-pdns4-postgres/a41ffb2/ : FAILURE in 29m 34s\n- designate-devstack-pdns4-manager-model http://logs.openstack.org/98/555398/1/check/designate-devstack-pdns4-manager-model/520cf57/ : FAILURE in 28m 06s\n- designate-devstack-pdns4-py35 http://logs.openstack.org/98/555398/1/check/designate-devstack-pdns4-py35/e6c4bbb/ : FAILURE in 26m 02s\n- designate-grenade-pdns4 http://logs.openstack.org/98/555398/1/check/designate-grenade-pdns4/5e463f3/ : FAILURE in 49m 25s\n- neutron-tempest-plugin-designate-scenario http://logs.openstack.org/98/555398/1/check/neutron-tempest-plugin-designate-scenario/4c722f3/ : SUCCESS in 25m 02s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"4d98736d63d286685e5198be0bf89abb80a7b670","author":{"_account_id":20104,"name":"Dmitrii Galkin","display_name":"Dmitry Galkin","email":"galkindmitrii@gmail.com","username":"galkindmitrii"},"date":"2018-03-26 11:54:21.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"85e1e98767d047f6b0c05cbcc5b1712f2070cf2d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-03-26 12:49:50.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 http://logs.openstack.org/98/555398/2/check/openstack-tox-pep8/fde60f7/ : SUCCESS in 4m 52s\n- openstack-tox-py27 http://logs.openstack.org/98/555398/2/check/openstack-tox-py27/e7bb738/ : SUCCESS in 5m 11s\n- openstack-tox-py35 http://logs.openstack.org/98/555398/2/check/openstack-tox-py35/6d520bc/ : SUCCESS in 5m 11s\n- build-openstack-sphinx-docs http://logs.openstack.org/98/555398/2/check/build-openstack-sphinx-docs/03feffc/html/ : SUCCESS in 4m 52s\n- openstack-tox-cover http://logs.openstack.org/98/555398/2/check/openstack-tox-cover/6485828/cover/ : SUCCESS in 5m 42s (non-voting)\n- designate-devstack-bind9 http://logs.openstack.org/98/555398/2/check/designate-devstack-bind9/5543fab/ : FAILURE in 24m 56s\n- designate-devstack-pdns4 http://logs.openstack.org/98/555398/2/check/designate-devstack-pdns4/352affe/ : FAILURE in 29m 47s\n- designate-devstack-pdns4-postgres http://logs.openstack.org/98/555398/2/check/designate-devstack-pdns4-postgres/417b7b3/ : FAILURE in 23m 45s\n- designate-devstack-pdns4-manager-model http://logs.openstack.org/98/555398/2/check/designate-devstack-pdns4-manager-model/40bb58e/ : FAILURE in 30m 38s\n- designate-devstack-pdns4-py35 http://logs.openstack.org/98/555398/2/check/designate-devstack-pdns4-py35/e1f283b/ : FAILURE in 25m 56s\n- designate-grenade-pdns4 http://logs.openstack.org/98/555398/2/check/designate-grenade-pdns4/7b29cbb/ : FAILURE in 49m 54s\n- neutron-tempest-plugin-designate-scenario http://logs.openstack.org/98/555398/2/check/neutron-tempest-plugin-designate-scenario/c97a618/ : SUCCESS in 23m 22s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"1dec0920fa378c98ff1f4f8923698796d6882ab6","author":{"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},"date":"2018-03-26 13:01:13.000000000","message":"Patch Set 2: Code-Review-1\n\nThis sort of validation should happen in the objects (the rrdata_txt.py file)\n\nLook at how the recordset and zone objects override the \"validate()\" method to insert custom logic, and then call the parent class validate at the end.","accounts_in_message":[],"_revision_number":2},{"id":"03855ac925a96acc09ea036d58094bcd36b80c64","author":{"_account_id":20104,"name":"Dmitrii Galkin","display_name":"Dmitry Galkin","email":"galkindmitrii@gmail.com","username":"galkindmitrii"},"date":"2018-04-09 15:12:53.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"82dec7907425fa13c55c4c1e13292a4a3b839100","author":{"_account_id":20104,"name":"Dmitrii Galkin","display_name":"Dmitry Galkin","email":"galkindmitrii@gmail.com","username":"galkindmitrii"},"date":"2018-04-09 15:17:35.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"244369cca4ec8ccc5dea74e1a0771a152bb319f7","author":{"_account_id":20104,"name":"Dmitrii Galkin","display_name":"Dmitry Galkin","email":"galkindmitrii@gmail.com","username":"galkindmitrii"},"date":"2018-04-09 15:28:40.000000000","message":"Patch Set 4:\n\n- \u003e Look at how the recordset and zone objects override the \"validate()\" method to insert custom logic, and then call the parent class validate at the end.\n\nSo, doing like super(TXT, self).validate() does not result in validation method being called. In this patch set it is called explicitly from recordset.py Any hints?","accounts_in_message":[],"_revision_number":4},{"id":"7046db45fbd75c2c1216818560613eff065ea4a1","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-04-09 16:05:58.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 http://logs.openstack.org/98/555398/4/check/openstack-tox-pep8/709301b/ : SUCCESS in 4m 51s\n- openstack-tox-py27 http://logs.openstack.org/98/555398/4/check/openstack-tox-py27/338d105/ : SUCCESS in 6m 09s\n- openstack-tox-py35 http://logs.openstack.org/98/555398/4/check/openstack-tox-py35/5affe8b/ : SUCCESS in 5m 40s\n- build-openstack-sphinx-docs http://logs.openstack.org/98/555398/4/check/build-openstack-sphinx-docs/2214566/html/ : SUCCESS in 5m 15s\n- openstack-tox-cover http://logs.openstack.org/98/555398/4/check/openstack-tox-cover/0e0b177/cover/ : SUCCESS in 6m 19s (non-voting)\n- designate-devstack-bind9 http://logs.openstack.org/98/555398/4/check/designate-devstack-bind9/806f19e/ : FAILURE in 28m 24s\n- designate-devstack-pdns4 http://logs.openstack.org/98/555398/4/check/designate-devstack-pdns4/c469247/ : FAILURE in 26m 15s\n- designate-devstack-pdns4-postgres http://logs.openstack.org/98/555398/4/check/designate-devstack-pdns4-postgres/13b8b6b/ : FAILURE in 23m 08s\n- designate-devstack-pdns4-manager-model http://logs.openstack.org/98/555398/4/check/designate-devstack-pdns4-manager-model/ceb4e2b/ : FAILURE in 32m 25s\n- designate-devstack-pdns4-py35 http://logs.openstack.org/98/555398/4/check/designate-devstack-pdns4-py35/04c50bd/ : FAILURE in 32m 55s\n- designate-grenade-pdns4 http://logs.openstack.org/98/555398/4/check/designate-grenade-pdns4/30c1479/ : FAILURE in 45m 23s\n- neutron-tempest-plugin-designate-scenario http://logs.openstack.org/98/555398/4/check/neutron-tempest-plugin-designate-scenario/f754489/ : SUCCESS in 27m 41s (non-voting)\n- openstack-tox-lower-constraints http://logs.openstack.org/98/555398/4/check/openstack-tox-lower-constraints/acf48b6/ : SUCCESS in 4m 47s","accounts_in_message":[],"_revision_number":4},{"id":"3fed13e19b23ecf320f6f0652fe381b7e0369c4c","author":{"_account_id":20104,"name":"Dmitrii Galkin","display_name":"Dmitry Galkin","email":"galkindmitrii@gmail.com","username":"galkindmitrii"},"date":"2018-05-03 15:47:21.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"100dcdc541e348ca88a480412f00c854bda78512","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-05-03 16:49:39.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 http://logs.openstack.org/98/555398/5/check/openstack-tox-pep8/ab93d0a/ : SUCCESS in 5m 06s\n- openstack-tox-py27 http://logs.openstack.org/98/555398/5/check/openstack-tox-py27/b1fde6f/ : SUCCESS in 5m 14s\n- openstack-tox-py35 http://logs.openstack.org/98/555398/5/check/openstack-tox-py35/12da7c2/ : SUCCESS in 5m 46s\n- build-openstack-sphinx-docs http://logs.openstack.org/98/555398/5/check/build-openstack-sphinx-docs/d3f0971/html/ : SUCCESS in 5m 44s\n- openstack-tox-cover http://logs.openstack.org/98/555398/5/check/openstack-tox-cover/2bd34dc/cover/ : SUCCESS in 6m 13s (non-voting)\n- designate-devstack-bind9 http://logs.openstack.org/98/555398/5/check/designate-devstack-bind9/12ee108/ : FAILURE in 26m 01s\n- designate-devstack-pdns4 http://logs.openstack.org/98/555398/5/check/designate-devstack-pdns4/06fb300/ : FAILURE in 30m 59s\n- designate-devstack-pdns4-postgres http://logs.openstack.org/98/555398/5/check/designate-devstack-pdns4-postgres/bcb5a4b/ : FAILURE in 37m 00s\n- designate-devstack-pdns4-manager-model http://logs.openstack.org/98/555398/5/check/designate-devstack-pdns4-manager-model/483cdfd/ : FAILURE in 25m 16s\n- designate-devstack-pdns4-py35 http://logs.openstack.org/98/555398/5/check/designate-devstack-pdns4-py35/de7a457/ : FAILURE in 26m 40s\n- designate-grenade-pdns4 http://logs.openstack.org/98/555398/5/check/designate-grenade-pdns4/28539a2/ : FAILURE in 56m 12s\n- neutron-tempest-plugin-designate-scenario http://logs.openstack.org/98/555398/5/check/neutron-tempest-plugin-designate-scenario/ab84b98/ : SUCCESS in 29m 14s (non-voting)\n- openstack-tox-lower-constraints http://logs.openstack.org/98/555398/5/check/openstack-tox-lower-constraints/7d1b183/ : SUCCESS in 4m 33s","accounts_in_message":[],"_revision_number":5},{"id":"5d1fcac435771f0a7d7f3439cfe001befa4613ff","author":{"_account_id":18418,"name":"Andy Botting","email":"andy@andybotting.com","username":"andybotting"},"date":"2018-06-18 02:30:12.000000000","message":"Patch Set 5: Code-Review+1\n\nWas just bitten by this!","accounts_in_message":[],"_revision_number":5},{"id":"4ff591af6ca679935ec26c1e365b737d70e61d73","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2018-07-22 03:01:00.000000000","message":"Patch Set 5:\n\nrecheck","accounts_in_message":[],"_revision_number":5},{"id":"e2697324273e51a3ab3e00a294b75fbaf11a748e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-07-22 03:48:19.000000000","message":"Patch Set 5:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 http://logs.openstack.org/98/555398/5/check/openstack-tox-pep8/605bb55/ : SUCCESS in 5m 04s\n- openstack-tox-py27 http://logs.openstack.org/98/555398/5/check/openstack-tox-py27/61d72ad/ : SUCCESS in 5m 29s\n- openstack-tox-py35 http://logs.openstack.org/98/555398/5/check/openstack-tox-py35/1046aee/ : SUCCESS in 5m 23s\n- build-openstack-sphinx-docs http://logs.openstack.org/98/555398/5/check/build-openstack-sphinx-docs/a5c2e14/html/ : SUCCESS in 5m 06s\n- openstack-tox-cover http://logs.openstack.org/98/555398/5/check/openstack-tox-cover/2c0cbb8/ : POST_FAILURE in 5m 46s (non-voting)\n- designate-bind9 http://logs.openstack.org/98/555398/5/check/designate-bind9/ff961ba/ : FAILURE in 29m 25s\n- designate-bind9-manager-model http://logs.openstack.org/98/555398/5/check/designate-bind9-manager-model/0cf636d/ : FAILURE in 27m 04s\n- designate-bind9-py35 http://logs.openstack.org/98/555398/5/check/designate-bind9-py35/43e4dd9/ : FAILURE in 35m 34s\n- designate-bind9-py36 http://logs.openstack.org/98/555398/5/check/designate-bind9-py36/f6b2a1a/ : FAILURE in 29m 31s (non-voting)\n- designate-devstack-pdns4 http://logs.openstack.org/98/555398/5/check/designate-devstack-pdns4/2504936/ : FAILURE in 27m 34s\n- designate-devstack-pdns4-postgres http://logs.openstack.org/98/555398/5/check/designate-devstack-pdns4-postgres/062cc1e/ : FAILURE in 26m 15s\n- designate-grenade-pdns4 http://logs.openstack.org/98/555398/5/check/designate-grenade-pdns4/0a3209d/ : FAILURE in 45m 22s\n- neutron-tempest-plugin-designate-scenario http://logs.openstack.org/98/555398/5/check/neutron-tempest-plugin-designate-scenario/146c916/ : SUCCESS in 32m 27s\n- openstack-tox-lower-constraints http://logs.openstack.org/98/555398/5/check/openstack-tox-lower-constraints/ecc09ed/ : SUCCESS in 3m 58s","accounts_in_message":[],"_revision_number":5},{"id":"dd4425762635211bd41d13f623d288c4ee249e3c","author":{"_account_id":18418,"name":"Andy Botting","email":"andy@andybotting.com","username":"andybotting"},"date":"2018-11-13 15:32:46.000000000","message":"Patch Set 5:\n\nrecheck","accounts_in_message":[],"_revision_number":5},{"id":"7d269bb25c1a42d4bea465b5f98925f74bf691c3","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-11-13 16:31:18.000000000","message":"Patch Set 5:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- designate-bind9 http://logs.openstack.org/98/555398/5/check/designate-bind9/531b5d4/ : FAILURE in 35m 17s\n- designate-bind9-manager-model http://logs.openstack.org/98/555398/5/check/designate-bind9-manager-model/89f3d16/ : FAILURE in 37m 53s\n- designate-bind9-py35 http://logs.openstack.org/98/555398/5/check/designate-bind9-py35/24fdbd6/ : FAILURE in 30m 09s\n- designate-bind9-py36 http://logs.openstack.org/98/555398/5/check/designate-bind9-py36/02b9b69/ : FAILURE in 32m 53s\n- designate-devstack-pdns4 http://logs.openstack.org/98/555398/5/check/designate-devstack-pdns4/4bc4a47/ : FAILURE in 29m 27s\n- designate-devstack-pdns4-postgres http://logs.openstack.org/98/555398/5/check/designate-devstack-pdns4-postgres/5e9c55e/ : FAILURE in 53m 44s\n- designate-grenade-pdns4 http://logs.openstack.org/98/555398/5/check/designate-grenade-pdns4/04129ae/ : FAILURE in 47m 57s\n- openstack-tox-cover http://logs.openstack.org/98/555398/5/check/openstack-tox-cover/97f9d27/cover/ : SUCCESS in 7m 06s\n- openstack-tox-lower-constraints http://logs.openstack.org/98/555398/5/check/openstack-tox-lower-constraints/e58ff2c/ : SUCCESS in 4m 19s\n- openstack-tox-pep8 http://logs.openstack.org/98/555398/5/check/openstack-tox-pep8/d8e553a/ : SUCCESS in 6m 08s\n- openstack-tox-py27 http://logs.openstack.org/98/555398/5/check/openstack-tox-py27/e50d7a8/ : SUCCESS in 5m 51s\n- openstack-tox-py35 http://logs.openstack.org/98/555398/5/check/openstack-tox-py35/3491f05/ : SUCCESS in 5m 36s\n- openstack-tox-py36 http://logs.openstack.org/98/555398/5/check/openstack-tox-py36/11373e8/ : FAILURE in 5m 41s\n- openstack-tox-py37 http://logs.openstack.org/98/555398/5/check/openstack-tox-py37/728d34c/ : SUCCESS in 6m 18s\n- openstack-tox-docs http://logs.openstack.org/98/555398/5/check/openstack-tox-docs/c784d62/html/ : SUCCESS in 8m 24s\n- neutron-tempest-plugin-designate-scenario http://logs.openstack.org/98/555398/5/check/neutron-tempest-plugin-designate-scenario/3df5691/ : SUCCESS in 29m 10s","accounts_in_message":[],"_revision_number":5},{"id":"b6c2d0388c1b9e7437040b687a179a57eb26340b","author":{"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},"date":"2018-11-19 14:00:58.000000000","message":"Patch Set 5:\n\nrecheck","accounts_in_message":[],"_revision_number":5},{"id":"5b40f6a01ac21f2cd3d8e2a7dea1ada5c33bc6ab","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-11-19 15:01:01.000000000","message":"Patch Set 5:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- designate-bind9 http://logs.openstack.org/98/555398/5/check/designate-bind9/54d6047/ : FAILURE in 46m 37s\n- designate-bind9-manager-model http://logs.openstack.org/98/555398/5/check/designate-bind9-manager-model/deadda3/ : FAILURE in 38m 37s\n- designate-bind9-py35 http://logs.openstack.org/98/555398/5/check/designate-bind9-py35/42821a3/ : FAILURE in 38m 51s\n- designate-bind9-py36 http://logs.openstack.org/98/555398/5/check/designate-bind9-py36/78c6930/ : FAILURE in 35m 56s\n- designate-devstack-pdns4 http://logs.openstack.org/98/555398/5/check/designate-devstack-pdns4/5e75827/ : FAILURE in 40m 52s\n- designate-devstack-pdns4-postgres http://logs.openstack.org/98/555398/5/check/designate-devstack-pdns4-postgres/2b40adb/ : FAILURE in 31m 03s\n- designate-grenade-pdns4 http://logs.openstack.org/98/555398/5/check/designate-grenade-pdns4/fb7b5f0/ : FAILURE in 53m 59s\n- openstack-tox-cover http://logs.openstack.org/98/555398/5/check/openstack-tox-cover/cb81b57/cover/ : SUCCESS in 6m 03s\n- openstack-tox-lower-constraints http://logs.openstack.org/98/555398/5/check/openstack-tox-lower-constraints/03f3cc1/ : SUCCESS in 6m 13s\n- openstack-tox-pep8 http://logs.openstack.org/98/555398/5/check/openstack-tox-pep8/c89518b/ : SUCCESS in 6m 49s\n- openstack-tox-py27 http://logs.openstack.org/98/555398/5/check/openstack-tox-py27/37cb99a/ : SUCCESS in 6m 02s\n- openstack-tox-py35 http://logs.openstack.org/98/555398/5/check/openstack-tox-py35/a88ed6c/ : SUCCESS in 5m 31s\n- openstack-tox-py36 http://logs.openstack.org/98/555398/5/check/openstack-tox-py36/29b987d/ : SUCCESS in 5m 35s\n- openstack-tox-py37 http://logs.openstack.org/98/555398/5/check/openstack-tox-py37/91b61f8/ : SUCCESS in 8m 08s\n- openstack-tox-docs http://logs.openstack.org/98/555398/5/check/openstack-tox-docs/58a7232/html/ : SUCCESS in 5m 03s\n- neutron-tempest-plugin-designate-scenario http://logs.openstack.org/98/555398/5/check/neutron-tempest-plugin-designate-scenario/9bac0de/ : SUCCESS in 39m 27s","accounts_in_message":[],"_revision_number":5},{"id":"87f32b4523a72058af31f938edd4a2396fc10305","author":{"_account_id":20104,"name":"Dmitrii Galkin","display_name":"Dmitry Galkin","email":"galkindmitrii@gmail.com","username":"galkindmitrii"},"date":"2018-11-19 15:09:01.000000000","message":"Patch Set 5:\n\nThe tempest changes depend on https://review.openstack.org/#/c/617809/","accounts_in_message":[],"_revision_number":5},{"id":"992114e70792083ea4b7cb4e5ccc4663a584dcf1","author":{"_account_id":20104,"name":"Dmitrii Galkin","display_name":"Dmitry Galkin","email":"galkindmitrii@gmail.com","username":"galkindmitrii"},"date":"2018-11-19 15:13:10.000000000","message":"Patch Set 5:\n\n^ I\u0027ve meant the tempest tests","accounts_in_message":[],"_revision_number":5},{"id":"13ba926792ad694c251b91530bf3053d9c473642","author":{"_account_id":15736,"name":"Ritesh Anand","email":"tellrit@gmail.com","username":"ritesh"},"date":"2019-02-12 22:53:01.000000000","message":"Patch Set 5: Code-Review+1\n\nrecheck","accounts_in_message":[],"_revision_number":5},{"id":"39da3c540952a200ba3d14c9a3ba7a2eadf8959e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-02-12 23:55:39.000000000","message":"Patch Set 5:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- designate-bind9 http://logs.openstack.org/98/555398/5/check/designate-bind9/80ad721/ : FAILURE in 30m 56s\n- designate-bind9-manager-model http://logs.openstack.org/98/555398/5/check/designate-bind9-manager-model/d4c1620/ : FAILURE in 34m 09s\n- designate-bind9-py35 http://logs.openstack.org/98/555398/5/check/designate-bind9-py35/096bfc4/ : FAILURE in 31m 38s\n- designate-bind9-py36 http://logs.openstack.org/98/555398/5/check/designate-bind9-py36/de30a6a/ : FAILURE in 30m 08s\n- designate-devstack-pdns4 http://logs.openstack.org/98/555398/5/check/designate-devstack-pdns4/bdcadf9/ : FAILURE in 31m 33s\n- designate-devstack-pdns4-postgres http://logs.openstack.org/98/555398/5/check/designate-devstack-pdns4-postgres/f4c1fde/ : FAILURE in 29m 59s\n- designate-grenade-pdns4 http://logs.openstack.org/98/555398/5/check/designate-grenade-pdns4/a5e3cbf/ : FAILURE in 54m 29s\n- openstack-tox-cover http://logs.openstack.org/98/555398/5/check/openstack-tox-cover/39a7713/cover/ : SUCCESS in 6m 23s\n- openstack-tox-lower-constraints http://logs.openstack.org/98/555398/5/check/openstack-tox-lower-constraints/6040d69/ : SUCCESS in 4m 41s\n- openstack-tox-pep8 http://logs.openstack.org/98/555398/5/check/openstack-tox-pep8/2e4f5a3/ : SUCCESS in 5m 32s\n- openstack-tox-py27 http://logs.openstack.org/98/555398/5/check/openstack-tox-py27/e742573/ : SUCCESS in 5m 58s\n- openstack-tox-py35 http://logs.openstack.org/98/555398/5/check/openstack-tox-py35/d51e6ac/ : SUCCESS in 6m 16s\n- openstack-tox-py36 http://logs.openstack.org/98/555398/5/check/openstack-tox-py36/2a795db/ : SUCCESS in 5m 24s\n- openstack-tox-py37 http://logs.openstack.org/98/555398/5/check/openstack-tox-py37/578a71e/ : SUCCESS in 6m 26s\n- openstack-tox-docs http://logs.openstack.org/98/555398/5/check/openstack-tox-docs/df27069/html/ : SUCCESS in 4m 57s\n- neutron-tempest-plugin-designate-scenario http://logs.openstack.org/98/555398/5/check/neutron-tempest-plugin-designate-scenario/1544701/ : SUCCESS in 32m 30s","accounts_in_message":[],"_revision_number":5},{"id":"e08dbe57098f080a0e0e1844da206f2653a7d204","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2019-03-07 12:50:58.000000000","message":"Patch Set 5: Code-Review-1\n\n(3 comments)\n\nI think you\u0027d need to depend on the change to the tempest-plugin https://review.openstack.org/#/c/617809/4 in order to pass tests.\n\nWe also need a release note and probably matching unit tests.","accounts_in_message":[],"_revision_number":5},{"id":"9fbfaadb6eaaa0d1c070d95ed74a7db28c846ff6","author":{"_account_id":20104,"name":"Dmitrii Galkin","display_name":"Dmitry Galkin","email":"galkindmitrii@gmail.com","username":"galkindmitrii"},"date":"2019-03-13 14:59:29.000000000","message":"Patch Set 5:\n\nOK. I\u0027ll implement when the tempest plugin change is merged and Zuul checks get green again.","accounts_in_message":[],"_revision_number":5},{"id":"dab9c9b82c0f0255dc58ceccfd3e5d113ca8d19d","author":{"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},"date":"2019-03-14 13:16:12.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"2734a7994c188f658768a96b8579db72c139e7c4","author":{"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},"date":"2019-03-14 13:17:03.000000000","message":"Patch Set 6:\n\nupdated the commit message to pull in the new tests, and rebased on master","accounts_in_message":[],"_revision_number":6},{"id":"16fbf4887cb3bd82cbf63d5d2e5a2d38f468635b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-03-14 14:21:00.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- designate-bind9 http://logs.openstack.org/98/555398/6/check/designate-bind9/54b3aba/ : SUCCESS in 43m 10s\n- designate-bind9-manager-model http://logs.openstack.org/98/555398/6/check/designate-bind9-manager-model/e061341/ : SUCCESS in 43m 51s\n- designate-bind9-py36 http://logs.openstack.org/98/555398/6/check/designate-bind9-py36/cddb4da/ : SUCCESS in 42m 49s\n- designate-devstack-pdns4 http://logs.openstack.org/98/555398/6/check/designate-devstack-pdns4/ca5557f/ : SUCCESS in 38m 28s\n- designate-devstack-pdns4-postgres http://logs.openstack.org/98/555398/6/check/designate-devstack-pdns4-postgres/b60c1a1/ : SUCCESS in 38m 50s\n- designate-grenade-pdns4 http://logs.openstack.org/98/555398/6/check/designate-grenade-pdns4/92381a6/ : FAILURE in 1h 01m 47s\n- openstack-tox-cover http://logs.openstack.org/98/555398/6/check/openstack-tox-cover/d5830d7/cover/ : SUCCESS in 8m 58s\n- openstack-tox-lower-constraints http://logs.openstack.org/98/555398/6/check/openstack-tox-lower-constraints/5987fb3/ : SUCCESS in 5m 21s\n- openstack-tox-pep8 http://logs.openstack.org/98/555398/6/check/openstack-tox-pep8/3809454/ : SUCCESS in 8m 24s\n- openstack-tox-py27 http://logs.openstack.org/98/555398/6/check/openstack-tox-py27/ee0ea51/ : SUCCESS in 6m 13s\n- openstack-tox-py35 http://logs.openstack.org/98/555398/6/check/openstack-tox-py35/68a561d/ : SUCCESS in 6m 24s\n- openstack-tox-py36 http://logs.openstack.org/98/555398/6/check/openstack-tox-py36/e230ab7/ : SUCCESS in 9m 23s\n- openstack-tox-py37 http://logs.openstack.org/98/555398/6/check/openstack-tox-py37/07b848f/ : SUCCESS in 6m 26s\n- openstack-tox-docs http://logs.openstack.org/98/555398/6/check/openstack-tox-docs/c013469/html/ : SUCCESS in 7m 08s\n- neutron-tempest-plugin-designate-scenario http://logs.openstack.org/98/555398/6/check/neutron-tempest-plugin-designate-scenario/c8b59ce/ : SUCCESS in 29m 51s","accounts_in_message":[],"_revision_number":6},{"id":"a750f8b4e541ad3129b0682bfb4c641831bd4e04","author":{"_account_id":20104,"name":"Dmitrii Galkin","display_name":"Dmitry Galkin","email":"galkindmitrii@gmail.com","username":"galkindmitrii"},"date":"2019-03-14 14:31:02.000000000","message":"Patch Set 6:\n\nrecheck","accounts_in_message":[],"_revision_number":6},{"id":"213013a2e187af4d198c425b7a689c3c74544167","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-03-14 15:43:10.000000000","message":"Patch Set 6:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- designate-bind9 http://logs.openstack.org/98/555398/6/check/designate-bind9/1cb7529/ : SUCCESS in 35m 42s\n- designate-bind9-manager-model http://logs.openstack.org/98/555398/6/check/designate-bind9-manager-model/9455293/ : SUCCESS in 41m 49s\n- designate-bind9-py36 http://logs.openstack.org/98/555398/6/check/designate-bind9-py36/24cbe9c/ : SUCCESS in 38m 34s\n- designate-devstack-pdns4 http://logs.openstack.org/98/555398/6/check/designate-devstack-pdns4/eeba3a3/ : SUCCESS in 38m 31s\n- designate-devstack-pdns4-postgres http://logs.openstack.org/98/555398/6/check/designate-devstack-pdns4-postgres/f834177/ : SUCCESS in 42m 25s\n- designate-grenade-pdns4 http://logs.openstack.org/98/555398/6/check/designate-grenade-pdns4/e0f3bd0/ : FAILURE in 1h 02m 20s\n- openstack-tox-cover http://logs.openstack.org/98/555398/6/check/openstack-tox-cover/8cbdee9/cover/ : SUCCESS in 6m 53s\n- openstack-tox-lower-constraints http://logs.openstack.org/98/555398/6/check/openstack-tox-lower-constraints/dbca17d/ : SUCCESS in 5m 25s\n- openstack-tox-pep8 http://logs.openstack.org/98/555398/6/check/openstack-tox-pep8/5c5e85c/ : SUCCESS in 5m 15s\n- openstack-tox-py27 http://logs.openstack.org/98/555398/6/check/openstack-tox-py27/1edfdbc/ : SUCCESS in 6m 18s\n- openstack-tox-py35 http://logs.openstack.org/98/555398/6/check/openstack-tox-py35/c270f7c/ : SUCCESS in 7m 07s\n- openstack-tox-py36 http://logs.openstack.org/98/555398/6/check/openstack-tox-py36/db6c0ec/ : SUCCESS in 6m 29s\n- openstack-tox-py37 http://logs.openstack.org/98/555398/6/check/openstack-tox-py37/14f42c3/ : SUCCESS in 6m 45s\n- openstack-tox-docs http://logs.openstack.org/98/555398/6/check/openstack-tox-docs/b11f402/html/ : SUCCESS in 5m 33s\n- neutron-tempest-plugin-designate-scenario http://logs.openstack.org/98/555398/6/check/neutron-tempest-plugin-designate-scenario/a4f1e46/ : SUCCESS in 33m 03s","accounts_in_message":[],"_revision_number":6},{"id":"c1f22ad4894dfbe1207f90f73d3b76fe855c24f2","author":{"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},"date":"2019-03-14 18:15:31.000000000","message":"Patch Set 6:\n\nrecheck","accounts_in_message":[],"_revision_number":6},{"id":"094f8e1c44f5140fb2773752ddeb939325401c38","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-03-14 19:13:19.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\n\n- designate-bind9 http://logs.openstack.org/98/555398/6/check/designate-bind9/7acd663/ : SUCCESS in 39m 01s\n- designate-bind9-manager-model http://logs.openstack.org/98/555398/6/check/designate-bind9-manager-model/e0d85bf/ : SUCCESS in 38m 38s\n- designate-bind9-py36 http://logs.openstack.org/98/555398/6/check/designate-bind9-py36/03ce34e/ : SUCCESS in 39m 33s\n- designate-devstack-pdns4 http://logs.openstack.org/98/555398/6/check/designate-devstack-pdns4/95d2eae/ : SUCCESS in 35m 03s\n- designate-devstack-pdns4-postgres http://logs.openstack.org/98/555398/6/check/designate-devstack-pdns4-postgres/075767b/ : SUCCESS in 40m 41s\n- designate-grenade-pdns4 http://logs.openstack.org/98/555398/6/check/designate-grenade-pdns4/faaaea4/ : SUCCESS in 55m 13s\n- openstack-tox-cover http://logs.openstack.org/98/555398/6/check/openstack-tox-cover/bd72681/cover/ : SUCCESS in 6m 29s\n- openstack-tox-lower-constraints http://logs.openstack.org/98/555398/6/check/openstack-tox-lower-constraints/d79bbba/ : SUCCESS in 4m 42s\n- openstack-tox-pep8 http://logs.openstack.org/98/555398/6/check/openstack-tox-pep8/7406cb5/ : SUCCESS in 5m 33s\n- openstack-tox-py27 http://logs.openstack.org/98/555398/6/check/openstack-tox-py27/98852ad/ : SUCCESS in 6m 16s\n- openstack-tox-py35 http://logs.openstack.org/98/555398/6/check/openstack-tox-py35/467d92d/ : SUCCESS in 6m 35s\n- openstack-tox-py36 http://logs.openstack.org/98/555398/6/check/openstack-tox-py36/cef6d71/ : SUCCESS in 5m 46s\n- openstack-tox-py37 http://logs.openstack.org/98/555398/6/check/openstack-tox-py37/493cff8/ : SUCCESS in 6m 48s\n- openstack-tox-docs http://logs.openstack.org/98/555398/6/check/openstack-tox-docs/026d6ed/html/ : SUCCESS in 5m 48s\n- neutron-tempest-plugin-designate-scenario http://logs.openstack.org/98/555398/6/check/neutron-tempest-plugin-designate-scenario/b0fe5e9/ : SUCCESS in 39m 14s","accounts_in_message":[],"_revision_number":6},{"id":"2a0b3738915a0115da677cd26b0d693805ab4b0d","author":{"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},"date":"2019-03-19 12:01:57.000000000","message":"Patch Set 6: Review-Priority+1","accounts_in_message":[],"_revision_number":6},{"id":"1ee28468fe41bbf08fbbfec3c06a599957d76dda","author":{"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},"date":"2019-03-19 13:34:58.000000000","message":"Patch Set 6:\n\n@dmitry - do you want to update this? I would love to get it merged this week so we can have it in stien","accounts_in_message":[],"_revision_number":6},{"id":"e930fe19777f31553de50144a95a95e01337634d","author":{"_account_id":20104,"name":"Dmitrii Galkin","display_name":"Dmitry Galkin","email":"galkindmitrii@gmail.com","username":"galkindmitrii"},"date":"2019-03-19 15:06:22.000000000","message":"Patch Set 6:\n\nHi Graham, yes. Will do tomorrow!","accounts_in_message":[],"_revision_number":6},{"id":"60e1f3c6bc65be7b580a6bc18f901bd97a9de6cb","author":{"_account_id":20104,"name":"Dmitrii Galkin","display_name":"Dmitry Galkin","email":"galkindmitrii@gmail.com","username":"galkindmitrii"},"date":"2019-03-24 16:33:43.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"0c4a5645c9f345fffd105e46c51899ce77544986","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-03-24 17:38:04.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- designate-bind9 http://logs.openstack.org/98/555398/7/check/designate-bind9/ca2c429/ : SUCCESS in 38m 18s\n- designate-bind9-manager-model http://logs.openstack.org/98/555398/7/check/designate-bind9-manager-model/6de1823/ : SUCCESS in 45m 46s\n- designate-bind9-py36 http://logs.openstack.org/98/555398/7/check/designate-bind9-py36/958f51e/ : SUCCESS in 36m 13s\n- designate-devstack-pdns4 http://logs.openstack.org/98/555398/7/check/designate-devstack-pdns4/0c57eed/ : SUCCESS in 37m 42s\n- designate-devstack-pdns4-postgres http://logs.openstack.org/98/555398/7/check/designate-devstack-pdns4-postgres/d6bc75a/ : SUCCESS in 34m 32s\n- designate-grenade-pdns4 http://logs.openstack.org/98/555398/7/check/designate-grenade-pdns4/423ba6e/ : SUCCESS in 57m 56s\n- openstack-tox-cover http://logs.openstack.org/98/555398/7/check/openstack-tox-cover/fb8097a/cover/ : SUCCESS in 6m 04s\n- openstack-tox-lower-constraints http://logs.openstack.org/98/555398/7/check/openstack-tox-lower-constraints/5c42f6c/ : SUCCESS in 4m 43s\n- openstack-tox-pep8 http://logs.openstack.org/98/555398/7/check/openstack-tox-pep8/1a7d2fb/ : FAILURE in 4m 53s\n- openstack-tox-py27 http://logs.openstack.org/98/555398/7/check/openstack-tox-py27/b5360bb/ : SUCCESS in 5m 52s\n- openstack-tox-py35 http://logs.openstack.org/98/555398/7/check/openstack-tox-py35/898c412/ : SUCCESS in 5m 40s\n- openstack-tox-py36 http://logs.openstack.org/98/555398/7/check/openstack-tox-py36/e6f2628/ : SUCCESS in 5m 33s\n- openstack-tox-py37 http://logs.openstack.org/98/555398/7/check/openstack-tox-py37/a584e2b/ : SUCCESS in 6m 05s\n- openstack-tox-docs http://logs.openstack.org/98/555398/7/check/openstack-tox-docs/fe2533e/html/ : SUCCESS in 5m 01s\n- build-openstack-releasenotes http://logs.openstack.org/98/555398/7/check/build-openstack-releasenotes/79c88ac/html/ : SUCCESS in 6m 08s\n- neutron-tempest-plugin-designate-scenario http://logs.openstack.org/98/555398/7/check/neutron-tempest-plugin-designate-scenario/86c70dc/ : SUCCESS in 34m 55s","accounts_in_message":[],"_revision_number":7},{"id":"3dd75597d38e2f8c7179e3bdb7e995f1e24dc348","author":{"_account_id":20104,"name":"Dmitrii Galkin","display_name":"Dmitry Galkin","email":"galkindmitrii@gmail.com","username":"galkindmitrii"},"date":"2019-03-25 14:02:43.000000000","message":"Uploaded patch set 8.","accounts_in_message":[],"_revision_number":8},{"id":"a202f3e1eb5d3a01fc349844c2567226d37ef394","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-03-25 15:15:36.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\n\n- designate-bind9 http://logs.openstack.org/98/555398/8/check/designate-bind9/42f9b6f/ : SUCCESS in 38m 30s\n- designate-bind9-manager-model http://logs.openstack.org/98/555398/8/check/designate-bind9-manager-model/672bc49/ : SUCCESS in 38m 51s\n- designate-bind9-py36 http://logs.openstack.org/98/555398/8/check/designate-bind9-py36/cdf5712/ : SUCCESS in 47m 17s\n- designate-devstack-pdns4 http://logs.openstack.org/98/555398/8/check/designate-devstack-pdns4/19db65c/ : SUCCESS in 38m 03s\n- designate-devstack-pdns4-postgres http://logs.openstack.org/98/555398/8/check/designate-devstack-pdns4-postgres/1cff162/ : SUCCESS in 41m 25s\n- designate-grenade-pdns4 http://logs.openstack.org/98/555398/8/check/designate-grenade-pdns4/8ff8089/ : SUCCESS in 1h 03m 07s\n- openstack-tox-cover http://logs.openstack.org/98/555398/8/check/openstack-tox-cover/f548a36/cover/ : SUCCESS in 7m 35s\n- openstack-tox-lower-constraints http://logs.openstack.org/98/555398/8/check/openstack-tox-lower-constraints/ca0c521/ : SUCCESS in 8m 03s\n- openstack-tox-pep8 http://logs.openstack.org/98/555398/8/check/openstack-tox-pep8/64999a1/ : SUCCESS in 6m 26s\n- openstack-tox-py27 http://logs.openstack.org/98/555398/8/check/openstack-tox-py27/95898fb/ : SUCCESS in 8m 29s\n- openstack-tox-py35 http://logs.openstack.org/98/555398/8/check/openstack-tox-py35/a1ee5e1/ : SUCCESS in 8m 18s\n- openstack-tox-py36 http://logs.openstack.org/98/555398/8/check/openstack-tox-py36/9ddd8f4/ : SUCCESS in 7m 54s\n- openstack-tox-py37 http://logs.openstack.org/98/555398/8/check/openstack-tox-py37/ffce5bc/ : SUCCESS in 9m 56s\n- openstack-tox-docs http://logs.openstack.org/98/555398/8/check/openstack-tox-docs/91ab762/html/ : SUCCESS in 7m 22s\n- build-openstack-releasenotes http://logs.openstack.org/98/555398/8/check/build-openstack-releasenotes/80cd521/html/ : SUCCESS in 7m 46s\n- neutron-tempest-plugin-designate-scenario http://logs.openstack.org/98/555398/8/check/neutron-tempest-plugin-designate-scenario/ddb2d54/ : SUCCESS in 37m 30s","accounts_in_message":[],"_revision_number":8},{"id":"cf90aca0873a4b7236ff075a8d8f82552e33ba8a","author":{"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},"date":"2019-03-25 16:04:00.000000000","message":"Patch Set 8: Code-Review+2 Review-Priority+1","accounts_in_message":[],"_revision_number":8},{"id":"e0e13138140c8b4594fbf1756afc94c595d5aec9","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2019-05-20 03:18:03.000000000","message":"Patch Set 8: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":8},{"id":"c13d52a75c0d20c36c17c46f948a6353ce7455a6","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-20 03:18:15.000000000","message":"Patch Set 8: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":8},{"id":"b0a9ddb99db6d052ec800339c70058877c3e82cc","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-20 04:18:20.000000000","message":"Patch Set 8: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- designate-bind9 http://logs.openstack.org/98/555398/8/gate/designate-bind9/dd89f3d/ : SUCCESS in 34m 53s\n- designate-pdns4 http://logs.openstack.org/98/555398/8/gate/designate-pdns4/097bb5d/ : SUCCESS in 34m 26s\n- designate-bind9-manager-model http://logs.openstack.org/98/555398/8/gate/designate-bind9-manager-model/81ee4f6/ : SUCCESS in 37m 51s\n- designate-bind9-py27 http://logs.openstack.org/98/555398/8/gate/designate-bind9-py27/aeecca0/ : SUCCESS in 37m 20s\n- designate-pdns4-py27 http://logs.openstack.org/98/555398/8/gate/designate-pdns4-py27/d6ab0ed/ : SUCCESS in 34m 42s\n- designate-pdns4-postgres http://logs.openstack.org/98/555398/8/gate/designate-pdns4-postgres/fcc5d13/ : SUCCESS in 36m 06s\n- designate-grenade-pdns4 http://logs.openstack.org/98/555398/8/gate/designate-grenade-pdns4/a150987/ : SUCCESS in 54m 18s\n- openstack-tox-lower-constraints http://logs.openstack.org/98/555398/8/gate/openstack-tox-lower-constraints/f661c4c/ : SUCCESS in 4m 23s\n- openstack-tox-pep8 http://logs.openstack.org/98/555398/8/gate/openstack-tox-pep8/74758c8/ : SUCCESS in 5m 45s\n- openstack-tox-py27 http://logs.openstack.org/98/555398/8/gate/openstack-tox-py27/ee35905/ : SUCCESS in 5m 17s\n- openstack-tox-py36 http://logs.openstack.org/98/555398/8/gate/openstack-tox-py36/7fcc878/ : SUCCESS in 5m 17s\n- openstack-tox-py37 http://logs.openstack.org/98/555398/8/gate/openstack-tox-py37/f260505/ : SUCCESS in 5m 27s\n- openstack-tox-docs http://logs.openstack.org/98/555398/8/gate/openstack-tox-docs/84f78d1/html/ : SUCCESS in 4m 31s\n- build-openstack-releasenotes http://logs.openstack.org/98/555398/8/gate/build-openstack-releasenotes/ea6cc8c/html/ : SUCCESS in 6m 00s","accounts_in_message":[],"_revision_number":8},{"id":"2aa5d5211056539ed3d901d89b3df984e6df5204","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-20 04:18:20.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":8}],"current_revision_number":8,"current_revision":"02ef0350af6bb2ee1aacb6e7e4ec9acb21eb3dc7","revisions":{"7d5be04ab8510d7e4d896f0caf53af28713529da":{"kind":"REWORK","_number":1,"created":"2018-03-22 18:05:31.000000000","uploader":{"_account_id":20104,"name":"Dmitrii Galkin","display_name":"Dmitry Galkin","email":"galkindmitrii@gmail.com","username":"galkindmitrii"},"ref":"refs/changes/98/555398/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/98/555398/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/98/555398/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/98/555398/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/98/555398/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/98/555398/1"}}},"commit":{"parents":[{"commit":"fb5c1a97797c66292151549b82519a682b6300b6","subject":"Merge \"Remove translation of log messages\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/fb5c1a97797c66292151549b82519a682b6300b6"}]}],"author":{"name":"Dmitry Galkin","email":"galkindmitrii@gmail.com","date":"2018-03-22 17:53:27.000000000","tz":0},"committer":{"name":"Dmitry Galkin","email":"galkindmitrii@gmail.com","date":"2018-03-22 18:04:50.000000000","tz":0},"subject":"Check TXT/SPF records for RFC1035 sec. 5.1","message":"Check TXT/SPF records for RFC1035 sec. 5.1\n\nIf record data has empty spaces it should be surrounded by double quotes.\nThis patch will raise an error during validation if record has whitespaces,\nempty spaces, tabs, etc., but not wrapped in \" \" (double quotes).\n\nCorresponding RFC part:\n\n\u003ccharacter-string\u003e is expressed in one or two ways: as a contiguous set\nof characters without interior spaces, or as a string beginning with a \"\nand ending with a \".  Inside a \" delimited string any character can\noccur, except for a \" itself, which must be quoted using \\ (back slash).\n\nCloses-Bug: 1755788\n\nChange-Id: I159d0732688ddf1337ab3602a84a43fd043dcaa2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/7d5be04ab8510d7e4d896f0caf53af28713529da"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/7d5be04ab8510d7e4d896f0caf53af28713529da"}]},"branch":"refs/heads/master"},"e2f32ada6bdd39fb1ab3fe08d6394f7dc2c4ca4b":{"kind":"REWORK","_number":2,"created":"2018-03-26 11:54:21.000000000","uploader":{"_account_id":20104,"name":"Dmitrii Galkin","display_name":"Dmitry Galkin","email":"galkindmitrii@gmail.com","username":"galkindmitrii"},"ref":"refs/changes/98/555398/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/98/555398/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/98/555398/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/98/555398/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/98/555398/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/98/555398/2"}}},"commit":{"parents":[{"commit":"fb5c1a97797c66292151549b82519a682b6300b6","subject":"Merge \"Remove translation of log messages\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/fb5c1a97797c66292151549b82519a682b6300b6"}]}],"author":{"name":"Dmitry Galkin","email":"galkindmitrii@gmail.com","date":"2018-03-22 17:53:27.000000000","tz":0},"committer":{"name":"Dmitry Galkin","email":"galkindmitrii@gmail.com","date":"2018-03-26 11:53:32.000000000","tz":0},"subject":"Check TXT/SPF records for RFC1035 sec. 5.1","message":"Check TXT/SPF records for RFC1035 sec. 5.1\n\nIf record data has empty spaces it should be surrounded by double quotes.\nThis patch will raise an error during validation if record has whitespaces,\nempty spaces, tabs, etc., but not wrapped in \" \" (double quotes).\n\nCorresponding RFC part:\n\n\u003ccharacter-string\u003e is expressed in one or two ways: as a contiguous set\nof characters without interior spaces, or as a string beginning with a \"\nand ending with a \".  Inside a \" delimited string any character can\noccur, except for a \" itself, which must be quoted using \\ (back slash).\n\nCloses-Bug: 1755788\n\nChange-Id: I159d0732688ddf1337ab3602a84a43fd043dcaa2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/e2f32ada6bdd39fb1ab3fe08d6394f7dc2c4ca4b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/e2f32ada6bdd39fb1ab3fe08d6394f7dc2c4ca4b"}]},"branch":"refs/heads/master"},"8283c9426653e4ee4a505f21325d736f36a58877":{"kind":"REWORK","_number":3,"created":"2018-04-09 15:12:53.000000000","uploader":{"_account_id":20104,"name":"Dmitrii Galkin","display_name":"Dmitry Galkin","email":"galkindmitrii@gmail.com","username":"galkindmitrii"},"ref":"refs/changes/98/555398/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/98/555398/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/98/555398/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/98/555398/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/98/555398/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/98/555398/3"}}},"commit":{"parents":[{"commit":"fb5c1a97797c66292151549b82519a682b6300b6","subject":"Merge \"Remove translation of log messages\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/fb5c1a97797c66292151549b82519a682b6300b6"}]}],"author":{"name":"Dmitry Galkin","email":"galkindmitrii@gmail.com","date":"2018-03-22 17:53:27.000000000","tz":0},"committer":{"name":"Dmitry Galkin","email":"galkindmitrii@gmail.com","date":"2018-04-09 15:12:37.000000000","tz":0},"subject":"Check TXT/SPF records for RFC1035 sec. 5.1","message":"Check TXT/SPF records for RFC1035 sec. 5.1\n\nIf record data has empty spaces it should be surrounded by double quotes.\nThis patch will raise an error during validation if record has whitespaces,\nempty spaces, tabs, etc., but not wrapped in \" \" (double quotes).\n\nCorresponding RFC part:\n\n\u003ccharacter-string\u003e is expressed in one or two ways: as a contiguous set\nof characters without interior spaces, or as a string beginning with a \"\nand ending with a \".  Inside a \" delimited string any character can\noccur, except for a \" itself, which must be quoted using \\ (back slash).\n\nCloses-Bug: 1755788\n\nChange-Id: I159d0732688ddf1337ab3602a84a43fd043dcaa2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/8283c9426653e4ee4a505f21325d736f36a58877"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/8283c9426653e4ee4a505f21325d736f36a58877"}]},"branch":"refs/heads/master"},"ea3eadd7f050919a961d859fe3b79d1d586b188b":{"kind":"REWORK","_number":4,"created":"2018-04-09 15:17:35.000000000","uploader":{"_account_id":20104,"name":"Dmitrii Galkin","display_name":"Dmitry Galkin","email":"galkindmitrii@gmail.com","username":"galkindmitrii"},"ref":"refs/changes/98/555398/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/98/555398/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/98/555398/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/98/555398/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/98/555398/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/98/555398/4"}}},"commit":{"parents":[{"commit":"fb5c1a97797c66292151549b82519a682b6300b6","subject":"Merge \"Remove translation of log messages\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/fb5c1a97797c66292151549b82519a682b6300b6"}]}],"author":{"name":"Dmitry Galkin","email":"galkindmitrii@gmail.com","date":"2018-03-22 17:53:27.000000000","tz":0},"committer":{"name":"Dmitry Galkin","email":"galkindmitrii@gmail.com","date":"2018-04-09 15:17:25.000000000","tz":0},"subject":"Check TXT/SPF records for RFC1035 sec. 5.1","message":"Check TXT/SPF records for RFC1035 sec. 5.1\n\nIf record data has empty spaces it should be surrounded by double quotes.\nThis patch will raise an error during validation if record has whitespaces,\nempty spaces, tabs, etc., but not wrapped in \" \" (double quotes).\n\nCorresponding RFC part:\n\n\u003ccharacter-string\u003e is expressed in one or two ways: as a contiguous set\nof characters without interior spaces, or as a string beginning with a \"\nand ending with a \".  Inside a \" delimited string any character can\noccur, except for a \" itself, which must be quoted using \\ (back slash).\n\nCloses-Bug: 1755788\n\nChange-Id: I159d0732688ddf1337ab3602a84a43fd043dcaa2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/ea3eadd7f050919a961d859fe3b79d1d586b188b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/ea3eadd7f050919a961d859fe3b79d1d586b188b"}]},"branch":"refs/heads/master"},"746f274532a8fb79e2a7533aa190bf9b9bb45bf7":{"kind":"REWORK","_number":5,"created":"2018-05-03 15:47:21.000000000","uploader":{"_account_id":20104,"name":"Dmitrii Galkin","display_name":"Dmitry Galkin","email":"galkindmitrii@gmail.com","username":"galkindmitrii"},"ref":"refs/changes/98/555398/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/98/555398/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/98/555398/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/98/555398/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/98/555398/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/98/555398/5"}}},"commit":{"parents":[{"commit":"fb5c1a97797c66292151549b82519a682b6300b6","subject":"Merge \"Remove translation of log messages\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/fb5c1a97797c66292151549b82519a682b6300b6"}]}],"author":{"name":"Dmitry Galkin","email":"galkindmitrii@gmail.com","date":"2018-03-22 17:53:27.000000000","tz":0},"committer":{"name":"Dmitry Galkin","email":"galkindmitrii@gmail.com","date":"2018-05-03 15:39:34.000000000","tz":0},"subject":"Check TXT/SPF records for RFC1035 sec. 5.1","message":"Check TXT/SPF records for RFC1035 sec. 5.1\n\nIf record data has empty spaces it should be surrounded by double quotes.\nThis patch will raise an error during validation if record has whitespaces,\nempty spaces, tabs, etc., but not wrapped in \" \" (double quotes).\n\nCorresponding RFC part:\n\n\u003ccharacter-string\u003e is expressed in one or two ways: as a contiguous set\nof characters without interior spaces, or as a string beginning with a \"\nand ending with a \".  Inside a \" delimited string any character can\noccur, except for a \" itself, which must be quoted using \\ (back slash).\n\nCloses-Bug: 1755788\n\nChange-Id: I159d0732688ddf1337ab3602a84a43fd043dcaa2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/746f274532a8fb79e2a7533aa190bf9b9bb45bf7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/746f274532a8fb79e2a7533aa190bf9b9bb45bf7"}]},"branch":"refs/heads/master"},"c89d5acb310cf9a77df99dd9ca91aff0caa9fb9e":{"kind":"TRIVIAL_REBASE_WITH_MESSAGE_UPDATE","_number":6,"created":"2019-03-14 13:16:12.000000000","uploader":{"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},"ref":"refs/changes/98/555398/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/98/555398/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/98/555398/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/98/555398/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/98/555398/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/98/555398/6"}}},"commit":{"parents":[{"commit":"59d8cab9bbe37529346728778e6dae79a3a57a2a","subject":"Allow non localhost connections to pdns api","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/59d8cab9bbe37529346728778e6dae79a3a57a2a"}]}],"author":{"name":"Dmitry Galkin","email":"galkindmitrii@gmail.com","date":"2018-03-22 17:53:27.000000000","tz":0},"committer":{"name":"Graham Hayes","email":"gr@ham.ie","date":"2019-03-14 13:15:52.000000000","tz":0},"subject":"Check TXT/SPF records for RFC1035 sec. 5.1","message":"Check TXT/SPF records for RFC1035 sec. 5.1\n\nIf record data has empty spaces it should be surrounded by double quotes.\nThis patch will raise an error during validation if record has whitespaces,\nempty spaces, tabs, etc., but not wrapped in \" \" (double quotes).\n\nCorresponding RFC part:\n\n\u003ccharacter-string\u003e is expressed in one or two ways: as a contiguous set\nof characters without interior spaces, or as a string beginning with a \"\nand ending with a \".  Inside a \" delimited string any character can\noccur, except for a \" itself, which must be quoted using \\ (back slash).\n\nCloses-Bug: 1755788\nDepends-On: https://review.openstack.org/#/c/617809/\nChange-Id: I159d0732688ddf1337ab3602a84a43fd043dcaa2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/c89d5acb310cf9a77df99dd9ca91aff0caa9fb9e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/c89d5acb310cf9a77df99dd9ca91aff0caa9fb9e"}]},"branch":"refs/heads/master"},"4ae5f3a794a5cd8c08e864ece32ad0ebc19705d5":{"kind":"REWORK","_number":7,"created":"2019-03-24 16:33:43.000000000","uploader":{"_account_id":20104,"name":"Dmitrii Galkin","display_name":"Dmitry Galkin","email":"galkindmitrii@gmail.com","username":"galkindmitrii"},"ref":"refs/changes/98/555398/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/98/555398/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/98/555398/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/98/555398/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/98/555398/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/98/555398/7"}}},"commit":{"parents":[{"commit":"59d8cab9bbe37529346728778e6dae79a3a57a2a","subject":"Allow non localhost connections to pdns api","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/59d8cab9bbe37529346728778e6dae79a3a57a2a"}]}],"author":{"name":"Dmitry Galkin","email":"galkindmitrii@gmail.com","date":"2018-03-22 17:53:27.000000000","tz":0},"committer":{"name":"Dmitry Galkin","email":"galkindmitrii@gmail.com","date":"2019-03-24 16:33:39.000000000","tz":60},"subject":"Check TXT/SPF records for RFC1035 sec. 5.1","message":"Check TXT/SPF records for RFC1035 sec. 5.1\n\nIf record data has empty spaces it should be surrounded by double quotes.\nThis patch will raise an error during validation if record has whitespaces,\nempty spaces, tabs, etc., but not wrapped in \" \" (double quotes).\n\nCorresponding RFC part:\n\n\u003ccharacter-string\u003e is expressed in one or two ways: as a contiguous set\nof characters without interior spaces, or as a string beginning with a \"\nand ending with a \".  Inside a \" delimited string any character can\noccur, except for a \" itself, which must be quoted using \\ (back slash).\n\nCloses-Bug: 1755788\nDepends-On: https://review.openstack.org/#/c/617809/\nChange-Id: I159d0732688ddf1337ab3602a84a43fd043dcaa2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/4ae5f3a794a5cd8c08e864ece32ad0ebc19705d5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/4ae5f3a794a5cd8c08e864ece32ad0ebc19705d5"}]},"branch":"refs/heads/master"},"02ef0350af6bb2ee1aacb6e7e4ec9acb21eb3dc7":{"kind":"REWORK","_number":8,"created":"2019-03-25 14:02:43.000000000","uploader":{"_account_id":20104,"name":"Dmitrii Galkin","display_name":"Dmitry Galkin","email":"galkindmitrii@gmail.com","username":"galkindmitrii"},"ref":"refs/changes/98/555398/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/98/555398/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/98/555398/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/98/555398/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/98/555398/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/98/555398/8"}}},"commit":{"parents":[{"commit":"59d8cab9bbe37529346728778e6dae79a3a57a2a","subject":"Allow non localhost connections to pdns api","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/59d8cab9bbe37529346728778e6dae79a3a57a2a"}]}],"author":{"name":"Dmitry Galkin","email":"galkindmitrii@gmail.com","date":"2018-03-22 17:53:27.000000000","tz":0},"committer":{"name":"Dmitry Galkin","email":"galkindmitrii@gmail.com","date":"2019-03-25 14:02:26.000000000","tz":60},"subject":"Check TXT/SPF records for RFC1035 sec. 5.1","message":"Check TXT/SPF records for RFC1035 sec. 5.1\n\nIf record data has empty spaces it should be surrounded by double quotes.\nThis patch will raise an error during validation if record has whitespaces,\nempty spaces, tabs, etc., but not wrapped in \" \" (double quotes).\n\nCorresponding RFC part:\n\n\u003ccharacter-string\u003e is expressed in one or two ways: as a contiguous set\nof characters without interior spaces, or as a string beginning with a \"\nand ending with a \".  Inside a \" delimited string any character can\noccur, except for a \" itself, which must be quoted using \\ (back slash).\n\nCloses-Bug: 1755788\nDepends-On: https://review.openstack.org/#/c/617809/\nChange-Id: I159d0732688ddf1337ab3602a84a43fd043dcaa2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/02ef0350af6bb2ee1aacb6e7e4ec9acb21eb3dc7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/02ef0350af6bb2ee1aacb6e7e4ec9acb21eb3dc7"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
