)]}'
{"id":"openstack%2Fdesignate~55871","triplet_id":"openstack%2Fdesignate~master~I9718489335e24cb24733d4c4fc2a966e490014d3","project":"openstack/designate","branch":"master","topic":"bug/1249396","hashtags":[],"change_id":"I9718489335e24cb24733d4c4fc2a966e490014d3","subject":"Add SLDs that act like TLDs","status":"MERGED","created":"2013-11-11 16:28:09.000000000","updated":"2013-11-19 15:54:23.000000000","submitted":"2013-11-19 15:54:23.000000000","submitter":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"total_comment_count":26,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"b3418421e53017a4af3b22b7d6aed05c88e68dbb","_number":55871,"virtual_id_number":55871,"owner":{"_account_id":8094,"name":"Vinod Mangalpally","email":"mrvinod@outlook.com","username":"vinodmr"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"all":[{"value":0,"_account_id":8864,"name":"Artom Lifshitz","email":"notartom@gmail.com","username":"artom"},{"value":0,"_account_id":8130,"name":"Betsy Luzader","email":"betsy.luzader@rackspace.com","username":"bluzader"},{"value":0,"_account_id":8098,"name":"Joe McBride","email":"jmcbride@rackspace.com","username":"mejoe"},{"value":2,"date":"2013-11-19 15:54:23.000000000","post_submit":true,"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":0,"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},{"value":0,"_account_id":741,"name":"Kiall Mac Innes","email":"kiall@macinnes.ie","username":"kiall"},{"value":0,"_account_id":8101,"name":"Emmanuel Ankutse","email":"emmanuel.ankutse@rackspace.com","username":"emmanuel.ankutse"},{"value":0,"_account_id":8094,"name":"Vinod Mangalpally","email":"mrvinod@outlook.com","username":"vinodmr"},{"value":0,"_account_id":8174,"name":"Tim Simmons","email":"tim.r.simmons@gmail.com","username":"TimSimmons"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":741,"name":"Kiall Mac Innes","email":"kiall@macinnes.ie","username":"kiall"},"all":[{"value":0,"_account_id":8864,"name":"Artom Lifshitz","email":"notartom@gmail.com","username":"artom"},{"value":0,"_account_id":8130,"name":"Betsy Luzader","email":"betsy.luzader@rackspace.com","username":"bluzader"},{"value":0,"_account_id":8098,"name":"Joe McBride","email":"jmcbride@rackspace.com","username":"mejoe"},{"value":0,"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":0,"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},{"value":2,"date":"2013-11-19 14:56:14.000000000","_account_id":741,"name":"Kiall Mac Innes","email":"kiall@macinnes.ie","username":"kiall"},{"value":0,"_account_id":8101,"name":"Emmanuel Ankutse","email":"emmanuel.ankutse@rackspace.com","username":"emmanuel.ankutse"},{"value":0,"_account_id":8094,"name":"Vinod Mangalpally","email":"mrvinod@outlook.com","username":"vinodmr"},{"value":0,"_account_id":8174,"name":"Tim Simmons","email":"tim.r.simmons@gmail.com","username":"TimSimmons"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":741,"name":"Kiall Mac Innes","email":"kiall@macinnes.ie","username":"kiall"},"all":[{"value":0,"_account_id":8864,"name":"Artom Lifshitz","email":"notartom@gmail.com","username":"artom"},{"value":0,"_account_id":8130,"name":"Betsy Luzader","email":"betsy.luzader@rackspace.com","username":"bluzader"},{"value":0,"_account_id":8098,"name":"Joe McBride","email":"jmcbride@rackspace.com","username":"mejoe"},{"value":0,"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":0,"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},{"value":1,"date":"2013-11-19 14:56:14.000000000","_account_id":741,"name":"Kiall Mac Innes","email":"kiall@macinnes.ie","username":"kiall"},{"value":0,"_account_id":8101,"name":"Emmanuel Ankutse","email":"emmanuel.ankutse@rackspace.com","username":"emmanuel.ankutse"},{"value":0,"_account_id":8094,"name":"Vinod Mangalpally","email":"mrvinod@outlook.com","username":"vinodmr"},{"value":0,"_account_id":8174,"name":"Tim Simmons","email":"tim.r.simmons@gmail.com","username":"TimSimmons"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":8864,"name":"Artom Lifshitz","email":"notartom@gmail.com","username":"artom"},{"value":0,"_account_id":8130,"name":"Betsy Luzader","email":"betsy.luzader@rackspace.com","username":"bluzader"},{"value":0,"_account_id":8098,"name":"Joe McBride","email":"jmcbride@rackspace.com","username":"mejoe"},{"value":0,"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":0,"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},{"value":0,"_account_id":741,"name":"Kiall Mac Innes","email":"kiall@macinnes.ie","username":"kiall"},{"value":0,"_account_id":8101,"name":"Emmanuel Ankutse","email":"emmanuel.ankutse@rackspace.com","username":"emmanuel.ankutse"},{"value":0,"_account_id":8094,"name":"Vinod Mangalpally","email":"mrvinod@outlook.com","username":"vinodmr"},{"value":0,"_account_id":8174,"name":"Tim Simmons","email":"tim.r.simmons@gmail.com","username":"TimSimmons"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":3,"name":"Jenkins","username":"jenkins"},{"_account_id":741,"name":"Kiall Mac Innes","email":"kiall@macinnes.ie","username":"kiall"},{"_account_id":8094,"name":"Vinod Mangalpally","email":"mrvinod@outlook.com","username":"vinodmr"},{"_account_id":8098,"name":"Joe McBride","email":"jmcbride@rackspace.com","username":"mejoe"},{"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},{"_account_id":8101,"name":"Emmanuel Ankutse","email":"emmanuel.ankutse@rackspace.com","username":"emmanuel.ankutse"},{"_account_id":8130,"name":"Betsy Luzader","email":"betsy.luzader@rackspace.com","username":"bluzader"},{"_account_id":8174,"name":"Tim Simmons","email":"tim.r.simmons@gmail.com","username":"TimSimmons"},{"_account_id":8864,"name":"Artom Lifshitz","email":"notartom@gmail.com","username":"artom"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2013-11-11 16:31:10.000000000","updated_by":{"_account_id":8864,"name":"Artom Lifshitz","email":"notartom@gmail.com","username":"artom"},"reviewer":{"_account_id":8864,"name":"Artom Lifshitz","email":"notartom@gmail.com","username":"artom"},"state":"REVIEWER"},{"updated":"2013-11-11 16:31:41.000000000","updated_by":{"_account_id":8174,"name":"Tim Simmons","email":"tim.r.simmons@gmail.com","username":"TimSimmons"},"reviewer":{"_account_id":8174,"name":"Tim Simmons","email":"tim.r.simmons@gmail.com","username":"TimSimmons"},"state":"REVIEWER"},{"updated":"2013-11-11 16:31:50.000000000","updated_by":{"_account_id":8098,"name":"Joe McBride","email":"jmcbride@rackspace.com","username":"mejoe"},"reviewer":{"_account_id":8098,"name":"Joe McBride","email":"jmcbride@rackspace.com","username":"mejoe"},"state":"REVIEWER"},{"updated":"2013-11-15 21:33:29.000000000","updated_by":{"_account_id":8130,"name":"Betsy Luzader","email":"betsy.luzader@rackspace.com","username":"bluzader"},"reviewer":{"_account_id":8130,"name":"Betsy Luzader","email":"betsy.luzader@rackspace.com","username":"bluzader"},"state":"REVIEWER"},{"updated":"2013-11-19 13:16:39.000000000","updated_by":{"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},"reviewer":{"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},"state":"REVIEWER"},{"updated":"2013-11-19 14:27:32.000000000","updated_by":{"_account_id":8101,"name":"Emmanuel Ankutse","email":"emmanuel.ankutse@rackspace.com","username":"emmanuel.ankutse"},"reviewer":{"_account_id":8101,"name":"Emmanuel Ankutse","email":"emmanuel.ankutse@rackspace.com","username":"emmanuel.ankutse"},"state":"REVIEWER"},{"updated":"2013-11-19 14:56:14.000000000","updated_by":{"_account_id":741,"name":"Kiall Mac Innes","email":"kiall@macinnes.ie","username":"kiall"},"reviewer":{"_account_id":741,"name":"Kiall Mac Innes","email":"kiall@macinnes.ie","username":"kiall"},"state":"REVIEWER"},{"updated":"2013-11-19 15:54:23.000000000","updated_by":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"reviewer":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"state":"REVIEWER"}],"messages":[{"id":"4ae16f961b86cdda14b864bfe142903bde42b020","author":{"_account_id":8864,"name":"Artom Lifshitz","email":"notartom@gmail.com","username":"artom"},"date":"2013-11-11 19:53:50.000000000","message":"Patch Set 1:\n\nTo summarize what was said in the IRC channel (and to keep discussion centralized and easy to find and follow):\n\n1. My opinion (for what it\u0027s worth) is that this seems like the kind of thing that should be split off into a separate module altogether, or at least its own class/component within Designate. Central would just call is_effective_tld(suffix) for example.\n\n2. There was debate around how to ship and keep current the list. There are multiple sources (IANA, Mozilla\u0027s publicsuffix). Do we ship the files with Desigante? Do we update them from the \u0027net? If the latter, how often? Also if the latter, operators should probably have the option to override/modify the lists and to control when it gets updated from the \u0027net.\n\nMy 0.62 rubles ;)","accounts_in_message":[],"_revision_number":1},{"id":"99e554338edf9989cd5d55ff99e0af9fe9e4be66","author":{"_account_id":741,"name":"Kiall Mac Innes","email":"kiall@macinnes.ie","username":"kiall"},"date":"2013-11-12 11:17:43.000000000","message":"Patch Set 1:\n\nI think having a database table for something as static as these seems excessive, and when we eventually get around to implementing live config reloading, we can do the same with these lists.. \n\nIt\u0027s disappointing that none of the public suffix libraries on pypi are useful :( Oh well.\n\nI like artom\u0027s idea from IRC, where we turn the bulk of this into a module on it\u0027s own (e.g designate/publicsuffix.py) - central/service.py is getting HUGE, and is due some refactoring anyway.. Let\u0027s not add more work to the pile :)\n\nI think \"public suffix\" is a suitable name for handling both the Mozilla and IANA lists, as it\u0027s descriptive of intent rather than the source of the information.\n\nRe including in-tree - I\u0027ve done some research, and I believe we can include both IANA and Mozilla lists in-tree.. The IANA list is public domain, and the Mozilla list is MPL 2.0. See [1].\n\nFinally, if we want to merge this into the \"stable/havana\" branch, we need to avoid changing the existing default behaviors. Having 2 config options for \"Use IANA list\" and \"Use Public Suffix list\", both enabled by default in master, and only the IANA one enabled in stable/havana is probably the way to go.\n\n[1]: http://www.mozilla.org/MPL/2.0/FAQ.html#mpl-bsd-and-apache","accounts_in_message":[],"_revision_number":1},{"id":"924c6b7c6de5017d9a3088242cefe98a305ef41a","author":{"_account_id":8094,"name":"Vinod Mangalpally","email":"mrvinod@outlook.com","username":"vinodmr"},"date":"2013-11-12 20:37:08.000000000","message":"Patch Set 1:\n\nI moved the tld related code to a separate module - central/effectivetld.py to reflect its intent.\n\nAs for punycode - I found that encoding the publicsuffix list to idna works correctly.  If I encode it to punycode, it does not put the \"xn--\" prefix.  Additionally in punycode normal ascii strings are suffixed with a -.  With the idna format everything works as expected.","accounts_in_message":[],"_revision_number":1},{"id":"cf5b3ed0fd54d4ad0f133ecff16052e777b834e2","author":{"_account_id":8094,"name":"Vinod Mangalpally","email":"mrvinod@outlook.com","username":"vinodmr"},"date":"2013-11-12 20:37:29.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"b8b95c20c674f55845d6ec5530888fb782349ffb","author":{"_account_id":8094,"name":"Vinod Mangalpally","email":"mrvinod@outlook.com","username":"vinodmr"},"date":"2013-11-12 20:39:13.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"7284ddf0805d1d15db88d3130fae4c5d9f0f0c21","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2013-11-12 20:40:39.000000000","message":"Patch Set 3: Doesn\u0027t seem to work\n\nBuild failed.  For information on how to proceed, see https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures\n\n- gate-designate-pep8 http://logs.openstack.org/71/55871/3/check/gate-designate-pep8/a826e63 : SUCCESS in 1m 12s\n- gate-designate-python26 http://logs.openstack.org/71/55871/3/check/gate-designate-python26/96e0f13 : FAILURE in 1m 21s\n- gate-designate-python27 http://logs.openstack.org/71/55871/3/check/gate-designate-python27/1cd125b : FAILURE in 1m 23s\n","accounts_in_message":[],"_revision_number":3},{"id":"1d1ad0c50dbcebfb1ef17878c821c6e259cd4af2","author":{"_account_id":8094,"name":"Vinod Mangalpally","email":"mrvinod@outlook.com","username":"vinodmr"},"date":"2013-11-12 20:44:44.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"85ab91b96b797b29fa97e74a8f8528753568edf1","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2013-11-12 20:47:11.000000000","message":"Patch Set 4: Works for me\n\nBuild succeeded.\n\n- gate-designate-pep8 http://logs.openstack.org/71/55871/4/check/gate-designate-pep8/338473a : SUCCESS in 1m 11s\n- gate-designate-python26 http://logs.openstack.org/71/55871/4/check/gate-designate-python26/ecacc23 : SUCCESS in 2m 24s\n- gate-designate-python27 http://logs.openstack.org/71/55871/4/check/gate-designate-python27/71ed0d5 : SUCCESS in 2m 03s\n","accounts_in_message":[],"_revision_number":4},{"id":"24eb243968bd2125125d743b5c116f3089e12837","author":{"_account_id":741,"name":"Kiall Mac Innes","email":"kiall@macinnes.ie","username":"kiall"},"date":"2013-11-13 11:25:36.000000000","message":"Patch Set 4: (6 inline comments)\n\nLooking good, a couple of inline comments.","accounts_in_message":[],"_revision_number":4},{"id":"9019425ea5b7f66f0d7365fc110fd08670b92858","author":{"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},"date":"2013-11-13 16:11:04.000000000","message":"Patch Set 4: I would prefer that you didn\u0027t merge this\n\n(1 inline comment)\n\nLooks good, I would just prefer the error be raised in central/service.py (see inline comment)","accounts_in_message":[],"_revision_number":4},{"id":"6e87667ec8a3cbe02e3fd85237fbd7f9153978c2","author":{"_account_id":8094,"name":"Vinod Mangalpally","email":"mrvinod@outlook.com","username":"vinodmr"},"date":"2013-11-13 21:49:58.000000000","message":"Patch Set 4: (7 inline comments)\n\n","accounts_in_message":[],"_revision_number":4},{"id":"644159eab83d45998fab3457f7a91c9ef07b65ab","author":{"_account_id":8094,"name":"Vinod Mangalpally","email":"mrvinod@outlook.com","username":"vinodmr"},"date":"2013-11-13 21:50:06.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"fd80742c1b79c7948cad56669f7b43e752048871","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2013-11-13 21:52:28.000000000","message":"Patch Set 5: Works for me\n\nBuild succeeded.\n\n- gate-designate-pep8 http://logs.openstack.org/71/55871/5/check/gate-designate-pep8/6810d8b : SUCCESS in 1m 17s\n- gate-designate-python26 http://logs.openstack.org/71/55871/5/check/gate-designate-python26/804c249 : SUCCESS in 2m 20s\n- gate-designate-python27 http://logs.openstack.org/71/55871/5/check/gate-designate-python27/9d6a6d9 : SUCCESS in 2m 14s\n","accounts_in_message":[],"_revision_number":5},{"id":"c2cdc88cc30f4040be1a163d0b0ea7232bbd1bde","author":{"_account_id":741,"name":"Kiall Mac Innes","email":"kiall@macinnes.ie","username":"kiall"},"date":"2013-11-14 09:34:36.000000000","message":"Patch Set 5: (3 inline comments)\n\n","accounts_in_message":[],"_revision_number":5},{"id":"2657edb324fc5d2c5bbed87d0f56f0fc3fd3d360","author":{"_account_id":8094,"name":"Vinod Mangalpally","email":"mrvinod@outlook.com","username":"vinodmr"},"date":"2013-11-14 16:01:48.000000000","message":"Patch Set 5: (3 inline comments)\n\n","accounts_in_message":[],"_revision_number":5},{"id":"10e2757f585b99c0ead0c017c50d82c14c6bf6c5","author":{"_account_id":8094,"name":"Vinod Mangalpally","email":"mrvinod@outlook.com","username":"vinodmr"},"date":"2013-11-14 16:09:21.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"a89e4b0abc65703ba1ea3f4ead4c43457c48bb43","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2013-11-14 16:12:00.000000000","message":"Patch Set 6: Works for me\n\nBuild succeeded.\n\n- gate-designate-pep8 http://logs.openstack.org/71/55871/6/check/gate-designate-pep8/67485d5 : SUCCESS in 1m 33s\n- gate-designate-python26 http://logs.openstack.org/71/55871/6/check/gate-designate-python26/b064160 : SUCCESS in 2m 36s\n- gate-designate-python27 http://logs.openstack.org/71/55871/6/check/gate-designate-python27/d7abf88 : SUCCESS in 2m 13s\n","accounts_in_message":[],"_revision_number":6},{"id":"d9c5539d37d893fb36709735484e9475716a6c27","author":{"_account_id":8101,"name":"Emmanuel Ankutse","email":"emmanuel.ankutse@rackspace.com","username":"emmanuel.ankutse"},"date":"2013-11-15 15:40:34.000000000","message":"Patch Set 6: I would prefer that you didn\u0027t merge this\n\n(2 inline comments)\n\n","accounts_in_message":[],"_revision_number":6},{"id":"76dbfcf9b888a77e387efd0bddc04e843de0a63f","author":{"_account_id":8094,"name":"Vinod Mangalpally","email":"mrvinod@outlook.com","username":"vinodmr"},"date":"2013-11-15 16:20:26.000000000","message":"Patch Set 6: (2 inline comments)\n\n","accounts_in_message":[],"_revision_number":6},{"id":"bf6b92dae3cf6ba758d79a2faaf87556ca6877a0","author":{"_account_id":8130,"name":"Betsy Luzader","email":"betsy.luzader@rackspace.com","username":"bluzader"},"date":"2013-11-15 21:33:29.000000000","message":"Patch Set 6: I would prefer that you didn\u0027t merge this\n\nI\u0027m not convinced that we should include the \"official\" effective_tld_names_.dat and tlds-alpha-by-domain.txt files in the Designate code base. I don\u0027t think Designate should be responsible for keeping this data up to date, or even make it look like we\u0027re keeping it up to date.\n\nI think it would be a better idea to put in a sample file of each one, like we do with the config file. For example, we could have an effective_tld_names.dat.sample file with a few examples to show the layout of the file and instructions on where to go get the latest data.","accounts_in_message":[],"_revision_number":6},{"id":"e51ebf3a01417f589f3132cbf3ea7f8297c75649","author":{"_account_id":741,"name":"Kiall Mac Innes","email":"kiall@macinnes.ie","username":"kiall"},"date":"2013-11-18 11:10:17.000000000","message":"Patch Set 6:\n\nI think I agree with Betsy here, I like the idea of working out of the box.. But don\u0027t want to be \"held responsible for\" keeping the list up to date.\n\n\nProviding the latest copy at the time of commit as a sample file, then allowing users to use it or download a up-to-date copy makes sense to me.","accounts_in_message":[],"_revision_number":6},{"id":"2567e10f712dcd8dec50fa11baa667229e1d2ec5","author":{"_account_id":8094,"name":"Vinod Mangalpally","email":"mrvinod@outlook.com","username":"vinodmr"},"date":"2013-11-18 21:56:50.000000000","message":"Patch Set 6:\n\nRenamed the accepted and effective lists as samples.  Updated the tests accordingly.","accounts_in_message":[],"_revision_number":6},{"id":"23f9bfe8754c020e05eacc9695dc58f2c9e18daa","author":{"_account_id":8094,"name":"Vinod Mangalpally","email":"mrvinod@outlook.com","username":"vinodmr"},"date":"2013-11-18 22:12:33.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"6c6b97cf6771839773fdafb74996d6e94a25e140","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2013-11-18 23:09:20.000000000","message":"Patch Set 7: Works for me\n\nBuild succeeded.\n\n- gate-designate-pep8 http://logs.openstack.org/71/55871/7/check/gate-designate-pep8/505dc67 : SUCCESS in 1m 36s\n- gate-designate-python26 http://logs.openstack.org/71/55871/7/check/gate-designate-python26/614ebd5 : SUCCESS in 1m 39s\n- gate-designate-python27 http://logs.openstack.org/71/55871/7/check/gate-designate-python27/1450ed6 : SUCCESS in 1m 49s\n","accounts_in_message":[],"_revision_number":7},{"id":"7046cfdfe94cf6f9961cfbb653712eedd500a42b","author":{"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},"date":"2013-11-19 13:16:39.000000000","message":"Patch Set 7: (1 inline comment)\n\nWith the commit message fix, looks good to me","accounts_in_message":[],"_revision_number":7},{"id":"f86abf81ad93277a15f7a2a7bf4171bb046eca94","author":{"_account_id":8101,"name":"Emmanuel Ankutse","email":"emmanuel.ankutse@rackspace.com","username":"emmanuel.ankutse"},"date":"2013-11-19 14:27:32.000000000","message":"Patch Set 7: Looks good to me, but someone else must approve\n\n(1 inline comment)\n\n","accounts_in_message":[],"_revision_number":7},{"id":"f94c95e1824459421b38b1eb8fb0ebb50226ed38","author":{"_account_id":8094,"name":"Vinod Mangalpally","email":"mrvinod@outlook.com","username":"vinodmr"},"date":"2013-11-19 14:54:10.000000000","message":"Uploaded patch set 8.","accounts_in_message":[],"_revision_number":8},{"id":"e0ab76f1d553820c9c43c5e32262526edd7d4521","author":{"_account_id":5494,"name":"Trivial Rebase","email":"trivial-rebase@review.openstack.org","username":"trivial-rebase"},"date":"2013-11-19 14:54:17.000000000","message":"Patch Set 8:\n\nNew patchset patch-id matches previous patchset, but commit message has changed.","accounts_in_message":[],"_revision_number":8},{"id":"6db5aad1b21f4335079d3cbd1539d66e6d22ffd8","author":{"_account_id":741,"name":"Kiall Mac Innes","email":"kiall@macinnes.ie","username":"kiall"},"date":"2013-11-19 14:56:14.000000000","message":"Patch Set 8: Looks good to me (core reviewer); Approved\n\nExcellent :) Looks good!\n\nThanks Vinod","accounts_in_message":[],"_revision_number":8},{"id":"275d3ba53d41f835f159d837f645fa18fcc0c515","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2013-11-19 14:56:15.000000000","message":"Patch Set 8:\n\nStarting gate jobs.\nhttp://status.openstack.org/zuul/","accounts_in_message":[],"_revision_number":8},{"id":"ece18abe1f333759491677373cdd72623cdaab91","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2013-11-19 15:52:53.000000000","message":"Patch Set 8: Works for me\n\nBuild succeeded.\n\n- gate-designate-pep8 http://logs.openstack.org/71/55871/8/check/gate-designate-pep8/d0ed464 : SUCCESS in 1m 25s\n- gate-designate-python26 http://logs.openstack.org/71/55871/8/check/gate-designate-python26/e1d11f5 : SUCCESS in 1m 44s\n- gate-designate-python27 http://logs.openstack.org/71/55871/8/check/gate-designate-python27/814da2c : SUCCESS in 1m 32s\n","accounts_in_message":[],"_revision_number":8},{"id":"240d8af402cc9169b74cc3a4a2aaa3e8af491637","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2013-11-19 15:54:23.000000000","message":"Change has been successfully merged into the git repository.","accounts_in_message":[],"_revision_number":8},{"id":"dec4397070a05312c0f58625fe02b6f9308e7b8d","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2013-11-19 15:54:23.000000000","message":"Patch Set 8: Verified\n\nBuild succeeded.\n\n- gate-designate-pep8 http://logs.openstack.org/71/55871/8/gate/gate-designate-pep8/ef7f667 : SUCCESS in 1m 30s\n- gate-designate-python26 http://logs.openstack.org/71/55871/8/gate/gate-designate-python26/591856c : SUCCESS in 1m 46s\n- gate-designate-python27 http://logs.openstack.org/71/55871/8/gate/gate-designate-python27/2d3c047 : SUCCESS in 1m 35s\n","accounts_in_message":[],"_revision_number":8}],"current_revision_number":8,"current_revision":"3db98ab1fd1596a544f84c4ad0a6314383ed6629","revisions":{"a643ead8cafea2694e16aae78497a0d1edd82645":{"kind":"REWORK","_number":1,"created":"2013-11-11 16:28:09.000000000","uploader":{"_account_id":8094,"name":"Vinod Mangalpally","email":"mrvinod@outlook.com","username":"vinodmr"},"ref":"refs/changes/71/55871/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/71/55871/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/71/55871/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/71/55871/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/71/55871/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/71/55871/1"}}},"commit":{"parents":[{"commit":"6df678bd14958ca287575f431e0ed6312337d18c","subject":"Merge \"Update TLD list per IANA list version 2013110700\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/6df678bd14958ca287575f431e0ed6312337d18c"}]}],"author":{"name":"Vinod Mangalpally","email":"vinod.mang@rackspace.com","date":"2013-11-11 16:21:40.000000000","tz":-360},"committer":{"name":"Vinod Mangalpally","email":"vinod.mang@rackspace.com","date":"2013-11-11 16:21:40.000000000","tz":-360},"subject":"Add SLDs that act like TLDs","message":"Add SLDs that act like TLDs\n\nThe SLD list is from publicsuffix.org.  This needs to be present at\netc/designate/effective_tld_names.dat.  This is parsed so that lines\nbeginning with a * are put in a regular expression list.  Lines\nbeginning\nwith a ! are ignored presently.  All other valid lines are put in a\ndictionary.  Entries with only 1 label are also ignored as we check that\nthe domain has more than 1 label.\n\nI also moved the TLDs to a separate list at\netc/designate/tlds-alpha-by-domain.txt.  This is so that it is easier\nto update the list.  Also it makes handling the TLDs and SLDs that\nact as TLDs similar.\n\nChange-Id: I9718489335e24cb24733d4c4fc2a966e490014d3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/a643ead8cafea2694e16aae78497a0d1edd82645"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/a643ead8cafea2694e16aae78497a0d1edd82645"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"6df678bd14958ca287575f431e0ed6312337d18c","is_merged_in_target_branch":true}],"branch":"refs/heads/master"},"c705c023aa37ecdf2dd77f7e5541dc63f6b44670":{"kind":"REWORK","_number":2,"created":"2013-11-12 20:37:29.000000000","uploader":{"_account_id":8094,"name":"Vinod Mangalpally","email":"mrvinod@outlook.com","username":"vinodmr"},"ref":"refs/changes/71/55871/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/71/55871/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/71/55871/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/71/55871/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/71/55871/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/71/55871/2"}}},"commit":{"parents":[{"commit":"20fd3a90fb5a97b3de43776063d529fa9b69f814","subject":"Update PPA URL","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/20fd3a90fb5a97b3de43776063d529fa9b69f814"}]}],"author":{"name":"Vinod Mangalpally","email":"vinod.mang@rackspace.com","date":"2013-11-12 20:33:20.000000000","tz":-360},"committer":{"name":"Vinod Mangalpally","email":"vinod.mang@rackspace.com","date":"2013-11-12 20:33:20.000000000","tz":-360},"subject":"Add SLDs that act like TLDs","message":"Add SLDs that act like TLDs\n\nThe SLD list is from publicsuffix.org.  This needs to be present at\netc/designate/effective_tld_names.dat.  This is parsed so that lines\nbeginning with a * are put in a regular expression list.  Lines\nbeginning\nwith a ! are ignored presently.  All other valid lines are put in a\ndictionary.  Entries with only 1 label are also ignored as we check that\nthe domain has more than 1 label.\n\nI also moved the TLDs to a separate list at\netc/designate/tlds-alpha-by-domain.txt.  This is so that it is easier\nto update the list.  Also it makes handling the TLDs and SLDs that\nact as TLDs similar.\n\nChange-Id: I9718489335e24cb24733d4c4fc2a966e490014d3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/c705c023aa37ecdf2dd77f7e5541dc63f6b44670"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/c705c023aa37ecdf2dd77f7e5541dc63f6b44670"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"20fd3a90fb5a97b3de43776063d529fa9b69f814","is_merged_in_target_branch":true,"change_id":"Idf4c1d401f3e9b4c84d3029f1f1935ab947c01f1","change_number":56054,"patch_set_number":1,"change_status":"MERGED"}],"branch":"refs/heads/master"},"2351275b62f8e44266b8770e9a8335a65d3e2ee3":{"kind":"REWORK","_number":3,"created":"2013-11-12 20:39:13.000000000","uploader":{"_account_id":8094,"name":"Vinod Mangalpally","email":"mrvinod@outlook.com","username":"vinodmr"},"ref":"refs/changes/71/55871/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/71/55871/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/71/55871/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/71/55871/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/71/55871/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/71/55871/3"}}},"commit":{"parents":[{"commit":"20fd3a90fb5a97b3de43776063d529fa9b69f814","subject":"Update PPA URL","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/20fd3a90fb5a97b3de43776063d529fa9b69f814"}]}],"author":{"name":"Vinod Mangalpally","email":"vinod.mang@rackspace.com","date":"2013-11-12 20:39:06.000000000","tz":-360},"committer":{"name":"Vinod Mangalpally","email":"vinod.mang@rackspace.com","date":"2013-11-12 20:39:06.000000000","tz":-360},"subject":"Add SLDs that act like TLDs","message":"Add SLDs that act like TLDs\n\nThe SLD list is from publicsuffix.org.  This needs to be present at\netc/designate/effective_tld_names.dat.  This is parsed so that lines\nbeginning with a * are put in a regular expression list.  Lines\nbeginning with a ! are ignored presently.  All other valid lines are\nput in a\ndictionary.  Entries with only 1 label are also ignored as we check that\nthe domain has more than 1 label.\n\nI also moved the TLDs to a separate list at\netc/designate/tlds-alpha-by-domain.txt.  This is so that it is easier\nto update the list.  Also it makes handling the TLDs and SLDs that\nact as TLDs similar.\n\nChange-Id: I9718489335e24cb24733d4c4fc2a966e490014d3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/2351275b62f8e44266b8770e9a8335a65d3e2ee3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/2351275b62f8e44266b8770e9a8335a65d3e2ee3"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"20fd3a90fb5a97b3de43776063d529fa9b69f814","is_merged_in_target_branch":true,"change_id":"Idf4c1d401f3e9b4c84d3029f1f1935ab947c01f1","change_number":56054,"patch_set_number":1,"change_status":"MERGED"}],"branch":"refs/heads/master"},"42bb6f8a44cccf1412644e7ad39935c9c58c595d":{"kind":"REWORK","_number":4,"created":"2013-11-12 20:44:44.000000000","uploader":{"_account_id":8094,"name":"Vinod Mangalpally","email":"mrvinod@outlook.com","username":"vinodmr"},"ref":"refs/changes/71/55871/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/71/55871/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/71/55871/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/71/55871/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/71/55871/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/71/55871/4"}}},"commit":{"parents":[{"commit":"20fd3a90fb5a97b3de43776063d529fa9b69f814","subject":"Update PPA URL","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/20fd3a90fb5a97b3de43776063d529fa9b69f814"}]}],"author":{"name":"Vinod Mangalpally","email":"vinod.mang@rackspace.com","date":"2013-11-12 20:44:12.000000000","tz":-360},"committer":{"name":"Vinod Mangalpally","email":"vinod.mang@rackspace.com","date":"2013-11-12 20:44:12.000000000","tz":-360},"subject":"Add SLDs that act like TLDs","message":"Add SLDs that act like TLDs\n\nThe SLD list is from publicsuffix.org.  This needs to be present at\netc/designate/effective_tld_names.dat.  This is parsed so that lines\nbeginning with a * are put in a regular expression list.\n\nLines beginning with a ! are ignored presently.  All other valid lines\nare put in a dictionary.  Entries with only 1 label are also ignored\nas we check that the domain has more than 1 label.\n\nAdditionally all the entries from publicsuffix.org are converted to\nidna format.\n\nI also moved the TLDs to a separate list at\netc/designate/tlds-alpha-by-domain.txt.  This is so that it is easier\nto update the list.  Also it makes handling the TLDs and SLDs that\nact as TLDs similar.\n\nChange-Id: I9718489335e24cb24733d4c4fc2a966e490014d3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/42bb6f8a44cccf1412644e7ad39935c9c58c595d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/42bb6f8a44cccf1412644e7ad39935c9c58c595d"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"20fd3a90fb5a97b3de43776063d529fa9b69f814","is_merged_in_target_branch":true,"change_id":"Idf4c1d401f3e9b4c84d3029f1f1935ab947c01f1","change_number":56054,"patch_set_number":1,"change_status":"MERGED"}],"branch":"refs/heads/master"},"fd5e3fe4c645b30193a112b6412d2d2d48874628":{"kind":"REWORK","_number":5,"created":"2013-11-13 21:50:06.000000000","uploader":{"_account_id":8094,"name":"Vinod Mangalpally","email":"mrvinod@outlook.com","username":"vinodmr"},"ref":"refs/changes/71/55871/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/71/55871/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/71/55871/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/71/55871/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/71/55871/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/71/55871/5"}}},"commit":{"parents":[{"commit":"20fd3a90fb5a97b3de43776063d529fa9b69f814","subject":"Update PPA URL","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/20fd3a90fb5a97b3de43776063d529fa9b69f814"}]}],"author":{"name":"Vinod Mangalpally","email":"vinod.mang@rackspace.com","date":"2013-11-13 21:42:24.000000000","tz":-360},"committer":{"name":"Vinod Mangalpally","email":"vinod.mang@rackspace.com","date":"2013-11-13 21:42:24.000000000","tz":-360},"subject":"Add SLDs that act like TLDs","message":"Add SLDs that act like TLDs\n\nThe SLD list is from publicsuffix.org.  This needs to be present at\netc/designate/effective_tld_names.dat.  This is parsed so that lines\nbeginning with a * are put in a regular expression list.\n\nLines beginning with a ! are ignored presently.  All other valid lines\nare put in a dictionary.  Entries with only 1 label are also ignored\nas we check that the domain has more than 1 label.\n\nAdditionally all the entries from publicsuffix.org are converted to\nidna format.\n\nI also moved the TLDs to a separate list at\netc/designate/tlds-alpha-by-domain.txt.  This is so that it is easier\nto update the list.  Also it makes handling the TLDs and SLDs that\nact as TLDs similar.\n\nChange-Id: I9718489335e24cb24733d4c4fc2a966e490014d3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/fd5e3fe4c645b30193a112b6412d2d2d48874628"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/fd5e3fe4c645b30193a112b6412d2d2d48874628"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"20fd3a90fb5a97b3de43776063d529fa9b69f814","is_merged_in_target_branch":true,"change_id":"Idf4c1d401f3e9b4c84d3029f1f1935ab947c01f1","change_number":56054,"patch_set_number":1,"change_status":"MERGED"}],"branch":"refs/heads/master"},"b591e15fc3a29e45ccfc9710aaab4c8727847b96":{"kind":"REWORK","_number":6,"created":"2013-11-14 16:09:21.000000000","uploader":{"_account_id":8094,"name":"Vinod Mangalpally","email":"mrvinod@outlook.com","username":"vinodmr"},"ref":"refs/changes/71/55871/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/71/55871/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/71/55871/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/71/55871/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/71/55871/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/71/55871/6"}}},"commit":{"parents":[{"commit":"20fd3a90fb5a97b3de43776063d529fa9b69f814","subject":"Update PPA URL","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/20fd3a90fb5a97b3de43776063d529fa9b69f814"}]}],"author":{"name":"Vinod Mangalpally","email":"vinod.mang@rackspace.com","date":"2013-11-14 16:08:10.000000000","tz":-360},"committer":{"name":"Vinod Mangalpally","email":"vinod.mang@rackspace.com","date":"2013-11-14 16:08:10.000000000","tz":-360},"subject":"Add SLDs that act like TLDs","message":"Add SLDs that act like TLDs\n\nThe SLD list is from publicsuffix.org.  This needs to be present at\netc/designate/effective_tld_names.dat.  This is parsed so that lines\nbeginning with a * are put in a regular expression list.\n\nLines beginning with a ! are ignored presently.  All other valid lines\nare put in a dictionary.  Entries with only 1 label are also ignored\nas we check that the domain has more than 1 label.\n\nAdditionally all the entries from publicsuffix.org are converted to\nidna format.\n\nI also moved the TLDs to a separate list at\netc/designate/tlds-alpha-by-domain.txt.  This is so that it is easier\nto update the list.  Also it makes handling the TLDs and SLDs that\nact as TLDs similar.\n\nChange-Id: I9718489335e24cb24733d4c4fc2a966e490014d3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/b591e15fc3a29e45ccfc9710aaab4c8727847b96"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/b591e15fc3a29e45ccfc9710aaab4c8727847b96"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"20fd3a90fb5a97b3de43776063d529fa9b69f814","is_merged_in_target_branch":true,"change_id":"Idf4c1d401f3e9b4c84d3029f1f1935ab947c01f1","change_number":56054,"patch_set_number":1,"change_status":"MERGED"}],"branch":"refs/heads/master"},"0370efe482576f9768077218fc7d1c1fec7cc8b5":{"kind":"REWORK","_number":7,"created":"2013-11-18 22:12:33.000000000","uploader":{"_account_id":8094,"name":"Vinod Mangalpally","email":"mrvinod@outlook.com","username":"vinodmr"},"ref":"refs/changes/71/55871/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/71/55871/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/71/55871/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/71/55871/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/71/55871/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/71/55871/7"}}},"commit":{"parents":[{"commit":"a5f6a1fb3d5e64553f07e3e67c2fd1094d43cab0","subject":"Merge \"Add .testrepository to ST2 project ignore\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/a5f6a1fb3d5e64553f07e3e67c2fd1094d43cab0"}]}],"author":{"name":"Vinod Mangalpally","email":"vinod.mang@rackspace.com","date":"2013-11-18 21:54:45.000000000","tz":-360},"committer":{"name":"Vinod Mangalpally","email":"vinod.mang@rackspace.com","date":"2013-11-18 21:54:45.000000000","tz":-360},"subject":"Add SLDs that act like TLDs","message":"Add SLDs that act like TLDs\n\nThe SLD list is from publicsuffix.org.  This needs to be present at\netc/designate/effective_tld_names.dat.  This is parsed so that lines\nbeginning with a * are put in a regular expression list.\n\nLines beginning with a ! are ignored presently.  All other valid lines\nare put in a dictionary.  Entries with only 1 label are also ignored\nas we check that the domain has more than 1 label.\n\nAdditionally all the entries from publicsuffix.org are converted to\nidna format.\n\nI also moved the TLDs to a separate list at\netc/designate/tlds-alpha-by-domain.txt.  This is so that it is easier\nto update the list.  Also it makes handling the TLDs and SLDs that\nact as TLDs similar.\n\nChange-Id: I9718489335e24cb24733d4c4fc2a966e490014d3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/0370efe482576f9768077218fc7d1c1fec7cc8b5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/0370efe482576f9768077218fc7d1c1fec7cc8b5"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"a5f6a1fb3d5e64553f07e3e67c2fd1094d43cab0","is_merged_in_target_branch":true}],"branch":"refs/heads/master"},"3db98ab1fd1596a544f84c4ad0a6314383ed6629":{"kind":"NO_CODE_CHANGE","_number":8,"created":"2013-11-19 14:54:10.000000000","uploader":{"_account_id":8094,"name":"Vinod Mangalpally","email":"mrvinod@outlook.com","username":"vinodmr"},"ref":"refs/changes/71/55871/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/71/55871/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/71/55871/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/71/55871/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/71/55871/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/71/55871/8"}}},"commit":{"parents":[{"commit":"a5f6a1fb3d5e64553f07e3e67c2fd1094d43cab0","subject":"Merge \"Add .testrepository to ST2 project ignore\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/a5f6a1fb3d5e64553f07e3e67c2fd1094d43cab0"}]}],"author":{"name":"Vinod Mangalpally","email":"vinod.mang@rackspace.com","date":"2013-11-19 14:53:53.000000000","tz":-360},"committer":{"name":"Vinod Mangalpally","email":"vinod.mang@rackspace.com","date":"2013-11-19 14:53:53.000000000","tz":-360},"subject":"Add SLDs that act like TLDs","message":"Add SLDs that act like TLDs\n\nThis change offers the designate administrator the ability to configure\n2 lists to indicate the TLDs(accepted_tlds_file) and SLDs and others\nthat\neffectively act as TLDs.(effective_tlds_file)\n\n2 sample files are included to show the format required for these files.\nThe TLD file is from http://data.iana.org/TLD/tlds-alpha-by-domain.txt\nCommented lines in this file begin with a #. There is one entry per line\nand the entries are in the IDNA format.\n\nThe sample effective TLD file is from http://publicsuffix.org/list/\nThis file is in the UTF-8 format.  Lines beginning with a // and ! are\nignored.  The wildcard character * (asterisk) may only be used to\nwildcard\nthe topmost level in a domain name.  All the entries  are converted to\nidna format.\n\nCloses-Bug: #1249396\n\nChange-Id: I9718489335e24cb24733d4c4fc2a966e490014d3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/3db98ab1fd1596a544f84c4ad0a6314383ed6629"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/3db98ab1fd1596a544f84c4ad0a6314383ed6629"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"a5f6a1fb3d5e64553f07e3e67c2fd1094d43cab0","is_merged_in_target_branch":true}],"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
