)]}'
{"id":"openstack%2Fdesignate~643725","triplet_id":"openstack%2Fdesignate~master~I3be68bf294157631a9bac128df0841220020de96","project":"openstack/designate","branch":"master","topic":"central_cfg","hashtags":[],"change_id":"I3be68bf294157631a9bac128df0841220020de96","subject":"Moved most configuration into designate.conf","status":"MERGED","created":"2019-03-17 01:05:31.000000000","updated":"2019-06-09 07:50:15.000000000","submitted":"2019-06-09 07:50:15.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"643725-1560066615504-0eece10d","meta_rev_id":"811654963315a3947bb4a905cc92e22f4ed513f5","_number":643725,"virtual_id_number":643725,"owner":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},{"value":0,"_account_id":16006,"name":"goldyfruit","email":"gaetan.trellu@incloudus.com","username":"goldyfruit"},{"value":2,"date":"2019-06-09 07:50:15.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},"all":[{"value":2,"date":"2019-06-04 17:07:45.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},{"value":1,"date":"2019-05-28 23:54:00.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16006,"name":"goldyfruit","email":"gaetan.trellu@incloudus.com","username":"goldyfruit"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"all":[{"value":0,"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},{"value":0,"_account_id":16006,"name":"goldyfruit","email":"gaetan.trellu@incloudus.com","username":"goldyfruit"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2019-06-09 06:44:07.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},{"value":0,"_account_id":16006,"name":"goldyfruit","email":"gaetan.trellu@incloudus.com","username":"goldyfruit"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},{"_account_id":16006,"name":"goldyfruit","email":"gaetan.trellu@incloudus.com","username":"goldyfruit"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-04-01 19:21:24.000000000","updated_by":{"_account_id":16006,"name":"goldyfruit","email":"gaetan.trellu@incloudus.com","username":"goldyfruit"},"reviewer":{"_account_id":16006,"name":"goldyfruit","email":"gaetan.trellu@incloudus.com","username":"goldyfruit"},"state":"REVIEWER"},{"updated":"2019-06-04 17:07:45.000000000","updated_by":{"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},"reviewer":{"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},"state":"REVIEWER"},{"updated":"2019-06-09 07:50:15.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"13b70ecb1bceec28cde78c8523795a3919487a9a","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2019-03-17 01:05:31.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"0c3d20d5bac1254fd911d70293ee5f33d0f3d060","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2019-03-17 01:38:16.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"cebdf3dd3253c08187a4de7421eafb5d2c6c3adc","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-03-17 02:56:03.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- test-release-openstack http://logs.openstack.org/25/643725/2/check/test-release-openstack/c025417/ : SUCCESS in 5m 33s\n- designate-bind9 http://logs.openstack.org/25/643725/2/check/designate-bind9/9a183e7/ : TIMED_OUT in 1h 12m 44s\n- designate-bind9-manager-model http://logs.openstack.org/25/643725/2/check/designate-bind9-manager-model/ff0bade/ : TIMED_OUT in 1h 12m 22s\n- designate-bind9-py36 http://logs.openstack.org/25/643725/2/check/designate-bind9-py36/fd223da/ : TIMED_OUT in 1h 14m 03s\n- designate-devstack-pdns4 http://logs.openstack.org/25/643725/2/check/designate-devstack-pdns4/97c90a4/ : SUCCESS in 49m 27s\n- designate-devstack-pdns4-postgres http://logs.openstack.org/25/643725/2/check/designate-devstack-pdns4-postgres/fd5d592/ : SUCCESS in 54m 19s\n- designate-grenade-pdns4 http://logs.openstack.org/25/643725/2/check/designate-grenade-pdns4/0567982/ : SUCCESS in 1h 09m 49s\n- openstack-tox-cover http://logs.openstack.org/25/643725/2/check/openstack-tox-cover/a66ad91/cover/ : SUCCESS in 8m 26s\n- openstack-tox-lower-constraints http://logs.openstack.org/25/643725/2/check/openstack-tox-lower-constraints/9cdccaa/ : SUCCESS in 5m 00s\n- openstack-tox-pep8 http://logs.openstack.org/25/643725/2/check/openstack-tox-pep8/f4f9fec/ : FAILURE in 5m 18s\n- openstack-tox-py27 http://logs.openstack.org/25/643725/2/check/openstack-tox-py27/bab8d46/ : SUCCESS in 9m 11s\n- openstack-tox-py35 http://logs.openstack.org/25/643725/2/check/openstack-tox-py35/5b0cb90/ : SUCCESS in 5m 08s\n- openstack-tox-py36 http://logs.openstack.org/25/643725/2/check/openstack-tox-py36/e87a3cd/ : SUCCESS in 5m 27s\n- openstack-tox-py37 http://logs.openstack.org/25/643725/2/check/openstack-tox-py37/8c912c3/ : SUCCESS in 8m 00s\n- openstack-tox-docs http://logs.openstack.org/25/643725/2/check/openstack-tox-docs/6bc492c/html/ : SUCCESS in 4m 36s\n- neutron-tempest-plugin-designate-scenario http://logs.openstack.org/25/643725/2/check/neutron-tempest-plugin-designate-scenario/11f0026/ : FAILURE in 38m 56s","accounts_in_message":[],"_revision_number":2},{"id":"ba0acd928725eacb46ce009b36371c72b5f593eb","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2019-03-17 04:00:45.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"414a3078061c059ed7b9cc9f66b1b87f11998d8c","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2019-03-17 04:10:41.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"3aaaf64499dd7c5a076d1ae22e80a19bd4c21863","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2019-03-17 04:11:59.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"edd3d707fee9801264ed693c62529cf1cf5d8883","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-03-17 05:08:29.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- test-release-openstack http://logs.openstack.org/25/643725/5/check/test-release-openstack/6068c53/ : SUCCESS in 3m 13s\n- designate-bind9 http://logs.openstack.org/25/643725/5/check/designate-bind9/7b79e39/ : SUCCESS in 33m 56s\n- designate-bind9-manager-model http://logs.openstack.org/25/643725/5/check/designate-bind9-manager-model/7f30796/ : SUCCESS in 38m 56s\n- designate-bind9-py36 http://logs.openstack.org/25/643725/5/check/designate-bind9-py36/0433b59/ : SUCCESS in 35m 15s\n- designate-devstack-pdns4 http://logs.openstack.org/25/643725/5/check/designate-devstack-pdns4/201c69c/ : SUCCESS in 31m 35s\n- designate-devstack-pdns4-postgres http://logs.openstack.org/25/643725/5/check/designate-devstack-pdns4-postgres/7935b4e/ : SUCCESS in 36m 21s\n- designate-grenade-pdns4 http://logs.openstack.org/25/643725/5/check/designate-grenade-pdns4/d999295/ : SUCCESS in 53m 14s\n- openstack-tox-cover http://logs.openstack.org/25/643725/5/check/openstack-tox-cover/f18c056/cover/ : SUCCESS in 6m 10s\n- openstack-tox-lower-constraints http://logs.openstack.org/25/643725/5/check/openstack-tox-lower-constraints/1168b38/ : SUCCESS in 3m 54s\n- openstack-tox-pep8 http://logs.openstack.org/25/643725/5/check/openstack-tox-pep8/99a06f6/ : SUCCESS in 5m 16s\n- openstack-tox-py27 http://logs.openstack.org/25/643725/5/check/openstack-tox-py27/ad247af/ : SUCCESS in 7m 09s\n- openstack-tox-py35 http://logs.openstack.org/25/643725/5/check/openstack-tox-py35/a8ed428/ : SUCCESS in 7m 06s\n- openstack-tox-py36 http://logs.openstack.org/25/643725/5/check/openstack-tox-py36/8298ad7/ : SUCCESS in 5m 25s\n- openstack-tox-py37 http://logs.openstack.org/25/643725/5/check/openstack-tox-py37/9cc02fd/ : SUCCESS in 5m 24s\n- openstack-tox-docs http://logs.openstack.org/25/643725/5/check/openstack-tox-docs/58bf5af/html/ : SUCCESS in 5m 03s\n- neutron-tempest-plugin-designate-scenario http://logs.openstack.org/25/643725/5/check/neutron-tempest-plugin-designate-scenario/3f7efed/ : SUCCESS in 32m 56s","accounts_in_message":[],"_revision_number":5},{"id":"d3178ef895d966cb4bcfe69f10df8e280596c836","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2019-03-17 06:09:23.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"f693e5a41bf50feac31c7c2f99940b7083b4a921","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-03-17 07:09:36.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\n\n- test-release-openstack http://logs.openstack.org/25/643725/6/check/test-release-openstack/d1aed8e/ : SUCCESS in 4m 00s\n- designate-bind9 http://logs.openstack.org/25/643725/6/check/designate-bind9/35d7261/ : SUCCESS in 39m 29s\n- designate-bind9-manager-model http://logs.openstack.org/25/643725/6/check/designate-bind9-manager-model/4678e5e/ : SUCCESS in 37m 36s\n- designate-bind9-py36 http://logs.openstack.org/25/643725/6/check/designate-bind9-py36/e606fd1/ : SUCCESS in 34m 39s\n- designate-devstack-pdns4 http://logs.openstack.org/25/643725/6/check/designate-devstack-pdns4/ec79494/ : SUCCESS in 33m 06s\n- designate-devstack-pdns4-postgres http://logs.openstack.org/25/643725/6/check/designate-devstack-pdns4-postgres/5bf5923/ : SUCCESS in 33m 10s\n- designate-grenade-pdns4 http://logs.openstack.org/25/643725/6/check/designate-grenade-pdns4/dc1e6e0/ : SUCCESS in 55m 12s\n- openstack-tox-cover http://logs.openstack.org/25/643725/6/check/openstack-tox-cover/939f406/cover/ : SUCCESS in 7m 00s\n- openstack-tox-lower-constraints http://logs.openstack.org/25/643725/6/check/openstack-tox-lower-constraints/d856865/ : SUCCESS in 5m 07s\n- openstack-tox-pep8 http://logs.openstack.org/25/643725/6/check/openstack-tox-pep8/07bf239/ : SUCCESS in 5m 43s\n- openstack-tox-py27 http://logs.openstack.org/25/643725/6/check/openstack-tox-py27/1e2630b/ : SUCCESS in 5m 27s\n- openstack-tox-py35 http://logs.openstack.org/25/643725/6/check/openstack-tox-py35/6c79131/ : SUCCESS in 6m 50s\n- openstack-tox-py36 http://logs.openstack.org/25/643725/6/check/openstack-tox-py36/956cd26/ : SUCCESS in 8m 09s\n- openstack-tox-py37 http://logs.openstack.org/25/643725/6/check/openstack-tox-py37/17b333a/ : SUCCESS in 6m 58s\n- openstack-tox-docs http://logs.openstack.org/25/643725/6/check/openstack-tox-docs/81c6588/html/ : SUCCESS in 5m 48s\n- neutron-tempest-plugin-designate-scenario http://logs.openstack.org/25/643725/6/check/neutron-tempest-plugin-designate-scenario/5a7da77/ : SUCCESS in 31m 58s","accounts_in_message":[],"_revision_number":6},{"id":"ceb4bba5270da053b75786da61b515bd8cb745b6","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2019-03-17 19:01:45.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"11a001e69c452cdc9ac254e75dc77cd93e8db04a","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2019-03-17 19:06:09.000000000","message":"Uploaded patch set 8.","accounts_in_message":[],"_revision_number":8},{"id":"67d237e2b84db3a128fec0c1693054fea4802b00","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2019-03-17 19:07:24.000000000","message":"Uploaded patch set 9.","accounts_in_message":[],"_revision_number":9},{"id":"f79575e34dc18d6c9c9edbd40f67497a270f94d6","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-03-17 20:04:32.000000000","message":"Patch Set 9: Verified+1\n\nBuild succeeded (check pipeline).\n\n- test-release-openstack http://logs.openstack.org/25/643725/9/check/test-release-openstack/2199343/ : SUCCESS in 3m 10s\n- designate-bind9 http://logs.openstack.org/25/643725/9/check/designate-bind9/050e899/ : SUCCESS in 32m 40s\n- designate-bind9-manager-model http://logs.openstack.org/25/643725/9/check/designate-bind9-manager-model/5dcfb65/ : SUCCESS in 35m 57s\n- designate-bind9-py36 http://logs.openstack.org/25/643725/9/check/designate-bind9-py36/ed3113d/ : SUCCESS in 35m 13s\n- designate-devstack-pdns4 http://logs.openstack.org/25/643725/9/check/designate-devstack-pdns4/96062b3/ : SUCCESS in 32m 05s\n- designate-devstack-pdns4-postgres http://logs.openstack.org/25/643725/9/check/designate-devstack-pdns4-postgres/92bd50a/ : SUCCESS in 31m 39s\n- designate-grenade-pdns4 http://logs.openstack.org/25/643725/9/check/designate-grenade-pdns4/2f75be7/ : SUCCESS in 52m 47s\n- openstack-tox-cover http://logs.openstack.org/25/643725/9/check/openstack-tox-cover/6cb018d/cover/ : SUCCESS in 5m 51s\n- openstack-tox-lower-constraints http://logs.openstack.org/25/643725/9/check/openstack-tox-lower-constraints/5640ca1/ : SUCCESS in 4m 25s\n- openstack-tox-pep8 http://logs.openstack.org/25/643725/9/check/openstack-tox-pep8/9a84bfc/ : SUCCESS in 4m 51s\n- openstack-tox-py27 http://logs.openstack.org/25/643725/9/check/openstack-tox-py27/51dc251/ : SUCCESS in 5m 34s\n- openstack-tox-py35 http://logs.openstack.org/25/643725/9/check/openstack-tox-py35/17be42e/ : SUCCESS in 5m 32s\n- openstack-tox-py36 http://logs.openstack.org/25/643725/9/check/openstack-tox-py36/3d70bf6/ : SUCCESS in 5m 39s\n- openstack-tox-py37 http://logs.openstack.org/25/643725/9/check/openstack-tox-py37/1cc6beb/ : SUCCESS in 5m 52s\n- openstack-tox-docs http://logs.openstack.org/25/643725/9/check/openstack-tox-docs/ba6d07a/html/ : SUCCESS in 4m 44s\n- neutron-tempest-plugin-designate-scenario http://logs.openstack.org/25/643725/9/check/neutron-tempest-plugin-designate-scenario/c97b9e9/ : SUCCESS in 30m 17s","accounts_in_message":[],"_revision_number":9},{"id":"55eda8146d50ce09d0b61f10f9bd317a5948c72e","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2019-03-17 20:37:34.000000000","message":"Uploaded patch set 10.","accounts_in_message":[],"_revision_number":10},{"id":"342bbacd682ea34b5c7fc717ad3dafe37820d2a8","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2019-03-17 20:50:04.000000000","message":"Uploaded patch set 11: Commit message was updated.","accounts_in_message":[],"_revision_number":11},{"id":"c807f4dfb9efbee5937b92076d520e3281d65bb7","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-03-17 21:44:04.000000000","message":"Patch Set 11: Verified+1\n\nBuild succeeded (check pipeline).\n\n- test-release-openstack http://logs.openstack.org/25/643725/11/check/test-release-openstack/07d229d/ : SUCCESS in 3m 27s\n- designate-bind9 http://logs.openstack.org/25/643725/11/check/designate-bind9/c4ae652/ : SUCCESS in 36m 33s\n- designate-bind9-manager-model http://logs.openstack.org/25/643725/11/check/designate-bind9-manager-model/911a9ad/ : SUCCESS in 34m 37s\n- designate-bind9-py36 http://logs.openstack.org/25/643725/11/check/designate-bind9-py36/3173fca/ : SUCCESS in 37m 58s\n- designate-devstack-pdns4 http://logs.openstack.org/25/643725/11/check/designate-devstack-pdns4/b809f0b/ : SUCCESS in 37m 59s\n- designate-devstack-pdns4-postgres http://logs.openstack.org/25/643725/11/check/designate-devstack-pdns4-postgres/a807f13/ : SUCCESS in 35m 51s\n- designate-grenade-pdns4 http://logs.openstack.org/25/643725/11/check/designate-grenade-pdns4/8b6ca8a/ : SUCCESS in 51m 25s\n- openstack-tox-cover http://logs.openstack.org/25/643725/11/check/openstack-tox-cover/8127d23/cover/ : SUCCESS in 6m 52s\n- openstack-tox-lower-constraints http://logs.openstack.org/25/643725/11/check/openstack-tox-lower-constraints/043c4fa/ : SUCCESS in 5m 28s\n- openstack-tox-pep8 http://logs.openstack.org/25/643725/11/check/openstack-tox-pep8/843e3b5/ : SUCCESS in 5m 27s\n- openstack-tox-py27 http://logs.openstack.org/25/643725/11/check/openstack-tox-py27/141adc8/ : SUCCESS in 7m 28s\n- openstack-tox-py35 http://logs.openstack.org/25/643725/11/check/openstack-tox-py35/6d37971/ : SUCCESS in 7m 35s\n- openstack-tox-py36 http://logs.openstack.org/25/643725/11/check/openstack-tox-py36/031c848/ : SUCCESS in 8m 34s\n- openstack-tox-py37 http://logs.openstack.org/25/643725/11/check/openstack-tox-py37/aa40b62/ : SUCCESS in 6m 51s\n- openstack-tox-docs http://logs.openstack.org/25/643725/11/check/openstack-tox-docs/2e5c999/html/ : SUCCESS in 4m 52s\n- neutron-tempest-plugin-designate-scenario http://logs.openstack.org/25/643725/11/check/neutron-tempest-plugin-designate-scenario/305a484/ : SUCCESS in 32m 36s","accounts_in_message":[],"_revision_number":11},{"id":"1eb77099e3b55f2d366236eed315d907b6169b60","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2019-04-01 16:59:07.000000000","message":"Patch Set 12: Patch Set 11 was rebased","accounts_in_message":[],"_revision_number":12},{"id":"c5c88f8ac2ff607ed99cc26577bf9140cf5673be","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-01 18:02:34.000000000","message":"Patch Set 12: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- test-release-openstack http://logs.openstack.org/25/643725/12/check/test-release-openstack/4659dad/ : SUCCESS in 5m 06s\n- designate-bind9 http://logs.openstack.org/25/643725/12/check/designate-bind9/c4d5902/ : SUCCESS in 36m 08s\n- designate-bind9-manager-model http://logs.openstack.org/25/643725/12/check/designate-bind9-manager-model/f4c9660/ : SUCCESS in 41m 42s\n- designate-bind9-py36 http://logs.openstack.org/25/643725/12/check/designate-bind9-py36/54fbdea/ : SUCCESS in 38m 10s\n- designate-devstack-pdns4 http://logs.openstack.org/25/643725/12/check/designate-devstack-pdns4/488a04c/ : SUCCESS in 33m 01s\n- designate-devstack-pdns4-postgres http://logs.openstack.org/25/643725/12/check/designate-devstack-pdns4-postgres/b7d0a80/ : SUCCESS in 37m 21s\n- designate-grenade-pdns4 http://logs.openstack.org/25/643725/12/check/designate-grenade-pdns4/5754bf6/ : FAILURE in 58m 19s\n- openstack-tox-cover http://logs.openstack.org/25/643725/12/check/openstack-tox-cover/8ca559d/cover/ : SUCCESS in 6m 51s\n- openstack-tox-lower-constraints http://logs.openstack.org/25/643725/12/check/openstack-tox-lower-constraints/8517d7c/ : SUCCESS in 5m 11s\n- openstack-tox-pep8 http://logs.openstack.org/25/643725/12/check/openstack-tox-pep8/eef5261/ : SUCCESS in 7m 46s\n- openstack-tox-py27 http://logs.openstack.org/25/643725/12/check/openstack-tox-py27/0f1305a/ : SUCCESS in 14m 52s\n- openstack-tox-py35 http://logs.openstack.org/25/643725/12/check/openstack-tox-py35/bff2454/ : SUCCESS in 7m 51s\n- openstack-tox-py36 http://logs.openstack.org/25/643725/12/check/openstack-tox-py36/0395f1b/ : SUCCESS in 8m 20s\n- openstack-tox-py37 http://logs.openstack.org/25/643725/12/check/openstack-tox-py37/b132d05/ : SUCCESS in 6m 08s\n- openstack-tox-docs http://logs.openstack.org/25/643725/12/check/openstack-tox-docs/efba620/html/ : SUCCESS in 6m 54s\n- neutron-tempest-plugin-designate-scenario http://logs.openstack.org/25/643725/12/check/neutron-tempest-plugin-designate-scenario/24fb7f0/ : SUCCESS in 28m 53s","accounts_in_message":[],"_revision_number":12},{"id":"44000514604ec928060f5f8749085653ea5cff6f","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2019-04-01 18:10:33.000000000","message":"Patch Set 12:\n\nrecheck","accounts_in_message":[],"_revision_number":12},{"id":"3cea6243599d193056fe4b58bfc3a867ef423a2c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-01 19:13:25.000000000","message":"Patch Set 12: Verified+1\n\nBuild succeeded (check pipeline).\n\n- test-release-openstack http://logs.openstack.org/25/643725/12/check/test-release-openstack/42987b1/ : SUCCESS in 3m 47s\n- designate-bind9 http://logs.openstack.org/25/643725/12/check/designate-bind9/4bd29fd/ : SUCCESS in 38m 14s\n- designate-bind9-manager-model http://logs.openstack.org/25/643725/12/check/designate-bind9-manager-model/148cc0c/ : SUCCESS in 42m 56s\n- designate-bind9-py36 http://logs.openstack.org/25/643725/12/check/designate-bind9-py36/aa48f8b/ : SUCCESS in 33m 29s\n- designate-devstack-pdns4 http://logs.openstack.org/25/643725/12/check/designate-devstack-pdns4/f108c4a/ : SUCCESS in 32m 00s\n- designate-devstack-pdns4-postgres http://logs.openstack.org/25/643725/12/check/designate-devstack-pdns4-postgres/6345328/ : SUCCESS in 37m 36s\n- designate-grenade-pdns4 http://logs.openstack.org/25/643725/12/check/designate-grenade-pdns4/7ae6866/ : SUCCESS in 59m 18s\n- openstack-tox-cover http://logs.openstack.org/25/643725/12/check/openstack-tox-cover/7670208/cover/ : SUCCESS in 6m 51s\n- openstack-tox-lower-constraints http://logs.openstack.org/25/643725/12/check/openstack-tox-lower-constraints/f3e3b6b/ : SUCCESS in 5m 29s\n- openstack-tox-pep8 http://logs.openstack.org/25/643725/12/check/openstack-tox-pep8/87ba68c/ : SUCCESS in 7m 30s\n- openstack-tox-py27 http://logs.openstack.org/25/643725/12/check/openstack-tox-py27/59c3fd8/ : SUCCESS in 6m 11s\n- openstack-tox-py35 http://logs.openstack.org/25/643725/12/check/openstack-tox-py35/6e02788/ : SUCCESS in 6m 38s\n- openstack-tox-py36 http://logs.openstack.org/25/643725/12/check/openstack-tox-py36/83ddafd/ : SUCCESS in 6m 15s\n- openstack-tox-py37 http://logs.openstack.org/25/643725/12/check/openstack-tox-py37/2acd03e/ : SUCCESS in 6m 48s\n- openstack-tox-docs http://logs.openstack.org/25/643725/12/check/openstack-tox-docs/4623f5f/html/ : SUCCESS in 5m 24s\n- neutron-tempest-plugin-designate-scenario http://logs.openstack.org/25/643725/12/check/neutron-tempest-plugin-designate-scenario/c290db3/ : SUCCESS in 36m 11s","accounts_in_message":[],"_revision_number":12},{"id":"4bd2c7344ace7c2ed1c00f4edbf2f671ad298136","author":{"_account_id":16006,"name":"goldyfruit","email":"gaetan.trellu@incloudus.com","username":"goldyfruit"},"date":"2019-04-01 19:21:24.000000000","message":"Patch Set 12: Code-Review+1","accounts_in_message":[],"_revision_number":12},{"id":"68efe9153d82e105a5db1e953fbc03d36111eb07","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2019-04-06 19:11:57.000000000","message":"Patch Set 12:\n\nThis is a diff between the config generated before and after this patch.\nhttp://paste.openstack.org/show/748965/","accounts_in_message":[],"_revision_number":12},{"id":"a24a42425ab2c4bdc17b288909da2a627180b719","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2019-05-20 00:10:16.000000000","message":"Patch Set 13: Patch Set 12 was rebased","accounts_in_message":[],"_revision_number":13},{"id":"393111de9a046ffd40a2f2ea6efd31a9775336bb","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-20 01:03:03.000000000","message":"Patch Set 13: Verified+1\n\nBuild succeeded (check pipeline).\n\n- test-release-openstack http://logs.openstack.org/25/643725/13/check/test-release-openstack/38de8b6/ : SUCCESS in 3m 01s\n- designate-bind9 http://logs.openstack.org/25/643725/13/check/designate-bind9/52f2424/ : SUCCESS in 46m 31s\n- designate-pdns4 http://logs.openstack.org/25/643725/13/check/designate-pdns4/f6e5443/ : SUCCESS in 47m 49s\n- designate-bind9-manager-model http://logs.openstack.org/25/643725/13/check/designate-bind9-manager-model/31193fb/ : SUCCESS in 34m 05s\n- designate-bind9-py27 http://logs.openstack.org/25/643725/13/check/designate-bind9-py27/55b0a46/ : SUCCESS in 30m 16s\n- designate-pdns4-py27 http://logs.openstack.org/25/643725/13/check/designate-pdns4-py27/b8d9794/ : SUCCESS in 30m 47s\n- designate-pdns4-postgres http://logs.openstack.org/25/643725/13/check/designate-pdns4-postgres/ecd6fbe/ : SUCCESS in 44m 12s\n- designate-grenade-pdns4 http://logs.openstack.org/25/643725/13/check/designate-grenade-pdns4/aae68b4/ : SUCCESS in 50m 42s\n- openstack-tox-cover http://logs.openstack.org/25/643725/13/check/openstack-tox-cover/008b157/cover/ : SUCCESS in 8m 15s\n- openstack-tox-lower-constraints http://logs.openstack.org/25/643725/13/check/openstack-tox-lower-constraints/a0b01eb/ : SUCCESS in 5m 56s\n- openstack-tox-pep8 http://logs.openstack.org/25/643725/13/check/openstack-tox-pep8/631a58a/ : SUCCESS in 8m 36s\n- openstack-tox-py27 http://logs.openstack.org/25/643725/13/check/openstack-tox-py27/d1ede9a/ : SUCCESS in 8m 31s\n- openstack-tox-py36 http://logs.openstack.org/25/643725/13/check/openstack-tox-py36/2730183/ : SUCCESS in 5m 47s\n- openstack-tox-py37 http://logs.openstack.org/25/643725/13/check/openstack-tox-py37/afa8cfe/ : SUCCESS in 7m 02s\n- openstack-tox-docs http://logs.openstack.org/25/643725/13/check/openstack-tox-docs/5da2dda/html/ : SUCCESS in 7m 57s\n- neutron-tempest-plugin-designate-scenario http://logs.openstack.org/25/643725/13/check/neutron-tempest-plugin-designate-scenario/2a5c282/ : SUCCESS in 40m 13s","accounts_in_message":[],"_revision_number":13},{"id":"6257a285eba443f4ac631c5fce62c8642c7ea1a7","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2019-05-20 20:06:52.000000000","message":"Patch Set 14: Patch Set 13 was rebased","accounts_in_message":[],"_revision_number":14},{"id":"109809e43edba544685b646cf26f03230a89b697","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-20 21:13:43.000000000","message":"Patch Set 14: Verified+1\n\nBuild succeeded (check pipeline).\n\n- test-release-openstack http://logs.openstack.org/25/643725/14/check/test-release-openstack/9290a5d/ : SUCCESS in 3m 57s\n- designate-bind9 http://logs.openstack.org/25/643725/14/check/designate-bind9/270128b/ : SUCCESS in 37m 46s\n- designate-pdns4 http://logs.openstack.org/25/643725/14/check/designate-pdns4/b72ec4a/ : SUCCESS in 36m 02s\n- designate-bind9-manager-model http://logs.openstack.org/25/643725/14/check/designate-bind9-manager-model/3569a1a/ : SUCCESS in 39m 55s\n- designate-bind9-py27 http://logs.openstack.org/25/643725/14/check/designate-bind9-py27/ecf909c/ : SUCCESS in 51m 31s\n- designate-pdns4-py27 http://logs.openstack.org/25/643725/14/check/designate-pdns4-py27/38cadc7/ : SUCCESS in 35m 43s\n- designate-pdns4-postgres http://logs.openstack.org/25/643725/14/check/designate-pdns4-postgres/0d1312e/ : SUCCESS in 34m 21s\n- designate-grenade-pdns4 http://logs.openstack.org/25/643725/14/check/designate-grenade-pdns4/f1c0672/ : SUCCESS in 55m 50s\n- openstack-tox-cover http://logs.openstack.org/25/643725/14/check/openstack-tox-cover/7e601ea/cover/ : SUCCESS in 6m 08s\n- openstack-tox-lower-constraints http://logs.openstack.org/25/643725/14/check/openstack-tox-lower-constraints/1aa9d6c/ : SUCCESS in 5m 57s\n- openstack-tox-pep8 http://logs.openstack.org/25/643725/14/check/openstack-tox-pep8/e1851a1/ : SUCCESS in 5m 24s\n- openstack-tox-py27 http://logs.openstack.org/25/643725/14/check/openstack-tox-py27/ee59ecb/ : SUCCESS in 5m 15s\n- openstack-tox-py36 http://logs.openstack.org/25/643725/14/check/openstack-tox-py36/66cb892/ : SUCCESS in 6m 38s\n- openstack-tox-py37 http://logs.openstack.org/25/643725/14/check/openstack-tox-py37/7041743/ : SUCCESS in 6m 21s\n- openstack-tox-docs http://logs.openstack.org/25/643725/14/check/openstack-tox-docs/78c11e8/html/ : SUCCESS in 4m 38s\n- neutron-tempest-plugin-designate-scenario http://logs.openstack.org/25/643725/14/check/neutron-tempest-plugin-designate-scenario/095ef09/ : SUCCESS in 34m 34s","accounts_in_message":[],"_revision_number":14},{"id":"a39c1288a9c1791d9ccd70a1c82c61bf260c0129","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2019-05-23 15:46:11.000000000","message":"Patch Set 15: Patch Set 14 was rebased","accounts_in_message":[],"_revision_number":15},{"id":"43ba17045367511510986904635e62ddda021dbb","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-23 16:41:24.000000000","message":"Patch Set 15: Verified+1\n\nBuild succeeded (check pipeline).\n\n- test-release-openstack http://logs.openstack.org/25/643725/15/check/test-release-openstack/cbb0140/ : SUCCESS in 3m 18s\n- designate-bind9 http://logs.openstack.org/25/643725/15/check/designate-bind9/7ad7a8e/ : SUCCESS in 36m 57s\n- designate-pdns4 http://logs.openstack.org/25/643725/15/check/designate-pdns4/8702842/ : SUCCESS in 34m 38s\n- designate-bind9-manager-model http://logs.openstack.org/25/643725/15/check/designate-bind9-manager-model/89095f6/ : SUCCESS in 47m 24s\n- designate-bind9-py27 http://logs.openstack.org/25/643725/15/check/designate-bind9-py27/92d84b3/ : SUCCESS in 38m 40s\n- designate-pdns4-py27 http://logs.openstack.org/25/643725/15/check/designate-pdns4-py27/229d395/ : SUCCESS in 35m 06s\n- designate-pdns4-postgres http://logs.openstack.org/25/643725/15/check/designate-pdns4-postgres/b84121e/ : SUCCESS in 36m 30s\n- designate-grenade-pdns4 http://logs.openstack.org/25/643725/15/check/designate-grenade-pdns4/0cbc9f8/ : SUCCESS in 45m 30s\n- openstack-tox-cover http://logs.openstack.org/25/643725/15/check/openstack-tox-cover/ca011d4/cover/ : SUCCESS in 5m 41s\n- openstack-tox-lower-constraints http://logs.openstack.org/25/643725/15/check/openstack-tox-lower-constraints/65f040a/ : SUCCESS in 5m 08s\n- openstack-tox-pep8 http://logs.openstack.org/25/643725/15/check/openstack-tox-pep8/4769bb2/ : SUCCESS in 6m 31s\n- openstack-tox-py27 http://logs.openstack.org/25/643725/15/check/openstack-tox-py27/79bb842/ : SUCCESS in 6m 43s\n- openstack-tox-py36 http://logs.openstack.org/25/643725/15/check/openstack-tox-py36/ed70d5b/ : SUCCESS in 6m 54s\n- openstack-tox-py37 http://logs.openstack.org/25/643725/15/check/openstack-tox-py37/4f00157/ : SUCCESS in 6m 51s\n- openstack-tox-docs http://logs.openstack.org/25/643725/15/check/openstack-tox-docs/4999d4b/html/ : SUCCESS in 4m 51s\n- neutron-tempest-plugin-designate-scenario http://logs.openstack.org/25/643725/15/check/neutron-tempest-plugin-designate-scenario/f977586/ : SUCCESS in 35m 27s","accounts_in_message":[],"_revision_number":15},{"id":"56d509413b817cd22874123a11509be7f2ee157c","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2019-05-28 23:54:00.000000000","message":"Patch Set 16: Patch Set 15 was rebased","accounts_in_message":[],"_revision_number":16},{"id":"58029b086de72fe6e0686695ad136c2f972f9621","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-29 01:56:47.000000000","message":"Patch Set 16: Verified+1\n\nBuild succeeded (check pipeline).\n\n- test-release-openstack http://logs.openstack.org/25/643725/16/check/test-release-openstack/c0601a6/ : SUCCESS in 2m 59s\n- designate-bind9 http://logs.openstack.org/25/643725/16/check/designate-bind9/54b6882/ : SUCCESS in 38m 09s\n- designate-pdns4 http://logs.openstack.org/25/643725/16/check/designate-pdns4/d09d1a8/ : SUCCESS in 36m 46s\n- designate-bind9-manager-model http://logs.openstack.org/25/643725/16/check/designate-bind9-manager-model/92381cd/ : SUCCESS in 44m 18s\n- designate-bind9-py27 http://logs.openstack.org/25/643725/16/check/designate-bind9-py27/775bf46/ : SUCCESS in 42m 53s\n- designate-pdns4-py27 http://logs.openstack.org/25/643725/16/check/designate-pdns4-py27/a757563/ : SUCCESS in 38m 53s\n- designate-pdns4-postgres http://logs.openstack.org/25/643725/16/check/designate-pdns4-postgres/e82823f/ : SUCCESS in 33m 58s\n- designate-grenade-pdns4 http://logs.openstack.org/25/643725/16/check/designate-grenade-pdns4/366e6cf/ : SUCCESS in 1h 04m 52s\n- openstack-tox-cover http://logs.openstack.org/25/643725/16/check/openstack-tox-cover/7cdf4df/cover/ : SUCCESS in 5m 42s\n- openstack-tox-lower-constraints http://logs.openstack.org/25/643725/16/check/openstack-tox-lower-constraints/79e4aba/ : SUCCESS in 4m 35s\n- openstack-tox-pep8 http://logs.openstack.org/25/643725/16/check/openstack-tox-pep8/db0c467/ : SUCCESS in 5m 31s\n- openstack-tox-py27 http://logs.openstack.org/25/643725/16/check/openstack-tox-py27/eab7911/ : SUCCESS in 6m 29s\n- openstack-tox-py36 http://logs.openstack.org/25/643725/16/check/openstack-tox-py36/ecb0c50/ : SUCCESS in 5m 50s\n- openstack-tox-py37 http://logs.openstack.org/25/643725/16/check/openstack-tox-py37/32ef749/ : SUCCESS in 6m 06s\n- openstack-tox-docs http://logs.openstack.org/25/643725/16/check/openstack-tox-docs/47a146a/html/ : SUCCESS in 4m 52s\n- neutron-tempest-plugin-designate-scenario http://logs.openstack.org/25/643725/16/check/neutron-tempest-plugin-designate-scenario/ea984c7/ : SUCCESS in 34m 53s","accounts_in_message":[],"_revision_number":16},{"id":"a78ed7cfcb5cc0baa8fc0e5d208e5f55ec660402","author":{"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},"date":"2019-06-04 17:07:45.000000000","message":"Patch Set 16: Code-Review+2","accounts_in_message":[],"_revision_number":16},{"id":"7ad72579fabc3b2755fa5254128f09725831b3e1","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2019-06-09 06:44:07.000000000","message":"Patch Set 16: Workflow+1","accounts_in_message":[],"_revision_number":16},{"id":"127ce2e85da5300b58ad7d9532649bb781461058","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-09 06:44:20.000000000","message":"Patch Set 16: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":16},{"id":"3daadc6191e9be6f2650e150b81515466fbe2060","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-09 07:50:15.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":16},{"id":"424c8f981ee5142fcfc9a1577147e4c724a3139a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-09 07:50:15.000000000","message":"Patch Set 16: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- test-release-openstack http://logs.openstack.org/25/643725/16/gate/test-release-openstack/df909fa/ : SUCCESS in 3m 41s\n- designate-bind9 http://logs.openstack.org/25/643725/16/gate/designate-bind9/94e97f8/ : SUCCESS in 38m 54s\n- designate-pdns4 http://logs.openstack.org/25/643725/16/gate/designate-pdns4/6e91575/ : SUCCESS in 38m 33s\n- designate-bind9-manager-model http://logs.openstack.org/25/643725/16/gate/designate-bind9-manager-model/c5b3154/ : SUCCESS in 37m 31s\n- designate-bind9-py27 http://logs.openstack.org/25/643725/16/gate/designate-bind9-py27/6629e4e/ : SUCCESS in 41m 22s\n- designate-pdns4-py27 http://logs.openstack.org/25/643725/16/gate/designate-pdns4-py27/c3b7d66/ : SUCCESS in 37m 09s\n- designate-pdns4-postgres http://logs.openstack.org/25/643725/16/gate/designate-pdns4-postgres/eb04e1e/ : SUCCESS in 35m 11s\n- designate-grenade-pdns4 http://logs.openstack.org/25/643725/16/gate/designate-grenade-pdns4/40bab4b/ : SUCCESS in 51m 44s\n- openstack-tox-lower-constraints http://logs.openstack.org/25/643725/16/gate/openstack-tox-lower-constraints/7885d72/ : SUCCESS in 5m 10s\n- openstack-tox-pep8 http://logs.openstack.org/25/643725/16/gate/openstack-tox-pep8/8ac8e5c/ : SUCCESS in 5m 17s\n- openstack-tox-py27 http://logs.openstack.org/25/643725/16/gate/openstack-tox-py27/4a8fa14/ : SUCCESS in 5m 50s\n- openstack-tox-py36 http://logs.openstack.org/25/643725/16/gate/openstack-tox-py36/ef07b1e/ : SUCCESS in 5m 52s\n- openstack-tox-py37 http://logs.openstack.org/25/643725/16/gate/openstack-tox-py37/274fdce/ : SUCCESS in 6m 12s\n- openstack-tox-docs http://logs.openstack.org/25/643725/16/gate/openstack-tox-docs/3c25f16/html/ : SUCCESS in 4m 41s","accounts_in_message":[],"_revision_number":16}],"current_revision_number":16,"current_revision":"a53c3daf16d366e3cc513c62f57042f43cbd4588","revisions":{"38a913ba45e4e081515b82a0a490b4f4a1bbde66":{"kind":"REWORK","_number":1,"created":"2019-03-17 01:05:31.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/25/643725/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/25/643725/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/25/643725/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/25/643725/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/25/643725/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/25/643725/1"}}},"commit":{"parents":[{"commit":"59d8cab9bbe37529346728778e6dae79a3a57a2a","subject":"Allow non localhost connections to pdns api","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/59d8cab9bbe37529346728778e6dae79a3a57a2a"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2019-03-16 23:50:02.000000000","tz":-420},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2019-03-17 01:04:52.000000000","tz":-420},"subject":"[WIP] Centralized all config","message":"[WIP] Centralized all config\n\n- Move all configruation to a central location.\n- Fixed multiple bugs with missing or incorrect config.\n\nChange-Id: I3be68bf294157631a9bac128df0841220020de96\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/38a913ba45e4e081515b82a0a490b4f4a1bbde66"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/38a913ba45e4e081515b82a0a490b4f4a1bbde66"}]},"branch":"refs/heads/master"},"cab08863f532650fbb9ecda25c8330d09ca9ae04":{"kind":"REWORK","_number":2,"created":"2019-03-17 01:38:16.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/25/643725/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/25/643725/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/25/643725/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/25/643725/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/25/643725/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/25/643725/2"}}},"commit":{"parents":[{"commit":"59d8cab9bbe37529346728778e6dae79a3a57a2a","subject":"Allow non localhost connections to pdns api","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/59d8cab9bbe37529346728778e6dae79a3a57a2a"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2019-03-16 23:50:02.000000000","tz":-420},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2019-03-17 01:38:02.000000000","tz":-420},"subject":"[WIP] Centralized all config","message":"[WIP] Centralized all config\n\n- Move all configruation to a central location.\n- Fixed multiple bugs with missing or incorrect config.\n\nChange-Id: I3be68bf294157631a9bac128df0841220020de96\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/cab08863f532650fbb9ecda25c8330d09ca9ae04"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/cab08863f532650fbb9ecda25c8330d09ca9ae04"}]},"branch":"refs/heads/master"},"1abc80e838a4357cc44a13e1c5db6326d83dd65b":{"kind":"REWORK","_number":3,"created":"2019-03-17 04:00:45.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/25/643725/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/25/643725/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/25/643725/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/25/643725/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/25/643725/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/25/643725/3"}}},"commit":{"parents":[{"commit":"59d8cab9bbe37529346728778e6dae79a3a57a2a","subject":"Allow non localhost connections to pdns api","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/59d8cab9bbe37529346728778e6dae79a3a57a2a"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2019-03-16 23:50:02.000000000","tz":-420},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2019-03-17 03:57:23.000000000","tz":-420},"subject":"[WIP] Centralized all config","message":"[WIP] Centralized all config\n\n- Move all configruation to a central location.\n- Fixed multiple bugs with missing or incorrect config.\n\nChange-Id: I3be68bf294157631a9bac128df0841220020de96\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/1abc80e838a4357cc44a13e1c5db6326d83dd65b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/1abc80e838a4357cc44a13e1c5db6326d83dd65b"}]},"branch":"refs/heads/master"},"e214ee2273ce21389f54d6786c5d26910cf5dccc":{"kind":"REWORK","_number":4,"created":"2019-03-17 04:10:41.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/25/643725/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/25/643725/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/25/643725/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/25/643725/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/25/643725/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/25/643725/4"}}},"commit":{"parents":[{"commit":"59d8cab9bbe37529346728778e6dae79a3a57a2a","subject":"Allow non localhost connections to pdns api","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/59d8cab9bbe37529346728778e6dae79a3a57a2a"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2019-03-16 23:50:02.000000000","tz":-420},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2019-03-17 04:10:30.000000000","tz":-420},"subject":"[WIP] Centralized all config","message":"[WIP] Centralized all config\n\n- Move all configruation to a central location.\n- Fixed multiple bugs with missing or incorrect config.\n\nChange-Id: I3be68bf294157631a9bac128df0841220020de96\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/e214ee2273ce21389f54d6786c5d26910cf5dccc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/e214ee2273ce21389f54d6786c5d26910cf5dccc"}]},"branch":"refs/heads/master"},"f8edfb5d94cce18ff1d75366cc1e4614ad058dd2":{"kind":"REWORK","_number":5,"created":"2019-03-17 04:11:59.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/25/643725/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/25/643725/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/25/643725/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/25/643725/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/25/643725/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/25/643725/5"}}},"commit":{"parents":[{"commit":"59d8cab9bbe37529346728778e6dae79a3a57a2a","subject":"Allow non localhost connections to pdns api","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/59d8cab9bbe37529346728778e6dae79a3a57a2a"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2019-03-16 23:50:02.000000000","tz":-420},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2019-03-17 04:11:51.000000000","tz":-420},"subject":"[WIP] Centralized all config","message":"[WIP] Centralized all config\n\n- Move all configruation to a central location.\n- Fixed multiple bugs with missing or incorrect config.\n\nChange-Id: I3be68bf294157631a9bac128df0841220020de96\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/f8edfb5d94cce18ff1d75366cc1e4614ad058dd2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/f8edfb5d94cce18ff1d75366cc1e4614ad058dd2"}]},"branch":"refs/heads/master"},"8e5d4c54ba8062cf3718fcdae4e948d95293e992":{"kind":"REWORK","_number":6,"created":"2019-03-17 06:09:23.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/25/643725/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/25/643725/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/25/643725/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/25/643725/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/25/643725/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/25/643725/6"}}},"commit":{"parents":[{"commit":"59d8cab9bbe37529346728778e6dae79a3a57a2a","subject":"Allow non localhost connections to pdns api","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/59d8cab9bbe37529346728778e6dae79a3a57a2a"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2019-03-16 23:50:02.000000000","tz":-420},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2019-03-17 06:08:36.000000000","tz":-420},"subject":"Moved most configuration into designate.conf","message":"Moved most configuration into designate.conf\n\nBased upon the established conf paradigm used in Nova,\nthis patch moves most of the configuration to a\ncentral location. Making config a lot easier to\nmanage, as well a lot easier to find.\n\nThere is still some configuration listed in classes,\nthat I would like to eventually move to\ndesignate.conf as well, but to lower complexity not\nall config was moved.\n\n* Centrelized most configuration under designate.conf.\n* Cleaned up configuration.\n* Simplified config generation.\n* Renamed generated config to designate.conf.sample.\n\n* xfr_timeout now in the appropriate section.\n* backend:infoblox section now configurable again.\n* allowed_remote_exmods listed in sample config.\n\nChange-Id: I3be68bf294157631a9bac128df0841220020de96\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/8e5d4c54ba8062cf3718fcdae4e948d95293e992"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/8e5d4c54ba8062cf3718fcdae4e948d95293e992"}]},"branch":"refs/heads/master"},"fa8c7fda9bf7efb3cfc4cd8de716f62eccbe51d6":{"kind":"REWORK","_number":7,"created":"2019-03-17 19:01:45.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/25/643725/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/25/643725/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/25/643725/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/25/643725/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/25/643725/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/25/643725/7"}}},"commit":{"parents":[{"commit":"59d8cab9bbe37529346728778e6dae79a3a57a2a","subject":"Allow non localhost connections to pdns api","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/59d8cab9bbe37529346728778e6dae79a3a57a2a"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2019-03-16 23:50:02.000000000","tz":-420},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2019-03-17 19:00:55.000000000","tz":-420},"subject":"Moved most configuration into designate.conf","message":"Moved most configuration into designate.conf\n\nBased upon the established conf paradigm used in Nova,\nthis patch moves most of the configuration to a\ncentral location. Making config a lot easier to\nmanage, as well a lot easier to locate.\n\nThere is still some configuration listed in classes,\nthat I would like to eventually move to\ndesignate.conf as well, but to lower complexity not\nall config was moved.\n\n* Centrelized most configuration under designate.conf.\n* Cleaned up configuration.\n* Simplified config generation.\n* Renamed generated config to designate.conf.sample.\n\n* xfr_timeout now in the appropriate section.\n* backend:infoblox section now listed in sample config.\n* allowed_remote_exmods listed in sample config.\n\nChange-Id: I3be68bf294157631a9bac128df0841220020de96\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/fa8c7fda9bf7efb3cfc4cd8de716f62eccbe51d6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/fa8c7fda9bf7efb3cfc4cd8de716f62eccbe51d6"}]},"branch":"refs/heads/master"},"88afa44f57955ac53703300d119a593a275f1069":{"kind":"REWORK","_number":8,"created":"2019-03-17 19:06:09.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/25/643725/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/25/643725/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/25/643725/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/25/643725/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/25/643725/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/25/643725/8"}}},"commit":{"parents":[{"commit":"59d8cab9bbe37529346728778e6dae79a3a57a2a","subject":"Allow non localhost connections to pdns api","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/59d8cab9bbe37529346728778e6dae79a3a57a2a"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2019-03-16 23:50:02.000000000","tz":-420},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2019-03-17 19:05:58.000000000","tz":-420},"subject":"Moved most configuration into designate.conf","message":"Moved most configuration into designate.conf\n\nBased upon the established conf paradigm used in Nova,\nthis patch moves most of the configuration to a\ncentral location. Making config a lot easier to\nmanage, as well a lot easier to locate.\n\nThere is still some configuration listed in classes,\nthat I would like to eventually move to\ndesignate.conf as well, but to lower complexity not\nall config was moved.\n\n* Centrelized most configuration under designate.conf.\n* Cleaned up configuration.\n* Simplified config generation.\n* Renamed generated config to designate.conf.sample.\n\n* xfr_timeout now in the appropriate section.\n* backend:infoblox section now listed in sample config.\n* allowed_remote_exmods listed in sample config.\n\nChange-Id: I3be68bf294157631a9bac128df0841220020de96\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/88afa44f57955ac53703300d119a593a275f1069"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/88afa44f57955ac53703300d119a593a275f1069"}]},"branch":"refs/heads/master"},"7590d2b29467a846efb35fba093334bba3bb0f35":{"kind":"REWORK","_number":9,"created":"2019-03-17 19:07:24.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/25/643725/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/25/643725/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/25/643725/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/25/643725/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/25/643725/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/25/643725/9"}}},"commit":{"parents":[{"commit":"59d8cab9bbe37529346728778e6dae79a3a57a2a","subject":"Allow non localhost connections to pdns api","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/59d8cab9bbe37529346728778e6dae79a3a57a2a"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2019-03-16 23:50:02.000000000","tz":-420},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2019-03-17 19:07:19.000000000","tz":-420},"subject":"Moved most configuration into designate.conf","message":"Moved most configuration into designate.conf\n\nBased upon the established conf paradigm used in Nova,\nthis patch moves most of the configuration to a\ncentral location. Making config a lot easier to\nmanage, as well a lot easier to locate.\n\nThere is still some configuration listed in classes,\nthat I would like to eventually move to\ndesignate.conf as well, but to lower complexity not\nall config was moved.\n\n* Centrelized most configuration under designate.conf.\n* Cleaned up configuration.\n* Simplified config generation.\n* Renamed generated config to designate.conf.sample.\n\n* xfr_timeout now in the appropriate section.\n* backend:infoblox section now listed in sample config.\n* allowed_remote_exmods listed in sample config.\n\nChange-Id: I3be68bf294157631a9bac128df0841220020de96\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/7590d2b29467a846efb35fba093334bba3bb0f35"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/7590d2b29467a846efb35fba093334bba3bb0f35"}]},"branch":"refs/heads/master"},"9ddfbad9c2792a4d13df300a597744c2e0c74eb1":{"kind":"REWORK","_number":10,"created":"2019-03-17 20:37:34.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/25/643725/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/25/643725/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/25/643725/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/25/643725/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/25/643725/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/25/643725/10"}}},"commit":{"parents":[{"commit":"59d8cab9bbe37529346728778e6dae79a3a57a2a","subject":"Allow non localhost connections to pdns api","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/59d8cab9bbe37529346728778e6dae79a3a57a2a"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2019-03-16 23:50:02.000000000","tz":-420},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2019-03-17 20:36:14.000000000","tz":-420},"subject":"Moved most configuration into designate.conf","message":"Moved most configuration into designate.conf\n\nBased upon the established conf paradigm used in Nova,\nthis patch moves most of the configuration to a\ncentral location. Making config a lot easier to\nmanage, as well a lot easier to locate.\n\nThere is still some configuration listed in classes,\nthat I would like to eventually move to\ndesignate.conf as well, but to lower complexity not\nall config was moved.\n\nGeneral changes\n* Centrelized most configuration under designate.conf.\n* Cleaned up configuration.\n* Simplified config generation.\n* Renamed generated config to designate.conf.sample.\n\nPreviously broken\n* xfr_timeout now in the appropriate section.\n* memcached_servers now in the appropriate section.\n\nPreviously missing from sample config\n* allowed_remote_exmods listed in sample config.\n* backend:akamai section listed in sample config.\n* backend:dynect section listed in sample config.\n* backend:infoblox section listed in sample config.\n\nChange-Id: I3be68bf294157631a9bac128df0841220020de96\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/9ddfbad9c2792a4d13df300a597744c2e0c74eb1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/9ddfbad9c2792a4d13df300a597744c2e0c74eb1"}]},"branch":"refs/heads/master"},"89ff506daef3b29c1975e17ccc4611f38cd41b03":{"kind":"NO_CODE_CHANGE","_number":11,"created":"2019-03-17 20:50:04.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/25/643725/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/25/643725/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/25/643725/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/25/643725/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/25/643725/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/25/643725/11"}}},"commit":{"parents":[{"commit":"59d8cab9bbe37529346728778e6dae79a3a57a2a","subject":"Allow non localhost connections to pdns api","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/59d8cab9bbe37529346728778e6dae79a3a57a2a"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2019-03-16 23:50:02.000000000","tz":-420},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2019-03-17 20:47:42.000000000","tz":-420},"subject":"Moved most configuration into designate.conf","message":"Moved most configuration into designate.conf\n\nBased upon the established conf paradigm used in Nova,\nthis patch moves most of the configuration to a\ncentral location. Making config a lot easier to\nmanage, as well a lot easier to locate.\n\nThere is still some configuration listed in classes,\nthat I would like to eventually move to\ndesignate.conf as well, but to lower complexity not\nall config was moved.\n\nGeneral changes\n* Centrelized most configuration under designate.conf.\n* Cleaned up configuration.\n* Simplified config generation.\n* Renamed generated config to designate.conf.sample.\n\nPreviously broken\n* xfr_timeout now in the appropriate section.\n* memcached_servers now in the appropriate section.\n\nPreviously missing from sample config\n* allowed_remote_exmods listed in sample config.\n* backend:akamai section listed in sample config.\n* backend:dynect section listed in sample config.\n* backend:infoblox section listed in sample config.\n\nblueprint centralize-config-designate\n\nChange-Id: I3be68bf294157631a9bac128df0841220020de96\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/89ff506daef3b29c1975e17ccc4611f38cd41b03"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/89ff506daef3b29c1975e17ccc4611f38cd41b03"}]},"branch":"refs/heads/master"},"28c772c28271db5f5c89f1a1471e27fb4bef42be":{"kind":"TRIVIAL_REBASE","_number":12,"created":"2019-04-01 16:59:07.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/25/643725/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/25/643725/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/25/643725/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/25/643725/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/25/643725/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/25/643725/12"}}},"commit":{"parents":[{"commit":"33ca0b87891228d86e440e929feeb324353b39d3","subject":"Merge \"Allow pdns4 backend to select tsigkey\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/33ca0b87891228d86e440e929feeb324353b39d3"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2019-03-16 23:50:02.000000000","tz":-420},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2019-04-01 16:59:07.000000000","tz":0},"subject":"Moved most configuration into designate.conf","message":"Moved most configuration into designate.conf\n\nBased upon the established conf paradigm used in Nova,\nthis patch moves most of the configuration to a\ncentral location. Making config a lot easier to\nmanage, as well a lot easier to locate.\n\nThere is still some configuration listed in classes,\nthat I would like to eventually move to\ndesignate.conf as well, but to lower complexity not\nall config was moved.\n\nGeneral changes\n* Centrelized most configuration under designate.conf.\n* Cleaned up configuration.\n* Simplified config generation.\n* Renamed generated config to designate.conf.sample.\n\nPreviously broken\n* xfr_timeout now in the appropriate section.\n* memcached_servers now in the appropriate section.\n\nPreviously missing from sample config\n* allowed_remote_exmods listed in sample config.\n* backend:akamai section listed in sample config.\n* backend:dynect section listed in sample config.\n* backend:infoblox section listed in sample config.\n\nblueprint centralize-config-designate\n\nChange-Id: I3be68bf294157631a9bac128df0841220020de96\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/28c772c28271db5f5c89f1a1471e27fb4bef42be"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/28c772c28271db5f5c89f1a1471e27fb4bef42be"}]},"branch":"refs/heads/master"},"95272c94917237f94d0fe44b4b70c35bab40af9c":{"kind":"TRIVIAL_REBASE","_number":13,"created":"2019-05-20 00:10:16.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/25/643725/13","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/25/643725/13","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/25/643725/13 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/25/643725/13 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/25/643725/13 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/25/643725/13"}}},"commit":{"parents":[{"commit":"178328f0a8d2485214814275f7b7fdaefe799aca","subject":"Merge \"Fixing stevedore warning when running devstack\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/178328f0a8d2485214814275f7b7fdaefe799aca"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2019-03-16 23:50:02.000000000","tz":-420},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2019-05-20 00:10:16.000000000","tz":0},"subject":"Moved most configuration into designate.conf","message":"Moved most configuration into designate.conf\n\nBased upon the established conf paradigm used in Nova,\nthis patch moves most of the configuration to a\ncentral location. Making config a lot easier to\nmanage, as well a lot easier to locate.\n\nThere is still some configuration listed in classes,\nthat I would like to eventually move to\ndesignate.conf as well, but to lower complexity not\nall config was moved.\n\nGeneral changes\n* Centrelized most configuration under designate.conf.\n* Cleaned up configuration.\n* Simplified config generation.\n* Renamed generated config to designate.conf.sample.\n\nPreviously broken\n* xfr_timeout now in the appropriate section.\n* memcached_servers now in the appropriate section.\n\nPreviously missing from sample config\n* allowed_remote_exmods listed in sample config.\n* backend:akamai section listed in sample config.\n* backend:dynect section listed in sample config.\n* backend:infoblox section listed in sample config.\n\nblueprint centralize-config-designate\n\nChange-Id: I3be68bf294157631a9bac128df0841220020de96\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/95272c94917237f94d0fe44b4b70c35bab40af9c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/95272c94917237f94d0fe44b4b70c35bab40af9c"}]},"branch":"refs/heads/master"},"9cd55579800d7b5bf0e11a7fa3391cf23daf0f6e":{"kind":"TRIVIAL_REBASE","_number":14,"created":"2019-05-20 20:06:52.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/25/643725/14","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/25/643725/14","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/25/643725/14 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/25/643725/14 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/25/643725/14 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/25/643725/14"}}},"commit":{"parents":[{"commit":"a705d055bd22fb9fe796e15001ed6db8175ff8db","subject":"Merge \"Update Python 3 test runtimes for Train\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/a705d055bd22fb9fe796e15001ed6db8175ff8db"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2019-03-16 23:50:02.000000000","tz":-420},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2019-05-20 20:06:52.000000000","tz":0},"subject":"Moved most configuration into designate.conf","message":"Moved most configuration into designate.conf\n\nBased upon the established conf paradigm used in Nova,\nthis patch moves most of the configuration to a\ncentral location. Making config a lot easier to\nmanage, as well a lot easier to locate.\n\nThere is still some configuration listed in classes,\nthat I would like to eventually move to\ndesignate.conf as well, but to lower complexity not\nall config was moved.\n\nGeneral changes\n* Centrelized most configuration under designate.conf.\n* Cleaned up configuration.\n* Simplified config generation.\n* Renamed generated config to designate.conf.sample.\n\nPreviously broken\n* xfr_timeout now in the appropriate section.\n* memcached_servers now in the appropriate section.\n\nPreviously missing from sample config\n* allowed_remote_exmods listed in sample config.\n* backend:akamai section listed in sample config.\n* backend:dynect section listed in sample config.\n* backend:infoblox section listed in sample config.\n\nblueprint centralize-config-designate\n\nChange-Id: I3be68bf294157631a9bac128df0841220020de96\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/9cd55579800d7b5bf0e11a7fa3391cf23daf0f6e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/9cd55579800d7b5bf0e11a7fa3391cf23daf0f6e"}]},"branch":"refs/heads/master"},"762b10dd452a87d64aaf13f812af90ff1699fd69":{"kind":"TRIVIAL_REBASE","_number":15,"created":"2019-05-23 15:46:11.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/25/643725/15","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/25/643725/15","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/25/643725/15 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/25/643725/15 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/25/643725/15 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/25/643725/15"}}},"commit":{"parents":[{"commit":"12fdd24e5bd997c7a444a1a343435fe28219de77","subject":"Merge \"Modernized object tests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/12fdd24e5bd997c7a444a1a343435fe28219de77"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2019-03-16 23:50:02.000000000","tz":-420},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2019-05-23 15:46:11.000000000","tz":0},"subject":"Moved most configuration into designate.conf","message":"Moved most configuration into designate.conf\n\nBased upon the established conf paradigm used in Nova,\nthis patch moves most of the configuration to a\ncentral location. Making config a lot easier to\nmanage, as well a lot easier to locate.\n\nThere is still some configuration listed in classes,\nthat I would like to eventually move to\ndesignate.conf as well, but to lower complexity not\nall config was moved.\n\nGeneral changes\n* Centrelized most configuration under designate.conf.\n* Cleaned up configuration.\n* Simplified config generation.\n* Renamed generated config to designate.conf.sample.\n\nPreviously broken\n* xfr_timeout now in the appropriate section.\n* memcached_servers now in the appropriate section.\n\nPreviously missing from sample config\n* allowed_remote_exmods listed in sample config.\n* backend:akamai section listed in sample config.\n* backend:dynect section listed in sample config.\n* backend:infoblox section listed in sample config.\n\nblueprint centralize-config-designate\n\nChange-Id: I3be68bf294157631a9bac128df0841220020de96\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/762b10dd452a87d64aaf13f812af90ff1699fd69"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/762b10dd452a87d64aaf13f812af90ff1699fd69"}]},"branch":"refs/heads/master"},"a53c3daf16d366e3cc513c62f57042f43cbd4588":{"kind":"TRIVIAL_REBASE","_number":16,"created":"2019-05-28 23:54:00.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/25/643725/16","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/25/643725/16","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/25/643725/16 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/25/643725/16 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/25/643725/16 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/25/643725/16"}}},"commit":{"parents":[{"commit":"7fe1c1e9b64efc3ecdbe8af72fd5827b84de02fe","subject":"Merge \"Fix api-ref building\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/7fe1c1e9b64efc3ecdbe8af72fd5827b84de02fe"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2019-03-16 23:50:02.000000000","tz":-420},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2019-05-28 23:54:00.000000000","tz":0},"subject":"Moved most configuration into designate.conf","message":"Moved most configuration into designate.conf\n\nBased upon the established conf paradigm used in Nova,\nthis patch moves most of the configuration to a\ncentral location. Making config a lot easier to\nmanage, as well a lot easier to locate.\n\nThere is still some configuration listed in classes,\nthat I would like to eventually move to\ndesignate.conf as well, but to lower complexity not\nall config was moved.\n\nGeneral changes\n* Centrelized most configuration under designate.conf.\n* Cleaned up configuration.\n* Simplified config generation.\n* Renamed generated config to designate.conf.sample.\n\nPreviously broken\n* xfr_timeout now in the appropriate section.\n* memcached_servers now in the appropriate section.\n\nPreviously missing from sample config\n* allowed_remote_exmods listed in sample config.\n* backend:akamai section listed in sample config.\n* backend:dynect section listed in sample config.\n* backend:infoblox section listed in sample config.\n\nblueprint centralize-config-designate\n\nChange-Id: I3be68bf294157631a9bac128df0841220020de96\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/a53c3daf16d366e3cc513c62f57042f43cbd4588"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/a53c3daf16d366e3cc513c62f57042f43cbd4588"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
