)]}'
{"id":"openstack%2Fdesignate~648314","triplet_id":"openstack%2Fdesignate~master~I22dfd49de4ea9efb9b7e791534e033271ff9efd1","project":"openstack/designate","branch":"master","topic":"project_id_neutron","hashtags":[],"change_id":"I22dfd49de4ea9efb9b7e791534e033271ff9efd1","subject":"Switch to using project_id for network_api","status":"MERGED","created":"2019-03-28 04:55:55.000000000","updated":"2019-05-20 11:54:50.000000000","submitted":"2019-05-20 11:54:50.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"648314-1558353290592-172dd170","meta_rev_id":"54c9943e96a9ab3822bd97ad96d5486a3547cd61","_number":648314,"virtual_id_number":648314,"owner":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},{"value":2,"date":"2019-05-20 11:54:50.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},"all":[{"value":2,"date":"2019-04-01 12:26:21.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},"all":[{"value":1,"date":"2019-05-20 10:52:58.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"recommended":{"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},"all":[{"value":1,"date":"2019-04-01 12:26:21.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","value":1,"default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-05-20 10:52:58.000000000","updated_by":{"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},"reviewer":{"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},"state":"REVIEWER"},{"updated":"2019-05-20 11:54:50.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"710724a532f7f5322dfdb6df6e4f28c465343397","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2019-03-28 04:55:55.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"57625e44227a27a13fcb6836942eda7f652b787e","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2019-03-28 04:58:45.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"0a8aaf7aa5fb8c8d16dd810c18e08c32a64a30bd","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-03-28 05:54:58.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- designate-bind9 http://logs.openstack.org/14/648314/2/check/designate-bind9/8e5fe58/ : SUCCESS in 36m 54s\n- designate-bind9-manager-model http://logs.openstack.org/14/648314/2/check/designate-bind9-manager-model/281fff9/ : SUCCESS in 35m 34s\n- designate-bind9-py36 http://logs.openstack.org/14/648314/2/check/designate-bind9-py36/9be4bad/ : SUCCESS in 35m 01s\n- designate-devstack-pdns4 http://logs.openstack.org/14/648314/2/check/designate-devstack-pdns4/e5af601/ : SUCCESS in 36m 13s\n- designate-devstack-pdns4-postgres http://logs.openstack.org/14/648314/2/check/designate-devstack-pdns4-postgres/25b82ae/ : SUCCESS in 34m 15s\n- designate-grenade-pdns4 http://logs.openstack.org/14/648314/2/check/designate-grenade-pdns4/efb0f96/ : SUCCESS in 53m 57s\n- openstack-tox-cover http://logs.openstack.org/14/648314/2/check/openstack-tox-cover/88279cd/cover/ : SUCCESS in 6m 25s\n- openstack-tox-lower-constraints http://logs.openstack.org/14/648314/2/check/openstack-tox-lower-constraints/b6aa496/ : SUCCESS in 6m 08s\n- openstack-tox-pep8 http://logs.openstack.org/14/648314/2/check/openstack-tox-pep8/efe6c70/ : SUCCESS in 5m 28s\n- openstack-tox-py27 http://logs.openstack.org/14/648314/2/check/openstack-tox-py27/d4905ea/ : SUCCESS in 5m 45s\n- openstack-tox-py35 http://logs.openstack.org/14/648314/2/check/openstack-tox-py35/7a67e35/ : SUCCESS in 5m 52s\n- openstack-tox-py36 http://logs.openstack.org/14/648314/2/check/openstack-tox-py36/1962876/ : SUCCESS in 7m 35s\n- openstack-tox-py37 http://logs.openstack.org/14/648314/2/check/openstack-tox-py37/c33b25e/ : SUCCESS in 7m 48s\n- openstack-tox-docs http://logs.openstack.org/14/648314/2/check/openstack-tox-docs/99623a9/html/ : SUCCESS in 7m 19s\n- neutron-tempest-plugin-designate-scenario http://logs.openstack.org/14/648314/2/check/neutron-tempest-plugin-designate-scenario/1f0fe90/ : SUCCESS in 34m 13s","accounts_in_message":[],"_revision_number":2},{"id":"25667a18992aa39949f6731a6eacb0fa9096c70d","author":{"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},"date":"2019-04-01 12:26:21.000000000","message":"Patch Set 2: Code-Review+2 Review-Priority+1","accounts_in_message":[],"_revision_number":2},{"id":"c66b87e88c9126f360b46fab092d00727b2c42a8","author":{"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},"date":"2019-05-20 10:52:58.000000000","message":"Patch Set 2: Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"84e0fcf3a4779bc6cc4cf3dfef197819373a96cb","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-20 10:53:10.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"fa563a1de519bc07fb80e6ff18a458d4b1234b4e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-20 11:54:50.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- designate-bind9 http://logs.openstack.org/14/648314/2/gate/designate-bind9/591c047/ : SUCCESS in 36m 19s\n- designate-pdns4 http://logs.openstack.org/14/648314/2/gate/designate-pdns4/c73c4f1/ : SUCCESS in 36m 56s\n- designate-bind9-manager-model http://logs.openstack.org/14/648314/2/gate/designate-bind9-manager-model/5cc97c6/ : SUCCESS in 36m 57s\n- designate-bind9-py27 http://logs.openstack.org/14/648314/2/gate/designate-bind9-py27/b374916/ : SUCCESS in 45m 46s\n- designate-pdns4-py27 http://logs.openstack.org/14/648314/2/gate/designate-pdns4-py27/8e9b829/ : SUCCESS in 43m 41s\n- designate-pdns4-postgres http://logs.openstack.org/14/648314/2/gate/designate-pdns4-postgres/42dc66d/ : SUCCESS in 37m 04s\n- designate-grenade-pdns4 http://logs.openstack.org/14/648314/2/gate/designate-grenade-pdns4/1343f96/ : SUCCESS in 55m 26s\n- openstack-tox-lower-constraints http://logs.openstack.org/14/648314/2/gate/openstack-tox-lower-constraints/ad73bca/ : SUCCESS in 5m 04s\n- openstack-tox-pep8 http://logs.openstack.org/14/648314/2/gate/openstack-tox-pep8/bea89d5/ : SUCCESS in 4m 48s\n- openstack-tox-py27 http://logs.openstack.org/14/648314/2/gate/openstack-tox-py27/52cc7e1/ : SUCCESS in 5m 00s\n- openstack-tox-py36 http://logs.openstack.org/14/648314/2/gate/openstack-tox-py36/0829536/ : SUCCESS in 5m 24s\n- openstack-tox-py37 http://logs.openstack.org/14/648314/2/gate/openstack-tox-py37/7b636f7/ : SUCCESS in 7m 58s\n- openstack-tox-docs http://logs.openstack.org/14/648314/2/gate/openstack-tox-docs/d5233ee/html/ : SUCCESS in 7m 15s","accounts_in_message":[],"_revision_number":2},{"id":"c0b44656e6edba57d008177f3719a1e2c2a13177","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-20 11:54:50.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"3f5b7fc7c94f2a4675d15c227fb2eeb79acaadf5","revisions":{"3f6b4bd0a0d3539d93cb7d74aa7dac5d78ffddfe":{"kind":"REWORK","_number":1,"created":"2019-03-28 04:55:55.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/14/648314/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/14/648314/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/14/648314/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/14/648314/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/14/648314/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/14/648314/1"}}},"commit":{"parents":[{"commit":"33ca0b87891228d86e440e929feeb324353b39d3","subject":"Merge \"Allow pdns4 backend to select tsigkey\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/33ca0b87891228d86e440e929feeb324353b39d3"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2019-03-28 04:55:33.000000000","tz":-420},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2019-03-28 04:55:33.000000000","tz":-420},"subject":"Switch to using project_id for network_api","message":"Switch to using project_id for network_api\n\nChange-Id: I22dfd49de4ea9efb9b7e791534e033271ff9efd1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/3f6b4bd0a0d3539d93cb7d74aa7dac5d78ffddfe"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/3f6b4bd0a0d3539d93cb7d74aa7dac5d78ffddfe"}]},"branch":"refs/heads/master"},"3f5b7fc7c94f2a4675d15c227fb2eeb79acaadf5":{"kind":"REWORK","_number":2,"created":"2019-03-28 04:58:45.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/14/648314/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/14/648314/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/14/648314/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/14/648314/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/14/648314/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/14/648314/2"}}},"commit":{"parents":[{"commit":"33ca0b87891228d86e440e929feeb324353b39d3","subject":"Merge \"Allow pdns4 backend to select tsigkey\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/33ca0b87891228d86e440e929feeb324353b39d3"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2019-03-28 04:55:33.000000000","tz":-420},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2019-03-28 04:58:31.000000000","tz":-420},"subject":"Switch to using project_id for network_api","message":"Switch to using project_id for network_api\n\nChange-Id: I22dfd49de4ea9efb9b7e791534e033271ff9efd1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/3f5b7fc7c94f2a4675d15c227fb2eeb79acaadf5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/3f5b7fc7c94f2a4675d15c227fb2eeb79acaadf5"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
