)]}'
{"id":"openstack%2Fdesignate~649031","triplet_id":"openstack%2Fdesignate~master~I99d228f850e03045febf196f12eef5927f7605b9","project":"openstack/designate","branch":"master","topic":"vagrant-updates","hashtags":[],"change_id":"I99d228f850e03045febf196f12eef5927f7605b9","subject":"Update vagrant config","status":"MERGED","created":"2019-04-01 13:50:23.000000000","updated":"2019-07-01 16:33:42.000000000","submitted":"2019-07-01 16:33:42.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"649031-1561998822928-427e186e","meta_rev_id":"2460cc5a375ada393c3811a99dcb841fc0014402","_number":649031,"virtual_id_number":649031,"owner":{"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},{"value":2,"date":"2019-07-01 16:33:42.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},"all":[{"value":2,"date":"2019-07-01 15:26:53.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2019-06-19 20:18:45.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},"all":[{"value":1,"date":"2019-07-01 15:26:53.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-06-19 20:18:45.000000000","updated_by":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"reviewer":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"state":"REVIEWER"},{"updated":"2019-07-01 16:33:42.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"4bbe7e11efb19b341e9997dc68de0519a6e47d53","author":{"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},"date":"2019-04-01 13:50:23.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"b79cce356082155d82ea0b07b0899a155fa4b0f2","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-01 14:56:21.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- designate-bind9 http://logs.openstack.org/31/649031/1/check/designate-bind9/35c456a/ : SUCCESS in 43m 36s\n- designate-bind9-manager-model http://logs.openstack.org/31/649031/1/check/designate-bind9-manager-model/dcdc47c/ : SUCCESS in 58m 08s\n- designate-bind9-py36 http://logs.openstack.org/31/649031/1/check/designate-bind9-py36/56174e6/ : SUCCESS in 41m 31s\n- designate-devstack-pdns4 http://logs.openstack.org/31/649031/1/check/designate-devstack-pdns4/4e859ec/ : SUCCESS in 37m 01s\n- designate-devstack-pdns4-postgres http://logs.openstack.org/31/649031/1/check/designate-devstack-pdns4-postgres/cafdd57/ : SUCCESS in 38m 47s\n- designate-grenade-pdns4 http://logs.openstack.org/31/649031/1/check/designate-grenade-pdns4/1f86c9d/ : SUCCESS in 57m 34s\n- openstack-tox-cover http://logs.openstack.org/31/649031/1/check/openstack-tox-cover/21725e3/cover/ : SUCCESS in 7m 18s\n- openstack-tox-lower-constraints http://logs.openstack.org/31/649031/1/check/openstack-tox-lower-constraints/c129444/ : SUCCESS in 5m 02s\n- openstack-tox-pep8 http://logs.openstack.org/31/649031/1/check/openstack-tox-pep8/4c49f43/ : SUCCESS in 5m 36s\n- openstack-tox-py27 http://logs.openstack.org/31/649031/1/check/openstack-tox-py27/8b49bd2/ : SUCCESS in 6m 14s\n- openstack-tox-py35 http://logs.openstack.org/31/649031/1/check/openstack-tox-py35/f5d3344/ : SUCCESS in 6m 25s\n- openstack-tox-py36 http://logs.openstack.org/31/649031/1/check/openstack-tox-py36/0bcda89/ : SUCCESS in 6m 18s\n- openstack-tox-py37 http://logs.openstack.org/31/649031/1/check/openstack-tox-py37/fc1cb45/ : SUCCESS in 6m 44s\n- openstack-tox-docs http://logs.openstack.org/31/649031/1/check/openstack-tox-docs/0990e2a/html/ : SUCCESS in 5m 19s\n- neutron-tempest-plugin-designate-scenario http://logs.openstack.org/31/649031/1/check/neutron-tempest-plugin-designate-scenario/a29c02d/ : FAILURE in 11m 35s","accounts_in_message":[],"_revision_number":1},{"id":"9f04e3a56fc926154c7a012f1de7f571be426ebd","author":{"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},"date":"2019-04-02 10:03:09.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"c66d3f55d5b0b8daf05a1bbab771a15a0732f579","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-02 11:04:14.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- designate-bind9 http://logs.openstack.org/31/649031/1/check/designate-bind9/f3a2840/ : SUCCESS in 50m 02s\n- designate-bind9-manager-model http://logs.openstack.org/31/649031/1/check/designate-bind9-manager-model/44cdd2c/ : SUCCESS in 44m 19s\n- designate-bind9-py36 http://logs.openstack.org/31/649031/1/check/designate-bind9-py36/f96a348/ : SUCCESS in 46m 08s\n- designate-devstack-pdns4 http://logs.openstack.org/31/649031/1/check/designate-devstack-pdns4/0eff09b/ : SUCCESS in 36m 42s\n- designate-devstack-pdns4-postgres http://logs.openstack.org/31/649031/1/check/designate-devstack-pdns4-postgres/f536c51/ : SUCCESS in 33m 10s\n- designate-grenade-pdns4 http://logs.openstack.org/31/649031/1/check/designate-grenade-pdns4/dc16eb3/ : SUCCESS in 52m 04s\n- openstack-tox-cover http://logs.openstack.org/31/649031/1/check/openstack-tox-cover/0c754cd/cover/ : SUCCESS in 7m 38s\n- openstack-tox-lower-constraints http://logs.openstack.org/31/649031/1/check/openstack-tox-lower-constraints/f1ee519/ : SUCCESS in 5m 37s\n- openstack-tox-pep8 http://logs.openstack.org/31/649031/1/check/openstack-tox-pep8/d0efb7f/ : SUCCESS in 6m 33s\n- openstack-tox-py27 http://logs.openstack.org/31/649031/1/check/openstack-tox-py27/99f7ddf/ : SUCCESS in 7m 57s\n- openstack-tox-py35 http://logs.openstack.org/31/649031/1/check/openstack-tox-py35/38d7bc5/ : SUCCESS in 7m 33s\n- openstack-tox-py36 http://logs.openstack.org/31/649031/1/check/openstack-tox-py36/d4c3cbc/ : SUCCESS in 7m 07s\n- openstack-tox-py37 http://logs.openstack.org/31/649031/1/check/openstack-tox-py37/5976740/ : SUCCESS in 8m 28s\n- openstack-tox-docs http://logs.openstack.org/31/649031/1/check/openstack-tox-docs/f284403/html/ : SUCCESS in 6m 26s\n- neutron-tempest-plugin-designate-scenario http://logs.openstack.org/31/649031/1/check/neutron-tempest-plugin-designate-scenario/f2847ee/ : SUCCESS in 41m 58s","accounts_in_message":[],"_revision_number":1},{"id":"547205c0e52fe62d5b5915362abb05cef9289f68","author":{"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},"date":"2019-05-21 13:14:50.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"1fc54330536906617eef587d1ea28768697d37fc","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-21 14:12:02.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- designate-bind9 http://logs.openstack.org/31/649031/2/check/designate-bind9/1717d33/ : SUCCESS in 37m 29s\n- designate-pdns4 http://logs.openstack.org/31/649031/2/check/designate-pdns4/465a5bb/ : SUCCESS in 42m 28s\n- designate-bind9-manager-model http://logs.openstack.org/31/649031/2/check/designate-bind9-manager-model/e60015d/ : SUCCESS in 38m 28s\n- designate-bind9-py27 http://logs.openstack.org/31/649031/2/check/designate-bind9-py27/35d5425/ : SUCCESS in 40m 04s\n- designate-pdns4-py27 http://logs.openstack.org/31/649031/2/check/designate-pdns4-py27/96defd4/ : SUCCESS in 34m 29s\n- designate-pdns4-postgres http://logs.openstack.org/31/649031/2/check/designate-pdns4-postgres/4c77018/ : SUCCESS in 36m 29s\n- designate-grenade-pdns4 http://logs.openstack.org/31/649031/2/check/designate-grenade-pdns4/7f74309/ : SUCCESS in 53m 17s\n- openstack-tox-cover http://logs.openstack.org/31/649031/2/check/openstack-tox-cover/e8ffc9d/cover/ : SUCCESS in 7m 08s\n- openstack-tox-lower-constraints http://logs.openstack.org/31/649031/2/check/openstack-tox-lower-constraints/41f863f/ : SUCCESS in 4m 13s\n- openstack-tox-pep8 http://logs.openstack.org/31/649031/2/check/openstack-tox-pep8/51db1bc/ : SUCCESS in 5m 22s\n- openstack-tox-py27 http://logs.openstack.org/31/649031/2/check/openstack-tox-py27/14b71b6/ : SUCCESS in 6m 40s\n- openstack-tox-py36 http://logs.openstack.org/31/649031/2/check/openstack-tox-py36/90e3cb2/ : SUCCESS in 5m 56s\n- openstack-tox-py37 http://logs.openstack.org/31/649031/2/check/openstack-tox-py37/6b1c3a2/ : SUCCESS in 5m 59s\n- openstack-tox-docs http://logs.openstack.org/31/649031/2/check/openstack-tox-docs/ab87718/html/ : SUCCESS in 5m 08s\n- neutron-tempest-plugin-designate-scenario http://logs.openstack.org/31/649031/2/check/neutron-tempest-plugin-designate-scenario/229dd10/ : SUCCESS in 29m 05s","accounts_in_message":[],"_revision_number":2},{"id":"43eb52e7e62d7d3a0f6b2ace1c96866abbf86088","author":{"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},"date":"2019-06-19 13:44:49.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"291a4dc9449d36f56ae341dfdf747afe54e04581","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-19 14:54:33.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- designate-bind9 http://logs.openstack.org/31/649031/3/check/designate-bind9/d6affff/ : SUCCESS in 35m 45s\n- designate-pdns4 http://logs.openstack.org/31/649031/3/check/designate-pdns4/4298128/ : SUCCESS in 38m 23s\n- designate-bind9-manager-model http://logs.openstack.org/31/649031/3/check/designate-bind9-manager-model/9985732/ : SUCCESS in 40m 45s\n- designate-bind9-py27 http://logs.openstack.org/31/649031/3/check/designate-bind9-py27/e616bcc/ : SUCCESS in 42m 25s\n- designate-pdns4-py27 http://logs.openstack.org/31/649031/3/check/designate-pdns4-py27/2273d85/ : SUCCESS in 41m 22s\n- designate-pdns4-postgres http://logs.openstack.org/31/649031/3/check/designate-pdns4-postgres/5b90d85/ : SUCCESS in 37m 02s\n- designate-grenade-pdns4 http://logs.openstack.org/31/649031/3/check/designate-grenade-pdns4/59c3de2/ : SUCCESS in 1h 01m 26s\n- openstack-tox-cover http://logs.openstack.org/31/649031/3/check/openstack-tox-cover/1124a2f/cover/ : SUCCESS in 6m 02s\n- openstack-tox-lower-constraints http://logs.openstack.org/31/649031/3/check/openstack-tox-lower-constraints/7ae2dbb/ : SUCCESS in 4m 48s\n- openstack-tox-pep8 http://logs.openstack.org/31/649031/3/check/openstack-tox-pep8/1e898eb/ : SUCCESS in 5m 26s\n- openstack-tox-py27 http://logs.openstack.org/31/649031/3/check/openstack-tox-py27/5ff291b/ : SUCCESS in 6m 05s\n- openstack-tox-py36 http://logs.openstack.org/31/649031/3/check/openstack-tox-py36/9535737/ : SUCCESS in 6m 15s\n- openstack-tox-py37 http://logs.openstack.org/31/649031/3/check/openstack-tox-py37/ab18e91/ : SUCCESS in 6m 02s\n- openstack-tox-docs http://logs.openstack.org/31/649031/3/check/openstack-tox-docs/85b206b/html/ : SUCCESS in 5m 20s\n- neutron-tempest-plugin-designate-scenario http://logs.openstack.org/31/649031/3/check/neutron-tempest-plugin-designate-scenario/65b02fd/ : SUCCESS in 37m 38s","accounts_in_message":[],"_revision_number":3},{"id":"c55cda8c97bfa4dce3c6f1afccceb537e8b13b11","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2019-06-19 20:18:45.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"442b32665e4add3fb19a43447d0d919f1a01730a","author":{"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},"date":"2019-07-01 15:26:53.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"48645bfeefaa4a04f3ae5a100951d1a3533d20e1","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-01 15:27:06.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"af7da02c700d4fbb5dbb5fecec763d044ae32503","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-01 16:33:42.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- designate-bind9 http://logs.openstack.org/31/649031/3/gate/designate-bind9/69d6a13/ : SUCCESS in 52m 59s\n- designate-pdns4 http://logs.openstack.org/31/649031/3/gate/designate-pdns4/90d1dd6/ : SUCCESS in 43m 57s\n- designate-bind9-manager-model http://logs.openstack.org/31/649031/3/gate/designate-bind9-manager-model/b04de02/ : SUCCESS in 40m 52s\n- designate-bind9-py27 http://logs.openstack.org/31/649031/3/gate/designate-bind9-py27/c81248d/ : SUCCESS in 44m 21s\n- designate-pdns4-py27 http://logs.openstack.org/31/649031/3/gate/designate-pdns4-py27/6d5d944/ : SUCCESS in 37m 51s\n- designate-pdns4-postgres http://logs.openstack.org/31/649031/3/gate/designate-pdns4-postgres/e6873f3/ : SUCCESS in 38m 59s\n- designate-grenade-pdns4 http://logs.openstack.org/31/649031/3/gate/designate-grenade-pdns4/de6cb98/ : SUCCESS in 59m 51s\n- openstack-tox-lower-constraints http://logs.openstack.org/31/649031/3/gate/openstack-tox-lower-constraints/bf6f49f/ : SUCCESS in 3m 37s\n- openstack-tox-pep8 http://logs.openstack.org/31/649031/3/gate/openstack-tox-pep8/7ac6e61/ : SUCCESS in 3m 26s\n- openstack-tox-py27 http://logs.openstack.org/31/649031/3/gate/openstack-tox-py27/a63ef3d/ : SUCCESS in 4m 24s\n- openstack-tox-py36 http://logs.openstack.org/31/649031/3/gate/openstack-tox-py36/ffc7b79/ : SUCCESS in 4m 20s\n- openstack-tox-py37 http://logs.openstack.org/31/649031/3/gate/openstack-tox-py37/c688b9f/ : SUCCESS in 4m 56s\n- openstack-tox-docs http://logs.openstack.org/31/649031/3/gate/openstack-tox-docs/4c9e387/html/ : SUCCESS in 3m 41s","accounts_in_message":[],"_revision_number":3},{"id":"0c6db4c878bd5f6de7fb83d5c3ecc5cb8b7b0d92","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-01 16:33:42.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"3734f2a13f094fe86c354c87e102543f9b1fa006","revisions":{"dc4d29a9309b27141897b8ef6e82f32b7cb3b533":{"kind":"REWORK","_number":1,"created":"2019-04-01 13:50:23.000000000","uploader":{"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},"ref":"refs/changes/31/649031/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/31/649031/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/31/649031/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/31/649031/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/31/649031/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/31/649031/1"}}},"commit":{"parents":[{"commit":"33ca0b87891228d86e440e929feeb324353b39d3","subject":"Merge \"Allow pdns4 backend to select tsigkey\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/33ca0b87891228d86e440e929feeb324353b39d3"}]}],"author":{"name":"Graham Hayes","email":"gr@ham.ie","date":"2019-04-01 13:48:48.000000000","tz":60},"committer":{"name":"Graham Hayes","email":"gr@ham.ie","date":"2019-04-01 13:48:48.000000000","tz":60},"subject":"Update vagrant config","message":"Update vagrant config\n\n* Move to bionic\n* Add openSUSE Leap 15\n* * Update Bind devstack backend plugin to work on Leap\n* Remove port mapping\n* Rmove the .gitconfig copy into the VM\n\nChange-Id: I99d228f850e03045febf196f12eef5927f7605b9\nSigned-off-by: Graham Hayes \u003cgr@ham.ie\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/dc4d29a9309b27141897b8ef6e82f32b7cb3b533"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/dc4d29a9309b27141897b8ef6e82f32b7cb3b533"}]},"branch":"refs/heads/master"},"8f82ba8fc12691e1252df52e9b8f6645b75090f7":{"kind":"REWORK","_number":2,"created":"2019-05-21 13:14:50.000000000","uploader":{"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},"ref":"refs/changes/31/649031/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/31/649031/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/31/649031/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/31/649031/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/31/649031/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/31/649031/2"}}},"commit":{"parents":[{"commit":"8594066e2b172b8338a5bbb3439cd75aeeef3e9f","subject":"Merge \"Added PowerDNS 4 unit tests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/8594066e2b172b8338a5bbb3439cd75aeeef3e9f"}]}],"author":{"name":"Graham Hayes","email":"gr@ham.ie","date":"2019-04-01 13:48:48.000000000","tz":60},"committer":{"name":"Graham Hayes","email":"gr@ham.ie","date":"2019-05-21 13:14:24.000000000","tz":60},"subject":"Update vagrant config","message":"Update vagrant config\n\n* Move to bionic\n* Add openSUSE Leap 15\n* * Update Bind devstack backend plugin to work on Leap\n* Remove port mapping\n* Rmove the .gitconfig copy into the VM\n\nChange-Id: I99d228f850e03045febf196f12eef5927f7605b9\nSigned-off-by: Graham Hayes \u003cgr@ham.ie\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/8f82ba8fc12691e1252df52e9b8f6645b75090f7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/8f82ba8fc12691e1252df52e9b8f6645b75090f7"}]},"branch":"refs/heads/master"},"3734f2a13f094fe86c354c87e102543f9b1fa006":{"kind":"REWORK","_number":3,"created":"2019-06-19 13:44:49.000000000","uploader":{"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},"ref":"refs/changes/31/649031/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/31/649031/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/31/649031/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/31/649031/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/31/649031/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/31/649031/3"}}},"commit":{"parents":[{"commit":"8594066e2b172b8338a5bbb3439cd75aeeef3e9f","subject":"Merge \"Added PowerDNS 4 unit tests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/8594066e2b172b8338a5bbb3439cd75aeeef3e9f"}]}],"author":{"name":"Graham Hayes","email":"gr@ham.ie","date":"2019-04-01 13:48:48.000000000","tz":60},"committer":{"name":"Graham Hayes","email":"gr@ham.ie","date":"2019-06-19 13:44:37.000000000","tz":60},"subject":"Update vagrant config","message":"Update vagrant config\n\n* Move to bionic\n* Add openSUSE Leap 15\n* * Update Bind devstack backend plugin to work on Leap\n* Remove port mapping\n* Rmove the .gitconfig copy into the VM\n\nChange-Id: I99d228f850e03045febf196f12eef5927f7605b9\nSigned-off-by: Graham Hayes \u003cgr@ham.ie\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/3734f2a13f094fe86c354c87e102543f9b1fa006"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/3734f2a13f094fe86c354c87e102543f9b1fa006"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
