)]}'
{"id":"openstack%2Fdesignate~649272","triplet_id":"openstack%2Fdesignate~stable%2Fstein~Iadb2660d306c8920c2202ae6bffec2ef621af36c","project":"openstack/designate","branch":"stable/stein","topic":"missed_constraint-stable/stein","hashtags":[],"change_id":"Iadb2660d306c8920c2202ae6bffec2ef621af36c","subject":"Added missing service_statuses metadata","status":"MERGED","created":"2019-04-02 10:02:28.000000000","updated":"2019-05-20 11:48:55.000000000","submitted":"2019-05-20 11:48:55.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"649272-1558352935674-d1a6a7d1","meta_rev_id":"a02d6bb905e5fa0c2a766047ff5e71b068cda7ea","_number":649272,"virtual_id_number":649272,"owner":{"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},{"value":2,"date":"2019-05-20 11:48:55.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":30381,"name":"Jean-Pierre REN","email":"renzp@awcloud.com","username":"jpren"},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},"all":[{"value":2,"date":"2019-05-20 10:42:24.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2019-05-13 04:01:00.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":30381,"name":"Jean-Pierre REN","email":"renzp@awcloud.com","username":"jpren"},{"value":2,"date":"2019-05-20 00:09:25.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},"all":[{"value":1,"date":"2019-05-20 10:42:24.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":30381,"name":"Jean-Pierre REN","email":"renzp@awcloud.com","username":"jpren"},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":30381,"name":"Jean-Pierre REN","email":"renzp@awcloud.com","username":"jpren"},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},{"_account_id":30381,"name":"Jean-Pierre REN","email":"renzp@awcloud.com","username":"jpren"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-05-13 04:01:00.000000000","updated_by":{"_account_id":30381,"name":"Jean-Pierre REN","email":"renzp@awcloud.com","username":"jpren"},"reviewer":{"_account_id":30381,"name":"Jean-Pierre REN","email":"renzp@awcloud.com","username":"jpren"},"state":"REVIEWER"},{"updated":"2019-05-20 00:09:25.000000000","updated_by":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"reviewer":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"state":"REVIEWER"},{"updated":"2019-05-20 11:48:55.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"7451dc8e915adac914802c5c1894375288500ad3","author":{"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},"date":"2019-04-02 10:02:28.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"9aef25b92834678d0e152d7c8fbd17f0b8d3d3a1","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-02 11:14:52.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- designate-bind9 http://logs.openstack.org/72/649272/1/check/designate-bind9/4725ac1/ : SUCCESS in 50m 49s\n- designate-bind9-manager-model http://logs.openstack.org/72/649272/1/check/designate-bind9-manager-model/e5fdb51/ : SUCCESS in 51m 14s\n- designate-bind9-py36 http://logs.openstack.org/72/649272/1/check/designate-bind9-py36/92e63bb/ : SUCCESS in 45m 20s\n- designate-devstack-pdns4 http://logs.openstack.org/72/649272/1/check/designate-devstack-pdns4/cbf203c/ : SUCCESS in 38m 57s\n- designate-devstack-pdns4-postgres http://logs.openstack.org/72/649272/1/check/designate-devstack-pdns4-postgres/8cc88b8/ : SUCCESS in 32m 17s\n- designate-grenade-pdns4 http://logs.openstack.org/72/649272/1/check/designate-grenade-pdns4/3a070d6/ : SUCCESS in 1h 00m 39s\n- openstack-tox-cover http://logs.openstack.org/72/649272/1/check/openstack-tox-cover/6131357/cover/ : SUCCESS in 7m 31s\n- openstack-tox-lower-constraints http://logs.openstack.org/72/649272/1/check/openstack-tox-lower-constraints/b1e9dfa/ : SUCCESS in 6m 13s\n- openstack-tox-pep8 http://logs.openstack.org/72/649272/1/check/openstack-tox-pep8/ba880a5/ : SUCCESS in 7m 44s\n- openstack-tox-py27 http://logs.openstack.org/72/649272/1/check/openstack-tox-py27/ac161fb/ : SUCCESS in 6m 54s\n- openstack-tox-py35 http://logs.openstack.org/72/649272/1/check/openstack-tox-py35/694e641/ : SUCCESS in 5m 35s\n- openstack-tox-py36 http://logs.openstack.org/72/649272/1/check/openstack-tox-py36/d18f38a/ : SUCCESS in 5m 32s\n- openstack-tox-py37 http://logs.openstack.org/72/649272/1/check/openstack-tox-py37/e3de833/ : SUCCESS in 7m 04s\n- openstack-tox-docs http://logs.openstack.org/72/649272/1/check/openstack-tox-docs/a966015/html/ : SUCCESS in 6m 02s\n- neutron-tempest-plugin-designate-scenario http://logs.openstack.org/72/649272/1/check/neutron-tempest-plugin-designate-scenario/4f73029/ : SUCCESS in 47m 25s","accounts_in_message":[],"_revision_number":1},{"id":"5f27fa1877c2bda347e5bee897824fe98cc36c11","author":{"_account_id":30381,"name":"Jean-Pierre REN","email":"renzp@awcloud.com","username":"jpren"},"date":"2019-05-13 04:01:00.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"2d6e9c3d70a5b0fe8e3b333b69acb2924089d4ec","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2019-05-20 00:09:25.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"efbf3a3720316789bd83bea5f522109ce214164a","author":{"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},"date":"2019-05-20 10:42:24.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"ac6a28ffaeb9928fca3ff39d6cf199e4f096eb32","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-20 10:42:36.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"aae94bc6f77e14aabd6194a05e61e9188a057be1","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-20 11:48:55.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- designate-bind9 http://logs.openstack.org/72/649272/1/gate/designate-bind9/485349a/ : SUCCESS in 40m 16s\n- designate-bind9-manager-model http://logs.openstack.org/72/649272/1/gate/designate-bind9-manager-model/3fc9dd5/ : SUCCESS in 36m 29s\n- designate-bind9-py36 http://logs.openstack.org/72/649272/1/gate/designate-bind9-py36/e51ecce/ : SUCCESS in 39m 26s\n- designate-devstack-pdns4 http://logs.openstack.org/72/649272/1/gate/designate-devstack-pdns4/32e83e5/ : SUCCESS in 38m 25s\n- designate-devstack-pdns4-postgres http://logs.openstack.org/72/649272/1/gate/designate-devstack-pdns4-postgres/f0afccb/ : SUCCESS in 36m 25s\n- designate-grenade-pdns4 http://logs.openstack.org/72/649272/1/gate/designate-grenade-pdns4/08ada10/ : SUCCESS in 59m 24s\n- openstack-tox-lower-constraints http://logs.openstack.org/72/649272/1/gate/openstack-tox-lower-constraints/6b9c122/ : SUCCESS in 4m 48s\n- openstack-tox-pep8 http://logs.openstack.org/72/649272/1/gate/openstack-tox-pep8/93177a8/ : SUCCESS in 4m 53s\n- openstack-tox-py27 http://logs.openstack.org/72/649272/1/gate/openstack-tox-py27/3c4a49e/ : SUCCESS in 5m 14s\n- openstack-tox-py35 http://logs.openstack.org/72/649272/1/gate/openstack-tox-py35/b3e8a5d/ : SUCCESS in 6m 24s\n- openstack-tox-py36 http://logs.openstack.org/72/649272/1/gate/openstack-tox-py36/f252801/ : SUCCESS in 6m 44s\n- openstack-tox-py37 http://logs.openstack.org/72/649272/1/gate/openstack-tox-py37/4b44f27/ : SUCCESS in 7m 27s\n- openstack-tox-docs http://logs.openstack.org/72/649272/1/gate/openstack-tox-docs/31265bf/html/ : SUCCESS in 6m 14s","accounts_in_message":[],"_revision_number":1},{"id":"215f3442a40e59eed6ef21dd06a43a528437e1ac","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-20 11:48:55.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"c6eb3fa7bffb5c5c2d296dee9de796d7a81af0fd","revisions":{"c6eb3fa7bffb5c5c2d296dee9de796d7a81af0fd":{"kind":"REWORK","_number":1,"created":"2019-04-02 10:02:28.000000000","uploader":{"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},"ref":"refs/changes/72/649272/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/72/649272/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/72/649272/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/72/649272/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/72/649272/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/72/649272/1"}}},"commit":{"parents":[{"commit":"f6cd521eb22d31bd41be58c2ba8a93baf619449f","subject":"Replace openstack.org git:// URLs with https://","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/f6cd521eb22d31bd41be58c2ba8a93baf619449f"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2019-03-28 04:42:53.000000000","tz":-420},"committer":{"name":"Graham Hayes","email":"gr@ham.ie","date":"2019-04-02 10:02:28.000000000","tz":0},"subject":"Added missing service_statuses metadata","message":"Added missing service_statuses metadata\n\nThis is a follow up to\nI307a8f7dd8b8a83effa447a846db3288efa32dba\n\nAdding the same constraint to the sql metadata.\n\nChange-Id: Iadb2660d306c8920c2202ae6bffec2ef621af36c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/c6eb3fa7bffb5c5c2d296dee9de796d7a81af0fd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/c6eb3fa7bffb5c5c2d296dee9de796d7a81af0fd"}]},"branch":"refs/heads/stable/stein"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
