)]}'
{"id":"openstack%2Fdesignate~650603","triplet_id":"openstack%2Fdesignate~stable%2Fstein~Icd6e690ac84a2fe0db0f4a8a513de47f7916f5ea","project":"openstack/designate","branch":"stable/stein","topic":"bug/1785459-stable/stein","hashtags":[],"change_id":"Icd6e690ac84a2fe0db0f4a8a513de47f7916f5ea","subject":"Fix DBDeadLock error resulting into 500","status":"MERGED","created":"2019-04-08 02:27:17.000000000","updated":"2019-05-08 18:34:42.000000000","submitted":"2019-05-08 18:34:42.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"650603-1557340482713-92f2513b","meta_rev_id":"540d0364a3e5d6a26d94dfe9889cf709ecb7a458","_number":650603,"virtual_id_number":650603,"owner":{"_account_id":28140,"name":"Takahito Hirose","email":"takahito.hirose0518@gmail.com","username":"thirose"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"date":"2019-05-06 13:27:57.000000000","_account_id":28140,"name":"Takahito Hirose","email":"takahito.hirose0518@gmail.com","username":"thirose"},{"value":2,"date":"2019-05-08 18:34:42.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},"all":[{"value":2,"date":"2019-05-08 12:15:49.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":28140,"name":"Takahito Hirose","email":"takahito.hirose0518@gmail.com","username":"thirose"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2019-04-08 03:44:51.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},"all":[{"value":1,"date":"2019-05-08 12:15:49.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":28140,"name":"Takahito Hirose","email":"takahito.hirose0518@gmail.com","username":"thirose"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},{"value":0,"date":"2019-05-07 02:55:28.000000000","permitted_voting_range":{"min":0,"max":2},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":28140,"name":"Takahito Hirose","email":"takahito.hirose0518@gmail.com","username":"thirose"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},{"_account_id":28140,"name":"Takahito Hirose","email":"takahito.hirose0518@gmail.com","username":"thirose"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-04-08 03:44:51.000000000","updated_by":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"reviewer":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"state":"REVIEWER"},{"updated":"2019-05-07 02:55:28.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"REVIEWER"},{"updated":"2019-05-08 12:15:49.000000000","updated_by":{"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},"reviewer":{"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},"state":"REVIEWER"},{"updated":"2019-05-08 18:34:42.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"18ab24dec03ecca592aa4007029dfb978db889bb","author":{"_account_id":28140,"name":"Takahito Hirose","email":"takahito.hirose0518@gmail.com","username":"thirose"},"date":"2019-04-08 02:27:17.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"20fcd9470a92c0b922102eed3901ed887d6dea64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-08 03:22:13.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- designate-bind9 http://logs.openstack.org/03/650603/1/check/designate-bind9/299dae6/ : SUCCESS in 40m 38s\n- designate-bind9-manager-model http://logs.openstack.org/03/650603/1/check/designate-bind9-manager-model/aff7756/ : SUCCESS in 36m 30s\n- designate-bind9-py36 http://logs.openstack.org/03/650603/1/check/designate-bind9-py36/2adc73d/ : SUCCESS in 39m 48s\n- designate-devstack-pdns4 http://logs.openstack.org/03/650603/1/check/designate-devstack-pdns4/0249983/ : SUCCESS in 33m 37s\n- designate-devstack-pdns4-postgres http://logs.openstack.org/03/650603/1/check/designate-devstack-pdns4-postgres/2a554f0/ : SUCCESS in 31m 37s\n- designate-grenade-pdns4 http://logs.openstack.org/03/650603/1/check/designate-grenade-pdns4/616aadb/ : SUCCESS in 52m 28s\n- openstack-tox-cover http://logs.openstack.org/03/650603/1/check/openstack-tox-cover/ba90f3d/cover/ : SUCCESS in 6m 30s\n- openstack-tox-lower-constraints http://logs.openstack.org/03/650603/1/check/openstack-tox-lower-constraints/e88a4a8/ : SUCCESS in 4m 44s\n- openstack-tox-pep8 http://logs.openstack.org/03/650603/1/check/openstack-tox-pep8/fb70866/ : SUCCESS in 5m 12s\n- openstack-tox-py27 http://logs.openstack.org/03/650603/1/check/openstack-tox-py27/5d5807e/ : SUCCESS in 5m 55s\n- openstack-tox-py35 http://logs.openstack.org/03/650603/1/check/openstack-tox-py35/c9351c2/ : SUCCESS in 5m 49s\n- openstack-tox-py36 http://logs.openstack.org/03/650603/1/check/openstack-tox-py36/ad30257/ : SUCCESS in 5m 58s\n- openstack-tox-py37 http://logs.openstack.org/03/650603/1/check/openstack-tox-py37/a8e1125/ : SUCCESS in 5m 57s\n- openstack-tox-docs http://logs.openstack.org/03/650603/1/check/openstack-tox-docs/d0700b5/html/ : SUCCESS in 5m 01s\n- neutron-tempest-plugin-designate-scenario http://logs.openstack.org/03/650603/1/check/neutron-tempest-plugin-designate-scenario/81355d3/ : SUCCESS in 31m 33s","accounts_in_message":[],"_revision_number":1},{"id":"59ad1b8a4284d3b07f5454cf72aa1a866f4705f7","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2019-04-08 03:44:51.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"25437b3dfea5ca938b8c0e31e406c419c1dbcfcb","author":{"_account_id":28140,"name":"Takahito Hirose","email":"takahito.hirose0518@gmail.com","username":"thirose"},"date":"2019-05-06 13:27:57.000000000","message":"Patch Set 1:\n\nAlready got two +2. if this seems good then can someone merge it.\nAnd check other version\u0027s same changes.","accounts_in_message":[],"_revision_number":1},{"id":"1e0f0d98afd0f978d776555d0cad66b17885e7ca","author":{"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},"date":"2019-05-08 12:15:49.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"dabff28641e6a35c8033c4d9bd395fe91419a25f","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-08 12:16:01.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"f4f46ec714acc05bc046bde00f86429e893fd62e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-08 13:16:11.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- designate-bind9 http://logs.openstack.org/03/650603/1/gate/designate-bind9/fa9ee94/ : SUCCESS in 36m 26s\n- designate-bind9-manager-model http://logs.openstack.org/03/650603/1/gate/designate-bind9-manager-model/dc67508/ : SUCCESS in 39m 21s\n- designate-bind9-py36 http://logs.openstack.org/03/650603/1/gate/designate-bind9-py36/64c6e7c/ : SUCCESS in 40m 50s\n- designate-devstack-pdns4 http://logs.openstack.org/03/650603/1/gate/designate-devstack-pdns4/c09f9d2/ : SUCCESS in 34m 02s\n- designate-devstack-pdns4-postgres http://logs.openstack.org/03/650603/1/gate/designate-devstack-pdns4-postgres/f0236b4/ : SUCCESS in 30m 32s\n- designate-grenade-pdns4 http://logs.openstack.org/03/650603/1/gate/designate-grenade-pdns4/0b1f89c/ : FAILURE in 57m 26s\n- openstack-tox-lower-constraints http://logs.openstack.org/03/650603/1/gate/openstack-tox-lower-constraints/5581d28/ : SUCCESS in 4m 49s\n- openstack-tox-pep8 http://logs.openstack.org/03/650603/1/gate/openstack-tox-pep8/ec9974e/ : SUCCESS in 9m 11s\n- openstack-tox-py27 http://logs.openstack.org/03/650603/1/gate/openstack-tox-py27/a2cafb3/ : SUCCESS in 6m 32s\n- openstack-tox-py35 http://logs.openstack.org/03/650603/1/gate/openstack-tox-py35/d6892da/ : SUCCESS in 6m 07s\n- openstack-tox-py36 http://logs.openstack.org/03/650603/1/gate/openstack-tox-py36/932aabe/ : SUCCESS in 5m 06s\n- openstack-tox-py37 http://logs.openstack.org/03/650603/1/gate/openstack-tox-py37/706db44/ : SUCCESS in 7m 38s\n- openstack-tox-docs http://logs.openstack.org/03/650603/1/gate/openstack-tox-docs/d822c18/html/ : SUCCESS in 4m 58s","accounts_in_message":[],"_revision_number":1},{"id":"4624ba425322c03d4b4366de688aa580714f075e","author":{"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},"date":"2019-05-08 15:01:50.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"c48f4dd98cb11cccd69dc686268d036605de56ec","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-08 16:06:28.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- designate-bind9 http://logs.openstack.org/03/650603/1/check/designate-bind9/136fd14/ : SUCCESS in 42m 12s\n- designate-bind9-manager-model http://logs.openstack.org/03/650603/1/check/designate-bind9-manager-model/311dfee/ : SUCCESS in 39m 45s\n- designate-bind9-py36 http://logs.openstack.org/03/650603/1/check/designate-bind9-py36/085194f/ : SUCCESS in 37m 00s\n- designate-devstack-pdns4 http://logs.openstack.org/03/650603/1/check/designate-devstack-pdns4/1e3acfb/ : SUCCESS in 36m 17s\n- designate-devstack-pdns4-postgres http://logs.openstack.org/03/650603/1/check/designate-devstack-pdns4-postgres/9182ac2/ : SUCCESS in 35m 54s\n- designate-grenade-pdns4 http://logs.openstack.org/03/650603/1/check/designate-grenade-pdns4/f081411/ : SUCCESS in 59m 03s\n- openstack-tox-cover http://logs.openstack.org/03/650603/1/check/openstack-tox-cover/17672de/cover/ : SUCCESS in 5m 49s\n- openstack-tox-lower-constraints http://logs.openstack.org/03/650603/1/check/openstack-tox-lower-constraints/04f169d/ : SUCCESS in 4m 02s\n- openstack-tox-pep8 http://logs.openstack.org/03/650603/1/check/openstack-tox-pep8/4049d37/ : SUCCESS in 5m 25s\n- openstack-tox-py27 http://logs.openstack.org/03/650603/1/check/openstack-tox-py27/130e05c/ : SUCCESS in 5m 01s\n- openstack-tox-py35 http://logs.openstack.org/03/650603/1/check/openstack-tox-py35/43e14af/ : FAILURE in 5m 17s\n- openstack-tox-py36 http://logs.openstack.org/03/650603/1/check/openstack-tox-py36/c5ad39e/ : SUCCESS in 6m 06s\n- openstack-tox-py37 http://logs.openstack.org/03/650603/1/check/openstack-tox-py37/0918d70/ : SUCCESS in 6m 38s\n- openstack-tox-docs http://logs.openstack.org/03/650603/1/check/openstack-tox-docs/4b7f7b9/html/ : SUCCESS in 4m 51s\n- neutron-tempest-plugin-designate-scenario http://logs.openstack.org/03/650603/1/check/neutron-tempest-plugin-designate-scenario/630c716/ : SUCCESS in 33m 39s","accounts_in_message":[],"_revision_number":1},{"id":"2dd01f4d01103a87840791b26c1d7e7f1381d072","author":{"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},"date":"2019-05-08 16:30:38.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"7e406f6c90b150a667511151ceac7857ef2ec68b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-08 17:39:45.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- designate-bind9 http://logs.openstack.org/03/650603/1/check/designate-bind9/d8ba6dc/ : SUCCESS in 37m 15s\n- designate-bind9-manager-model http://logs.openstack.org/03/650603/1/check/designate-bind9-manager-model/252811d/ : SUCCESS in 44m 24s\n- designate-bind9-py36 http://logs.openstack.org/03/650603/1/check/designate-bind9-py36/f3a3000/ : SUCCESS in 36m 50s\n- designate-devstack-pdns4 http://logs.openstack.org/03/650603/1/check/designate-devstack-pdns4/0b15126/ : SUCCESS in 35m 02s\n- designate-devstack-pdns4-postgres http://logs.openstack.org/03/650603/1/check/designate-devstack-pdns4-postgres/f32cfca/ : SUCCESS in 35m 54s\n- designate-grenade-pdns4 http://logs.openstack.org/03/650603/1/check/designate-grenade-pdns4/78dd14a/ : SUCCESS in 59m 36s\n- openstack-tox-cover http://logs.openstack.org/03/650603/1/check/openstack-tox-cover/343c87e/cover/ : SUCCESS in 7m 29s\n- openstack-tox-lower-constraints http://logs.openstack.org/03/650603/1/check/openstack-tox-lower-constraints/2421529/ : SUCCESS in 4m 47s\n- openstack-tox-pep8 http://logs.openstack.org/03/650603/1/check/openstack-tox-pep8/59d073f/ : SUCCESS in 4m 42s\n- openstack-tox-py27 http://logs.openstack.org/03/650603/1/check/openstack-tox-py27/8316f74/ : SUCCESS in 6m 49s\n- openstack-tox-py35 http://logs.openstack.org/03/650603/1/check/openstack-tox-py35/f0b8fa1/ : SUCCESS in 5m 17s\n- openstack-tox-py36 http://logs.openstack.org/03/650603/1/check/openstack-tox-py36/d0ff0e6/ : SUCCESS in 5m 30s\n- openstack-tox-py37 http://logs.openstack.org/03/650603/1/check/openstack-tox-py37/1e3940a/ : SUCCESS in 5m 57s\n- openstack-tox-docs http://logs.openstack.org/03/650603/1/check/openstack-tox-docs/a32e469/html/ : SUCCESS in 4m 44s\n- neutron-tempest-plugin-designate-scenario http://logs.openstack.org/03/650603/1/check/neutron-tempest-plugin-designate-scenario/c0d4f10/ : SUCCESS in 34m 59s","accounts_in_message":[],"_revision_number":1},{"id":"2184c081c4574e6f57a62ce902b1b4435b9c5fe7","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-08 17:39:58.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"476a64d601b5ca7d2928eddddf4b016a657e5eff","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-08 18:34:42.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"94b1656ca4aa343e376d52497c997ec337094b41","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-08 18:34:42.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- designate-bind9 http://logs.openstack.org/03/650603/1/gate/designate-bind9/297efa8/ : SUCCESS in 49m 47s\n- designate-bind9-manager-model http://logs.openstack.org/03/650603/1/gate/designate-bind9-manager-model/6207db4/ : SUCCESS in 32m 23s\n- designate-bind9-py36 http://logs.openstack.org/03/650603/1/gate/designate-bind9-py36/af6a966/ : SUCCESS in 36m 24s\n- designate-devstack-pdns4 http://logs.openstack.org/03/650603/1/gate/designate-devstack-pdns4/49f61ac/ : SUCCESS in 32m 57s\n- designate-devstack-pdns4-postgres http://logs.openstack.org/03/650603/1/gate/designate-devstack-pdns4-postgres/8b915ac/ : SUCCESS in 53m 35s\n- designate-grenade-pdns4 http://logs.openstack.org/03/650603/1/gate/designate-grenade-pdns4/d8cdd9d/ : SUCCESS in 47m 43s\n- openstack-tox-lower-constraints http://logs.openstack.org/03/650603/1/gate/openstack-tox-lower-constraints/a2745a5/ : SUCCESS in 4m 11s\n- openstack-tox-pep8 http://logs.openstack.org/03/650603/1/gate/openstack-tox-pep8/aca58e7/ : SUCCESS in 4m 56s\n- openstack-tox-py27 http://logs.openstack.org/03/650603/1/gate/openstack-tox-py27/8ff3150/ : SUCCESS in 7m 36s\n- openstack-tox-py35 http://logs.openstack.org/03/650603/1/gate/openstack-tox-py35/8ef0b33/ : SUCCESS in 8m 01s\n- openstack-tox-py36 http://logs.openstack.org/03/650603/1/gate/openstack-tox-py36/44ca485/ : SUCCESS in 5m 55s\n- openstack-tox-py37 http://logs.openstack.org/03/650603/1/gate/openstack-tox-py37/9a331bd/ : SUCCESS in 6m 54s\n- openstack-tox-docs http://logs.openstack.org/03/650603/1/gate/openstack-tox-docs/94054d2/html/ : SUCCESS in 4m 28s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"9b4b876d6ae3298110e15b6998306a2cf5660c3f","revisions":{"9b4b876d6ae3298110e15b6998306a2cf5660c3f":{"kind":"REWORK","_number":1,"created":"2019-04-08 02:27:17.000000000","uploader":{"_account_id":28140,"name":"Takahito Hirose","email":"takahito.hirose0518@gmail.com","username":"thirose"},"ref":"refs/changes/03/650603/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/03/650603/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/03/650603/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/03/650603/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/03/650603/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/03/650603/1"}}},"commit":{"parents":[{"commit":"f6cd521eb22d31bd41be58c2ba8a93baf619449f","subject":"Replace openstack.org git:// URLs with https://","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/f6cd521eb22d31bd41be58c2ba8a93baf619449f"}]}],"author":{"name":"Takahito Hirose","email":"takahito.hirose0518@gmail.com","date":"2019-03-26 10:52:33.000000000","tz":540},"committer":{"name":"Takahito Hirose","email":"takahito.hirose0518@gmail.com","date":"2019-04-08 02:27:17.000000000","tz":0},"subject":"Fix DBDeadLock error resulting into 500","message":"Fix DBDeadLock error resulting into 500\n\nWhen user requests the record registration request continuously,\nsometimes designate hits DBDeadLock resuting into 500 InternalServerError.\n\nWe get below error:\n\n2019-02-21 21:30:39.925 49752 ERROR designate.api.middleware RemoteError:\nRemote error: DBDeadlock (pymysql.err.InternalError)\n(1213, u\u0027Deadlock found when trying to get lock; try restarting transaction\u0027)\n[SQL: u\u0027UPDATE records SET version\u003d(records.version + %(version_1)s),\nupdated_at\u003d%(updated_at)s, data\u003d%(data)s, hash\u003d%(hash)s, status\u003d%(status)s,\naction\u003d%(action)s, serial\u003d%(serial)s WHERE records.id \u003d %(id_1)s\u0027]\n[parameters: {\u0027status\u0027: \u0027PENDING\u0027, \u0027hash\u0027: \u002739795ee18c6e3c9ad1c0190c6a3d8d4f\u0027,\n\u0027updated_at\u0027: datetime.datetime(2019, 2, 21, 12, 30, 39, 909846), u\u0027version_1\u0027: 1,\nu\u0027id_1\u0027: \u00277a655eeda4d446cdaa81caf19ab55fcc\u0027, \u0027action\u0027: \u0027UPDATE\u0027,\n\u0027serial\u0027: 1550752338,\n\u0027data\u0027: u\u0027ns2.example.jp. domain.example.com. 1550752338 3552 600 86400 3600\u0027}]\n\nIn the process of record registeration, designate first tried to update\nthe reocrd and then update the zone status.\n\nUpdating the zone_status and registering the record process[1] and after synced\nupdate record_status and zone_status process[2] are in reverse order. So If user\nrequest the registering record many time and same time, Designate will get the\nDBDeadLock, when these processes run the same time.\n\nWe observed that changing the order of the operations solves this issue.\n\n[1] https://github.com/openstack/designate/blob/master/designate/central/service.py#L1292-L1320\n[2] https://github.com/openstack/designate/blob/master/designate/central/service.py#L2310-L2322\n\n1. transaction [1]-1 updating zone status process \u003c- run    ---\u003e table_name-zone\n2. transaction [2]-1 updating record status process \u003c- run  ---\u003e table_name-record\n3. transaction [1]-2 registering record process \u003c- run and wait  ---\u003e table_name-record\n4. transaction [2]-2 updating zone process \u003c-deadlock!           ---\u003e table_name-zone\n\nChange-Id: Icd6e690ac84a2fe0db0f4a8a513de47f7916f5ea\nRelated-Bug: #1785459\n(cherry picked from commit f828654a3d40476cac7eb24a09a36e9978c2d708)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/9b4b876d6ae3298110e15b6998306a2cf5660c3f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/9b4b876d6ae3298110e15b6998306a2cf5660c3f"}]},"branch":"refs/heads/stable/stein"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
