)]}'
{"id":"openstack%2Fdesignate~650605","triplet_id":"openstack%2Fdesignate~stable%2Fqueens~Icd6e690ac84a2fe0db0f4a8a513de47f7916f5ea","project":"openstack/designate","branch":"stable/queens","topic":"bug/1785459-stable/queens","hashtags":[],"change_id":"Icd6e690ac84a2fe0db0f4a8a513de47f7916f5ea","subject":"Fix DBDeadLock error resulting into 500","status":"MERGED","created":"2019-04-08 02:29:22.000000000","updated":"2019-06-17 20:16:00.000000000","submitted":"2019-06-17 20:16:00.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"650605-1560802560812-ad3aa802","meta_rev_id":"8163af0d3deada93f681615d380e9b7a3006edf2","_number":650605,"virtual_id_number":650605,"owner":{"_account_id":28140,"name":"Takahito Hirose","email":"takahito.hirose0518@gmail.com","username":"thirose"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},{"value":0,"date":"2019-05-16 06:00:47.000000000","_account_id":28140,"name":"Takahito Hirose","email":"takahito.hirose0518@gmail.com","username":"thirose"},{"value":2,"date":"2019-06-17 20:16:00.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},"all":[{"value":2,"date":"2019-05-09 09:51:29.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},{"value":0,"_account_id":28140,"name":"Takahito Hirose","email":"takahito.hirose0518@gmail.com","username":"thirose"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2019-05-09 00:15:27.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},"all":[{"value":1,"date":"2019-05-09 09:51:29.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},{"value":0,"_account_id":28140,"name":"Takahito Hirose","email":"takahito.hirose0518@gmail.com","username":"thirose"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},{"value":0,"_account_id":28140,"name":"Takahito Hirose","email":"takahito.hirose0518@gmail.com","username":"thirose"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},{"_account_id":28140,"name":"Takahito Hirose","email":"takahito.hirose0518@gmail.com","username":"thirose"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-05-09 00:15:27.000000000","updated_by":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"reviewer":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"state":"REVIEWER"},{"updated":"2019-05-09 09:51:29.000000000","updated_by":{"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},"reviewer":{"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},"state":"REVIEWER"},{"updated":"2019-06-17 20:16:00.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"e24b31c34fbbf19859e1c6fc3104d5cb9ca512cd","author":{"_account_id":28140,"name":"Takahito Hirose","email":"takahito.hirose0518@gmail.com","username":"thirose"},"date":"2019-04-08 02:29:22.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"fac8922b482849d13c7bf55457ef87e8060ecab0","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-08 03:44:42.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- designate-devstack-bind9 http://logs.openstack.org/05/650605/1/check/designate-devstack-bind9/ffcd1be/ : SUCCESS in 36m 01s\n- designate-devstack-pdns4 http://logs.openstack.org/05/650605/1/check/designate-devstack-pdns4/657ca1d/ : SUCCESS in 30m 07s\n- designate-devstack-pdns4-postgres http://logs.openstack.org/05/650605/1/check/designate-devstack-pdns4-postgres/7cf8bbe/ : SUCCESS in 33m 08s\n- designate-devstack-pdns4-manager-model http://logs.openstack.org/05/650605/1/check/designate-devstack-pdns4-manager-model/5c5456d/ : FAILURE in 46m 19s\n- designate-devstack-pdns4-py35 http://logs.openstack.org/05/650605/1/check/designate-devstack-pdns4-py35/88d9246/ : SUCCESS in 33m 41s\n- designate-grenade-pdns4 http://logs.openstack.org/05/650605/1/check/designate-grenade-pdns4/024f9d9/ : SUCCESS in 46m 38s\n- openstack-tox-pep8 http://logs.openstack.org/05/650605/1/check/openstack-tox-pep8/d3eb61b/ : SUCCESS in 5m 47s\n- openstack-tox-py27 http://logs.openstack.org/05/650605/1/check/openstack-tox-py27/44d742d/ : SUCCESS in 5m 42s\n- openstack-tox-py35 http://logs.openstack.org/05/650605/1/check/openstack-tox-py35/84b5cfd/ : SUCCESS in 6m 42s\n- build-openstack-sphinx-docs http://logs.openstack.org/05/650605/1/check/build-openstack-sphinx-docs/024de8f/html/ : SUCCESS in 5m 40s\n- neutron-tempest-plugin-designate-scenario http://logs.openstack.org/05/650605/1/check/neutron-tempest-plugin-designate-scenario/d7c7362/ : FAILURE in 26m 20s (non-voting)\n- openstack-tox-cover http://logs.openstack.org/05/650605/1/check/openstack-tox-cover/ae6f3d0/cover/ : SUCCESS in 6m 42s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"65f08a23ecdb88ddbc712a068f17c2caf7c99f15","author":{"_account_id":28140,"name":"Takahito Hirose","email":"takahito.hirose0518@gmail.com","username":"thirose"},"date":"2019-04-08 05:12:56.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"717d4a078314a2fa76dbfb5da82ea93e4f67d255","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-08 06:05:04.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- designate-devstack-bind9 http://logs.openstack.org/05/650605/1/check/designate-devstack-bind9/7473b3c/ : SUCCESS in 34m 50s\n- designate-devstack-pdns4 http://logs.openstack.org/05/650605/1/check/designate-devstack-pdns4/06d16da/ : SUCCESS in 31m 46s\n- designate-devstack-pdns4-postgres http://logs.openstack.org/05/650605/1/check/designate-devstack-pdns4-postgres/bacd1ae/ : SUCCESS in 27m 52s\n- designate-devstack-pdns4-manager-model http://logs.openstack.org/05/650605/1/check/designate-devstack-pdns4-manager-model/c9ec4b3/ : FAILURE in 37m 22s\n- designate-devstack-pdns4-py35 http://logs.openstack.org/05/650605/1/check/designate-devstack-pdns4-py35/5f9faa0/ : SUCCESS in 39m 12s\n- designate-grenade-pdns4 http://logs.openstack.org/05/650605/1/check/designate-grenade-pdns4/c9c3d13/ : SUCCESS in 43m 55s\n- openstack-tox-pep8 http://logs.openstack.org/05/650605/1/check/openstack-tox-pep8/8de9d00/ : SUCCESS in 5m 11s\n- openstack-tox-py27 http://logs.openstack.org/05/650605/1/check/openstack-tox-py27/8019bb6/ : SUCCESS in 5m 35s\n- openstack-tox-py35 http://logs.openstack.org/05/650605/1/check/openstack-tox-py35/a80b0cc/ : SUCCESS in 5m 50s\n- build-openstack-sphinx-docs http://logs.openstack.org/05/650605/1/check/build-openstack-sphinx-docs/957cdc6/html/ : SUCCESS in 6m 41s\n- neutron-tempest-plugin-designate-scenario http://logs.openstack.org/05/650605/1/check/neutron-tempest-plugin-designate-scenario/a6c2c2d/ : FAILURE in 25m 43s (non-voting)\n- openstack-tox-cover http://logs.openstack.org/05/650605/1/check/openstack-tox-cover/9325849/cover/ : SUCCESS in 5m 18s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"88046c82c88fd697e03669f941c41dabce6af3c1","author":{"_account_id":28140,"name":"Takahito Hirose","email":"takahito.hirose0518@gmail.com","username":"thirose"},"date":"2019-04-08 14:47:10.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"bd353338f130d3ac382ad9e24c48956ec07e20b4","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-08 15:45:01.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- designate-devstack-bind9 http://logs.openstack.org/05/650605/1/check/designate-devstack-bind9/2341667/ : SUCCESS in 41m 03s\n- designate-devstack-pdns4 http://logs.openstack.org/05/650605/1/check/designate-devstack-pdns4/a2ea015/ : SUCCESS in 34m 51s\n- designate-devstack-pdns4-postgres http://logs.openstack.org/05/650605/1/check/designate-devstack-pdns4-postgres/3d1fd2c/ : SUCCESS in 35m 42s\n- designate-devstack-pdns4-manager-model http://logs.openstack.org/05/650605/1/check/designate-devstack-pdns4-manager-model/7675b2c/ : SUCCESS in 37m 36s\n- designate-devstack-pdns4-py35 http://logs.openstack.org/05/650605/1/check/designate-devstack-pdns4-py35/23fe33b/ : SUCCESS in 34m 42s\n- designate-grenade-pdns4 http://logs.openstack.org/05/650605/1/check/designate-grenade-pdns4/80e967c/ : FAILURE in 53m 21s\n- openstack-tox-pep8 http://logs.openstack.org/05/650605/1/check/openstack-tox-pep8/8778aa1/ : SUCCESS in 7m 00s\n- openstack-tox-py27 http://logs.openstack.org/05/650605/1/check/openstack-tox-py27/a1e2532/ : SUCCESS in 7m 16s\n- openstack-tox-py35 http://logs.openstack.org/05/650605/1/check/openstack-tox-py35/9a4eeb9/ : SUCCESS in 7m 11s\n- build-openstack-sphinx-docs http://logs.openstack.org/05/650605/1/check/build-openstack-sphinx-docs/b008212/html/ : SUCCESS in 7m 53s\n- neutron-tempest-plugin-designate-scenario http://logs.openstack.org/05/650605/1/check/neutron-tempest-plugin-designate-scenario/0e97873/ : FAILURE in 33m 31s (non-voting)\n- openstack-tox-cover http://logs.openstack.org/05/650605/1/check/openstack-tox-cover/04a8880/cover/ : SUCCESS in 8m 01s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"ceda815bd737d4121804af2755eb4c9097402b8c","author":{"_account_id":28140,"name":"Takahito Hirose","email":"takahito.hirose0518@gmail.com","username":"thirose"},"date":"2019-04-11 15:36:41.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"5396973945ccc99d9b45351dd21c880173d990d6","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-11 16:43:56.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- designate-devstack-bind9 http://logs.openstack.org/05/650605/1/check/designate-devstack-bind9/16b643f/ : SUCCESS in 38m 55s\n- designate-devstack-pdns4 http://logs.openstack.org/05/650605/1/check/designate-devstack-pdns4/20834f6/ : SUCCESS in 35m 08s\n- designate-devstack-pdns4-postgres http://logs.openstack.org/05/650605/1/check/designate-devstack-pdns4-postgres/da4c511/ : SUCCESS in 36m 47s\n- designate-devstack-pdns4-manager-model http://logs.openstack.org/05/650605/1/check/designate-devstack-pdns4-manager-model/d4fe2c0/ : SUCCESS in 39m 59s\n- designate-devstack-pdns4-py35 http://logs.openstack.org/05/650605/1/check/designate-devstack-pdns4-py35/5fb783b/ : SUCCESS in 36m 59s\n- designate-grenade-pdns4 http://logs.openstack.org/05/650605/1/check/designate-grenade-pdns4/d27b997/ : SUCCESS in 1h 00m 28s\n- openstack-tox-pep8 http://logs.openstack.org/05/650605/1/check/openstack-tox-pep8/a570c64/ : SUCCESS in 5m 43s\n- openstack-tox-py27 http://logs.openstack.org/05/650605/1/check/openstack-tox-py27/b9eb5b0/ : SUCCESS in 6m 41s\n- openstack-tox-py35 http://logs.openstack.org/05/650605/1/check/openstack-tox-py35/8368c54/ : SUCCESS in 6m 00s\n- build-openstack-sphinx-docs http://logs.openstack.org/05/650605/1/check/build-openstack-sphinx-docs/5ef675d/html/ : SUCCESS in 5m 47s\n- neutron-tempest-plugin-designate-scenario http://logs.openstack.org/05/650605/1/check/neutron-tempest-plugin-designate-scenario/695a4b2/ : FAILURE in 27m 27s (non-voting)\n- openstack-tox-cover http://logs.openstack.org/05/650605/1/check/openstack-tox-cover/b60a331/cover/ : SUCCESS in 6m 24s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"0cd10a3becf616588ef0abccc388f5ff3c41250e","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2019-05-09 00:15:27.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"5388833f4d1933e153de3bd45e0122ef60228c57","author":{"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},"date":"2019-05-09 09:51:29.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"444f3e0933570d6847b016ab07c76c2e5722e15a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-09 09:51:42.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"b3caac9bc1c69b04e9f0fb54f4b0c1b7da95bd51","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-09 10:38:56.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- designate-devstack-bind9 http://logs.openstack.org/05/650605/1/gate/designate-devstack-bind9/360b97d/ : SUCCESS in 34m 06s\n- designate-devstack-pdns4 http://logs.openstack.org/05/650605/1/gate/designate-devstack-pdns4/7ce456f/ : SUCCESS in 38m 37s\n- designate-devstack-pdns4-postgres http://logs.openstack.org/05/650605/1/gate/designate-devstack-pdns4-postgres/ca1f96d/ : FAILURE in 38m 18s\n- designate-devstack-pdns4-manager-model http://logs.openstack.org/05/650605/1/gate/designate-devstack-pdns4-manager-model/77377b3/ : SUCCESS in 41m 53s\n- designate-grenade-pdns4 http://logs.openstack.org/05/650605/1/gate/designate-grenade-pdns4/5ff18c3/ : SUCCESS in 45m 50s\n- openstack-tox-pep8 http://logs.openstack.org/05/650605/1/gate/openstack-tox-pep8/71626d1/ : FAILURE in 5m 32s\n- openstack-tox-py27 http://logs.openstack.org/05/650605/1/gate/openstack-tox-py27/2d32e5c/ : SUCCESS in 10m 39s\n- openstack-tox-py35 http://logs.openstack.org/05/650605/1/gate/openstack-tox-py35/1fb144e/ : SUCCESS in 6m 02s\n- build-openstack-sphinx-docs http://logs.openstack.org/05/650605/1/gate/build-openstack-sphinx-docs/de32f33/html/ : SUCCESS in 6m 44s","accounts_in_message":[],"_revision_number":1},{"id":"14e8edd466c1b843a305fd34ebd788382b7a750e","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2019-05-09 23:36:01.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"c32d053bd0a99718e393a21d1c6bb69c01d2bddb","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-10 00:39:15.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- designate-devstack-bind9 http://logs.openstack.org/05/650605/1/check/designate-devstack-bind9/fcabd41/ : SUCCESS in 34m 54s\n- designate-devstack-pdns4 http://logs.openstack.org/05/650605/1/check/designate-devstack-pdns4/ce11c93/ : SUCCESS in 39m 24s\n- designate-devstack-pdns4-postgres http://logs.openstack.org/05/650605/1/check/designate-devstack-pdns4-postgres/7f66466/ : SUCCESS in 37m 20s\n- designate-devstack-pdns4-manager-model http://logs.openstack.org/05/650605/1/check/designate-devstack-pdns4-manager-model/4d04c82/ : FAILURE in 38m 06s\n- designate-devstack-pdns4-py35 http://logs.openstack.org/05/650605/1/check/designate-devstack-pdns4-py35/f24f1f7/ : SUCCESS in 41m 18s\n- designate-grenade-pdns4 http://logs.openstack.org/05/650605/1/check/designate-grenade-pdns4/52c6edf/ : SUCCESS in 57m 07s\n- openstack-tox-pep8 http://logs.openstack.org/05/650605/1/check/openstack-tox-pep8/5e08855/ : FAILURE in 5m 05s\n- openstack-tox-py27 http://logs.openstack.org/05/650605/1/check/openstack-tox-py27/244a84d/ : SUCCESS in 5m 18s\n- openstack-tox-py35 http://logs.openstack.org/05/650605/1/check/openstack-tox-py35/3921bc9/ : SUCCESS in 6m 20s\n- build-openstack-sphinx-docs http://logs.openstack.org/05/650605/1/check/build-openstack-sphinx-docs/d06b614/html/ : SUCCESS in 4m 53s\n- neutron-tempest-plugin-designate-scenario http://logs.openstack.org/05/650605/1/check/neutron-tempest-plugin-designate-scenario/8d4cd35/ : FAILURE in 27m 19s (non-voting)\n- openstack-tox-cover http://logs.openstack.org/05/650605/1/check/openstack-tox-cover/ddbb676/cover/ : SUCCESS in 6m 17s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"257da9cb56cc976ebdab224be2b56e899222d10c","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2019-05-10 02:41:00.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"7c9319de0bce37bdd6c620bfd272cfd8cbfa61e9","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-10 03:30:29.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- designate-devstack-bind9 http://logs.openstack.org/05/650605/1/check/designate-devstack-bind9/5a225ad/ : SUCCESS in 30m 22s\n- designate-devstack-pdns4 http://logs.openstack.org/05/650605/1/check/designate-devstack-pdns4/e09ddd0/ : SUCCESS in 48m 39s\n- designate-devstack-pdns4-postgres http://logs.openstack.org/05/650605/1/check/designate-devstack-pdns4-postgres/f546d5a/ : SUCCESS in 31m 07s\n- designate-devstack-pdns4-manager-model http://logs.openstack.org/05/650605/1/check/designate-devstack-pdns4-manager-model/75f570b/ : SUCCESS in 35m 33s\n- designate-devstack-pdns4-py35 http://logs.openstack.org/05/650605/1/check/designate-devstack-pdns4-py35/4f68763/ : SUCCESS in 38m 14s\n- designate-grenade-pdns4 http://logs.openstack.org/05/650605/1/check/designate-grenade-pdns4/2a840d2/ : SUCCESS in 44m 33s\n- openstack-tox-pep8 http://logs.openstack.org/05/650605/1/check/openstack-tox-pep8/ffe7c84/ : FAILURE in 7m 14s\n- openstack-tox-py27 http://logs.openstack.org/05/650605/1/check/openstack-tox-py27/a8feaaf/ : SUCCESS in 5m 38s\n- openstack-tox-py35 http://logs.openstack.org/05/650605/1/check/openstack-tox-py35/42e4954/ : SUCCESS in 7m 58s\n- build-openstack-sphinx-docs http://logs.openstack.org/05/650605/1/check/build-openstack-sphinx-docs/59c8663/html/ : SUCCESS in 5m 31s\n- neutron-tempest-plugin-designate-scenario http://logs.openstack.org/05/650605/1/check/neutron-tempest-plugin-designate-scenario/3553306/ : FAILURE in 30m 48s (non-voting)\n- openstack-tox-cover http://logs.openstack.org/05/650605/1/check/openstack-tox-cover/26cbcfe/cover/ : SUCCESS in 6m 24s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"b4e9eb81eed03f31864f34b4e6d2f74869155258","author":{"_account_id":28140,"name":"Takahito Hirose","email":"takahito.hirose0518@gmail.com","username":"thirose"},"date":"2019-05-16 06:00:47.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"b96bc91c0ee7da0426de4c51b6d5cf2c5d1b7f8f","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-16 07:06:50.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- designate-devstack-bind9 http://logs.openstack.org/05/650605/1/check/designate-devstack-bind9/cc11f15/ : SUCCESS in 30m 57s\n- designate-devstack-pdns4 http://logs.openstack.org/05/650605/1/check/designate-devstack-pdns4/5cbf509/ : SUCCESS in 32m 24s\n- designate-devstack-pdns4-postgres http://logs.openstack.org/05/650605/1/check/designate-devstack-pdns4-postgres/be3728b/ : SUCCESS in 30m 34s\n- designate-devstack-pdns4-manager-model http://logs.openstack.org/05/650605/1/check/designate-devstack-pdns4-manager-model/63e45d2/ : SUCCESS in 33m 41s\n- designate-devstack-pdns4-py35 http://logs.openstack.org/05/650605/1/check/designate-devstack-pdns4-py35/4c19af4/ : SUCCESS in 44m 57s\n- designate-grenade-pdns4 http://logs.openstack.org/05/650605/1/check/designate-grenade-pdns4/185f92f/ : SUCCESS in 1h 00m 08s\n- openstack-tox-pep8 http://logs.openstack.org/05/650605/1/check/openstack-tox-pep8/d17831e/ : FAILURE in 6m 36s\n- openstack-tox-py27 http://logs.openstack.org/05/650605/1/check/openstack-tox-py27/6a1090c/ : SUCCESS in 6m 47s\n- openstack-tox-py35 http://logs.openstack.org/05/650605/1/check/openstack-tox-py35/aa8bb22/ : SUCCESS in 7m 04s\n- build-openstack-sphinx-docs http://logs.openstack.org/05/650605/1/check/build-openstack-sphinx-docs/21c6658/html/ : SUCCESS in 7m 01s\n- neutron-tempest-plugin-designate-scenario http://logs.openstack.org/05/650605/1/check/neutron-tempest-plugin-designate-scenario/38fed3a/ : FAILURE in 29m 03s (non-voting)\n- openstack-tox-cover http://logs.openstack.org/05/650605/1/check/openstack-tox-cover/cf2ca70/cover/ : SUCCESS in 8m 13s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"da55cb1eac6709b923dc6fb3f40df7b6f2d6f604","author":{"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},"date":"2019-05-17 17:24:59.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"b23aaadac62bc22c565a9d8e297c74e8142e9dfb","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-17 18:30:32.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- designate-devstack-bind9 http://logs.openstack.org/05/650605/1/check/designate-devstack-bind9/3f758ac/ : SUCCESS in 36m 45s\n- designate-devstack-pdns4 http://logs.openstack.org/05/650605/1/check/designate-devstack-pdns4/c84141d/ : SUCCESS in 40m 28s\n- designate-devstack-pdns4-postgres http://logs.openstack.org/05/650605/1/check/designate-devstack-pdns4-postgres/3dc9de8/ : SUCCESS in 36m 10s\n- designate-devstack-pdns4-manager-model http://logs.openstack.org/05/650605/1/check/designate-devstack-pdns4-manager-model/4e35b69/ : FAILURE in 41m 21s\n- designate-devstack-pdns4-py35 http://logs.openstack.org/05/650605/1/check/designate-devstack-pdns4-py35/0d56596/ : SUCCESS in 39m 18s\n- designate-grenade-pdns4 http://logs.openstack.org/05/650605/1/check/designate-grenade-pdns4/4f03cc6/ : SUCCESS in 59m 26s\n- openstack-tox-pep8 http://logs.openstack.org/05/650605/1/check/openstack-tox-pep8/fa4027a/ : FAILURE in 6m 05s\n- openstack-tox-py27 http://logs.openstack.org/05/650605/1/check/openstack-tox-py27/8d6916d/ : SUCCESS in 5m 18s\n- openstack-tox-py35 http://logs.openstack.org/05/650605/1/check/openstack-tox-py35/3d23086/ : SUCCESS in 7m 57s\n- build-openstack-sphinx-docs http://logs.openstack.org/05/650605/1/check/build-openstack-sphinx-docs/06a6ec5/html/ : SUCCESS in 5m 40s\n- neutron-tempest-plugin-designate-scenario http://logs.openstack.org/05/650605/1/check/neutron-tempest-plugin-designate-scenario/5daa94b/ : FAILURE in 25m 55s (non-voting)\n- openstack-tox-cover http://logs.openstack.org/05/650605/1/check/openstack-tox-cover/b475974/cover/ : SUCCESS in 5m 39s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"409b0b358b8ea78c0dc174e64dfd7d10dc7ee6af","author":{"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},"date":"2019-05-21 13:07:59.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"d5c6b2d9252e18fbaa223371764e900c941129af","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-21 14:14:09.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- designate-devstack-bind9 http://logs.openstack.org/05/650605/1/check/designate-devstack-bind9/0baa777/ : SUCCESS in 36m 21s\n- designate-devstack-pdns4 http://logs.openstack.org/05/650605/1/check/designate-devstack-pdns4/ddd295a/ : SUCCESS in 35m 18s\n- designate-devstack-pdns4-postgres http://logs.openstack.org/05/650605/1/check/designate-devstack-pdns4-postgres/83e047d/ : SUCCESS in 36m 03s\n- designate-devstack-pdns4-manager-model http://logs.openstack.org/05/650605/1/check/designate-devstack-pdns4-manager-model/ce5e087/ : SUCCESS in 42m 13s\n- designate-devstack-pdns4-py35 http://logs.openstack.org/05/650605/1/check/designate-devstack-pdns4-py35/8edbf0a/ : SUCCESS in 37m 55s\n- designate-grenade-pdns4 http://logs.openstack.org/05/650605/1/check/designate-grenade-pdns4/e0542b9/ : SUCCESS in 1h 00m 23s\n- openstack-tox-pep8 http://logs.openstack.org/05/650605/1/check/openstack-tox-pep8/bc8d7b1/ : FAILURE in 5m 14s\n- openstack-tox-py27 http://logs.openstack.org/05/650605/1/check/openstack-tox-py27/3238b96/ : SUCCESS in 5m 36s\n- openstack-tox-py35 http://logs.openstack.org/05/650605/1/check/openstack-tox-py35/c581c22/ : SUCCESS in 5m 41s\n- build-openstack-sphinx-docs http://logs.openstack.org/05/650605/1/check/build-openstack-sphinx-docs/7c4deb8/html/ : SUCCESS in 5m 30s\n- neutron-tempest-plugin-designate-scenario http://logs.openstack.org/05/650605/1/check/neutron-tempest-plugin-designate-scenario/1ae264c/ : FAILURE in 22m 58s (non-voting)\n- openstack-tox-cover http://logs.openstack.org/05/650605/1/check/openstack-tox-cover/ef694c3/cover/ : SUCCESS in 6m 00s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"763f3fb4964df15b94d3d482b1014ed094cde9ef","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2019-05-21 18:20:35.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"44b9f134ecd0f5d0cb809816449cc662d0ab0e86","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-21 19:34:49.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- designate-devstack-bind9 http://logs.openstack.org/05/650605/1/check/designate-devstack-bind9/20ef7cb/ : SUCCESS in 32m 10s\n- designate-devstack-pdns4 http://logs.openstack.org/05/650605/1/check/designate-devstack-pdns4/a5127d0/ : SUCCESS in 37m 59s\n- designate-devstack-pdns4-postgres http://logs.openstack.org/05/650605/1/check/designate-devstack-pdns4-postgres/7972a62/ : SUCCESS in 35m 25s\n- designate-devstack-pdns4-manager-model http://logs.openstack.org/05/650605/1/check/designate-devstack-pdns4-manager-model/5c383f9/ : SUCCESS in 42m 00s\n- designate-devstack-pdns4-py35 http://logs.openstack.org/05/650605/1/check/designate-devstack-pdns4-py35/e685fc4/ : SUCCESS in 40m 15s\n- designate-grenade-pdns4 http://logs.openstack.org/05/650605/1/check/designate-grenade-pdns4/6f07a2b/ : SUCCESS in 56m 21s\n- openstack-tox-pep8 http://logs.openstack.org/05/650605/1/check/openstack-tox-pep8/164359a/ : FAILURE in 4m 54s\n- openstack-tox-py27 http://logs.openstack.org/05/650605/1/check/openstack-tox-py27/4c62ed0/ : SUCCESS in 5m 34s\n- openstack-tox-py35 http://logs.openstack.org/05/650605/1/check/openstack-tox-py35/d51320b/ : SUCCESS in 5m 29s\n- build-openstack-sphinx-docs http://logs.openstack.org/05/650605/1/check/build-openstack-sphinx-docs/691027c/html/ : SUCCESS in 5m 16s\n- neutron-tempest-plugin-designate-scenario http://logs.openstack.org/05/650605/1/check/neutron-tempest-plugin-designate-scenario/b20b424/ : FAILURE in 24m 59s (non-voting)\n- openstack-tox-cover http://logs.openstack.org/05/650605/1/check/openstack-tox-cover/07e5c05/cover/ : SUCCESS in 6m 07s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"57bdad40e09576e9b91007dc0b56e69931ad3fef","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2019-06-17 04:49:03.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"2c7882b837c7e241b492452a7b9d67c2e3e7151a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-17 05:50:11.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- designate-devstack-bind9 http://logs.openstack.org/05/650605/1/check/designate-devstack-bind9/0710ec9/ : SUCCESS in 37m 49s\n- designate-devstack-pdns4 http://logs.openstack.org/05/650605/1/check/designate-devstack-pdns4/2d8eb2d/ : SUCCESS in 34m 48s\n- designate-devstack-pdns4-postgres http://logs.openstack.org/05/650605/1/check/designate-devstack-pdns4-postgres/b258c35/ : SUCCESS in 30m 39s\n- designate-devstack-pdns4-manager-model http://logs.openstack.org/05/650605/1/check/designate-devstack-pdns4-manager-model/7642128/ : FAILURE in 32m 19s\n- designate-devstack-pdns4-py35 http://logs.openstack.org/05/650605/1/check/designate-devstack-pdns4-py35/d7f1197/ : SUCCESS in 40m 40s\n- designate-grenade-pdns4 http://logs.openstack.org/05/650605/1/check/designate-grenade-pdns4/ec4f5b0/ : FAILURE in 1h 00m 20s\n- openstack-tox-pep8 http://logs.openstack.org/05/650605/1/check/openstack-tox-pep8/ac5d039/ : SUCCESS in 5m 36s\n- openstack-tox-py27 http://logs.openstack.org/05/650605/1/check/openstack-tox-py27/3000930/ : SUCCESS in 8m 36s\n- openstack-tox-py35 http://logs.openstack.org/05/650605/1/check/openstack-tox-py35/ee06c7b/ : SUCCESS in 5m 34s\n- build-openstack-sphinx-docs http://logs.openstack.org/05/650605/1/check/build-openstack-sphinx-docs/03b31c5/html/ : SUCCESS in 5m 12s\n- neutron-tempest-plugin-designate-scenario http://logs.openstack.org/05/650605/1/check/neutron-tempest-plugin-designate-scenario/824287f/ : FAILURE in 28m 54s (non-voting)\n- openstack-tox-cover http://logs.openstack.org/05/650605/1/check/openstack-tox-cover/c075651/cover/ : SUCCESS in 5m 52s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"dc7d05bdfdf723bdd1cfd0d52aeaf780782dab69","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2019-06-17 16:42:25.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"540556d22211144921c4eded2965ecf499dbf777","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-17 17:50:09.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- designate-devstack-bind9 http://logs.openstack.org/05/650605/1/check/designate-devstack-bind9/025656d/ : SUCCESS in 33m 13s\n- designate-devstack-pdns4 http://logs.openstack.org/05/650605/1/check/designate-devstack-pdns4/4e47635/ : SUCCESS in 32m 08s\n- designate-devstack-pdns4-postgres http://logs.openstack.org/05/650605/1/check/designate-devstack-pdns4-postgres/9a3ab61/ : SUCCESS in 36m 31s\n- designate-devstack-pdns4-manager-model http://logs.openstack.org/05/650605/1/check/designate-devstack-pdns4-manager-model/cedfba4/ : SUCCESS in 43m 13s\n- designate-devstack-pdns4-py35 http://logs.openstack.org/05/650605/1/check/designate-devstack-pdns4-py35/96d817e/ : SUCCESS in 40m 07s\n- designate-grenade-pdns4 http://logs.openstack.org/05/650605/1/check/designate-grenade-pdns4/a6fe163/ : FAILURE in 59m 46s\n- openstack-tox-pep8 http://logs.openstack.org/05/650605/1/check/openstack-tox-pep8/1841252/ : SUCCESS in 5m 34s\n- openstack-tox-py27 http://logs.openstack.org/05/650605/1/check/openstack-tox-py27/a46c136/ : SUCCESS in 6m 21s\n- openstack-tox-py35 http://logs.openstack.org/05/650605/1/check/openstack-tox-py35/51878f2/ : SUCCESS in 5m 45s\n- build-openstack-sphinx-docs http://logs.openstack.org/05/650605/1/check/build-openstack-sphinx-docs/abddea9/html/ : SUCCESS in 5m 30s\n- neutron-tempest-plugin-designate-scenario http://logs.openstack.org/05/650605/1/check/neutron-tempest-plugin-designate-scenario/3cea780/ : FAILURE in 28m 35s (non-voting)\n- openstack-tox-cover http://logs.openstack.org/05/650605/1/check/openstack-tox-cover/1cf3f77/cover/ : SUCCESS in 6m 03s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"01d33105926e23f04fe769e3803571ce2424dfb5","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2019-06-17 17:51:08.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"d2234567aa4aeae21cab8889dea5a5539d2310d1","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-17 18:56:20.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- designate-devstack-bind9 http://logs.openstack.org/05/650605/1/check/designate-devstack-bind9/304a991/ : SUCCESS in 37m 19s\n- designate-devstack-pdns4 http://logs.openstack.org/05/650605/1/check/designate-devstack-pdns4/de716b2/ : SUCCESS in 39m 04s\n- designate-devstack-pdns4-postgres http://logs.openstack.org/05/650605/1/check/designate-devstack-pdns4-postgres/18960d5/ : SUCCESS in 38m 37s\n- designate-devstack-pdns4-manager-model http://logs.openstack.org/05/650605/1/check/designate-devstack-pdns4-manager-model/479dda9/ : SUCCESS in 43m 28s\n- designate-devstack-pdns4-py35 http://logs.openstack.org/05/650605/1/check/designate-devstack-pdns4-py35/c073a28/ : SUCCESS in 42m 44s\n- designate-grenade-pdns4 http://logs.openstack.org/05/650605/1/check/designate-grenade-pdns4/95d7810/ : SUCCESS in 56m 29s\n- openstack-tox-pep8 http://logs.openstack.org/05/650605/1/check/openstack-tox-pep8/7043429/ : SUCCESS in 5m 37s\n- openstack-tox-py27 http://logs.openstack.org/05/650605/1/check/openstack-tox-py27/5402b0f/ : SUCCESS in 6m 10s\n- openstack-tox-py35 http://logs.openstack.org/05/650605/1/check/openstack-tox-py35/b849b5f/ : SUCCESS in 6m 17s\n- build-openstack-sphinx-docs http://logs.openstack.org/05/650605/1/check/build-openstack-sphinx-docs/713554e/html/ : SUCCESS in 6m 06s\n- neutron-tempest-plugin-designate-scenario http://logs.openstack.org/05/650605/1/check/neutron-tempest-plugin-designate-scenario/ab21d7b/ : FAILURE in 30m 37s (non-voting)\n- openstack-tox-cover http://logs.openstack.org/05/650605/1/check/openstack-tox-cover/7e08a8f/cover/ : SUCCESS in 7m 03s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"4f2460cfb380da6c9d18d548bf0686f33770cf2a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-17 18:56:33.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"3d52978a0d21c8971c2221c7e49fb86e8a047da2","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-17 20:16:00.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"c99814438a2ea7fa521a6ae2265d2efa0db25016","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-17 20:16:00.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- designate-devstack-bind9 http://logs.openstack.org/05/650605/1/gate/designate-devstack-bind9/629b6d5/ : SUCCESS in 40m 10s\n- designate-devstack-pdns4 http://logs.openstack.org/05/650605/1/gate/designate-devstack-pdns4/3b08c6e/ : SUCCESS in 33m 16s\n- designate-devstack-pdns4-postgres http://logs.openstack.org/05/650605/1/gate/designate-devstack-pdns4-postgres/8a1ad13/ : SUCCESS in 39m 54s\n- designate-devstack-pdns4-manager-model http://logs.openstack.org/05/650605/1/gate/designate-devstack-pdns4-manager-model/3c9f2a5/ : SUCCESS in 39m 05s\n- designate-grenade-pdns4 http://logs.openstack.org/05/650605/1/gate/designate-grenade-pdns4/db94c9a/ : SUCCESS in 1h 05m 21s\n- openstack-tox-pep8 http://logs.openstack.org/05/650605/1/gate/openstack-tox-pep8/e902795/ : SUCCESS in 5m 31s\n- openstack-tox-py27 http://logs.openstack.org/05/650605/1/gate/openstack-tox-py27/a31a030/ : SUCCESS in 5m 59s\n- openstack-tox-py35 http://logs.openstack.org/05/650605/1/gate/openstack-tox-py35/e81d9c7/ : SUCCESS in 6m 30s\n- build-openstack-sphinx-docs http://logs.openstack.org/05/650605/1/gate/build-openstack-sphinx-docs/5b275f1/html/ : SUCCESS in 5m 49s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"00d0cb71e63e870fdf09833eec02f29b95542b18","revisions":{"00d0cb71e63e870fdf09833eec02f29b95542b18":{"kind":"REWORK","_number":1,"created":"2019-04-08 02:29:22.000000000","uploader":{"_account_id":28140,"name":"Takahito Hirose","email":"takahito.hirose0518@gmail.com","username":"thirose"},"ref":"refs/changes/05/650605/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/05/650605/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/05/650605/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/05/650605/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/05/650605/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/05/650605/1"}}},"commit":{"parents":[{"commit":"59c5c664654b88772a6309fbb319e446423317c0","subject":"Make update_status use synchronized_zone","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/59c5c664654b88772a6309fbb319e446423317c0"}]}],"author":{"name":"Takahito Hirose","email":"takahito.hirose0518@gmail.com","date":"2019-03-26 10:52:33.000000000","tz":540},"committer":{"name":"Takahito Hirose","email":"takahito.hirose0518@gmail.com","date":"2019-04-08 02:29:22.000000000","tz":0},"subject":"Fix DBDeadLock error resulting into 500","message":"Fix DBDeadLock error resulting into 500\n\nWhen user requests the record registration request continuously,\nsometimes designate hits DBDeadLock resuting into 500 InternalServerError.\n\nWe get below error:\n\n2019-02-21 21:30:39.925 49752 ERROR designate.api.middleware RemoteError:\nRemote error: DBDeadlock (pymysql.err.InternalError)\n(1213, u\u0027Deadlock found when trying to get lock; try restarting transaction\u0027)\n[SQL: u\u0027UPDATE records SET version\u003d(records.version + %(version_1)s),\nupdated_at\u003d%(updated_at)s, data\u003d%(data)s, hash\u003d%(hash)s, status\u003d%(status)s,\naction\u003d%(action)s, serial\u003d%(serial)s WHERE records.id \u003d %(id_1)s\u0027]\n[parameters: {\u0027status\u0027: \u0027PENDING\u0027, \u0027hash\u0027: \u002739795ee18c6e3c9ad1c0190c6a3d8d4f\u0027,\n\u0027updated_at\u0027: datetime.datetime(2019, 2, 21, 12, 30, 39, 909846), u\u0027version_1\u0027: 1,\nu\u0027id_1\u0027: \u00277a655eeda4d446cdaa81caf19ab55fcc\u0027, \u0027action\u0027: \u0027UPDATE\u0027,\n\u0027serial\u0027: 1550752338,\n\u0027data\u0027: u\u0027ns2.example.jp. domain.example.com. 1550752338 3552 600 86400 3600\u0027}]\n\nIn the process of record registeration, designate first tried to update\nthe reocrd and then update the zone status.\n\nUpdating the zone_status and registering the record process[1] and after synced\nupdate record_status and zone_status process[2] are in reverse order. So If user\nrequest the registering record many time and same time, Designate will get the\nDBDeadLock, when these processes run the same time.\n\nWe observed that changing the order of the operations solves this issue.\n\n[1] https://github.com/openstack/designate/blob/master/designate/central/service.py#L1292-L1320\n[2] https://github.com/openstack/designate/blob/master/designate/central/service.py#L2310-L2322\n\n1. transaction [1]-1 updating zone status process \u003c- run    ---\u003e table_name-zone\n2. transaction [2]-1 updating record status process \u003c- run  ---\u003e table_name-record\n3. transaction [1]-2 registering record process \u003c- run and wait  ---\u003e table_name-record\n4. transaction [2]-2 updating zone process \u003c-deadlock!           ---\u003e table_name-zone\n\nChange-Id: Icd6e690ac84a2fe0db0f4a8a513de47f7916f5ea\nRelated-Bug: #1785459\n(cherry picked from commit f828654a3d40476cac7eb24a09a36e9978c2d708)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/00d0cb71e63e870fdf09833eec02f29b95542b18"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/00d0cb71e63e870fdf09833eec02f29b95542b18"}]},"branch":"refs/heads/stable/queens"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
