)]}'
{"id":"openstack%2Fdesignate~654997","triplet_id":"openstack%2Fdesignate~master~I6e5ee791775198b9f5feaf61534f0d18e5c78ce3","project":"openstack/designate","branch":"master","hashtags":[],"change_id":"I6e5ee791775198b9f5feaf61534f0d18e5c78ce3","subject":"Replace git.openstack.org URLs with opendev.org URLs","status":"ABANDONED","created":"2019-04-23 11:37:44.000000000","updated":"2019-04-26 10:15:14.000000000","total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"d9de2e0e036b32ca3446b19b2be3848cb596e07a","_number":654997,"virtual_id_number":654997,"owner":{"_account_id":22165,"name":"caoyuan","email":"cao.yingjunz@gmail.com","username":"caoyuan"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":-1,"date":"2019-04-23 15:39:47.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"disliked":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"all":[{"value":-1,"date":"2019-04-24 03:33:43.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":-1,"default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-04-23 15:39:47.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-04-24 03:33:43.000000000","updated_by":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"reviewer":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"state":"REVIEWER"}],"messages":[{"id":"81f551dcbc6e4a293ec004c4dfcffd673abae0fd","author":{"_account_id":22165,"name":"caoyuan","email":"cao.yingjunz@gmail.com","username":"caoyuan"},"date":"2019-04-23 11:37:44.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"73cff7faab557f15b401fd933edd2dd6ca4889ec","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-23 15:39:47.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- test-release-openstack http://logs.openstack.org/97/654997/1/check/test-release-openstack/49d6933/ : SUCCESS in 5m 20s\n- designate-bind9 http://logs.openstack.org/97/654997/1/check/designate-bind9/2094630/ : SUCCESS in 43m 30s\n- designate-pdns4 http://logs.openstack.org/97/654997/1/check/designate-pdns4/258f047/ : SUCCESS in 35m 05s\n- designate-bind9-manager-model finger://ze11.openstack.org/5be262a649a34e8284ef6cfc3e083808 : RETRY_LIMIT in 42m 53s\n- designate-bind9-py27 http://logs.openstack.org/97/654997/1/check/designate-bind9-py27/8879aab/ : SUCCESS in 46m 38s\n- designate-pdns4-py27 http://logs.openstack.org/97/654997/1/check/designate-pdns4-py27/ef684e0/ : POST_FAILURE in 44m 14s\n- designate-pdns4-postgres http://logs.openstack.org/97/654997/1/check/designate-pdns4-postgres/971f338/ : POST_FAILURE in 44m 00s\n- designate-grenade-pdns4 finger://ze11.openstack.org/205be6ad2dd0476fa5ef7514e6331851 : RETRY_LIMIT in 42m 15s\n- openstack-tox-cover http://logs.openstack.org/97/654997/1/check/openstack-tox-cover/73a1b7d/cover/ : SUCCESS in 7m 11s\n- openstack-tox-lower-constraints http://logs.openstack.org/97/654997/1/check/openstack-tox-lower-constraints/8a86f59/ : SUCCESS in 6m 00s\n- openstack-tox-pep8 http://logs.openstack.org/97/654997/1/check/openstack-tox-pep8/ea918f0/ : SUCCESS in 6m 49s\n- openstack-tox-py27 http://logs.openstack.org/97/654997/1/check/openstack-tox-py27/852fc5e/ : SUCCESS in 7m 09s\n- openstack-tox-py35 http://logs.openstack.org/97/654997/1/check/openstack-tox-py35/8e9d348/ : SUCCESS in 6m 03s\n- openstack-tox-py36 http://logs.openstack.org/97/654997/1/check/openstack-tox-py36/6ca8aa9/ : SUCCESS in 7m 37s\n- openstack-tox-py37 http://logs.openstack.org/97/654997/1/check/openstack-tox-py37/05c0e4d/ : SUCCESS in 6m 12s\n- openstack-tox-docs http://logs.openstack.org/97/654997/1/check/openstack-tox-docs/3f69a35/html/ : SUCCESS in 7m 10s\n- build-openstack-releasenotes http://logs.openstack.org/97/654997/1/check/build-openstack-releasenotes/c95ddec/html/ : SUCCESS in 7m 39s","accounts_in_message":[],"_revision_number":1},{"id":"3aa6d340384373ee207d2cb4634d6c076463a8fb","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2019-04-24 03:33:43.000000000","message":"Patch Set 1: Code-Review-1\n\n(3 comments)","accounts_in_message":[],"_revision_number":1},{"id":"d9de2e0e036b32ca3446b19b2be3848cb596e07a","author":{"_account_id":22165,"name":"caoyuan","email":"cao.yingjunz@gmail.com","username":"caoyuan"},"date":"2019-04-26 10:15:14.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"87ed91275cd401e749e4be5060bfd72de45afb8b","revisions":{"87ed91275cd401e749e4be5060bfd72de45afb8b":{"kind":"REWORK","_number":1,"created":"2019-04-23 11:37:44.000000000","uploader":{"_account_id":22165,"name":"caoyuan","email":"cao.yingjunz@gmail.com","username":"caoyuan"},"ref":"refs/changes/97/654997/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/97/654997/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/97/654997/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/97/654997/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/97/654997/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/97/654997/1"}}},"commit":{"parents":[{"commit":"fffdf29a3c409b5c5df31540d3cedb27fe4ae074","subject":"OpenDev Migration Patch","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/fffdf29a3c409b5c5df31540d3cedb27fe4ae074"}]}],"author":{"name":"caoyuan","email":"cao.yuan@99cloud.net","date":"2019-04-23 11:37:14.000000000","tz":480},"committer":{"name":"caoyuan","email":"cao.yuan@99cloud.net","date":"2019-04-23 11:37:40.000000000","tz":480},"subject":"Replace git.openstack.org URLs with opendev.org URLs","message":"Replace git.openstack.org URLs with opendev.org URLs\n\nChange-Id: I6e5ee791775198b9f5feaf61534f0d18e5c78ce3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/87ed91275cd401e749e4be5060bfd72de45afb8b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/87ed91275cd401e749e4be5060bfd72de45afb8b"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
